)]}'
{"id":"openstack%2Fkolla~974312","triplet_id":"openstack%2Fkolla~master~I6e30b749488a0ef2f247adf182db40cda8fa8ca4","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2026-01-22 15:08:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6e30b749488a0ef2f247adf182db40cda8fa8ca4","subject":"Fix ironic-pxe syslinux-tftpboot on aarch64 Rocky Linux","status":"MERGED","created":"2026-01-22 11:56:16.000000000","updated":"2026-01-22 15:11:06.000000000","submitted":"2026-01-22 15:08:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"974312","meta_rev_id":"a71339ec949bd2273682863de42f785a4aeb58fa","_number":974312,"virtual_id_number":974312,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-22 15:08:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2026-01-22 13:35:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2026-01-22 12:04:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2026-01-22 13:35:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2026-01-22 13:35:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-22 12:04:07.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-01-22 13:28:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-22 13:35:37.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"57f6cf793cabd9f46ba6e2a6117e8eb267a007ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-01-22 11:56:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0082af21af5691fb863a9423844aaf76f07f23f1","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-01-22 12:04:07.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"4a30b8dfe0dd9f4cf86b34eb88a716ac67b4d23e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 13:28:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21410a7b9194486dba917161208ba030\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/7b0211b2ebca4e65b08b72c2e19d7175 : SUCCESS in 41m 58s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/48fded4291a24441b26bd7a49904f0e7 : SUCCESS in 33m 49s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/ffbce22fa7b74121b78b288d74fd6c71 : SUCCESS in 13m 04s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/fb9b66362f03422b991e79756859cf1e : SUCCESS in 41m 16s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/ba0e1b91231e46b882a7843223b32aa2 : SUCCESS in 35m 09s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8afd1ac39e324aa187c36c14127c94d0 : SUCCESS in 35m 27s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/41ec4d3046f24bd4adbef0b5ed78d43b : SUCCESS in 31m 51s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/10ff39f77f21437fbd7899fe0a04ca9a : SUCCESS in 42m 04s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/6460907f29bc47b1af310e8a393dec92 : SUCCESS in 1h 02m 43s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/cd2c1786b42041f8b9792224489a55b0 : SUCCESS in 1h 11m 40s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/7205cf57ea434f40ae45fe16962d1123 : SUCCESS in 1h 09m 19s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/187ff67092d44c44a4d1805cbc72c4a3 : SUCCESS in 1h 29m 58s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6813b3acba6a41c592d18fc9005272ca : SUCCESS in 1h 04m 54s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/01173e9dc2434b819702fe7e8a0e4abd : SUCCESS in 1h 16m 01s\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/2ca40eaf99af4da7ba275078cf4c0bf0 : FAILURE in 1h 15m 31s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/30d46038b37f4e2f8fb4cdd55684564b : FAILURE in 1h 13m 14s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/3b68238a46d14aa2b11e87266ed86b92 : SUCCESS in 2m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d01cf4c599f4c3f8529135184bfe2fc : SUCCESS in 2m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/19f9d677b8f04ff98130f3d1d4056d34 : SUCCESS in 3m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/01fcd6b931ba40e1a97b5a533fac2436 : SUCCESS in 3m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d48fb1bc231c4662808bdcbcd2d7f012 : SUCCESS in 2m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/304c33f2a1ab445aa5bfa32f5350536b : SUCCESS in 3m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/61a7a0325d604ed084808675eea0067c : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":1},{"id":"a72c1d8bc93af04c6821da3d36c1cbb1a11983f1","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-01-22 13:35:37.000000000","message":"Patch Set 1: Backport-Candidate+1 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6133c41b9a80ce49b99e38680d72d065776b1ade","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 13:35:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2968d7a5a51c17a882e0df22282cf563be00e6d7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 14:29:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6fa770dd3704bfa80b764cbb626d605\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/b4d6f4d6d77d45faa0f79abcf04ddde1 : SUCCESS in 1h 46m 54s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/6e5435c04e1c457aad8908d49a6fbecc : SUCCESS in 2h 00m 45s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/ee51a6c55c7f4d7985a0a02a31e02326 : SUCCESS in 1h 17m 27s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/521e805864b14f9584c9a06d6a603874 : FAILURE in 7m 43s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/34bfea1ec1b8472aa58ea02af83d1e48 : FAILURE in 2h 07m 46s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2240323220e54b5c913a224371c507e9 : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/dead88aa0af14c46ad6d7ec59b9e2f4f : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2018e19138ab477fa57750f98ed4f9db : RETRY_LIMIT in 2m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dac616d2a0800e327589e4d5f692ddb51abf0086","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 15:08:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5dc2cd26338408cb668a823d622fa80\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/9f0c6d2b0a6b4058a132482455caef12 : SUCCESS in 49m 12s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/932ad9ef7a6d4a4d801d4e6d08113f3b : SUCCESS in 30m 28s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/b6e3a8bc494941a88b9c44d493eb3ef1 : SUCCESS in 43m 18s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/b42628e8307b417c80c2bab428675761 : SUCCESS in 41m 33s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/28d8b184051f4771a486f1a67adb2977 : SUCCESS in 39m 04s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b618e74b37f74e269d1a1017770b99fc : SUCCESS in 28m 00s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/bacc7c2743854a24bcc9e649579b27a1 : SUCCESS in 1h 02m 05s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/5c6e8a40bbed49718f8ce26b1f0dbe28 : SUCCESS in 1h 08m 07s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/d461693e85c04b05b9cb946eb0ab9241 : SUCCESS in 1h 07m 25s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/566a488fac3c431ab8badb2c5aeda045 : SUCCESS in 1h 31m 56s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fdcbc419445144909071438d73ef120b : SUCCESS in 1h 03m 37s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/efc0f373a8cc488fbfb17ae2ee632d81 : SUCCESS in 1h 29m 10s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/f24867d252b8464d90f5f6f9031e0137 : SUCCESS in 3m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/591c3389da4c4b679915a72921698c77 : SUCCESS in 2m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/039c2387e2a74c98b59042c11aa21f88 : SUCCESS in 2m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4cd2f98463bd46f39c6ea9d438f495fd : SUCCESS in 3m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4a7b5b180cb46a7a3b1fb2126aa95ea : SUCCESS in 4m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/707a1ed19c06444babcc93ad50b930bc : SUCCESS in 2m 46s","accounts_in_message":[],"_revision_number":1},{"id":"5a1e3c71b4a1a7835bade1ecd64ffc371e8ba660","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 15:08:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a71339ec949bd2273682863de42f785a4aeb58fa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 15:11:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7db198e8e5894b56b0dc47af5dfc0215\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ef5e0cc97b14dccb5f84721ed9ecace : SUCCESS in 1m 20s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e205ffae6db44d0e94c22cd09be45cfa : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"900fd8002a6a6538d48127e232b9ac0ba4499b99","revisions":{"900fd8002a6a6538d48127e232b9ac0ba4499b99":{"kind":"REWORK","_number":1,"created":"2026-01-22 11:56:16.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/12/974312/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/974312/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/974312/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/974312/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/974312/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/974312/1"}}},"commit":{"parents":[{"commit":"1a3987372c857f8ca5e4ecd9c4be7b617df0e758","subject":"Update README with new mission statement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1a3987372c857f8ca5e4ecd9c4be7b617df0e758"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2026-01-22 11:51:55.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2026-01-22 11:55:28.000000000","tz":60},"subject":"Fix ironic-pxe syslinux-tftpboot on aarch64 Rocky Linux","message":"Fix ironic-pxe syslinux-tftpboot on aarch64 Rocky Linux\n\nCloses-Bug: #2138872\n\nChange-Id: I6e30b749488a0ef2f247adf182db40cda8fa8ca4\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/900fd8002a6a6538d48127e232b9ac0ba4499b99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/900fd8002a6a6538d48127e232b9ac0ba4499b99"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d1"],"failing_atoms":["label:Backport-Candidate\u003d-1"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
