)]}'
{"id":"openstack%2Fkolla~974429","triplet_id":"openstack%2Fkolla~stable%2F2025.1~I064ef350fc8ec35f88606f7e8c3fd6f62bc6a6c0","project":"openstack/kolla","branch":"stable/2025.1","topic":"proxysql-3-epoxy","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-03-06 20:10:51.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-03-25 14:40:22.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2026-03-30 10:10:37.000000000","reason":"Change was submitted"},"36765":{"account":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"last_update":"2026-03-30 10:10:37.000000000","reason":"Change was submitted"},"37855":{"account":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"last_update":"2026-03-30 10:10:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I064ef350fc8ec35f88606f7e8c3fd6f62bc6a6c0","subject":"[2025.1 only] Build ProxySQL 3.0.x in 2025.1","status":"MERGED","created":"2026-01-23 14:58:31.000000000","updated":"2026-03-30 10:11:47.000000000","submitted":"2026-03-30 10:10:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":16,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"974429-proxysql-3-epoxy","meta_rev_id":"b86214ee00b78059c4a03adb2241814aca17cacb","_number":974429,"virtual_id_number":974429,"owner":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-30 10:10:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2026-03-25 14:38:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2026-03-25 14:40:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-03-30 07:52:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-03-30 07:52:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"all":[{"value":1,"date":"2026-03-25 14:38:34.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2026-03-26 21:41:34.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}],"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-23 17:59:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-24 10:57:52.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-01-26 10:48:53.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2026-02-09 15:09:18.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2026-02-11 08:25:59.000000000","updated_by":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"reviewer":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"state":"REVIEWER"},{"updated":"2026-02-17 16:02:23.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2026-03-26 21:40:25.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"CC"},{"updated":"2026-03-26 21:41:34.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2026-03-30 07:52:09.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"9348ebe255374637a16a65548533e5bbb53d6947","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2026-01-23 14:58:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e915d722d4e25976c479e8820c97168628cc2dae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2026-01-23 15:40:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"46c6fb4155ab6a4c4656e7449f9c48a999dfbefd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 17:59:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d65bca8bf7b49a59442b090f585cfeb\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/aece46a05f2f4bd38ed0af4837792275 : SUCCESS in 1h 04m 29s (non-voting)\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/289d8f5c056d4be79d4e747aacab038e : SUCCESS in 49m 22s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/f487a576b9244a5e8a12faee9bc27947 : SUCCESS in 1h 27m 33s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/6a7c58234bf54829904fd6a7b37b07e7 : SUCCESS in 47m 43s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/247843662f59458fa5390741c94d8c82 : SUCCESS in 49m 02s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/42fd0f3677324b4c85efdcbed6cc168a : SUCCESS in 33m 08s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/3bb5e24a1e0a40c79e6dab1a0fd81e98 : SUCCESS in 58m 21s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/c23447ccccf14b77ac214ff0900d2c39 : SUCCESS in 1h 26m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1cb91d4a58a4b6ba52025d06593e8c3 : SUCCESS in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e1912a710b4c4336934ff3239da0210a : SUCCESS in 2m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/80d6f0142094446a8cd0a748014fc539 : SUCCESS in 2m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e2f576aa0bb9445aacf9d2fcb8f95573 : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/783f8bc59ee6409dafa88be9668b1f2f : SUCCESS in 7m 05s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/f24f92e2667b49a793c3aff04cfb74d6 : SUCCESS in 3m 39s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/a13dbf7ab17e4ebca76c8debd6b38061 : SUCCESS in 53m 09s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/110b1e402c094095b74abb6949a837ef : SUCCESS in 1h 12m 15s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/34a39cdee31c40b692cfa58f3cabc62c : SUCCESS in 59m 40s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/8c850352038042468da3c5b644e6bc8e : SUCCESS in 1h 57m 49s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/78b474b1eed14860a2b6c17b6d101aa7 : SUCCESS in 1h 36m 44s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/bbccbfd51f7a42b9a888afbde4dca87a : FAILURE in 2h 04m 12s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/2c37af64fd8146d885c94745f4ea08ff : FAILURE in 1h 10m 14s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/58deea1916a141efa48fa33558d464bb : SUCCESS in 30m 39s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/bc01c3fc85134305ad84eca93547cbcb : SUCCESS in 1h 11m 52s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/2613346d4ba94ec488bd07318c65b729 : SUCCESS in 1h 25m 46s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/edb32bac493b409b869d1db434857521 : SUCCESS in 2h 06m 07s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/d567c319d774442c92e30db43f66d0a1 : SUCCESS in 26m 10s (non-voting)\n- kolla-ansible-rocky9-cells https://zuul.opendev.org/t/openstack/build/380951e9f7f94f51a6c9c8d9451c1b90 : SUCCESS in 1h 02m 45s (non-voting)\n- kolla-build-rocky10 https://zuul.opendev.org/t/openstack/build/5242a44af4374b43956a1a3937a58c10 : SUCCESS in 54m 39s\n- kolla-build-rocky10-podman https://zuul.opendev.org/t/openstack/build/3993eb7e5e404f2c8f09e372fbef353a : SUCCESS in 45m 07s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/2433c0d7a5374a51bef795645284814b : SUCCESS in 48m 04s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/c78c5194399e4fa190560023164a3544 : SUCCESS in 26m 11s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/89ff3a341e614b1cad339f19db00d729 : SUCCESS in 1h 04m 01s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/08a43551bf484179a90592c1adb4728d : SUCCESS in 1h 41m 12s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/e94706e26a8b420180fdbe72d537a5a4 : SUCCESS in 1h 52m 33s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/2fbe49ead4a3497694ab0d0897db6f51 : FAILURE in 2h 15m 08s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/2957780dc6534256b6a1b562628cac02 : FAILURE in 42m 02s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/fa1d73bac0814640a46edeef1939d769 : SUCCESS in 23m 29s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/f66675ace010465d978d81e28487bdb3 : SUCCESS in 50m 06s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/75e54c3ae0a84cc6b4589eebf54c0e4c : SUCCESS in 1h 19m 01s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/ca8a28335bbb444f9b28e5484b8a813c : FAILURE in 1h 17m 07s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/fc1af0caa6ef43e6b8dbed969a377adc : SUCCESS in 19m 34s (non-voting)\n- kolla-ansible-ubuntu-cells https://zuul.opendev.org/t/openstack/build/29ad32f879fd4277a3b32e0878fe08d1 : FAILURE in 1h 06m 58s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/7facc72330654265a4ed07f78386795f : FAILURE in 1h 28m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"23efc9f140372563a11f32e89ab658b4f70de481","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 20:25:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d64dfd714724486a4567985748bd1a1\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/c79bdfd3e53d4b218e478e13c195f231 : SUCCESS in 2h 02m 40s (non-voting)\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/95c09b7c7d394333b3d35a12b6e4c050 : SUCCESS in 2h 11m 14s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2508eb5512614a5e85f01b99b24e62b2 : FAILURE in 1h 40m 47s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/07c9d7c35bd04a95983fc5842168d681 : SUCCESS in 2h 14m 48s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/12fc3a6231984506a278d948d49651ae : SUCCESS in 2h 04m 49s (non-voting)\n- kolla-build-rocky10-aarch64 https://zuul.opendev.org/t/openstack/build/0d79d2e2cd3c4982b48617032f18b79c : SUCCESS in 2h 05m 55s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/88702d83f70c46c88ff33d289df6a7d5 : TIMED_OUT in 3h 04m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9b5270a14bc446c001eb4c76a02d69f0180fa40d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-01-24 10:57:52.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"bf8ce13f78f97dfc444d089a91617497f38eb37f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-01-26 10:48:53.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"71fd877f7a17587b007696b60e801eadd8df054b","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2026-01-26 15:12:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"514f7cb2514f48033e1b836d77d2841b5a98c790","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-01-26 16:49:41.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4ecbb1d10b589d726167e0a386dbaabf68b7ff50","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2026-01-27 11:35:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c0262d2d73af0d991a6053beb40d657940bbb910","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2026-01-27 11:36:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2c518c65c773086e9516429ae9e6496546b97ef1","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-01-27 11:37:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"39099f4b46043569b10815d668a72e3592d0e16e","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2026-01-27 11:39:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cd1aae956e7f5a2a2a3f4fe95504fc036c01f6bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2026-01-27 12:02:29.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"f6253a9604f0c161bee85dff3436da559ba45ea1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2026-01-27 13:29:47.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"86abe6ec4caa043e0d960bafbd0e2730abcd1fec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 18:17:45.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e942fd08b8704d04bca696f0f35b3b63\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/e15b3bc66ab14855a18336fd708da1ab : SUCCESS in 1h 00m 47s (non-voting)\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/dca561248ccd4e4c8633ad121aa5be99 : SUCCESS in 43m 27s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/f53f5c4869194f4f91ef5c018c5cad5e : SUCCESS in 1h 31m 17s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/f31230744f0d40f9bc5eec64f4ecf5a9 : SUCCESS in 1h 07m 13s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b5f9c36c99bc45acb320373a0e778c15 : SUCCESS in 46m 48s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/7d71fee3bb0a4351b23dfe1e5bdca32e : SUCCESS in 28m 18s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/92f3ba7e76784f06b3a9b8dd946656a2 : SUCCESS in 1h 02m 41s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/f03e195de15e4e6aaa5752c5fefc42d1 : SUCCESS in 1h 35m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f73ccc62be6a4b199015767bc2290a53 : SUCCESS in 3m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9631e753c2ab4388abfc44640741b1b5 : SUCCESS in 6m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/40cdf00b2dac45d9bed216d1886c95e7 : SUCCESS in 3m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/87abffe6420243beabc423babdaf9a72 : SUCCESS in 3m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d452e20d62040a7803be4004c7d9640 : SUCCESS in 6m 25s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/54d4dc76f20f471798e92930ce2266f8 : SUCCESS in 3m 01s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/e73bcc161848474fbabab40140073957 : SUCCESS in 1h 00m 03s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/b6200670b3594264990623abab57fe50 : SUCCESS in 43m 35s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/36fec23380504414b9aa3a87eeca1761 : SUCCESS in 1h 26m 48s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/a0b7ac255e1e4bd982bab1df6d382d7b : SUCCESS in 2h 12m 13s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/842c9c48c3a640db9bbd28ab02880b78 : SUCCESS in 2h 36m 09s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/e9f24af7e88b4727ae767ee4d89312d0 : SUCCESS in 1h 49m 23s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/6bb1574a042a4aa19c7b425c909a724a : FAILURE in 1h 11m 52s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/b8f09da96a0348aeafbb31ca25960229 : SUCCESS in 24m 17s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/4e09f58240214a2b81b6683d47cc8d4f : SUCCESS in 44m 49s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/d02bbf42fc0447ecbe1a3ef15025b95c : SUCCESS in 1h 28m 27s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/134a30381de448f694e61984005eb2d9 : SUCCESS in 1h 43m 32s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/1d9cdc727bb24d6e8442f7cdf4da9a01 : SUCCESS in 33m 14s (non-voting)\n- kolla-ansible-rocky9-cells https://zuul.opendev.org/t/openstack/build/d13b573db14c48a49f94c3566cb3f380 : SUCCESS in 1h 47m 59s (non-voting)\n- kolla-build-rocky10 https://zuul.opendev.org/t/openstack/build/d5799b4a318e4054b72b467d2c460c6a : SUCCESS in 40m 36s\n- kolla-build-rocky10-podman https://zuul.opendev.org/t/openstack/build/7a72f62c881c4fb19a027bb4db573909 : SUCCESS in 42m 35s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/c90511739de347f887dbb850c53670cc : SUCCESS in 50m 03s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/32c7305510274bed9b731ed01b81dd64 : SUCCESS in 28m 30s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/54646c8837ec4a37a6483aebbed96aa2 : SUCCESS in 1h 21m 17s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/cc2112885c564da0857d584a765efc49 : SUCCESS in 1h 54m 51s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/034441b3908d43d6b767d8766afac1a0 : SUCCESS in 2h 16m 56s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/0f83d024349f4d79a9021a2d643b9a30 : TIMED_OUT in 3h 03m 27s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/94be35e23dfa4cde8a16cafd24c43c19 : FAILURE in 41m 06s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/830a91d1978d446cb181acd5d4f4951f : SUCCESS in 28m 27s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/cfec8b2206d549cf81378ca576912867 : SUCCESS in 1h 10m 51s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/aa86bbca4adf4bcc946c4361c2932b34 : SUCCESS in 1h 28m 05s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/fef47c3fc5af4a9ba00206d8095f8761 : SUCCESS in 1h 56m 46s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/6d50fb5bc0034098baac858aba56b25b : SUCCESS in 25m 35s (non-voting)\n- kolla-ansible-ubuntu-cells https://zuul.opendev.org/t/openstack/build/0e7b1114ff40451fbca8a2c477c23047 : SUCCESS in 1h 31m 03s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/c31d91bbaef54c1080199a41bcffb2bb : FAILURE in 1h 34m 56s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9da8ff1b235773768a378baa01ddcbc54b0ab288","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 18:36:25.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94d6c9055eaf4243b7f4ceae3f849fdc\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/28fa1bd077fe4e7f8a52b35587fdfade : SUCCESS in 1h 40m 13s (non-voting)\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/294bc9f9d7194191a1f42a0b770623d9 : SUCCESS in 1h 54m 02s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8607e5fa908946028a0e852104fe7c4a : FAILURE in 1h 32m 46s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/75836ed370ea469ead4a1305ae29b102 : SUCCESS in 2h 04m 58s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/02ed3c850c2b4d17be2fc09a7eb37d3f : SUCCESS in 1h 48m 58s (non-voting)\n- kolla-build-rocky10-aarch64 https://zuul.opendev.org/t/openstack/build/03947322be6e4a65ae27007fb66d5341 : SUCCESS in 1h 53m 23s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/e120332181e34d308d688b687463b3ce : TIMED_OUT in 3h 03m 44s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"99bf982e4a02399e560769f851c0856d69ccf664","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-01-27 19:37:47.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"15e521826c61e7a03391c514c943ea3e19536015","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-02-09 15:09:18.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"335207f2a8f75076bac4d2db0f2c943a8d1d88f6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 17:10:48.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7755618a6014359b39bdb135dcdbb18\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/0b163ccedc4b45cf9a62b387951df68c : SUCCESS in 1h 28m 20s (non-voting)\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/c5069617414e401a8344604a707e643c : SUCCESS in 1h 32m 08s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f63b4eba7771465a97e13594f4a5b330 : FAILURE in 1h 16m 59s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9e628cbde0b94a8fa2d548a19b8f71ac : SUCCESS in 1h 45m 21s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/1f2dac54176c426da92e524d269c7353 : SUCCESS in 1h 25m 07s (non-voting)\n- kolla-build-rocky10-aarch64 https://zuul.opendev.org/t/openstack/build/fd40c93be73b456ebb2af474ee37e6c4 : SUCCESS in 59m 13s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/8aef1ffd9b0a47b182d6609adc366903 : SUCCESS in 1h 01m 10s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8a53b3fa52ef1c1b04225634639641701471d00b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 17:48:04.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4aed4822731d43a783321485b6a51a3f\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/489f442e1386453aac841e93c91fbece : SUCCESS in 54m 25s (non-voting)\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/6f276014fe054bf69ffa2a5c2e8cb48f : FAILURE in 55m 50s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/89b8dc8edeb44509afae47ed7556a98d : SUCCESS in 47m 38s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/87ff206a814b4907a22cd06dbe53c37f : SUCCESS in 1h 17m 21s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c73fcf88e47040019ff07d6022d91d0c : SUCCESS in 52m 05s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/ce291ff21cb4421fa66926b9a7a797b9 : SUCCESS in 26m 38s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/765b0401956646c9a7a49914b1f91416 : SUCCESS in 1h 02m 56s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/402608ef739040f9b1618276dc0dd314 : SUCCESS in 1h 31m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72c6a134f36b4dccb3e13620618b047f : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3151ca6beaf54159a9b93149d3f7b014 : SUCCESS in 2m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/69a487e56b8542c18763de2347f83df7 : FAILURE in 3m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4aca52e7729341dd8b87b69241e3988a : FAILURE in 2m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/340011fc97ca49a9a793b7dafc122df6 : FAILURE in 4m 24s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/3e512b52918a4649bdf6a8c18bfcc888 : FAILURE in 1m 50s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/63f9089d30e548d2bd569fc894976a30 : SUCCESS in 42m 30s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/694f3e392ab24c81a1f64136079c4505 : SUCCESS in 43m 39s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/760785d1e9674856b410692f34a386e4 : SUCCESS in 48m 34s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/02eaa4cc650c45a7ba29e5ee355908f6 : SUCCESS in 2h 25m 28s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/68ed07cd6828433ba14eca6d5ef87a07 : SUCCESS in 2h 19m 06s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/91e270d034a74221a19261137776f708 : FAILURE in 1h 22m 39s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/907b37b5a20648a0859d755b6bf43139 : FAILURE in 1h 31m 18s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/88827e238481423fb6276f6d7754d6a9 : SUCCESS in 34m 13s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/1c5a2816aa094000a25c49a92bf9dc04 : SUCCESS in 47m 56s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/68ffadd17f844416afb64d1d43a102cf : SUCCESS in 1h 40m 42s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/74fba3c4cf34414cae395ef7ec4d6451 : SUCCESS in 2h 26m 15s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/ca7ca3ccd69f4a38b08c4b1e8364c867 : SUCCESS in 39m 24s (non-voting)\n- kolla-ansible-rocky9-cells https://zuul.opendev.org/t/openstack/build/e9f6e3423a7645099487ff2c34f377ed : SUCCESS in 1h 06m 29s (non-voting)\n- kolla-build-rocky10 https://zuul.opendev.org/t/openstack/build/77fe20027197454f89c9bfeb8895aecd : SUCCESS in 42m 24s\n- kolla-build-rocky10-podman https://zuul.opendev.org/t/openstack/build/1dab8ace64f9480890b20b40318ba4ad : SUCCESS in 57m 07s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/63ed29ac36a9482d83f174808a1ea94e : SUCCESS in 57m 40s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/d43bda8769e9488c84bf258f59446f52 : SUCCESS in 28m 49s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/890373eb6af64f2dae41bad1010b5acf : SUCCESS in 1h 09m 42s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/daccdbabd7314f069d90e3d7cf7ed357 : SUCCESS in 1h 40m 46s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/9ee644c8e2c5421083674972f0e0160f : SUCCESS in 1h 55m 16s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/b34adfbd48924c08a425a501d8bdd7fc : SUCCESS in 2h 35m 43s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/3f7effd4045b484ba658103982cc9397 : FAILURE in 1h 21m 11s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/5a20e6b16e3a4e3d8a41a92c4a1d079c : SUCCESS in 29m 04s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/f8732f23860c4ad7a5f35a0daf71a8d5 : SUCCESS in 1h 16m 08s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/0cf62ffeb33545ac825852cea14222b2 : SUCCESS in 1h 14m 44s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/f411ffb8b18a4e34ac2cbb6f39c4820d : SUCCESS in 1h 38m 10s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/9b8b653bab064fc6b32be0ee09f969ee : SUCCESS in 37m 47s (non-voting)\n- kolla-ansible-ubuntu-cells https://zuul.opendev.org/t/openstack/build/9628d87fb81e45739d1f9ac77b36d23a : SUCCESS in 1h 23m 15s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/b5de39964ff94d6eaba1171d3d00c580 : FAILURE in 1h 35m 38s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f37272969041d896cf38778abd8900835ccc738f","author":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"date":"2026-02-11 08:25:59.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"1e19f9ad8851cd0909ee2b5cad0a99de33e53e91","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-02-11 14:35:42.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7d513d3cf2c8d15b49a14eb2ec6646d46f539698","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 17:25:33.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41fe87f2123f4a37b6a708698dd46437\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/79568580b7994b1b930cb8c2407a372d : SUCCESS in 2h 36m 24s (non-voting)\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/72b04bfd7e6e4d40bad2f4a873dd8ce1 : SUCCESS in 58m 11s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e47257aee6864e1aa987fb1e52994626 : FAILURE in 2h 19m 06s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/74d24196e4b6445a9c71ec2d2ba55f91 : SUCCESS in 2h 47m 44s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/25dc1626c5394b8f8c337d9d577eac3a : SUCCESS in 2h 31m 48s (non-voting)\n- kolla-build-rocky10-aarch64 https://zuul.opendev.org/t/openstack/build/fccb2c766c01400eb1c523e2667bcf0e : SUCCESS in 2h 44m 12s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/0662e192f64944118c3c8706a170e66c : SUCCESS in 1h 23m 56s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"0352b5b602e1d78ba00bd5487d9f731179af3580","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 19:26:39.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4de6395f8086453486f62f02c34e2b07\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/a428c222de7f491cbe1faf30badc74fc : SUCCESS in 59m 14s (non-voting)\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/4b3841e156db4152ab21fb333fe0ed9d : FAILURE in 40m 23s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/81729318266545d1a9bc4040f42277be : SUCCESS in 1h 30m 14s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/cd8d9f29f4764409825e568af64dcfed : SUCCESS in 52m 14s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/16b195c0fdfa4cb885cbc4e067ec24cd : SUCCESS in 42m 22s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/f77b4ed1560d4bb5a1f8738f5aeb6faa : SUCCESS in 27m 13s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/fc30f08d955c4ab4825b42ddc829f8e2 : SUCCESS in 1h 00m 50s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/cfb71ff7740a4340b0f7a62878d48184 : SUCCESS in 1h 27m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b145fc3219f4aa0853964bfc3fab828 : SUCCESS in 3m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cb2428fcbd5c4f0a9cd6428c33253205 : SUCCESS in 1m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ffaee90d9cf14e379c195ee598cecd04 : FAILURE in 3m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8977b5aad4f048daa134e958cde03bbd : FAILURE in 2m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/854d56731b604da295ead98963962610 : FAILURE in 3m 42s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/e02742041005427292701aa5e436f2ef : FAILURE in 2m 56s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/e19582a0321c48088a04780551bb58f1 : SUCCESS in 1h 02m 54s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/a6f11b8e6ec64d72aae8a96bd1e3598a : SUCCESS in 1h 35m 10s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/6f7733e3e9a84c0686866ca343762daf : SUCCESS in 1h 24m 46s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/d8d49b57f3554a3399c1bc97dcf3ba11 : SUCCESS in 1h 59m 12s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/c0acff4cd16b4c889d8a18dd0e270694 : SUCCESS in 2h 19m 57s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/01c45db4144f4c3ca14cdc41ded1584e : FAILURE in 1h 59m 29s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/a26673a97b364188844fb7371100bb97 : FAILURE in 45m 31s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/ce7ace98cd8e49168ffd47e893499e82 : FAILURE in 18m 02s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/f4e7e44ac98e4b8fb21a1d7fb62bbd87 : SUCCESS in 1h 40m 22s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/ca915190ae79412f990dcdde8570ac16 : SUCCESS in 1h 32m 45s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/ddd1eeee949745e98566176eb311ce8e : SUCCESS in 1h 11m 44s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/5a33fb993c4b4527957f494e6d099d82 : SUCCESS in 43m 08s (non-voting)\n- kolla-ansible-rocky9-cells https://zuul.opendev.org/t/openstack/build/d0a4b9dd33c34b38a7a63554c8d724a8 : SUCCESS in 1h 45m 17s (non-voting)\n- kolla-build-rocky10 https://zuul.opendev.org/t/openstack/build/34050ab36fe74bb883393e199fa61986 : SUCCESS in 49m 24s\n- kolla-build-rocky10-podman https://zuul.opendev.org/t/openstack/build/ea4e7219d2aa4c1f9f6839e799fa2a1c : SUCCESS in 1h 04m 42s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/5c799560ca6740ec8ee147f62472a173 : SUCCESS in 42m 02s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/ce9a5dd06b11456b947d2ba6f21e54ca : SUCCESS in 37m 43s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/5fc5faf24b104e9f9ed944c23e0902d3 : SUCCESS in 1h 05m 58s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/3bd6e970d8ff497ba10eb1b3b57ce2e0 : SUCCESS in 1h 49m 52s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/8d41b58784ac4796a781b690abf2393a : SUCCESS in 1h 51m 06s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/91e90e98234841c3a0f9774458c59852 : SUCCESS in 2h 37m 42s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/562992c39e454409a4ecb91b20ffb1ab : FAILURE in 1h 10m 28s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/382f0584e2e94df69aed47004989be71 : SUCCESS in 25m 14s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/bb259abb75db40a7ba7cccbef3407cc9 : SUCCESS in 1h 16m 18s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/430b87d8cc4b4a8ba7f7f4daa7ff5c95 : SUCCESS in 1h 10m 13s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/59e015339c6d41b8a8424918e087e8bf : SUCCESS in 1h 43m 10s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/eb999fc82fc6430780ba601dd87d4c7f : SUCCESS in 19m 33s (non-voting)\n- kolla-ansible-ubuntu-cells https://zuul.opendev.org/t/openstack/build/12f333d276054dfa8688f735e448da08 : SUCCESS in 1h 35m 44s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/4c8734b326d643a6a22f0005060d9f28 : FAILURE in 1h 01m 05s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"79ff727c880879c0e366869e640fe9efcd41b264","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-17 16:02:23.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5b685f6d125926097b8b9d9b2a6c50f83f6797cf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 18:30:55.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e52e363be74b4d759bd3305bfb680358\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/ac13ec7aa3294765b26371730e02b2c2 : SUCCESS in 54m 26s (non-voting)\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/f880e61408614950a2c6d7ab6c3c48cb : SUCCESS in 49m 17s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a82a6f0ee9374e00b17c7f6a5330736a : FAILURE in 41m 32s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f8253d8e4bb94a0ea23437258f5195ed : SUCCESS in 1h 01m 26s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/5b4c767b43c34d9fa712a6d0b331a493 : SUCCESS in 51m 53s (non-voting)\n- kolla-build-rocky10-aarch64 https://zuul.opendev.org/t/openstack/build/d7089dd9a56d47829ce556bd7ed4a40a : SUCCESS in 50m 34s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/4c1e44c16e7345468c2ac9dd73e52c9a : SUCCESS in 59m 15s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"0d54117af2a44349a6542b1e9bf061fd2acffc94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 18:32:52.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1310711f94264182811dbc1bd7d81b78\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/b8493b7bb9a24ef4a46a0a9c511e4f46 : SUCCESS in 1h 00m 47s (non-voting)\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/e2b1570be3c046ec9bb57224a0083e24 : FAILURE in 58m 35s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/8beba6ee1a6648919050ff4643eb6d78 : SUCCESS in 1h 42m 29s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/5b53833ef7c841fcb7979f618e7b57bc : SUCCESS in 1h 15m 34s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/3c8ef52a863646c1a2054686cb36d4d3 : SUCCESS in 52m 27s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/53c4fb33816a471ea09143169de8a24e : SUCCESS in 38m 47s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/1da9716a9af3415daafeb26e9f3165df : SUCCESS in 59m 09s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/f937063c9e774a7f9c3dbc94dddc119e : SUCCESS in 1h 28m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7729f1d51fc74abc87deaa9c029e1c45 : SUCCESS in 2m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ac216497ac4042fba4850743bb61916e : FAILURE in 3m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b8fd76c466d445998f2b4222c6655ead : FAILURE in 3m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc03f4bff47b4be089461eab66251c33 : FAILURE in 3m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89af6148623f4b489d36933f5bb5642d : FAILURE in 6m 15s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/0f9b056edcbb49a983083b6a1e984c26 : FAILURE in 4m 05s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b6f45104c0184a0ea1b82142738e7a6c : SUCCESS in 1h 06m 18s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/e1447b1d4eb142c984cc82e28d6f5546 : SUCCESS in 42m 16s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/6e0d48e89edf4c42a4eff514ca6a4b53 : SUCCESS in 1h 14m 50s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/52681719dc2f45f5a0dd5e56c1ade2b8 : SUCCESS in 2h 04m 10s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/44d3db86924242c28a8dd6f9a60853a2 : SUCCESS in 2h 18m 31s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/f5ca06a96b8b464f8e303174cc00437b : POST_FAILURE in 2h 01m 38s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/7070c172465c4ee3ac0d933a78b093f1 : FAILURE in 46m 37s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/421463724ff54d6fbbcb02ca2fb143f6 : SUCCESS in 18m 30s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/8c8586b30b464044a9790f81f5e98522 : SUCCESS in 1h 33m 20s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/d4f14867712f423790528d856639084d : SUCCESS in 1h 40m 25s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/76148d0b9d55403797debae9cae7794c : SUCCESS in 2h 02m 30s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/930f80f8ba3541d485f59d77ca06416f : SUCCESS in 39m 09s (non-voting)\n- kolla-ansible-rocky9-cells https://zuul.opendev.org/t/openstack/build/44faf7d954194e7dabd9cbbd8d2da167 : SUCCESS in 1h 55m 28s (non-voting)\n- kolla-build-rocky10 https://zuul.opendev.org/t/openstack/build/993bb89f8095402a91eb2fe91f68281a : SUCCESS in 56m 21s\n- kolla-build-rocky10-podman https://zuul.opendev.org/t/openstack/build/82bc73e6baaa453cb2486bc891410684 : SUCCESS in 59m 14s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a9076f796ac548729b30b71dd2fa658c : SUCCESS in 58m 07s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/c02c75153ede4ee38126ac5c1280111b : SUCCESS in 30m 05s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/265932f5bd9f4586b2616c7674a972f7 : SUCCESS in 1h 19m 02s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/26fde642dd2b4e2d8a05ababb794be1d : SUCCESS in 1h 47m 47s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/05d663cb282947c3adf1e19404db9e48 : SUCCESS in 2h 12m 04s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/09fd88a7f539462cad23f523d46c2231 : FAILURE in 2h 16m 46s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/ca92e63c80f94f31911d8c8b0e9593f0 : FAILURE in 1h 20m 28s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/8bdb3f65d0e6427fbda935d37f754229 : SUCCESS in 19m 15s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/77f25d3933c54fa9bf621dab7f812c4a : SUCCESS in 1h 05m 15s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/b4e17334c9e44ac289037f918bde9acd : SUCCESS in 1h 12m 21s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/7c47510d42394208b6b047a2eabc9eea : SUCCESS in 1h 46m 24s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/84b1494569d44d59906e16ecd4cecb8d : SUCCESS in 35m 37s (non-voting)\n- kolla-ansible-ubuntu-cells https://zuul.opendev.org/t/openstack/build/808b6a3d3bdd4e65a37cdc7f8581e343 : SUCCESS in 1h 29m 29s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/e2741ed8eec040a3bc084754c44ea551 : FAILURE in 57m 35s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f0868aea561d7aeca7bb651b34fdbc163396be41","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2026-02-25 10:13:53.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a37e2e7b2b53eb7ac0972ff6963ebfb200dd6130","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 12:50:12.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a94c40506c3c4b5a8be3e697d7d59bf6\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/8d9544f18a464d0fbc2ac996e2572b4c : SUCCESS in 2h 21m 59s (non-voting)\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/66976f7ffecf43099f6b3eb08b743071 : SUCCESS in 2h 30m 48s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/af59b8efe29542a6929760bb46a40319 : FAILURE in 2h 11m 57s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/28e6f1c96b724b939e1224b34b40f8ad : SUCCESS in 2h 13m 31s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/61d562b98f4d4538a4d0484956f9050b : SUCCESS in 2h 13m 13s (non-voting)\n- kolla-build-rocky10-aarch64 https://zuul.opendev.org/t/openstack/build/c84d13f884984c54a030fbb177e135ea : SUCCESS in 52m 49s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/193b1dab786e42888abace08d5e8b036 : SUCCESS in 2h 25m 00s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"828bab378921e51441ead0bf758566b62db09712","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 12:59:54.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/49400698698d4a93a163fa910bc60fde\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/20e7dec78950472e866fffa73e93eafe : SUCCESS in 1h 09m 23s (non-voting)\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/0e6eb97afa5546339dd2ece28240fb58 : FAILURE in 54m 57s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/e177efe593414bc6a65feadbc19057f4 : SUCCESS in 1h 26m 01s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/6c25cdfad52b4d7090b06def554d8367 : SUCCESS in 1h 22m 09s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/3d72eece4bbd435fbf8b66c87f55d463 : SUCCESS in 50m 46s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/e39a3e3c9b9a40ffb9b52f1b5d96c547 : SUCCESS in 31m 21s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/5217a240ed084207adfe482b31e83f64 : SUCCESS in 1h 02m 18s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/4d646018cf13410cb64be2a2b9d56ac0 : SUCCESS in 1h 30m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4cab4dd8668241d9952e4d542b548a3f : SUCCESS in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/98e3dddee0304c71ae9a50e52fb6eaa8 : FAILURE in 3m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1e2b6287228c456ead8324b40aee070c : FAILURE in 3m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7f23adeccb7b49658706406e83994fac : FAILURE in 3m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b63a089b1c42464398cf89c5ccf6aa1a : FAILURE in 3m 24s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/8b7a565daf774f32bb68b91500d42866 : FAILURE in 3m 21s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b2924834605640db9eefe1c29d326616 : SUCCESS in 1h 06m 57s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/0e7dc7aaaa8d4e0b885efef973742c36 : SUCCESS in 1h 23m 56s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/f0a30c8bdb924350a2ece47dcac75aee : SUCCESS in 1h 24m 32s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/415dde5adb034619a79c423bc6283a56 : SUCCESS in 1h 50m 00s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/21c03938a5e4436baa50c57a5d7a7af6 : SUCCESS in 2h 34m 12s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/a42c42a7335a423ea6584af3f44a8525 : POST_FAILURE in 2h 35m 57s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/e8bd8ae61010480e83d50ddf97f06206 : FAILURE in 1h 28m 20s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/503b5f224f9e4e01ae5e6675e8d4da6f : SUCCESS in 29m 59s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/77f6c9136c0a4c1eb4a6425eb85f6fbb : SUCCESS in 1h 18m 50s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/8f2ee11941094efc9b4b0c09501aacf7 : SUCCESS in 1h 37m 39s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/90fdd6a7ebaf4bd3b5010f5480adadc7 : SUCCESS in 2h 15m 07s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/e6ae2a9235e74e9599bd8de44eaf2692 : SUCCESS in 41m 06s (non-voting)\n- kolla-ansible-rocky9-cells https://zuul.opendev.org/t/openstack/build/effbe3db87ac40c99ef4f2f3f474b9f4 : SUCCESS in 1h 47m 03s (non-voting)\n- kolla-build-rocky10 https://zuul.opendev.org/t/openstack/build/090a6bb8ae0e457f99f33845dcb8ee4e : SUCCESS in 46m 19s\n- kolla-build-rocky10-podman https://zuul.opendev.org/t/openstack/build/51377eae076d44258a8822bfb26c3dd8 : SUCCESS in 42m 33s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/677bd75ec020419681ed6e4e6c9846ff : SUCCESS in 53m 14s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/7b9d7642a743459db6aa60c37bbaf307 : SUCCESS in 32m 00s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/0e81462d7b9a4b5e9cc5664a9be6e336 : SUCCESS in 1h 15m 14s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b3f7f5f5dc1c40138263d2d2223b0c38 : SUCCESS in 1h 38m 08s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/cea56622e5a74a2d9c9dc254e00d98bf : SUCCESS in 1h 53m 08s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/aee2e3320f29452d892e1ae8b690cc16 : FAILURE in 2h 14m 01s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/385c76b5b5e0440da032137a8739510d : FAILURE in 40m 18s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/7e69a3718d4d414cbaf55e04da1cbbcd : SUCCESS in 32m 00s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/bce21e06c4084be58a194880fd44e762 : SUCCESS in 1h 31m 26s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/c3ef7e2bd9f34b99807ebc482d05baee : SUCCESS in 1h 20m 03s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/d68926b2d59044468d7207e0a3befb40 : SUCCESS in 1h 44m 11s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/e16c65fdbe214d0294b47eaf7b46a0b1 : SUCCESS in 36m 01s (non-voting)\n- kolla-ansible-ubuntu-cells https://zuul.opendev.org/t/openstack/build/d96c23500a9f42f6963216bb925461e3 : FAILURE in 56m 28s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/dbd506769cea43d8b2f121f26d0c7f9a : FAILURE in 1h 28m 30s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"90885e9e4f7a731781f82f984ece759190cf4a5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2026-02-26 12:07:24.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"70789da047cf8976da190b05432fb0112384c080","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 13:34:41.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92141652b80c4c618bd632f3b8d58c0c\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/3e615084fc5e4dbea673377fff0093ac : SUCCESS in 58m 53s (non-voting)\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/199699816f6a44a8ac20011bbf7d23aa : SUCCESS in 1h 08m 15s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a027e4ba5d8247a1838c58db38441c4f : FAILURE in 56m 51s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/973542f7b0bb42a99e8dea5f4fb21401 : SUCCESS in 1h 13m 06s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/953e3e471e2f4c62b014c336f02f5c74 : SUCCESS in 55m 57s (non-voting)\n- kolla-build-rocky10-aarch64 https://zuul.opendev.org/t/openstack/build/0e584760d6b1469dbc0c569bc3a5b87e : SUCCESS in 1h 07m 12s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/0fb7d9013483409ebefcb52ebacc8f62 : SUCCESS in 1h 23m 03s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"0961b7641d0d7cf8b6a7b1ca13d6eb0ce6fe3ddb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 14:50:51.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f1d1ef4e605e46ca8a1264650c6f6ac9\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/7d375f4a5b1646a2abb8d77c1fcc3ce6 : SUCCESS in 40m 57s (non-voting)\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/7ec6c6206a49416b85484796568244fe : FAILURE in 49m 55s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/de406181464b4cbf9b01a1833bdd50dd : SUCCESS in 1h 25m 59s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/711db56e7d0a4a26b8a6b615c1c543e2 : SUCCESS in 1h 32m 19s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/2169500bea2f418e9c1106946cf92459 : SUCCESS in 47m 35s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/beafbd731b2d4e2991d6ad3a0c6a3606 : SUCCESS in 26m 55s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/58c7996823ce4178b4640ec8385d36d3 : SUCCESS in 57m 41s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/a7b6e790ce6f42ff95b99055674e4831 : SUCCESS in 1h 24m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/115ee0ee86cb4994b68e4bbe83267e33 : SUCCESS in 2m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/afc3ea8e08cd49a0b4763e57839bcec7 : FAILURE in 4m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e93688b6eaa14a7fb3df8a3a34cced0f : FAILURE in 4m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4bdd402402f14b32a6fc23fa3a8998dd : FAILURE in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4a995e421fe452a9fab99fc85d6a7b0 : FAILURE in 6m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/154c63f947e94df88eef56a90993dd77 : SUCCESS in 5m 54s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/85cf90378a7241848e24dfd42a2f3514 : FAILURE in 3m 20s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/c7cd3c260bad4d339334ea32a83da3cb : SUCCESS in 49m 03s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/df5e7129d0f344a3bb0ee0e12800597d : SUCCESS in 1h 26m 07s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/4b30785d7c0b4505b5bd46a048f0ebe3 : SUCCESS in 1h 36m 52s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/c37c4297d17f4a6e9e6e530b38cabe4b : SUCCESS in 2h 14m 41s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/eb5c6158b39f4624aedbcbcb8fca0225 : SUCCESS in 2h 20m 16s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/f0a1231e994d42328cb590cedb3fc706 : FAILURE in 1h 35m 54s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/67df3068fc334d7fa752fab058f5e8c7 : FAILURE in 46m 15s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/35aa05532c674a039f112ede446536de : FAILURE in 18m 45s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/71ac12d3e5a743c6ac2fd8659ff5abce : SUCCESS in 1h 05m 08s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/3a8f0e9cc9ef43faa2b896456f84524d : SUCCESS in 1h 29m 02s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/eb32b2f583834ec7a5944d47316f4940 : SUCCESS in 1h 54m 59s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/f8684f369ac5464c88b647c450451b2b : SUCCESS in 20m 42s (non-voting)\n- kolla-ansible-rocky9-cells https://zuul.opendev.org/t/openstack/build/51a914b8c7c54f34a0348473858c4377 : SUCCESS in 1h 03m 46s (non-voting)\n- kolla-build-rocky10 https://zuul.opendev.org/t/openstack/build/41688b1c70764915aab6fe58ec4681e3 : SUCCESS in 43m 38s\n- kolla-build-rocky10-podman https://zuul.opendev.org/t/openstack/build/ad19b65119854b8eb25b48ded98eb65d : SUCCESS in 1h 16m 07s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b9ba3531f13c45aeb67147156038b36d : SUCCESS in 46m 39s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/9bc87b6785ca4526a4e89984f573af7b : SUCCESS in 25m 18s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/2d6772a3d1134323b8de825959a32771 : SUCCESS in 1h 19m 47s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/dac4075f9da84816bbbe9861aa5635b2 : SUCCESS in 1h 51m 11s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/1fa8f73832be4f1a9656a8ae389d229b : SUCCESS in 1h 47m 29s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/7663fce8957e48289afdac681c1a47ad : FAILURE in 2h 09m 44s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/e555e20a27524119901c4ab121911fde : FAILURE in 1h 01m 49s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/b9fca1359da64f4ca644f992a360bd3b : SUCCESS in 30m 22s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/f27d7e2f68a04dd99a4eef9558260e9a : SUCCESS in 1h 16m 38s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/787bf7c5a8ba467f936de9c6211f8011 : SUCCESS in 1h 08m 53s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/f4128765b3d543679122fe8e4d9ac284 : SUCCESS in 1h 39m 23s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/ea292539458e40fabb45692e845457e2 : SUCCESS in 34m 47s (non-voting)\n- kolla-ansible-ubuntu-cells https://zuul.opendev.org/t/openstack/build/7fca8646a9d040de91041cb2c3f4cf0c : SUCCESS in 1h 27m 07s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/f1e5289a68cb49d2a09a412856cf02c0 : FAILURE in 1h 24m 01s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"cd36884b8db317892e76c1eacfa8b6a2710ce25f","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2026-02-27 15:52:43.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"136afa61329365d3b0d5be3126743d99cf827c35","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 17:50:19.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9e09463c88e4b5ea71928e32077d14c\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/e90f68f7bcc64eca9d88e27380080ba0 : SUCCESS in 1h 08m 31s (non-voting)\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/c980510b95f54529b5c5760f3dfc33ad : SUCCESS in 1h 35m 57s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/03cd87df0b2e4e76a67a3a339ce49158 : FAILURE in 1h 04m 55s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8311a63b570e43ed9fd386ee5835b714 : SUCCESS in 1h 45m 49s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/752717aa26e940e7b912c7a21bb81c07 : SUCCESS in 1h 24m 57s (non-voting)\n- kolla-build-rocky10-aarch64 https://zuul.opendev.org/t/openstack/build/57afd60286eb4b959269c077d55ff3c9 : SUCCESS in 1h 28m 18s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/ac2bbb7318384f2cb2f704c8eb1854ed : SUCCESS in 1h 01m 29s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"298456f70b94252b00f609f4fb6f296c76a9d593","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 18:42:24.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ba7461ff75f9464aae1ce2ed574b4f52\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/fddb7e4b636d4b8aa659d81db64ed48c : SUCCESS in 49m 22s (non-voting)\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/d405d21b19e545f2b5c905e3aa1c01f1 : FAILURE in 43m 55s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/f6bcb1852e8c48ddb9f3dd211d10c773 : SUCCESS in 1h 36m 55s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/1c3c3774132c46f792d56f87fba5e86a : SUCCESS in 47m 23s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/e209f7d64888493c8144ead21c720085 : SUCCESS in 51m 07s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/80d298b448df4fb7b526b701ae3b49f9 : SUCCESS in 18m 13s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ed2b4477039341be81b0f0b3b1d7eb51 : SUCCESS in 1h 03m 28s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/3225f4d39ebf4468b4202960fc4e2b81 : SUCCESS in 1h 26m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/506105cea4af49ea95c3da161e8b8545 : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f39e6406256e43109ccb1533ffec758d : FAILURE in 5m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fcefa8d50f2f41b2a9ebf81bc912de22 : FAILURE in 3m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/796c2a296398421398008771dbf54a89 : FAILURE in 2m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dca12098d3df49be9e8e76e2c185d7ad : FAILURE in 6m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0149d272391a4e6aa9f5c3f85a007cca : SUCCESS in 3m 16s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/34d371a217df4bcd92430672907265d6 : FAILURE in 3m 42s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/fee4e9ef3e45480f9a5b05c829478ebb : SUCCESS in 55m 38s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/914f3c2d9b4f4e8aa8cae00600fe39e5 : SUCCESS in 1h 26m 32s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/9e40b0a2710847eb8c66d8ead4f366ca : SUCCESS in 50m 13s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/436c1fd5056146148ba8eeee678d0616 : SUCCESS in 2h 03m 33s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/349174cbe2154eb7a03640fa40edc33d : SUCCESS in 2h 34m 26s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/f5ca36bc065444a69e8a997d59370b06 : POST_FAILURE in 2h 23m 39s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/344b7388dfd84ee4a1db8e2118321da5 : FAILURE in 48m 42s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/a2d6397ccdb340fa98ae4c503622d960 : SUCCESS in 37m 41s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/640b2a8c7d7444388f5d442583289a16 : SUCCESS in 1h 13m 57s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/be84d40d819842db92510d15b6d84600 : SUCCESS in 1h 28m 02s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/fc2f7227d8e9417eb8e8a5aa55683f0b : SUCCESS in 1h 21m 27s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/1791033ebc3841fdae798b2f6d9c4666 : SUCCESS in 41m 14s (non-voting)\n- kolla-ansible-rocky9-cells https://zuul.opendev.org/t/openstack/build/524edc03c0df417b9d10d0aa1acebf7d : SUCCESS in 1h 59m 19s (non-voting)\n- kolla-build-rocky10 https://zuul.opendev.org/t/openstack/build/f933936756ec4710aa45e95e54783042 : SUCCESS in 40m 55s\n- kolla-build-rocky10-podman https://zuul.opendev.org/t/openstack/build/ad57fc5068d54ccc9dafc16edd07594b : SUCCESS in 49m 17s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/4fcceac1eae74d8392a105c00a729477 : SUCCESS in 42m 53s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/e52de31a2f454f8c9e828f06508fdcbf : SUCCESS in 32m 37s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/4d2a069b67f24e4c97195c322f944678 : SUCCESS in 1h 12m 29s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/9e46e0814f214d3fb973ad362b78e8b6 : SUCCESS in 1h 38m 19s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/bedcfb08ffc945a4a8ba572a1a2d63a3 : SUCCESS in 2h 28m 50s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/b37cb425caaa463cb05629af1dfb57dc : SUCCESS in 2h 30m 24s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/a3c0fc9eebe54abba78e10cf7f01f188 : FAILURE in 1h 14m 52s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/3921ae6de5ab46dd89fff22e074a4d3c : SUCCESS in 29m 44s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/2213d8fc6c7844df9bc7c5e7a3045843 : SUCCESS in 1h 20m 37s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/73b251ac07274dcca1fdd1bf1619ad36 : SUCCESS in 1h 25m 42s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/971611f778eb4a9990b05d976dec0009 : FAILURE in 1h 24m 02s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/32a5bcfff05f430fb4e5cbaa5b4cffba : SUCCESS in 36m 55s (non-voting)\n- kolla-ansible-ubuntu-cells https://zuul.opendev.org/t/openstack/build/82bc7f77847d444dbd1c3341f2b0b509 : SUCCESS in 59m 50s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/aeb2f61a2fc541818d79550ee3804774 : FAILURE in 1h 01m 43s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"9bc11387217dbe890adfc15489105b3723ca2c67","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2026-03-06 15:55:41.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"8ea203ce5a2219597d07a59f0a149d02c9fc04ed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-06 17:45:57.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a41ce4e54284403b5a1b64b7fbe9895\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/3dad165a1de649e98754572218415bea : SUCCESS in 53m 43s (non-voting)\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/7a7c6dfe2c1d471cb682a247830a32a4 : SUCCESS in 55m 17s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a1a5047b376e4fcbaee12b079a3726fb : SUCCESS in 48m 08s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b42a01032ece486db36ee4aeac04586e : SUCCESS in 1h 18m 50s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/af5811bf06234f19b698bd157463e08e : SUCCESS in 1h 03m 39s (non-voting)\n- kolla-build-rocky10-aarch64 https://zuul.opendev.org/t/openstack/build/8b15aa203fc748e885b6f31b91de3b88 : SUCCESS in 1h 21m 50s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/0eaf7d8e96f746c1bb931a7e262d5538 : SUCCESS in 56m 40s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"6a02a4147759293b12d8298a5c2bbfccf269e128","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-06 19:12:53.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b52f28eebc774cc18ea731fcdbe935fa\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/2b694043874b4cfdbcb79ea08a3794db : SUCCESS in 53m 01s (non-voting)\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/a9e9511a67194948b7f3fe60c4da66ec : FAILURE in 52m 45s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/dbfc7c246206403ba3385cc06395e136 : SUCCESS in 1h 27m 52s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/27beca4030ba470cbc1a958bb68c9ae4 : SUCCESS in 47m 51s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/64531d933a7842839af5ce8f435bfbb9 : SUCCESS in 41m 29s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/e3a8830576ee46a8af289eb2861135f0 : SUCCESS in 27m 20s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/3b1c28def3ea40a6a30c5ce9b4e2ffbd : SUCCESS in 1h 02m 47s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/0c916ee0433b4a049cd816f11d964246 : SUCCESS in 1h 34m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ed28ddc48f34b6ab9aab98fdc8573ef : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/367f449e4d1b4ea997317baea58dfb84 : SUCCESS in 3m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/eb19842b508d46a19872b9795d0fce9d : SUCCESS in 3m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e0ce55335d284155ba4223aebdc26d6e : FAILURE in 3m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76f9b80aea01402ea3b51771a54d08bf : SUCCESS in 6m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b7865ec228da4ad4bec6a7742ffb8745 : SUCCESS in 3m 56s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/df7551a1419a4ca98da3a67504665c92 : SUCCESS in 2m 06s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/ebb9554359f74a74b1c798f45536875d : SUCCESS in 1h 11m 59s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/9056c96b8f64419a862fc3315e04612e : SUCCESS in 1h 23m 02s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/14cea072f7d248ec822f61c3d482bf59 : SUCCESS in 1h 38m 24s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/a812d6c0698e45f1a37d67fd3278b899 : SUCCESS in 2h 12m 03s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/b3bab44828bb42a8a319309717a8f301 : SUCCESS in 2h 01m 23s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/36517fc61f2c48ad9b63af2ccf7820bf : TIMED_OUT in 3h 03m 13s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/0358568792c348bd93201b69256a9ad2 : FAILURE in 55m 39s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/c8f7ec50e4414e78b851f86dc9505d97 : SUCCESS in 35m 08s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/0b77f76a445446c0afc5271b8e29501e : SUCCESS in 1h 24m 01s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/3c6c03f19f6945f7b8830f752e72f3a8 : SUCCESS in 59m 08s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/477d7653096b47f19d351921269fa345 : SUCCESS in 2h 18m 32s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/f863dcfadc83467abf9f2f1e708066db : SUCCESS in 35m 06s (non-voting)\n- kolla-ansible-rocky9-cells https://zuul.opendev.org/t/openstack/build/4825cb65bcd74bf4b9234d7bafc294d6 : SUCCESS in 1h 18m 10s (non-voting)\n- kolla-build-rocky10 https://zuul.opendev.org/t/openstack/build/993317eec1714a9889d9a1fe32f030e6 : SUCCESS in 44m 19s\n- kolla-build-rocky10-podman https://zuul.opendev.org/t/openstack/build/5a62896d14934b9381cf42e064cc0927 : SUCCESS in 43m 42s\n- kolla-ansible-rocky10 https://zuul.opendev.org/t/openstack/build/05403319c09a4ccaae6a87ab1be276cf : SUCCESS in 1h 29m 23s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/1ae10431dafc46ba89f94a3343692794 : SUCCESS in 52m 38s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/167f71a32e6541009fa4e60ee5cf8c36 : SUCCESS in 29m 38s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/f396ff85c729416ebdfc02925801a4ab : SUCCESS in 1h 10m 58s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/7e34bc37f4c043a6ba593569bd50482d : SUCCESS in 1h 43m 25s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/f55d12e327f545df816d203cb27e4f2f : SUCCESS in 2h 01m 08s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/4c704f6bdf1445fa8b016fcdde740154 : SUCCESS in 2h 56m 16s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/58b3c11a22a348fbb56302ebdf2302b3 : FAILURE in 1h 14m 00s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/cf0443610b164e7c9a840eb16440ccdc : SUCCESS in 35m 45s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/182292d6040c4b91ae61d5358a7784be : SUCCESS in 48m 14s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/907a53462cab4d06a2fb419a62b7c47d : SUCCESS in 1h 10m 59s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/c5b20083e2ac4905bb48a43d88c5c3de : FAILURE in 54m 21s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/5bccfd8b7c834b35a1bd47f4099b5155 : SUCCESS in 52m 32s (non-voting)\n- kolla-ansible-ubuntu-cells https://zuul.opendev.org/t/openstack/build/1c404896099140869ed4946caaf92c62 : SUCCESS in 1h 27m 56s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/2cdab9c5de8649748887fa1a4fc255c2 : FAILURE in 1h 35m 09s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"820c79f76fa123228dc9da3ea6c69cb587d47807","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-03-06 20:10:51.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f044a60c54052cdcd5d19bbe7bdc52068a32d5f0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-06 21:51:23.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64283cdd7a854829a3ccfe3d8804f950\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/310906944ae04c41a56ee12cb7eb60a2 : SUCCESS in 1h 28m 18s (non-voting)\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/5a0d930a4638419ba8c44256dcd5e3b8 : SUCCESS in 1h 01m 55s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/0f7f9ed1753347a59ffcaf2ddaec4322 : SUCCESS in 46m 21s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5eac16ee072b4514b9eb914f74ab7f74 : SUCCESS in 1h 21m 23s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/c44dfaef89e34ab0bdfaf7f45da9c843 : SUCCESS in 50m 22s (non-voting)\n- kolla-build-rocky10-aarch64 https://zuul.opendev.org/t/openstack/build/94f7479f2dae4d7398f85324df6cb28f : SUCCESS in 1h 38m 17s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/9ff8e9f926074c0dbe805b05982a9493 : SUCCESS in 56m 08s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"630ac2a7c3259e5b44159627c54f8e5738a64e92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-06 22:39:12.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8bc6bbbb52464030baa15ab76ea5b393\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/45090e09897a492a8e5a9d663ac6105d : SUCCESS in 59m 10s (non-voting)\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/cf4ec302c47d46f7af91997aa028ee8a : FAILURE in 42m 23s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/83ae36f748a74d6f92e40d01319500e5 : SUCCESS in 46m 09s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/0db886abe29a4f27a57566fd0e2a104c : SUCCESS in 1h 14m 45s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b082109723da4cd0993e99c1f4c838c2 : SUCCESS in 49m 17s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/94cfd6941a914bf79026810738145502 : SUCCESS in 16m 29s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/a840778fc7bd4a418ce7daf50377ec17 : SUCCESS in 57m 36s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/625962fe639147b984db123d488691c6 : SUCCESS in 1h 25m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f9be4d537034552abc7a65def29fe41 : SUCCESS in 2m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/41bf29bbdeae4ec385145c7cf425bf06 : SUCCESS in 2m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9b4ea269c6f34fa488ac8e771b2cdc3a : SUCCESS in 5m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3936b40809b94e98a62eb6653d35fae5 : SUCCESS in 6m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/858cf24c67d1449bbc4d2f3e800fdb1c : SUCCESS in 5m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9c9043e2976649659fa23463f51f3ddb : SUCCESS in 4m 58s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/eccf359971264dd49fbcd41e9d10955d : SUCCESS in 4m 54s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/7c15f0a7b421440bbabcd8445024b6f4 : SUCCESS in 1h 11m 31s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/40802b291f0542df934994990ca97a51 : SUCCESS in 1h 17m 47s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/07bdedd47fba4d36a032d186a32bdcc5 : SUCCESS in 48m 04s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/2e214573cea54f228c380ec3f587fd9f : SUCCESS in 1h 34m 16s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/d38273fe805e48759ace6c22e9899073 : SUCCESS in 2h 22m 50s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/9a376ad6b76e45da8dcdee90c4ead2a8 : FAILURE in 2h 03m 35s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/531dac2c76744d40b4e7e882ce57da00 : FAILURE in 1h 22m 19s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/e3347b6b52464ad79eaccddb50769dd5 : SUCCESS in 30m 48s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/53da0df0f5f547d39496375f56e22891 : SUCCESS in 1h 34m 12s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/5c52d29b495e43dd8cba2bef8aa2adaf : SUCCESS in 1h 39m 53s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/95892e8eed1f41e7aa57c280fcf5efb7 : SUCCESS in 1h 37m 54s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/c4b991b6f12e41668340e82da024788e : SUCCESS in 33m 06s (non-voting)\n- kolla-ansible-rocky9-cells https://zuul.opendev.org/t/openstack/build/985594c6d71f4b3980519b33fd5e4953 : SUCCESS in 1h 36m 29s (non-voting)\n- kolla-build-rocky10 https://zuul.opendev.org/t/openstack/build/f71b4267392d4c5ca4e0f80c15d7edda : SUCCESS in 45m 21s\n- kolla-build-rocky10-podman https://zuul.opendev.org/t/openstack/build/21ce42b087a5483eae788aef20384bf9 : SUCCESS in 45m 39s\n- kolla-ansible-rocky10 https://zuul.opendev.org/t/openstack/build/923793cdbc4e4676898633da6508171d : SUCCESS in 47m 37s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/2491c7f3ddd54ff5b2703d1c6a9b1cd6 : SUCCESS in 51m 35s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/3a477e1e452c46fbb52a3b1ae7aa9304 : SUCCESS in 24m 46s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/625f0fb22cfc49eaa2dc1f87eb317cc6 : SUCCESS in 1h 04m 48s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/49c6d15b830941c8b46cdd81c06c1ac6 : SUCCESS in 1h 44m 08s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/76820b44ddb64949869c1d3c308e22c1 : SUCCESS in 2h 00m 35s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/7dcd58ddded742409c6b651d17913d57 : FAILURE in 2h 19m 24s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/e71448cc2fe24d4da16aac8794ee2af8 : FAILURE in 40m 37s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/db497006e361468ca914ca384d063656 : SUCCESS in 29m 12s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/b7fd21588261446098735107fa343e17 : SUCCESS in 1h 16m 07s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/6765cdecbb934e838a586fed1c3b4407 : SUCCESS in 1h 20m 33s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/3c4ca70ec50b4df7b5c957c1f63d3339 : SUCCESS in 1h 40m 33s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/f5a283aec8124266ab9ce8960f70dd3e : SUCCESS in 37m 12s (non-voting)\n- kolla-ansible-ubuntu-cells https://zuul.opendev.org/t/openstack/build/e2671a9d435a43b181e66844d0aec52b : SUCCESS in 1h 29m 02s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/3c24d9b0b1d940a8b8935d07bbc29476 : FAILURE in 1h 03m 47s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"0d745e01873fea078458c2e75dece0055a735289","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2026-03-09 10:50:22.000000000","message":"Topic build-proxysql-3-in-stable removed","accounts_in_message":[],"_revision_number":5},{"id":"a1a38a767d9479f00ee8f75deae99c689b4d5c14","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2026-03-09 10:50:30.000000000","message":"Topic set to proxysql-3-epoxy","accounts_in_message":[],"_revision_number":5},{"id":"2b088eb748653320b2fa7520da39f873e0811940","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-03-25 14:38:34.000000000","message":"Patch Set 5: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":5},{"id":"103a3f49155d7870474e4bde6896112dafdfbab9","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-03-25 14:40:22.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"d66998fc3317441a2ccb745c1a2c77f8bc8e74e3","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-03-26 21:40:25.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"3b491474b5d7adc37612be0416c3127c3077f0d7","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-03-26 21:41:34.000000000","message":"Patch Set 5: Review-Priority+1","accounts_in_message":[],"_revision_number":5},{"id":"eae6124998f34fb84181aeb81e1e974e12098b37","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2026-03-27 09:41:31.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"2379f50af4efe868c1794105af846d77d482a1d2","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-03-30 07:52:09.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"78b728df80cb3334644909695191fba0cc1a515e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 07:52:21.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"418d1b77a1e52faa399dbdf345db2c2cd9bc54c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 10:10:37.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32f48a40a9d645e3820385cd6eac282b\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/adf6f1d36e1e4808b38fd812991f97ee : SUCCESS in 40m 13s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/52946ee4216d4ab39443b823bdaa94df : SUCCESS in 17m 36s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/92b33e55901445a6946671231182f1c1 : SUCCESS in 59m 44s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/ce40639f22a0469fb8853b94026ada37 : SUCCESS in 1h 30m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6b33614d98c4352b1f2cb8424f073d2 : SUCCESS in 3m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3f9370ce9bd2459090ff395a9b32ab8c : SUCCESS in 3m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1087edeaf0374795b6dfa5a08543ebbc : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4d268dd17a14efaad99919229b2179d : SUCCESS in 7m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a1e05abebebf4f359de23d5153eccef0 : SUCCESS in 3m 41s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/cf761772a0c842038ec4a8c5fb96edb6 : SUCCESS in 1m 52s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/f51fd8fe7f9a4bbf90b9ffb07b3bb618 : SUCCESS in 43m 00s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/cb564410c740410eb261af7f04d3a3b3 : SUCCESS in 1h 30m 27s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/a42a209517d44cfc8c1c2878b78c3654 : SUCCESS in 1h 30m 51s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/85931c68c06541319885b16de8dbdab9 : SUCCESS in 2h 10m 07s\n- kolla-build-rocky10 https://zuul.opendev.org/t/openstack/build/3fdce49c75ee4e6ab0b4ccd35f934476 : SUCCESS in 53m 56s\n- kolla-build-rocky10-podman https://zuul.opendev.org/t/openstack/build/d73d18e4908644cd8ed6b440b94fbb11 : SUCCESS in 54m 12s\n- kolla-ansible-rocky10 https://zuul.opendev.org/t/openstack/build/06dafb368a03428fb73585fb5ca42158 : SUCCESS in 46m 41s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/20ddb36fb56f409f8b05bcea7aae45d8 : SUCCESS in 56m 05s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/a8cac81fb7e94009a9da71d9a08006dc : SUCCESS in 31m 19s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/bd7ef367b84b477a9df3685bae9d99a5 : SUCCESS in 1h 12m 57s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/1b3c54226b8e42d3b0cf492db4710b78 : SUCCESS in 1h 49m 48s","accounts_in_message":[],"_revision_number":5},{"id":"ec69800dbe024a5f5fda44c0c5c61bf310ac995d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 10:10:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"b86214ee00b78059c4a03adb2241814aca17cacb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 10:11:47.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d9a6c8d62964bedb65da4a6ab9906eb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bf74ad8e9034054b9f1c439efa9a633 : SUCCESS in 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f72e3a3834c846f58c42758f52f485b7 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"dbbbcdc1d1c74af40a7e4afe47143918366661f4","revisions":{"af1a6be18b7396b2b5132bf350d351dabe81f47a":{"kind":"REWORK","_number":1,"created":"2026-01-23 14:58:31.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/29/974429/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/29/974429/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/974429/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/974429/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/974429/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/29/974429/1"}}},"commit":{"parents":[{"commit":"7905d651e5193d3865bdbbc13530d5f768fe928a","subject":"Merge \"Update README with new mission statement\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7905d651e5193d3865bdbbc13530d5f768fe928a"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2026-01-23 14:51:35.000000000","tz":0},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2026-01-23 14:51:35.000000000","tz":0},"subject":"Build ProxySQL 3.0.x in stable releases","message":"Build ProxySQL 3.0.x in stable releases\n\nProxySQL 2.7.x has a bug that it fails to verify DB certificate when DB\nuses chain of certificates (e.g. intermediate certs) [1].\nThis was fixed in 3.0.x but it was decided not to backport the bug fix\nto 2.7.x [2].\n\nThis patch adds ProxySQL 3.0.x to stable releases, so users who use\nProxySQL as DB loadbalancer can choose to upgrade it to get the bug\nfixed.\n\n[1] https://github.com/sysown/proxysql/issues/4877\n[2] https://github.com/sysown/proxysql/issues/5014\n\nChange-Id: I064ef350fc8ec35f88606f7e8c3fd6f62bc6a6c0\nSigned-off-by: Seunghun Lee \u003cseunghun@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/af1a6be18b7396b2b5132bf350d351dabe81f47a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/af1a6be18b7396b2b5132bf350d351dabe81f47a"}]},"branch":"refs/heads/stable/2025.1"},"528a76f32b2d6fc5369131a4fca604cfaab1e4dd":{"kind":"REWORK","_number":2,"created":"2026-01-23 15:40:34.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/29/974429/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/29/974429/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/974429/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/974429/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/974429/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/29/974429/2"}}},"commit":{"parents":[{"commit":"7905d651e5193d3865bdbbc13530d5f768fe928a","subject":"Merge \"Update README with new mission statement\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7905d651e5193d3865bdbbc13530d5f768fe928a"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2026-01-23 14:51:35.000000000","tz":0},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2026-01-23 15:38:40.000000000","tz":0},"subject":"Build ProxySQL 3.0.x in stable releases","message":"Build ProxySQL 3.0.x in stable releases\n\nProxySQL 2.7.x has a bug that it fails to verify DB certificate when DB\nuses chain of certificates (e.g. intermediate certs) [1].\nThis was fixed in 3.0.x but it was decided not to backport the bug fix\nto 2.7.x [2].\n\nThis patch adds ProxySQL 3.0.x to stable releases, so users who use\nProxySQL as DB loadbalancer can choose to upgrade it to get the bug\nfixed.\n\n[1] https://github.com/sysown/proxysql/issues/4877\n[2] https://github.com/sysown/proxysql/issues/5014\n\nChange-Id: I064ef350fc8ec35f88606f7e8c3fd6f62bc6a6c0\nSigned-off-by: Seunghun Lee \u003cseunghun@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/528a76f32b2d6fc5369131a4fca604cfaab1e4dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/528a76f32b2d6fc5369131a4fca604cfaab1e4dd"}]},"branch":"refs/heads/stable/2025.1"},"8a76d146d83c80de9ce85adaeba6bb7556b921b4":{"kind":"REWORK","_number":3,"created":"2026-01-27 12:02:29.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/29/974429/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/29/974429/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/974429/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/974429/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/974429/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/29/974429/3"}}},"commit":{"parents":[{"commit":"7905d651e5193d3865bdbbc13530d5f768fe928a","subject":"Merge \"Update README with new mission statement\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7905d651e5193d3865bdbbc13530d5f768fe928a"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2026-01-23 14:51:35.000000000","tz":0},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2026-01-27 12:02:20.000000000","tz":0},"subject":"Build ProxySQL 3.0.x in stable releases","message":"Build ProxySQL 3.0.x in stable releases\n\nProxySQL 2.7.x has a bug that it fails to verify DB certificate when DB\nuses chain of certificates (e.g. intermediate certs) [1].\nThis was fixed in 3.0.x but it was decided not to backport the bug fix\nto 2.7.x [2].\n\nThis patch adds ProxySQL 3.0.x to stable releases, so users who use\nProxySQL as DB loadbalancer can choose to upgrade it to get the bug\nfixed.\n\n[1] https://github.com/sysown/proxysql/issues/4877\n[2] https://github.com/sysown/proxysql/issues/5014\n\nChange-Id: I064ef350fc8ec35f88606f7e8c3fd6f62bc6a6c0\nSigned-off-by: Seunghun Lee \u003cseunghun@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8a76d146d83c80de9ce85adaeba6bb7556b921b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8a76d146d83c80de9ce85adaeba6bb7556b921b4"}]},"branch":"refs/heads/stable/2025.1"},"df3c0ddc36cddd188ad1906358bcafd90136b9d5":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2026-01-27 13:29:47.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/29/974429/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/29/974429/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/974429/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/974429/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/974429/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/29/974429/4"}}},"commit":{"parents":[{"commit":"7905d651e5193d3865bdbbc13530d5f768fe928a","subject":"Merge \"Update README with new mission statement\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7905d651e5193d3865bdbbc13530d5f768fe928a"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2026-01-23 14:51:35.000000000","tz":0},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2026-01-27 13:29:09.000000000","tz":0},"subject":"[2025.1 only] Build ProxySQL 3.0.x in 2025.1","message":"[2025.1 only] Build ProxySQL 3.0.x in 2025.1\n\nProxySQL 2.7.x has a bug that it fails to verify DB certificate when DB\nuses chain of certificates (e.g. intermediate certs) [1].\nThis was fixed in 3.0.x but it was decided not to backport the bug fix\nto 2.7.x [2].\n\nThis patch adds ProxySQL 3.0.x to stable releases, so users who use\nProxySQL as DB loadbalancer can choose to upgrade it to get the bug\nfixed.\n\n[1] https://github.com/sysown/proxysql/issues/4877\n[2] https://github.com/sysown/proxysql/issues/5014\n\nChange-Id: I064ef350fc8ec35f88606f7e8c3fd6f62bc6a6c0\nSigned-off-by: Seunghun Lee \u003cseunghun@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/df3c0ddc36cddd188ad1906358bcafd90136b9d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/df3c0ddc36cddd188ad1906358bcafd90136b9d5"}]},"branch":"refs/heads/stable/2025.1"},"dbbbcdc1d1c74af40a7e4afe47143918366661f4":{"kind":"REWORK","_number":5,"created":"2026-02-26 12:07:24.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/29/974429/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/29/974429/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/974429/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/974429/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/974429/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/29/974429/5"}}},"commit":{"parents":[{"commit":"7905d651e5193d3865bdbbc13530d5f768fe928a","subject":"Merge \"Update README with new mission statement\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7905d651e5193d3865bdbbc13530d5f768fe928a"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2026-01-23 14:51:35.000000000","tz":0},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2026-02-26 12:07:18.000000000","tz":0},"subject":"[2025.1 only] Build ProxySQL 3.0.x in 2025.1","message":"[2025.1 only] Build ProxySQL 3.0.x in 2025.1\n\nProxySQL 2.7.x has a bug that it fails to verify DB certificate when DB\nuses chain of certificates (e.g. intermediate certs) [1].\nThis was fixed in 3.0.x but it was decided not to backport the bug fix\nto 2.7.x [2].\n\nThis patch adds ProxySQL 3.0.x to stable releases, so users who use\nProxySQL as DB loadbalancer can choose to upgrade it to get the bug\nfixed.\n\n[1] https://github.com/sysown/proxysql/issues/4877\n[2] https://github.com/sysown/proxysql/issues/5014\n\nChange-Id: I064ef350fc8ec35f88606f7e8c3fd6f62bc6a6c0\nSigned-off-by: Seunghun Lee \u003cseunghun@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dbbbcdc1d1c74af40a7e4afe47143918366661f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dbbbcdc1d1c74af40a7e4afe47143918366661f4"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
