)]}'
{"id":"openstack%2Fkolla~975940","triplet_id":"openstack%2Fkolla~master~Iee851c09e63817cd03b780163392517bf35cdfef","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2026-02-10 12:18:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iee851c09e63817cd03b780163392517bf35cdfef","subject":"[release] Use UCA Gazpacho","status":"MERGED","created":"2026-02-06 18:21:39.000000000","updated":"2026-02-10 12:19:23.000000000","submitted":"2026-02-10 12:18:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"975940","meta_rev_id":"2a062b99fc8fb1ae101de3f202c4f38280459ed6","_number":975940,"virtual_id_number":975940,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-10 12:18:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2026-02-10 10:57:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2026-02-10 11:35:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2026-02-10 10:57:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":-1,"date":"2026-02-10 10:57:59.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2026-02-06 18:23:07.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-06 20:00:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-02-06 20:31:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-10 10:57:59.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2026-02-10 11:35:09.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"2e7de874e501fe1ca4eedd42b123fa7eb1235c01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-02-06 18:21:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9ab4903870bd73588fc5789c52118e0344ddc04d","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-02-06 18:23:07.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"63645e7bf02d995d45e91e572c799fd361c38995","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-06 20:00:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01b1cf842e644d899671498404873885\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/979789304e744bcd8e5de86c3bf3497b : SUCCESS in 1h 29m 45s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/6557ccecf73d4b06a7a1c230ac1d56a6 : SUCCESS in 1h 23m 59s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/10fb00597862499f9a44051ccd2c35dc : SUCCESS in 1h 22m 39s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/d3c71b2c4d704b519425bc1a544cf9d9 : FAILURE in 8m 59s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/3a68b559814e4d02a09c9b017dfa0edd : FAILURE in 1h 36m 49s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6c174f0c1fc048f1a1d3b533b40ab14b : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d5cc570670b04c0f98cab295b7beb470 : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7bac5b6176d641da8b904eb04475f1bc : RETRY_LIMIT in 2m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5035c0534e38944f60114b1560803d09519b5325","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-06 20:31:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b674ab8707e34390b72fcfd68f9ba8db\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/7fe421648b4d4dbbaa3fc3d6bc49b793 : SUCCESS in 51m 49s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/4e8b052eab094ecb981ed856fd0abf51 : SUCCESS in 48m 46s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/45867642cbec4b959ef25a11b8f605a6 : SUCCESS in 29m 02s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/db10d50cf16f45eab3a6a20b07d4f53e : SUCCESS in 40m 44s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/ef1fd165c69a4c67964fd8cafc90327a : SUCCESS in 1h 10m 36s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ce08eb7555a34106a8541441d67d85ce : FAILURE in 4m 54s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b0ed269485b245b598732c3d97117eb4 : FAILURE in 2m 47s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/0d5473c84d0b4deb9578398a5be62ac4 : FAILURE in 1h 05m 15s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/2740e88d806247c6a3c5517ff111094a : SUCCESS in 1h 04m 48s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/f325f4307d8e411f8b5d9bff5dd8d283 : SUCCESS in 1h 22m 57s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/3e78e67dafd04cf998ee22cc961b43fe : FAILURE in 8m 50s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/a3adeef7343a4a77b25c32a685ae736b : SUCCESS in 1h 27m 25s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/103ce02398aa45e095bc8b0ad3bd6b22 : FAILURE in 10m 21s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/863e721a685c418a93fd25de953767a4 : FAILURE in 10m 11s\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/903c7c651ed34879a135a524c277bcfd : SUCCESS in 1h 34m 22s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/1f5886b32339493fa790ef8240ee394a : SUCCESS in 2h 08m 32s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/5570becd6f674ec0a9deb56104a77bc2 : FAILURE in 12m 28s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/4b8b9c9a11b148c1bd2ccfa2d18216d4 : FAILURE in 10m 22s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/759d5b7729b54f99b12769dce93f9fc8 : FAILURE in 1h 11m 09s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/42f3e5df64ad4cd98ac04eee27684a16 : FAILURE in 9m 57s (non-voting)\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/800ef822aeea45b4bc7a18653a1e5391 : SUCCESS in 30m 28s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/54b9a03a7a574b299c1701cee7c26511 : FAILURE in 10m 36s\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/aec49ab6ad6c4efb81f153802230bc1a : SUCCESS in 33m 01s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/66998ba2a1fd47649983b17d9467f1de : FAILURE in 10m 26s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/3b1a6f39bb824780b898adc77ad12581 : FAILURE in 8m 21s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/ff5f101e559e41739bad3444ff7bb9c8 : FAILURE in 36m 51s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/1ac8e0f20e714c79a300a6de76ed1785 : FAILURE in 8m 33s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/7868b63704524d8286bc397d219dca40 : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ff0cf33124848d2b5b6720cae904d98 : SUCCESS in 3m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bab925c7721b4b91824575c57c44aa58 : SUCCESS in 2m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c888e9b3ac244a98bf92f02a1f39b244 : SUCCESS in 3m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f1d3368aacb04cfe85181d72e6d199f1 : SUCCESS in 3m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7281ee858adf494c972c334ab1cadc80 : SUCCESS in 6m 29s","accounts_in_message":[],"_revision_number":1},{"id":"70208d844b3ff26827ecb97983d44c6d222eba01","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-02-06 21:31:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"006d5a0160661a8bc567d4f9844edbac7ba04579","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-02-09 08:19:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"88de22be57d90e67e27067fc63c938ef4b499b08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 09:20:03.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c14430e27f3d4451b18035e89ecdae74\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/6370b96316c24d8d9e875ae9fe1f4c78 : FAILURE in 50m 12s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/9c6bef7b73684367a6172b0565c1f5eb : FAILURE in 55m 59s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/784207cc95584b7cb8a53712f6e07e0d : FAILURE in 28m 06s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/1daa31c066be483987d8a4bc478d0807 : FAILURE in 41m 32s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/65ef5b4349c745c7bd306d9bb01d33f3 : FAILURE in 47m 24s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cf210af9c97f4a1384ac97f62a744134 : FAILURE in 5m 15s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0520a81072494d70a24612a373104b0a : FAILURE in 4m 32s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/074d0984a21c4cc1ac547f23a694ceff : FAILURE in 23m 52s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/9b74e5706d9440dea04b88837fd6c0ca : FAILURE in 15m 17s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/2dbe952207584d0d95224c8b4cb0fe93 : FAILURE in 14m 26s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/c385076ddd4a4933b8271e529ca6f378 : FAILURE in 16m 33s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/566e23020762438f8d0302865f9ac055 : FAILURE in 17m 32s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/508ca058e21f4bc082828ce1ed0cfd2a : FAILURE in 7m 12s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/99b5d348b3e94c5297c1d6438d705e31 : FAILURE in 7m 53s\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/93f74114fc6a4160aaf29dd89eeb4859 : FAILURE in 18m 24s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/215d222315684249afc7ae9b6243115f : FAILURE in 20m 45s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/049467180f884aa3a79b48c8897d801a : FAILURE in 9m 19s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/f3f70d35d28340a7a40846db301c2ef1 : FAILURE in 10m 26s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/ffbc16b5906e4df8bc6c01e5c30911a6 : FAILURE in 19m 40s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/89cf0437b80145b5b994ca363ff95f61 : FAILURE in 8m 05s (non-voting)\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/6ba0b4596e0248d7a6c0f62d86147203 : SUCCESS in 25m 07s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/2ecd831b1c7e4fa29dcf74c418d463e3 : FAILURE in 9m 47s\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/dfd1257452b14c4caee8a59582144d9d : FAILURE in 22m 05s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/9d475fd4e0b0437f8e764ce231eff637 : FAILURE in 9m 31s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/eb0521380a3e454e8bbeb4b390853761 : FAILURE in 7m 59s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/20f6e8999f7042c2990c87435c076fd4 : FAILURE in 17m 42s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/ec36b683822144a599a5e156982c98a1 : FAILURE in 8m 13s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/d39b0fe8a6874dcc9ba72d3265fd4b10 : SUCCESS in 2m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc5c1c42437f480e957988dd5b98ca2f : SUCCESS in 3m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/41bfe4edc183409784f8132e6f1039a0 : SUCCESS in 2m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dbbc6110bafb423e8950029a0f381a12 : SUCCESS in 4m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3acf9d4a8bf0494ba6cdc2e48cb329c5 : SUCCESS in 3m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d0aef9becd8488b9d2521a6afc919eb : SUCCESS in 6m 24s","accounts_in_message":[],"_revision_number":1},{"id":"38f18673eb23b2200fbe481bc7cc2d15273af3bd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 10:09:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d7d2af21db747188177803597aef9db\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/17241c60d4b04898a4bee713166b9578 : FAILURE in 1h 09m 15s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/d900e051f6a442d8a5cbf1fabb3532c6 : FAILURE in 1h 38m 58s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/edebd072d5fb4b9783a9077b97123eb7 : FAILURE in 1h 45m 16s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/6e150ad696a74a3599e2d90c080390d3 : FAILURE in 6m 25s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/5ec64853a387467ea2e748d9b1144152 : FAILURE in 34m 54s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/eb178842bdb7418e86b84df42df26b15 : SUCCESS in 3m 59s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8c28e4bb3d1841fd91d2fef6d4f3b3b9 : SUCCESS in 4m 11s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7ff980218fdb482285583936719fa40d : RETRY_LIMIT in 2m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f4d01acc18261c78a7980fb9e980f7edfd856bb1","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-02-10 07:48:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6f3c8d0e864d4037920c4b9a06e2c686dbf38f0d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-10 09:47:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b018f94142547b5a42057169eef2abf\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/9c655d0334134088877ffdaff63235f6 : SUCCESS in 1h 56m 35s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/17422702e62a4677aa7042dcb4351481 : SUCCESS in 1h 21m 13s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/cfd9bb69f5024684a2b0396a8edc85c6 : SUCCESS in 1h 01m 13s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/6aad182052c74893b3185b9ef37f08b3 : SUCCESS in 1h 01m 07s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/5d456b35c6374537ac4a39a3136b894c : FAILURE in 1h 55m 07s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0467aeac9a024f9caec869406bfb5858 : SUCCESS in 4m 04s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ef02000413d54da687b7a01ab2eff22f : SUCCESS in 4m 52s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d3cdd56d4cbb421a9f955037c2804dc9 : RETRY_LIMIT in 2m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1535e2668de864326088e1bc10b95a67601f7b91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-10 10:12:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f66d0d503adc41129f5fc69a625e7d1e\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/daebbe73d9f34326a76bdece3b64eae8 : SUCCESS in 41m 41s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/65917d2ae6e0405cb16b7a258e65eb40 : SUCCESS in 38m 22s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/9ce4332f644a47e9892225ab72af490a : SUCCESS in 26m 01s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/53e626bc0918411eb18712772f279ebc : SUCCESS in 43m 56s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/59282943b37946f59f568947a4fbe7e6 : SUCCESS in 35m 13s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5b388c5cdd2f48228c50f182bab26c74 : SUCCESS in 37m 53s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/68603b5cba89430b9d42ae7002bbb54a : SUCCESS in 27m 52s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/fb32ad537e2342ec81f96153e7c59bd6 : FAILURE in 1h 04m 35s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/bf2e08cc0f644ce980e1f4d01ab543a5 : SUCCESS in 57m 34s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/6d5cead26186459f99760ca21693aaa1 : SUCCESS in 1h 15m 05s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/18b030434c644e00b83c98664eb4ae8d : SUCCESS in 1h 01m 52s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/9f73f4fab7aa44f587560d1c23478293 : SUCCESS in 1h 22m 37s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1aab255789004a688595132e374ce2fe : SUCCESS in 1h 13m 51s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/18d31657446645bbb7980fe86ac40ff2 : SUCCESS in 1h 35m 03s\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/c04bf4a538a74a538701c8d6ed633da9 : SUCCESS in 1h 44m 41s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/84bcbfcd2cc54f8ca48738e02f06d25f : SUCCESS in 2h 21m 25s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/a213394ac37143a882e2937b0de2d684 : SUCCESS in 1h 46m 57s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/df959373ab2748089a75968cedf1e821 : SUCCESS in 2h 21m 52s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/f50b5a37467947fab0a5c192c984ee5d : FAILURE in 1h 27m 34s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/30dac0a0b62c4bd7a9dad0223919b8e9 : FAILURE in 1h 29m 03s (non-voting)\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/cc668fd2ab1246bcb8af08a00c5c6169 : SUCCESS in 36m 24s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/373f33af175b415badd9cbab252deaad : SUCCESS in 30m 14s\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/160ea8ae211a439796b2bc1a05681c1f : SUCCESS in 49m 49s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/8d01cb569cee4121b35ac812c8a5ec86 : SUCCESS in 53m 14s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/18a460c95f504d598337697e171fe774 : SUCCESS in 1h 20m 26s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/4de27c98a39441c8b50131e269cea5ad : FAILURE in 53m 15s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/0d8ed66d435e45af9e3391b13f37f4dd : FAILURE in 53m 43s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/fe40403ea1ba40938dd0d742ee30c9ff : SUCCESS in 2m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/713cea04aa8648e89e53b46a9b9063a1 : SUCCESS in 3m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/de8932d46d2c4409a26be8fc736f2571 : SUCCESS in 1m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cd766490c28646608711a073530f9079 : SUCCESS in 3m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/662e908cc42044c99d8426ce96b524fc : SUCCESS in 2m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b11dd70ef9447269b86d0d97c508524 : SUCCESS in 6m 20s","accounts_in_message":[],"_revision_number":1},{"id":"a0950051fc476d30835a10acd0db8ac998c1d328","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-02-10 10:57:59.000000000","message":"Patch Set 1: Backport-Candidate-1 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"14eded46e9becf8de11e00d436cb87c3ad7ca42f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-10 10:58:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"511d5d708ac7856bbf74236b47a4cb0ce7f003b2","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-02-10 11:35:09.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e9ac8804e763cfa17883114420f796e0a6d76861","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-10 12:18:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc2ea4bc85a64fed891afbb6f7c88948\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/c1bea4c198fe40d6bdee2a40a92443d7 : SUCCESS in 50m 34s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/0974370d88c546808751484bba2f1d22 : SUCCESS in 26m 33s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/1cc561b4355145eaad0ed83c27c1477a : SUCCESS in 43m 53s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/7f426f1911d747f9ae91a0497b31f407 : SUCCESS in 39m 12s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/22aa64323ae14195beb180942358162a : SUCCESS in 57m 38s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/97e5c98cf21f43e188e8ccd74243012e : SUCCESS in 26m 54s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/ae01eff0ca1846cb953db8427db54be8 : SUCCESS in 55m 48s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/473f86b2c5734038a1c8600c049aaa3e : SUCCESS in 1h 15m 53s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/af9671a1268346aaad761afba178a89b : SUCCESS in 1h 01m 47s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/b2160db1d1cf46789d4764a2e6446e06 : SUCCESS in 1h 19m 02s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aa06e94e643b431291925ffb19ea1a16 : SUCCESS in 57m 57s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/ad05d5d4f1ad4555b73874761d8be9e0 : SUCCESS in 1h 19m 46s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/9b8b64058af549e1ac541b732472d743 : SUCCESS in 2m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fec296000505419486acd9c7c489a322 : SUCCESS in 5m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c66b0ee72ba14b7c89c245d0f345575f : SUCCESS in 2m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/699fb878317546819dac878ed8c47d52 : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6fc62f0cbf74b239ecc7e090b621924 : SUCCESS in 7m 50s","accounts_in_message":[],"_revision_number":1},{"id":"d65efb9511b64e035299327d2f258c89d8e5efb5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-10 12:18:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2a062b99fc8fb1ae101de3f202c4f38280459ed6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-10 12:19:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f664eef246944e4a38a2bad6c6f1420\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76caea02ece24136911f4ab8e9c211e8 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a056cf58775ef51d927222ff5a2b236e668e843e","revisions":{"a056cf58775ef51d927222ff5a2b236e668e843e":{"kind":"REWORK","_number":1,"created":"2026-02-06 18:21:39.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/40/975940/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/40/975940/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/975940/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/975940/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/975940/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/40/975940/1"}}},"commit":{"parents":[{"commit":"f6f52cdcc2857bde404a4abe7d772f96cd0d61a8","subject":"Fix Misinformation on ProxySQL version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f6f52cdcc2857bde404a4abe7d772f96cd0d61a8"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2026-02-06 18:21:17.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2026-02-06 18:21:38.000000000","tz":60},"subject":"[release] Use UCA Gazpacho","message":"[release] Use UCA Gazpacho\n\nChange-Id: Iee851c09e63817cd03b780163392517bf35cdfef\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a056cf58775ef51d927222ff5a2b236e668e843e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a056cf58775ef51d927222ff5a2b236e668e843e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-1"],"failing_atoms":["label:Backport-Candidate\u003d1"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
