)]}'
{"id":"openstack%2Fkolla~977311","triplet_id":"openstack%2Fkolla~master~I323ea4e7dabecfcf20fa890e028eeb1c3766be0b","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"17669":{"account":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"last_update":"2026-03-30 16:53:19.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-03-24 10:52:00.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-03-25 07:39:36.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2026-03-30 16:53:19.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-03-30 16:53:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I323ea4e7dabecfcf20fa890e028eeb1c3766be0b","subject":"Bump OVN to 25.09 in Rocky Linux 10 images","status":"MERGED","created":"2026-02-19 11:35:51.000000000","updated":"2026-03-30 16:54:45.000000000","submitted":"2026-03-30 16:53:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"977311","meta_rev_id":"16591836537cbbbb7844a07d9d0ee7de6506a4b0","_number":977311,"virtual_id_number":977311,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-30 16:53:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2026-03-25 05:09:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2026-03-25 07:39:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2026-03-30 14:47:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":-1,"date":"2026-03-25 14:14:03.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2026-03-25 14:14:03.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-19 11:40:11.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2026-02-19 11:45:32.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2026-02-19 14:18:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-02-19 16:00:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-20 16:56:21.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-03-24 13:26:02.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"781e79ee8e90e2460b905bbc8b5b2bfec037763f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-19 11:35:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"deaf323ed67b48717522fdc564a7a4690cdf14bd","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-02-19 11:45:32.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f2999aa4e8d7e5aca6df68d8b076ed2f3ee4803f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-19 11:46:28.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b5849fc6b0cb8e5a9110ffbe32a875a1373db6cf","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-02-19 11:48:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d8ae6c4c986984e0e298ce7215e27e888811078e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-19 12:41:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4d233c00040f83ed31bd407b61646863dd6e8364","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-19 12:54:02.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"678524bc1afaede039cb0a7c2e3b2a07981f2f39","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-19 12:54:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0592207d344a35c641b36d243530be8d878cea65","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 14:18:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89ba48fc08d64895b3428b264a602b75\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/085e52dfdcab4783bf22646c2dc88a3e : SUCCESS in 56m 31s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/678e6a4d34134e12be4ff56273e75f55 : SUCCESS in 1h 02m 17s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/511fa643fdc14edcbe4d68a8a41aac84 : SUCCESS in 1h 02m 55s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/cefd728beb884a2c864d47aa45320912 : SUCCESS in 1h 04m 04s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/1e40d6e3cad14443870e66c4f65b04cf : FAILURE in 1h 22m 05s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/95f5085f69cc449a889ff9207c46f79e : POST_FAILURE in 4m 31s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/349d013209134492af2093c6894622ff : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4c5813e0877c439d97bce4dfbb623f5f : RETRY_LIMIT in 2m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fac0317e6f1fd2250009db9227c0ba04672cb593","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 16:00:00.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/11b5e26f2b4b4054808236acd838c59c\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/8349b51031a64f54a7680ceb646e8a69 : SUCCESS in 37m 29s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/dbfcd49ecc564ac697f86d1730372ce8 : SUCCESS in 52m 30s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/b942a523e97a460abcaa1aa320a9ec43 : SUCCESS in 12m 19s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/8ae726d729ee4182a239094def693350 : SUCCESS in 40m 02s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/a81fa880bfb04fe685f5a799eb18bf98 : SUCCESS in 35m 12s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ff67f40b5873413ea5a79734d7869bb9 : SUCCESS in 56m 41s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3bbf40c22325480ab0750bd3ce352940 : SUCCESS in 16m 12s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/7510bb35fee54d28bd4e8e667f486595 : FAILURE in 1h 08m 50s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/751cab293ec641729e3739de1c0c5763 : SUCCESS in 1h 13m 17s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/cb844d79480b41c7bf731c45eec31d40 : SUCCESS in 1h 29m 09s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/482ab2ff833c45c29803935661b84105 : SUCCESS in 1h 02m 14s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/b1e65e1fa08a413a86792c8821d4728c : SUCCESS in 1h 21m 13s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fce4214fa1644a3f9ae48c2b58d62233 : SUCCESS in 1h 11m 18s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/46e2a6ff4f0c4b2ab520ca6bfc1f0c9f : SUCCESS in 1h 41m 48s\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/b73ab89b95664d02bdbc4336fc46e783 : SUCCESS in 1h 38m 29s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/52975cf7277c45a298746e0dc138241b : SUCCESS in 2h 20m 50s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/66339c69a8d94c14a260fa4f2e230d29 : TIMED_OUT in 3h 04m 23s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/7270666f65734c3ea02475dc33555e7a : SUCCESS in 2h 06m 34s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/38f849e7368c4addb8134296533f0f26 : FAILURE in 1h 13m 11s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/1e55e526af3648fd8dd9d8853e590cdc : FAILURE in 1h 10m 30s (non-voting)\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/0de7301390cf4d9c9a86d9a8b46f7250 : SUCCESS in 29m 41s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/bf967416dbbf4e298af6cc516b90ec93 : SUCCESS in 19m 29s\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/b399566607eb4a419bf2d4e3c802278e : SUCCESS in 1h 16m 01s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/e8e5f41edffc4b1eb5278ece4e727f6a : SUCCESS in 1h 14m 17s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/3214613e2ebb4a32bd0f846b72061c44 : SUCCESS in 58m 03s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/a046462007de41699578294f1b3d384f : FAILURE in 1h 15m 15s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/bf4d13bc62a04e2f8328ffa0b98f8915 : FAILURE in 1h 05m 45s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/a676efbca8c34b4b8aa2c5c1cfe7ccd6 : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff7edec1f4cb49b79f432fb18b8413ca : SUCCESS in 2m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0a4bc184b7f2492c89a77c9640573f44 : POST_FAILURE in 2m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/96680a155ecd4fe9956ab30683ac43b4 : SUCCESS in 3m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4c48783ab3f44209ab39c726cfba9633 : SUCCESS in 2m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a85e5ec0df894ce3836e41a1233638d0 : SUCCESS in 6m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b5b4afbeaa4547f4af08ecce396aad91 : SUCCESS in 2m 46s","accounts_in_message":[],"_revision_number":2},{"id":"8dee45e668f7e8d01caadc3c862332a9cb74819c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-20 08:08:38.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1c5e0a1fb8fdda2f383f3df0c8f8c2d79b3943cf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-20 09:43:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f85d870fe8664398ae11edf4191e5f8a\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/39c69dcf5f264e38a85e58aff56612e1 : SUCCESS in 1h 11m 10s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/2bb22040e6f24c2ea15723a0bd36a660 : SUCCESS in 1h 16m 20s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/2a38cebc0cae4bf7b31654bbdd1df148 : SUCCESS in 1h 10m 52s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/6d00341a97f7479a84a6b4fb27f17857 : SUCCESS in 1h 32m 59s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/96d082d588d94adeb377a87cf1690167 : FAILURE in 1h 23m 16s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6e147708308e48d3a1af8c9663189219 : POST_FAILURE in 4m 22s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d19b919f33a7401196711e2aa9432263 : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/dd4a58ada4044fa3a409d5f6ac22a2d3 : RETRY_LIMIT in 2m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e69894919ae4ab3c5426ac394169547e1530a983","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-20 10:46:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d6ec196c68d4d76a174108872fccc16\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/168ebde2def344cdab7d5c26a67efaa5 : SUCCESS in 38m 00s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/d058e14b7f6f40dfa5addc150fdf221d : SUCCESS in 55m 33s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/bd692c6f97f848598739b860fa8e2095 : SUCCESS in 13m 00s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/54728b7d807c493d8fcaf32f185e3c09 : SUCCESS in 36m 40s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/baec93ad3f7346969d0050b0461d85ed : SUCCESS in 36m 58s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/94b260c3f5064915a86db8cf7429fca8 : SUCCESS in 53m 55s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a6b0cee45c1f4b9f8ff3f760ec90bba4 : SUCCESS in 30m 02s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/ed152ad9edd2416f98c96698bba289ef : FAILURE in 46m 15s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/18ee2489cbc148579f7319f768ab19b1 : SUCCESS in 1h 27m 07s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/93ff499f28044079a0ba1590d996d4c7 : SUCCESS in 1h 44m 19s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/f1d8e82428e34bf7b40d1227091d1f7b : SUCCESS in 1h 28m 17s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/9e1bff2678134728bfab2bd60612de3c : SUCCESS in 1h 53m 14s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ef8dbc775b2340eaa7ed225ee32a81d5 : SUCCESS in 1h 29m 40s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/9b86d2ffa9404345b5e50bc1b93ce56f : SUCCESS in 1h 50m 45s\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/7202ba197e6e4a21b9fd3f00d8b25173 : SUCCESS in 1h 53m 28s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/9903bc26bf204f9d92ed22cdfdc3bfd3 : SUCCESS in 2h 14m 12s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/06d6a6427a7349c6a6f461941e934ffa : SUCCESS in 1h 49m 39s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/fdddcb1a329741278fda5e4c0886db88 : SUCCESS in 2h 35m 51s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/81d04966c37047e381e7b6ecabba07cf : FAILURE in 1h 21m 27s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/7b4769d99af843be945b32a8882bd829 : FAILURE in 1h 12m 25s (non-voting)\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/148276f272f4462e9509feebde4f52a7 : SUCCESS in 29m 30s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/323f497248ff499c8ecc9ee9fdcdd9c6 : SUCCESS in 31m 15s\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/be382b9c5cc644b09b99f4a793ad0f8b : SUCCESS in 34m 49s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/10a7c3ee1d9a4deb88637071a5a6d451 : SUCCESS in 1h 14m 17s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/c56e3fb620644f3db15a84955eca238c : SUCCESS in 1h 01m 49s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/81c2fcb05d1240f0b4389eb6f84b4e6b : FAILURE in 1h 05m 12s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/f40f7612254844d19da8276d520a2fac : FAILURE in 1h 08m 45s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/163a948f5f294474b32f020ed4e2a0e9 : SUCCESS in 3m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5176060e5a447569f9f5e4a1cb66d14 : SUCCESS in 3m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/be84de9e0ce64a40a6db065d9d81e156 : SUCCESS in 1m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b33202348a474888b15b560c8a124736 : SUCCESS in 3m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3907f5c211ad4ed7976d779b58fded88 : SUCCESS in 3m 22s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/de0feae46d594302a4c4099bfc561dc0 : SUCCESS in 2m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/995fc3611d4344c08efc4275830d0f36 : SUCCESS in 6m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/964776bee805458898a0e9c456f43d31 : SUCCESS in 2m 11s","accounts_in_message":[],"_revision_number":2},{"id":"c61e0000eb721f420f60c9f192e6e1fc8af0a39d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-02-20 16:56:21.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"1b5d2c95a789b60afc6ea59d4a806e28f0175522","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-02-23 09:54:11.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3cae8b0422be2e3e1583839c3831bb2a869a0266","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-23 10:42:41.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":2},{"id":"8c09fa1fa31c2831f3d01ffcb661a569ef7ad0dd","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-23 10:42:50.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"726ba625285ff862c41298026ed325f64ecbd17a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-03-06 17:26:09.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":3},{"id":"755196d8e06cac124676a57df9fb692dffdf5016","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-06 19:21:18.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27435b572d0b411caa630cd4bb2c6297\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/33c753aa1491493594295b7b1bc7a99f : SUCCESS in 51m 08s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/7310e77c852641e4bda8fc0e2b14fa01 : SUCCESS in 1h 16m 12s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/ef25f4afc29449fcbf1e8d2edf494809 : SUCCESS in 52m 24s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/c7d7a4fb3c2e4490a95b11058e65467e : SUCCESS in 53m 24s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/d6ccf7993de84780b2b27536e4bcd785 : FAILURE in 50m 49s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b2be29864075458bbe75a5a68d1fe3d1 : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4f84eddcdffd4a2c941a280058da4bba : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/4880ab914db4485ca466c2afce542db0 : SUCCESS in 7m 41s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e1abf245703759355d3a01c42867ba65a3b63b3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-06 19:53:42.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/be74ef0adc2d496a8455f317a51fe386\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/fad6d392724a4b7b8436022b8293532a : SUCCESS in 39m 48s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/30b2c02d5c7a48e2ac0e2d316f3cf62c : SUCCESS in 35m 05s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/8ef81883ecc64787b7e722e75bd54ee5 : SUCCESS in 13m 51s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/647f5f9c8bee432a912853c0a738c687 : SUCCESS in 40m 32s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/8f2293d8bbaa424695b43c4152a2447e : SUCCESS in 35m 01s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f8b6fb2b08d1491a81ec962588cc035b : SUCCESS in 1h 02m 52s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/571dbc22d3ca4b9ca00480b591a45cb5 : SUCCESS in 28m 21s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/05de12f9fa2f404e8341369e87d4db9d : FAILURE in 25m 08s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/5f7306d3241746068374a0ff4ce93837 : FAILURE in 40m 43s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/e5ff4eaa1da34c2ca13eb9ecb56bcb1b : FAILURE in 1h 21m 06s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/0dd8982166eb4b8499190c008f50b3c9 : FAILURE in 45m 13s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/9a38ae1482bf45bbb76956b7beb331a1 : FAILURE in 1h 28m 26s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/518bb159896e46fc853f862ed4f1cebd : FAILURE in 59m 05s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/d95acad63a9c42a992a85ac386df0591 : FAILURE in 1h 45m 17s\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/b16334427381475f8ab0923d68147ceb : FAILURE in 1h 21m 33s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/dbc94e32396c4111b5d62a2b25763d10 : FAILURE in 2h 25m 20s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/e2b2d140a62c421692753b0d072712c3 : FAILURE in 1h 21m 03s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/6e99b82a72fb437cb2ea01ba6ec056db : FAILURE in 2h 25m 50s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/56b99f9f9ea8417282a062318beeb80a : SUCCESS in 1h 31m 28s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/a85e8b6e3d0c4497af345b46cad5753f : SUCCESS in 1h 08m 06s (non-voting)\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/18c618aa2ee749958ddd6f8210f95c29 : SUCCESS in 25m 49s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/6699e433466647f5aab30c081de0e4be : SUCCESS in 17m 19s\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/4f5acd6522b445dc9dcda4b42d18e021 : FAILURE in 34m 05s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/861a7b9c70eb4da8af6bbc5d6f3d96f4 : FAILURE in 36m 26s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/374c4fce292d4562aeb86eea8b49fa41 : SUCCESS in 57m 00s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/6eec240b83624c0eb13dcad93b118bdb : FAILURE in 54m 30s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/277962c83e644ea98eb67711daf7027c : FAILURE in 52m 43s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/0d208605dded43ab9f0fe1ca70a630e4 : SUCCESS in 2m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb623281c1134928b86dc29a0e1fc380 : SUCCESS in 2m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/774f0708730448d18983494ba97060e5 : SUCCESS in 2m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/92610757a24a4c2bbdc229e5224931dc : SUCCESS in 2m 37s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/aeacfd5fd8494de19fd5e016e3e7de24 : SUCCESS in 4m 40s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e09164cc93a94777b2ae422e8f115a05 : SUCCESS in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b318597551794bbcbd8f0b7327c90410 : SUCCESS in 10m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0a8ad1ed7d7d41ba819aaeecb6ec58d9 : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":3},{"id":"7a0afd5b3add701d67ce468805e0691c496d97e8","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-03-07 08:22:08.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"aa8fde0648d90865fee5088b054a8e772fdd60d8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-07 09:58:54.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c32dc571cce4f9ba55aec96d9755721\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/5548ee1cefbf421f97480f8bc55bb91c : SUCCESS in 51m 11s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/f619fe4b158b46e7a3870cfe6374f192 : SUCCESS in 1h 34m 30s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/2fa9a7e414dd4ef0886e6e003a0138a0 : SUCCESS in 1h 21m 33s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/de3c616c0eae4f70a8b63d9d79eeb0ec : SUCCESS in 1h 00m 41s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/1c906c2b021e4665afa187fcb4936500 : FAILURE in 1h 33m 32s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0baa878947224650bea01ed0cc6b09ab : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/26b6f281e97940388d709e7884b7c7aa : SUCCESS in 5m 47s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/9cd1e835e5054d0a9f2c576ed7b56ea4 : SUCCESS in 10m 00s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8ecb213b9f9e0a98c414919032d73a13444deb0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-07 11:05:50.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ae9b71a2229446a1bed2869a7b2ed53a\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/9b222534f44a4c2d89637d50a9d30605 : SUCCESS in 43m 17s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/3577553b8d9844f3bb6e5f9171ffb789 : SUCCESS in 53m 10s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/36a37c401d6a4a7d8ee6afc33545ca43 : SUCCESS in 24m 11s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/53135609e40c488b862fe21fc01898af : SUCCESS in 41m 28s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/6228d3185db94013a8e57f625b8ebd8c : SUCCESS in 36m 50s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/48af2e1462ef46f5a8794fda913f6966 : SUCCESS in 52m 03s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4ca3fd31e1704f79a4b38e5f2a52f73c : SUCCESS in 17m 22s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/5f398e2e5f37463fad08f460d707bd6d : FAILURE in 35m 00s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/9414a00e3db84bf0bd3451040a2e5e3a : FAILURE in 1h 06m 23s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/7605401e6f6a408699dae53d2fc2cd33 : FAILURE in 1h 13m 38s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/e110ff6369424a349616f4b25980c86e : FAILURE in 1h 07m 24s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/13053c397d2541e5a2ed953182c4c2bd : FAILURE in 1h 23m 06s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ce82e7b4abd64d2aa46e8e036eda799b : FAILURE in 34m 19s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/730a1fecef164815be8204b251d5e9d3 : FAILURE in 1h 47m 37s\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/8abe78aea1a649a891413cae5c5476aa : FAILURE in 31m 10s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/3c567d3182064424ab32419e865437ee : FAILURE in 2h 42m 22s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/a7e50b34b2b3467a82527f2eab26432b : FAILURE in 1h 31m 33s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/14909fe02873479b88f6428b38b79f9a : FAILURE in 2h 34m 58s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/2e6ccfc3769642859005ff594d4f031f : SUCCESS in 1h 40m 23s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/446ba13aac3a4b2b82d42f1ce0c2b3b3 : SUCCESS in 1h 09m 22s (non-voting)\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/89cd092dd0bb4505929865fcf010798b : SUCCESS in 32m 43s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/5ff809c65090449b865f0b3db8fe8acd : SUCCESS in 30m 59s\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/81b22c79fd7f44ecbc44415441242f00 : FAILURE in 42m 23s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/dfa5ca264eee4a00b270e312e4ad05b2 : SUCCESS in 1h 09m 32s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/48646421267846fbbb1b3cff4766fc15 : FAILURE in 36m 23s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/2cac822c62c64945b43cf12c94e3e044 : FAILURE in 53m 20s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/b7525e27aaf64073a38eb089b7d14e87 : FAILURE in 52m 01s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/057f301bbd394e9588867605a9e95dfb : SUCCESS in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8504ef030bda46668dd503aa6ea6d3b2 : SUCCESS in 3m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ffde18b1f7054736ae94edb8e06825e1 : SUCCESS in 3m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c1044acc7d134bc28de1cef53994701f : SUCCESS in 3m 20s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f609697fc3ab495ba14afbd513d7a919 : SUCCESS in 5m 46s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/03057fe1099644a3b892c16aa4700069 : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce54e7e610894006bdbfa5f8b5a0c012 : SUCCESS in 4m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b799e019260946e2b8dc0bf788d12631 : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":3},{"id":"3ef0d207d95d990acc0911a37cb2d39d3c601fa4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-03-18 11:02:22.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"d009e89434f196f24a671b7a51e79fb241d7dcde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 13:41:58.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/68dd532f6aa347c9812b84db8608cdac\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/027bcdb38fc841fcbf15a6d60f922576 : SUCCESS in 53m 31s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/b8a6bfff365d45fe8d15bdb6f5545f44 : SUCCESS in 46m 38s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/e7e96083b2154db88a5f9c11bfcb0d38 : SUCCESS in 23m 52s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/67d98ab470fe44ccb346516f052eab7c : SUCCESS in 47m 57s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/f203b6b7ddba45418cfc488940efa310 : SUCCESS in 59m 14s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/28835363a900452ba80d7a6b84996822 : SUCCESS in 39m 09s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d1583ea9cd0145bc961bd2986fb51d23 : SUCCESS in 22m 59s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/aa2e20697fcb484ea3b6f9aaafd36a8d : FAILURE in 54m 05s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/620b2d083b894191a1fc04d3d5b7ba7b : SUCCESS in 1h 24m 01s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/c4fbfb1a044a40d7b82afeb8174bf0ff : SUCCESS in 1h 31m 05s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/f2760541748e425da5e3d205c131069d : SUCCESS in 1h 31m 21s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/110979eaf6bd417c877cb069f602e441 : SUCCESS in 1h 57m 21s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d67427d900894438901229dfea83e3de : FAILURE in 1h 22m 27s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/f82d2c0eedbd43f8b4ac2725254769da : SUCCESS in 1h 36m 26s\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/d4b59d1051bf4d9bb4f5153ecbaa04eb : SUCCESS in 1h 47m 14s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/346562f494444e5abf83c52dbf6d3d48 : SUCCESS in 2h 31m 11s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/a835bccd7afc4dd284651d4e8d003a1a : SUCCESS in 1h 50m 20s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/7efc5a43d5c14ba9844f1640b05cc8b2 : FAILURE in 2h 01m 24s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/91168e95ede84b1ba7ae8cf2ec63a9d8 : SUCCESS in 1h 12m 25s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/922e87f6fbb6445fbfc71bcc8d9ad56b : SUCCESS in 1h 43m 34s (non-voting)\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/610d38281977473aaef4d983ed939561 : SUCCESS in 47m 05s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/18eab77dd72a47b9a221190634874e6e : SUCCESS in 26m 26s\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/1ec547ccaa6f46e0bfb6825fcc9b7b46 : SUCCESS in 36m 15s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/4cb6f4cc20a64bacae162fcc94f66b81 : FAILURE in 35m 11s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/e0d3f865d35c4b78952dd42915818bb2 : SUCCESS in 57m 46s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/bdb548d8c7db4eadbe67bdcc5c3ea4b3 : FAILURE in 57m 24s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/4483e4c14db947c18cc48e75e67aa2e7 : SUCCESS in 1h 42m 37s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/8fb748d448574ddcbc733c97d62e0381 : SUCCESS in 2m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2998833c8af44b709bcb4b75735cf728 : SUCCESS in 3m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/76ad52485b2b45ccafb30cc34ab2926a : SUCCESS in 3m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/882109d375294d51908f7dcba7d4b57f : SUCCESS in 3m 29s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b4abd3cd8ca24c22bda419ba9bf503c0 : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cfec5b0640de419ab259091fade6c834 : SUCCESS in 3m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab20310f31f248bca6ba213373ca29fd : SUCCESS in 6m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3dc086275e7e4b439e90d96d35961ca6 : SUCCESS in 2m 29s","accounts_in_message":[],"_revision_number":4},{"id":"4cef7c28935464e7a8c0db6717b646f6a541ea6f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-03-18 13:47:13.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"781f1f76a3c310e8251ef71d61fca6292e717437","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-03-18 13:48:36.000000000","message":"Patch Set 4: -Code-Review","accounts_in_message":[],"_revision_number":4},{"id":"c797772c5a1281dc844d6b69ded3dfbd02d87b2c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 14:21:41.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/287997c72a094d28a1dca90f9f6b8724\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/2373b2f4be254b0faeccee15e32c5f78 : SUCCESS in 1h 04m 58s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/786a034a7c194b8e8e9651ea03a9971f : SUCCESS in 1h 03m 33s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/eebacee58d47497f8110d7a1d988c461 : SUCCESS in 1h 54m 30s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/758ba5a73bb04966bfe41fd9ba2c79b3 : SUCCESS in 1h 04m 13s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/e35a1deaef994975b077b608bcee8c98 : FAILURE in 1h 19m 53s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b592b4a2492c4ea2a868ddf4172d8ebb : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e60f49fff8cf4dc2af4969631d12aad0 : SUCCESS in 4m 33s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/e1028965e3544eae9a6160c6d6cfe549 : SUCCESS in 11m 06s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"fd588bc060fb494b0d78dcb5703a4c07e96c462a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 16:17:34.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77277cb7dd964b47b0ef0eb8654d46ee\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/6ce4cbebdead4d14ba99acc2235c464d : SUCCESS in 40m 01s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/f1c32735f71f430ea6689596df908372 : SUCCESS in 37m 02s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/84d856f78c6a4b9f8d8df02d8372c321 : SUCCESS in 28m 54s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/15a9fe9030ee409eb03000109e27660a : SUCCESS in 39m 08s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/6986d16c74be4af68a9a455b6cf4a115 : SUCCESS in 55m 44s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b388f11e260946c6967d0864662c1b72 : SUCCESS in 55m 20s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f52ef649fead49be8276b41e0e01c38d : SUCCESS in 28m 50s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/4c651f59da2544719ae52717d0f74a7e : FAILURE in 55m 00s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/7d94fb4ae20145c09a63dae28a95a525 : SUCCESS in 53m 54s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/033a99205ccc44a991f742f87b345cc7 : SUCCESS in 1h 19m 02s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/2177998662a146ccb0418c1b9b47bd5a : SUCCESS in 1h 00m 14s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/518cb3e2fb91463eb74b9fa31ee46b70 : SUCCESS in 1h 24m 12s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/402dafbdb0c64f418f51fecfa3390947 : SUCCESS in 1h 17m 33s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/6cdf2eefa1d84dbe951741a1964d9c79 : SUCCESS in 1h 27m 58s\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/af3daea6cf6240d0b099796aa95927b6 : SUCCESS in 1h 44m 16s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/e1b1d46fc12d4e518a5582bfed173867 : SUCCESS in 2h 23m 09s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/b6201f445e744aa79d2e4ff1da991e26 : FAILURE in 1h 27m 25s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/6c8ad918841045618768a789610fe119 : SUCCESS in 2h 19m 42s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/a6735f59636643028c31dbce70023378 : SUCCESS in 1h 15m 49s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/48bab0fcb65f4ff9b3c3fff4bd7f04f9 : SUCCESS in 1h 09m 56s (non-voting)\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/e1519cddb1d24913ac827e957e87880e : SUCCESS in 34m 47s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/db01d40a3b904f52a5b2724dbe73030a : SUCCESS in 32m 14s\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/ceddd7ccd3544fa6ba6668f97e08d061 : SUCCESS in 1h 01m 58s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/b72bcb9a9b4d46ccbc4219f148a84088 : FAILURE in 35m 41s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/7defb42ed0e1420998fb1895c434d6f3 : SUCCESS in 57m 47s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/bad0c3410df04f6dbfce1da6d5ae1eb6 : SUCCESS in 1h 31m 06s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/ebdb8bbad5274a9781981242541d4a42 : SUCCESS in 1h 40m 55s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/f6fd177947094ca3a5093f17de7f0726 : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f13e364d600f451ebb8a81ffe7d17d65 : SUCCESS in 2m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c4233d7d80bc4cab82256400975c0a31 : SUCCESS in 3m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a8676e1a6ca2443298c9c3fee39d71cd : SUCCESS in 3m 28s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/eaa7952b79a34b97b3ff7724d06b3917 : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8a3c1fbb3b224270a5a9d57b19e5aff3 : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48b9f99ff52c4505b2662e5ab4a0b15a : SUCCESS in 6m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/863244d63a0a469fb295ff8623700610 : SUCCESS in 3m 06s","accounts_in_message":[],"_revision_number":4},{"id":"a4c68677a25c40314f8320db78ce594cee9ca437","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-03-23 05:36:50.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"eb4423b5b83b3ce42b14eeb6a5f229bd3ca9dee3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-23 07:15:09.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb37e973a41b46419f439c16e667bb6b\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/48463cc2eeeb491b9eba6ad069a804ef : RETRY_LIMIT in 3m 58s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/0ea7e269547748f1b55cd98f6d1a72ff : RETRY_LIMIT in 3m 42s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/353522c7f35940d98008a436372a951a : RETRY_LIMIT in 4m 29s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/30e0725478344faca7af25a1bcaeab25 : RETRY_LIMIT in 3m 21s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/bcbbd21339f24c5ba8f473df984407a4 : FAILURE in 1h 35m 23s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5eb14b77937242f989ff6885bd930600 : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/aa85d50bde2f427882f9101210a4d513 : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/3921eaff496244139fbf8044813a23a0 : SUCCESS in 6m 43s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a43205105c260fb531af6bba218cb149ee5c6bbe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-23 08:20:44.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e802c87a23e44c5a96c8a720d294b4f6\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/abf3cf75f44f4ffd8adc97ae3bb693f8 : RETRY_LIMIT in 2m 43s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/2395c34533f1456e85ac40165866ce02 : RETRY_LIMIT in 2m 21s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/64071c06890441b594c34accbb76be01 : SUCCESS in 26m 05s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/71a525592dd945baa36ba7b92757a001 : RETRY_LIMIT in 3m 23s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/6a635a564c614bc4ad319ca50bba8180 : SUCCESS in 47m 03s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8a36920a2ed4496b84951350b617528e : RETRY_LIMIT in 2m 57s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0b3d08ff9804439496fbbe6010913b12 : SUCCESS in 30m 16s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/94115114d7e24360bfa6ca850cd5445f : SUCCESS in 1h 08m 31s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/8559dbca32354f42a9363d58fd96cb9b : SUCCESS in 1h 12m 15s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/142a7ccbe39a42909c6990ccff15ea89 : SUCCESS in 1h 37m 03s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/74922ffe6c2e4af6a649bea70e312bf5 : SUCCESS in 1h 18m 52s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/27d1fba96e0741b9a8642b93fbc79d0f : SUCCESS in 1h 44m 40s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/79d91bbf0e0b4c9d81e57912243739b8 : SUCCESS in 1h 16m 32s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/559f1843274a4905bae4c2bbac4f8b3e : SUCCESS in 2h 18m 42s\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/635dc83182cf4d60b4a094d06ce6c639 : FAILURE in 2h 05m 10s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/cfa7e6e3cbd843b195799f2ed8fb6075 : SUCCESS in 2h 42m 37s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/05230d9555014af98495c98e2a4a2583 : SUCCESS in 1h 58m 39s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/b27bef9322cd4f08ad54c1aa73bc4eb8 : SUCCESS in 2h 28m 45s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/d5514a0a01af42e88a39532030c4e767 : SUCCESS in 1h 34m 24s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/d6d6b71b4bf149589dfe82bca6803287 : SUCCESS in 1h 39m 23s (non-voting)\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/3883183974464c70aaea4a05844d1de6 : SUCCESS in 32m 04s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/3876626a5b494ba5a42dfdd23522802d : SUCCESS in 32m 57s\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/e1c71f8c2f8d4dd5b9e96320afdadc6c : SUCCESS in 1h 08m 03s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/883cc361a4a8433eb404edb9243a1e01 : SUCCESS in 1h 07m 21s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/cfe620dc1a1c45bfb7296d58f5238f16 : SUCCESS in 1h 18m 16s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/c113403430154eb7b3d99299670c6b72 : SUCCESS in 1h 55m 39s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/f19602424a94485a80fc19a4973db198 : SUCCESS in 1h 50m 19s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/c6cdbd2877484b3fa1953f36ec557320 : SUCCESS in 1m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02005757e5e74822bee2d145f3e1ebf1 : SUCCESS in 3m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8f0be7c4b84e4bb1a4bbba9ce5e73245 : SUCCESS in 3m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5a249b0406b04337aa84fc00861027d0 : SUCCESS in 2m 44s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1235eadb69734eb1a7069abeb625e025 : SUCCESS in 4m 10s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1bd60c92183f4f5085ae97aae662523b : SUCCESS in 2m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f80e6d630d804a1184a3a2d5913e3dad : SUCCESS in 7m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e6001969e6cc4a948ef7ea696d38905e : SUCCESS in 2m 31s","accounts_in_message":[],"_revision_number":5},{"id":"d2644ada5805464cddb540fe28ec09a60696ba95","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-03-24 09:24:02.000000000","message":"Patch Set 5:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":5},{"id":"b669405459e564d6d4b22151afc6fd6f78a9cb43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-03-24 09:27:20.000000000","message":"Uploaded patch set 6: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"ed056e4e33ec0ceddbc5c29ba845e1d494112318","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-03-24 10:52:00.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"8a9e9f2df96100491cad5cd316f0c7eaa15b2617","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 11:33:39.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e14c8b0b2f44dc2a1b0c15e920a309d\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/4bc4d3fb1e4a432f94ccc731b14edd32 : SUCCESS in 1h 10m 19s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/2f5489910d5c4316ae89350931411e97 : SUCCESS in 1h 28m 36s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/e62f3f7e94cb4a9383878dabe02a0411 : SUCCESS in 1h 26m 58s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/18ed2d6c2f524357a9f5882424c6c14c : SUCCESS in 1h 15m 11s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/085d0d795bb844a8a7570638daf5adce : FAILURE in 1h 34m 29s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/94033482e22e4de8a684b074f5a47960 : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0d335306e2234b0bb80e70fa4959bc13 : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b5cc181e7d964455abb4ec7fba9e3d81 : SUCCESS in 8m 13s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"79aa1dc3bcccb0a53b4d2096f1b2d36abc4cc843","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 12:14:55.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/895410a086aa41b99a0fc3e7394816fb\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/e92026a0077642fc8749d0c7c76ac99a : SUCCESS in 41m 55s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/fefde949a49440eaa0eefd899a095fae : SUCCESS in 50m 58s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/934b36d2bd9b4b5e9564b38fd559bdd0 : SUCCESS in 29m 44s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/db700d7463c54f6d85c04c41779d720a : SUCCESS in 41m 19s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/747d2d302a0447f9aef0b6ffee89a04f : SUCCESS in 35m 12s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a497dbc9a7974169bf4a6a3006ad08b1 : SUCCESS in 59m 21s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8cd9411731aa40d6a01c9cb322d6803c : SUCCESS in 43m 06s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/bf31c09af1a647babe82a54705b94d4a : SUCCESS in 1h 06m 08s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/643a8323614943a296a4bb3ee1e82169 : SUCCESS in 1h 12m 24s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/95f63a3c55df414ea205019cffa32e1e : SUCCESS in 1h 33m 43s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/52af6a5fdaa34d66afd2af897c91fe06 : SUCCESS in 1h 19m 16s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/af22ef3ab57b4c279882de0e3da2ca20 : SUCCESS in 1h 56m 30s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/edb094ac26334c698fac34b21ca5a5ec : SUCCESS in 1h 19m 53s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/d816fa4dae8945a29c290ccf6d88ca7b : SUCCESS in 1h 50m 19s\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/99ba68617c624ec98eccf502e935e024 : SUCCESS in 2h 00m 28s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/c7dd7b13273c455cb0eae2111def232c : SUCCESS in 2h 45m 07s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/7b16dba09948431d8c33ef14fa7d3752 : SUCCESS in 2h 07m 18s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/f9f124137c894994b9ab6e227f7cff14 : SUCCESS in 2h 42m 17s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/77b8b2cb557f44ba9979bc5bd6d97a88 : SUCCESS in 1h 49m 15s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/4e442bf2a67149cc946e2b374864e69e : SUCCESS in 1h 38m 44s (non-voting)\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/af56a0ae28e14ecea947c314c99e540d : SUCCESS in 28m 41s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/79c5949f01904e8bae858cdc6993f101 : SUCCESS in 34m 30s\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/b82f1d0ccd18463aa753879608aed09e : SUCCESS in 45m 37s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/c686117bdc82433c85705d1702b9c373 : SUCCESS in 1h 35m 21s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/d33e965a5d11491b9cb119b1542033fb : SUCCESS in 1h 30m 38s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/39101693990546debe0dfa050efdde72 : SUCCESS in 1h 49m 25s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/b56d7a46def64eba80b1ac06d2c037e1 : FAILURE in 1h 16m 49s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/528bb9f8749948c1bb00a494189b0ae7 : SUCCESS in 3m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45a9bcb178eb4982938e239bdae97b44 : SUCCESS in 3m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b54d5e934c694d1285321a95bbc7cabc : SUCCESS in 3m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9e3d6ab0a33f4d9b8cd42e9e2d1b0a8e : SUCCESS in 3m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0df44db66f0c4049891752158b77bb29 : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/06ae00056cc6477994b880e315baca73 : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a893c852a2042dd902e3264cd911b62 : SUCCESS in 4m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6db5d18f89ab454e96d95436fd43d9ac : SUCCESS in 3m 00s","accounts_in_message":[],"_revision_number":6},{"id":"f88c541488773d1633b841856e2306edf88ae77d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-03-25 05:09:27.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"6cbddb5fa1df1ea8cca8102cc8c6b2a7c83ad372","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-03-25 07:39:36.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"7007fa98b056a5ab5abbd7350a99c98cf9b06199","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-03-25 14:14:03.000000000","message":"Patch Set 6: Backport-Candidate-1 Review-Priority+1","accounts_in_message":[],"_revision_number":6},{"id":"c438a820c0d9836cfb54fea148120a27fb8524d6","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-03-30 14:47:52.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"78395659b901040d29e6ae6a4f8ae1ad73fccfec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 14:48:10.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"6b4d7089a764d587d9f7371f3ce797096dd8aefd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 16:53:18.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fcee7893e31441e882f1a9eae739da6f\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/e74d092d94c24dd4be6239008b55d9b4 : SUCCESS in 46m 46s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/3680cc1fae374c73b31471f0256f073e : SUCCESS in 12m 47s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/fc5eb14519e6404a98022f4dedad6ee4 : SUCCESS in 41m 31s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/c7ada4d19896457e9d9a9ba3d8ee43c6 : SUCCESS in 49m 50s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4ddbcab53486416f8b44a1980895478f : SUCCESS in 50m 03s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/24d0f04e5788436d84b86eeb22c8dea6 : SUCCESS in 21m 27s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/5f037e5ee2da489385eee7b22cdae8ee : SUCCESS in 1h 22m 02s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/3134939843304c1684439637165eb3f9 : SUCCESS in 1h 48m 59s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/11d28bd20c554a3095b6057a015fb000 : SUCCESS in 1h 31m 59s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/bd65d5de826b4044a642cef596f4e66f : SUCCESS in 2h 01m 04s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1e07ea7be85e4d4ab8b3b1be25251857 : SUCCESS in 1h 25m 46s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/0fcb4797831b4899898c6846b2496bd3 : SUCCESS in 2h 04m 12s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/1b8024605b324c5f92face7dfb7e82df : SUCCESS in 2m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ff9173a3129443996a0c62276a6d932 : SUCCESS in 3m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b21e43abc0eb481da82a48d164daa8e8 : SUCCESS in 2m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b5e25bd3f6ab4a978633ff04dd16d872 : SUCCESS in 2m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/303ad83d531f41e69e62989e1488cff0 : SUCCESS in 3m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/93bf7ae38cbf447194669f79c37679ab : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":6},{"id":"a9fed81cc1298ac510c112c65035c08f9834a184","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 16:53:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"16591836537cbbbb7844a07d9d0ee7de6506a4b0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 16:54:45.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f7f586817a7497c8c4db0dec5660325\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4c70b70edda409988f6ac9c307340bb : SUCCESS in 1m 09s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/473b3b35654749feb34c2004c57b9e03 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"250cdef268c725cbeddc3298541250860f940e1c","revisions":{"3e4cfad9b8d0aecef88fa24018f91685b84b815c":{"kind":"REWORK","_number":1,"created":"2026-02-19 11:35:51.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/11/977311/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/11/977311/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/977311/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/977311/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/977311/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/11/977311/1"}}},"commit":{"parents":[{"commit":"98c8bd8e848742474f9764844fd0f882d6539f82","subject":"Merge \"tox: Drop basepython\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/98c8bd8e848742474f9764844fd0f882d6539f82"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-02-19 11:32:35.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-02-19 11:32:35.000000000","tz":60},"subject":"Bump OVN to 25.09 in Rocky Linux 10 images","message":"Bump OVN to 25.09 in Rocky Linux 10 images\n\nThis matches the version used by Ubuntu images since Flamingo [1].\nDebian Trixie still uses 25.03 [2].\n\n[1] https://openstack-ci-reports.ubuntu.com/reports/cloud-archive/gazpacho_versions.html\n[2] https://packages.debian.org/source/trixie/ovn\n\nChange-Id: I323ea4e7dabecfcf20fa890e028eeb1c3766be0b\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3e4cfad9b8d0aecef88fa24018f91685b84b815c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3e4cfad9b8d0aecef88fa24018f91685b84b815c"}]},"branch":"refs/heads/master"},"3075dc50cfeb6c0a7867cb1015f61bd633089160":{"kind":"REWORK","_number":2,"created":"2026-02-19 12:54:02.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/11/977311/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/11/977311/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/977311/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/977311/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/977311/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/11/977311/2"}}},"commit":{"parents":[{"commit":"0f8932b5a6a83853e8d39e46067ec6288750814e","subject":"Add Open vSwitch/OVN versions to support matrix","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0f8932b5a6a83853e8d39e46067ec6288750814e"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-02-19 12:49:21.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-02-19 12:49:21.000000000","tz":60},"subject":"Bump OVN to 25.09 in Rocky Linux 10 images","message":"Bump OVN to 25.09 in Rocky Linux 10 images\n\nThis matches the version used by Ubuntu images since Flamingo [1][2].\nDebian Trixie still uses 25.03 [3].\n\nNote that support for 25.03 is planned to end on 7 March, 2026 [4].\n\n[1] https://openstack-ci-reports.ubuntu.com/reports/cloud-archive/flamingo_versions.html\n[2] https://openstack-ci-reports.ubuntu.com/reports/cloud-archive/gazpacho_versions.html\n[3] https://packages.debian.org/source/trixie/ovn\n[4] https://www.ovn.org/en/releases/25.03/\n\nChange-Id: I323ea4e7dabecfcf20fa890e028eeb1c3766be0b\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3075dc50cfeb6c0a7867cb1015f61bd633089160"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3075dc50cfeb6c0a7867cb1015f61bd633089160"}]},"branch":"refs/heads/master"},"d47fc0c21b9c9e639d90d11ed40ad629b9fb7f11":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-03-06 17:26:09.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/11/977311/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/11/977311/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/977311/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/977311/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/977311/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/11/977311/3"}}},"commit":{"parents":[{"commit":"761b0c0beb564d81fcbd420099ea7512e83bb789","subject":"Fix MariaDB version number","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/761b0c0beb564d81fcbd420099ea7512e83bb789"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-02-19 12:49:21.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-03-06 17:26:09.000000000","tz":0},"subject":"Bump OVN to 25.09 in Rocky Linux 10 images","message":"Bump OVN to 25.09 in Rocky Linux 10 images\n\nThis matches the version used by Ubuntu images since Flamingo [1][2].\nDebian Trixie still uses 25.03 [3].\n\nNote that support for 25.03 is planned to end on 7 March, 2026 [4].\n\n[1] https://openstack-ci-reports.ubuntu.com/reports/cloud-archive/flamingo_versions.html\n[2] https://openstack-ci-reports.ubuntu.com/reports/cloud-archive/gazpacho_versions.html\n[3] https://packages.debian.org/source/trixie/ovn\n[4] https://www.ovn.org/en/releases/25.03/\n\nChange-Id: I323ea4e7dabecfcf20fa890e028eeb1c3766be0b\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d47fc0c21b9c9e639d90d11ed40ad629b9fb7f11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d47fc0c21b9c9e639d90d11ed40ad629b9fb7f11"}]},"branch":"refs/heads/master","description":"Rebase"},"4fb596948d747126dd4afba5d14658bfac6c2323":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-03-18 11:02:22.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/11/977311/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/11/977311/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/977311/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/977311/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/977311/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/11/977311/4"}}},"commit":{"parents":[{"commit":"acda933879243491e77dd9238bc47633126bd0bc","subject":"Merge \"Update Prometheus to version 3.5.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/acda933879243491e77dd9238bc47633126bd0bc"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-02-19 12:49:21.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-03-18 11:02:22.000000000","tz":0},"subject":"Bump OVN to 25.09 in Rocky Linux 10 images","message":"Bump OVN to 25.09 in Rocky Linux 10 images\n\nThis matches the version used by Ubuntu images since Flamingo [1][2].\nDebian Trixie still uses 25.03 [3].\n\nNote that support for 25.03 is planned to end on 7 March, 2026 [4].\n\n[1] https://openstack-ci-reports.ubuntu.com/reports/cloud-archive/flamingo_versions.html\n[2] https://openstack-ci-reports.ubuntu.com/reports/cloud-archive/gazpacho_versions.html\n[3] https://packages.debian.org/source/trixie/ovn\n[4] https://www.ovn.org/en/releases/25.03/\n\nChange-Id: I323ea4e7dabecfcf20fa890e028eeb1c3766be0b\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4fb596948d747126dd4afba5d14658bfac6c2323"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4fb596948d747126dd4afba5d14658bfac6c2323"}]},"branch":"refs/heads/master","description":"Rebase"},"9a3e268283f453699fecabb37c088e3b4f62015a":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-03-23 05:36:50.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/11/977311/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/11/977311/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/977311/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/977311/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/977311/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/11/977311/5"}}},"commit":{"parents":[{"commit":"482aec6a69e47034bc63bbb9d0d80d59cc710ffc","subject":"Merge \"Update Prometheus Alertmanager and exporters to the latest\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/482aec6a69e47034bc63bbb9d0d80d59cc710ffc"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-02-19 12:49:21.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-03-23 05:36:50.000000000","tz":0},"subject":"Bump OVN to 25.09 in Rocky Linux 10 images","message":"Bump OVN to 25.09 in Rocky Linux 10 images\n\nThis matches the version used by Ubuntu images since Flamingo [1][2].\nDebian Trixie still uses 25.03 [3].\n\nNote that support for 25.03 is planned to end on 7 March, 2026 [4].\n\n[1] https://openstack-ci-reports.ubuntu.com/reports/cloud-archive/flamingo_versions.html\n[2] https://openstack-ci-reports.ubuntu.com/reports/cloud-archive/gazpacho_versions.html\n[3] https://packages.debian.org/source/trixie/ovn\n[4] https://www.ovn.org/en/releases/25.03/\n\nChange-Id: I323ea4e7dabecfcf20fa890e028eeb1c3766be0b\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9a3e268283f453699fecabb37c088e3b4f62015a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9a3e268283f453699fecabb37c088e3b4f62015a"}]},"branch":"refs/heads/master","description":"Rebase"},"250cdef268c725cbeddc3298541250860f940e1c":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2026-03-24 09:27:20.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/11/977311/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/11/977311/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/977311/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/977311/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/977311/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/11/977311/6"}}},"commit":{"parents":[{"commit":"482aec6a69e47034bc63bbb9d0d80d59cc710ffc","subject":"Merge \"Update Prometheus Alertmanager and exporters to the latest\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/482aec6a69e47034bc63bbb9d0d80d59cc710ffc"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-03-24 09:24:51.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-03-24 09:24:51.000000000","tz":60},"subject":"Bump OVN to 25.09 in Rocky Linux 10 images","message":"Bump OVN to 25.09 in Rocky Linux 10 images\n\nThis matches the version used by Ubuntu images since Flamingo [1][2].\nDebian Trixie still uses 25.03 [3].\n\nSupport for 25.03 has ended on 7 March, 2026 [4] while 25.09 is\nsupported until 5 September, 2026 [5].\n\n[1] https://openstack-ci-reports.ubuntu.com/reports/cloud-archive/flamingo_versions.html\n[2] https://openstack-ci-reports.ubuntu.com/reports/cloud-archive/gazpacho_versions.html\n[3] https://packages.debian.org/source/trixie/ovn\n[4] https://www.ovn.org/en/releases/25.03/\n[5] https://www.ovn.org/en/releases/25.09/\n\nChange-Id: I323ea4e7dabecfcf20fa890e028eeb1c3766be0b\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/250cdef268c725cbeddc3298541250860f940e1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/250cdef268c725cbeddc3298541250860f940e1c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-1"],"failing_atoms":["label:Backport-Candidate\u003d1"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
