)]}'
{"id":"openstack%2Fkolla~977609","triplet_id":"openstack%2Fkolla~master~I66015b45f2c1f915e1aadc57af9c3d239e9c2472","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-03-16 06:23:44.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2026-03-13 14:11:15.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-03-16 08:31:54.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I66015b45f2c1f915e1aadc57af9c3d239e9c2472","subject":"Update Prometheus to version 3.5.1","status":"MERGED","created":"2026-02-23 10:41:01.000000000","updated":"2026-03-16 08:32:59.000000000","submitted":"2026-03-16 08:31:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"977609","meta_rev_id":"ba0520813df356a0a27a07c3ce0619a299980d0d","_number":977609,"virtual_id_number":977609,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":37313,"name":"Alma MAILLET-CONTOZ","display_name":"Alma MAILLET-CONTOZ","email":"alma.maillet-contoz@infomaniak.com","username":"nicolasmc","status":"Intern SRE at Infomaniak"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-16 08:31:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},{"value":0,"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"recommended":{"_account_id":37313,"name":"Alma MAILLET-CONTOZ","display_name":"Alma MAILLET-CONTOZ","email":"alma.maillet-contoz@infomaniak.com","username":"nicolasmc","status":"Intern SRE at Infomaniak"},"all":[{"value":1,"date":"2026-03-03 21:42:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37313,"name":"Alma MAILLET-CONTOZ","display_name":"Alma MAILLET-CONTOZ","email":"alma.maillet-contoz@infomaniak.com","username":"nicolasmc","status":"Intern SRE at Infomaniak"},{"value":2,"date":"2026-03-06 17:51:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2026-02-23 19:50:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2026-03-13 14:11:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-03-12 15:27:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},{"value":1,"date":"2026-02-27 10:29:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":37313,"name":"Alma MAILLET-CONTOZ","display_name":"Alma MAILLET-CONTOZ","email":"alma.maillet-contoz@infomaniak.com","username":"nicolasmc","status":"Intern SRE at Infomaniak"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2026-03-13 14:11:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},{"value":0,"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":37313,"name":"Alma MAILLET-CONTOZ","display_name":"Alma MAILLET-CONTOZ","email":"alma.maillet-contoz@infomaniak.com","username":"nicolasmc","status":"Intern SRE at Infomaniak"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":-1,"date":"2026-03-13 14:11:15.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},{"value":0,"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":37313,"name":"Alma MAILLET-CONTOZ","display_name":"Alma MAILLET-CONTOZ","email":"alma.maillet-contoz@infomaniak.com","username":"nicolasmc","status":"Intern SRE at Infomaniak"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},{"value":0,"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},{"_account_id":37313,"name":"Alma MAILLET-CONTOZ","display_name":"Alma MAILLET-CONTOZ","email":"alma.maillet-contoz@infomaniak.com","username":"nicolasmc","status":"Intern SRE at Infomaniak"},{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-23 11:40:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-23 19:50:40.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-02-27 10:29:53.000000000","updated_by":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"reviewer":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"state":"REVIEWER"},{"updated":"2026-03-03 21:42:21.000000000","updated_by":{"_account_id":37313,"name":"Alma MAILLET-CONTOZ","display_name":"Alma MAILLET-CONTOZ","email":"alma.maillet-contoz@infomaniak.com","username":"nicolasmc","status":"Intern SRE at Infomaniak"},"reviewer":{"_account_id":37313,"name":"Alma MAILLET-CONTOZ","display_name":"Alma MAILLET-CONTOZ","email":"alma.maillet-contoz@infomaniak.com","username":"nicolasmc","status":"Intern SRE at Infomaniak"},"state":"REVIEWER"},{"updated":"2026-03-06 17:22:28.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2026-03-06 17:22:28.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2026-03-06 17:51:51.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2026-03-12 15:27:28.000000000","updated_by":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"reviewer":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"12dd5d3ceb47c2d9714daca433a7b60cf98f4cb9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-23 10:41:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3000e91c8eb24cf3e3e60f863f062ff03cc14b58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-23 11:40:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad47c032fb8d4241a029bf1547d9fc29\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/72b4dd2c80f744b98ee51ce885b2d94c : SUCCESS in 41m 12s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/bdd55590b0ae451c9e27aee0ae633fbb : SUCCESS in 42m 10s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/d4ebbf97e817487bb718d38891437bf9 : SUCCESS in 20m 11s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/e882fe551b9549a2bccc62dd7a2c1c7e : SUCCESS in 42m 59s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/b6649cb5dcd34fa99bf714bca8f3dd83 : SUCCESS in 52m 13s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e09125293b14431b8a3e88a61dd261ee : SUCCESS in 53m 46s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8dbfd03b8d814618916ac4e1c6aa8200 : SUCCESS in 21m 46s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/812fadc8b9f742e68a4cb4e799417310 : SUCCESS in 2m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a205404dbbab4690847e2a401d0a8870 : SUCCESS in 3m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4524d25b708b472d89691576a5d29ab4 : SUCCESS in 2m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/993dc22e5cb84c6ab4ae3e97a79f463b : SUCCESS in 2m 53s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2a7d980d939743ff8f6a5961edec5d7c : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dfab5fbbaf1e44dfb78ebdd70927a180 : SUCCESS in 2m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3a776956ecb4392ab9bcf3aadef44ed : SUCCESS in 4m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b5954118f8dd4850a5f85cd2559d0cf4 : SUCCESS in 2m 07s","accounts_in_message":[],"_revision_number":1},{"id":"55c031ac6ce4b43aecdf652bfb9f3b68faf4ce98","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-23 11:52:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/658be39ed7f340329c1f65e4c08f947c\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/486339edac6d42b3be3b20024d69c745 : SUCCESS in 1h 00m 28s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/74ba6f87e761485c908b061a810c7043 : SUCCESS in 51m 41s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/079a2ddf4bbb4fa9a30224b3af44752b : SUCCESS in 54m 02s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/08239a3002e443d684417f39046d9bb5 : SUCCESS in 1h 09m 44s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1df86624a3a74813b0a765aee91d44cb : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/53f26eea2ebf4e4bbd3351c1c826edb4 : SUCCESS in 3m 53s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/26105e2d2a4a4f8d93d9396a70d764d7 : RETRY_LIMIT in 2m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5c5db880855e2cf5cf20eaaae4b641a8b58e8553","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-02-23 19:50:40.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"97292f35e8c731f5ffb6c7c0cc1f16db1ac1d7dd","author":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"date":"2026-02-27 10:29:53.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"806507ab138296118b1f2bdc2c527332be78dd43","author":{"_account_id":37313,"name":"Alma MAILLET-CONTOZ","display_name":"Alma MAILLET-CONTOZ","email":"alma.maillet-contoz@infomaniak.com","username":"nicolasmc","status":"Intern SRE at Infomaniak"},"date":"2026-03-03 21:42:21.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d87e4418df0754461f95191e35299034f6fb6428","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2026-03-06 17:51:51.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"be2dc9813867d11844eba26c2852b1b4c366e53f","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2026-03-12 15:27:28.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f127340ad9d299b0db78841277763a669ee2c1a4","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-03-13 14:11:15.000000000","message":"Patch Set 1: Backport-Candidate-1 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8fcfd0ed31c33696cefbccf74c970fc829cbff82","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-13 14:11:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4b183db7cb858a01c1ef6b2b65d65c32e5abb711","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-13 15:14:50.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ece34801b8b2455599bd99495f8cccb9\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/d7aa53142b65470099af898bc98caa18 : SUCCESS in 33m 30s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/756649e937874d06995c17c9d344e8f6 : SUCCESS in 14m 24s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/12e348615ec24ff3b45382dc80e452ba : SUCCESS in 43m 06s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/e5e13c337b294b858861b5c20e04c6d2 : FAILURE in 34m 39s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8a8e380876da498c9cee02694b52bc11 : SUCCESS in 58m 01s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5ff5da7ac9c5489ab387dd6b36de2576 : SUCCESS in 28m 31s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/94df39ea0f7e4cae9f1267f639f6a897 : SUCCESS in 3m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/328ccebfdb1c4982a1545cb7f02ee124 : SUCCESS in 3m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6d5b2538833046f48ebc379a4d213777 : SUCCESS in 3m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/31b2ee622c57464cab21e38903f36aa9 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e5a2c993f5d49a5b925242ed519b8ca : SUCCESS in 6m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/482a9838194841d988ecddb0b820c740 : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":1},{"id":"760f2756d54f62eab58c980acfea0800ced787c1","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-03-16 06:23:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dc4de90b04a5af5dec626b5d2db76967a3e8f335","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-16 07:22:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/561ad1c70f4242829725b36298814b3d\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/0a4265ae3ebb4820b478caf85b76a429 : SUCCESS in 38m 39s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/16a1a1823ea54491b156e7ecaf533a26 : SUCCESS in 54m 26s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/86aeb98b1a494492ba73e5b8f23187d1 : SUCCESS in 19m 15s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/65d8861112994f499be971353d51ba3e : SUCCESS in 39m 00s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/7270ea5dfc01409ca6dbb519fa5e2e5d : SUCCESS in 47m 55s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b881abfec5e545869391259596beee64 : SUCCESS in 33m 40s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/dfd5b0b6fc1b49ce950cbd551946abbd : SUCCESS in 29m 07s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/96b90504a3244a169f43bbdbb1e255b1 : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/226e193cbb704f72be203e8ca144b7a7 : SUCCESS in 3m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/18683e86de444aacb7e491a5e4e90fd1 : SUCCESS in 2m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/36c5aeaae4524865ac719f7441b00218 : SUCCESS in 3m 37s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/85977c946041450c9f3e480a4e9eeff2 : SUCCESS in 7m 03s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a45cc9abcdab441b94ec175219a1e7f4 : SUCCESS in 3m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65ee5ebd4f584c739d48dd53ef22e933 : SUCCESS in 7m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6aa736488dbd4dcc9bd69f0289611d4e : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":1},{"id":"548eba8cb102cbc2c402ffe0b4912306cffcf4fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-16 07:22:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"528f23807e78b50f1ae79c46e699d2eaca2a7aa7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-16 08:02:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ddf3413813e4b278bd1e067d81cd0f1\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/c8a8561ccae748a99ec7fbeba35f3ab1 : SUCCESS in 1h 36m 48s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/53e169328c124422829a8383ba811e27 : SUCCESS in 56m 13s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/7571024e2859406cbf4bed5407bb219d : SUCCESS in 55m 03s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/cb379077c42c47bdb1509b21a38dc361 : SUCCESS in 1h 20m 24s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6d72fc51e54a46f5ac6d38a9d832ad35 : SUCCESS in 5m 40s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/36b1e30a81bd4519965c0ea486b708f2 : SUCCESS in 5m 45s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/83569973eb78400fa2c182b0975d8333 : SUCCESS in 8m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c01ecdcd30b561c583ad26aa281b6cce48f9fc48","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-16 08:31:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f295b6ffa1d34ee58ec21d601034b5f4\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/d0b05f6ffa7d445fa77263badd0b4d8e : SUCCESS in 40m 17s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/df197d981f3640b587e6c638dce39e1f : SUCCESS in 27m 17s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/f0256ac0c595469ea741766c91b3e549 : SUCCESS in 40m 52s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/07bd11f7c84648a8b3aa09de48a53cf5 : SUCCESS in 1h 04m 41s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b0ea99a0bcdd468c8edab8393ff92e58 : SUCCESS in 39m 59s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/63e207b2ea684f6fafa1d94343c57c0d : SUCCESS in 16m 13s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/677258d1c56e4576b400169b70507473 : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/932c36062c3d485b8b8f231b7f8afa0b : SUCCESS in 3m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d745f0e54e3542969372df47b25f3970 : SUCCESS in 4m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a7eb66c54d32439590eb39e2896ae126 : SUCCESS in 3m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a68037638051474d8e881a0afd39a295 : SUCCESS in 5m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f4f05a490774327aba7aff29b220375 : SUCCESS in 3m 23s","accounts_in_message":[],"_revision_number":1},{"id":"73ca597da513799549769a30491687fb9afe12d3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-16 08:31:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ba0520813df356a0a27a07c3ce0619a299980d0d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-16 08:32:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c760343e5483446db843bb2c271caf44\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73355521dd4c4f1eadddf639dec4e950 : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9c2627f9bf204fdf808581d1ac5c0e87 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5a0e470152471e27ed04419f27523355abfc6d89","revisions":{"5a0e470152471e27ed04419f27523355abfc6d89":{"kind":"REWORK","_number":1,"created":"2026-02-23 10:41:01.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/09/977609/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/09/977609/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/977609/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/977609/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/977609/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/09/977609/1"}}},"commit":{"parents":[{"commit":"98c8bd8e848742474f9764844fd0f882d6539f82","subject":"Merge \"tox: Drop basepython\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/98c8bd8e848742474f9764844fd0f882d6539f82"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-02-23 10:40:19.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-02-23 10:40:19.000000000","tz":60},"subject":"Update Prometheus to version 3.5.1","message":"Update Prometheus to version 3.5.1\n\nChange-Id: I66015b45f2c1f915e1aadc57af9c3d239e9c2472\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5a0e470152471e27ed04419f27523355abfc6d89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5a0e470152471e27ed04419f27523355abfc6d89"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-1"],"failing_atoms":["label:Backport-Candidate\u003d1"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
