)]}'
{"id":"openstack%2Fkolla~978812","triplet_id":"openstack%2Fkolla~stable%2F2025.1~I846e4f19696ca28758cb9c0400be96c01b079daa","project":"openstack/kolla","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-03-04 13:37:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I846e4f19696ca28758cb9c0400be96c01b079daa","subject":"debian: use valkey from bookworm-backports","status":"MERGED","created":"2026-03-03 19:02:14.000000000","updated":"2026-03-04 13:38:05.000000000","submitted":"2026-03-04 13:37:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"978812","cherry_pick_of_change":965907,"cherry_pick_of_patch_set":15,"meta_rev_id":"628e474d0624f852844da50b3ebcd5ccef82b355","_number":978812,"virtual_id_number":978812,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-04 13:37:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2026-03-04 11:30:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2026-03-03 21:12:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2026-03-04 11:07:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2026-03-04 11:30:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-03 19:02:14.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2026-03-03 21:06:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-04 10:25:02.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"CC"},{"updated":"2026-03-04 11:07:48.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-03-04 11:30:31.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"}],"messages":[{"id":"2bbf915c4d81b8cd57b8480963218a0ab7de4da1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-03-03 19:02:14.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"65744728b0d8cd1cf5f3d3c5ba4d8a954ceff36a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-03-03 19:36:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"41729da31678a37ff94eaff73d1018ad38ed4235","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 21:06:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5011e18970e04bf48380ca6bb9dda349\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/ac2a98b5ac544559a638496972ce93e2 : SUCCESS in 59m 18s (non-voting)\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/a9303f460356455cb0447a91d7e468e2 : FAILURE in 44m 56s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/1a8ab9dfbf8749c690510c55e3ac8bf0 : SUCCESS in 51m 16s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/19285c2e799043fab5aff0502354f27d : SUCCESS in 1h 07m 15s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/152823d629a5479994fef543b5dbe354 : SUCCESS in 45m 57s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/585ac0dcfd9d4801afb64aaafdcba3c8 : SUCCESS in 31m 47s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/342f66e7ffdd45d6a7bb5551d4eafd17 : SUCCESS in 59m 31s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/ffed7f8531fc444a9fbf17db88b5aed3 : SUCCESS in 1h 30m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6c7e85f03ba4024a1fb33c5584e347b : SUCCESS in 3m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0f337d190956435ab62fe4534ed24c79 : FAILURE in 2m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6db48faaba5b42dd9a01432451aaef02 : FAILURE in 2m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a7a482d17f194647a72201d5c413ec97 : FAILURE in 2m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/518c53c9ebf0433aaf30bc7e3ff64fa2 : FAILURE in 2m 53s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/d6d3e96e95df47ce8b20342330d28ef2 : FAILURE in 3m 00s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/271a560711ea4472b5589a9fd5bee824 : SUCCESS in 43m 14s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/b1777b6dc87944aaa49551e209a5a24b : SUCCESS in 1h 21m 31s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/972783eaaad6469f8b6be3c37e700379 : SUCCESS in 1h 16m 46s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/9861b02fbe5e4d81b5687ff0472ce8bc : SUCCESS in 1h 50m 19s\n- kolla-build-rocky10 https://zuul.opendev.org/t/openstack/build/ad2ef7fb0f224a908e7033f55beb0adf : SUCCESS in 42m 26s\n- kolla-build-rocky10-podman https://zuul.opendev.org/t/openstack/build/d4a0a6d1d2c64844affb50c215e552ad : SUCCESS in 1h 05m 04s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/86a2f64554a24002b5d5e19d80ab93fa : SUCCESS in 46m 29s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/4a0ec8f0e8ff423fbd30c7dac8e97bd6 : SUCCESS in 25m 47s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/cacba4140d294ff2bb10722e91113a12 : SUCCESS in 1h 09m 39s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/f5831ba8fd6c46169ae8b28a4dfe2b72 : SUCCESS in 1h 34m 21s\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/3e308428face4d7e923c49d693926ae1 : FAILURE in 1h 23m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"83b83b4bb946fc1fb9dd2b37737b124b498ab0b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"real_author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-03-03 21:12:25.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_22629\u003e\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"_revision_number":2},{"id":"33e9d8ea978b25adb7ef22c6dded61c3a0c2b70b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 21:17:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a360fa1d03044a2e9e34912c42074a46\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/fc9f7cffdfb546f2aecad7cb8dd093e1 : NODE_FAILURE Node(set) request 9a87eb8009914d8d9142fe3676b3f31b failed in 0s (non-voting)\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/1002a3eb46fd4661a6ef63d260f3003a : NODE_FAILURE Node(set) request 95b2a4ee4bd14ec3b859e98ce565119c failed in 0s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/941a024185d44011aed23b2808819246 : NODE_FAILURE Node(set) request 27e7c07b859c43a98e84fd342350aac3 failed in 0s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/fc2fb4aa5fd447b7b4df5af732915e11 : NODE_FAILURE Node(set) request 4f89b940f6254b32830e4fa669e266d3 failed in 0s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/2f7dd3eedd64422b9887ec1300be50fc : NODE_FAILURE Node(set) request 6c704e76579d4e54bac2a255e80fbc43 failed in 0s (non-voting)\n- kolla-build-rocky10-aarch64 https://zuul.opendev.org/t/openstack/build/d84ba32609e94c24807efd5a09aeb62a : NODE_FAILURE Node(set) request 20745d80cdc842168dd6c09f5dd838db failed in 0s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/b9826076bb0a46f5b45fdeb8dc01ec22 : NODE_FAILURE Node(set) request b81abcd212754125b642ddc86fa62a6b failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"123671f681829e3a9d7556012c3a428cb8d0ea80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 23:51:12.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/64c55b8fd76c4391a14ba539d0c39f0f\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/82ddde6f8f1042989fba31f1186dec72 : SUCCESS in 1h 00m 00s (non-voting)\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/5674288a3d2347f5ad9e4d6eb965c1a4 : FAILURE in 40m 19s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/34b0ff2be23e4601a9d6dbe36f0e86d5 : SUCCESS in 1h 32m 50s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/006dcba5eb824aba8152a93f306a2225 : SUCCESS in 48m 25s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/9959567391f348029304310086d54e28 : SUCCESS in 57m 29s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/dafd094c373a4fcaa1508f874199d7f0 : SUCCESS in 34m 24s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/70e5eb36068346beac7d1013e875608d : TIMED_OUT in 2h 32m 05s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/e11194cd75a14c13b355908a9a7586fc : SUCCESS in 1h 24m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0c58ea35d8a42bdb4be0a5e0d89e3a8 : SUCCESS in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6cfed7bec5914bedb6fb6929d451075a : SUCCESS in 3m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0316e6d0c23c4e4cab770c9e7f0a4917 : SUCCESS in 3m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a5cdddd00b8b456aa28fc39d5e92eb9d : SUCCESS in 2m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26a0470076e4449aaae243582c7b58c0 : SUCCESS in 7m 16s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/3e80e3ab66594a38856e3a35c057a035 : SUCCESS in 3m 12s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/8ddfacd69dbe49f7a3e660b3e812318a : SUCCESS in 1h 02m 03s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/b5e6a57dac0f4aa488c8d87343ae37a1 : SUCCESS in 1h 25m 11s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/2b55172dc8f840048a311235bb2159b7 : SUCCESS in 1h 39m 19s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/7a641fe6fb294925bbd557050f6f8352 : SUCCESS in 2h 08m 58s\n- kolla-build-rocky10 https://zuul.opendev.org/t/openstack/build/7723f1409cc74615bd172ccdb3844d5d : SUCCESS in 58m 13s\n- kolla-build-rocky10-podman https://zuul.opendev.org/t/openstack/build/76653dd938b14c5fb67a7ec9d675fbff : SUCCESS in 58m 03s\n- kolla-ansible-rocky10 https://zuul.opendev.org/t/openstack/build/0d6092e879cc48f3af0d21bf9dc44341 : SUCCESS in 49m 51s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d0a4886e86214616b03e10aaf8471e80 : SUCCESS in 49m 34s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/18000f8c04414f908e72d3db451458eb : SUCCESS in 37m 54s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/1ffaf0aa704147af802b12a80db729ff : SUCCESS in 1h 10m 14s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/aa351c8cc14f496b863a4a21df09259e : SUCCESS in 1h 39m 03s\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/4b5e1a7bc7724b6bb4407353556dd22d : FAILURE in 1h 28m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b07c7a25819a877a16505facff88bcc7fd3bf268","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-03-04 07:40:38.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e9c0a6a50947ea389161a45da18e818ecff4539a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-04 09:25:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63353aeeb8194357ae4f5c9d281a8e8d\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/144e266322b34463814129067a374197 : SUCCESS in 52m 39s (non-voting)\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/7492ac78fae449c7b84c9d20018fd558 : FAILURE in 1h 01m 41s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/c3271345146d45ecb93d144d686cef05 : SUCCESS in 1h 24m 48s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/17853be4667a4577a5aafc7882780a4f : SUCCESS in 1h 28m 43s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f7e0664d072b4c8cac14d86bc90a7fb4 : SUCCESS in 55m 47s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/ef6bf01146a34f7fa7b3335c717be1ce : SUCCESS in 19m 00s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/14a19ba3433b42cab2d93d457bf78ae5 : SUCCESS in 1h 01m 15s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/8b9caa236db6476cbdeb510a78c57fb7 : SUCCESS in 1h 32m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/783f1adba98d4021860c24f0535cd165 : SUCCESS in 2m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a8197eacc1ad437ca0dc9cbb0fe528ec : SUCCESS in 3m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4d3908ab055f4d23b7103960eec5b257 : SUCCESS in 2m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2f70d7e3c8544aa683242b59fd05ffa8 : SUCCESS in 3m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/611efa8b64714573a6516dc3d2678bca : SUCCESS in 4m 48s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/5e43659f3706410ebbc2f9c37ecb363a : SUCCESS in 2m 54s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/795552e8149340bfa3ae8d2e269760e2 : SUCCESS in 47m 02s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/6b85cc31ab294f199c494f36ce8b4d0a : SUCCESS in 1h 23m 14s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/deb635a6ae5d4c86b16aab6fdf5c0258 : SUCCESS in 1h 27m 49s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/8dcd3a6794e84e6ab848d5137f6c066e : SUCCESS in 1h 36m 26s\n- kolla-build-rocky10 https://zuul.opendev.org/t/openstack/build/02deda6303e04d399df2aa351916abf5 : SUCCESS in 45m 54s\n- kolla-build-rocky10-podman https://zuul.opendev.org/t/openstack/build/414423e2a91c432ca8907bbbabfce833 : SUCCESS in 1h 01m 35s\n- kolla-ansible-rocky10 https://zuul.opendev.org/t/openstack/build/36bd943c3329495f9b2ebb932eeb097d : SUCCESS in 1h 11m 47s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/e0825d19f7bb49c0b4c37c17bda0b51b : SUCCESS in 59m 06s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/acf65180ce004d678267c56f65aea840 : SUCCESS in 26m 45s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/26701bc98d6f4bb6aaf1a683d2f712f8 : SUCCESS in 1h 04m 08s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/c2f13ed052c0400bb7ead437d5ab329d : SUCCESS in 1h 43m 22s\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/4c16897680674f05a6a675256246498d : FAILURE in 1h 32m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1ed9647dadb4bd8bd3aa0dd43f5866fac278bf7c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-04 09:38:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae66b298169a483fac663d54d4118128\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/5cb5f917b09447a4abbe80ea3b368394 : SUCCESS in 56m 59s (non-voting)\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/3deb01928f024fb6a026217c6ae1cafe : SUCCESS in 57m 31s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4a3d95f79b1a4ae98a47e1505e9f72de : SUCCESS in 1h 08m 10s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4760c7493e194e5e91c5258503455362 : SUCCESS in 1h 55m 50s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/6e5dd9c7ab4a4307ac271609fb4dc5ed : SUCCESS in 55m 57s (non-voting)\n- kolla-build-rocky10-aarch64 https://zuul.opendev.org/t/openstack/build/f22f690978c24f0bb9edb3c3a2e7f381 : SUCCESS in 1h 28m 05s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/6b3029bde4864efcad04b0430b8a1dc0 : SUCCESS in 1h 30m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c0eead07db0e28cee6768e173799034064b91d3d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-03-04 11:07:48.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"d12acfce0feeca330f18863e09b8a6be1ec3deab","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2026-03-04 11:30:31.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5935b05f52bd61cc3cc3df7f611172e0d525e8a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-04 11:31:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9c4be78769086f7b241712ffe0a6806252199a6a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-04 13:37:13.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4e73044bbe24eb7b23a49bfb261430d\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/984c8b7a6ef34ef0b28c4df99ef7a20e : SUCCESS in 43m 07s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/82e1e63bf7fe408286960333fd39aa4d : SUCCESS in 24m 52s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/2515fb355fac40fc9da08733b5274e68 : SUCCESS in 1h 01m 43s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/3b00566ffeb34fb58ff23c574fc34365 : SUCCESS in 1h 31m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64a9c8b2b8ea4b4f8d5ab326ae43c9a7 : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/08eef565ff844059a0fa63dbb3e4e3d6 : SUCCESS in 2m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b49aa759065247739a0ed57f7b95fba3 : SUCCESS in 3m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3c6b9bd344d4c14aa00d63395b4f6dd : SUCCESS in 6m 27s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/1eb6ed0713bc47719ef93a692dad3d7c : SUCCESS in 5m 11s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/dfe30f86701243069e21a3dbf1ba0ed5 : SUCCESS in 1h 01m 03s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/cd28d18c54aa4dd8beb5979dbbe48c47 : SUCCESS in 1h 07m 55s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/d9608384265c4d53858a5e5957994abb : SUCCESS in 1h 30m 22s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/c615329cd0104c68b908979ac67c74cf : SUCCESS in 2h 01m 13s\n- kolla-build-rocky10 https://zuul.opendev.org/t/openstack/build/fff838efa27f4d359c008a796b6a855a : SUCCESS in 46m 08s\n- kolla-build-rocky10-podman https://zuul.opendev.org/t/openstack/build/ab3c43a899654a38992decd68a82b351 : SUCCESS in 1h 21m 24s\n- kolla-ansible-rocky10 https://zuul.opendev.org/t/openstack/build/c68777c4cb7c4a83a05dba834e63b258 : SUCCESS in 45m 52s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/25762e32d7f344ccba5676d2f3f6047b : SUCCESS in 46m 40s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/4f9107d8fac549798b2cf5c6f43e0b26 : SUCCESS in 17m 55s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/5369b44274994a05a90a14740f7bf7c5 : SUCCESS in 1h 12m 27s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/f970becc32fc4b9cbd2b81fbd7a7f346 : SUCCESS in 1h 35m 01s","accounts_in_message":[],"_revision_number":2},{"id":"e4c94e1b20b1c5733582da7dbd6df79942c2307c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-04 13:37:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"628e474d0624f852844da50b3ebcd5ccef82b355","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-04 13:38:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f58863787a68438cbb76b86872334f9b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12a5f9bef08a4a6aa0c07a1b5e9f65b4 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7b4dddce681980af06d67d54d900d56e72ea4937","revisions":{"006e99f6ea8c5ac5acc17d32fae085670f062d98":{"kind":"REWORK","_number":1,"created":"2026-03-03 19:02:14.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/12/978812/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/978812/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/978812/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/978812/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/978812/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/978812/1"}}},"commit":{"parents":[{"commit":"7905d651e5193d3865bdbbc13530d5f768fe928a","subject":"Merge \"Update README with new mission statement\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7905d651e5193d3865bdbbc13530d5f768fe928a"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-11-03 10:43:59.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-03-03 19:02:14.000000000","tz":0},"subject":"debian: use valkey from bookworm-backports","message":"debian: use valkey from bookworm-backports\n\nTo fix arm64 valkey build.\nThe virtual `valkey` resolves to `valkey-server` in bookworm-backports,\nbut our global -1000 pin blocks it. Amd64 worked only because extrepo\nshipped it. Use valkey directly from the backports packages instead.\n\nfollow up to https://review.opendev.org/c/openstack/kolla/+/932174\n\nChange-Id: I846e4f19696ca28758cb9c0400be96c01b079daa\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n(cherry picked from commit 477f71b6ba8dd9472ecccd16ac9a5aa79d82eaf0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/006e99f6ea8c5ac5acc17d32fae085670f062d98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/006e99f6ea8c5ac5acc17d32fae085670f062d98"}]},"branch":"refs/heads/stable/2025.1"},"7b4dddce681980af06d67d54d900d56e72ea4937":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-03-03 21:12:25.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"real_uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/12/978812/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/978812/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/978812/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/978812/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/978812/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/978812/2"}}},"commit":{"parents":[{"commit":"7130543722bedc1a621800fb09ced1f69485d310","subject":"[2025.1] CI: enable k-a rocky10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7130543722bedc1a621800fb09ced1f69485d310"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-11-03 10:43:59.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-03-03 21:12:25.000000000","tz":0},"subject":"debian: use valkey from bookworm-backports","message":"debian: use valkey from bookworm-backports\n\nTo fix arm64 valkey build.\nThe virtual `valkey` resolves to `valkey-server` in bookworm-backports,\nbut our global -1000 pin blocks it. Amd64 worked only because extrepo\nshipped it. Use valkey directly from the backports packages instead.\n\nfollow up to https://review.opendev.org/c/openstack/kolla/+/932174\n\nChange-Id: I846e4f19696ca28758cb9c0400be96c01b079daa\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n(cherry picked from commit 477f71b6ba8dd9472ecccd16ac9a5aa79d82eaf0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7b4dddce681980af06d67d54d900d56e72ea4937"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7b4dddce681980af06d67d54d900d56e72ea4937"}]},"branch":"refs/heads/stable/2025.1","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
