)]}'
{"id":"openstack%2Fkolla~980311","triplet_id":"openstack%2Fkolla~master~If4f6bed2254f01eb7f9e3aec01cfc81c493ae787","project":"openstack/kolla","branch":"master","topic":"bump-prometheus-exporters","attention_set":{},"removed_from_attention_set":{"36765":{"account":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"last_update":"2026-03-18 16:14:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If4f6bed2254f01eb7f9e3aec01cfc81c493ae787","subject":"Update Prometheus Alertmanager and exporters to the latest","status":"MERGED","created":"2026-03-12 15:41:50.000000000","updated":"2026-03-18 16:15:39.000000000","submitted":"2026-03-18 16:14:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"980311-bump-prometheus-exporters","meta_rev_id":"610d76424e60a5bdaf73125c0313492b1c5ea958","_number":980311,"virtual_id_number":980311,"owner":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-18 16:14:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"recommended":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"all":[{"value":1,"date":"2026-03-12 19:09:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":2,"date":"2026-03-13 10:15:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2026-03-18 13:44:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2026-03-13 12:17:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2026-03-18 13:44:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2026-03-18 13:44:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-12 16:53:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-12 19:09:43.000000000","updated_by":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"reviewer":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"state":"REVIEWER"},{"updated":"2026-03-13 10:15:01.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2026-03-13 12:17:18.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-03-18 13:44:58.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"240ac61f8ffb2077e56e6d8e628cca59ab40bf45","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2026-03-12 15:41:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"95ab6ad7583efd5b309f97fd9f9755418218a68a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 16:53:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5040b5da34764b929a66d164d1fb5008\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/73006b2972b342c8b115465d11c70a13 : SUCCESS in 40m 10s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/ab763ca27ba844769d00f4f6bdc4f069 : SUCCESS in 53m 23s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/e32bd0c027f445cbb8307cefc73843ac : SUCCESS in 14m 10s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/a904eab591194d56b57196e1e108b0cc : SUCCESS in 53m 45s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/d6c5d886fcc54bb5a311f5188350ca50 : SUCCESS in 55m 14s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/95646b78b8134113a9fa96ea22a5649e : SUCCESS in 56m 31s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1601721975be44fc8babb69b4f64a3a2 : SUCCESS in 15m 54s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/09ac276b6d374a91a290d7fd879f64f0 : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/353861d6e512463eaeade65f1017cdcc : SUCCESS in 4m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d4745f25489e4937a92d8249c87b86d2 : SUCCESS in 3m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2a8dd5cbc2424774b09016435ba57d25 : SUCCESS in 3m 31s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/56698e9d4d794822bfdc234623eb315b : SUCCESS in 7m 51s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d142be9b8b7c4d43b5da15ff2aaccb86 : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51ae792cbe3547dfa5fc78fc23960cc5 : SUCCESS in 3m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bcba7715cc8b47b3b37a2f449770a463 : SUCCESS in 2m 39s","accounts_in_message":[],"_revision_number":1},{"id":"6ffc7e6623d57827fe00327e82ed73c3b503a629","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 18:58:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39168f9afd474526b094bafbd4c5be60\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/2ffff1314ab54eceaa0acafc7d6454f5 : SUCCESS in 1h 01m 26s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/45496e4ed1d1417ca43bd8ade1977581 : SUCCESS in 59m 50s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/fece5568c3124ac198557c74be44ce30 : SUCCESS in 1h 08m 59s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/159918ce06fa4601b23416307fb3fb38 : SUCCESS in 1h 04m 36s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8941d77f8ae34860b00ffbf9d5be545e : SUCCESS in 4m 24s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5efba609842f43a48991770aa1dfa382 : SUCCESS in 5m 03s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/c01ad14588ef46bcb52e2ebd7bbcbf6e : SUCCESS in 6m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3d4426ad81059cadb36274c87fafa99997d3a694","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2026-03-12 19:09:43.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"72c1bbd4a16eefb6ac2315bb0f0c9bf98e6a4eee","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2026-03-13 10:15:01.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0ccfe36bf5122060dcf8840c7c4a7b79a01ea245","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-03-13 12:17:18.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c93d910c5aba3b8ccbd848dca0318af2decdedda","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-03-18 13:44:58.000000000","message":"Patch Set 1: Backport-Candidate+1 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5de72e2313208076bf509ebfb16ada6ff197bc21","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 13:45:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a2c4578c2214a294a2fe5b81d92caba93c9dbafa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 16:14:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a56169396aa4e7fbea5d848855861fb\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/54ef8893a581443a9a34ec888e51e03b : SUCCESS in 56m 37s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/a7ddcd93914a49b98f13f06cce7b9f5d : SUCCESS in 28m 08s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/b967c751cf104e5583b855a6c7fd1a83 : SUCCESS in 53m 44s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/46cf5aa4c41541b58e0fe2625db57c3f : SUCCESS in 1h 11m 17s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6657720a98754d6b9456be30f744c853 : SUCCESS in 1h 00m 45s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2b1b1fb899344274b9508944620368a5 : SUCCESS in 27m 52s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/b27e28d02d0445279adf63fea6b25c33 : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8291a8b5be44cf484197088ae80ba0a : SUCCESS in 5m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/df5e1cade3a24264abe775fa86220a01 : SUCCESS in 3m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7445dfa38a6f4c13bdd3de1f19a7ea36 : SUCCESS in 2m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d892cb8e0f9f4024a6d32b0346e50a39 : SUCCESS in 7m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dbe9bfca286e498d82a76b3db51eb465 : SUCCESS in 5m 53s","accounts_in_message":[],"_revision_number":1},{"id":"b3ad07c3a79484e0be813c363f0e5b0853dc155e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 16:14:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"610d76424e60a5bdaf73125c0313492b1c5ea958","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 16:15:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d569284a2bc14c65ad2a641afac841f2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d57833a82ede4b8d8f5b14516ff5718d : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1f8f4f1bf7e947b1bb6ff4088a92d965 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ac049f84d9a12d87f64464f6de0afc66f99480b0","revisions":{"ac049f84d9a12d87f64464f6de0afc66f99480b0":{"kind":"REWORK","_number":1,"created":"2026-03-12 15:41:50.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/11/980311/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/11/980311/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/980311/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/980311/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/980311/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/11/980311/1"}}},"commit":{"parents":[{"commit":"761b0c0beb564d81fcbd420099ea7512e83bb789","subject":"Fix MariaDB version number","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/761b0c0beb564d81fcbd420099ea7512e83bb789"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2026-03-12 15:38:42.000000000","tz":0},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2026-03-12 15:41:44.000000000","tz":0},"subject":"Update Prometheus Alertmanager and exporters to the latest","message":"Update Prometheus Alertmanager and exporters to the latest\n\n* prometheus-alertmanager: 0.28.1 -\u003e 0.31.1\n* prometheus-blackbox-exporter: 0.27.0 -\u003e 0.28.0\n* prometheus-cadvisor: 0.53.0 -\u003e 0.56.2\n* prometheus-elasticsearch-exporter: 1.9.0 -\u003e 1.10.0\n* prometheus-libvirt-exporter: 2.2.0 -\u003e 2.3.1\n* prometheus-memcached-exporter: 0.15.3 -\u003e 0.15.5\n* prometheus-mysqld-exporter: 0.17.2 -\u003e 0.18.0\n* prometheus-node-exporter: 1.9.1 -\u003e 1.10.2\n\nprometheus-mtail and prometheus-openstack-exporter were not updated as\nthere are no later stable releases.\n\nChange-Id: If4f6bed2254f01eb7f9e3aec01cfc81c493ae787\nSigned-off-by: Seunghun Lee \u003cseunghun@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac049f84d9a12d87f64464f6de0afc66f99480b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac049f84d9a12d87f64464f6de0afc66f99480b0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d1"],"failing_atoms":["label:Backport-Candidate\u003d-1"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
