)]}'
{"id":"openstack%2Fkolla~980663","triplet_id":"openstack%2Fkolla~stable%2F2025.2~I66015b45f2c1f915e1aadc57af9c3d239e9c2472","project":"openstack/kolla","branch":"stable/2025.2","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-03-18 15:30:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I66015b45f2c1f915e1aadc57af9c3d239e9c2472","subject":"Update Prometheus to version 3.5.1","status":"MERGED","created":"2026-03-16 08:45:59.000000000","updated":"2026-03-18 15:32:57.000000000","submitted":"2026-03-18 15:30:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"980663","cherry_pick_of_change":977609,"cherry_pick_of_patch_set":1,"meta_rev_id":"b98380fdad9c32186c3beb3d848bd0fa060d4218","_number":980663,"virtual_id_number":980663,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-18 15:30:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2026-03-18 13:38:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2026-03-16 09:34:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2026-03-18 13:38:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-16 09:31:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-16 09:34:47.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-03-18 13:38:02.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"851ec540b828a3e528a7fd9427037e113436f201","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-03-16 08:45:59.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"2cff8756247c5fce0b3b17cedca6242e309006ba","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-03-16 08:46:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bb1c1b02cf0255dc35569202ba9c9c00514d799d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-16 09:31:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a85d4198d8ef44eda704fa8ff547bfaa\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/12d814e1148e4adca6765cd7d3c7b822 : SUCCESS in 37m 54s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/f91d222612b747d4b4bd9a873029dde5 : SUCCESS in 39m 12s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/5418cf3ce7c64b819d6260552aeb10e0 : SUCCESS in 17m 13s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/ab9bc8d1d93246b39146a950f12d1b6a : SUCCESS in 39m 58s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/b2942ed5dec24c5aa9460bec428450e3 : SUCCESS in 37m 25s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b0dd5c923e6a4c0290349801a50cc3c5 : SUCCESS in 37m 04s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4c656e468dad481f86ec4be8dfc2632c : SUCCESS in 27m 37s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/e93eea23e17c408991d6c7e1bd9de863 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/937827994e64455db9a7e3fb7f59675a : SUCCESS in 3m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eafef47c1625400ba5b6ba4cbafa7291 : SUCCESS in 3m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/27ca811f3bbd40afb246d97d5ebdbd6f : SUCCESS in 2m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0185d45d750a43e49727eb91d852bf65 : SUCCESS in 7m 23s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/044f8168ed88420ab6cea03d1c688689 : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cad4d9e899845948e84597eff9a1b84 : SUCCESS in 6m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0860cd847f904df6bbec435e38df103b : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":1},{"id":"c012ce0dcf5d98b3e0d9136d64b4d4f9f06a4249","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-03-16 09:34:47.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0ee52d50fa6e36fce52be9fa308fba073446707f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-16 10:05:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e273324215f4b5d8ebce144aa329bc4\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/9985f79f425b4e668e7fcdf2a19780b0 : SUCCESS in 48m 28s (non-voting)\n- kolla-build-debian-bookworm-arm64 https://zuul.opendev.org/t/openstack/build/25e8e5a0ac154a498e48371e33a25922 : SUCCESS in 52m 45s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/25c69792059040b1b711a197d4eae127 : SUCCESS in 1h 06m 19s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/1302021eecec4086b8613f23217753c0 : SUCCESS in 58m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"43b07c033b0ff9af9d60f3d225ab205a95599110","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-03-18 13:38:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"55d4a276c7b86d64c39220e27b6a7adc43535523","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 13:39:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"755dee55bb5f5537e7203f27f72fea18e4656619","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 15:30:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e304810e686b4bb082f3f3cfd4fd7703\n\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/fcbec7dbfff14d53a36bb655f70c4c9e : SUCCESS in 52m 33s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/c8cd7b9623f84d90b3a9eb11466f0465 : SUCCESS in 33m 06s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/83cfacaf604244ce8ad0ae4550bf895c : SUCCESS in 51m 10s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/1a034eaa75cc4cc7be51920058be1a74 : SUCCESS in 53m 13s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e116f6797e92473b8e3704a618175401 : SUCCESS in 54m 09s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/765d7005b2c44c8297060ace9fafa8ff : SUCCESS in 30m 37s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/88b2f761b8914bce88e64b7f96c64ced : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae0d752cfbb24d88ad428ee69d5bf043 : SUCCESS in 2m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3812880e67be4d1dad1c5d4a3b8c6b54 : SUCCESS in 5m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e623e0800fda49c690f4050a13cef4e4 : SUCCESS in 3m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98ba75b81a4b442096d10850c5ce31f7 : SUCCESS in 4m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e6e1a761e3584887910afc3ad729a38c : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":1},{"id":"07f43e3e0b6651f89b233eb1bbbe93891f59c2c6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 15:30:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b98380fdad9c32186c3beb3d848bd0fa060d4218","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 15:32:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d2a0f944831459d9d79e66d4e74f0bb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d07e03dba93d47c7a258a9002a9a3107 : SUCCESS in 1m 57s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d0822af8663e4e45bd82be4b14e34a6b : SUCCESS in 1m 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cfe1c404d1320c2f40e2f43f2b5624c4f17b4d91","revisions":{"cfe1c404d1320c2f40e2f43f2b5624c4f17b4d91":{"kind":"REWORK","_number":1,"created":"2026-03-16 08:45:59.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/63/980663/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/63/980663/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/980663/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/980663/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/980663/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/63/980663/1"}}},"commit":{"parents":[{"commit":"6f429307d2d48c878f3039a66530c7cc91889849","subject":"Fix MariaDB version number","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6f429307d2d48c878f3039a66530c7cc91889849"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-02-23 10:40:19.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-03-16 08:45:59.000000000","tz":0},"subject":"Update Prometheus to version 3.5.1","message":"Update Prometheus to version 3.5.1\n\nChange-Id: I66015b45f2c1f915e1aadc57af9c3d239e9c2472\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n(cherry picked from commit 5a0e470152471e27ed04419f27523355abfc6d89)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cfe1c404d1320c2f40e2f43f2b5624c4f17b4d91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cfe1c404d1320c2f40e2f43f2b5624c4f17b4d91"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
