)]}'
{"id":"openstack%2Fkolla~981096","triplet_id":"openstack%2Fkolla~stable%2F2025.1~I2bba43407a10edb283d8605d579667cb8d8e3126","project":"openstack/kolla","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-03-30 15:16:48.000000000","reason":"Change was submitted"},"35345":{"account":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"last_update":"2026-03-30 13:02:05.000000000","reason":"\u003cGERRIT_ACCOUNT_35345\u003e replied on the change","reason_account":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"}},"17669":{"account":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"last_update":"2026-03-30 15:16:48.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-03-25 16:50:16.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-03-27 12:07:44.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}}},"hashtags":[],"change_id":"I2bba43407a10edb283d8605d579667cb8d8e3126","subject":"Install iptables-nft in Rocky 9 containers","status":"MERGED","created":"2026-03-18 11:33:54.000000000","updated":"2026-03-30 15:17:47.000000000","submitted":"2026-03-30 15:16:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"981096","meta_rev_id":"dbaa501121ce3d1c6ca47bbec22d95fc48d9f45d","_number":981096,"virtual_id_number":981096,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-30 15:16:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"recommended":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"all":[{"value":1,"date":"2026-03-30 13:02:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2026-03-27 11:46:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2026-03-27 12:07:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-03-30 13:26:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-03-30 13:26:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2026-03-27 11:46:01.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-18 13:38:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-20 08:04:36.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-03-20 16:06:16.000000000","updated_by":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"reviewer":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"state":"REVIEWER"},{"updated":"2026-03-20 16:07:52.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2026-03-25 05:41:28.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2026-03-30 13:26:56.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"fff9a483beb5950f3921bb0fae1a7f975fe40d8f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-18 11:33:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b57459ff9d091383be4ca49a8967b04d20a74955","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-18 11:39:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fe2fd59fdddad3274bbb01ee27521be3f46e1382","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-18 11:39:12.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"c68d6fc148106d50c031c46cd69d241c0bb4aac4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-18 11:53:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2fe65b21947597534e31cc66a6f95dba04390e4f","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-18 11:59:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0b5d9e2a24c37336f9862ef98b45c05080844dd6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 13:38:43.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/07222dd04ba4414e9beb0936877d8380\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/04981e865f0f4373a6a5376ec4e57708 : FAILURE in 59m 30s (non-voting)\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/a42a0d04ad034ed798ba39f3ce31b46d : FAILURE in 40m 25s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/f216928738994dacb5fc4814bf8db6ae : FAILURE in 15m 58s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/96c3034744524c8caeba09511905aa4d : FAILURE in 28m 34s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ac0e735d23aa48789ce773a480391789 : SUCCESS in 50m 32s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/ba167b9593634cd481076b4e6a5be6bd : SUCCESS in 41m 04s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/7cead48457c24bad8ea6f99fa3005875 : SUCCESS in 1h 00m 50s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/0d6f65cc4a054fc58f033cf2b7cd7498 : SUCCESS in 1h 30m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f4d23126b704845bd25c59a863f53f0 : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a0a722c99e0b4b3c94acf18753ea8a41 : SUCCESS in 4m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6ee9173c06024d57bbde53d67b44aa71 : SUCCESS in 3m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2dc79814c8f14c5295d45a39dd3ea2d1 : SUCCESS in 3m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ea79bfa84f949ac94a8507a78aedf65 : SUCCESS in 6m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/918da257c3444e689a82867c1c9a22eb : SUCCESS in 3m 36s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/ed2caa4a14314eaaa9b035c23c58014b : SUCCESS in 4m 24s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/c7e1b40de7d7444bb25db53dabdbd32a : FAILURE in 52m 15s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/20d7b455f20745d0acec0d668f328e02 : FAILURE in 1h 16m 58s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/b6be751f6d6f4941b84ffe74f858cc26 : FAILURE in 29m 53s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/fccbb3a13a754d2892546cb1f44223da : FAILURE in 41m 17s\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/71178700ca0a47969c237190422258e5 : FAILURE in 32m 10s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/8c65c00fee854b9bad81d5682c8a5598 : FAILURE in 31m 03s (non-voting)\n- kolla-build-rocky10 https://zuul.opendev.org/t/openstack/build/bc31006da13f4122914cffcc7dc86fee : SUCCESS in 1h 03m 47s\n- kolla-build-rocky10-podman https://zuul.opendev.org/t/openstack/build/5e8f68565a99424b900eca1bac59c611 : SUCCESS in 42m 13s\n- kolla-ansible-rocky10 https://zuul.opendev.org/t/openstack/build/eeb0b15946e04d5983f5da63a15076cd : FAILURE in 23m 27s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/763a91c64767456580935fafd367e479 : SUCCESS in 55m 55s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/ab5ae2d89e6349dcb9ac68e50dadba6c : SUCCESS in 26m 39s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/32e43fa62daf435ba636a831fef16ae5 : SUCCESS in 1h 05m 48s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/cde1ed73132b4124a9978ccf17eb80ca : SUCCESS in 1h 44m 04s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/34c15a42fda34782b4b6ec8af8b1c2d6 : SUCCESS in 1h 16m 36s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/a8dce3a140b34327bbd487d2f8e9e8fb : FAILURE in 1h 31m 20s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/7b1dcce4d59347d19bc533c57fc02678 : FAILURE in 1h 34m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"32dc6dc6f94aab026558a7790604f403b6d58e6d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-18 13:48:44.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"f03642d63cd73080f8f95b868cc6bfafd71c20e9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-18 13:56:20.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c1677d72b9b9e7fe4150163a72829970aa26e2d8","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-18 13:57:13.000000000","message":"Patch Set 4:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":4},{"id":"2e076516ad0a25ec40beacc246e073302e4a5c73","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-18 13:58:58.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a127b8653344d59a153ae773fc7b63a04c8e9a6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-18 14:52:01.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"5e5eaae2769303eb051a50d830c882f257d14758","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 17:22:29.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9236d07eb0f3464cbdbc2c9905bb8ce9\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/9d3fac35d0654a96b4fdbab3e6dc7de2 : SUCCESS in 51m 16s (non-voting)\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/eb1a374758ec4c989ebbfcacadd4865c : FAILURE in 53m 00s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/30449cbb01d845e49b675e04d89ddfbe : SUCCESS in 1h 35m 19s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/9e03cdb11e9c40b2952761c2dea2dc51 : SUCCESS in 1h 18m 51s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5a26a5cfc447474786c208fd9503c9b8 : SUCCESS in 51m 41s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/40461e47c23845f7b3f7153a6e9f696c : SUCCESS in 34m 24s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/a8e516f518e14181a5e94c01d0c67577 : SUCCESS in 1h 06m 21s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/96c96e3000cd4387b84f117e74a7de0a : SUCCESS in 1h 38m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59d389234bba44b3af4ca4b7bc5ee122 : SUCCESS in 2m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c27edada85ff49a882ca8d6cad1b8df2 : SUCCESS in 3m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9a1f1858def442da81697937828bff96 : SUCCESS in 4m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b1888b18a6f44393b4ee83c31b098278 : SUCCESS in 3m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35a0351c43bf404db07b5508c50fdd6a : SUCCESS in 6m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/853dc3452b9a4c0fa0cd5f3dac626bbe : SUCCESS in 3m 34s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/46c8d409fca342de85c9e7062a1485b1 : SUCCESS in 2m 56s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/031ac02c8d1f491aa4b18dbad0d4da20 : SUCCESS in 52m 21s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/2bd4a4d9b7a24df4835309c7de398027 : SUCCESS in 53m 22s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/2f0053e444a64736a2c07d6827c79708 : SUCCESS in 1h 40m 02s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/c0ca281d9eef4a2189bb88df1bbd8b91 : SUCCESS in 2h 13m 29s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/ea10bd159efd44cb8839677b6a80ea89 : FAILURE in 1h 12m 09s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/40dbbb83fbce427eacd274105f481948 : SUCCESS in 2h 05m 06s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/3842bd8108d44fcc8b90311bfdf09b33 : FAILURE in 1h 22m 52s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/3a6a764c5d61457e91bc282d10d59593 : SUCCESS in 32m 27s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/7781cae689074fd098db435275090ff6 : SUCCESS in 1h 20m 01s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/78b3130195144ef2b8741d84e50745a4 : SUCCESS in 1h 23m 10s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/d082524ab1084f6587d066a998b23472 : SUCCESS in 2h 10m 16s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/e9da4e2adab04682ab8340c978684fc7 : SUCCESS in 22m 20s (non-voting)\n- kolla-build-rocky10 https://zuul.opendev.org/t/openstack/build/3321250f748041ba9f534e5801297528 : SUCCESS in 41m 46s\n- kolla-build-rocky10-podman https://zuul.opendev.org/t/openstack/build/b7d61db9e2b84800bcce74dd559285a4 : SUCCESS in 53m 53s\n- kolla-ansible-rocky10 https://zuul.opendev.org/t/openstack/build/7520b1cbe6e8430497e5ed104a959416 : SUCCESS in 1h 33m 53s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/cbe5de02052742d78e213a644cb13280 : SUCCESS in 1h 05m 52s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/84c259854a8446a3bae65b584bfbe938 : SUCCESS in 14m 01s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/6aa9f96a8f884183ba084dc999d8ea28 : SUCCESS in 1h 07m 09s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/c82a242aa3564d5d835ac9809e62dc91 : SUCCESS in 1h 35m 54s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/b5e492518f744cbcae6a82afdb3f978c : SUCCESS in 1h 49m 47s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/a017ab78e4054e45b891eb0c17b688f7 : FAILURE in 2h 22m 32s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/1449680b15484efd8658016742418263 : FAILURE in 1h 26m 01s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/08e5b860a75b40ba864f0c6cfa7f29ab : SUCCESS in 35m 30s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/744cf55c5fec4e6eaeabe25fcfd37372 : SUCCESS in 1h 08m 44s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/489acac7c7dc4727871809c52947c4de : SUCCESS in 1h 11m 21s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/cfead446a8014f0a9bc20e0daa12366c : SUCCESS in 1h 53m 53s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/46e3fe16b388478ba78788d5e7df6ea2 : SUCCESS in 18m 33s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/e3969af503ca4714abf66630fecf724b : FAILURE in 57m 22s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"035e4ac02ca59a89f252ca2a055c9807a5a40027","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 21:14:58.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61324dc718544a7ab36da6647a93fe7c\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/f43c01a1d98a4d00a4d5cb11e34f6b32 : SUCCESS in 1h 26m 04s (non-voting)\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/a1ec38d332384b2ebedbfa6489ec0783 : SUCCESS in 1h 16m 11s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c4b181e9beab4e5f81e38d46f31b690c : SUCCESS in 1h 56m 41s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9e023aa2873e4a3c9d45b59e6aab3bb7 : SUCCESS in 1h 26m 28s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/ddd400dff4b7441faedf2decfe6e9ed9 : SUCCESS in 1h 30m 30s (non-voting)\n- kolla-build-rocky10-aarch64 https://zuul.opendev.org/t/openstack/build/f31f69991a5644a0b9032d76842a4dab : SUCCESS in 1h 36m 13s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/a89e0e42bcfc4cdaad28956e51a524c0 : SUCCESS in 59m 12s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"6e7cebb5b9693866351e475e007e27f0d1490feb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-19 11:31:26.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"9152c22a0858424e0705a3a2492357b26ca9088b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 14:08:10.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/053686c3c40e4db4a17a691ab98465fd\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/e051f5cd7cc54a91be452b5c04d3b3ec : FAILURE in 40m 17s (non-voting)\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/6bfe1ee50b484c6886721d8e49283c54 : FAILURE in 46m 12s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/25dff35cc62548aab44f0798fcf7257c : FAILURE in 20m 52s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/478bcf93fd424b2c8e938a1637638863 : FAILURE in 30m 24s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f3e5a1d604e34c8b95cff03be467ddd5 : SUCCESS in 43m 59s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/7dfd1c26e85b432cbc4fb5f0ec232c96 : SUCCESS in 14m 42s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/758616699108450d8c2c892024dd8e30 : SUCCESS in 1h 02m 20s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/5ab899d7642c4c7d9a70117537620174 : SUCCESS in 1h 34m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d365ce1f2d1047e4b3a57b2de376d14e : SUCCESS in 2m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dd33a495af3d4828befe3a83eb141920 : SUCCESS in 1m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4253df481a0c4ad0920a0228fddddc29 : SUCCESS in 3m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/25dc40f974a846988833c2586c005aaa : SUCCESS in 3m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f36240a06d24f8b93e898bf4f5dcbeb : SUCCESS in 10m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fd11ef4760924252b63aba5be6458984 : SUCCESS in 3m 17s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/5ecab16ef59b4d1a99f6845a1d845594 : SUCCESS in 2m 08s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/1a51e7d2258a4584b0172ed0e3009959 : FAILURE in 41m 51s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/87dc76153c384689ab28859cb17eae49 : FAILURE in 1h 16m 16s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/8a6c521a9822491695f519e676c8446c : FAILURE in 32m 48s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/30b43e5e0c6c4f9fa832595e93e95f23 : FAILURE in 16m 14s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/b26d86f63f8e435f8d03d92a274e382d : FAILURE in 26m 14s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/dc13556fe6a24cf58b8c1246a267b18f : FAILURE in 33m 07s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/8160179e80e54aafa875e25c2866621d : FAILURE in 30m 59s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/45e4a5dc6a174b8ca2915d6631f53de7 : SUCCESS in 35m 07s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/351f09809655447c96f49e8d58cdcaf9 : FAILURE in 35m 14s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/252adcbe8fbf404e9cc4495abd90601d : FAILURE in 19m 14s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/051fc5adac474547b951c096ad0c7004 : FAILURE in 18m 54s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/40a75379249a43729dda17baeba49365 : SUCCESS in 43m 07s (non-voting)\n- kolla-build-rocky10 https://zuul.opendev.org/t/openstack/build/f4445ae2e3d3455590f9aeb77a7644ce : SUCCESS in 50m 20s\n- kolla-build-rocky10-podman https://zuul.opendev.org/t/openstack/build/6a542b67940b455c88ebec069ab414be : SUCCESS in 1h 06m 37s\n- kolla-ansible-rocky10 https://zuul.opendev.org/t/openstack/build/1c6fd0c0a57047169f09d3d66163b70c : FAILURE in 23m 27s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/c038c5c017dc42dca86792aefcbe110e : SUCCESS in 54m 03s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/2b90e43188ea4039b4893ed4145563c1 : SUCCESS in 28m 20s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/7ec83b0c429248b5838a4e7331eb1c46 : SUCCESS in 1h 16m 10s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/18aa76a9863d4e0b8fbab678b9cb5de8 : SUCCESS in 1h 51m 29s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/c0ee5038781346aba9ff53990c6ab447 : SUCCESS in 2h 11m 11s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/acfb752b9cc54d9ea2416973687e8bcd : FAILURE in 2h 35m 40s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/fec3a742f2c94a0694a176c28b64a2e5 : FAILURE in 1h 14m 10s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/fe93be7b7a064449975c4a0a20f0cdc9 : SUCCESS in 30m 47s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/2e133b9d9c7d4f1b907e79bd521e6e7b : SUCCESS in 1h 10m 37s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/ba1c1c41722f433aa113ad341084e5a2 : SUCCESS in 1h 10m 07s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/6b9b5f9dee1d4f6d8d189dca528ab7ea : SUCCESS in 1h 44m 47s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/90ca672c03e24526a246224f70465599 : SUCCESS in 23m 23s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/16c13b34abbe407d84374f9109c959e7 : FAILURE in 1h 14m 36s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"285ce95603219cbcbf825adb014d5756852146dd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 14:52:23.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/992671bb394b4679b77bf87d0de32e90\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/480bc84495e949b29b3b63728d91563b : FAILURE in 2h 08m 35s (non-voting)\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/7ccc464132a648ef96495173a3bb3ca9 : SUCCESS in 2h 27m 12s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/711818b388a34fee97fcc3938d11ab46 : SUCCESS in 1h 55m 07s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7bdb33a55d3e4d2195f549e83e501216 : SUCCESS in 2h 37m 30s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/d18469ded4e74c3c8b883296d03095a5 : FAILURE in 1h 55m 45s (non-voting)\n- kolla-build-rocky10-aarch64 https://zuul.opendev.org/t/openstack/build/298dee819bda477ab3547a214b3a8092 : SUCCESS in 2h 10m 34s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/24d7d6759c6346f4b06d73678d5f0413 : SUCCESS in 2h 09m 40s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"f0b6d6482c305c1f6b325f9b70bca51e6ccd6a9a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-19 15:34:09.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"2f4faea0c04be83157d11f854342f551a075e99f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 18:14:13.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ad7947567894468a10b33ffeff92ff8\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/1ebfde4ca23648f2a5b7bf44958382fe : SUCCESS in 47m 57s (non-voting)\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/82cc8abb46424fd3a340ebb3034b35b4 : FAILURE in 57m 30s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/6ceb3dbcfe73476c8271ed5983c5dab9 : SUCCESS in 55m 49s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/e4e14ed232544b41987bbd30066c1b5d : SUCCESS in 56m 34s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ea1d79da156e40ff8aa1153a09c48c26 : SUCCESS in 50m 53s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/b342d9f0516e477fb70862832cb19393 : SUCCESS in 30m 50s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/046b254ddecf4ba5ab2d8a368cb89031 : SUCCESS in 1h 05m 20s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/f8b0e5c2caa34d90aa53e87c4ca8dd6d : SUCCESS in 1h 27m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/815db31a12914a97b21ae8c819c45251 : SUCCESS in 3m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6292d4d19b9e4823990c38096ddcabd0 : SUCCESS in 7m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b2f6686a082c4df1b663b373ed13ad65 : SUCCESS in 4m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cdc578975ad348e49dbd187d4f4ad319 : SUCCESS in 2m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80b293fed1d04dc7b8b8c10f3267341b : SUCCESS in 5m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/982fd67f57114891997b57d76d2f6b90 : SUCCESS in 4m 01s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/2216650d3649482ab24b95c66109ae90 : SUCCESS in 4m 01s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/1b0664d4d6524e84a140147237ef536a : SUCCESS in 57m 19s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/41d977a0135a471fabeec2051d18ed4a : SUCCESS in 1h 27m 50s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/24aad0b75e744a9381820a4e3104afcd : SUCCESS in 1h 32m 53s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/7257c0bb2659407ea90dbd0c0d50a5b6 : SUCCESS in 1h 18m 21s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/8453989aeb7744ae8f512cd7091b4875 : SUCCESS in 2h 28m 56s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/3f68a65e5f03442db9c212aaf359d261 : FAILURE in 2h 00m 11s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/14813f69ba714d67bb89f2a8841b437c : FAILURE in 47m 26s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/4f2c86db3d744242a0a9f26a612b94e5 : SUCCESS in 19m 23s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/60be1652e0d748c6878a3225ec87809a : SUCCESS in 1h 33m 06s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/822a0126b8234487ae9a9e784a780da5 : SUCCESS in 1h 39m 20s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/e415d6cc5e5845f98a2e77323294fe4e : SUCCESS in 2h 17m 58s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/db4a028729114c79916645f40d965785 : SUCCESS in 35m 39s (non-voting)\n- kolla-build-rocky10 https://zuul.opendev.org/t/openstack/build/5272f288d2264249b10889808ac3472e : SUCCESS in 43m 18s\n- kolla-build-rocky10-podman https://zuul.opendev.org/t/openstack/build/ec212982e2b0475e9064a5153e46b917 : SUCCESS in 1h 16m 00s\n- kolla-ansible-rocky10 https://zuul.opendev.org/t/openstack/build/ff5aa48fd2cd49a38bddfa4dc863f318 : SUCCESS in 1h 34m 12s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/02c63eea20974bda9c6649f668934d07 : SUCCESS in 56m 25s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/649e2b519a45420e968cf1173adef5db : SUCCESS in 35m 01s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/4e39cde1fe344420bfc58c22c6a4ae6e : SUCCESS in 1h 06m 14s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b5e7e162d1074ad1a1208c771acb843d : SUCCESS in 1h 35m 38s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/dbaf30ffc7ca48dcbffd9f42900978d1 : SUCCESS in 2h 02m 14s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/674148d33cd8473ca36ab3e94d0d4a9d : FAILURE in 2h 19m 21s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/93781270002e475788746d21bfc1cc66 : FAILURE in 1h 18m 20s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/2986722104904bf4bcff94d12de2d048 : SUCCESS in 25m 32s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/e1f21453cf8a49149c1c2f5b6a5610ed : SUCCESS in 1h 10m 14s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/640b7270e4614952b4dbe68182018c54 : SUCCESS in 1h 21m 49s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/6dee7c5f1b6046d1ab892c266754032c : SUCCESS in 1h 56m 02s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/a570b0c75bed4fb4bddad77c322f4a41 : SUCCESS in 37m 23s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/92809cb3a9cb415882b13c775c4627d5 : FAILURE in 1h 38m 38s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"9f89fa614d69f760c9a3fd35a05b4ce316333309","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 00:53:56.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac29826840504d0595d9b20cb9074fca\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/9e4c95db01ed4e20ab97e3888fd9312c : SUCCESS in 57m 21s (non-voting)\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/be992a2364804942b87d2c7c0d929701 : SUCCESS in 1h 11m 22s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a3a5329d7c7444e382be93d2f8cef55f : SUCCESS in 1h 29m 31s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6e640d320b2b4e0f93426bd960b561fb : SUCCESS in 1h 30m 24s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/e9534106d90e4b8c8fefe376bc9eb5e6 : SUCCESS in 1h 37m 39s (non-voting)\n- kolla-build-rocky10-aarch64 https://zuul.opendev.org/t/openstack/build/68e40c7e9d2b4cbeb30205f34d06f12d : SUCCESS in 1h 14m 48s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/ba8155e5e503434b9009cf0b02205b59 : SUCCESS in 2h 17m 42s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"5d8dd22b10ed5715fb89a250a3aa657dead3b78b","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-03-20 08:04:36.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"dca309f9fb7f5327c28c830c2d2797de07db35fe","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2026-03-20 16:06:16.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"b734dcd536588eb500a15470f4f9991652c5695c","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2026-03-20 16:07:52.000000000","message":"Patch Set 7: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"115d9574a19b6cdb5094b6d193e8d88683dc0417","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-20 16:29:18.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"0d18b17fe9a7029c3694be9d9e087c53adffc28e","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2026-03-20 16:47:45.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"243acc15478194e7a48926542dff483f84ec8905","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-03-25 05:41:28.000000000","message":"Patch Set 7: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"38c0212744a9d8d9ca4d6c722871958fbe481423","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-25 16:33:59.000000000","message":"Patch Set 8: Published edit on patch set 7.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"32d6f1b608fe8ef2d013ea8400e1d0fe47962287","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-25 16:38:25.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"afa0983a86a10a9d2d060967b0b2873d95e1e191","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-03-25 16:50:16.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"169109e9cec02c31247139b20cd3c94f4b03af6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-25 21:11:55.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d110d8f2b1c4482a191944833e81fd4\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/61fdc17129b54cbe9770bbc91703794c : SUCCESS in 1h 05m 39s (non-voting)\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/ba28e5b0490e4ccc9a0c24eef83d5b9c : FAILURE in 55m 58s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/1ca67929273d4a1996b5ca54f2a639e1 : SUCCESS in 1h 35m 07s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/b5c198d0daeb4959a5ecb9ae0e4e964c : SUCCESS in 1h 34m 47s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c07d1968000c4b019ce09c36001734d1 : SUCCESS in 52m 18s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/3cfdba4bc46346bebb52c32389bccd45 : SUCCESS in 35m 09s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/8f06ed00bb304f189f61658741c42b59 : SUCCESS in 1h 03m 05s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/744c7911aa5943669026bb0bdaa53f4e : SUCCESS in 1h 30m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b671819f97804ce0a6f777edc5deeb7e : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3c5146524aa24333b44d129d38f26375 : SUCCESS in 3m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b77c637b85f54fff8845c76632d5cbf4 : SUCCESS in 3m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e391eb3fc89a41519cc44a7209fb9de8 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/343e440db14b4f78aa5da039f7c601ec : SUCCESS in 7m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3a22f9e87f7c4654887ed9f7c728f529 : SUCCESS in 5m 35s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/61bbff5860df4b22b3f97702d00640de : SUCCESS in 4m 55s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/803e42dd11ef4a92ba0e3d8648da0000 : SUCCESS in 1h 02m 27s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/72e2b8b744964fc297eba42727c605c0 : SUCCESS in 46m 02s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/78b61dfcdcf8433b913f4fe0821664e8 : SUCCESS in 51m 45s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/db68131f527349fcbf4e263475728652 : SUCCESS in 2h 21m 55s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/a4c02c001ca447d096e12d2aea5682a0 : SUCCESS in 2h 24m 21s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/82bdd937abbf4608a7fb0a10fcf7c90b : FAILURE in 2h 02m 18s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/dd55b0e9f7794a68b33300d5dc5b00e4 : FAILURE in 1h 34m 58s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/7d18bcef829f402e86482e7cd6b727d5 : SUCCESS in 19m 20s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/c14ea703d5b6473cb676888bdefeaa11 : SUCCESS in 1h 22m 25s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/3ab4fe82f8e144589a8fc1d1c845c023 : SUCCESS in 1h 35m 19s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/b4bcf3ac64494ae3a123c13deb48ceaa : SUCCESS in 2h 03m 00s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/4019f7ed9a7242fca3da514b34e6aff6 : SUCCESS in 35m 49s (non-voting)\n- kolla-build-rocky10 https://zuul.opendev.org/t/openstack/build/3cfd48de846d4269ae0effc5a7d1f2ba : SUCCESS in 48m 48s\n- kolla-build-rocky10-podman https://zuul.opendev.org/t/openstack/build/a93b043bf0c64ab69b4c94c1ee4766b6 : SUCCESS in 1h 01m 15s\n- kolla-ansible-rocky10 https://zuul.opendev.org/t/openstack/build/e028c18f3d1446cf8c39dfd6f4988915 : SUCCESS in 1h 16m 11s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/1544969622984cabbbc9c42b244c6a3d : SUCCESS in 49m 43s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/b023a89ad54d4cd98acbee9903771a47 : SUCCESS in 27m 42s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/cea954206567417d99758ac2b4ae2d27 : SUCCESS in 1h 07m 41s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/a6a81eff840344a3afcc3d2cccc9c1b3 : SUCCESS in 1h 39m 46s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/94549b92295e4532b57b5d25c251e63b : SUCCESS in 1h 56m 26s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/71aa886fb2e64b74b612d65c11680f20 : FAILURE in 2h 05m 02s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/c2a152a2b36d4cb2b684277ad46e496d : FAILURE in 1h 16m 28s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/5a571740c8104d5698d4ea55f0fafd18 : SUCCESS in 23m 42s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/33ef453243d84b0cb71534cacd8c343d : SUCCESS in 1h 17m 35s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/4f9eaa6bfb9a41acb12db93ea432f810 : SUCCESS in 1h 12m 56s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/e8fd4d9b0f5e4a95a983597cb28d31d6 : SUCCESS in 1h 45m 13s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/7d2f69a7e3ce468192102411697cac19 : SUCCESS in 41m 29s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/b0fc4fe3ce2743d896fd01547e63bd85 : FAILURE in 1h 34m 17s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"1396e9fa98cd2f6dcba3ca808cd013b975390955","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 11:01:38.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53c8a600ce7545958565544b81b2dfd9\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/d754ae7fa741420881454fbe12173813 : SUCCESS in 1h 07m 20s (non-voting)\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/68cb601c5d674015b85de358d2ce8704 : SUCCESS in 1h 18m 10s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b66c9e217a7c4cb6825554c9665fe8d2 : SUCCESS in 1h 05m 41s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/415287cca24b41d5a9f18f4c117fcfe0 : SUCCESS in 1h 30m 07s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/9ffaa895a07446d2b5e27954498d7f4e : SUCCESS in 58m 11s (non-voting)\n- kolla-build-rocky10-aarch64 https://zuul.opendev.org/t/openstack/build/7a40d1603e614955a6ce119b11ab9803 : SUCCESS in 1h 11m 06s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/c1f5d25c78df4751a6c666e7c79774eb : SUCCESS in 1h 11m 05s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"abd4759ca428fa2145b16b797e326d85b0838bf3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-26 15:26:37.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"ed2feae1f8c9cefd95346eb808c5901bdab39056","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 16:40:23.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ca4082e6113544a09c294bdec078544d\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/7b28eb42a83a41baae363ba2626ba0f6 : FAILURE in 1h 06m 37s (non-voting)\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/d13b60f3e0c5464ba609b7cbf8cc9453 : FAILURE in 38m 21s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/7bf93a0b29974efdb53aac4b3f9b693b : FAILURE in 30m 48s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/2914628a223648b6b2440ea8e5011206 : FAILURE in 17m 38s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a0787510d0174dc88dd7a467fc0cad44 : FAILURE in 35m 35s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/f33c3e8c761d4844b663797cd3348d5f : FAILURE in 27m 55s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/847b169c0a7b488a832a5da81aeb28f2 : FAILURE in 18m 48s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/664f24c6af8845d2817f64ccea683716 : FAILURE in 19m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ff3f7b396aa4eadb7020356afdae73d : SUCCESS in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/724bc47181094f189300f111dae73639 : SUCCESS in 4m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d9e29d8bf4984f91ae1341717171d375 : SUCCESS in 3m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd41144ffd9c443caeafbb6b27aadd1a : SUCCESS in 3m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e135d30d1da445859a7358c60880b719 : SUCCESS in 7m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7e1d5b69d17048bd9ae4a0d15c96f3ef : SUCCESS in 3m 51s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/8a0cf60366f04690ba7853061b89eef8 : SUCCESS in 3m 02s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/7181b973b8234eb2af9d49abd1b431c2 : FAILURE in 37m 44s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/6b35c167a79e445aa7cb2ec69d223247 : FAILURE in 1h 04m 40s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/5052a99391ec42de997b6f3eb95f37db : FAILURE in 31m 46s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/5647567167a84002a6be97d793dce80e : FAILURE in 29m 15s\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/d47a64181dda49ad8a8ac8a6c0eee166 : FAILURE in 28m 38s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/46484759feff4546a489b18ba233d215 : FAILURE in 19m 35s (non-voting)\n- kolla-build-rocky10 https://zuul.opendev.org/t/openstack/build/ec8b6886e94f4663b7f84c56d5726c95 : FAILURE in 37m 41s\n- kolla-build-rocky10-podman https://zuul.opendev.org/t/openstack/build/a3a477f4bd4a4da2891cee8d6ac51550 : FAILURE in 45m 11s\n- kolla-ansible-rocky10 https://zuul.opendev.org/t/openstack/build/35d4ac043cde447d97a2c447e5997c20 : FAILURE in 17m 04s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/c6a1a0e9e5184767bf3f8c8753299a3c : FAILURE in 43m 47s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/2c8973d750914fc5ae4fd2ec5b65eafb : FAILURE in 26m 02s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/6d02a0b83eae433d9e1082fbd942f296 : FAILURE in 31m 18s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/15d31cac85804ec180bd1266b0019578 : FAILURE in 34m 57s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/5c942186333c419a8d09157fd5182342 : FAILURE in 41m 58s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/b569a6b8c8c74c2caa3692608988db4b : FAILURE in 40m 30s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/0d520d53386a4ee79892400966951ee5 : FAILURE in 28m 58s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"85b86ff8df5b48c60822995f32cefbdbfe19d80f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-26 18:29:08.000000000","message":"Patch Set 10: Published edit on patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"08a50a578caac29fc9f7dd018c2598d07635fb59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 20:13:18.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4fb7fae82d064dd5af9e8835ba730270\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/4ad1f7fe2ce3417290246d9936161a64 : SUCCESS in 1h 04m 43s (non-voting)\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/e4aa69170f1441e293da40b47345432a : FAILURE in 41m 27s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/d3f7df6fc38c46d4860c796edc674fe0 : SUCCESS in 1h 19m 49s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/aba8f6a9aef84020ae6db3b8a685d1b9 : SUCCESS in 1h 04m 00s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/e2668c70f6bf4453b4bfe18156a0fe5a : SUCCESS in 41m 02s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/0c9151607fda43e4a27e3acb792adaa7 : SUCCESS in 18m 33s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/39e7ee23d8cd461db25337d62d3bc2b0 : SUCCESS in 56m 52s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/5e76ac014e7b46c6a2a21591d3ba48cb : SUCCESS in 1h 28m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30dbc803f48f4f33a33f15756ad5852f : SUCCESS in 2m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4a978a1f1e6e4d62922803bc2ddab525 : SUCCESS in 3m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/082b703d77d047c6a7a7f78cab97ba77 : SUCCESS in 3m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ae3b499cb754a74b3e04973e139ebb0 : SUCCESS in 2m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f3a61c958d049c3895b3ad1198fe1a3 : SUCCESS in 6m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d30affa6059346e4ae3d2c8afa6a85f6 : SUCCESS in 3m 32s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/7186c2918d394ca18fe4a14435f7f11b : SUCCESS in 3m 31s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/76e471a2ae3147aeb5bf9d9fb811c2b7 : SUCCESS in 1h 03m 02s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/9964dab5d8744aa0b1d26159164c79cf : SUCCESS in 1h 10m 07s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/d018fba7a04d474e94963edc19165df1 : SUCCESS in 1h 29m 36s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/2849654dd0254b4b82e5ed588c056e23 : FAILURE in 1h 08m 25s\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/aa2de4dc9a204fc5b12a330ce33c90dc : SUCCESS in 49m 09s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/eab24050bcbb420ca61b140d7f11b591 : SUCCESS in 1h 12m 55s (non-voting)\n- kolla-build-rocky10 https://zuul.opendev.org/t/openstack/build/720b064a14644c54bb9176cec5a75f10 : SUCCESS in 47m 20s\n- kolla-build-rocky10-podman https://zuul.opendev.org/t/openstack/build/4e71d9fb32fd4024a2779e5aa09ebb37 : SUCCESS in 42m 13s\n- kolla-ansible-rocky10 https://zuul.opendev.org/t/openstack/build/9bcce8d58e0944a98878454c2075c2ea : SUCCESS in 50m 28s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/1340fe5f01a64bb7be926a51f4e4ed52 : SUCCESS in 56m 14s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/c3228394e13e4ce489e612cd8ea9dfd6 : SUCCESS in 31m 59s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/71c1be58de9b4f7cba3f9174d268ccad : SUCCESS in 1h 08m 12s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/0f695892703e4b9ba83e021b69366104 : SUCCESS in 1h 36m 43s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/c56e840851fc42fcb07e26e2a12a53b9 : SUCCESS in 1h 08m 51s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/4adee4a1d3c840a6bd8b301350ea5111 : SUCCESS in 1h 39m 38s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/2ad87dba62be410e88fafac7b10e20b8 : FAILURE in 1h 35m 17s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"cecfec3614b63f6ef44af8e50d92a1b2069468c0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-27 02:13:42.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3ca92ffd2c147728b746eafa96702d8\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/7839c1b7e7f149bda8efbfc32885f53d : SUCCESS in 1h 08m 29s (non-voting)\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/c32b90363a904e00a16ecc53c42bf2d4 : SUCCESS in 1h 04m 36s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/071ad87f10b9467ca935cbd8d0c9b985 : SUCCESS in 1h 01m 56s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/791217833d72453c8790ce3542bcdc9a : SUCCESS in 1h 51m 10s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/dcfed0c10e804526b22888aa15654b5e : SUCCESS in 1h 10m 45s (non-voting)\n- kolla-build-rocky10-aarch64 https://zuul.opendev.org/t/openstack/build/0676e34610f14153a83f68f8957c2e23 : SUCCESS in 1h 16m 55s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/70a0659a0fad43b787860b2f5ecc5c5b : SUCCESS in 1h 41m 47s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"267f086816d87145ebb9124b5312e33d10f4eab5","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-27 09:30:22.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"6c4fe24e82a24c448fd322b28d71477c3b85d8b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-27 11:42:41.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/46421f7679594defb162dd46f5bd655f\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/6d49f2b88d6946278374491ede54d128 : SUCCESS in 52m 03s (non-voting)\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/7eafe2dc81304cf28beb03d1f646cfb4 : FAILURE in 52m 57s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/e87d385db6454722b19e4aa824e9b929 : SUCCESS in 1h 23m 39s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/98e063bedebf4de5a55b692265ebc18a : SUCCESS in 46m 11s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/7c36c11e850e456b8ab0434cfd72b1fa : SUCCESS in 36m 03s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/81d28134abcf4b639670a9af3ec7ec09 : SUCCESS in 30m 28s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/5ab9b60279a0437a92a6794f98f1ba38 : SUCCESS in 58m 10s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/c38e64b4838749dda869fee030a329d4 : SUCCESS in 1h 35m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5f1f05f0c244713a9646fd7b9deb75f : SUCCESS in 2m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b0f730586f7f48e38609203184c8079b : SUCCESS in 2m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6c651244d93e4a0fb374f78664f64247 : SUCCESS in 3m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4c42da186edb435abb7b035defded4d4 : SUCCESS in 3m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66ecf0ae3a9d4c36a251d3a27cdfbb1d : SUCCESS in 6m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2ee919c3cee74b29bf397c299ba9c9cc : SUCCESS in 3m 41s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/b7d43b921072455d936ae21f93a7653a : SUCCESS in 2m 01s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/92e906f57dfe4cc88a1e2791233aa342 : SUCCESS in 52m 25s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/6ec60a390c4e43e4bd748f368b5334f5 : SUCCESS in 1h 15m 42s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/065ecf44600a47f69bfd125a9b307cb2 : SUCCESS in 59m 27s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/f0de19b10819453eb6aecbb617f222df : SUCCESS in 1h 21m 07s\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/ed82c4f9734c40ab92a7dfd9c039f05f : SUCCESS in 1h 58m 13s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/c1818e418825485f907068a7f1e8927b : SUCCESS in 2h 07m 06s (non-voting)\n- kolla-build-rocky10 https://zuul.opendev.org/t/openstack/build/dce657c7f29f42dda780c1bdf6687bb5 : SUCCESS in 44m 08s\n- kolla-build-rocky10-podman https://zuul.opendev.org/t/openstack/build/37bf046cab69401cacf3094737827cff : SUCCESS in 53m 41s\n- kolla-ansible-rocky10 https://zuul.opendev.org/t/openstack/build/0fa09257514b4ae2825f3aeba8400649 : SUCCESS in 59m 34s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8e735657f3d64a359d93f0e7cd393064 : SUCCESS in 34m 56s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/4dc142b0093640c58448f6598605cff3 : SUCCESS in 41m 11s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/072e238e58b0472a8f4003e09cc55213 : SUCCESS in 1h 13m 46s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/2f02817e64b24207839d71c5780895b7 : SUCCESS in 1h 44m 45s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/e2d03627d0694410a74a3ab2a722d038 : SUCCESS in 1h 14m 30s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/0be6a6b7847c4260ad8a621f1d5c682e : SUCCESS in 1h 38m 44s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/184ce51af7c844d7b7e3ecf4528d9fbd : FAILURE in 1h 29m 46s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"f3f7b6f272a7e7333fc83734341bd701cbabeacf","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-03-27 11:46:01.000000000","message":"Patch Set 10: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":10},{"id":"6c4c32f861a938ea6e2164752df08074c45f6761","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-03-27 11:46:12.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"dcda67ac31752365a1525ad738f3cd94535402d7","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-03-27 12:07:44.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"643fda408a007c0412021fea81278c5a79abdfb1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-27 12:27:09.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c928abc191c4258a79e758046371122\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/a4cbebd9fe6b43429c3ad24892506d9a : SUCCESS in 1h 36m 45s (non-voting)\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/9d69b94f84b54eac8a7bb9138aecc899 : SUCCESS in 1h 32m 12s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6831ee86e19c4097b12f6b35093bc53f : SUCCESS in 1h 24m 33s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c2b0a9a66b634eaf95322a390671f0e7 : SUCCESS in 1h 30m 42s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/3cd4a37050b24069b733b046ef4899b7 : SUCCESS in 1h 11m 12s (non-voting)\n- kolla-build-rocky10-aarch64 https://zuul.opendev.org/t/openstack/build/81d344c67eea4db1ac89d62c78f1acce : SUCCESS in 1h 42m 07s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/52c6ae75b68d47739562c3721fac2557 : SUCCESS in 1h 35m 25s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"c9db93075469322f1f8880b0902bf3520ccf0054","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2026-03-30 13:02:05.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"c343b3980b1cfce79bf31b64ac336ede1429627d","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-03-30 13:26:56.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"9d75c03f19642800890f0457d63f9f063205ce1f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 13:27:08.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"e7b868c6af9f09be2211cda219f8f06b5ad2ac0c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 15:16:48.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6e0d64e586b4adbbf93ae2515b6dd3c\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/1d5b1b81cdc74ae9ac0079b09e3c1a57 : SUCCESS in 47m 46s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/c2f4d3c57ca34ede9d1782a90c2f1402 : SUCCESS in 31m 50s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/7c20a5e400b74a96846705356e17b667 : SUCCESS in 1h 10m 32s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/0d34fe34a3ee4ee6a80ef7f8e5d2dca2 : SUCCESS in 1h 40m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2a04497c9ce4bc18f9caf91f1d83961 : SUCCESS in 2m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c0d043e9f4f948ad81ab2a15dfc89266 : SUCCESS in 4m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f7606e8e7e704ac4b19bc0a967d3442b : SUCCESS in 3m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61f5c38247fa444a8bd8ad8fd38ae657 : SUCCESS in 8m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0d5ee13992ca46749ad60d4c6a7e3800 : SUCCESS in 2m 08s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/c69128a421534f959f4de718d50b0ca7 : SUCCESS in 3m 13s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/c5abaf7a724c45d885b49742c6f2f1c4 : SUCCESS in 1h 13m 30s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/4f9285cf06da4ced84053e09988e1abe : SUCCESS in 43m 17s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/33a7a2f1af624b858b6114978f46ec68 : SUCCESS in 1h 00m 54s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/66f0a11fa4e04d3f89e57b1123daaf12 : SUCCESS in 1h 09m 06s\n- kolla-build-rocky10 https://zuul.opendev.org/t/openstack/build/ae204598d9324c789aa25d04ef4986ce : SUCCESS in 50m 33s\n- kolla-build-rocky10-podman https://zuul.opendev.org/t/openstack/build/54fe5893378c4d40b400b4001aa1bb50 : SUCCESS in 42m 43s\n- kolla-ansible-rocky10 https://zuul.opendev.org/t/openstack/build/6dbb58e4244147b6ab09bb0c3e80e098 : SUCCESS in 59m 58s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/6ee3b992868240b3ad72d993cad338f4 : SUCCESS in 45m 06s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/76082786976c4ccaa7212942f887083f : SUCCESS in 46m 54s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/46d8af9e594348ffbef2deb0a13c6570 : SUCCESS in 1h 15m 02s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/739cc5d96fd64232a015bce17f0da4bf : SUCCESS in 1h 48m 58s","accounts_in_message":[],"_revision_number":10},{"id":"19c4ea1d30962a3ca702dc308c22c8920db2431a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 15:16:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"dbaa501121ce3d1c6ca47bbec22d95fc48d9f45d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 15:17:47.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/426ce1c26ce24f028468db0f059c84e4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38eeaca3e8cb4330b89f3305dd737e62 : SUCCESS in 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1b60418c88424335a9120de127f5b039 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"bc638acab491f6f4e0c46a958ccc43fa8982ff4d","revisions":{"7b83f6514576f8a95ef32be7cebb446d40154d16":{"kind":"REWORK","_number":1,"created":"2026-03-18 11:33:54.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/96/981096/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/96/981096/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/981096/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/981096/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/981096/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/96/981096/1"}}},"commit":{"parents":[{"commit":"7b4dddce681980af06d67d54d900d56e72ea4937","subject":"debian: use valkey from bookworm-backports","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7b4dddce681980af06d67d54d900d56e72ea4937"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-03-16 13:39:16.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-03-18 11:33:45.000000000","tz":0},"subject":"Install iptables-nft and iptables-legacy explicitly","message":"Install iptables-nft and iptables-legacy explicitly\n\nWe need both packages to be installed so that we can support both\nnf_tables and iptables based rules. The behaviour is switchable with the\nKOLLA_LEGACY_IPTABLES environment variable.\n\nThere was a regression in 2025.1 where we switched back to using\niptables-legacy. Explicitly install iptables-nft restores the\nold behaviour.\n\nThe iptables metapackage doesn\u0027t install iptables-nft if iptables-legacy\nhas already been installed previously.\n\nCloses-Bug: #2144562\nChange-Id: I2bba43407a10edb283d8605d579667cb8d8e3126\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7b83f6514576f8a95ef32be7cebb446d40154d16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7b83f6514576f8a95ef32be7cebb446d40154d16"}]},"branch":"refs/heads/stable/2025.1"},"7602e7870be831f6e92a2ac7e4e385b081f31001":{"kind":"REWORK","_number":2,"created":"2026-03-18 11:53:54.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/96/981096/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/96/981096/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/981096/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/981096/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/981096/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/96/981096/2"}}},"commit":{"parents":[{"commit":"7b4dddce681980af06d67d54d900d56e72ea4937","subject":"debian: use valkey from bookworm-backports","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7b4dddce681980af06d67d54d900d56e72ea4937"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-03-16 13:39:16.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-03-18 11:53:44.000000000","tz":0},"subject":"Install iptables-nft in Rocky 9 containers","message":"Install iptables-nft in Rocky 9 containers\n\nThere was a regression in 2025.1 where we switched back to using\niptables-legacy. Explicitly installing iptables-nft restores the\nold behaviour.\n\nCloses-Bug: #2144562\nChange-Id: I2bba43407a10edb283d8605d579667cb8d8e3126\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7602e7870be831f6e92a2ac7e4e385b081f31001"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7602e7870be831f6e92a2ac7e4e385b081f31001"}]},"branch":"refs/heads/stable/2025.1"},"0e50fc53406db1b851b5cb12dbbb6162d5c37370":{"kind":"REWORK","_number":3,"created":"2026-03-18 13:48:44.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/96/981096/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/96/981096/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/981096/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/981096/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/981096/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/96/981096/3"}}},"commit":{"parents":[{"commit":"7b4dddce681980af06d67d54d900d56e72ea4937","subject":"debian: use valkey from bookworm-backports","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7b4dddce681980af06d67d54d900d56e72ea4937"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-03-16 13:39:16.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-03-18 13:48:17.000000000","tz":0},"subject":"Install iptables-nft in Rocky 9 containers","message":"Install iptables-nft in Rocky 9 containers\n\nThere was a regression in 2025.1 where we switched back to using\niptables-legacy. Explicitly installing iptables-nft restores the\nold behaviour.\n\nCloses-Bug: #2144562\nChange-Id: I2bba43407a10edb283d8605d579667cb8d8e3126\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0e50fc53406db1b851b5cb12dbbb6162d5c37370"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0e50fc53406db1b851b5cb12dbbb6162d5c37370"}]},"branch":"refs/heads/stable/2025.1"},"995416869f4074cb4352e1753b158e5e12a1de5b":{"kind":"REWORK","_number":4,"created":"2026-03-18 13:56:20.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/96/981096/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/96/981096/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/981096/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/981096/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/981096/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/96/981096/4"}}},"commit":{"parents":[{"commit":"7b4dddce681980af06d67d54d900d56e72ea4937","subject":"debian: use valkey from bookworm-backports","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7b4dddce681980af06d67d54d900d56e72ea4937"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-03-16 13:39:16.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-03-18 13:56:09.000000000","tz":0},"subject":"Install iptables-nft in Rocky 9 containers","message":"Install iptables-nft in Rocky 9 containers\n\nThere was a regression in 2025.1 where we switched back to using\niptables-legacy. Explicitly installing iptables-nft restores the\nold behaviour.\n\nCloses-Bug: #2144562\nChange-Id: I2bba43407a10edb283d8605d579667cb8d8e3126\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/995416869f4074cb4352e1753b158e5e12a1de5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/995416869f4074cb4352e1753b158e5e12a1de5b"}]},"branch":"refs/heads/stable/2025.1"},"f1e405d16843992085dae06d31faba1658aec2e9":{"kind":"REWORK","_number":5,"created":"2026-03-18 14:52:01.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/96/981096/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/96/981096/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/981096/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/981096/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/981096/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/96/981096/5"}}},"commit":{"parents":[{"commit":"7b4dddce681980af06d67d54d900d56e72ea4937","subject":"debian: use valkey from bookworm-backports","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7b4dddce681980af06d67d54d900d56e72ea4937"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-03-16 13:39:16.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-03-18 14:51:37.000000000","tz":0},"subject":"Install iptables-nft in Rocky 9 containers","message":"Install iptables-nft in Rocky 9 containers\n\nThere was a regression in 2025.1 where we switched back to using\niptables-legacy. Explicitly installing iptables-nft restores the\nold behaviour.\n\nCloses-Bug: #2144562\nChange-Id: I2bba43407a10edb283d8605d579667cb8d8e3126\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f1e405d16843992085dae06d31faba1658aec2e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f1e405d16843992085dae06d31faba1658aec2e9"}]},"branch":"refs/heads/stable/2025.1"},"568fc43180c81d5b3bfe5176472c0c418b06d736":{"kind":"REWORK","_number":6,"created":"2026-03-19 11:31:26.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/96/981096/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/96/981096/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/981096/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/981096/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/981096/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/96/981096/6"}}},"commit":{"parents":[{"commit":"7b4dddce681980af06d67d54d900d56e72ea4937","subject":"debian: use valkey from bookworm-backports","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7b4dddce681980af06d67d54d900d56e72ea4937"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-03-16 13:39:16.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-03-19 11:31:20.000000000","tz":0},"subject":"Install iptables-nft in Rocky 9 containers","message":"Install iptables-nft in Rocky 9 containers\n\nThere was a regression in 2025.1 where we switched back to using\niptables-legacy. Explicitly installing iptables-nft restores the\nold behaviour.\n\nCloses-Bug: #2144562\nChange-Id: I2bba43407a10edb283d8605d579667cb8d8e3126\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/568fc43180c81d5b3bfe5176472c0c418b06d736"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/568fc43180c81d5b3bfe5176472c0c418b06d736"}]},"branch":"refs/heads/stable/2025.1"},"6e77e8a6828a43277a6c24e8565912b0a7782242":{"kind":"REWORK","_number":7,"created":"2026-03-19 15:34:09.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/96/981096/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/96/981096/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/981096/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/981096/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/981096/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/96/981096/7"}}},"commit":{"parents":[{"commit":"7b4dddce681980af06d67d54d900d56e72ea4937","subject":"debian: use valkey from bookworm-backports","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7b4dddce681980af06d67d54d900d56e72ea4937"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-03-16 13:39:16.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-03-19 15:33:44.000000000","tz":0},"subject":"Install iptables-nft in Rocky 9 containers","message":"Install iptables-nft in Rocky 9 containers\n\nThere was a regression in 2025.1 where we switched back to using\niptables-legacy. Explicitly installing iptables-nft restores the\nold behaviour.\n\nCloses-Bug: #2144562\nChange-Id: I2bba43407a10edb283d8605d579667cb8d8e3126\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6e77e8a6828a43277a6c24e8565912b0a7782242"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6e77e8a6828a43277a6c24e8565912b0a7782242"}]},"branch":"refs/heads/stable/2025.1"},"75cef149f65154638f5bad851a071edb7f676e42":{"kind":"REWORK","_number":8,"created":"2026-03-25 16:33:59.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/96/981096/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/96/981096/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/981096/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/981096/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/981096/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/96/981096/8"}}},"commit":{"parents":[{"commit":"7b4dddce681980af06d67d54d900d56e72ea4937","subject":"debian: use valkey from bookworm-backports","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7b4dddce681980af06d67d54d900d56e72ea4937"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-03-16 13:39:16.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-03-25 16:33:52.000000000","tz":0},"subject":"Install iptables-nft in Rocky 9 containers","message":"Install iptables-nft in Rocky 9 containers\n\nThere was a regression in 2025.1 where we switched back to using\niptables-legacy. Explicitly installing iptables-nft restores the\nold behaviour.\n\nCloses-Bug: #2144562\nChange-Id: I2bba43407a10edb283d8605d579667cb8d8e3126\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/75cef149f65154638f5bad851a071edb7f676e42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/75cef149f65154638f5bad851a071edb7f676e42"}]},"branch":"refs/heads/stable/2025.1"},"8f260212fa6d8339219071c4c07e30d9e5a635d6":{"kind":"REWORK","_number":9,"created":"2026-03-26 15:26:37.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/96/981096/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/96/981096/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/981096/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/981096/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/981096/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/96/981096/9"}}},"commit":{"parents":[{"commit":"7b4dddce681980af06d67d54d900d56e72ea4937","subject":"debian: use valkey from bookworm-backports","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7b4dddce681980af06d67d54d900d56e72ea4937"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-03-16 13:39:16.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-03-26 15:26:32.000000000","tz":0},"subject":"Install iptables-nft in Rocky 9 containers","message":"Install iptables-nft in Rocky 9 containers\n\nThere was a regression in 2025.1 where we switched back to using\niptables-legacy. Explicitly installing iptables-nft restores the\nold behaviour.\n\nCloses-Bug: #2144562\nChange-Id: I2bba43407a10edb283d8605d579667cb8d8e3126\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f260212fa6d8339219071c4c07e30d9e5a635d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f260212fa6d8339219071c4c07e30d9e5a635d6"}]},"branch":"refs/heads/stable/2025.1"},"bc638acab491f6f4e0c46a958ccc43fa8982ff4d":{"kind":"REWORK","_number":10,"created":"2026-03-26 18:29:08.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/96/981096/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/96/981096/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/981096/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/981096/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/981096/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/96/981096/10"}}},"commit":{"parents":[{"commit":"7b4dddce681980af06d67d54d900d56e72ea4937","subject":"debian: use valkey from bookworm-backports","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7b4dddce681980af06d67d54d900d56e72ea4937"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-03-16 13:39:16.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-03-26 18:29:08.000000000","tz":0},"subject":"Install iptables-nft in Rocky 9 containers","message":"Install iptables-nft in Rocky 9 containers\n\nThere was a regression in 2025.1 where we switched back to using\niptables-legacy. Explicitly installing iptables-nft restores the\nold behaviour.\n\nCloses-Bug: #2144562\nChange-Id: I2bba43407a10edb283d8605d579667cb8d8e3126\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bc638acab491f6f4e0c46a958ccc43fa8982ff4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bc638acab491f6f4e0c46a958ccc43fa8982ff4d"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
