)]}'
{"id":"openstack%2Fkolla~981123","triplet_id":"openstack%2Fkolla~master~I9b6eba0c55fa9914e8db3576127f7d53c13e97d5","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"37203":{"account":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"last_update":"2026-03-26 17:11:58.000000000","reason":"Change was submitted"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2026-03-26 11:04:59.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-03-26 17:11:58.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-03-26 13:14:31.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2026-03-26 17:11:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9b6eba0c55fa9914e8db3576127f7d53c13e97d5","subject":"Switch to Gazpacho/2026.1 sources","status":"MERGED","created":"2026-03-18 13:56:46.000000000","updated":"2026-03-26 18:44:40.000000000","submitted":"2026-03-26 17:11:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"981123","meta_rev_id":"ee26931e2635a99b2fc92e695f53ef428aedcbea","_number":981123,"virtual_id_number":981123,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-26 17:11:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},{"value":2,"date":"2026-03-26 11:04:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2026-03-26 10:19:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2026-03-26 13:14:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2026-03-26 10:19:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},{"value":-1,"date":"2026-03-26 11:04:59.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":-1,"date":"2026-03-26 10:19:13.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"recommended":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},{"value":1,"date":"2026-03-26 11:04:59.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2026-03-26 10:19:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-18 14:29:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-25 14:11:44.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2026-03-25 14:17:09.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-03-26 09:20:18.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2026-03-26 09:37:38.000000000","updated_by":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"reviewer":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"state":"REVIEWER"}],"messages":[{"id":"33965796dd93cb2b83e5ae676cdd2fd40a467321","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-03-18 13:56:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"59545b4dc166d0e0c09dbc60209748263b001a8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 14:29:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/460b55997621402e9f68f36a96808ebb\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/526a6fc51da4413e978f1854c70dc1e2 : FAILURE in 19m 41s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/44b2f3e4d94948628b2f411e704c1a88 : FAILURE in 14m 51s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/235a41feeaa94575a03fe02dda28938c : FAILURE in 12m 53s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/0d8a25c3dc2249838d60d8245c53d1cf : FAILURE in 13m 00s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/80f776097705489dbad81236aaa9c20e : FAILURE in 17m 09s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/64b45355c2d84a8bb0150da486aca1e5 : FAILURE in 18m 34s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f65ec6e7471a415bbf5376024df5491c : FAILURE in 14m 41s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/486529c2e5a444a086cf6d54fd920082 : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f63297df970045508658511f23cfd5cb : SUCCESS in 4m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0fd5f24df4c14b578b4f59e9dac293a8 : SUCCESS in 3m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/31e7c388a4a7469686717bddbbdb0b18 : SUCCESS in 3m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8f28adc080664eda85b4c81f5f396424 : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/87c498e31f4f4d95ba6d9716c23563b3 : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84863ebb1f27456d9372d7e591ff90e7 : SUCCESS in 7m 46s","accounts_in_message":[],"_revision_number":1},{"id":"bc6efa804cf3003cb019066b90d09ae5e5f5077d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 19:22:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60ed2ee063ca4e8db8124fa29ebe8844\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/7d6057d961224a56b3f8193224e9fecf : FAILURE in 30m 45s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/1b7ca0274de2429d9aa680424777b92f : FAILURE in 22m 34s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/2040ebfd584f4ae78af84de4b20e8ff0 : FAILURE in 26m 45s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/736edb6c297f4e909c40f53593af82b2 : FAILURE in 32m 27s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4aa2595c45134933b30bcad1a0e73f75 : SUCCESS in 4m 15s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1cf9e39dc7374cf7a85411acb6bfdae8 : SUCCESS in 4m 54s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/1f1e0f1d036a488ca41455d0fdeaea6c : SUCCESS in 6m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6c3c75ec626369eb31e64ba7e941561d5cee46e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-03-25 14:10:15.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"811cfb8aa698c64f5223373c57b9d739e3742945","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-03-25 14:11:44.000000000","message":"Patch Set 2: Backport-Candidate-1 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c5ef74106891d7daf8aa4e700ce61c3180f74ecb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-25 14:34:10.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/84d75d4557d1450f985bdc0c068de52a\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/213df322a2bd4a0ab8ebf63dd82c8a56 : FAILURE in 18m 07s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/eabb7c4432a44ae989bc45a1dca35185 : FAILURE in 10m 12s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/d9e135a2e98c491eb5595d05f1e9dd5c : FAILURE in 12m 12s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/0b911a12d7744e3382b8817f41bc2957 : FAILURE in 18m 23s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/a097ba1ecacb455ebede43280d1e97fe : FAILURE in 15m 42s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0f213963be2248e2a2e68e66de721d9d : FAILURE in 18m 17s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1c6801ead74f4d50aae5b2fede76eeac : FAILURE in 9m 04s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/c1cedfeffff6464890ead751d3c0de51 : SUCCESS in 3m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7562a2622b42428f9596a0cdd046cdab : SUCCESS in 3m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/43a4a113499b466babb665f7284977ab : SUCCESS in 3m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0506f1a67f7f406db7b815690d7275af : SUCCESS in 3m 05s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/cc5f43fe6100452bbffaec469637fd9d : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/38e2a866b16a4c26bcf094b7e43ec9cb : SUCCESS in 2m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/313d0b5fc139489d8460c67d47ab049d : SUCCESS in 6m 21s","accounts_in_message":[],"_revision_number":2},{"id":"9aa15298334ecf383101049b3cce72824355d647","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-03-25 14:46:45.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"40d8b2129fefab31e24d622fc64599b8211e7b23","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-03-25 14:47:31.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":4},{"id":"32c5f25d8d9287b964a1cdf83d869c25d058ac7d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-03-25 14:49:30.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":5},{"id":"acc44eea6f07ede3f955be5fe61434309778777e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-25 15:51:01.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d014bd2bfcbf4279942a3b75f663a8ff\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/0c74acd4411e4fa7b9f7fa28e6795a91 : FAILURE in 53m 30s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/2aec11d2f0964aecb430831e4527f28e : FAILURE in 47m 30s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/f4fe715ca6f74900a6ee60e22b88e924 : FAILURE in 27m 29s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/7836e3a2d92c45afacb946673e585ace : FAILURE in 39m 01s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/2aa738b3502c4435b43794b8320617c7 : FAILURE in 54m 10s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/002e6de195a04981982a14a87075a929 : FAILURE in 56m 04s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4d0f8489a9b148d783cc24cd8d34eb6d : FAILURE in 37m 33s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/4a6bb614992c4ce7acf71480d4327e6b : FAILURE in 25m 07s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/84939b8125a94255a11736a0bdab3a9f : FAILURE in 26m 38s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/ba11dab9d4074ff99b154a796afbcc3b : FAILURE in 21m 15s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/ad151238ab6a487eab585e72a5d1658c : FAILURE in 25m 53s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/0fe1047bb04e49acbc86e4eb3a5dfa16 : FAILURE in 27m 19s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/01f677fd45a0445296f8d420fbe68fac : FAILURE in 26m 40s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/83fe278eb46d4567bd270b6fe9a35621 : FAILURE in 25m 03s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/71c4fe9bb60a40aebf07299e4966919b : SUCCESS in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a49c8b7541b4667bd51b0df21f999a3 : SUCCESS in 4m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3fedab86a54f47188514cb0b7b12671e : SUCCESS in 3m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0067d56bec2948c590e886bc66a4ebd1 : SUCCESS in 5m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ae5f1507448f45b7b6ec05a65c97ba4f : SUCCESS in 6m 02s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9ca59764e09d472db264cb8bfeb145ec : SUCCESS in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea0a8bd96b3a4317ac23240fda48b914 : SUCCESS in 6m 54s","accounts_in_message":[],"_revision_number":5},{"id":"2024402e1ec911487d7b43906b852f1e1ddaccc9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-03-25 16:24:56.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"1af30961d970b7fdf54d9568e786496b5b1b7e9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-25 17:35:34.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4f734ce25c5a4ba490b57c6ad277fad4\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/d23ba0c5f5cf4819a2061620a9175ea2 : FAILURE in 44m 54s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/a9fdadebbe9c42e1b84006622dbc2097 : FAILURE in 47m 08s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/9c04402c1d2941959a243bed414aba48 : FAILURE in 26m 33s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/e285aaf6676f45e088ce04c472264548 : FAILURE in 39m 12s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/9350f6a6e8ae4649bf508d9f9419402c : FAILURE in 51m 07s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cb96e79e95094a038b2fcf47b1af1820 : FAILURE in 34m 14s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e35f206970084b37967dd8e6f03c768f : FAILURE in 34m 55s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/d10efbe8cdf744d6976e64cfbc03e0f6 : FAILURE in 28m 16s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/23c03fc17de549eb89e700e6aa16fb74 : FAILURE in 23m 52s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/b675c47e5b864a0a924f9aabfa87093c : FAILURE in 23m 28s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/0d37a3f4b6114377ab59c4f8012f4a9e : FAILURE in 30m 04s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/b4c5d4f9a2bd4c2bab27fd6617fd4d5b : FAILURE in 28m 28s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fe35740c572441778803bf87ad459f53 : FAILURE in 25m 11s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/ab852730421847249ab49aec68328d36 : FAILURE in 1h 08m 40s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/38bfa3220f5b4b89a3c9903487b602f8 : SUCCESS in 3m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2af4c3612ec4dd3aed4394aa9752c42 : SUCCESS in 3m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b200bed43dc44695b473c0bf3f48df1e : SUCCESS in 2m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/51df39ad428d4251bb1156453a1179f6 : SUCCESS in 3m 02s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/324b015f1ed9472a968f38f0de6065f1 : SUCCESS in 7m 29s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4781f47fa75b4253be5208baa6a054f6 : SUCCESS in 3m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/621f94c6fdf047f1b927a409f3166143 : SUCCESS in 7m 43s","accounts_in_message":[],"_revision_number":6},{"id":"94b945b65c361f6757d49e68422d3abdbddced08","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-03-25 19:19:38.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"6d0fa16903229627e03e15682988778304eca670","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-03-25 19:20:27.000000000","message":"Patch Set 7: Published edit on patch set 6.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"1a06c3f5cb90fd4d54904647037df841dac2954c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-25 20:35:38.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a1d757482b5e4b0fb185cd8aa6cb801c\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/f78c3ab2a9e84a5f9a6ddb0982b1fe39 : FAILURE in 51m 05s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/a1405ce8ab17401088ae28fa9c04bd3f : FAILURE in 46m 24s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/065700f0e3574f4ca00fdc0ffa7d35af : FAILURE in 22m 42s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/4fe67bb7eaa7475f80bb5dae4f896065 : FAILURE in 39m 53s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/cf5f1043ca344f32bee27146c7c77294 : FAILURE in 1h 07m 07s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8add42042b9b4c62b61252925589090a : FAILURE in 53m 11s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6c123845fe614e85aaad1d4857225a02 : FAILURE in 4m 54s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/033948b805ad45129fd8146e04cb59f6 : FAILURE in 32m 44s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/cb140fa6463c43adb021dc193dac01bd : FAILURE in 24m 26s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/dcc62b9a3404499fa9d2bd867c28d6d2 : FAILURE in 29m 36s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/3605467ac49241619ee06b2d7e7670bf : FAILURE in 34m 24s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/c9d6b7c1ea774db2a9e9f73ca027223f : FAILURE in 35m 27s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/53bd1bf9ce26481e9e2b1a90ca9903b5 : FAILURE in 35m 56s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/e1a6ddcc9b78430e94ba7d5298d637bc : FAILURE in 34m 38s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/6ffc069f5d5549c79e09dd969a1460b7 : SUCCESS in 3m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c690b133a874e28bef76261ba6eeced : SUCCESS in 4m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eb6e78958b3b4408a1e834477c4862dd : SUCCESS in 3m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/60ba4d58089f4482ad47720e86439fc0 : SUCCESS in 3m 15s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/60caaf0f0e864983ba081301496a044e : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/40ee5f9314ff4d048ee72e7ec409663c : SUCCESS in 3m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c5eb85997834572b5890446ad6a510d : SUCCESS in 7m 32s","accounts_in_message":[],"_revision_number":7},{"id":"538af8fab36661f8375204faf9aea69172fe4c1a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-03-26 07:54:52.000000000","message":"Patch Set 8: Published edit on patch set 7.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"1837b01dc9372f3a710652a902659d636b9fa176","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-03-26 09:17:38.000000000","message":"Patch Set 8: Review-Priority+2","accounts_in_message":[],"_revision_number":8},{"id":"0bc98aedfc7c1cecb97cb415edf603b1a380bf19","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-03-26 09:20:18.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"1752bf3a17ac654aae248ee59eb6c1fee8ebfb81","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-03-26 09:33:05.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"b284aecf31128246e0bf1dde2849aaeb7fed1a82","author":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"date":"2026-03-26 09:37:38.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"103b3849c6b419234a54ce4c8acc752f5354388a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 09:43:37.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/959bc42a960b4c6f9b1246c28d28a561\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/0418459525354948b70feabcf50489d1 : SUCCESS in 40m 56s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/d2b20df9fc4245e499bd11a3c3972cdd : SUCCESS in 37m 55s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/8bb20bdd3bc64bcfb7e33a6e5078516c : SUCCESS in 24m 05s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/c73dabecb4c248b6a883399c25aa279a : SUCCESS in 37m 52s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/15cb608570284c029dacb1a05b93edbd : SUCCESS in 55m 38s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e2d0b52c16bb40069b881c0f8bd8a08a : SUCCESS in 54m 44s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fedf93b0038b4477860f7406db9f0296 : SUCCESS in 24m 40s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/43d8386914904358ac4bec67dd819566 : SUCCESS in 1h 09m 48s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/624dd63a37ba4f2abe651566d2f04f41 : SUCCESS in 1h 11m 25s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/50a4fbdbb24c4933bdc9f26dbb60c995 : SUCCESS in 1h 33m 11s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/2e7fc24d32cd42eabdde232408a2fe06 : SUCCESS in 1h 17m 57s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/f042f9a9de304381bec6c1b944e54e90 : SUCCESS in 1h 47m 47s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/42b1c22df80f45ec890c59584b4a9d73 : SUCCESS in 1h 14m 12s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/049cf656eac04bc28e2c93fe1f8e2630 : SUCCESS in 1h 43m 45s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/fbfd22ff538048f381eba7b23a6f69c8 : SUCCESS in 3m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec42a3ff21dc42f198082b49b2ff8292 : SUCCESS in 2m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5451fea0a20a42eb9a4ae352fd601f2b : SUCCESS in 2m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/422ef4ee849245749b0e62316e6b6028 : SUCCESS in 3m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5f7487f39d6942e0be8ec98e788c78ee : SUCCESS in 4m 04s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d36d00eba6954eda85d16add2f99fb52 : SUCCESS in 3m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/413e5703aa3c4d288eb6c6de32bdfc3b : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":8},{"id":"108a8c6e607dcd82f28f8886cb785d922a539e32","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-03-26 09:45:15.000000000","message":"Patch Set 8: Backport-Candidate-1 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"ac599827cada6de195b9ebc36e48cefb1a024c7f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 09:45:29.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"7bec95f9b6dfdb6a4a548349240635e5daf81f05","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-03-26 10:14:24.000000000","message":"Patch Set 8: Backport-Candidate-1 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"f3e6b77c6bf970447f3090f8ea860e0a39038b50","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-03-26 10:19:08.000000000","message":"Patch Set 9: Published edit on patch set 8.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1, Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Workflow+1\n","accounts_in_message":[],"_revision_number":9},{"id":"d888d0162f125aaad3d8fe62922db079f5d9c4d2","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-03-26 10:19:13.000000000","message":"Patch Set 9: Backport-Candidate-1 Code-Review+2 Review-Priority+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"25c09580e7655eff10b1e740fce4052305d201a1","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-03-26 11:04:59.000000000","message":"Patch Set 9: Backport-Candidate-1 Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"7ba7a8cd363087421e741fe59aced1b396fc902f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 11:08:07.000000000","message":"Patch Set 9:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"1775ddd33f953ed9afede3d14b95cd2c61af2c73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 12:24:23.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9f9ab12fb28d44ffad29adf505f3c806\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/46ae916bca7a497ca423f7d5d002b703 : SUCCESS in 44m 21s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/33cac1beefc846eb8cf16b1cf438fcce : SUCCESS in 1h 06m 55s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/228cb62505d54dd8880af87c691100a6 : SUCCESS in 27m 52s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/a8929ffe140742e88f7422ec66db3eb1 : SUCCESS in 38m 39s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/897bfc9be507405b820b647bb3d3c028 : SUCCESS in 1h 06m 38s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8d430d55e3424f1391a9e13fd7bc7deb : SUCCESS in 1h 08m 36s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9d5c4654f4e74aa685960a9b8f333d0e : SUCCESS in 35m 21s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/6aacc2897b89433e9d08bc576c37c082 : SUCCESS in 1h 15m 35s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/3318eccfcaf44245b39f75104f045730 : SUCCESS in 1h 13m 42s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/208b75c3716c42cfa461661fd6d5a31c : SUCCESS in 1h 39m 24s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/453e4ca020f3443197c6658f4effc6f6 : SUCCESS in 1h 23m 12s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/ac09d24ddea54074b1f4a103ca3b748f : SUCCESS in 1h 51m 00s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/987a979f82344982bb30066323ecd225 : SUCCESS in 1h 20m 40s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/4f8566806df74a988fc42b55de038bea : FAILURE in 34m 17s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/ee9a9c22d6a0482caaa78c9f28eb24e6 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72836164c9e7486b945e8a783b3e1dd8 : SUCCESS in 3m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b11325270e454182939b826166c86c3c : SUCCESS in 3m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e2bb1b81c36f4bb597744e60fb9dab4d : SUCCESS in 3m 50s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/35dfc5b22e70438eb52c2593fb6b94c2 : SUCCESS in 5m 58s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1121eb58fa4f42b0936153845ce9cc20 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea01fb6b34f34f96887643ff836b4232 : SUCCESS in 6m 33s","accounts_in_message":[],"_revision_number":9},{"id":"eb4765186a0473952757da27d2c265f6c1e457c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 13:08:03.000000000","message":"Patch Set 9: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8825a3857b7e4dc19b0e4d92c3ea4e0e\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/88acbcd8411f4593bdd5e4017c2151d6 : SUCCESS in 37m 42s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/fea7a6073fc2454f9c8ae8fac90b9a72 : SUCCESS in 24m 29s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/bb39be72dff04338b118b8ddad2426ba : SUCCESS in 48m 43s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/61f55f33808a40e1bc0a648c4937575e : SUCCESS in 51m 17s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/418b886a79764e5f8cf1f1912be4ac55 : SUCCESS in 52m 55s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/aff37c0337574395b9ca26ea1fa46d1b : SUCCESS in 27m 45s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/c04c6ea9ca0e4bd480892333a280668b : SUCCESS in 1h 09m 28s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/d3cc1738c55641cd87b10ae5019d3acf : SUCCESS in 1h 34m 37s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/80147f19628e4f7aaa09a6f12058b7a1 : SUCCESS in 1h 21m 02s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/dc43472266a247f18e969df4f25eb5e7 : FAILURE in 1h 59m 38s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6b94d38bd52a44e083cbdc3816dba720 : SUCCESS in 1h 20m 04s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/dfb26649a2e04f1f92949726e4a15ae8 : SUCCESS in 1h 47m 34s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/5a90b055b17247f1ad4983f3b225f725 : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9597764e0e8348dc87983285657a61ed : SUCCESS in 4m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/82e0143dae1d46d1b789621bec0c55fb : SUCCESS in 3m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2be47db070e646648c8f0ce7b865c4f5 : SUCCESS in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5862c82b4464edb983f77600ec2cfb9 : SUCCESS in 6m 08s","accounts_in_message":[],"_revision_number":9},{"id":"70e488b42fa62c34dc4a930fefd9e87e9dbf2129","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-03-26 13:08:15.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"51024ea154ddbccb8db5e753058dc6a27b712878","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-03-26 13:14:31.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"45fd8730a56037e18a34e0728972834abde9a5b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 15:09:40.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c816ef52e1c8475a9c40a79eb6b5512f\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/173ba1160729448bbf7edbb095ac08b8 : SUCCESS in 52m 23s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/15fd44d6c66a45c1a0b4dded293db946 : SUCCESS in 53m 52s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/ddbd71f661e44d58a2ae133d415eb49a : SUCCESS in 12m 53s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/2d97ed8968eb4500a9342bc3bf53698d : SUCCESS in 40m 10s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/1c5c60695fbb4fcda027bafee116919c : SUCCESS in 37m 19s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2ecf05257ea7459086e3a2478e1681bd : SUCCESS in 34m 44s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5dba0305a60940e08858bbe9369521f5 : SUCCESS in 15m 02s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/c2cd028f952b459ba56725d3e466ddd6 : SUCCESS in 1h 24m 58s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/3afebbb1b35b43bb9cd8735b0c12a866 : SUCCESS in 1h 28m 41s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/d28e2142836b4fcba4423826791f82f5 : SUCCESS in 1h 43m 25s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/44f6a32c3e81437290ca2ad7fd30a38e : SUCCESS in 1h 31m 17s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/143bf4fd156945feb23f4c0a5a7cb809 : SUCCESS in 2h 00m 19s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/983fcd6c3c1c40428086e34f9e4c0231 : SUCCESS in 1h 27m 31s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/90c2ad2db492458f956ff983a2e55f09 : SUCCESS in 1h 59m 20s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/f787ae01886740aca096b8ff2114bbc8 : SUCCESS in 2m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/effffdd5773c4f92a7cae24ac07757d2 : SUCCESS in 4m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5351bc00b3424daa8c1008600047f535 : SUCCESS in 4m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/91c43b807d4e4f6c9a99aca492e2aed9 : SUCCESS in 3m 24s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/91200feb6ce44a1da35f2a58bd9f5384 : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c2ba324d2a44488bad7e2d5ad56795b6 : SUCCESS in 3m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cf3b9f88268440a9b5a3cfaefd12057 : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":9},{"id":"eb034f3bd8bdd63270f0548bf87f6dcd98bbbf1b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 15:10:17.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"c443db577171a36f466372c1247fa1088fa9f280","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 17:11:57.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d40107f4fab0417dabec1a045b0c55a1\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/f77f88f62b034cab8898df8aded1c063 : SUCCESS in 47m 19s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/30683054b68942b8aafb72a96c8c8279 : SUCCESS in 26m 07s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/56552db989fd442386303b0d62202326 : SUCCESS in 47m 16s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/680f1639d7394c0ab923cb95bc99a560 : SUCCESS in 51m 36s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a1e02402d7cf436990a33d19e97c2585 : SUCCESS in 1h 28m 27s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/628886ee861b40b5b8cf21dbc88bf684 : SUCCESS in 32m 11s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/153fd3427dcb45e49091bc58d58354f6 : SUCCESS in 1h 25m 12s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/945ccd67148046b9beb463205d6ed289 : SUCCESS in 1h 52m 11s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/8e8809b5b84240e8a2895c5eaa1a67f3 : SUCCESS in 1h 31m 25s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/63e9234a385845c7b34219017f7d235a : SUCCESS in 2h 00m 38s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7c1c22e031694d9881d0f74c68768420 : SUCCESS in 1h 27m 30s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/5b37d30a69e64d7f8bea5a173e773b32 : SUCCESS in 1h 58m 05s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/742e62f374ae45188b8e7ef11c2b01fd : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/228b9b456b7742c49215123cb91acf62 : SUCCESS in 3m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/597828f9cc4e4788824f6a76764c5338 : SUCCESS in 2m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/15dba88aac444a0ab3fd2f5cbac572a6 : SUCCESS in 2m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3cee694a3494da68788052460e38bd5 : SUCCESS in 6m 31s","accounts_in_message":[],"_revision_number":9},{"id":"d967b4e416bc643914a5805f2359d445d9079ab2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 17:11:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"c25487db0255b887e6840cdd1545dfb4952e1083","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 17:13:01.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3ec4e73f0e443a98903abeed12438ae\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61770a2366364268ac517352567e4ca5 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":9},{"id":"ee26931e2635a99b2fc92e695f53ef428aedcbea","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 18:44:40.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/776af217381745b5b3c11f63e75fa0d7\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/7b12e887be9b4a29b5d76173140fed6b : SUCCESS in 54m 40s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/7c6a76ec0c044b679317eb5ba68d91ce : SUCCESS in 1h 05m 03s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/ebf11155553a46d2a8df3f56ac12c8af : SUCCESS in 1h 02m 27s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/9f792218b5704be493d85de25d602c6e : SUCCESS in 1h 00m 29s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/9eac18219edf40b5bd50c82ff9fa15fb : FAILURE in 1h 13m 59s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/cbc2675070b645baac0b5851e28ec651 : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c8d7f2325d664a0b8f8e6a59216236e7 : SUCCESS in 5m 40s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/05e085e9933b4c1bb221a519273f0a45 : SUCCESS in 7m 05s (non-voting)","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"352b7fce8d71de404ee53019144b9ab15e8bf539","revisions":{"6306487bdfcdc6c26661005e32f59c6e729a8166":{"kind":"REWORK","_number":1,"created":"2026-03-18 13:56:46.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/23/981123/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/981123/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/981123/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/981123/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/981123/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/981123/1"}}},"commit":{"parents":[{"commit":"acda933879243491e77dd9238bc47633126bd0bc","subject":"Merge \"Update Prometheus to version 3.5.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/acda933879243491e77dd9238bc47633126bd0bc"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-03-18 13:56:27.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-03-18 13:56:43.000000000","tz":60},"subject":"Switch to Gazpacho/2026.1 sources","message":"Switch to Gazpacho/2026.1 sources\n\nChange-Id: I9b6eba0c55fa9914e8db3576127f7d53c13e97d5\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6306487bdfcdc6c26661005e32f59c6e729a8166"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6306487bdfcdc6c26661005e32f59c6e729a8166"}]},"branch":"refs/heads/master"},"93443a3084de4576a73988b273620af8f974765a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-03-25 14:10:15.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/23/981123/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/981123/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/981123/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/981123/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/981123/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/981123/2"}}},"commit":{"parents":[{"commit":"4dd97a9d56f31ed3e109496e96e208c75d82d818","subject":"Merge \"Update component list and versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4dd97a9d56f31ed3e109496e96e208c75d82d818"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-03-18 13:56:27.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-03-25 14:10:15.000000000","tz":0},"subject":"Switch to Gazpacho/2026.1 sources","message":"Switch to Gazpacho/2026.1 sources\n\nChange-Id: I9b6eba0c55fa9914e8db3576127f7d53c13e97d5\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/93443a3084de4576a73988b273620af8f974765a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/93443a3084de4576a73988b273620af8f974765a"}]},"branch":"refs/heads/master","description":"Rebase"},"799b41a8b674cd06984c7a7ad34e7b466663f765":{"kind":"REWORK","_number":3,"created":"2026-03-25 14:46:45.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/23/981123/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/981123/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/981123/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/981123/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/981123/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/981123/3"}}},"commit":{"parents":[{"commit":"4dd97a9d56f31ed3e109496e96e208c75d82d818","subject":"Merge \"Update component list and versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4dd97a9d56f31ed3e109496e96e208c75d82d818"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-03-18 13:56:27.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-03-25 14:46:41.000000000","tz":0},"subject":"Switch to Gazpacho/2026.1 sources","message":"Switch to Gazpacho/2026.1 sources\n\nChange-Id: I9b6eba0c55fa9914e8db3576127f7d53c13e97d5\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/799b41a8b674cd06984c7a7ad34e7b466663f765"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/799b41a8b674cd06984c7a7ad34e7b466663f765"}]},"branch":"refs/heads/master"},"96368b273ccc539dd585f0ffd0571ce0f62d0651":{"kind":"REWORK","_number":4,"created":"2026-03-25 14:47:31.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/23/981123/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/981123/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/981123/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/981123/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/981123/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/981123/4"}}},"commit":{"parents":[{"commit":"4dd97a9d56f31ed3e109496e96e208c75d82d818","subject":"Merge \"Update component list and versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4dd97a9d56f31ed3e109496e96e208c75d82d818"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-03-18 13:56:27.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-03-25 14:47:31.000000000","tz":0},"subject":"Switch to Gazpacho/2026.1 sources","message":"Switch to Gazpacho/2026.1 sources\n\nChange-Id: I9b6eba0c55fa9914e8db3576127f7d53c13e97d5\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/96368b273ccc539dd585f0ffd0571ce0f62d0651"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/96368b273ccc539dd585f0ffd0571ce0f62d0651"}]},"branch":"refs/heads/master"},"baff681900c43e22892cef28571d2750875f18f7":{"kind":"REWORK","_number":5,"created":"2026-03-25 14:49:30.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/23/981123/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/981123/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/981123/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/981123/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/981123/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/981123/5"}}},"commit":{"parents":[{"commit":"4dd97a9d56f31ed3e109496e96e208c75d82d818","subject":"Merge \"Update component list and versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4dd97a9d56f31ed3e109496e96e208c75d82d818"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-03-18 13:56:27.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-03-25 14:49:30.000000000","tz":0},"subject":"Switch to Gazpacho/2026.1 sources","message":"Switch to Gazpacho/2026.1 sources\n\nChange-Id: I9b6eba0c55fa9914e8db3576127f7d53c13e97d5\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/baff681900c43e22892cef28571d2750875f18f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/baff681900c43e22892cef28571d2750875f18f7"}]},"branch":"refs/heads/master"},"07b8cdc28836674523f58276fded0beb07934f37":{"kind":"REWORK","_number":6,"created":"2026-03-25 16:24:56.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/23/981123/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/981123/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/981123/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/981123/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/981123/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/981123/6"}}},"commit":{"parents":[{"commit":"4dd97a9d56f31ed3e109496e96e208c75d82d818","subject":"Merge \"Update component list and versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4dd97a9d56f31ed3e109496e96e208c75d82d818"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-03-18 13:56:27.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-03-25 16:24:56.000000000","tz":0},"subject":"Switch to Gazpacho/2026.1 sources","message":"Switch to Gazpacho/2026.1 sources\n\nChange-Id: I9b6eba0c55fa9914e8db3576127f7d53c13e97d5\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/07b8cdc28836674523f58276fded0beb07934f37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/07b8cdc28836674523f58276fded0beb07934f37"}]},"branch":"refs/heads/master"},"6819a294a5ddf9ec9f21d1f905e69e329bd26cc0":{"kind":"REWORK","_number":7,"created":"2026-03-25 19:20:27.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/23/981123/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/981123/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/981123/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/981123/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/981123/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/981123/7"}}},"commit":{"parents":[{"commit":"4dd97a9d56f31ed3e109496e96e208c75d82d818","subject":"Merge \"Update component list and versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4dd97a9d56f31ed3e109496e96e208c75d82d818"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-03-18 13:56:27.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-03-25 19:20:27.000000000","tz":0},"subject":"Switch to Gazpacho/2026.1 sources","message":"Switch to Gazpacho/2026.1 sources\n\nChange-Id: I9b6eba0c55fa9914e8db3576127f7d53c13e97d5\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6819a294a5ddf9ec9f21d1f905e69e329bd26cc0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6819a294a5ddf9ec9f21d1f905e69e329bd26cc0"}]},"branch":"refs/heads/master"},"790e44867107ea9af8412130f1af25888c7ba48b":{"kind":"REWORK","_number":8,"created":"2026-03-26 07:54:52.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/23/981123/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/981123/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/981123/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/981123/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/981123/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/981123/8"}}},"commit":{"parents":[{"commit":"4dd97a9d56f31ed3e109496e96e208c75d82d818","subject":"Merge \"Update component list and versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4dd97a9d56f31ed3e109496e96e208c75d82d818"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-03-18 13:56:27.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-03-26 07:54:52.000000000","tz":0},"subject":"Switch to Gazpacho/2026.1 sources","message":"Switch to Gazpacho/2026.1 sources\n\nChange-Id: I9b6eba0c55fa9914e8db3576127f7d53c13e97d5\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/790e44867107ea9af8412130f1af25888c7ba48b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/790e44867107ea9af8412130f1af25888c7ba48b"}]},"branch":"refs/heads/master"},"352b7fce8d71de404ee53019144b9ab15e8bf539":{"kind":"REWORK","_number":9,"created":"2026-03-26 10:19:08.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/23/981123/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/981123/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/981123/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/981123/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/981123/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/981123/9"}}},"commit":{"parents":[{"commit":"4dd97a9d56f31ed3e109496e96e208c75d82d818","subject":"Merge \"Update component list and versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4dd97a9d56f31ed3e109496e96e208c75d82d818"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-03-18 13:56:27.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-03-26 10:19:07.000000000","tz":0},"subject":"Switch to Gazpacho/2026.1 sources","message":"Switch to Gazpacho/2026.1 sources\n\nChange-Id: I9b6eba0c55fa9914e8db3576127f7d53c13e97d5\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/352b7fce8d71de404ee53019144b9ab15e8bf539"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/352b7fce8d71de404ee53019144b9ab15e8bf539"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-1"],"failing_atoms":["label:Backport-Candidate\u003d1"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
