)]}'
{"id":"openstack%2Fkolla~981168","triplet_id":"openstack%2Fkolla~master~I1f6694fe2948b82cdf6089c699fbcca08291958b","project":"openstack/kolla","branch":"master","attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-03-23 21:11:59.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I1f6694fe2948b82cdf6089c699fbcca08291958b","subject":"Allow KOLLA_LEGACY_IPTABLES on CentOS/Redhat","status":"NEW","created":"2026-03-18 17:53:32.000000000","updated":"2026-06-08 04:00:35.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2338cd3fb8efebcc2881ed608e6faf961833d156","_number":981168,"virtual_id_number":981168,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-06-07 22:29:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-18 19:32:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"992f5928fbe677686b9c65e0d44ccb99ab469402","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-18 17:53:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f3570d06a34e62b5d4c8dfec1790279fdf5d53b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 19:32:44.000000000","message":"Patch Set 1: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2b86416aca2d485a95be388beb489827\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/62092a79699e49a29540f180d7ae1bdb : SUCCESS in 38m 00s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/1beb84199ff242c6a37589752dc76fb1 : SUCCESS in 45m 39s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/6812ce7da2a4479682f66695d3d589bc : SUCCESS in 32m 19s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/c0b7d8395ca648b6b43a678b23adfa83 : SUCCESS in 37m 10s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/298230b8d02240a7b6d6653d8b661e53 : SUCCESS in 51m 01s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0e8c67859a304e4a93488708974ad3f8 : SUCCESS in 50m 47s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b1f70924f8ba4654ad5603aede47de10 : SUCCESS in 31m 32s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/24e04a3420e845d284b687a9d2521c5e : FAILURE in 1h 04m 11s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/d29fd86ffc804ef9a74233e0489aee10 : SUCCESS in 1h 09m 14s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/6899b03d7ecc405fa6fc2c0b57646fd6 : SUCCESS in 1h 30m 48s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/c83f6298d5fa48bab60d38d449c6bae0 : SUCCESS in 1h 12m 53s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/68668aabb1934161bd75fa9130008af8 : SUCCESS in 1h 37m 36s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/36010029f2b9400ea375284d9c496356 : SUCCESS in 1h 10m 55s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/7b925bb57dea46ebace1b2dde3f51a26 : SUCCESS in 1h 20m 37s\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/54fcc0f7383f4c9daee7c41c0511e077 : SUCCESS in 1h 13m 20s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/3080e5361392433ba012ffcdd948a641 : SUCCESS in 1h 21m 36s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/d60a9f8db08e4e3c8a1f64c23be84462 : SUCCESS in 1h 26m 14s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/29b68b9656be4c94b085c196ad9caf9f : SUCCESS in 2m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/613d70a5a1a442a381d9ad69db5d3948 : FAILURE in 3m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/74c5222be74949c08c206a70f3a29290 : SUCCESS in 3m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cd0765b6b57e49959b251086705bac66 : SUCCESS in 4m 41s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4c6c53a9927e463ea93ce9fdb1f18d41 : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d73a3b7981a54bd3bedf3926fe46174f : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b92cb2f23fb43ca9af6ac88134d27bb : SUCCESS in 6m 54s","accounts_in_message":[],"_revision_number":1},{"id":"261694f01b313b61361b34bc8a437941cf0e1027","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 22:33:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa2660c08c474a5a982607b81cf0cf18\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/5ea00a87083c4a878b712b1cbf39d7cb : SUCCESS in 59m 44s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/7c14691f6e54408dae9109094b8d0dc1 : SUCCESS in 59m 58s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/94575e5d2a6040deb9fa39aa666e862c : SUCCESS in 1h 24m 10s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/7b30bc183ce84b5a900694df6160a706 : SUCCESS in 1h 18m 08s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/21a411abe2864627b49fd5349395043e : FAILURE in 1h 39m 55s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4b7b79a394f34fec9e0268b3a5073579 : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/51a9d150010a47a5b48f11e06df2dfa5 : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d6db6b2d6ed846bd8707a2132b3a9f03 : SUCCESS in 6m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fa358f4bc1c2bd6f6891d6e5e203c0cfbda479f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-19 11:37:05.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"a30b853b2b3062f20cf7c21716016e9a4a3a6031","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 13:13:22.000000000","message":"Patch Set 2: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e105fa36c68f451a85330efa22f6ef2a\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/b7b0f2ed9bf546dda78ecc11d6843a4b : SUCCESS in 45m 10s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/d3d8a9bae8f54b57b850897711aa9f33 : SUCCESS in 53m 49s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/6828c7dd9f934905884edeaf01412e23 : SUCCESS in 29m 58s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/bf037ba7d68f470d8e53e6dd956945ed : SUCCESS in 53m 27s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/c6a3afd8d6594d579a730d17d14a4488 : SUCCESS in 1h 05m 26s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6d3a92347cd04309b32c374e9cb36701 : SUCCESS in 55m 40s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2534760eca914aa0b69743583bba9d27 : SUCCESS in 32m 25s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/a757eebb6ebb40aa82526e3935301b3f : FAILURE in 1h 06m 43s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/e75ef5cbe653447cadfb184e8af0e814 : SUCCESS in 1h 03m 08s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/73489fb9177f439dbac3423268acdf10 : SUCCESS in 1h 15m 09s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/3367bc926019438387ce05710bdbf8ee : SUCCESS in 1h 02m 07s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/e65868173ed542fbba3f303b2785981d : SUCCESS in 1h 31m 07s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/47292613cb0747d886c15950acd89628 : SUCCESS in 1h 07m 03s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/418d285cf874435da0435548bbbcbbe7 : SUCCESS in 1h 31m 30s\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/44f15cfb7434405eb2d2e6bb7b3e9761 : SUCCESS in 56m 41s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/b902fed26bac4e0da406369c30c0775e : SUCCESS in 1h 32m 05s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/e6279a28673147129c888ff7f329345b : SUCCESS in 1h 30m 49s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/ec3b8502f8bd442883172fbf9482b46d : SUCCESS in 3m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a72558ce1514e5c9cde02eff0519668 : FAILURE in 2m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6ff3e8e1f50b49709c657e80ad03ec30 : SUCCESS in 3m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/71787e3111d64b4da6515f08eb804557 : SUCCESS in 2m 58s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1ada27908f5648b2b5b8f4015d05a38a : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0d55367c0d8b401a92a5f66b66fdf4c3 : SUCCESS in 3m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b11a6aacea8473e902509e991263524 : SUCCESS in 6m 03s","accounts_in_message":[],"_revision_number":2},{"id":"46e0d4a06a4a18c88eeb4ef15e6bacdef4b77bea","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 15:54:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4f6c67fe71e42eba232836c9d297f24\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/e3c69fe1feda4c8f9262b28ed31f5b70 : SUCCESS in 1h 28m 19s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/239e63d1c29443318acac10620898069 : SUCCESS in 2h 16m 21s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/638b4a615a15441dae89b5f8a9c55b3d : SUCCESS in 1h 23m 05s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/6959834c17b446e18ebb6dd62df065c8 : SUCCESS in 1h 31m 20s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/c5c64b15708e47e4920204b64853de29 : FAILURE in 1h 46m 13s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2cb3900f64bf4d6a9c189c25535a7bc1 : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5b6e7e1856d3481e947e549ad98f6724 : SUCCESS in 5m 47s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/e78cef0776bf4267a9a908b0eca01db6 : SUCCESS in 6m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"79d66802af60709de09ae200348cd5d345c57d97","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-20 16:33:54.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"d8106edb1058e36dc5746cb41b4981dfbb068362","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 18:40:31.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f009fec35d844625b7816720400e10ca\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/df6ca18c9064443a9faa8b6221fdc58f : RETRY_LIMIT in 5m 28s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/2a69bb0017b3470cb83a7ba47dad2e55 : RETRY_LIMIT in 2m 53s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/c6674b115f2f4a4f863dff1ee0692200 : SUCCESS in 30m 49s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/d0f10e1740a7463ab67732705dc794d5 : RETRY_LIMIT in 3m 30s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/2eff954ecf774e94999d6d98aa2752a1 : SUCCESS in 37m 36s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7ace57f751ee4ff8b70661c0e198b261 : RETRY_LIMIT in 2m 37s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/88a652d6f0034b3f910d905ead5794c8 : SUCCESS in 39m 13s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/490f3160b6e040d9a117e7aacfaea522 : FAILURE in 1h 00m 37s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/a94dcd69ed534349a776c3584ebf044a : SUCCESS in 1h 15m 59s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/d3b10f3739b94e8d9c71a55fe15414b7 : SUCCESS in 1h 38m 36s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/ed6d0590cf1b41f6af7b22cb17c24977 : SUCCESS in 1h 25m 28s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/63e366f41253499ab9d81aab4f8a5746 : SUCCESS in 1h 57m 21s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f2209815ff1542439bdcfd10151420c0 : SUCCESS in 1h 16m 42s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/1eddf14446d2483da0bc49f81f503c21 : SUCCESS in 1h 57m 31s\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/1a12c0992fa54ec9aa66b21fa9eef019 : SUCCESS in 1h 18m 04s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/b79c17a539934c0eb2d3e02b1a203bc6 : FAILURE in 30m 32s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/aeceed4a8e1645859e236507395aceae : SUCCESS in 1h 51m 56s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/dc95a2859d2740a4b94aa4af95561b32 : SUCCESS in 3m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96db6263f506434a8955e74c9dcb324f : SUCCESS in 3m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3ccb963dfa41468482e6326ff0cb63f5 : SUCCESS in 2m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2d849e1ac3944751a5b25952b1b3729d : SUCCESS in 3m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fe772ae0490c4f598581ca8f32b624ad : SUCCESS in 5m 46s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aada318bb20e49dcaa8982b9eafc380e : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8c4c766ec6a4425a2eddbd430d3902a : SUCCESS in 6m 24s","accounts_in_message":[],"_revision_number":3},{"id":"f9e76d8210bd73a8e5210f1fc756f81a570c2218","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 22:18:56.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d53efbba68274940a64f3425670ba86c\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/e83a4939b3734eaa827d1aa341f216fa : RETRY_LIMIT in 4m 53s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/b24408fd3aaf49eb9e176421be7da882 : RETRY_LIMIT in 4m 36s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/7a41c96a2570403bb12ff9bd33a0125b : RETRY_LIMIT in 7m 57s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/3cddb16b2a6a4cb198a60130e3885eb5 : RETRY_LIMIT in 4m 22s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/05963f0595544caa85265073a3b7da54 : FAILURE in 1h 16m 29s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9e156521cd1e4ab18ac11fe0df0cc7a6 : SUCCESS in 7m 05s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/cedf019a93b246cc8c9d12d12cc0e245 : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/755107004de544f4bb0cabefb804860f : SUCCESS in 12m 32s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"54c816b16d3d62fabb4f68c104a05814d3be55e9","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-23 19:21:02.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1795a529ad397a72cfd048f1a0418216a22398f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-23 21:11:59.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a3f8fa6b58164f618fcc4085d9235b4a\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/80ac54514de0405897d43a9f4a53248a : RETRY_LIMIT in 2m 05s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/e9a78c4b7dbe40f68bacf9d58df921f4 : RETRY_LIMIT in 3m 22s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/948eec88a5924739a752c38e1f5d21e0 : SUCCESS in 14m 45s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/838f09f5db6e4c1b9d2411a1e8c7ec45 : RETRY_LIMIT in 3m 30s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/f067c9cb2ae74084a09634f0813670c3 : SUCCESS in 49m 47s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/67230caa18394c25bbbd00b855cac382 : RETRY_LIMIT in 1m 36s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1b8ebd5294ad4d8f9eb3d45f694f9250 : SUCCESS in 28m 20s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/869dd757444a47f8833dee9438ec598d : SUCCESS in 1h 10m 44s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/e3dbf06425fd4eb2a3067c93e38abe76 : SUCCESS in 1h 11m 16s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/194d1f999f47417a95d068d7c4dbdb23 : SUCCESS in 1h 36m 43s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/6a9a6802fa224bf7af3347cf133fb0d5 : SUCCESS in 1h 20m 30s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/e77edba2b54f4b76a9dbef38fa883cf8 : SUCCESS in 1h 48m 01s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/803487e305ec494fa145cd5e6aa0beca : SUCCESS in 1h 15m 11s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/b3cf289dce954a1cbcadd00d6cd6dbad : SUCCESS in 1h 49m 34s\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/2223dcce41f04b4faedd5eeca2b2efe2 : SUCCESS in 1h 15m 28s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/57ec2857991447ee98fd09be1551ee8d : SUCCESS in 1h 46m 12s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/65f637c3302f4a5cb6d010657b5185d8 : SUCCESS in 1h 46m 28s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/e3653e26a9cb4d11927f6f0077122c7b : SUCCESS in 2m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6fdd4d1140274909b56e194d9452c61c : SUCCESS in 3m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6f92ac818cf84d608ae79c1eebe48d2e : SUCCESS in 2m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d03c5915c85c48e9b5d84aa7a054eb53 : SUCCESS in 3m 09s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a681f88e137c4197b015526ec4a8200b : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e38338b76320496ab08dbf6b274ebc61 : SUCCESS in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58209232818f49f8a2daf3d8dc73f575 : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":3},{"id":"7a496cce44c32c1c7236ebf798f75f750d225b6c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 00:24:37.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3300c09c73bd4ca6a4f65c738a44942c\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/cbb6ba63db87472fbcb4e9c22bcff639 : RETRY_LIMIT in 4m 10s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/22154a97a9b347b68c964fef7a93ddae : RETRY_LIMIT in 3m 31s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/43a8265ffc704a7da703988c0bb79271 : RETRY_LIMIT in 4m 42s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/d15ffd26e68b4a2f8f7050c5b5e9002d : RETRY_LIMIT in 3m 48s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/eb32c31372194d56b2c30f8ebe4fd916 : FAILURE in 1h 40m 02s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/89f85c78cd41407aa8506fe1aa22eb7e : SUCCESS in 6m 47s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a18ee1b19fdb48f68227b9dcd8c55ec6 : SUCCESS in 6m 15s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b1002f36991e477bb9cc1d2272fb1224 : SUCCESS in 9m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"bb38c9a0ccaf246551d99ecb226d75380ffb5d98","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"real_author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-07 19:53:50.000000000","message":"Patch Set 4: Patch Set 3 was rebased on behalf of \u003cGERRIT_ACCOUNT_28048\u003e\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}],"_revision_number":4},{"id":"f731b1040cd7d74a82515be4716669e45bf66232","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-07 22:29:04.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c41c95a7874d422aa9af49c96f9a695c\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/7d90ebf8af3a43af92b4c0a3f902bd66 : SUCCESS in 1h 21m 28s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/24d04ba3f46744879eff9b232e536710 : SUCCESS in 1h 08m 56s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/b88e57862c3643c3a4f1ea8f9370ee88 : SUCCESS in 38m 58s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/3a46d1a6920447349b98cc7781d15bd7 : SUCCESS in 1h 06m 08s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/dab02825a90b42e9aafdde85d2406a7a : SUCCESS in 38m 01s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/28bfe3f4b21c494ab7c76aa38d0e0bca : SUCCESS in 1h 04m 46s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/41f7df596b764f1c8837a58bad9f862a : SUCCESS in 22m 23s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/c9d091fca80c48b9931e8d03b29e3d6c : SUCCESS in 1h 16m 28s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/68d0e5462f564f689a481f941538210f : SUCCESS in 1h 21m 56s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/f52ab9366ec1432780f79909c327876b : SUCCESS in 1h 20m 01s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/059b9a891fbd4851b3c4ea01c543bfeb : SUCCESS in 2h 20m 26s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bdc069e6f90345189c042c4bbfee2e6e : SUCCESS in 1h 52m 31s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/7e61ed43cf7b488199dba52d0f181b43 : SUCCESS in 2h 31m 33s\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/cd17a10bdcf34fdbb30011c231c826a4 : SUCCESS in 1h 25m 28s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/3923c25644e7484a9c95cef932c5ce1f : SUCCESS in 1h 27m 12s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/e2d7bab7e4904341aa1eb901416a0479 : FAILURE in 1h 24m 44s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/8bbfecfb1c1447e9b52564f6ce09db7f : TIMED_OUT in 2h 33m 11s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/b970c80ac3c04baba1d08593d605188a : FAILURE in 50m 52s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/d4df6bd87b0f494493be2031efd34591 : SUCCESS in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc1d0e7f2de74ab9a5228e182f3b857d : SUCCESS in 3m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a8f4d84c18944c1ead72a0732962bdda : SUCCESS in 3m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/030f84dd9d914eb7aaf56c1e5475c732 : SUCCESS in 3m 24s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/04781b425fa04e36b93b942179f34634 : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/11eecb84eef14bb690a23e28dab6b7a6 : SUCCESS in 3m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/455a19507e51407f82c8bc2b345bc4fe : SUCCESS in 7m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ba1d89edcc5b441fa35d130c9676dacd : SUCCESS in 9m 50s","accounts_in_message":[],"_revision_number":4},{"id":"2338cd3fb8efebcc2881ed608e6faf961833d156","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 04:00:35.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a27098460c347158fd32327c9955a42\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/59dd1a632b654484a8c399151bd1dc9b : SUCCESS in 2h 36m 27s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/b70f2640c9b14490afc57515157a8f19 : TIMED_OUT in 3h 10m 54s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/2e8c2c839e1649309c7ff216e1147ace : SUCCESS in 1h 49m 34s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/e6989beb4e3f4e4aa0da9029bbf311bb : TIMED_OUT in 3h 16m 59s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/0b0a774ee5ad495db07b66cdbee7fbdd : TIMED_OUT in 2h 35m 12s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/fb44ff3a3cbc4271bdba07d2bb85820a : SUCCESS in 2h 31m 52s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/70f927db10f748049e459764c1ef98e9 : SUCCESS in 4m 33s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8bc48969c1da4422a2ca3b6d765f602d : SUCCESS in 4m 52s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/8710751231bb460fb3cfbb13a4446125 : SUCCESS in 11m 04s (non-voting)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"6251810a04702d80165fec1efea11324b08a9c18","revisions":{"89eb5f1544d7d824828c211e9303ff52ced09687":{"kind":"REWORK","_number":1,"created":"2026-03-18 17:53:32.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/68/981168/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/68/981168/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/981168/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/981168/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/981168/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/68/981168/1"}}},"commit":{"parents":[{"commit":"482aec6a69e47034bc63bbb9d0d80d59cc710ffc","subject":"Merge \"Update Prometheus Alertmanager and exporters to the latest\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/482aec6a69e47034bc63bbb9d0d80d59cc710ffc"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-03-18 17:46:25.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-03-18 17:53:26.000000000","tz":0},"subject":"Allow KOLLA_LEGACY_IPTABLES on CentOS/Redhat","message":"Allow KOLLA_LEGACY_IPTABLES on CentOS/Redhat\n\nEnterprise Linux use the iptables provider to also configure ip6tables:\n\n   ()[root@3e5de7e954fb /]# update-alternatives --display iptables\n   iptables - status is auto.\n    link currently points to /usr/sbin/iptables-nft\n   /usr/sbin/iptables-nft - priority 10\n    follower ip6tables: /usr/sbin/ip6tables-nft\n    follower iptables-restore: /usr/sbin/iptables-nft-restore\n    follower iptables-save: /usr/sbin/iptables-nft-save\n    follower ip6tables-restore: /usr/sbin/ip6tables-nft-restore\n    follower ip6tables-save: /usr/sbin/ip6tables-nft-save\n   Current `best\u0027 version is /usr/sbin/iptables-nft.\n\nThis allows use to also set KOLLA_LEGACY_IPTABLES for EL. Useful if\nyou provided your own iptables-legacy package or when backed to EL9,\nwhere iptables-legacy exists in EPEL.\n\nChange-Id: I1f6694fe2948b82cdf6089c699fbcca08291958b\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/89eb5f1544d7d824828c211e9303ff52ced09687"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/89eb5f1544d7d824828c211e9303ff52ced09687"}]},"branch":"refs/heads/master"},"d174bad25a71e72e177504368fd2a49765028637":{"kind":"REWORK","_number":2,"created":"2026-03-19 11:37:05.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/68/981168/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/68/981168/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/981168/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/981168/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/981168/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/68/981168/2"}}},"commit":{"parents":[{"commit":"482aec6a69e47034bc63bbb9d0d80d59cc710ffc","subject":"Merge \"Update Prometheus Alertmanager and exporters to the latest\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/482aec6a69e47034bc63bbb9d0d80d59cc710ffc"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-03-18 17:46:25.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-03-19 11:36:59.000000000","tz":0},"subject":"Allow KOLLA_LEGACY_IPTABLES on CentOS/Redhat","message":"Allow KOLLA_LEGACY_IPTABLES on CentOS/Redhat\n\nEnterprise Linux use the iptables provider to also configure ip6tables:\n\n   ()[root@3e5de7e954fb /]# update-alternatives --display iptables\n   iptables - status is auto.\n    link currently points to /usr/sbin/iptables-nft\n   /usr/sbin/iptables-nft - priority 10\n    follower ip6tables: /usr/sbin/ip6tables-nft\n    follower iptables-restore: /usr/sbin/iptables-nft-restore\n    follower iptables-save: /usr/sbin/iptables-nft-save\n    follower ip6tables-restore: /usr/sbin/ip6tables-nft-restore\n    follower ip6tables-save: /usr/sbin/ip6tables-nft-save\n   Current `best\u0027 version is /usr/sbin/iptables-nft.\n\nThis allows use to also set KOLLA_LEGACY_IPTABLES for EL. Useful if\nyou provided your own iptables-legacy package or when backed to EL9,\nwhere iptables-legacy exists in EPEL.\n\nChange-Id: I1f6694fe2948b82cdf6089c699fbcca08291958b\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d174bad25a71e72e177504368fd2a49765028637"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d174bad25a71e72e177504368fd2a49765028637"}]},"branch":"refs/heads/master"},"83ae068fbeca221d82b8c1f3b327d04f594136ce":{"kind":"REWORK","_number":3,"created":"2026-03-20 16:33:54.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/68/981168/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/68/981168/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/981168/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/981168/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/981168/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/68/981168/3"}}},"commit":{"parents":[{"commit":"482aec6a69e47034bc63bbb9d0d80d59cc710ffc","subject":"Merge \"Update Prometheus Alertmanager and exporters to the latest\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/482aec6a69e47034bc63bbb9d0d80d59cc710ffc"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-03-18 17:46:25.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-03-20 16:33:47.000000000","tz":0},"subject":"Allow KOLLA_LEGACY_IPTABLES on CentOS/Redhat","message":"Allow KOLLA_LEGACY_IPTABLES on CentOS/Redhat\n\nEnterprise Linux use the iptables provider to also configure ip6tables:\n\n   ()[root@3e5de7e954fb /]# update-alternatives --display iptables\n   iptables - status is auto.\n    link currently points to /usr/sbin/iptables-nft\n   /usr/sbin/iptables-nft - priority 10\n    follower ip6tables: /usr/sbin/ip6tables-nft\n    follower iptables-restore: /usr/sbin/iptables-nft-restore\n    follower iptables-save: /usr/sbin/iptables-nft-save\n    follower ip6tables-restore: /usr/sbin/ip6tables-nft-restore\n    follower ip6tables-save: /usr/sbin/ip6tables-nft-save\n   Current `best\u0027 version is /usr/sbin/iptables-nft.\n\nThis allows use to also set KOLLA_LEGACY_IPTABLES for EL. Useful if\nyou provided your own iptables-legacy package or when backed to EL9,\nwhere iptables-legacy exists in EPEL.\n\nChange-Id: I1f6694fe2948b82cdf6089c699fbcca08291958b\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/83ae068fbeca221d82b8c1f3b327d04f594136ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/83ae068fbeca221d82b8c1f3b327d04f594136ce"}]},"branch":"refs/heads/master"},"6251810a04702d80165fec1efea11324b08a9c18":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-06-07 19:53:50.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"real_uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/68/981168/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/68/981168/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/981168/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/981168/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/981168/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/68/981168/4"}}},"commit":{"parents":[{"commit":"cbdbd894b06b049e36ecd4fab4d0ecc3a49e0c00","subject":"Merge \"masakari: Add crmadmin and cibadmin to monitors sudoers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cbdbd894b06b049e36ecd4fab4d0ecc3a49e0c00"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-03-18 17:46:25.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-06-07 19:53:50.000000000","tz":0},"subject":"Allow KOLLA_LEGACY_IPTABLES on CentOS/Redhat","message":"Allow KOLLA_LEGACY_IPTABLES on CentOS/Redhat\n\nEnterprise Linux use the iptables provider to also configure ip6tables:\n\n   ()[root@3e5de7e954fb /]# update-alternatives --display iptables\n   iptables - status is auto.\n    link currently points to /usr/sbin/iptables-nft\n   /usr/sbin/iptables-nft - priority 10\n    follower ip6tables: /usr/sbin/ip6tables-nft\n    follower iptables-restore: /usr/sbin/iptables-nft-restore\n    follower iptables-save: /usr/sbin/iptables-nft-save\n    follower ip6tables-restore: /usr/sbin/ip6tables-nft-restore\n    follower ip6tables-save: /usr/sbin/ip6tables-nft-save\n   Current `best\u0027 version is /usr/sbin/iptables-nft.\n\nThis allows use to also set KOLLA_LEGACY_IPTABLES for EL. Useful if\nyou provided your own iptables-legacy package or when backed to EL9,\nwhere iptables-legacy exists in EPEL.\n\nChange-Id: I1f6694fe2948b82cdf6089c699fbcca08291958b\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6251810a04702d80165fec1efea11324b08a9c18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6251810a04702d80165fec1efea11324b08a9c18"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"482aec6a69e47034bc63bbb9d0d80d59cc710ffc","ours":"83ae068fbeca221d82b8c1f3b327d04f594136ce","theirs":"cbdbd894b06b049e36ecd4fab4d0ecc3a49e0c00","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
