)]}'
{"id":"openstack%2Fkolla~981186","triplet_id":"openstack%2Fkolla~master~If185fc29eabc831a7cc1b28b435cc5ecfd63732e","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"36962":{"account":{"_account_id":36962,"name":"Volodymyr Boiko","display_name":"Volodymyr","email":"volodymyr.boiko@vastdata.com","username":"fnn45"},"last_update":"2026-03-24 11:24:28.000000000","reason":"Change was submitted"},"17669":{"account":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"last_update":"2026-03-24 11:24:28.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-03-24 09:34:07.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-03-20 12:18:21.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"10910":{"account":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"last_update":"2026-03-24 11:24:28.000000000","reason":"Change was submitted"},"36238":{"account":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"last_update":"2026-03-24 11:24:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If185fc29eabc831a7cc1b28b435cc5ecfd63732e","subject":"Add NVME CLI to Glance API container","status":"MERGED","created":"2026-03-18 20:02:19.000000000","updated":"2026-03-24 11:25:34.000000000","submitted":"2026-03-24 11:24:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"981186","meta_rev_id":"991a5982915296a7a86554cd1e3a1a481d75923d","_number":981186,"virtual_id_number":981186,"owner":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-24 11:24:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2026-03-20 10:24:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2026-03-24 09:34:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2026-03-20 12:18:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2026-03-24 09:34:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2026-03-24 09:34:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":36962,"name":"Volodymyr Boiko","display_name":"Volodymyr","email":"volodymyr.boiko@vastdata.com","username":"fnn45"},{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},{"_account_id":36962,"name":"Volodymyr Boiko","display_name":"Volodymyr","email":"volodymyr.boiko@vastdata.com","username":"fnn45"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-18 20:03:31.000000000","updated_by":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"reviewer":{"_account_id":36962,"name":"Volodymyr Boiko","display_name":"Volodymyr","email":"volodymyr.boiko@vastdata.com","username":"fnn45"},"state":"REVIEWER"},{"updated":"2026-03-18 20:05:34.000000000","updated_by":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"reviewer":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"state":"REVIEWER"},{"updated":"2026-03-18 22:14:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-19 07:46:10.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-03-20 10:24:23.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2026-03-24 06:24:24.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2026-03-24 09:34:07.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"c7105052b5be4f7b9007c0765d84bbf2162994fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"date":"2026-03-18 20:02:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0a6e55ec69bfb8daf88c61c6b5f74cb9265e3613","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 22:14:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f7e9bf61ad384203a8af5c94a2567066\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/23b684221c7c482e9049f39a17c51ae0 : SUCCESS in 45m 38s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/c461ad6ff82d4a3ba23c4bdee7fdecc9 : SUCCESS in 38m 34s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/73d3e834d10441b18081cc19cded1066 : SUCCESS in 13m 13s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/dcce0701efb741ab80aca3334399a01d : SUCCESS in 37m 01s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/db42890aa4ba4e0384666aeb5d560a6b : SUCCESS in 53m 32s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/843ed200a51a41a5a3587f4ff8b809c9 : SUCCESS in 32m 41s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d037047f9c3144918c3a708bdbcea2c2 : SUCCESS in 28m 30s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/ec7892057dd84b59b0e654e35e1a302d : FAILURE in 35m 49s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/2c81d2a43f7e490498cdb3a078c47581 : SUCCESS in 51m 48s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/ac817e265f4d4d968eb2975176be790b : SUCCESS in 1h 23m 43s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/6cae321e78234bec8ee4f3c0efeb803b : SUCCESS in 58m 31s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/8771659a2abc410e8753ee73ba2b9bee : FAILURE in 57m 36s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b19773b9e94547a393d3ebca509b4e2a : SUCCESS in 1h 07m 49s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/ce45571241b64d6d8a6a9affd9d8e79e : SUCCESS in 1h 18m 14s\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/50fe449f54274af8b889cf44eb45c055 : SUCCESS in 1h 40m 44s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/e439fcdd51d34ce8b8399668a54d10ef : SUCCESS in 2h 10m 15s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/88997bb786ee43c7b6a84cf923a5967c : SUCCESS in 1h 37m 13s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/990ae1dc68f945608e5bdeeb145fdfd3 : FAILURE in 1h 18m 27s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/b4f670ef44fe4227ac97ec0493c55c55 : SUCCESS in 2m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c4a9164419843d6a5389ce6d403a9cc : SUCCESS in 3m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/77b30111ed97487dba81fd0f83b9fd14 : SUCCESS in 3m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d86319156f3d47b2844fa86be1e0159b : SUCCESS in 3m 04s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d02b2c997fd24c3a88d2192243c90c7f : SUCCESS in 3m 42s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2dd5a62eb1b84fb4bc9ef3ffe7a16a27 : SUCCESS in 3m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bd69b1e5ee34a88898bec532dd1a0ae : SUCCESS in 6m 58s","accounts_in_message":[],"_revision_number":1},{"id":"5ee352c9afef805e3f484ad87a2191cf1aee066c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 00:37:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5fb587a8b37f4dd89d5b9881dc771c63\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/2382871cff2740e0a8c8de752fae7f84 : SUCCESS in 58m 42s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/17efc4499a7645da823bb0322136eafe : SUCCESS in 1h 00m 43s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/39cd28803f384e2cb135ce4c6f66f5d1 : SUCCESS in 58m 55s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/1447d46ac3e5478c978f5c2df402ea0c : SUCCESS in 1h 15m 31s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/774e9f43963341a4926aa379f29da2dc : FAILURE in 1h 11m 20s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c11273b075b8471ebfe95a010afad7d0 : SUCCESS in 5m 34s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/704edea65f7d42aea26a6f2332db9eb9 : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/34aa0471f8324b559fe524657bd13b3f : SUCCESS in 6m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d0a949a222edbd274f618a06e2c288a30fd47412","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-03-19 07:46:10.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d15aaaa8f34ff2c0d82257c2c6b214ddc6ea905a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 09:53:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee2865e28de44d3ea76dd54a66a9be20\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/b0efd6d6d5d1494d9db39ce450763f02 : SUCCESS in 2h 04m 42s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/7cb260ba635e4450be020bce7763f78c : SUCCESS in 1h 58m 13s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/da18cf9a7ead46ab84cde33a48284b28 : SUCCESS in 1h 10m 14s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/451bdfc5ef784a85a4c4d503344bb448 : SUCCESS in 49m 09s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/19b15244d9864fbd9d59af425c95ad1f : FAILURE in 1h 55m 52s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ffc301ab062946c1a3f9afd58cf6e49f : SUCCESS in 5m 40s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/19d264eaa9214bccae51cc44ae8d7695 : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/9fa5e9d5f6d543f8a2be6e09c11186e4 : SUCCESS in 6m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"177dae3f60d6df7c4bed5b2c7ce29bc491a501ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 10:29:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98a2d901f4a64c4fa195d57b4da15dd1\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/2078d24691cc4b18b339c7f341b256c4 : SUCCESS in 46m 09s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/744ea228760348029a5921c87712911d : SUCCESS in 46m 55s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/412d99d7c0d6459aa24a4800d6e5ef8d : SUCCESS in 26m 09s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/e43765c2ed7740aba01ded90f08b6417 : SUCCESS in 48m 32s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/2974991d3eb84912b81c3c7a81f09e1c : SUCCESS in 1h 06m 56s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8608fcb38b1249a3aa82db5359297cf3 : SUCCESS in 55m 30s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a6809c820c864f3fbd6e3078560ecb68 : SUCCESS in 13m 16s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/a502f752c10d41169fa9daf901893d10 : FAILURE in 38m 27s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/68338d675e9c47bfaadb63bf156799ae : SUCCESS in 1h 19m 48s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/5c62d7f7d883451e9d9bd21fdd8f1132 : SUCCESS in 1h 45m 39s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/aa5c2c4b9f214c8fbda6ba6bba82a91b : SUCCESS in 1h 23m 41s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/1751402287dc4b1fa169568f6e49b517 : SUCCESS in 1h 27m 15s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bcb2d4cabd5548b7a99f3e997be8b1fb : SUCCESS in 1h 01m 42s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/3e8af4684cef4654a67a2a821c1b58cb : SUCCESS in 1h 43m 05s\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/9a4438b077074db79a55590917737c49 : SUCCESS in 2h 01m 34s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/27c3d7130bf447fa95b069662815b3b1 : SUCCESS in 2h 35m 00s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/106897fbb058454f83bec387d29aaaf5 : SUCCESS in 1h 42m 37s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/6c8d4f376b4f45cc9781d4fcd41a1d67 : SUCCESS in 2h 37m 30s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/af63c5f053b14d71874217e7eefb7440 : SUCCESS in 3m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d0ade6340f44104b331aa0472c3c39f : SUCCESS in 3m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1af584082e744fa68f61ec624199d8bb : SUCCESS in 1m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4839795bdd0842d9a2a44a9d31416b64 : SUCCESS in 3m 08s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1aaa38262c3a4952a7b4ed4d4e27c95a : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/11695455aea64f73a7ea1e66a47a9768 : SUCCESS in 1m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8187c25f15994da7abe69fbc920688d1 : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":1},{"id":"b7ce3c719d866b6b182c4f64cbe9c9ff089635c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2026-03-20 10:23:42.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"47cce206f61cc1ae6c4b2ec8306bfdb0200450e7","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2026-03-20 10:24:23.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b8020c1e310e486bed3fca87b08ee88dd12e8200","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-03-20 12:18:21.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"4802768774dff3839e2eb89942eadc915f5e965e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 12:54:49.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b8cf6af1f8a945cfa12742641406f08e\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/bbc11305d6194586a8305ea096c73c69 : RETRY_LIMIT in 2m 22s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/f4950d419f904255893a951df9db88cc : RETRY_LIMIT in 2m 09s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/3cea9f55c9324466960a151239888829 : SUCCESS in 12m 38s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/8303200d2cf64a1eaacbae8ea7d7013c : RETRY_LIMIT in 4m 19s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/4208ba2bb8c74e35ad7c86d0fc1c0655 : SUCCESS in 1h 06m 00s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e4e76b860f7d410c8552ff9d0d4a1266 : RETRY_LIMIT in 2m 37s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/776463d42d9a44cdb1bab4c072627ece : SUCCESS in 27m 12s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/009e8779cd394eef913d7897ed1fa2e8 : FAILURE in 1h 05m 21s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/2a594bc252024733adf09cac5ef6805c : SUCCESS in 1h 09m 28s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/e1f4cf59d5ef4db5b554bf49b8bc7938 : SUCCESS in 1h 33m 53s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/dd5596be2e7644828bea407e04364852 : SUCCESS in 1h 18m 43s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/c3ce388d580b4c2f9c45c31dedeeefb2 : SUCCESS in 1h 47m 37s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a27106030afe4902aa2655a6859c3a66 : SUCCESS in 1h 26m 02s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/58f3cd98bfbc4c93a676a2b39872f586 : SUCCESS in 1h 56m 55s\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/27770f59d6824e7fbc190b3a84563d46 : SUCCESS in 1h 49m 54s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/df4323acd01547f2a1ffecb1381760dd : FAILURE in 2h 28m 54s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/3bb9ee22aa8e476181c395a14973d538 : FAILURE in 2h 00m 20s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/023f249a727c4ef58e6c061ffd01ceca : FAILURE in 2h 16m 23s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/ca74bdb2b1ab4842971cd55d15a96408 : SUCCESS in 2m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e124d09bbf94be1a5f049f1e9c7274a : SUCCESS in 3m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/362254e3af0c41fc988748b657a7af2a : SUCCESS in 3m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/90fb2e43b88f48c18c853f9a4ec920d9 : SUCCESS in 3m 15s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9d3180fa53c04e199aa9c51566e4ac78 : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aa0b8676ad2d413cb3cf7c9cb0658d89 : SUCCESS in 3m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65d5e0f77bba452c8ec380ea51128773 : SUCCESS in 6m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/19a571baef3f436ba6917919c0701908 : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":2},{"id":"705481d36d948e8e54c4ad963dbfd04fa954684e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 15:08:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3526505ec5804d46b5bf0475e76bad33\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/958a24ec6c6d486fbebb13ce2fd33bc2 : RETRY_LIMIT in 4m 14s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/284dba0f2b9a47f7b51c396541dc0880 : RETRY_LIMIT in 3m 29s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/e53701533c12424fb76b0ac13fe3ca51 : RETRY_LIMIT in 5m 02s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/fac6226f4de846ec95abd9d2d99a0efc : RETRY_LIMIT in 4m 53s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/a3aeac6914b74583acf2303357b15119 : FAILURE in 1h 44m 30s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9ec363ee9934478c9706f37d3ee5b57a : SUCCESS in 7m 03s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0bdc5c7d9c524a58a95b5eefed9fca55 : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/59dd388da7d74e838f85b9bbbe75b71d : SUCCESS in 9m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7beefef15ea234f35b6164d33764004ab49fa3f6","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-03-24 06:24:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1eb8ae0898f63092148de767a8912d544a02df94","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-03-24 06:25:38.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b48e025db0babf91aafbc9f4ba26ed3b6aa94840","author":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"date":"2026-03-24 08:27:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b774840df120ea2cfa6b15891439fb8ae5d461a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 09:33:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2c1d91abd3444dd811fa1c9469e4ec8\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/b921ecbd4e5a47249eb3c7c8e1d948d3 : SUCCESS in 40m 58s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/6a62c3211a1e45f6893690c985d6e9d2 : SUCCESS in 1h 00m 42s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/221d9ac70796410da86ae9b7d1190c77 : SUCCESS in 27m 26s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/0c59ae69272848c088a5eb0647d610b0 : SUCCESS in 48m 33s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/983c077b54db4f78b8dec32d220ce1ad : SUCCESS in 55m 41s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a90271579d234aed931ea500efadbc0b : SUCCESS in 52m 52s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/52960a2816db406da67d0028827358d7 : SUCCESS in 28m 39s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/b03d5ccbbdba4c409ff67fda529281c4 : SUCCESS in 1h 17m 48s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/f0bf30ab394a450f9916ae81a7c6640d : SUCCESS in 1h 14m 59s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/babf8ef15399498588ff949dd7fd2778 : SUCCESS in 1h 44m 01s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/6a91b39ee6054084a99d28dab753cf2d : SUCCESS in 1h 18m 49s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/48ea3e51e04f4b20bd215a12335496b6 : SUCCESS in 1h 55m 09s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/54d9f0c774924a88863e123aed057470 : SUCCESS in 1h 34m 38s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/c52d6a4f3f054201a03cfdf290ff06a4 : SUCCESS in 2h 03m 31s\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/82e863139fc84c79bb6bab77f4131788 : SUCCESS in 1h 57m 48s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/72f4cd8477d44fa9b3a60843cc121b93 : TIMED_OUT in 3h 03m 56s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/88ca4cb46a3a405ba54167b32d5063fa : SUCCESS in 1h 56m 41s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/03f9907f01bc4e68b1e97262617c106b : FAILURE in 1h 42m 01s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/fcdef0c1645240e9a66d1b106e561901 : SUCCESS in 2m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d680b33c5ec42fdb112c5dcfceb559b : SUCCESS in 3m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1241f758d3dd4993abd21f093b04d9ce : SUCCESS in 3m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6061d90a21cf4b11a22e5d9e98d588a8 : SUCCESS in 3m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3632a2a4ee2f44c6a706fc338cfa0736 : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/025a9b0489944a628e029355c6735e65 : SUCCESS in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5af43c475a32483c96f60adc97d12f9e : SUCCESS in 3m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5d1f52f2a95a421d93b7a8c2279f2d0c : SUCCESS in 2m 06s","accounts_in_message":[],"_revision_number":2},{"id":"fdaf2ff2c1fda933068304a6342ddc763592c16e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-03-24 09:34:07.000000000","message":"Patch Set 2: Backport-Candidate+1 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"03e3250d241f8a35ef09b7f88c991164641d35af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 09:34:21.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"491e8ea9216e58faafc562bc751efb82fbc4626c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 09:51:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f295f3a91bf4293af701c5c2d5be9c1\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/5a251233019d47bcb7287cb9895b40d5 : SUCCESS in 2h 47m 39s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/e9a02f53cb404cf5aaed6c72cdad26e6 : SUCCESS in 3h 02m 29s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/1bce8973ed7c4878b9daba629e9ff666 : SUCCESS in 2h 58m 38s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/0db080ebecd4406aa889e336a409d5b4 : SUCCESS in 1h 46m 29s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/f697b01224674777af4eaaf6a63b4fed : TIMED_OUT in 2h 32m 07s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8263068dc8a945c99154ce625590f615 : SUCCESS in 7m 00s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c6c01286ef7f465d8ba7696b3039e8fa : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/3c3d8d89e2af4edcbe4454e92bab7ed4 : SUCCESS in 8m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b8dbd652f255b87f578cb45961691101441edf72","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 11:24:28.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bea38f32f41f4a0fac2c431c205cc98b\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/9fd0b1a9345c49b4bde99274ac64f81f : SUCCESS in 50m 55s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/6a4acf211a7e453680fac75aa7c55716 : SUCCESS in 30m 18s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/5fd4ac0816014e04befde7b66d6799d2 : SUCCESS in 47m 22s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/a7c27eb34823471989e0dec8fcd0859b : SUCCESS in 34m 39s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a79585f65ac2486ea8600745241773a9 : SUCCESS in 50m 29s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8350febf493a46809ea5fe3337178d34 : SUCCESS in 35m 08s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/c2b7e7d3baa94e2ea37ef167762a67bb : SUCCESS in 1h 13m 24s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/8b7fe12347604118a93efbd4dc9dc38c : SUCCESS in 1h 38m 44s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/12076f7bcf3b43648973b6230c092e57 : SUCCESS in 1h 16m 09s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/a5e43ebaef514353b3d59549981af834 : SUCCESS in 1h 48m 22s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/364164c1053e4bb5912607c01ccb7acd : SUCCESS in 1h 18m 26s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/e1bf598aa9c34c46b47b00914cba44f0 : SUCCESS in 1h 48m 14s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/bf0d61673ec346f69f26267f8dbbb0c0 : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe1105262f1d41319c2ccc904c924247 : SUCCESS in 3m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7ae08443b7a84b6a9415f55be9c2b121 : SUCCESS in 4m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b33ca6469702472fbaa1e3c2e4d1049c : SUCCESS in 3m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2252f386ee9b46a5adfe4db2b925093f : SUCCESS in 7m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/19ba0bdc73a647119d941edefb7e08f5 : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":2},{"id":"f4d391cb817fadec59ff37b39ac4f7d32f28e326","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 11:24:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"991a5982915296a7a86554cd1e3a1a481d75923d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 11:25:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea20f785aee346d48dc5b88a20d43447\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8155668f74584fc2a0da9ba7053faa8b : SUCCESS in 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa65444f0f344c16b183a33f88fd5778 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1636abcfe1f877e78ffe4207f95e84c46da7cdba","revisions":{"3914bebc11788736f4591dd1cecdee3ecabd625d":{"kind":"REWORK","_number":1,"created":"2026-03-18 20:02:19.000000000","uploader":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"ref":"refs/changes/86/981186/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/86/981186/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/981186/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/981186/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/981186/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/86/981186/1"}}},"commit":{"parents":[{"commit":"482aec6a69e47034bc63bbb9d0d80d59cc710ffc","subject":"Merge \"Update Prometheus Alertmanager and exporters to the latest\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/482aec6a69e47034bc63bbb9d0d80d59cc710ffc"}]}],"author":{"name":"Stig Telfer","email":"stig@stackhpc.com","date":"2026-03-18 19:59:27.000000000","tz":0},"committer":{"name":"Stig Telfer","email":"stig@stackhpc.com","date":"2026-03-18 19:59:27.000000000","tz":0},"subject":"Add NVME CLI to Glance API container","message":"Add NVME CLI to Glance API container\n\nVAST uses NVME-over-TCP to implement Cinder volumes.  When\nCinder uses VAST as a backend driver and Glance uses Cinder\nas store, the nvme-cli package is needed in glance-api,\nto generate NVME NQNs.\n\nPartial-Bug: #2035975\nChange-Id: If185fc29eabc831a7cc1b28b435cc5ecfd63732e\nSigned-off-by: Stig Telfer \u003cstig@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3914bebc11788736f4591dd1cecdee3ecabd625d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3914bebc11788736f4591dd1cecdee3ecabd625d"}]},"branch":"refs/heads/master"},"1636abcfe1f877e78ffe4207f95e84c46da7cdba":{"kind":"REWORK","_number":2,"created":"2026-03-20 10:23:42.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/86/981186/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/86/981186/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/981186/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/981186/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/981186/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/86/981186/2"}}},"commit":{"parents":[{"commit":"482aec6a69e47034bc63bbb9d0d80d59cc710ffc","subject":"Merge \"Update Prometheus Alertmanager and exporters to the latest\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/482aec6a69e47034bc63bbb9d0d80d59cc710ffc"}]}],"author":{"name":"Stig Telfer","email":"stig@stackhpc.com","date":"2026-03-18 19:59:27.000000000","tz":0},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2026-03-20 10:23:24.000000000","tz":0},"subject":"Add NVME CLI to Glance API container","message":"Add NVME CLI to Glance API container\n\nVAST uses NVME-over-TCP to implement Cinder volumes.  When\nCinder uses VAST as a backend driver and Glance uses Cinder\nas store, the nvme-cli package is needed in glance-api,\nto generate NVME NQNs.\n\nPartial-Bug: #2035975\nChange-Id: If185fc29eabc831a7cc1b28b435cc5ecfd63732e\nSigned-off-by: Stig Telfer \u003cstig@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1636abcfe1f877e78ffe4207f95e84c46da7cdba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1636abcfe1f877e78ffe4207f95e84c46da7cdba"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d1"],"failing_atoms":["label:Backport-Candidate\u003d-1"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
