)]}'
{"id":"openstack%2Fkolla~981705","triplet_id":"openstack%2Fkolla~master~Ic6ea2f92cb57043585eb722aeabf664589b78454","project":"openstack/kolla","branch":"master","attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-03-23 11:37:41.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ic6ea2f92cb57043585eb722aeabf664589b78454","subject":"Ensure requests is present for docker role","status":"NEW","created":"2026-03-23 05:44:04.000000000","updated":"2026-03-23 13:37:54.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3b99cdf9c2dbacbfe9cba684200d470d57a26c5c","_number":981705,"virtual_id_number":981705,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-03-23 11:37:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-23 07:32:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bbb945bb5c16ce9ad8bb4880c56096a37bc05432","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-03-23 05:44:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9ef9a1aa0202461537892932060be285cb7e27b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-23 07:32:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0baac22269b94cbdb925425e2664e8b4\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/41d9973f13de422fbef97283074739ca : SUCCESS in 37m 18s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/e9d944bb80fb461b9b3d94d60b4e3eb3 : SUCCESS in 43m 25s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/36473f0b227e46cea15faed64258e905 : SUCCESS in 26m 13s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/b1cb26789a4f4692ba7c2a4219bf42c1 : SUCCESS in 48m 47s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/7bdb6482f564457a8b92054793f87daa : SUCCESS in 37m 22s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/635c5e353e2443aba938933ba234d103 : SUCCESS in 37m 54s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8d32545744b343488ef9616e330b46ce : SUCCESS in 21m 45s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/275e1ce641154a338bd08d83cfea9753 : SUCCESS in 1h 09m 00s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/6a6dacf7af6142269991c4f95ea453c8 : SUCCESS in 1h 09m 57s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/7654d63c697042fabe924f8d44716195 : SUCCESS in 1h 37m 08s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/a796699a7b5c4724bdcef528ece3e585 : SUCCESS in 1h 21m 03s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/5bcc457bc2874965b5939bfe73f6904a : SUCCESS in 1h 45m 31s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/150690a4f8ff4f2988b19299b5a521e6 : SUCCESS in 1h 12m 39s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/e38a4390bab848d0ba530a8a3e1355f9 : SUCCESS in 1h 46m 15s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/aa23cde2deec4ef4984faae1e004133b : SUCCESS in 2m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50e5b44f6e344f3b822b607da65ba96b : SUCCESS in 3m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/243d7d964c1b4ad9a6fc0833dae4208f : SUCCESS in 3m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/503665a1d18d4b8bb8a41f267b02c061 : SUCCESS in 2m 52s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9604f15828f0439db6ffe44518afa0ba : SUCCESS in 3m 19s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/da723e34ec9c4a63a5518fe12b01c99a : SUCCESS in 2m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b301fc251c8c4f4c81e46f6cdc6edb59 : SUCCESS in 6m 20s","accounts_in_message":[],"_revision_number":1},{"id":"01ea715890578f23cbddd1926f697b6a7216e3fb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-23 07:54:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd0f773ecbe844319aa4ca0cfcc6fad1\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/0d596ea46ad345e184e7afba4e6e346c : SUCCESS in 1h 39m 38s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/f493220687eb4d9494d245c8d204e6fc : SUCCESS in 1h 43m 29s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/55455552c40e452097eeeb5dbfa55991 : SUCCESS in 1h 55m 32s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/bdc90ddc56154fae8b0d5b086f546618 : SUCCESS in 1h 51m 10s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/6272c8f3a3144de59dee3e50cfc9ecd5 : FAILURE in 1h 50m 46s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/035e945c18d542ea97a993836e0559b9 : SUCCESS in 4m 04s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4101bf7e7318404e9c69e938de2de961 : SUCCESS in 5m 03s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/3be33a66c6a247899821daaf861c91bb : SUCCESS in 6m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"beb65c98e5f4c9afca62a93d606bffdd0b6a4bb6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-03-23 10:18:55.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"fe7f2b5401da5836c89ab21427ea6868846321e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-23 11:37:41.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c52469dc75544cfead23efcbf2d2e925\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/aae51731eec44db098ea1c4f364ecd29 : RETRY_LIMIT in 2m 25s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/9809f717487e439c8565fc70872c305a : RETRY_LIMIT in 2m 00s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/3384939935514c6182512633d0175d79 : SUCCESS in 25m 26s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/bc2b9f6c4cc749268ddcfc5d43ece187 : RETRY_LIMIT in 6m 26s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/4ddc94d7b9984f28aeb332c940912908 : SUCCESS in 1h 03m 21s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d3f64c08a2ce4a6fa9167818d62d78dd : RETRY_LIMIT in 2m 02s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/aec80291eb2346e1bc2f6519ac199974 : SUCCESS in 31m 22s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/5b95609daf3641199fedceefb777da83 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06eaa59bd311446da61b49291f0ede78 : SUCCESS in 3m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2b2b10b155044c27a7e2116a26900d01 : SUCCESS in 3m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7990901fe0764ebba58a00ba0ded5074 : SUCCESS in 5m 37s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4d1785a24ef84e458c41f79ff260b4ce : SUCCESS in 4m 01s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d6f7089aa304f6da0a105c19dd11fba : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97866e8adc4048e09eb19885750e16bd : SUCCESS in 7m 13s","accounts_in_message":[],"_revision_number":2},{"id":"3b99cdf9c2dbacbfe9cba684200d470d57a26c5c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-23 13:37:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ec1bdef6edd4433a24197587d9c5107\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/180a6d030f584f108ee03939174a1fd8 : RETRY_LIMIT in 5m 01s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/518a603301654430bd443bc7691791e4 : RETRY_LIMIT in 5m 02s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/bbc7150cb8754cf790f9986f53522fca : RETRY_LIMIT in 6m 48s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/accf38248be64467937a6431c04f2f32 : RETRY_LIMIT in 4m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/54214bf692de4a7eb765c5717e19924b : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/72f991b1acff48f5a22863dc270aa67e : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/cfb9ba4e7a2e4c49815e02fb66c793ce : SUCCESS in 8m 10s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fd32e61899a798dff4a2456aa9b90cdb428f8e5f","revisions":{"7ecd5d936cb066ec9737f31f3d0a447ea56de9b7":{"kind":"REWORK","_number":1,"created":"2026-03-23 05:44:04.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/05/981705/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/05/981705/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/981705/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/981705/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/981705/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/05/981705/1"}}},"commit":{"parents":[{"commit":"482aec6a69e47034bc63bbb9d0d80d59cc710ffc","subject":"Merge \"Update Prometheus Alertmanager and exporters to the latest\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/482aec6a69e47034bc63bbb9d0d80d59cc710ffc"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-03-23 05:43:57.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-03-23 05:43:57.000000000","tz":60},"subject":"[DNM] Test a-c-k requests fix","message":"[DNM] Test a-c-k requests fix\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/981703\nChange-Id: Ic6ea2f92cb57043585eb722aeabf664589b78454\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7ecd5d936cb066ec9737f31f3d0a447ea56de9b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7ecd5d936cb066ec9737f31f3d0a447ea56de9b7"}]},"branch":"refs/heads/master"},"fd32e61899a798dff4a2456aa9b90cdb428f8e5f":{"kind":"REWORK","_number":2,"created":"2026-03-23 10:18:55.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/05/981705/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/05/981705/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/981705/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/981705/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/981705/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/05/981705/2"}}},"commit":{"parents":[{"commit":"482aec6a69e47034bc63bbb9d0d80d59cc710ffc","subject":"Merge \"Update Prometheus Alertmanager and exporters to the latest\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/482aec6a69e47034bc63bbb9d0d80d59cc710ffc"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-03-23 10:17:10.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-03-23 10:17:10.000000000","tz":60},"subject":"Ensure requests is present for docker role","message":"Ensure requests is present for docker role\n\nThe docker role uses the community.docker.docker_host_info module [1]\nwhich requires the requests library. Ensure it is present by calling the\ndocker_sdk role earlier, which installs requests.\n\n[1] https://docs.ansible.com/projects/ansible/latest/collections/community/docker/docker_host_info_module.html\n\nChange-Id: Ic6ea2f92cb57043585eb722aeabf664589b78454\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fd32e61899a798dff4a2456aa9b90cdb428f8e5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fd32e61899a798dff4a2456aa9b90cdb428f8e5f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
