)]}'
{"id":"openstack%2Fkolla~983118","triplet_id":"openstack%2Fkolla~stable%2F2025.2~Ib4ced4a04353b9c11c6f0abb7cfa91001b239e53","project":"openstack/kolla","branch":"stable/2025.2","topic":"bug/2138186","attention_set":{},"removed_from_attention_set":{"2271":{"account":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"last_update":"2026-04-07 11:14:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib4ced4a04353b9c11c6f0abb7cfa91001b239e53","subject":"Re-enable SPICE support on Ubuntu Noble.","status":"MERGED","created":"2026-04-01 19:57:59.000000000","updated":"2026-04-07 11:15:40.000000000","submitted":"2026-04-07 11:14:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"983118-bug/2138186","meta_rev_id":"416c6883bb6d03cc5e352ca7e440a8892cfbcd5d","_number":983118,"virtual_id_number":983118,"owner":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-07 11:14:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2026-04-07 09:35:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2026-04-07 10:22:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2026-04-07 09:35:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-01 21:16:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-07 09:35:45.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2026-04-07 10:22:00.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"14837430de581c5a7b8dda4044d90813155c2777","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2026-04-01 19:57:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e748894621263b91b02694c9c079de669d5bf99e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2026-04-01 20:15:25.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"3ad6eacf0c42a676eb941f32056a60a73b12e639","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2026-04-01 20:16:46.000000000","message":"Uploaded patch set 3: New patch set was added with same tree, parent tree, and commit message as Patch Set 2.","accounts_in_message":[],"_revision_number":3},{"id":"3a79f67227ea3a614a268c0d3081b1ba81b5a7c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 21:16:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4b4327cf51e4088b308eccf617362fa\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/f5a885abfa11437383c8572476a8524e : SUCCESS in 41m 21s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/8c66ef659b5b47939ba306f9b7cfa712 : SUCCESS in 50m 21s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/fc4a5ab2f38f4fb1a1a30be307d62210 : SUCCESS in 30m 25s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/70bb0a83d83446efaf2489159a5a70de : SUCCESS in 39m 49s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/dcf1579bb2d540349088e83ac3b4ce23 : SUCCESS in 36m 26s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ca7069bf805f4087a2002e2f685f5761 : SUCCESS in 37m 05s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8d01efcfcbec4a4db018abe6e76792bb : SUCCESS in 27m 33s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/ba206f0d9f4c423e913d18b2054df7ac : SUCCESS in 2m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f637a568e03e48dcb607f14aaab4ae36 : SUCCESS in 4m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2e7903c046904fccab06f145474c4e46 : SUCCESS in 2m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b23442ea6cf94bb8bf1330f679a2cf53 : SUCCESS in 3m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/712fb385301a489ab8be4b65b2f5973a : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cfdb5c091db644bf97e7bae77a578279 : SUCCESS in 3m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f996b869df264c67b21a7e962ea2cdf3 : SUCCESS in 3m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f71855ffb44542afba3f2e05248fdac8 : SUCCESS in 4m 40s","accounts_in_message":[],"_revision_number":3},{"id":"624a670f7db86a9330054ab953cdc44f2d9bdc0a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 22:28:48.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2893b40492564e6296129d7d2f1dd41c\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/738d01d23d4c493da2b2c7a7321c9f05 : SUCCESS in 51m 11s (non-voting)\n- kolla-build-debian-bookworm-arm64 https://zuul.opendev.org/t/openstack/build/4919af27487446109200b2f991d495d8 : SUCCESS in 49m 35s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/fa665f1848c14f0e805e910efd6d9980 : SUCCESS in 55m 42s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/8c6215e9e6c44d988cb755227a4d98f7 : SUCCESS in 1h 01m 26s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"dc1d75b83c97ef5f0aaff4474de085e61f30f577","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-07 09:35:45.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"78953ee53734e6334543ca1fbfddae1c08a3d6f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 09:36:15.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"24dc5ea6a7988ff5cbffe1cb1c1d80355706f520","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-07 10:22:00.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"89c7aa3f5b9a2038c5a6ea72e6ec19bbfdf55d31","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 11:14:35.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f25f4281f0342fc859c413a0ba5366b\n\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/593c0426ae7f4fed9b4a766e2c2c5ade : SUCCESS in 35m 49s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/2f75eed3caf9444e8d88c26ddcc1d804 : SUCCESS in 33m 01s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/2e4c889d8598471aaad66561867c85a4 : SUCCESS in 41m 38s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/4d7a485908cc43dda54d0e459bb27607 : SUCCESS in 36m 28s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ad98d31ec80d43e5a7e03572cf63e72d : SUCCESS in 1h 08m 22s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6022b466cf484285bff6888526931d84 : SUCCESS in 31m 56s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/420f7b3471b842509bef0ace4b9f7f9c : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5df155cfa5e411685d4836c708a5225 : SUCCESS in 3m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fd7bd71db0c743dfa60f1251b984a483 : SUCCESS in 3m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/182e96ed0e7e468eadcbce7615f02505 : SUCCESS in 2m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19b938d3c454498cb7db01c7fd81a153 : SUCCESS in 6m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a7792b9b2fc540d99beb017434539dcb : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":3},{"id":"55332aeb4227977692995dfef0b8b2bb9a07f42d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 11:14:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"416c6883bb6d03cc5e352ca7e440a8892cfbcd5d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 11:15:40.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ca23b6f15f6416894133064623aa6e1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa782f10b2ea48a782c9512280bf3ea2 : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/63319504953442fd8b7c7ce63c863e6d : SUCCESS in 49s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"91e37843fefaf46f6358093c28c966c46e6f8d2f","revisions":{"83d649bad4f7ab1b9e200dc9b3b154f6856b0995":{"kind":"REWORK","_number":1,"created":"2026-04-01 19:57:59.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/18/983118/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/18/983118/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/983118/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/983118/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/983118/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/18/983118/1"}}},"commit":{"parents":[{"commit":"25fb3945aea4000911760c22389c169a7cf8fc26","subject":"Merge \"Add Open vSwitch/OVN versions to support matrix\" into stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/25fb3945aea4000911760c22389c169a7cf8fc26"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2026-04-01 19:55:39.000000000","tz":660},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2026-04-01 19:57:56.000000000","tz":660},"subject":"Re-enable SPICE support on Ubuntu Noble.","message":"Re-enable SPICE support on Ubuntu Noble.\n\nThe move to Ubuntu Noble for Kolla 2025.1 onwards accidentally broke SPICE\nconsoles for Nova (both Horizon and spice-direct). This is because Debian\nand Ubuntu now package the SPICE support in a separate package. Without\nthis package, an attempt to start an instance in Nova requiring SPICE will\nresult in messages like this in the nova-libvirt logs:\n\n    libvirt.libvirtError: unsupported configuration: spice graphics are\n    not supported with this QEMU\n\nAnd the instance will enter an error state.\n\nThe fix is to install the `qemu-system-modules-spice` package as well in\nthe nova-libvirt image. This re-arrangement of packages does not affect\nDebian Bookworm unless backports are explicitly enabled later.\n\nCloses-Bug: #2138186\nChange-Id: Ib4ced4a04353b9c11c6f0abb7cfa91001b239e53\nSigned-off-by: Michael Still \u003cmikal@stillhq.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/83d649bad4f7ab1b9e200dc9b3b154f6856b0995"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/83d649bad4f7ab1b9e200dc9b3b154f6856b0995"}]},"branch":"refs/heads/stable/2025.2"},"283193d8b9508cf6b72272923a937ce7ec3be4f6":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-04-01 20:15:25.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/18/983118/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/18/983118/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/983118/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/983118/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/983118/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/18/983118/2"}}},"commit":{"parents":[{"commit":"25fb3945aea4000911760c22389c169a7cf8fc26","subject":"Merge \"Add Open vSwitch/OVN versions to support matrix\" into stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/25fb3945aea4000911760c22389c169a7cf8fc26"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2026-04-01 20:13:39.000000000","tz":660},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2026-04-01 20:13:39.000000000","tz":660},"subject":"Re-enable SPICE support on Ubuntu Noble.","message":"Re-enable SPICE support on Ubuntu Noble.\n\nThe move to Ubuntu Noble for Kolla 2025.1 onwards accidentally broke SPICE\nconsoles for Nova (both Horizon and spice-direct). This is because Debian\nand Ubuntu now package the SPICE support in a separate package. Without\nthis package, an attempt to start an instance in Nova requiring SPICE will\nresult in messages like this in the nova-libvirt logs:\n\n    libvirt.libvirtError: unsupported configuration: spice graphics are\n    not supported with this QEMU\n\nAnd the instance will enter an error state.\n\nThe fix is to install the `qemu-system-modules-spice` package as well in\nthe nova-libvirt image.\n\nCloses-Bug: #2138186\nChange-Id: Ib4ced4a04353b9c11c6f0abb7cfa91001b239e53\nSigned-off-by: Michael Still \u003cmikal@stillhq.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/283193d8b9508cf6b72272923a937ce7ec3be4f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/283193d8b9508cf6b72272923a937ce7ec3be4f6"}]},"branch":"refs/heads/stable/2025.2"},"91e37843fefaf46f6358093c28c966c46e6f8d2f":{"kind":"NO_CHANGE","_number":3,"created":"2026-04-01 20:16:46.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/18/983118/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/18/983118/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/983118/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/983118/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/983118/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/18/983118/3"}}},"commit":{"parents":[{"commit":"25fb3945aea4000911760c22389c169a7cf8fc26","subject":"Merge \"Add Open vSwitch/OVN versions to support matrix\" into stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/25fb3945aea4000911760c22389c169a7cf8fc26"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2026-04-01 20:16:07.000000000","tz":660},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2026-04-01 20:16:07.000000000","tz":660},"subject":"Re-enable SPICE support on Ubuntu Noble.","message":"Re-enable SPICE support on Ubuntu Noble.\n\nThe move to Ubuntu Noble for Kolla 2025.1 onwards accidentally broke SPICE\nconsoles for Nova (both Horizon and spice-direct). This is because Debian\nand Ubuntu now package the SPICE support in a separate package. Without\nthis package, an attempt to start an instance in Nova requiring SPICE will\nresult in messages like this in the nova-libvirt logs:\n\n    libvirt.libvirtError: unsupported configuration: spice graphics are\n    not supported with this QEMU\n\nAnd the instance will enter an error state.\n\nThe fix is to install the `qemu-system-modules-spice` package as well in\nthe nova-libvirt image.\n\nCloses-Bug: #2138186\nChange-Id: Ib4ced4a04353b9c11c6f0abb7cfa91001b239e53\nSigned-off-by: Michael Still \u003cmikal@stillhq.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/91e37843fefaf46f6358093c28c966c46e6f8d2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/91e37843fefaf46f6358093c28c966c46e6f8d2f"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
