)]}'
{"id":"openstack%2Fkolla~983516","triplet_id":"openstack%2Fkolla~master~If5c1905a096d776811ce932cb040e20f5dfebf82","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-04-07 11:01:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If5c1905a096d776811ce932cb040e20f5dfebf82","subject":"CI: Drop changing visibility using uri","status":"MERGED","created":"2026-04-07 07:53:40.000000000","updated":"2026-04-07 11:03:04.000000000","submitted":"2026-04-07 11:01:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"983516","meta_rev_id":"8ba4f6f9f8bcfdefd1fb9921ea157138ad0f2f42","_number":983516,"virtual_id_number":983516,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-07 11:01:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-04-07 09:28:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2026-04-07 10:22:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-04-07 09:28:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-04-07 09:28:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-04-07 09:28:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-07 09:19:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-07 09:28:42.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2026-04-07 10:22:40.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"6cc7adde4e1d8e6b8c7c292d0f25fc5faaf8ff4b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-07 07:53:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"40eced739a413ee6b367af8a94fe3dc8fb293bc1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-07 07:55:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c6a1a21dd74d96386fad5c1b9139433da93ee908","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 09:19:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9770781560514bb2b45c9db8c0d692c1\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/379030344e3c41aea724c1e40224adfd : SUCCESS in 1h 08m 40s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/e41d4a96342a4258999328b08481dc6b : SUCCESS in 1h 01m 13s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/4378f2d4e0134cb08d55a15de97d1a88 : SUCCESS in 26m 35s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/2535a8fef3f24dbf9f4f142c76896519 : SUCCESS in 1h 07m 03s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/15b5573862334593a5fb44e837fe47b2 : SUCCESS in 49m 48s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dc5839d126794239b6041ae8b584240b : SUCCESS in 1h 20m 01s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/cd43863293c54a6cb1e163703383c020 : SUCCESS in 28m 38s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/e259fbe689654988a066e36f275e45dd : SUCCESS in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9386eed65cb04c90b00d2f4a87338872 : SUCCESS in 4m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a0cdde8dedc346ce9c598a5fd992375f : SUCCESS in 2m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ca65abfbef6345d2887470ba68bbc7b2 : SUCCESS in 2m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2ce8682ffd7a432c9d2f7cafbcf38c86 : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/71bbaad8d35545379c5ed1b691c7773d : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/765015847f1241ffbfca515da25e0c18 : SUCCESS in 6m 50s","accounts_in_message":[],"_revision_number":2},{"id":"2dd554122efa45fd19a4e494caa6ada13787a1c8","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-07 09:28:42.000000000","message":"Patch Set 2: Backport-Candidate+1 Code-Review+2 Review-Priority+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6bb0f65f8de2ef31d4fb929872178dda03379c63","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 09:29:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6aca5b24ef8328a13658231be50b5c109bb6f7e4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 09:29:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c7a09bbe1054a959b87088fa6c28581\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/b61ee55cf50b4c789d1494a508527b50 : SUCCESS in 1h 28m 08s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/1554ce0db40e4cbaac6c750d310ef01a : SUCCESS in 1h 16m 08s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/87c60f3bfefe46258cbdcbcd729e5da7 : SUCCESS in 1h 30m 42s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/6b4b7315f0f04782bb49a1809b64fb33 : SUCCESS in 1h 25m 25s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ed8e43d2fcce4e3082f55381ab493684 : SUCCESS in 3m 56s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c1ec0cb7e4f54787a0bd244cb8f99c7a : SUCCESS in 4m 17s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/9f032f306ebd437a912647bf749e95bd : SUCCESS in 7m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"23033e822976b744a55716982b0524f4fc377b1f","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-07 10:22:40.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"17cfd5bedcdf82fc95a188b9658d08de13ac36a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 11:01:04.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b40fc655df3d46f3b032115dace7f534\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/0e616b68d6714769a03e0da986e62f0b : SUCCESS in 1h 27m 58s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/eab8cf8b3178484d87ef51c8514af885 : SUCCESS in 27m 58s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/be7082f481bf4b8099d4d30ee5e5ffaf : SUCCESS in 1h 10m 57s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/89b5d91b4d7f498ca63fb3567bcb0101 : SUCCESS in 38m 18s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/41db7798e6f34e5bb1a7f1342d573ae1 : SUCCESS in 1h 25m 07s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c9bcb7085eef4d18994dbf01864573e5 : SUCCESS in 28m 05s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/a8b8ad09de124497bffcbf5eb8cd7467 : SUCCESS in 2m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a02bdb2e5f2d4368aac827c5937fd275 : SUCCESS in 3m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/714ef16da9a342b9b5ee0fcc183215a9 : SUCCESS in 3m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4ef8a069cf2b497294e2f7460a53fa72 : SUCCESS in 3m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71396c658b694c3a911ddd54a88386f9 : SUCCESS in 6m 26s","accounts_in_message":[],"_revision_number":2},{"id":"e64a28947b00f4729c76b70fc3e4c58343bd4ef8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 11:01:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"8ba4f6f9f8bcfdefd1fb9921ea157138ad0f2f42","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 11:03:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d33087aacbe9459db02dbf13b2883668\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff599190606e4bc38830844af7d47f6b : SUCCESS in 46s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8745919e6ce5b9648776ab8f8a9d8947b4d1f252","revisions":{"2c59b95c328d32d26d2ca1c29fe577b6eb8319fa":{"kind":"REWORK","_number":1,"created":"2026-04-07 07:53:40.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/983516/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/983516/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/983516/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/983516/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/983516/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/983516/1"}}},"commit":{"parents":[{"commit":"fb21c890c01c91f62fd9554dbcfcb9d2c98d393e","subject":"Merge \"add venv path check when set dev mode\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fb21c890c01c91f62fd9554dbcfcb9d2c98d393e"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-07 07:51:19.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-07 07:53:37.000000000","tz":120},"subject":"CI: Drop changing visibility using uri","message":"CI: Drop changing visibility using uri\n\nIt has not been working and leaks secret\n\nRelated-Bug: #2147406\n\nChange-Id: If5c1905a096d776811ce932cb040e20f5dfebf82\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2c59b95c328d32d26d2ca1c29fe577b6eb8319fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2c59b95c328d32d26d2ca1c29fe577b6eb8319fa"}]},"branch":"refs/heads/master"},"8745919e6ce5b9648776ab8f8a9d8947b4d1f252":{"kind":"REWORK","_number":2,"created":"2026-04-07 07:55:00.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/983516/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/983516/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/983516/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/983516/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/983516/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/983516/2"}}},"commit":{"parents":[{"commit":"fb21c890c01c91f62fd9554dbcfcb9d2c98d393e","subject":"Merge \"add venv path check when set dev mode\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fb21c890c01c91f62fd9554dbcfcb9d2c98d393e"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-07 07:51:19.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-07 07:54:54.000000000","tz":120},"subject":"CI: Drop changing visibility using uri","message":"CI: Drop changing visibility using uri\n\nIt has not been working and leaks secret\n\nRelated-Bug: #2147406\n\nChange-Id: If5c1905a096d776811ce932cb040e20f5dfebf82\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8745919e6ce5b9648776ab8f8a9d8947b4d1f252"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8745919e6ce5b9648776ab8f8a9d8947b4d1f252"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d1"],"failing_atoms":["label:Backport-Candidate\u003d-1"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
