)]}'
{"id":"openstack%2Fkolla~984826","triplet_id":"openstack%2Fkolla~stable%2F2025.2~I7b0c25d361fc61992089ee88625fb3ab63c12fb4","project":"openstack/kolla","branch":"stable/2025.2","topic":"cinder-udev-stable/2025.2","attention_set":{},"removed_from_attention_set":{"32688":{"account":{"_account_id":32688,"name":"Victor Chembaev","email":"chembervint@gmail.com","username":"chembervint"},"last_update":"2026-04-20 09:15:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7b0c25d361fc61992089ee88625fb3ab63c12fb4","subject":"Return missing systemd-udev into cinder and glance containers","status":"MERGED","created":"2026-04-15 17:04:40.000000000","updated":"2026-04-20 09:17:37.000000000","submitted":"2026-04-20 09:15:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"984826-cinder-udev-stable/2025.2","cherry_pick_of_change":984396,"cherry_pick_of_patch_set":3,"meta_rev_id":"558bf35464c0dd77e782bdd9a0f73a921810f92a","_number":984826,"virtual_id_number":984826,"owner":{"_account_id":32688,"name":"Victor Chembaev","email":"chembervint@gmail.com","username":"chembervint"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-20 09:15:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2026-04-17 08:15:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2026-04-15 18:20:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2026-04-20 07:54:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2026-04-20 07:54:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-15 18:20:28.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-04-15 18:54:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-17 08:15:36.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2026-04-20 07:54:08.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"0736378d64b77ac07c81161d390e334ca2394a7f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":32688,"name":"Victor Chembaev","email":"chembervint@gmail.com","username":"chembervint"},"date":"2026-04-15 17:04:40.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* docker/glance/glance-api/Dockerfile.j2","accounts_in_message":[],"_revision_number":1},{"id":"75a12a25d20f89a8137b1cb7da9ce68a98228f8a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":32688,"name":"Victor Chembaev","email":"chembervint@gmail.com","username":"chembervint"},"date":"2026-04-15 17:07:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1e8e44ca45bb9634a2494dd6b6950263d154450e","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":32688,"name":"Victor Chembaev","email":"chembervint@gmail.com","username":"chembervint"},"date":"2026-04-15 17:07:32.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"071add05b961092bdf7087505ade0af7ccd7c202","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-15 18:20:28.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"d868d2f3b67e89bd4fb9ddf5d07f542261ed9604","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 18:54:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/30b0de351e05452281d9cb89a816870d\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/f8a85e43ea9a4e7d905d1a3b0f97cd1e : SUCCESS in 50m 35s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/2655c938328c4f6c92ce27817b112644 : SUCCESS in 41m 20s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/2c7f4819a6af42229b5f93f703843182 : SUCCESS in 16m 13s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/bd176a34883e43ee8b633be22ace8622 : SUCCESS in 43m 47s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/3f6e1cd88ade40a3b017cc2304ecd890 : POST_FAILURE in 35m 59s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8073053181914af3bca7d335ecad3669 : SUCCESS in 59m 54s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/875cef45cb46421098bb54a39d6a2431 : SUCCESS in 31m 26s\n- kolla-ansible-debian-bookworm-cephadm https://zuul.opendev.org/t/openstack/build/ed414b9412e0474a8525f702a1f6b078 : FAILURE in 1h 13m 46s (non-voting)\n- kolla-ansible-debian-bookworm-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/dcea4734b64448919807197284a916e1 : FAILURE in 41m 33s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/14766e62cb8c470a924632732f40015d : SUCCESS in 1h 46m 27s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/5a2417e0cdc7420284121c57cbc7647d : FAILURE in 55m 20s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/81206b37e33f4f0494794fe1a7357aa7 : SUCCESS in 3m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82eee26e4ff74536a9b6aa021b1c7e85 : SUCCESS in 3m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/75de2181cece49dd81fe7f562f8a4fc8 : SUCCESS in 3m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4682fa88d21541ec9802229657170ff4 : SUCCESS in 3m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6a0217c21748495d97077b68d621be42 : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/02b7df8a76ff46b6bca30459e6f8b42d : SUCCESS in 2m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/926146866ce94eaa94f039c60a1ee12e : SUCCESS in 5m 28s","accounts_in_message":[],"_revision_number":2},{"id":"1747e50abfd18f2edab0aa4079446c62b2d01735","author":{"_account_id":32688,"name":"Victor Chembaev","email":"chembervint@gmail.com","username":"chembervint"},"date":"2026-04-15 19:03:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"19e845dffa33965a07dce2d9b3b1816b3dd8c64c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 20:13:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/368926ae5b8a44d38e174bf9474faa55\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/368e6d67676e4045a47b6a81730e8858 : SUCCESS in 50m 13s (non-voting)\n- kolla-build-debian-bookworm-arm64 https://zuul.opendev.org/t/openstack/build/f02ab5b8bad34740a8e33db0061c7297 : SUCCESS in 50m 08s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/10c1d81a8337441cb30ca34103a61de8 : SUCCESS in 1h 00m 20s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/2db82d57e3c04757b0c605466009673c : SUCCESS in 59m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a7c5cdfd0e8d9211ef7b2657aa1b81b932c2be60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 20:59:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6649dfcfccee46439a028c0807a81c1f\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/55c8c79abb5546878cdc283b1c97a511 : SUCCESS in 43m 21s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/91a6e3d2a6234fe0a411ed6240f96293 : SUCCESS in 39m 25s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/f479ee18b57f497da79fa8672d54d1fe : SUCCESS in 15m 03s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/dd6f98724f6c451a9c9ff7681c1ec955 : SUCCESS in 39m 20s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/133a5ca97d40406ebdc4594f89a92f3d : SUCCESS in 38m 37s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0ecefb98910e4d4c8e17f72c5a039b27 : SUCCESS in 53m 53s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9afa4125521b4f7e8f27a10705900cf9 : SUCCESS in 33m 20s\n- kolla-ansible-debian-bookworm-cephadm https://zuul.opendev.org/t/openstack/build/5b52c6c9631e47cf96b86778c95e029d : SUCCESS in 1h 44m 06s (non-voting)\n- kolla-ansible-debian-bookworm-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/98811c71176140248c9f9bb5f951a31f : FAILURE in 1h 00m 04s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/a299e1c3aed4409cb6202619ae7b0a2d : SUCCESS in 1h 55m 23s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/66b6bd4ed9204ae0b200f781fa26ec39 : FAILURE in 57m 47s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/3ce5779209374f1ab9c4045dc64b37a3 : SUCCESS in 2m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38f3b811d36146828004a8df173795b5 : SUCCESS in 3m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5071ecba9ce345ada9d8bd92c2cef174 : SUCCESS in 3m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/40ca867e5a044a9ba13769e197bda4d7 : SUCCESS in 3m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/582584dc0a3741a7a8497bd6dfd330f7 : SUCCESS in 4m 26s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b10f28d254724126b75e73bdf6989738 : SUCCESS in 3m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b8fff97129f4c4faa803f2773a8441f : SUCCESS in 6m 18s","accounts_in_message":[],"_revision_number":2},{"id":"4ae032a9bd6af41b5a2791762c33a7f0b074a6d2","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-17 08:15:36.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c0cf9ba8b1c650cea81452d5a06873f01a4630cf","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-04-20 07:54:08.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8363e4e709c430330e2930cb359ed8e00ac569b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 07:54:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"abed31acf8b98d8461be1131d9d90b1365b67fdb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 09:15:43.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e45b2f149db4448967074d11cc09f8d\n\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/095df8b2269f431c8dd04238a509e0d9 : SUCCESS in 51m 54s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/7f7da338f4524281b6a7af735b31783f : SUCCESS in 27m 58s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/d1a6c7c92ab54d6fa44555e43437e38b : SUCCESS in 38m 19s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/7a648ecf4eb0444db30d719ab1b23864 : SUCCESS in 37m 50s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6674235c2f264a7ca5962eed4d48df93 : SUCCESS in 51m 31s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4030c79196b44810ad0545893483c616 : SUCCESS in 28m 49s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/6aaa682fbff64bf591ddcb9011adfabd : SUCCESS in 2m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b400ba28fe7248a980b4b781140a836d : SUCCESS in 4m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aaf089b3e4714dee9fc0dab786e8b9c7 : SUCCESS in 3m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/71059af63eed44a28b360130386189f2 : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bac6fb2e4a4748648cabb64dc15aaa48 : SUCCESS in 7m 34s","accounts_in_message":[],"_revision_number":2},{"id":"e107d6266ffa90603ac23708370486ebfa68ffb2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 09:15:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"558bf35464c0dd77e782bdd9a0f73a921810f92a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 09:17:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10b0bb6f09c44b0982f3a9f1f5b48922\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90d1dfc24bf24acd8e95638dac7cc4a9 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"436395ae3523ee925abac3338e63fc5822208744","revisions":{"79f1ce1f268bfe4485339a3fe6a6e9a1b67cf00d":{"kind":"REWORK","_number":1,"created":"2026-04-15 17:04:40.000000000","uploader":{"_account_id":32688,"name":"Victor Chembaev","email":"chembervint@gmail.com","username":"chembervint"},"ref":"refs/changes/26/984826/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/26/984826/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/984826/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/984826/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/984826/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/26/984826/1"}}},"commit":{"parents":[{"commit":"8ee5a4bb2430c8df3c32a0e249be47a1fa319bb2","subject":"Remove dsa from haproxy-ssh host key types","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8ee5a4bb2430c8df3c32a0e249be47a1fa319bb2"}]}],"author":{"name":"Victor Chembaev","email":"chembervint@gmail.com","date":"2026-04-13 14:03:22.000000000","tz":180},"committer":{"name":"Victor Chembaev","email":"chembervint@gmail.com","date":"2026-04-15 17:04:40.000000000","tz":0},"subject":"Return missing systemd-udev into cinder and glance containers","message":"Return missing systemd-udev into cinder and glance containers\n\nsystemd-udev package provides /usr/lib/udev/scsi_id tool,\nrequired for operating with scsi-attached storage\n\nCloses-Bug: #2148205\nChange-Id: I7b0c25d361fc61992089ee88625fb3ab63c12fb4\nSigned-off-by: Victor Chembaev \u003cchembervint@gmail.com\u003e\n(cherry picked from commit 5096d8bff2519e983cb1c1b06a96a55f1849d87a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/79f1ce1f268bfe4485339a3fe6a6e9a1b67cf00d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/79f1ce1f268bfe4485339a3fe6a6e9a1b67cf00d"}]},"branch":"refs/heads/stable/2025.2","conflicts":{"ours":"8ee5a4bb2430c8df3c32a0e249be47a1fa319bb2","theirs":"5096d8bff2519e983cb1c1b06a96a55f1849d87a","no_base_reason":"HISTORIC_DATA_WITHOUT_BASE","contains_conflicts":true}},"436395ae3523ee925abac3338e63fc5822208744":{"kind":"REWORK","_number":2,"created":"2026-04-15 17:07:18.000000000","uploader":{"_account_id":32688,"name":"Victor Chembaev","email":"chembervint@gmail.com","username":"chembervint"},"ref":"refs/changes/26/984826/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/26/984826/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/984826/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/984826/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/984826/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/26/984826/2"}}},"commit":{"parents":[{"commit":"8ee5a4bb2430c8df3c32a0e249be47a1fa319bb2","subject":"Remove dsa from haproxy-ssh host key types","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8ee5a4bb2430c8df3c32a0e249be47a1fa319bb2"}]}],"author":{"name":"Victor Chembaev","email":"chembervint@gmail.com","date":"2026-04-13 14:03:22.000000000","tz":180},"committer":{"name":"Victor Chembaev","email":"chembervint@gmail.com","date":"2026-04-15 17:07:07.000000000","tz":180},"subject":"Return missing systemd-udev into cinder and glance containers","message":"Return missing systemd-udev into cinder and glance containers\n\nsystemd-udev package provides /usr/lib/udev/scsi_id tool,\nrequired for operating with scsi-attached storage\n\nCloses-Bug: #2148205\nChange-Id: I7b0c25d361fc61992089ee88625fb3ab63c12fb4\nSigned-off-by: Victor Chembaev \u003cchembervint@gmail.com\u003e\n(cherry picked from commit 5096d8bff2519e983cb1c1b06a96a55f1849d87a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/436395ae3523ee925abac3338e63fc5822208744"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/436395ae3523ee925abac3338e63fc5822208744"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
