)]}'
{"id":"openstack%2Fkolla~984827","triplet_id":"openstack%2Fkolla~stable%2F2025.1~I7b0c25d361fc61992089ee88625fb3ab63c12fb4","project":"openstack/kolla","branch":"stable/2025.1","topic":"cinder-udev-stable/2025.1","attention_set":{},"removed_from_attention_set":{"32688":{"account":{"_account_id":32688,"name":"Victor Chembaev","email":"chembervint@gmail.com","username":"chembervint"},"last_update":"2026-04-20 10:24:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7b0c25d361fc61992089ee88625fb3ab63c12fb4","subject":"Return missing systemd-udev into cinder and glance containers","status":"MERGED","created":"2026-04-15 17:04:55.000000000","updated":"2026-04-20 10:26:21.000000000","submitted":"2026-04-20 10:24:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"984827-cinder-udev-stable/2025.1","cherry_pick_of_change":984396,"cherry_pick_of_patch_set":3,"meta_rev_id":"1351ce0e0907ce1369f885aeff60b59ab3d515ba","_number":984827,"virtual_id_number":984827,"owner":{"_account_id":32688,"name":"Victor Chembaev","email":"chembervint@gmail.com","username":"chembervint"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-20 10:24:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2026-04-17 08:15:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2026-04-15 18:20:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2026-04-20 07:54:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2026-04-20 07:54:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-15 18:20:33.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-04-15 21:40:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-17 08:15:41.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2026-04-20 07:54:13.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"f7fb09f6f281667788205ad30013c045e06f7fa1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":32688,"name":"Victor Chembaev","email":"chembervint@gmail.com","username":"chembervint"},"date":"2026-04-15 17:04:55.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* docker/glance/glance-api/Dockerfile.j2","accounts_in_message":[],"_revision_number":1},{"id":"9662e908ffa3e693b7c697096155e4e19aeb6787","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":32688,"name":"Victor Chembaev","email":"chembervint@gmail.com","username":"chembervint"},"date":"2026-04-15 17:08:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a8324dee4864942a4ce07362889a4e902cf9d06c","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":32688,"name":"Victor Chembaev","email":"chembervint@gmail.com","username":"chembervint"},"date":"2026-04-15 17:08:28.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"7e8ae5fba8d9f41631c9df27c4247e27396682b2","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-15 18:20:33.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"c8fd5477d69b7a45f8d9fb046007d3f1c703f0cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 21:40:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d31e4cca02004d3c84d11707353092cb\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/6b45eb0bea7b465385b1699bff0c292c : SUCCESS in 44m 23s (non-voting)\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/5e94eabf9e104884a212bd079acb96d9 : FAILURE in 1h 03m 35s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/cae85d0256f24d08bd067015d3353abc : SUCCESS in 46m 51s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/76114d7309964b6f80c4451637d3e876 : SUCCESS in 48m 32s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f24ea9cc1f0143c5a493d38d5a97175f : SUCCESS in 37m 41s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/e32669fccd134a7f8605d4aa9b5a770f : SUCCESS in 17m 25s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/fd38a00e2c0b4cbc96911a7550462ec1 : SUCCESS in 1h 02m 58s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/9ac24bbf05e34a3fbe3421181fe57927 : SUCCESS in 1h 32m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9d5e3dc01a7493f906f5da87d3ae862 : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be6a96b8891442dba37a6ba0bd43097f : SUCCESS in 2m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dcb34113ee7b46b3b96f5add7b4b3aa5 : SUCCESS in 3m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c8cea6584b014702bea329c34855c31e : SUCCESS in 2m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bd07fccb1cd433280f356296fd9ea26 : SUCCESS in 6m 40s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/8adce4284cd44cb1b7b42007ee603321 : SUCCESS in 3m 01s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/36594662ee1948fabb6d95aed5c6fb9d : SUCCESS in 2h 01m 04s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/aecf127574e8424fa64ea1fe9296124f : SUCCESS in 1h 33m 41s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/5288d65dfd524dd0bd6995549544cdbc : SUCCESS in 1h 30m 19s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/c737dac867094c63bb5b03cd588cfc46 : SUCCESS in 2h 08m 02s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/370b249444b941ec93fc004814dd7428 : FAILURE in 1h 32m 52s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/c449230f541748c5a8fb75f2dff3d130 : POST_FAILURE in 2h 31m 29s (non-voting)\n- kolla-build-rocky10 https://zuul.opendev.org/t/openstack/build/21c4335067294a3d85ce937a42498310 : SUCCESS in 49m 11s\n- kolla-build-rocky10-podman https://zuul.opendev.org/t/openstack/build/0cbbc289175347b09788cea7a293da7d : SUCCESS in 57m 36s\n- kolla-ansible-rocky10 https://zuul.opendev.org/t/openstack/build/81196f000a154f1f8f1099706128924c : SUCCESS in 1h 33m 01s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/31487b1e26dc42dcb97d8adaac7ff3d2 : SUCCESS in 48m 22s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/61ae9052cb84472eb76c853b9f8dc3d2 : SUCCESS in 29m 27s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/e510e03958274c36b3e0b349aba0d424 : SUCCESS in 1h 10m 43s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/af492c0d21994a51ab1bfea5ce17b593 : SUCCESS in 1h 42m 15s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/9278d29022474f8698b5621c1578263f : SUCCESS in 2h 03m 02s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/cf43e486bb4a4a72b002b67bf2632aeb : FAILURE in 2h 16m 28s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/7228c7b92c7c464bb32b2ef3b9ec347c : FAILURE in 1h 29m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ae9c04693aa888389515eb7b3c7d1c7f8bcae90d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 00:35:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b5ee77b72574f8d8112ccb6249dee5d\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/6995b760aaab4936a9534ae16dbb79ab : SUCCESS in 59m 13s (non-voting)\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/5732cac172124dca9bb2c69c79a30ea8 : SUCCESS in 1h 01m 48s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/353f0b343a3e4b20a2366b28f07cfdaf : SUCCESS in 52m 28s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c7d2996c5a1843b7a9bfd734346b7252 : SUCCESS in 1h 13m 47s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/653cbae712a54e2eaeb52472db7530e6 : FAILURE in 58m 44s (non-voting)\n- kolla-build-rocky10-aarch64 https://zuul.opendev.org/t/openstack/build/b02df55732554be99356b246912143e9 : FAILURE in 50m 51s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/07ab1dc519c94897880cad91ea731ee8 : SUCCESS in 55m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e2d4f4e0aebf796af69073eb6d249659937406ab","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-17 08:15:41.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a8b09a07f2e01d14c1e84477f7e3935f76d939bb","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-04-20 07:54:13.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"13485fa95b893e0a0ba14545805ca47f02b9e2c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 07:54:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"983e4a688abe164cc55a251284e8fa77bde58ad7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 10:24:45.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1aa3aa240ec478eb7286319c4c85b91\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/bc35303867984bae929ae349228251ad : SUCCESS in 48m 42s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/7898d51dbaa14072b6d07d366d3dc198 : SUCCESS in 29m 17s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/b747d9b8c67642e48a507f06c23821ca : SUCCESS in 1h 00m 45s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/2fc74f5b30fd44a49c1407404ca59e69 : SUCCESS in 1h 29m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93c8158f20e94d17bbce9d825bc96ec7 : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3150b12e2107491996c56dcea0766648 : SUCCESS in 3m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ac997f28c63d442584429e22457f8cd2 : SUCCESS in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b6e90b9bd714e5a80e6abbce29b3d8b : SUCCESS in 6m 44s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/d349aceb8ac84f69a6a79aa6aa01a17c : SUCCESS in 4m 47s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/d53a287e2a89498ab590f01d05d446a8 : SUCCESS in 1h 04m 33s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/0fbc780068fe483388daa060819fb6f3 : SUCCESS in 1h 24m 49s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/fc03ce97c2a24e3cbabf2b5734c157ca : SUCCESS in 1h 34m 09s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/d0d24cec25474a42b8a043337962a279 : SUCCESS in 2h 24m 48s\n- kolla-build-rocky10 https://zuul.opendev.org/t/openstack/build/9833a4ad4bc8436588af38c67d2a1f24 : SUCCESS in 44m 27s\n- kolla-build-rocky10-podman https://zuul.opendev.org/t/openstack/build/1a3a8ebea1364d82920d93b9044f1b3d : SUCCESS in 44m 35s\n- kolla-ansible-rocky10 https://zuul.opendev.org/t/openstack/build/fa6b15de06fb41be8f9d318bf520c658 : SUCCESS in 1h 08m 13s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/df89230d56c645b6aa59ce06b453516a : SUCCESS in 38m 20s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/e478a15906854858bde2c3f591b3e048 : SUCCESS in 31m 22s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/c8f49cbaa60a4c25968234440d7f9210 : SUCCESS in 1h 08m 08s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/fabe0219372d4d07812968dff59bc551 : SUCCESS in 1h 42m 41s","accounts_in_message":[],"_revision_number":2},{"id":"0af6050672b42a4332ee1c97272d717afdf31826","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 10:24:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"1351ce0e0907ce1369f885aeff60b59ab3d515ba","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 10:26:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8322f6652ee4470cbeb9889c8908d4d5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65b8a1038691433b9600405c75c565ff : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"99b84ab9b9223b10130e3b5da5c8dc00f6e01ef5","revisions":{"cbcfa2f467dee2edbce78e890561336959842027":{"kind":"REWORK","_number":1,"created":"2026-04-15 17:04:55.000000000","uploader":{"_account_id":32688,"name":"Victor Chembaev","email":"chembervint@gmail.com","username":"chembervint"},"ref":"refs/changes/27/984827/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/27/984827/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/984827/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/984827/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/984827/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/27/984827/1"}}},"commit":{"parents":[{"commit":"5930c99e85170144864cf008e57a8c2546edf2c0","subject":"Remove dsa from haproxy-ssh host key types","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5930c99e85170144864cf008e57a8c2546edf2c0"}]}],"author":{"name":"Victor Chembaev","email":"chembervint@gmail.com","date":"2026-04-13 14:03:22.000000000","tz":180},"committer":{"name":"Victor Chembaev","email":"chembervint@gmail.com","date":"2026-04-15 17:04:55.000000000","tz":0},"subject":"Return missing systemd-udev into cinder and glance containers","message":"Return missing systemd-udev into cinder and glance containers\n\nsystemd-udev package provides /usr/lib/udev/scsi_id tool,\nrequired for operating with scsi-attached storage\n\nCloses-Bug: #2148205\nChange-Id: I7b0c25d361fc61992089ee88625fb3ab63c12fb4\nSigned-off-by: Victor Chembaev \u003cchembervint@gmail.com\u003e\n(cherry picked from commit 5096d8bff2519e983cb1c1b06a96a55f1849d87a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cbcfa2f467dee2edbce78e890561336959842027"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cbcfa2f467dee2edbce78e890561336959842027"}]},"branch":"refs/heads/stable/2025.1","conflicts":{"ours":"5930c99e85170144864cf008e57a8c2546edf2c0","theirs":"5096d8bff2519e983cb1c1b06a96a55f1849d87a","no_base_reason":"HISTORIC_DATA_WITHOUT_BASE","contains_conflicts":true}},"99b84ab9b9223b10130e3b5da5c8dc00f6e01ef5":{"kind":"REWORK","_number":2,"created":"2026-04-15 17:08:14.000000000","uploader":{"_account_id":32688,"name":"Victor Chembaev","email":"chembervint@gmail.com","username":"chembervint"},"ref":"refs/changes/27/984827/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/27/984827/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/984827/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/984827/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/984827/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/27/984827/2"}}},"commit":{"parents":[{"commit":"5930c99e85170144864cf008e57a8c2546edf2c0","subject":"Remove dsa from haproxy-ssh host key types","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5930c99e85170144864cf008e57a8c2546edf2c0"}]}],"author":{"name":"Victor Chembaev","email":"chembervint@gmail.com","date":"2026-04-13 14:03:22.000000000","tz":180},"committer":{"name":"Victor Chembaev","email":"chembervint@gmail.com","date":"2026-04-15 17:08:04.000000000","tz":180},"subject":"Return missing systemd-udev into cinder and glance containers","message":"Return missing systemd-udev into cinder and glance containers\n\nsystemd-udev package provides /usr/lib/udev/scsi_id tool,\nrequired for operating with scsi-attached storage\n\nCloses-Bug: #2148205\nChange-Id: I7b0c25d361fc61992089ee88625fb3ab63c12fb4\nSigned-off-by: Victor Chembaev \u003cchembervint@gmail.com\u003e\n(cherry picked from commit 5096d8bff2519e983cb1c1b06a96a55f1849d87a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/99b84ab9b9223b10130e3b5da5c8dc00f6e01ef5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/99b84ab9b9223b10130e3b5da5c8dc00f6e01ef5"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
