)]}'
{"id":"openstack%2Fkolla~985300","triplet_id":"openstack%2Fkolla~master~I36290404c86b7e4ed0688fda87e76cf0b18a5ae4","project":"openstack/kolla","branch":"master","topic":"fluentd-rework","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-05-25 23:39:09.000000000","reason":"Change was abandoned"},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2026-05-25 23:39:09.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I36290404c86b7e4ed0688fda87e76cf0b18a5ae4","subject":"Prepare /var/log/kolla in the base start script","status":"ABANDONED","created":"2026-04-19 23:40:49.000000000","updated":"2026-05-25 23:39:09.000000000","total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"595b7efda5648ab97d726c71e3f845910646b44d","_number":985300,"virtual_id_number":985300,"owner":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-05-17 20:13:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-20 01:28:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-04-20 02:50:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-24 05:49:16.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"}],"messages":[{"id":"85e28fb8fb252d756990d43d3d1bdc6f1be3bd57","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-04-19 23:40:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f1b712a446bbae3fa42caf8fd3bcacdd26e2ed9e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-04-19 23:52:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a8baee5423c01bc5018c69692e22794e3b4a363a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 01:28:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c129c48b7bba42d1b55e6ef98770df79\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/9cab7c903a5e4e4fb594d57333eb8ca8 : SUCCESS in 1h 25m 33s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/7739ba486b0c4454a867ac3153824084 : SUCCESS in 1h 25m 10s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/4f862f7b10534da7bf6c912905c611e6 : SUCCESS in 1h 26m 47s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/558ae4f62bf3494b8dfdc28f0d3bf7ee : SUCCESS in 1h 23m 47s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/ad3933710a5d49c49e7cc8604a911bdd : FAILURE in 50m 37s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/74bcdd3f3ebf43caa22b93c8494dfb4b : SUCCESS in 4m 15s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/19884262088f4be08758fc2ca8bba638 : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/61959e9722af4b4980ce5f91bd7c12ec : SUCCESS in 7m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bc2f169d206510811dc1e4eab80fd7e7831b45d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 02:50:36.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/36e0c2be5b7f4e45a24d57b78ac398cc\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/8aca22463a81410db2b87c38b22293f0 : SUCCESS in 1h 06m 27s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/257e8a322306434fae96380823e94cd8 : SUCCESS in 1h 20m 58s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/2e98163107e349dd94b76b808ede708d : SUCCESS in 28m 50s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/215642e3fde74f4f975b9b86f7bbf4e1 : SUCCESS in 1h 09m 56s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/a4dc4d720c144cf8a3613e520f491860 : SUCCESS in 55m 27s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/25a565ac9d14449a9e87eac494147eff : SUCCESS in 1h 46m 14s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a8818daf4e94414485e9f03cbddd1660 : SUCCESS in 32m 14s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/c1d1a552f109461d9b2f8424118a1615 : FAILURE in 38m 09s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/1ce652e2eed54094887068b895c6f854 : FAILURE in 47m 41s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/b8f4a4a59f7d42da8fb74bfb30df2634 : SUCCESS in 1h 40m 02s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/5d755615ebda4988a262a4c98efbdbaa : FAILURE in 53m 25s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/2230467cb77049ed9b12b7872dd2242e : SUCCESS in 2h 02m 36s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1af9aff78eb84ecabfe7d04a5a1987a5 : FAILURE in 45m 53s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/1e28885f061647d6a946fb97c1e0fe34 : SUCCESS in 2h 02m 15s\n- kolla-ansible-rocky-10-cells https://zuul.opendev.org/t/openstack/build/7ca70820f8994f55a708d96c332def55 : FAILURE in 56m 06s (non-voting)\n- kolla-ansible-ubuntu-noble-cells https://zuul.opendev.org/t/openstack/build/9ff733c1e0434cf4b63d71d722059525 : FAILURE in 56m 10s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/b2d5d9dfef5c47d8bf8ee74bd9642517 : FAILURE in 1h 09m 27s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/83e829a7afd949c19b40290f60bf3bf3 : SUCCESS in 2h 54m 49s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/32d3f6c8db09400281f1a4251c1fbf8d : FAILURE in 1h 00m 28s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/34ee1e50c9aa4dd7b183eb2820b095ff : SUCCESS in 2h 50m 07s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/16a4bb65179549349f966c16be8adf80 : FAILURE in 1h 00m 56s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/58976b8e7e544a1bb162aa273ba8c0c9 : FAILURE in 48m 17s (non-voting)\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/7407861c1e064e6ebb4c732c9dfce790 : FAILURE in 29m 22s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/80355e73ca8d41b7992b0ad2979d0e0c : FAILURE in 27m 39s\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/39fe5519ab2542bca689820372d1f3a1 : FAILURE in 54m 31s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/7be76543bfaa4f91b35c6fa551351f59 : FAILURE in 55m 21s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/31c6df1a5ebf4c859f526e3c985df19e : FAILURE in 45m 20s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/6c4fe8c574ba489ab33024512c9812af : FAILURE in 55m 05s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/8b36a990078e43558886964b3c6b5c7f : FAILURE in 44m 11s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/3ff4a51735964166a225437863f75cd9 : SUCCESS in 2m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd415813e34e40f999d8699bce2bc4f7 : FAILURE in 2m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0fd3597382a54b99b1ff5a5507e483e7 : SUCCESS in 2m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1f8008256d9a4846b3b3309f51702458 : SUCCESS in 3m 53s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/475960db11bc4f37b051d13bb60198a7 : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/febf609fa2cc44ce82b692bfdf991232 : SUCCESS in 3m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01873f0d6a274f1185b9bc21611f1741 : SUCCESS in 7m 32s","accounts_in_message":[],"_revision_number":2},{"id":"cfeeb396c4a883514e8b847e7d1b1fd709b8c9a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-04-20 09:19:49.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"bf21c15764ec0924d669544d59ed948413f35253","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-04-20 10:02:16.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"9e04482d97496c0a4c3064b30779463fadbc7a24","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-04-20 10:40:19.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"7d26b978cf8d1db9bc59ec30a8752ae003e9b242","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-04-20 14:38:08.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"ee7966e03183d6f3a530e0cdfe70a0e488fa003d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-04-20 15:14:09.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"f0604992e909bd3c35e4a273f612b5004f2fb42d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 18:24:56.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9657e51735564d5ba6803ee737bd4466\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/1dbca63a3014450888faf3576cc7dca1 : SUCCESS in 1h 11m 59s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/2112c6cfac98414194b4c9a98abcdbd6 : SUCCESS in 1h 25m 20s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/bfd1ab55f9924798b9dd220edfac48b9 : SUCCESS in 27m 37s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/100274343efe498a8d150715bed61b2c : SUCCESS in 1h 10m 08s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/136065b2883c49658fa4c0c0767066d6 : SUCCESS in 52m 58s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/477495cb613748f88026480d9b240eba : SUCCESS in 1h 15m 16s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b8394578e2024d329f2ea56abb4f9eb6 : SUCCESS in 29m 07s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/b75267de3ee445f39186323811c4eb9d : FAILURE in 1h 03m 35s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/3dcf6b55a6bd49098fe47ca138bc93f5 : FAILURE in 1h 25m 10s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/2b846125268d4be9a13d8b8528cd856a : SUCCESS in 2h 23m 26s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/3c0b5931ce54474e8c8d21820216bf86 : FAILURE in 55m 35s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/6b9e398653204eba9218dc840180fea3 : SUCCESS in 1h 32m 27s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b3b89e7c875a4e8996904cc5f27111a0 : FAILURE in 1h 40m 06s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/ede61256d0d240c8bf66ddf05be114d0 : SUCCESS in 2h 34m 44s\n- kolla-ansible-rocky-10-cells https://zuul.opendev.org/t/openstack/build/45e8b15f459d4c739b06c30de701c85f : FAILURE in 1h 02m 51s (non-voting)\n- kolla-ansible-ubuntu-noble-cells https://zuul.opendev.org/t/openstack/build/cbe86c67cc8b44d3830372a9ea06b7c4 : FAILURE in 2h 15m 38s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/11f8c7daa85b496cadbcacdb9b5411fc : FAILURE in 1h 07m 47s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/4c21156b953e49a0bcfae78226b07eed : SUCCESS in 2h 06m 50s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/ba5670fce48f408dbb5b343792c8f4f7 : FAILURE in 2h 46m 16s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/50d5a0e9df4148e7adadd0ce9bd0cc19 : TIMED_OUT in 3h 06m 09s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/ee6af78181b34b199f5cd5f300e90c6f : FAILURE in 1h 06m 06s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/07863e364f594d1cbb36d16a08a66fba : FAILURE in 2h 12m 40s (non-voting)\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/64628700893d4d16a524d1ac4da6f345 : SUCCESS in 32m 02s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/cecc9300a1b74e3691a9d9dc80a639aa : FAILURE in 34m 33s\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/a267c2c00d5d4c9d9e8783c6bb1ef64d : FAILURE in 1h 04m 38s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/c87047e9697441a4bb7ee010ee6eae9b : FAILURE in 1h 01m 29s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/c05f2fde07d8418ebc340d5c3f86dce0 : FAILURE in 45m 40s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/b3d496b9c8a044d49e48b266e19c5f6f : FAILURE in 51m 53s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/61b5ecd4b56c42a0ba5f73e900e3ef69 : FAILURE in 29m 50s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/90bd574ca1314080990b893f6f644575 : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abbc7a28d4d045cfade54e17152895cc : SUCCESS in 4m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a5a1dc79ecf147d48e7a482a3ef43e3f : SUCCESS in 4m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5fe571b7a98c47ac8ce78e4887b53e10 : SUCCESS in 4m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/efaa6164d16042b983850f3a233bc157 : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bad6ee2332404e01aa66c381bfe13f36 : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ef3a216907844e4a371f38021446613 : SUCCESS in 6m 59s","accounts_in_message":[],"_revision_number":7},{"id":"d69f5989d8743ef7c20cd506a09758b773a60754","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 21:05:09.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3d153f0f4204086814040a27fc7020c\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/46f078ad9f9c4b2baeaaa5e0275ac67b : SUCCESS in 2h 07m 39s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/d994d5eeda894df6918e55b9cc121dc0 : SUCCESS in 1h 16m 59s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/ea4cf6b8c7e745e486b4b40d20e8ac0f : SUCCESS in 1h 31m 00s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/b3bfe7359cd249f681371d4e04825f29 : SUCCESS in 1h 16m 48s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/80836015f2ca4441846b1e726f11baf7 : FAILURE in 1h 45m 56s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7bfffee8174c4216b068c40714672aa7 : SUCCESS in 4m 32s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/31995ef5b225419c8506007f43b88c87 : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/7d02acf396864e45a6dcc9b64746d38a : SUCCESS in 8m 26s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"d3486235116f344406629c93bcc5644f38c3ba30","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-04-21 00:05:26.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"fabca5f0f04b7bae60904b483d9a0c85f07b0665","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 03:33:19.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ad0041169014a88b733db96a2550ec4\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/1bc53137360b4e48b4b13b1f198f2b20 : SUCCESS in 1h 23m 41s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/76889847d6984c61a176088c3d5fa9da : SUCCESS in 1h 31m 45s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/14366ad51efb4751b2160cd92340e896 : SUCCESS in 1h 34m 26s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/75d727272b984087af3bd0614feb9f83 : SUCCESS in 1h 37m 06s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/693a9470c0e64fd19a1720d46bb5ec58 : FAILURE in 2h 08m 46s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d455ebc678684e9f8bca271e126f3b27 : SUCCESS in 4m 00s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/12b5fca179224af4a4a3fb93dd04423d : SUCCESS in 4m 03s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/e5cf6183286d4092b337822107f04ab6 : SUCCESS in 6m 47s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"afae89869cb313dd6913008e934c2eb1b9dc4efb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 04:19:00.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cea786f2a55e42eba99cd78e93a35c77\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/6f77e2a6c8344297b8c736d63e2ee950 : SUCCESS in 1h 14m 17s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/ce429941a80446d79e415fa7aad0baea : SUCCESS in 1h 28m 52s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/d1fb4e1156bb469e85b482274d04d443 : SUCCESS in 27m 21s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/6bf8424b35c64bcf8c7354988bfa1c08 : SUCCESS in 1h 18m 08s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/b0e400b2b34a486fb76a7d8989158aab : SUCCESS in 1h 00m 56s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2ff7d834e17f42de9733cdb035ccf012 : SUCCESS in 1h 24m 23s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3c2afa7679d3498399ff567b0821e502 : SUCCESS in 17m 27s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/bfce02398d93431dab1fbaa35b0e7131 : FAILURE in 1h 21m 31s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/cf4665218c5f453e9f2023562445d21b : FAILURE in 1h 24m 59s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/d1b469eb563c4317b3634dec4d6db959 : SUCCESS in 1h 45m 22s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/5169def43d2b44668cbb00e7f0518b5a : FAILURE in 1h 13m 55s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/31842d52fb1345cc82c9fc5029818b23 : SUCCESS in 2h 17m 26s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/717c4f094a094d72acda93c890e318a1 : FAILURE in 1h 28m 59s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/48049caf32004eaab69d5dbf521816cb : SUCCESS in 2h 01m 12s\n- kolla-ansible-rocky-10-cells https://zuul.opendev.org/t/openstack/build/3ce8391b7f1b48d39ef70bcae4dcaa20 : FAILURE in 1h 33m 37s (non-voting)\n- kolla-ansible-ubuntu-noble-cells https://zuul.opendev.org/t/openstack/build/7a610466e8db42c5969602a57acbbb0b : FAILURE in 1h 33m 02s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/1b3532d4b91843d5b4179448b6470cdb : FAILURE in 1h 53m 04s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/91c86e1a5b7e4fa2981f7bf8c47649d7 : SUCCESS in 2h 58m 05s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/71640a2a30fd49b6907718bbc9aaf3ba : FAILURE in 1h 46m 40s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/8e539cb1e0024b4daa7020c6d6c8af04 : FAILURE in 1h 50m 07s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/0995aaf4a47b4d68bf010eff556bda3a : FAILURE in 1h 45m 34s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/c3c89a445d0846cab97cf9486548d646 : FAILURE in 1h 43m 00s (non-voting)\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/8d0062f694f349bc9ee7803cb6413ee4 : SUCCESS in 36m 33s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/9a6f173152fe47268968291ef1422a99 : SUCCESS in 37m 20s\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/b3d8bc1ca7ea4707824b4320e1c2795a : SUCCESS in 1h 24m 39s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/2fb50aea798e47b4ba7a05b4f6ce4ed1 : SUCCESS in 1h 38m 11s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/dd7923e205dd4f9d85d29838834c1c4b : FAILURE in 1h 35m 50s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/2900fe4f913f4bf3810bd1d1782b3373 : FAILURE in 55m 38s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/9a2d2c87934540b0b106be3f4da1a757 : FAILURE in 50m 30s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/56e8c970c7bc4141ae7008e44beaa65e : SUCCESS in 3m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2f58a119171440ba2bab4eeab1cd35e : SUCCESS in 5m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6b89c85ca68f4ab39a7b8a1f608f5f0b : SUCCESS in 3m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/45af10af2287408fb04726d555cd1062 : SUCCESS in 3m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/feb1150c127f4e9bbd9a5ae32c18dd61 : SUCCESS in 6m 47s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3e0a16b72c4246bcaa2d92aee1ab8fb5 : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a8fe5dfc0b34f739415bf60dcd7f566 : SUCCESS in 6m 41s","accounts_in_message":[],"_revision_number":8},{"id":"ba90927c6dffdee7b7a4dd3889e6ef87473407a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-04-22 16:13:03.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"e8bb435ede037dc30b5fd7b7d7da8279e50962b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-22 19:07:54.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e3f0284fda04c4eba0ff1fddecfeb0c\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/a3243af6a7c548afb2f49226b46eb296 : SUCCESS in 1h 14m 07s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/6d496fc6a82645c6905a771fd76734c0 : SUCCESS in 1h 25m 39s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/e938a14511ac403fa9de0c22fb606e1e : SUCCESS in 23m 15s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/73950334cd374fc98473cd4470bf1d8a : SUCCESS in 1h 05m 57s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/16a12b2edac74aa582183a47ad913a8a : SUCCESS in 39m 27s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/22f6367955e44457864255dbcf9e26eb : SUCCESS in 1h 22m 08s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/61d2c295d445491ebbb59168ed04d405 : SUCCESS in 31m 43s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/c9b92a19c8d547baad33a935c8b08587 : SUCCESS in 1h 20m 32s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/5123a2e6af214c708a713e50297fafac : SUCCESS in 1h 27m 45s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/145893d4a9a24c7ea7ec990979ff72eb : SUCCESS in 1h 45m 23s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/6725321af2a14269973c8824f2735343 : SUCCESS in 1h 33m 16s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/3d074a5274994875ac33f78dcdd324aa : SUCCESS in 2h 00m 18s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6b8c69a0ba9340beaae60d2a575930ee : SUCCESS in 1h 26m 04s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/e18ac18078e646968363f21d5133b4cb : SUCCESS in 1h 59m 07s\n- kolla-ansible-rocky-10-cells https://zuul.opendev.org/t/openstack/build/b9d2544457134294b21d34103a7918c0 : SUCCESS in 1h 49m 22s (non-voting)\n- kolla-ansible-ubuntu-noble-cells https://zuul.opendev.org/t/openstack/build/3fbf5f336cdc4b6b945e368111c543e0 : SUCCESS in 1h 43m 00s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/c9d70c92b012407ca3e3047a8aee91a0 : SUCCESS in 2h 14m 27s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/0dad3bf6631a4d91bb651aaad01e91ef : SUCCESS in 2h 51m 35s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/67100b446a5644ac8a57b3da6be62edf : SUCCESS in 2h 10m 54s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/a62b99d893a5422fb065d9da752d2cb1 : FAILURE in 2h 32m 15s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/70ee6cc5fb1443a0a99204a3a7d2c9e2 : SUCCESS in 1h 58m 13s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/604546bb43f64afdb19aacfaaf5096a5 : SUCCESS in 1h 58m 29s (non-voting)\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/0d9a70556ab844289348a89c397abd50 : SUCCESS in 32m 49s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/3ec679cb09f242baa25d76e1379a956b : SUCCESS in 30m 39s\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/ce13d0b8b22642f181180028de0dba16 : SUCCESS in 1h 43m 19s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/7cd64d40c19f47c59d1303835f2ca43a : SUCCESS in 1h 55m 42s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/ab80c822dcf8455ba57f882d677c0e9d : SUCCESS in 1h 32m 10s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/f74b4946856f45cc9aa85ce9cd944f9e : SUCCESS in 2h 30m 46s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/b90be0c98a0842b9a1d6a91a9612c1e0 : FAILURE in 1h 48m 38s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/f0c3033e5d764b4fbacc2fbf4e2aa758 : SUCCESS in 3m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37259c1059d941ca9e38558705b72621 : SUCCESS in 4m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ea5f130a1fbe4f7a89f7280c2d2332cc : SUCCESS in 3m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c424c6bb395942e097daa98dd8d3cfd9 : SUCCESS in 4m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e1c3559cda9840aa93df91dea034f4b1 : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c2e2612df6e243a88b1c2fa323267e72 : SUCCESS in 2m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1fd264375374593861ade1375b518a3 : SUCCESS in 6m 29s","accounts_in_message":[],"_revision_number":9},{"id":"f514f982e7d753c15c0be094c9543f01658439c7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-22 23:31:17.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10912a835aa74e46b5e8cebf8914d93b\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/3153b273fb604c568e762507ccafe98c : SUCCESS in 1h 40m 22s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/b0ba3eb11f22449f83ff80e481ddc163 : SUCCESS in 1h 47m 30s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/fdddf385e6a54762a3be461b891b1008 : SUCCESS in 1h 38m 21s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/4c3dd223c3614771b18444d43afa7d11 : SUCCESS in 1h 24m 48s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/399a4f3a108a4cccbd65318dcf3a1c92 : FAILURE in 1h 59m 46s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a7aa93336e7c46baada83c4ca9624640 : SUCCESS in 3m 51s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/19ad5c98f26442fb90266beb2ed48cc6 : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/728be1db4a074fc0938882430abadd5f : SUCCESS in 5m 29s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"45651bb5776378de67224adbf31cc31c9127247a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-04-24 04:23:21.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"f56b40f048657d1b9aab292572f05b58b62fd15e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-24 05:49:16.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"9d2c67699a5e61390b2a879fe36af49f852aa27d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-24 07:20:55.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6a9933b7fd143c89eb148f872479ac2\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/eaa7630ed0424182b6905763b7699457 : SUCCESS in 1h 19m 47s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/251a294df081476287ee4ded7a4cc58c : SUCCESS in 1h 27m 43s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/459c608447b04c0792f21eb93645087d : SUCCESS in 14m 11s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/09f55dafbb564cb9a57abb5b20a4e488 : SUCCESS in 1h 15m 12s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/09545d1616bb4319b67a7a287c0bd65f : SUCCESS in 40m 34s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/257af44d84eb4e2583512290a24d959f : SUCCESS in 1h 07m 45s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/aa772139ffaf4384aac11ad1354386c1 : SUCCESS in 16m 46s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/f9ccd213dda44de2a980efe886e4fcc0 : SUCCESS in 1h 28m 24s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/ea38fed707d144c6ac850fc1eaab4f22 : SUCCESS in 1h 57m 59s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/558c4fdc6cf84321ab5162651cf6c499 : SUCCESS in 2h 19m 42s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/6b48c7bef8be450c8276418255f6e57c : SUCCESS in 2h 01m 43s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/7aaf82a10975428d8e74cb3ffb864e4a : SUCCESS in 2h 28m 15s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e1c90bb112ad42d5b647413088d3a509 : SUCCESS in 1h 32m 18s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/46ac73b01d83448983a398d42aa45ddc : SUCCESS in 2h 05m 32s\n- kolla-ansible-debian-trixie-cells https://zuul.opendev.org/t/openstack/build/e16f4a6518914a17bb236681d00c6de3 : FAILURE in 1h 23m 59s (non-voting)\n- kolla-ansible-rocky-10-cells https://zuul.opendev.org/t/openstack/build/48809c7fbb824c608e264a4f072a7bb8 : SUCCESS in 2h 00m 37s (non-voting)\n- kolla-ansible-ubuntu-noble-cells https://zuul.opendev.org/t/openstack/build/5c75d0eb8633441c98a42503f4fc7dcc : SUCCESS in 2h 13m 40s (non-voting)\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/64c1ee212fa54a29a3686cdc1fa61155 : FAILURE in 2h 10m 08s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/8c21eecf7f0945e1929dc391b8177721 : SUCCESS in 2h 20m 08s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/9b8cfd1dc7724d369d7f4e76430d9cb8 : SUCCESS in 2h 55m 47s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/54eae3b9805c4ea9815775cd41788556 : SUCCESS in 2h 03m 48s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/0980b4b21b6944d087e7110ec9b1c15d : SUCCESS in 2h 42m 35s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/d9f7e80c4b164ab1aca21dd26f5ece2a : SUCCESS in 1h 58m 52s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/864787fdfb154f798a0c5c19f8c4928b : SUCCESS in 2h 00m 51s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/11976b4c79904891a009ab6fa4749ab3 : SUCCESS in 1h 51m 16s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/ef4ad3a8723a4c5998fbbdf873c3a83a : SUCCESS in 38m 18s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/753cfc53808b4566a08d7a294a104acd : SUCCESS in 30m 26s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/5183675301314f3eae5d9a167c55bf95 : SUCCESS in 38m 32s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/30d97d9bd7d24861ad8a1de172781bda : FAILURE in 58m 26s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/97dc610857724f08a3fa11a016e37c5f : SUCCESS in 56m 22s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/ef00a313c27c4806b79e18c7aafbae3f : SUCCESS in 1h 02m 14s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/aeac0b830aea4468bf96a41ccc9c48e8 : SUCCESS in 1h 29m 54s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/652d1907c08149c082d7da2ce9467aec : SUCCESS in 1h 29m 06s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/ee66c21fc70f45eb82fe9e32f8e448dd : FAILURE in 1h 43m 49s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/e4d4646db37a4828b9ec38db67e2da5d : SUCCESS in 2h 22m 20s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/10341e605d1e47a584f04a70cd3b6517 : FAILURE in 1h 32m 03s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/5776f26c0e67477296e8460b6be44ba1 : SUCCESS in 2m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a9d1ec579b04ee7bf5946939a980979 : SUCCESS in 3m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/81b1fc279f2d452e97d3fa7b93589dc7 : SUCCESS in 4m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/deca38389490492cb21e37eed4644f56 : SUCCESS in 2m 30s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a4395fe910874a1d84e34bed286d8561 : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0c7679bb0c4e4d349143e974bd112944 : SUCCESS in 2m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/231bcc471460432fb51801f2d46ff3e1 : SUCCESS in 6m 35s","accounts_in_message":[],"_revision_number":10},{"id":"d0ecd1a300cdc07201472607be383e48d07712c0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-24 08:20:19.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1ac8ea7630c4a19a4b1ebb27b3cb531\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/d2575e2fa799488da4e39d967c3d517c : SUCCESS in 1h 28m 43s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/9748808accc249d199415474c8b42ccc : SUCCESS in 1h 15m 45s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/cd6794a8237d4561ac7f963f4f9939e6 : SUCCESS in 1h 23m 01s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/5fcf8c7d15fe4ea4986bcd4714510598 : SUCCESS in 1h 49m 54s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/8c12b06c96b846c982bf6f4bda636328 : FAILURE in 1h 43m 08s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/33f6d154e4824cceac86880742f95a64 : SUCCESS in 1h 41m 24s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fb0525f4b9494443ba175b55d6ea23dc : SUCCESS in 3m 41s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/98d54838cc2044cfbd971fc4dbd09416 : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/71123ae896974c3bad2af7092e12e8e5 : SUCCESS in 6m 12s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"93fc86c258a790895a1d1c9c88adcf6da84ddf66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-05-12 21:49:36.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":11},{"id":"97aa478f3347a29f71eb613e7b26dd7509ccebca","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-05-12 22:11:24.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"12bae49d38b15a6f97f249b6194728a85c87cdac","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-05-12 22:29:58.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"c888762eb62c9bbacfe5b67742c13ba2599466ac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 00:01:12.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d803d0dc7004bd892e7b3783a3c5fbe\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/f8a6d6cf9fd549fbb8cca933e6c0b129 : SUCCESS in 1h 20m 25s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/8c13737235ec414da0a02c8a16f26e02 : SUCCESS in 1h 17m 29s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/a9686c5b5a3549c9b804e0cb04b66b3d : SUCCESS in 1h 21m 50s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/cc83260b386949a887b9e630fdb93742 : SUCCESS in 1h 15m 51s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/ffba577fc8114289911e79616340c215 : FAILURE in 1h 36m 28s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/cfc3a402bcc34740a9e9ac0e06b8a4fe : SUCCESS in 2h 10m 37s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b299489288a74ede9807cc006fd0a4a1 : SUCCESS in 3m 53s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f8e752e19ebd4004afff3bdb9679b0d7 : SUCCESS in 4m 18s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b94a311c34a04ebd9488c17f2d071281 : SUCCESS in 6m 35s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"4eb9c692f91fc471d918d5560a49cdc49662e8e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 00:58:00.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/47c6565134714c7b86cd7b7c18b1e1bc\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/9f9422a98c284b4ea618573943b28f11 : SUCCESS in 1h 12m 13s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/038fea700a304b9e8c19980565cb4640 : SUCCESS in 1h 19m 35s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/7529d2eb8ae547bbb2e2f758558f52a4 : SUCCESS in 20m 47s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/786241acc03040839a2b6d9f69eff682 : SUCCESS in 1h 08m 39s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/973eb8059713436d89bfec2b86cd4f4a : SUCCESS in 56m 33s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e1581af763144694b7708cf7e2062b87 : SUCCESS in 1h 05m 02s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/39ef7d0a806a43a79448e44459dc6784 : SUCCESS in 13m 42s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/a4f8e38755424d25ac06a256f5adbdb3 : SUCCESS in 1h 18m 28s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/09f45951fa6e44e99884f9680de13e7c : SUCCESS in 1h 23m 44s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/ee4cf9515c844d1a978023e08e7d355d : SUCCESS in 1h 44m 49s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/2d245808426048f8b2d7df80f110df53 : SUCCESS in 1h 33m 18s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/f2ad88722c844df38d22437a755e2083 : SUCCESS in 2h 15m 25s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f793845b49554fefa924483294898e47 : SUCCESS in 1h 34m 48s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/fc1b0be5f3444460b7e0c4b04c60bd23 : SUCCESS in 2h 07m 16s\n- kolla-ansible-debian-trixie-cells https://zuul.opendev.org/t/openstack/build/4eff6d61ed2c44cc93a91c40a0f9f803 : SUCCESS in 1h 53m 32s (non-voting)\n- kolla-ansible-rocky-10-cells https://zuul.opendev.org/t/openstack/build/808c642fb7364a7cbf74be76e485c249 : SUCCESS in 2h 01m 29s (non-voting)\n- kolla-ansible-ubuntu-noble-cells https://zuul.opendev.org/t/openstack/build/f1c1747a2ff54759ac910d2b0a91d1d7 : SUCCESS in 1h 38m 54s (non-voting)\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/3e8a0ae22cad42d881315f32dc96d134 : SUCCESS in 2h 00m 47s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/d6f66af935254dbe99efb33b44121cf5 : SUCCESS in 2h 18m 06s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/b8a240cfadc24ca2b50dd3beb72765cf : TIMED_OUT in 3h 06m 16s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/a263528f06de4f36b2caf2480d3a5082 : SUCCESS in 1h 59m 31s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/bad9ace7034a467bbdd8ef539d61a4aa : SUCCESS in 2h 49m 00s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/80fc55bb725e42758bd542e6b429df26 : SUCCESS in 1h 50m 11s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/b2780693ee4e428a900feed0bb4b9ffb : SUCCESS in 1h 57m 41s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/62f68c1d3d3946398b154237a6362806 : SUCCESS in 1h 52m 38s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/e19ecc0a792a435092fe039398d0d790 : SUCCESS in 30m 32s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/166e263439454d808f3da9c27b44efd4 : SUCCESS in 33m 04s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/250a53f2d90f41af989bc3566a042004 : SUCCESS in 21m 42s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/2c86aa723a414b98b38cde3a411429f4 : FAILURE in 1h 08m 40s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/7cc68bb2a6d2448faec7f10951b89a7a : SUCCESS in 56m 13s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/4eb94e8785c948cb8836b3ddc9dc9e80 : SUCCESS in 1h 08m 50s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/8e06c70293374dd7bbbd01ddb1fdc7b4 : SUCCESS in 1h 24m 21s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/27f6625b5b174ec3a07f78bdf1c9d231 : SUCCESS in 1h 29m 21s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/ddf9b7d8e3114d23a88fc2329b2e1f5c : FAILURE in 1h 42m 20s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/30bb04bf5a8a41dcb7d65f696262f535 : SUCCESS in 2h 20m 58s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/a066e0a513fc4a39a20238cc5df6d381 : SUCCESS in 2h 20m 54s (non-voting)\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/441de08e4be441bf94e3618e543e2726 : SUCCESS in 1h 55m 24s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/9f0d35f0e79c4064bce8d6732fde6bf1 : SUCCESS in 1h 11m 22s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/dae8952b1024433698b8d5bd46f9f7b7 : SUCCESS in 2h 01m 54s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/eb023d3adb164538a737f6833731523b : SUCCESS in 2m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6938ec41abd84a4bbb564c9e93bec339 : SUCCESS in 4m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c86f0558e7664d07a2e446760e7bd657 : SUCCESS in 2m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7e77128b87c24f3e8705526fc1b8f169 : SUCCESS in 3m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/be09c4aba4b54365a6aabc3ebf790155 : SUCCESS in 4m 00s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/600c71e8f74048469f05f913f04153c7 : SUCCESS in 3m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f1f2d16604d451aa95580a19d2525fc : SUCCESS in 6m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b4ec8e9331794a9f93d59cbf739fa78d : FAILURE in 2m 46s","accounts_in_message":[],"_revision_number":11},{"id":"30f4c0a7660a929e1188033561c46f9734fbaec0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-05-13 07:53:19.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":12},{"id":"fa0365a9f71684a352b06c224fd2e9a02821de99","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-05-13 08:31:40.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.","accounts_in_message":[],"_revision_number":13},{"id":"c52000f8be35257845e706fa48fe8c49f50ebbe8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-05-13 09:47:29.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.","accounts_in_message":[],"_revision_number":14},{"id":"16b001e10ae956533dc62aa36f42db97646fb9c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 12:59:45.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cde2d8cf6f75433396af36ed0cd3162f\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/7bc6746e7160415eac90ebed7f2be621 : SUCCESS in 1h 19m 00s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/42735ebd2fea4a07818924673fcab87e : SUCCESS in 54m 24s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/86c7a71ba1254c16bcdd14c62437a573 : SUCCESS in 26m 16s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/68ac22bfc6984e92aa389a4a6fb0e4f3 : SUCCESS in 1h 14m 35s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/a222106346d74e0dbaed80524a1b8d46 : SUCCESS in 46m 39s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/42e3834195de4fadb9509f06f0541119 : SUCCESS in 1h 22m 03s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fd5db0b112724cc686fce240e0b040bd : SUCCESS in 24m 48s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/fbd15db81ce24c08a1581e7a6630502b : SUCCESS in 1h 24m 14s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/acdc6106a2b148a588ef1c19b5e594da : SUCCESS in 1h 22m 33s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/5fd0e755b9074cddbbd64dc7f3129d57 : SUCCESS in 1h 49m 39s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/0e48de9b160a49e38b2c041e4362c871 : SUCCESS in 1h 32m 26s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/86f8671ff4994b50a09b7677d75a6c3f : SUCCESS in 2h 06m 41s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3773787e7391417ba19173e603888a34 : SUCCESS in 1h 26m 14s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/0828cccf9de14ee69d2b48ed5c4fa737 : SUCCESS in 2h 03m 18s\n- kolla-ansible-debian-trixie-cells https://zuul.opendev.org/t/openstack/build/5ca0a5fed54442a69843cf7c823581fd : SUCCESS in 1h 40m 24s (non-voting)\n- kolla-ansible-rocky-10-cells https://zuul.opendev.org/t/openstack/build/a930342c000746138d0be78e5e4cc6f2 : SUCCESS in 1h 54m 33s (non-voting)\n- kolla-ansible-ubuntu-noble-cells https://zuul.opendev.org/t/openstack/build/0ab2a50661a8422f9775de93dbe14f11 : SUCCESS in 1h 43m 24s (non-voting)\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/18b72776cc394a03a7c51097cee39b6e : SUCCESS in 1h 59m 28s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/392e281691364f758c8ea11cb1e6c5a3 : SUCCESS in 2h 13m 12s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/f6f18cd4a96c471d93a993b6b4f3ebfa : TIMED_OUT in 3h 04m 27s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/d3b374ea4c5048f6b5a816cdb1ddcd4c : FAILURE in 2h 09m 11s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/7b556fa03355415a8cddde5a01996cd4 : FAILURE in 2h 01m 14s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/265481b14362479ca1d02b7e2e2ec547 : SUCCESS in 1h 54m 14s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/f2c65fce0a1e413d9db30637004b9980 : SUCCESS in 1h 59m 15s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/548aaedaf1f2419bae944811a9997e0a : SUCCESS in 1h 53m 36s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/db035703691240e9a14ea2d7f6758ba2 : SUCCESS in 27m 16s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/59fc3d8766c0455ea7c9b082ef3d19b7 : SUCCESS in 33m 21s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/ed4d3589013c4c2aabaf95d6cec5ddf0 : SUCCESS in 33m 08s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/78748ae3fafa449bab120ea00123f335 : FAILURE in 1h 05m 54s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/6097aab265cf493bbbeba26c78aaa4ac : SUCCESS in 1h 08m 51s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/04f4722ff7044daeba9d80a3e82033e8 : SUCCESS in 1h 32m 30s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/de175ecf2f944d10b5b428cd6b4508de : SUCCESS in 1h 29m 42s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/4807995691d64fa99607ea7e337fd42a : SUCCESS in 1h 30m 46s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/a60bbbab942c408eb5629c62d64849c2 : FAILURE in 1h 44m 00s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/0d28d63ac2d84290b1d55c7f6e2dbbae : SUCCESS in 2h 33m 08s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/eaa2bdd38fc9445cb99a6258131773f4 : SUCCESS in 2h 21m 15s (non-voting)\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/7c8821908b7446b3a51abdb503fe9171 : SUCCESS in 1h 52m 06s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/cf15c8d1ff8b46a1bf6d4febda854491 : SUCCESS in 1h 03m 30s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/1afa2e0238bc45bbaea460a7a4f3369f : SUCCESS in 1h 50m 56s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/a90d1fd26e234a949a06110711a00d50 : SUCCESS in 3m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58621bfa49114faeaaa964783299ec73 : SUCCESS in 4m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ca83c17df8844e2c9b1321998fd874b6 : SUCCESS in 4m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/17d78ef82dce4cee9417b9b539399332 : SUCCESS in 4m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4d3296b5eb7e4fc99c77b03fefdf7351 : SUCCESS in 3m 44s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e81f85f092f7417e9addf201983949c5 : SUCCESS in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10c2ef9fade7432394c7a619644059df : SUCCESS in 7m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3d7d22dfb29445e3b702d3f19f9b0339 : SUCCESS in 7m 43s","accounts_in_message":[],"_revision_number":14},{"id":"d49b49f7e4ae536995a72d9d50b10c1428fe563f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 14:08:56.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/461234d0b759409a842f9c866e060ea7\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/a325a8bb6a1641dea88c1644e597b655 : SUCCESS in 1h 20m 10s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/35555b8802f4454ca46b052792093888 : SUCCESS in 1h 14m 51s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/b9ff5a2b19ea4e76b04755baab09d5d2 : SUCCESS in 1h 15m 38s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/3464f6eaa2b04ef78fa5f594f1b090fc : SUCCESS in 1h 21m 21s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/8a2db176c95443538cd1ec30a129e2f3 : FAILURE in 1h 55m 27s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/aa6769b4b240474cbd9767cc4193bc6e : SUCCESS in 1h 29m 14s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/164e1aa31f044a118113388f91fc602c : SUCCESS in 4m 09s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c6c9077b1aba4eb4ace6374d248ccae2 : SUCCESS in 4m 03s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/be947ad34d884aee8ac58c6dbafb00f0 : SUCCESS in 7m 24s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"419d63a4cf96182e01a37ffe265d06ebf9af98d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-05-14 14:24:11.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":15},{"id":"53e21b891ac5601087c17dd577780ae7efd6fd53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 17:46:11.000000000","message":"Patch Set 15: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/34794862b046411f81b0881e834126a9\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/41d3ba0432674fa79cef512fbe78eadc : SUCCESS in 1h 17m 09s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/8f9c5a6cf69342e694823001dcd7c7c5 : SUCCESS in 1h 19m 56s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/5febd8c1d1294d05ad5186fb2410d33a : SUCCESS in 29m 06s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/da6c3f7027384542be9cd14740a3c2f6 : SUCCESS in 1h 11m 29s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/f2a26a1dec3f419194d00ae5753d048d : SUCCESS in 50m 44s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/34c97cb0f8274eca8885bdcaaccc35f0 : FAILURE in 1h 16m 55s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3804a574d09d407ab9d1ef32bba1b99c : SUCCESS in 24m 23s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/ceef72729401446884aa641e8ee572c4 : SUCCESS in 1h 41m 27s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/edc0b624ee4742cab797a9e7b68065b4 : SUCCESS in 1h 23m 08s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/fd5ab3dad14a444e9f58890e44ad0911 : FAILURE in 1h 17m 35s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/ca5d212625b84966af34cf52a7eda9ba : FAILURE in 46m 15s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/79532fc41ce947b9bdc7a04dfb06e859 : SUCCESS in 2h 27m 32s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ecc75e93bee84424a913c5c63c40fdcd : SUCCESS in 1h 32m 31s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/b5ae764d83b74fe09ee8c62907b61d03 : SUCCESS in 2h 05m 05s\n- kolla-ansible-debian-trixie-cells https://zuul.opendev.org/t/openstack/build/c74594bb11a04b6ba65d1ff2d46d942d : SUCCESS in 1h 57m 13s (non-voting)\n- kolla-ansible-rocky-10-cells https://zuul.opendev.org/t/openstack/build/234d12ba5d1146268f005f32e4489bda : FAILURE in 1h 09m 29s (non-voting)\n- kolla-ansible-ubuntu-noble-cells https://zuul.opendev.org/t/openstack/build/ab5ae0c12cfa4ead94fce95b61cb0c31 : SUCCESS in 2h 17m 37s (non-voting)\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/ed0be64bd61c4d11a75870a9ef071b4f : SUCCESS in 2h 25m 25s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/5193714f526c4e3cbf67ab74ffde8d7e : SUCCESS in 2h 39m 06s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/4c011826d32b479db2d70de85198037c : FAILURE in 2h 29m 01s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/cdc167876f474ec78b0b91f95febf45e : FAILURE in 1h 56m 41s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/3974b3ad3626413fa236654b90bb5716 : TIMED_OUT in 3h 04m 19s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/6950d1e2269747bdb0e7e5a50ea0ae10 : TIMED_OUT in 2h 33m 59s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/c73be0f43f3041d5b4574068777ec483 : TIMED_OUT in 2h 33m 15s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/3d0259509ddc40099a90e69514ce62f0 : TIMED_OUT in 2h 33m 32s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/a5a896ff425046d5aec41a5fd55769de : SUCCESS in 33m 44s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/0365da6072114e01a1dae2ddf4b0a4c7 : FAILURE in 35m 46s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/48686181bdfc41b1a48ec76c23db47b6 : SUCCESS in 25m 49s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/f75735c8c2434cfbbb07962ce90904df : FAILURE in 1h 12m 12s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/439425ca096f42d78222ef3c442eb4f9 : SUCCESS in 1h 17m 37s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/4afb4dfb83b54011865b104fe4bc7993 : SUCCESS in 1h 46m 30s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/e13fd9356c044d67a5d3558abae253e1 : SUCCESS in 2h 08m 59s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/5b55d4f8db764cc2b6adcbdab1ac5def : SUCCESS in 2h 08m 25s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/600532bd0ca94b34a44f652bc164186a : FAILURE in 2h 28m 31s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/7a5a3b632c8f42bc889a66d672267d8a : TIMED_OUT in 2h 33m 28s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/3895592e4eff4e56ade025ac70dedc8e : TIMED_OUT in 2h 36m 40s (non-voting)\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/6e45f3fba1a14881b5f61233fa27bb60 : FAILURE in 1h 38m 57s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/1ca1c3e91838426dac86e07e23adb365 : SUCCESS in 1h 23m 43s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/4fee8d27a074491c9e12028bb0dbd422 : SUCCESS in 1h 33m 33s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/a829e53b1eab4f97a4e2249df221abf7 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78914f9656e2461daca80f57457fd80f : FAILURE in 3m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4d9f3bb251354a30a82f3dd8b42f13bd : SUCCESS in 3m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cc3e5db64df24bcf9dff3e14cc7a88a8 : SUCCESS in 2m 24s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5d54f2596902400fa6338ef17e82cd94 : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a1d8d8c7f06642a489cf901c17c980e0 : SUCCESS in 3m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63e9676615bc44858b6da4efcec1c01d : SUCCESS in 8m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a40a7c9d550472089a1641006a73768 : SUCCESS in 9m 55s","accounts_in_message":[],"_revision_number":15},{"id":"4187495aa1e375104a80f5fb87fef73514482764","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-05-14 18:12:50.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":16},{"id":"ad463ea761589d816b45ede43cff803e95906cb0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 22:29:10.000000000","message":"Patch Set 16: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7efa648a1bc44e9686cc293d7324d02d\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/d8690f3621ec4385b06f71e386960ff7 : SUCCESS in 1h 28m 40s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/b2e06147219f4813a0a6b7070e5b1d3d : SUCCESS in 1h 23m 47s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/a5e6a83767ab412bb06f32effa543849 : SUCCESS in 29m 10s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/b5bd5b31cfbe45fcbe875dbf6a310ebc : SUCCESS in 1h 28m 15s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/cdda384e3ab04e2da639acb5dad0341f : SUCCESS in 55m 03s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1469beecb9d9410f8438f03fc02234c7 : SUCCESS in 1h 17m 14s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6f48d05bc2e0405b8501c8f51fcb3bb3 : SUCCESS in 31m 19s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/8808db5164b1450c8b50a78c026b8e24 : SUCCESS in 1h 05m 46s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/609638667c8b47ab855bbaba5c78461a : SUCCESS in 1h 44m 35s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/50d96ee8ba2443a6baa716146850e8a1 : SUCCESS in 2h 09m 44s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/22db51e867004f23972fc2d2e7674fdd : SUCCESS in 1h 50m 01s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/9bd9deac0e13464e89c0f08a88c782ec : SUCCESS in 2h 05m 44s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fd2ee77639704c08a3b85779b82cbb7a : SUCCESS in 2h 09m 26s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/02f356455fd0434398f53df536050cf4 : SUCCESS in 2h 57m 29s\n- kolla-ansible-debian-trixie-cells https://zuul.opendev.org/t/openstack/build/2514ef8f08244329bf214b7bb8ec4bf6 : SUCCESS in 2h 13m 07s (non-voting)\n- kolla-ansible-rocky-10-cells https://zuul.opendev.org/t/openstack/build/04ff29fd1db14dcba383b716266e12b4 : SUCCESS in 1h 46m 19s (non-voting)\n- kolla-ansible-ubuntu-noble-cells https://zuul.opendev.org/t/openstack/build/0f02012694e9471bb22e12bfe55a774e : SUCCESS in 2h 09m 37s (non-voting)\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/0211393809eb453b834e7920d59fde98 : SUCCESS in 1h 48m 24s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/103872686f99484cb560ceed6ad681a6 : SUCCESS in 2h 17m 21s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/b00c2b4d8c6e4d7caa1584964067e569 : FAILURE in 1h 06m 11s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/0f7b6e93193e44e7a16f5410a208fe83 : SUCCESS in 2h 45m 37s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/7c0d71c207f64776828e85fc75583bc4 : SUCCESS in 2h 49m 59s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/50432f03b86a4851aa5e7f57f5483715 : TIMED_OUT in 2h 33m 43s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/e8681f0f70de4d7f886af7c640d3c785 : SUCCESS in 2h 02m 49s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/fec07cc2cc2742c1af03fa042975c5cc : SUCCESS in 1h 25m 13s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/b8c41ead8b4d41ef984260dcbd07543f : FAILURE in 5m 36s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/5cd627842db140acaf661c287ea8e390 : SUCCESS in 24m 02s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/cadb108310564ba9b926bdb02027cdfc : SUCCESS in 41m 31s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/95b47666e7f64b5787dc9b3dd53de32c : FAILURE in 1h 09m 31s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/408f2ae48c144a22a0e9fc250ae793bf : SUCCESS in 1h 48m 03s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/1f6d626292a3484b871012f0d570fc11 : SUCCESS in 1h 43m 14s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/943d3f8acf7e4b2fb53e530fe981a23c : TIMED_OUT in 2h 32m 12s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/f1aa4e53233a40cba8256a6b1a0b4e7b : SUCCESS in 2h 03m 03s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/991241c9893a419cb699b64e6eec16ef : FAILURE in 2h 12m 07s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/9a9216e2ac194083a49eab524be1ac03 : SUCCESS in 1h 28m 10s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/31f83605cb91439b9c2be28a5e008ada : TIMED_OUT in 2h 34m 38s (non-voting)\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/40c0b51d3b4b413dbf6d08bcc886cf11 : FAILURE in 1h 35m 38s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/03b5ef5e26d747fcb74b95f9288d5caa : SUCCESS in 1h 45m 35s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/61009e7be21f483b94a2055be1aa6f4e : TIMED_OUT in 2h 33m 15s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/c1dc610a868140c193f036dc201a87c2 : SUCCESS in 3m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e897269e0e94349a4b0a8679b5c6eba : FAILURE in 5m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ccac4c8070d74201bd27e544e5fddb34 : SUCCESS in 4m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f4f7c847dfe64121a36af40728440c03 : SUCCESS in 3m 51s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5be3ff07c4d64de48d30d41607280c3d : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4540a32bc17744d88f03a52839077185 : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c29f6887ce1d4475a1032a03d753b685 : SUCCESS in 6m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/52b38139ce4c47308c790062ffabd8f5 : SUCCESS in 11m 53s","accounts_in_message":[],"_revision_number":16},{"id":"0a589c42eca7c727e64cb5de702a2d2057efcbe2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-05-15 10:45:47.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":17},{"id":"d9568ec40a3d90fb34766d921772ae63a603cb29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-15 13:57:12.000000000","message":"Patch Set 17: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/681cde1fa611422d9dc91ea57f8ba006\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/009d5c139df542c5ae79117db3d61828 : SUCCESS in 1h 20m 58s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/71638acfb7b74a4a8a67ee0ed892c047 : SUCCESS in 1h 17m 21s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/0235b44e81ce46c5b7edf31c4e8b87c6 : SUCCESS in 23m 41s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/6a635424cdf145e2859a327e44324f25 : SUCCESS in 1h 22m 48s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/22fd3d2f32ac4eccb0e2bd3531b606fd : SUCCESS in 1h 03m 01s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e9792747460f4c8694fe190c3f584979 : SUCCESS in 1h 21m 45s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/55d03c1e7e9449759cbebb277bae4191 : SUCCESS in 31m 40s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/9154a0b5965e4a15b0f837c93c552092 : SUCCESS in 1h 29m 58s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/647b24626d50468698807317618daa96 : SUCCESS in 1h 29m 16s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/52814a4de4a944e69475f21356221577 : FAILURE in 1h 40m 36s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/7d88280ff6e043a1964320e7eb895742 : SUCCESS in 1h 38m 40s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/a63e579b2c334ca890c5a612a6ebe15b : SUCCESS in 2h 11m 58s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b59f158adf934215a3a24f97020517e1 : SUCCESS in 1h 30m 21s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/d4b22db433964e9bb1c629567b516454 : SUCCESS in 2h 03m 49s\n- kolla-ansible-debian-trixie-cells https://zuul.opendev.org/t/openstack/build/680480991a4f4e94b3ab64970d5a8365 : SUCCESS in 1h 44m 55s (non-voting)\n- kolla-ansible-rocky-10-cells https://zuul.opendev.org/t/openstack/build/1d15a7d6766c488b8764b46997d1b863 : SUCCESS in 1h 55m 07s (non-voting)\n- kolla-ansible-ubuntu-noble-cells https://zuul.opendev.org/t/openstack/build/998e7d133776486080487b460a46f9d0 : SUCCESS in 1h 52m 44s (non-voting)\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/85fd5a81bddb4f82af0700e383eb870c : FAILURE in 1h 44m 14s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/9c23e850abd341d3a6d07d8b084bf306 : SUCCESS in 2h 18m 52s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/a3529660f61e41e28006d5c8ea410c75 : TIMED_OUT in 3h 04m 42s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/29b60b999e1f40458855987c57091e54 : SUCCESS in 2h 10m 26s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/21a86eead479420b8fc3aa52f3faa279 : TIMED_OUT in 3h 05m 00s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/976a99a14c1a4351bddfb5ae5205499b : SUCCESS in 1h 57m 22s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/4f4cf9f5e81d4ad784c575e3edac1bce : SUCCESS in 2h 01m 35s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/f2cd7963a7514eeba12905f476f84091 : SUCCESS in 1h 58m 00s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/45a635c9806a4aa1b3eaafe8108b88d5 : SUCCESS in 31m 21s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/be3d5b31c1544a068a5ffa2f53f3dcc2 : SUCCESS in 35m 48s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/fd6f8ddeb6c842158a79ef922411bd56 : SUCCESS in 32m 03s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/35a394fecd41471a880ea63c431978ea : FAILURE in 1h 13m 52s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/aac66f5b8df6440aa313c0bcdfce105a : SUCCESS in 1h 32m 26s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/f03eecd47ae14de9ab2aa1fa66633a48 : SUCCESS in 1h 40m 59s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/01754bcf15b445a892902345f5af3066 : SUCCESS in 1h 27m 12s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/b5c6bdd026d7456886843c0ff9a0c0d8 : SUCCESS in 1h 29m 50s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/9e8d3d78d63449428782482367a9ec9d : FAILURE in 1h 40m 41s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/51bd0f2b9b1346848f62828a8847dca6 : SUCCESS in 2h 25m 08s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/a0c40f586a664a45aa79b57c9955b836 : SUCCESS in 2h 19m 52s (non-voting)\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/a4716265af2d4b3097dd2956e1b76303 : SUCCESS in 1h 49m 50s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/995d40772bc643a6b150516b7b0be747 : SUCCESS in 1h 40m 35s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/c7cafc228ca540eeb1d0dc483f83bd02 : SUCCESS in 2h 06m 32s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/ce096b9940ec49298101190c5f4c5a70 : SUCCESS in 3m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/637f0e45aec44e118527109d8ce828fe : FAILURE in 3m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2185dca3d0714bb184ad0250279f9d69 : SUCCESS in 3m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bf81dd024a4043a4a72683b43a48502f : SUCCESS in 3m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a7002d3bc67b48e3afa8ad833f06b8f1 : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c4ec95829494c48a6c9736e945ba210 : SUCCESS in 3m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e246e331f2d4ae4880f39479ee002d9 : SUCCESS in 6m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/914da859052e47f5b3d690ba726822f1 : SUCCESS in 8m 59s","accounts_in_message":[],"_revision_number":17},{"id":"31111ed085a115e2a4c9621b91359ccbaaeec0ac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-16 10:01:47.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ac4e011d595484da71c996871e506e7\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/558689b15be546d19144849822130cb2 : SUCCESS in 2h 52m 36s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/451954d56a0b4754aadb8dd332859ab6 : SUCCESS in 2h 33m 27s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/e9e99f416140443c9776b5f7eff2fa92 : SUCCESS in 2h 22m 58s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/7b3231924551424aa41d841bb47b0467 : SUCCESS in 2h 18m 36s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/7033eb47df62445b89987e4eb696c66b : FAILURE in 2h 12m 19s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/1a3025f852ed4a4ab87234a88b820ab6 : SUCCESS in 2h 31m 34s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a245d14ca443425bb6427fdb07db2e79 : SUCCESS in 3m 26s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/da181ca19a2a45f9aa5babf21cd7e5e1 : SUCCESS in 4m 43s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/9ccb55a9f63a4c0f9b12571d4c9e252d : SUCCESS in 11m 40s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"62401f3de8d332be9eb7f7f5a6c9087be435c418","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-05-17 17:15:53.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":18},{"id":"e740858e0a2d225ed22fa71578d2d8d392227fcc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-17 20:13:53.000000000","message":"Patch Set 18: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/16f95aff19c843639d2b0e75bea0fe65\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/64d4552e5c064f14b74a9cca9fb4d86a : SUCCESS in 1h 07m 37s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/df578b97518c4915bc7ce7d84625b1c1 : FAILURE in 16m 36s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/f1d5a8923ae14666bce2424a0e60edb8 : FAILURE in 15m 55s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/1434060467b347958f04aaac88eadb8e : SUCCESS in 1h 32m 26s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/6f895920ab4448fc922afd169f0586b2 : SUCCESS in 59m 16s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3e665ecf541248f99c00be415992be6a : SUCCESS in 1h 10m 49s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/898ecbcbf7424c63859227b06f59489b : SUCCESS in 15m 24s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/c0fb7c7cbb2144adacdce85aa942ff5d : SUCCESS in 1h 24m 17s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/8c380101887d489bb43415edca55ed55 : FAILURE in 13m 52s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/d7645abdb0024bfc9d698749cfbfbaed : SUCCESS in 1h 34m 38s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/f0b82c447fca4ff1b8ce64a70b90e120 : SUCCESS in 2h 20m 11s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e085ba78b79546fbadaa8fa29cd61c88 : SUCCESS in 1h 51m 19s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/9728f9b36e414635a38006fbee24553d : SUCCESS in 1h 45m 47s\n- kolla-ansible-debian-trixie-cells https://zuul.opendev.org/t/openstack/build/04743161d6094b28a64d8481d52308a5 : FAILURE in 12m 07s (non-voting)\n- kolla-ansible-rocky-10-cells https://zuul.opendev.org/t/openstack/build/af2d7d9ce5c74da5be00e87485f10e20 : SUCCESS in 1h 27m 31s (non-voting)\n- kolla-ansible-ubuntu-noble-cells https://zuul.opendev.org/t/openstack/build/f2bbfec655704aa297a34b30f5856572 : SUCCESS in 2h 16m 08s (non-voting)\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/674b194422244ced980c73e716f8035e : FAILURE in 8m 05s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/85f37fa2c324449c9be7ae269bdaa2d5 : SUCCESS in 1h 19m 49s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/61b627f188ce4678a17b720f6e3385c1 : SUCCESS in 2h 56m 30s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/92ee1e595fde4d70ac9b8fb055fbf14b : SUCCESS in 2h 52m 44s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/9e19b1ac350848f586db569d31b44fb6 : SUCCESS in 2h 50m 42s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/c71df04258e24149ab4929b59711e4ad : FAILURE in 17m 00s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/6ef7d5e362b44d489f23468f1fcf5b87 : SUCCESS in 1h 56m 41s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/66aa522c994141d38e07e8ddee04975d : SUCCESS in 2h 00m 39s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/055a0d718c9a4acf8d698c56a224f49e : FAILURE in 12m 45s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/95621a28b5bf48d281a53ecf80c37c9f : SUCCESS in 28m 13s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/22c9a2111f65472791c505c356ad4689 : FAILURE in 33m 51s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/3fde56b332ad48cc90570f7df5541f6e : FAILURE in 17m 16s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/f77d8ae0ca664f05b5f1528b7d42f6e5 : SUCCESS in 1h 40m 59s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/c04832b779ad4ea2ab6b004c17572b79 : SUCCESS in 1h 37m 28s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/423954ca8b944469ad1faa5286535fcd : FAILURE in 12m 04s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/d7c1abf470d04aa7a13afccdeb999884 : SUCCESS in 1h 58m 30s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/fc1aa4b24ee44c43be60628d6cd11661 : FAILURE in 23m 20s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/e410e2dbf2a14998b264d67771e73b23 : FAILURE in 1h 45m 00s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/2654e69337ae49358d970a8ce1c14fce : SUCCESS in 2h 26m 17s (non-voting)\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/d04be145b78c4089bb0a8e754bc1369b : FAILURE in 14m 05s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/486d8d2870cd496bb13a55e73f406a91 : SUCCESS in 1h 33m 49s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/6e013b64eee04e159e29b888d0c92922 : SUCCESS in 2h 08m 33s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/eb69d62d81bd4c03b6435c9c136bcd85 : SUCCESS in 2m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98bb007b985845d18ceca722987f9405 : FAILURE in 6m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1c21f94c68174305a913dbf5159be0a3 : SUCCESS in 3m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/92c42db8780b4e9e97c15244e65f7fc5 : SUCCESS in 2m 58s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f74ac835683142b2a3af1272ad47103e : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/354a71579f4843449c0d1b73d9d0d167 : SUCCESS in 3m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b136560c54104f21879dfab40a3aa2c0 : SUCCESS in 7m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/48a429b2685a4380a39dd605bc2f76b9 : SUCCESS in 9m 39s","accounts_in_message":[],"_revision_number":18},{"id":"57fab82cb3120f6de16db14a494a58686cf0c97e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-17 20:26:24.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d46242971cc4251ac387393c667d9e1\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/3c1c698d51ce44b6b43d79cf66c905a1 : TIMED_OUT in 3h 07m 54s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/6005c9da015d4b768038b67703d85ae5 : FAILURE in 34m 03s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/6a689404a97a45ed8aa4e1f6be027c17 : SUCCESS in 2h 22m 40s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/b49e4ffcae864186954fe35c0d4fb57b : TIMED_OUT in 3h 05m 53s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/068fee2e62d14bf39d4c5613c53020eb : TIMED_OUT in 2h 31m 44s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/e4a307f2dcb24614b5d5c26fb8c7ec44 : FAILURE in 24m 44s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e0103a45d11e44f99f5be6f1c305d1af : SUCCESS in 4m 22s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b83ad8d41cfa49368aec41538661960a : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/7fd48919f0c24fb882a8cf591a4f76b0 : SUCCESS in 9m 44s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"595b7efda5648ab97d726c71e3f845910646b44d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-05-25 23:39:09.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":18}],"current_revision_number":18,"current_revision":"7b3847154b4c8fa11b694ae4ba1a32f3fad7a184","revisions":{"324c5dd8041295f2b0588e0189c6f5493be3a14c":{"kind":"REWORK","_number":1,"created":"2026-04-19 23:40:49.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/00/985300/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/985300/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/985300/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/985300/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/985300/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/985300/1"}}},"commit":{"parents":[{"commit":"a9bae9fe56dced112e71de51bd9f382f00d584ff","subject":"Remove tox-py310 ansible-core version in test-reqs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a9bae9fe56dced112e71de51bd9f382f00d584ff"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-04-19 22:44:05.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-04-19 23:40:03.000000000","tz":120},"subject":"Prepare /var/log/kolla in the base start script","message":"Prepare /var/log/kolla in the base start script\n\nAdd a shared helper for preparing /var/log/kolla at container startup\nand remove duplicated per-service log directory setup from extend_start\nscripts.\n\nThis makes log directory handling consistent for bind mounts and\nvolumes, ensures the log tree uses the kolla group, and applies SGID and\ngroup-writable permissions for shared log directories.\n\nChange-Id: I36290404c86b7e4ed0688fda87e76cf0b18a5ae4\nSigned-off-by: Michal Arbet \u003cmichal.arbet@ultimum.io\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/324c5dd8041295f2b0588e0189c6f5493be3a14c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/324c5dd8041295f2b0588e0189c6f5493be3a14c"}]},"branch":"refs/heads/master"},"62cc4a005d187c8e2c48989dd1f2b4abc3fe2023":{"kind":"REWORK","_number":2,"created":"2026-04-19 23:52:57.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/00/985300/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/985300/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/985300/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/985300/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/985300/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/985300/2"}}},"commit":{"parents":[{"commit":"a9bae9fe56dced112e71de51bd9f382f00d584ff","subject":"Remove tox-py310 ansible-core version in test-reqs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a9bae9fe56dced112e71de51bd9f382f00d584ff"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-04-19 22:44:05.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-04-19 23:52:48.000000000","tz":120},"subject":"Prepare /var/log/kolla in the base start script","message":"Prepare /var/log/kolla in the base start script\n\nAdd a shared helper for preparing /var/log/kolla at container startup\nand remove duplicated per-service log directory setup from extend_start\nscripts.\n\nThis makes log directory handling consistent for bind mounts and\nvolumes, ensures the log tree uses the kolla group, and applies SGID and\ngroup-writable permissions for shared log directories.\n\nChange-Id: I36290404c86b7e4ed0688fda87e76cf0b18a5ae4\nSigned-off-by: Michal Arbet \u003cmichal.arbet@ultimum.io\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/62cc4a005d187c8e2c48989dd1f2b4abc3fe2023"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/62cc4a005d187c8e2c48989dd1f2b4abc3fe2023"}]},"branch":"refs/heads/master"},"ee87482a9d585d7b6f08d9ee286b4174a9c04fc5":{"kind":"REWORK","_number":3,"created":"2026-04-20 09:19:49.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/00/985300/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/985300/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/985300/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/985300/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/985300/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/985300/3"}}},"commit":{"parents":[{"commit":"a9bae9fe56dced112e71de51bd9f382f00d584ff","subject":"Remove tox-py310 ansible-core version in test-reqs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a9bae9fe56dced112e71de51bd9f382f00d584ff"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-04-19 22:44:05.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-04-20 09:19:33.000000000","tz":120},"subject":"Prepare /var/log/kolla in the base start script","message":"Prepare /var/log/kolla in the base start script\n\nAdd a shared helper for preparing /var/log/kolla at container startup\nand remove duplicated per-service log directory setup from extend_start\nscripts.\n\nThis makes log directory handling consistent for bind mounts and\nvolumes, ensures the log tree uses the kolla group, and applies SGID and\ngroup-writable permissions for shared log directories.\n\nChange-Id: I36290404c86b7e4ed0688fda87e76cf0b18a5ae4\nSigned-off-by: Michal Arbet \u003cmichal.arbet@ultimum.io\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ee87482a9d585d7b6f08d9ee286b4174a9c04fc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ee87482a9d585d7b6f08d9ee286b4174a9c04fc5"}]},"branch":"refs/heads/master"},"76dd53e52b231515f89b1776fc5614afc8ef7092":{"kind":"REWORK","_number":4,"created":"2026-04-20 10:02:16.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/00/985300/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/985300/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/985300/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/985300/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/985300/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/985300/4"}}},"commit":{"parents":[{"commit":"a9bae9fe56dced112e71de51bd9f382f00d584ff","subject":"Remove tox-py310 ansible-core version in test-reqs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a9bae9fe56dced112e71de51bd9f382f00d584ff"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-04-19 22:44:05.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-04-20 10:02:07.000000000","tz":120},"subject":"Prepare /var/log/kolla in the base start script","message":"Prepare /var/log/kolla in the base start script\n\nAdd a shared helper for preparing /var/log/kolla at container startup\nand remove duplicated per-service log directory setup from extend_start\nscripts.\n\nThis makes log directory handling consistent for bind mounts and\nvolumes, ensures the log tree uses the kolla group, and applies SGID and\ngroup-writable permissions for shared log directories.\n\nChange-Id: I36290404c86b7e4ed0688fda87e76cf0b18a5ae4\nSigned-off-by: Michal Arbet \u003cmichal.arbet@ultimum.io\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/76dd53e52b231515f89b1776fc5614afc8ef7092"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/76dd53e52b231515f89b1776fc5614afc8ef7092"}]},"branch":"refs/heads/master"},"ff49bdd353321f52d691f83d58756015e383662b":{"kind":"REWORK","_number":5,"created":"2026-04-20 10:40:19.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/00/985300/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/985300/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/985300/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/985300/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/985300/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/985300/5"}}},"commit":{"parents":[{"commit":"a9bae9fe56dced112e71de51bd9f382f00d584ff","subject":"Remove tox-py310 ansible-core version in test-reqs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a9bae9fe56dced112e71de51bd9f382f00d584ff"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-04-19 22:44:05.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-04-20 10:35:44.000000000","tz":120},"subject":"Prepare /var/log/kolla in the base start script","message":"Prepare /var/log/kolla in the base start script\n\nAdd a shared helper for preparing /var/log/kolla at container startup\nand remove duplicated per-service log directory setup from extend_start\nscripts.\n\nThis makes log directory handling consistent for bind mounts and\nvolumes, ensures the log tree uses the kolla group, and applies SGID and\ngroup-writable permissions for shared log directories.\n\nChange-Id: I36290404c86b7e4ed0688fda87e76cf0b18a5ae4\nSigned-off-by: Michal Arbet \u003cmichal.arbet@ultimum.io\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ff49bdd353321f52d691f83d58756015e383662b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ff49bdd353321f52d691f83d58756015e383662b"}]},"branch":"refs/heads/master"},"7507924e99341534e445e0abc152f6d9355310af":{"kind":"REWORK","_number":6,"created":"2026-04-20 14:38:08.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/00/985300/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/985300/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/985300/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/985300/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/985300/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/985300/6"}}},"commit":{"parents":[{"commit":"a9bae9fe56dced112e71de51bd9f382f00d584ff","subject":"Remove tox-py310 ansible-core version in test-reqs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a9bae9fe56dced112e71de51bd9f382f00d584ff"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-04-19 22:44:05.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-04-20 14:37:39.000000000","tz":120},"subject":"Prepare /var/log/kolla in the base start script","message":"Prepare /var/log/kolla in the base start script\n\nAdd a shared helper for preparing /var/log/kolla at container startup\nand remove duplicated per-service log directory setup from extend_start\nscripts.\n\nThis makes log directory handling consistent for bind mounts and\nvolumes, ensures the log tree uses the kolla group, and applies SGID and\ngroup-writable permissions for shared log directories.\n\nChange-Id: I36290404c86b7e4ed0688fda87e76cf0b18a5ae4\nSigned-off-by: Michal Arbet \u003cmichal.arbet@ultimum.io\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7507924e99341534e445e0abc152f6d9355310af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7507924e99341534e445e0abc152f6d9355310af"}]},"branch":"refs/heads/master"},"fd47abce29305e1faa897245a6c42c55cbfff36e":{"kind":"REWORK","_number":7,"created":"2026-04-20 15:14:09.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/00/985300/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/985300/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/985300/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/985300/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/985300/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/985300/7"}}},"commit":{"parents":[{"commit":"a9bae9fe56dced112e71de51bd9f382f00d584ff","subject":"Remove tox-py310 ansible-core version in test-reqs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a9bae9fe56dced112e71de51bd9f382f00d584ff"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-04-19 22:44:05.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-04-20 15:13:57.000000000","tz":120},"subject":"Prepare /var/log/kolla in the base start script","message":"Prepare /var/log/kolla in the base start script\n\nAdd a shared helper for preparing /var/log/kolla at container startup\nand remove duplicated per-service log directory setup from extend_start\nscripts.\n\nThis makes log directory handling consistent for bind mounts and\nvolumes, ensures the log tree uses the kolla group, and applies SGID and\ngroup-writable permissions for shared log directories.\n\nChange-Id: I36290404c86b7e4ed0688fda87e76cf0b18a5ae4\nSigned-off-by: Michal Arbet \u003cmichal.arbet@ultimum.io\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fd47abce29305e1faa897245a6c42c55cbfff36e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fd47abce29305e1faa897245a6c42c55cbfff36e"}]},"branch":"refs/heads/master"},"93486068634ff82cdd1fca0e3d6da542666f6321":{"kind":"REWORK","_number":8,"created":"2026-04-21 00:05:26.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/00/985300/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/985300/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/985300/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/985300/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/985300/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/985300/8"}}},"commit":{"parents":[{"commit":"a9bae9fe56dced112e71de51bd9f382f00d584ff","subject":"Remove tox-py310 ansible-core version in test-reqs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a9bae9fe56dced112e71de51bd9f382f00d584ff"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-04-19 22:44:05.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-04-21 00:05:11.000000000","tz":120},"subject":"Prepare /var/log/kolla in the base start script","message":"Prepare /var/log/kolla in the base start script\n\nAdd a shared helper for preparing /var/log/kolla at container startup\nand remove duplicated per-service log directory setup from extend_start\nscripts.\n\nThis makes log directory handling consistent for bind mounts and\nvolumes, ensures the log tree uses the kolla group, and applies SGID and\ngroup-writable permissions for shared log directories.\n\nChange-Id: I36290404c86b7e4ed0688fda87e76cf0b18a5ae4\nSigned-off-by: Michal Arbet \u003cmichal.arbet@ultimum.io\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/93486068634ff82cdd1fca0e3d6da542666f6321"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/93486068634ff82cdd1fca0e3d6da542666f6321"}]},"branch":"refs/heads/master"},"c098f40a37b970910d9dbc7a3aaa54acf8582fac":{"kind":"REWORK","_number":9,"created":"2026-04-22 16:13:03.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/00/985300/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/985300/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/985300/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/985300/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/985300/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/985300/9"}}},"commit":{"parents":[{"commit":"a9bae9fe56dced112e71de51bd9f382f00d584ff","subject":"Remove tox-py310 ansible-core version in test-reqs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a9bae9fe56dced112e71de51bd9f382f00d584ff"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-04-19 22:44:05.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-04-22 15:58:08.000000000","tz":120},"subject":"Prepare /var/log/kolla in the base start script","message":"Prepare /var/log/kolla in the base start script\n\nAdd a shared helper for preparing /var/log/kolla at container startup\nand remove duplicated per-service log directory setup from extend_start\nscripts.\n\nThis makes log directory handling consistent for bind mounts and\nvolumes, ensures the log tree uses the kolla group, and applies SGID and\ngroup-writable permissions for shared log directories.\n\nChange-Id: I36290404c86b7e4ed0688fda87e76cf0b18a5ae4\nSigned-off-by: Michal Arbet \u003cmichal.arbet@ultimum.io\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c098f40a37b970910d9dbc7a3aaa54acf8582fac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c098f40a37b970910d9dbc7a3aaa54acf8582fac"}]},"branch":"refs/heads/master"},"96dc9ddf2524e841c36b5b031d54f45904e4852d":{"kind":"REWORK","_number":10,"created":"2026-04-24 04:23:21.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/00/985300/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/985300/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/985300/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/985300/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/985300/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/985300/10"}}},"commit":{"parents":[{"commit":"4a7e5ee0dd0fe3ea31a3e6d59a51fe0ca72bda2c","subject":"CI: sync Debian scenario jobs with kolla-ansible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4a7e5ee0dd0fe3ea31a3e6d59a51fe0ca72bda2c"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-04-19 22:44:05.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-04-24 04:11:32.000000000","tz":120},"subject":"Prepare /var/log/kolla in the base start script","message":"Prepare /var/log/kolla in the base start script\n\nAdd a shared helper for preparing /var/log/kolla at container startup\nand remove duplicated per-service log directory setup from extend_start\nscripts.\n\nThis makes log directory handling consistent for bind mounts and\nvolumes, ensures the log tree uses the kolla group, and applies SGID and\ngroup-writable permissions for shared log directories.\n\nChange-Id: I36290404c86b7e4ed0688fda87e76cf0b18a5ae4\nSigned-off-by: Michal Arbet \u003cmichal.arbet@ultimum.io\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/96dc9ddf2524e841c36b5b031d54f45904e4852d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/96dc9ddf2524e841c36b5b031d54f45904e4852d"}]},"branch":"refs/heads/master"},"9d78f724698a0e51d6fe3f5d6a38cdc3e4f3d15b":{"kind":"REWORK","_number":11,"created":"2026-05-12 21:49:36.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/00/985300/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/985300/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/985300/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/985300/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/985300/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/985300/11"}}},"commit":{"parents":[{"commit":"1ff4715aa6d7e2ecb84ec7a608b87b3ba95f4ea6","subject":"Merge \"ironic-pxe: Remove httpd setup call\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1ff4715aa6d7e2ecb84ec7a608b87b3ba95f4ea6"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-04-19 22:44:05.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-05-12 21:49:13.000000000","tz":120},"subject":"Prepare /var/log/kolla in the base start script","message":"Prepare /var/log/kolla in the base start script\n\nAdd a shared helper for preparing /var/log/kolla at container startup\nand remove duplicated per-service log directory setup from extend_start\nscripts.\n\nThis makes log directory handling consistent for bind mounts and\nvolumes, ensures the log tree uses the kolla group, and applies SGID and\ngroup-writable permissions for shared log directories.\n\nChange-Id: I36290404c86b7e4ed0688fda87e76cf0b18a5ae4\nSigned-off-by: Michal Arbet \u003cmichal.arbet@ultimum.io\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9d78f724698a0e51d6fe3f5d6a38cdc3e4f3d15b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9d78f724698a0e51d6fe3f5d6a38cdc3e4f3d15b"}]},"branch":"refs/heads/master"},"c810f8b92ddacecddab0c9a19d1d9abb791a33af":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2026-05-13 07:53:19.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/00/985300/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/985300/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/985300/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/985300/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/985300/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/985300/12"}}},"commit":{"parents":[{"commit":"f7654078378b446384b19f077c1a1531e3ec23e3","subject":"Fix the number of characters in the lines [trivial]","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f7654078378b446384b19f077c1a1531e3ec23e3"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-04-19 22:44:05.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-05-13 07:53:09.000000000","tz":120},"subject":"Prepare /var/log/kolla in the base start script","message":"Prepare /var/log/kolla in the base start script\n\nAdd a shared helper for preparing /var/log/kolla at container startup\nand remove duplicated per-service log directory setup from extend_start\nscripts.\n\nThis makes log directory handling consistent for bind mounts and\nvolumes, ensures the log tree uses the kolla group, and applies SGID and\ngroup-writable permissions for shared log directories.\n\nChange-Id: I36290404c86b7e4ed0688fda87e76cf0b18a5ae4\nSigned-off-by: Michal Arbet \u003cmichal.arbet@ultimum.io\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c810f8b92ddacecddab0c9a19d1d9abb791a33af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c810f8b92ddacecddab0c9a19d1d9abb791a33af"}]},"branch":"refs/heads/master"},"4874998d47036ef9de4a208fa5f70ecbafee4fed":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2026-05-13 08:31:40.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/00/985300/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/985300/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/985300/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/985300/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/985300/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/985300/13"}}},"commit":{"parents":[{"commit":"bc2a7ae0b790df435882adafe114424cd1cb81d2","subject":"Ignore D001 Line too long errors in release notes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bc2a7ae0b790df435882adafe114424cd1cb81d2"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-04-19 22:44:05.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-05-13 08:31:22.000000000","tz":120},"subject":"Prepare /var/log/kolla in the base start script","message":"Prepare /var/log/kolla in the base start script\n\nAdd a shared helper for preparing /var/log/kolla at container startup\nand remove duplicated per-service log directory setup from extend_start\nscripts.\n\nThis makes log directory handling consistent for bind mounts and\nvolumes, ensures the log tree uses the kolla group, and applies SGID and\ngroup-writable permissions for shared log directories.\n\nChange-Id: I36290404c86b7e4ed0688fda87e76cf0b18a5ae4\nSigned-off-by: Michal Arbet \u003cmichal.arbet@ultimum.io\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4874998d47036ef9de4a208fa5f70ecbafee4fed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4874998d47036ef9de4a208fa5f70ecbafee4fed"}]},"branch":"refs/heads/master"},"4a586632fb0170ddb31901534b0241ff9b6013c7":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2026-05-13 09:47:29.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/00/985300/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/985300/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/985300/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/985300/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/985300/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/985300/14"}}},"commit":{"parents":[{"commit":"5c3d61abbbc3b5c0c484f5054def1ba294145eed","subject":"Ignore D001 Line too long errors in release notes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5c3d61abbbc3b5c0c484f5054def1ba294145eed"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-04-19 22:44:05.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-05-13 09:47:01.000000000","tz":120},"subject":"Prepare /var/log/kolla in the base start script","message":"Prepare /var/log/kolla in the base start script\n\nAdd a shared helper for preparing /var/log/kolla at container startup\nand remove duplicated per-service log directory setup from extend_start\nscripts.\n\nThis makes log directory handling consistent for bind mounts and\nvolumes, ensures the log tree uses the kolla group, and applies SGID and\ngroup-writable permissions for shared log directories.\n\nChange-Id: I36290404c86b7e4ed0688fda87e76cf0b18a5ae4\nSigned-off-by: Michal Arbet \u003cmichal.arbet@ultimum.io\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4a586632fb0170ddb31901534b0241ff9b6013c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4a586632fb0170ddb31901534b0241ff9b6013c7"}]},"branch":"refs/heads/master"},"e9925e897deea70da34bcd9bf0946bccc2d7710d":{"kind":"REWORK","_number":15,"created":"2026-05-14 14:24:11.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/00/985300/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/985300/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/985300/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/985300/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/985300/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/985300/15"}}},"commit":{"parents":[{"commit":"129e88848d511ebdac4e42124e1192bc5c38d4b5","subject":"Merge \"Update mission statement in docs index\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/129e88848d511ebdac4e42124e1192bc5c38d4b5"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-04-19 22:44:05.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-05-14 14:23:22.000000000","tz":120},"subject":"Prepare /var/log/kolla in the base start script","message":"Prepare /var/log/kolla in the base start script\n\nAdd a shared helper for preparing /var/log/kolla at container startup\nand remove duplicated per-service log directory setup from extend_start\nscripts.\n\nThis makes log directory handling consistent for bind mounts and\nvolumes, ensures the log tree uses the kolla group, and applies SGID and\ngroup-writable permissions for shared log directories.\n\nChange-Id: I36290404c86b7e4ed0688fda87e76cf0b18a5ae4\nSigned-off-by: Michal Arbet \u003cmichal.arbet@ultimum.io\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e9925e897deea70da34bcd9bf0946bccc2d7710d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e9925e897deea70da34bcd9bf0946bccc2d7710d"}]},"branch":"refs/heads/master"},"806592d7f4d5ebfd8b809a846b22974f8dfd8076":{"kind":"REWORK","_number":16,"created":"2026-05-14 18:12:50.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/00/985300/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/985300/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/985300/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/985300/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/985300/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/985300/16"}}},"commit":{"parents":[{"commit":"129e88848d511ebdac4e42124e1192bc5c38d4b5","subject":"Merge \"Update mission statement in docs index\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/129e88848d511ebdac4e42124e1192bc5c38d4b5"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-04-19 22:44:05.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-05-14 18:12:37.000000000","tz":120},"subject":"Prepare /var/log/kolla in the base start script","message":"Prepare /var/log/kolla in the base start script\n\nAdd a shared helper for preparing /var/log/kolla at container startup\nand remove duplicated per-service log directory setup from extend_start\nscripts.\n\nThis makes log directory handling consistent for bind mounts and\nvolumes, ensures the log tree uses the kolla group, and applies SGID and\ngroup-writable permissions for shared log directories.\n\nChange-Id: I36290404c86b7e4ed0688fda87e76cf0b18a5ae4\nSigned-off-by: Michal Arbet \u003cmichal.arbet@ultimum.io\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/806592d7f4d5ebfd8b809a846b22974f8dfd8076"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/806592d7f4d5ebfd8b809a846b22974f8dfd8076"}]},"branch":"refs/heads/master"},"db748a2cf0f4e0cfadc05e72c36c310e6498fa3e":{"kind":"REWORK","_number":17,"created":"2026-05-15 10:45:47.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/00/985300/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/985300/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/985300/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/985300/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/985300/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/985300/17"}}},"commit":{"parents":[{"commit":"129e88848d511ebdac4e42124e1192bc5c38d4b5","subject":"Merge \"Update mission statement in docs index\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/129e88848d511ebdac4e42124e1192bc5c38d4b5"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-04-19 22:44:05.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-05-15 10:43:16.000000000","tz":120},"subject":"Prepare /var/log/kolla in the base start script","message":"Prepare /var/log/kolla in the base start script\n\nAdd a shared helper for preparing /var/log/kolla at container startup\nand remove duplicated per-service log directory setup from extend_start\nscripts.\n\nThis makes log directory handling consistent for bind mounts and\nvolumes, ensures the log tree uses the kolla group, and applies SGID and\ngroup-writable permissions for shared log directories.\n\nChange-Id: I36290404c86b7e4ed0688fda87e76cf0b18a5ae4\nSigned-off-by: Michal Arbet \u003cmichal.arbet@ultimum.io\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/db748a2cf0f4e0cfadc05e72c36c310e6498fa3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/db748a2cf0f4e0cfadc05e72c36c310e6498fa3e"}]},"branch":"refs/heads/master"},"7b3847154b4c8fa11b694ae4ba1a32f3fad7a184":{"kind":"REWORK","_number":18,"created":"2026-05-17 17:15:53.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/00/985300/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/985300/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/985300/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/985300/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/985300/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/985300/18"}}},"commit":{"parents":[{"commit":"129e88848d511ebdac4e42124e1192bc5c38d4b5","subject":"Merge \"Update mission statement in docs index\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/129e88848d511ebdac4e42124e1192bc5c38d4b5"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-04-19 22:44:05.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-05-17 17:15:42.000000000","tz":120},"subject":"Prepare /var/log/kolla in the base start script","message":"Prepare /var/log/kolla in the base start script\n\nAdd a shared helper for preparing /var/log/kolla at container startup\nand remove duplicated per-service log directory setup from extend_start\nscripts.\n\nThis makes log directory handling consistent for bind mounts and\nvolumes, ensures the log tree uses the kolla group, and applies SGID and\ngroup-writable permissions for shared log directories.\n\nChange-Id: I36290404c86b7e4ed0688fda87e76cf0b18a5ae4\nSigned-off-by: Michal Arbet \u003cmichal.arbet@ultimum.io\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7b3847154b4c8fa11b694ae4ba1a32f3fad7a184"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7b3847154b4c8fa11b694ae4ba1a32f3fad7a184"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
