)]}'
{"id":"openstack%2Fkolla~985940","triplet_id":"openstack%2Fkolla~master~Ia1e3dbac1ac104eceba79e6387d9692cc9ac56ee","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"17669":{"account":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"last_update":"2026-05-13 14:32:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia1e3dbac1ac104eceba79e6387d9692cc9ac56ee","subject":"Drop ironic-inspector from Zuul file trigger","status":"MERGED","created":"2026-04-23 11:12:31.000000000","updated":"2026-05-13 14:32:56.000000000","submitted":"2026-05-13 14:32:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"985940","meta_rev_id":"c9d38a42a150c0e13fe77f029149d4e844c7fde2","_number":985940,"virtual_id_number":985940,"owner":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-13 14:32:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2026-05-13 12:24:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2026-05-13 12:24:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2026-05-13 12:24:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-23 11:25:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-13 12:24:33.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"4a06c21e84ec3a3ad281a508a257dded53d5e1d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2026-04-23 11:12:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c72b2d09c07c7f37e0e281150f4efbe84f94255b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-23 11:25:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53a42099b3fc462aa4258c0b50d3900b\n\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/0f956ba9e95c452c9aa488dc3ef65086 : SUCCESS in 2m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58b0c9231bec48ebbf0615487e507f62 : SUCCESS in 3m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1435ebd38cdd4117bad86d17f4d16f53 : SUCCESS in 3m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6447f742e7d64ddab591a7daa46c21d9 : SUCCESS in 4m 29s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/721fd9ae1a554ebf9b1a1434990b17a0 : SUCCESS in 7m 18s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4739edfa3ffa41a0af6d72717fda27e4 : SUCCESS in 3m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a09b98139c0a42d09e60dc3724c04e0e : SUCCESS in 6m 31s","accounts_in_message":[],"_revision_number":1},{"id":"4038e007e1caa23472c40bc40ba8d878baf051b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2026-04-23 13:02:44.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"ad02b664327fccb238332ebd3c2093fdfc97ea6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-23 13:13:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cea769dbe57d41378bf26a3e05456d79\n\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/c34c83e75aa34e3eb5862dfe73ad59a6 : SUCCESS in 2m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b53289ca88f84667a45b5516aae762bb : SUCCESS in 4m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dcb6d8dc82cc43fb9a9ffbd179478ae0 : SUCCESS in 3m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/edfd7c432b2e4ac1936ed6ebb22f46e1 : SUCCESS in 3m 09s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0ade3b0295fb48debf8855ffd370b918 : SUCCESS in 4m 15s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/726d7db2c8674083bcf1af609249d7d6 : SUCCESS in 2m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71ec2f4b912140aeab3159c76194cb16 : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":2},{"id":"dc5cea1e08325cdb4066358f04c073b354f93ac2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-23 13:48:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26b7e26ea8454b26af1d76ad02b4fb25\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a14354bd1dea4a92b976d8c2d1e7c237 : SUCCESS in 3m 13s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5461d19a1dd34832b6c58c64161ee33b : SUCCESS in 4m 54s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/94df057cdc814ba5995fffaaa489f668 : SUCCESS in 6m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"48f078b7e7ceb3753ea066f6a435e936d8e17ce5","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-13 12:24:33.000000000","message":"Patch Set 2: Backport-Candidate+1 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a1b47cc46864a3fc89a6e232eaa04f1c08911894","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 12:25:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6657e73e53947f58d72e2f4ea113623935e554ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 14:32:56.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/713c5323ecea475887bf5c576ec5f852\n\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/b52a7db4e2ce405ca34b34be78aef95f : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2543893d8c2f4087bb946999a9bb555f : SUCCESS in 2m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9476e9610f2e42e3a0f7d4ac2eba04ec : SUCCESS in 2m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1d69dfeee2e940628a479822b19986e6 : SUCCESS in 3m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/991a5a8ee9e44e86b2472bade74605ae : SUCCESS in 6m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/39af5656411744b2bbf06c9272f370b7 : SUCCESS in 9m 09s","accounts_in_message":[],"_revision_number":2},{"id":"c9d38a42a150c0e13fe77f029149d4e844c7fde2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 14:32:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f931316818755dc9e82e026ec8a14b1649a6c8ae","revisions":{"5ff3271b2464502f74737cfac831957c0c6cfc6c":{"kind":"REWORK","_number":1,"created":"2026-04-23 11:12:31.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/40/985940/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/40/985940/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/985940/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/985940/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/985940/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/40/985940/1"}}},"commit":{"parents":[{"commit":"a9bae9fe56dced112e71de51bd9f382f00d584ff","subject":"Remove tox-py310 ansible-core version in test-reqs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a9bae9fe56dced112e71de51bd9f382f00d584ff"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2026-04-23 11:08:02.000000000","tz":60},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2026-04-23 11:12:24.000000000","tz":60},"subject":"Drop ironic-inspector from Zuul file trigger","message":"Drop ironic-inspector from Zuul file trigger\n\nThis container is gone - we use Ironic built in inspector.\n\nTrivial-fix\n\nChange-Id: Ia1e3dbac1ac104eceba79e6387d9692cc9ac56ee\nSigned-off-by: Doug Szumski \u003cdoug@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5ff3271b2464502f74737cfac831957c0c6cfc6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5ff3271b2464502f74737cfac831957c0c6cfc6c"}]},"branch":"refs/heads/master"},"f931316818755dc9e82e026ec8a14b1649a6c8ae":{"kind":"REWORK","_number":2,"created":"2026-04-23 13:02:44.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/40/985940/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/40/985940/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/985940/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/985940/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/985940/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/40/985940/2"}}},"commit":{"parents":[{"commit":"4a7e5ee0dd0fe3ea31a3e6d59a51fe0ca72bda2c","subject":"CI: sync Debian scenario jobs with kolla-ansible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4a7e5ee0dd0fe3ea31a3e6d59a51fe0ca72bda2c"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2026-04-23 11:08:02.000000000","tz":60},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2026-04-23 13:02:06.000000000","tz":60},"subject":"Drop ironic-inspector from Zuul file trigger","message":"Drop ironic-inspector from Zuul file trigger\n\nThis container is gone - we use Ironic built in inspector.\n\nTrivial-fix\n\nChange-Id: Ia1e3dbac1ac104eceba79e6387d9692cc9ac56ee\nSigned-off-by: Doug Szumski \u003cdoug@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f931316818755dc9e82e026ec8a14b1649a6c8ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f931316818755dc9e82e026ec8a14b1649a6c8ae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d1"],"failing_atoms":["label:Backport-Candidate\u003d-1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
