)]}'
{"id":"openstack%2Fkolla~986036","triplet_id":"openstack%2Fkolla~master~I681644e317e6341654895d7d47c79f379253a5e2","project":"openstack/kolla","branch":"master","hashtags":["kolla_monitoring_refresh"],"change_id":"I681644e317e6341654895d7d47c79f379253a5e2","subject":"WIP: Add vector","status":"NEW","created":"2026-04-24 07:16:52.000000000","updated":"2026-05-13 20:08:33.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":false,"meta_rev_id":"297633a645c00e9685214b23f12cb59c3f4fae12","_number":986036,"virtual_id_number":986036,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-05-13 15:25:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"reviewer_updates":[{"updated":"2026-04-24 10:14:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-24 11:12:46.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"}],"messages":[{"id":"1ac244a616e9b6b0503ebb75645a5abee4a704cb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-24 07:16:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"36199c39dfb8fc20d1ceb176d47a209df30ff267","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-24 10:14:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/69db3e86aa4a426abff5def0e6933773\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/d8b43a75d5cc4e4a8d7ae0f1c21e72e8 : FAILURE in 1h 15m 23s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/39a28cdaa0bd414db6dbe6c37e261a6e : FAILURE in 5m 27s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/ab06e61e08ea4eb5b87862d6769f86bf : FAILURE in 7m 13s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/5f984e47999c46e5ae42c68d6c45d4c0 : FAILURE in 1h 13m 05s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/639e2612c0cf4411ad8f40f7c293be7a : FAILURE in 39m 49s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/99bd0b139bd6441f8b76a85ed11a6b77 : FAILURE in 6m 27s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b3211e0c30c840daa878934ca56b634b : FAILURE in 8m 11s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/7f129448e5334e9da3f99b9a7f63815e : SUCCESS in 1h 27m 37s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/cb38213caf554b5bbc39ba08fed11d56 : FAILURE in 12m 36s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/513a801f05224a9e853aca5d737afe81 : FAILURE in 13m 31s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/08c55544295340ac8d3d8f4a07d9d889 : SUCCESS in 1h 33m 18s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/483ba7e9837145c29d87558251eeef73 : SUCCESS in 2h 42m 38s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/caa1c3363fe74043b823bd0ad078b771 : FAILURE in 9m 33s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/b49c9d49285d4023956a146c482eb159 : FAILURE in 14m 08s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/491e9f9c2a0640fb9bb43c8142fe134e : FAILURE in 12m 53s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/9e504329f6084d848c7356cf175aba34 : SUCCESS in 2h 09m 44s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/0013e7342bdf4929aeae21c76cbed702 : FAILURE in 2h 55m 30s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/b459a84566284252a77c9db549ec885d : FAILURE in 12m 03s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/7caaca02db304ff7abe8a7dbe90c200a : FAILURE in 16m 27s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/89f5f052c4c54086af75ca403937da6a : FAILURE in 15m 37s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/6a29a95cdec54915904041f5ed90658a : TIMED_OUT in 2h 33m 46s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/eb52a37425a84a0f8c0a556c38e228f7 : FAILURE in 15m 10s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/e43ad5ef520d46aa9793508c41791b89 : FAILURE in 15m 19s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/a25819e9e4474b65a3b345daf4455596 : SUCCESS in 30m 23s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/1e6cc4626c41411e802d06046ce9c01d : FAILURE in 12m 03s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/5d6c0207293e4276950dcea3047ebed5 : FAILURE in 13m 01s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/afc243558cf8490d894a0efe0beaf32d : SUCCESS in 56m 02s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/52cf6452f2e04034999bc30a888e1a95 : FAILURE in 7m 46s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/c566d7c38edf486ba3b3be20340b9944 : FAILURE in 10m 05s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/adf8360f2bba42de8aac67c379fe46dd : FAILURE in 10m 41s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/a4db33e38e374f8ba2c6391b82a48136 : FAILURE in 16m 31s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/25e4c35e690944db9e9e0e004d662375 : SUCCESS in 2h 25m 12s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/a17c46ef454e4985a5d3e47cf7a0a409 : FAILURE in 10m 49s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/2c552496d0c5421cb077f156add1c5db : SUCCESS in 3m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da317ea8b29445fbaf38f3c2bace5c11 : SUCCESS in 3m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5074cf9e2077469f87a6914b97d3f952 : SUCCESS in 3m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7555a8fc80a449e4bb5aa899dea98123 : SUCCESS in 3m 44s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/29b4c73dd57541f38636b730c162ecf9 : SUCCESS in 3m 44s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dda2d70539184c0bab7ec33dc7592226 : SUCCESS in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2bbd8db22824433a8d51de3fd0d6240c : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":1},{"id":"6292640162016b9a218fb5f781f684795fd4052e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-24 10:24:12.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"5eaa73568050e0c666490a1633894040994e5f2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-24 15:47:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e5ced510f7be45df80dc91b6440424c5\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/1f5ee487e7504af7ba7246c41e019dbb : FAILURE in 1h 28m 33s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/c8ee6449b4954426910921c0855c7359 : SUCCESS in 1h 21m 36s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/541ddefb58a44af8bcfaa559f0968e08 : SUCCESS in 27m 48s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/7b72c2a1fdb646e5a82076c59a1e42dc : FAILURE in 1h 22m 03s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/af97d888e54d461facbf30a846c4f25f : FAILURE in 1h 09m 49s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7f5161e27b6f469591f8cb1e8579378b : SUCCESS in 1h 17m 26s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b4f6068147de4a46a240275a0394623d : SUCCESS in 29m 54s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/c1857098fc5f4d0b81e921ef69b15ba6 : SUCCESS in 1h 00m 36s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/ce35d70967c74be399c084fd49e6aeaa : SUCCESS in 1h 20m 34s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/dcb60ef3c4b24d6085f8631e39d22410 : SUCCESS in 2h 36m 31s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/1c950b197f244424811f09999a82bf51 : SUCCESS in 1h 00m 20s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/02a71cd9e05b4b26948ee0bdfd94094a : SUCCESS in 1h 20m 18s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e9bc8a4eea6a43308bd9b84740761305 : TIMED_OUT in 2h 32m 43s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/8138844f05b448059cdd382e798b2821 : SUCCESS in 2h 32m 14s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/a97739e9748149d3a509990168524b46 : SUCCESS in 2h 35m 34s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/2d9ff109cb274f4b94a209f4ef5ba54b : SUCCESS in 1h 33m 34s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/aebda53780284d08a9f0d5cc5527ccfe : SUCCESS in 2h 51m 20s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/6adc4da1ea2e482696724c3a44b1f099 : SUCCESS in 1h 15m 12s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/1669892449a740acb14005d9efc86aa1 : FAILURE in 2h 36m 49s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/860e40269d6240148251bd3cc1c0497e : SUCCESS in 1h 10m 28s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/ee4d4b10bd9d492f84865547c04aa7ba : SUCCESS in 2h 05m 16s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/3c463ed870cb461da4e825e1567a1522 : SUCCESS in 1h 29m 03s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/e1e9c2fc3fe54b22803d8e421bf38b85 : SUCCESS in 21m 13s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/6c5fefa021d34affba38596a2a75d682 : SUCCESS in 21m 38s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/a2d6ad3d23af4bffb29841556fc427dd : SUCCESS in 31m 51s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/4343bcffffb94875812f3f036cfac473 : FAILURE in 1h 11m 33s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/98f2eeef9ed5405b8e946b46753afdc5 : SUCCESS in 1h 36m 53s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/f1e8c8637cd9471fb7a9816bc851b23c : SUCCESS in 1h 29m 22s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/4fc106b4a9504d4e9873f43c2c489002 : TIMED_OUT in 2h 32m 02s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/d1f141499ef54fea860728ccbbd5baa7 : SUCCESS in 2h 03m 38s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/d607a485cd7b4b36b2069cf431fd9a6b : FAILURE in 2h 06m 33s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/9292eed13e954c15b231c3676d46501d : SUCCESS in 2h 23m 04s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/63b678cdaa29417fa796c3c9ac6a952b : SUCCESS in 1h 31m 25s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/3867eee85ad34b67bc6baedcf52f3e8f : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/206fbaa8ffff4b86910cf40f9668b02a : SUCCESS in 3m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e28a19655f504ea4877f26891f7979f0 : SUCCESS in 3m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d46ace8e36cb43e79cda0893b1838050 : SUCCESS in 6m 15s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2b9eebc5bfd7463caaecc31452cdf2c2 : SUCCESS in 8m 10s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2acf90e6b1d547ad8b25d36a863510ac : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df1db0c6126446e99032f840376448e9 : SUCCESS in 6m 31s","accounts_in_message":[],"_revision_number":2},{"id":"65edac60774f95bd818d93fe8370cddeafd028d2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-24 15:54:14.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"e44d260fc1b5d71aed129133d0a62b6512a89ce4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-24 15:55:01.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"782b8e70fa670974d9ba998dacb4e3800e80f8d1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-24 18:02:29.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"b85cbd165dd969d53899f4789165be5df7e3e5a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-24 21:07:39.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/100cd1584bdd4558a9cbae5070331411\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/f8f0093d267641298679abe32497331e : SUCCESS in 1h 24m 27s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/79ca84127e8b4fcd85eb99f037b64bfa : SUCCESS in 1h 34m 22s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/a255fb7649ba47458752685659989026 : SUCCESS in 1h 20m 56s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/25d2e629a01c4f0e9c4ccf86a9acb35a : SUCCESS in 1h 10m 47s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/2e99efc9a80e413c878f6177c6aa1354 : TIMED_OUT in 3h 02m 08s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b0a3eabe24d142f1aa1e39c38d9f5132 : SUCCESS in 1h 39m 02s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b6caecca69e748bfb8c9130cfaee80ab : SUCCESS in 19m 28s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/4bc80857f4844289b93222d4694455b1 : SUCCESS in 1h 23m 09s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/ae5b6306c3044a9bac59af151a9f19e5 : SUCCESS in 1h 27m 28s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/e33c374b6e64462985034f8444fc7f8d : SUCCESS in 1h 49m 32s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/ff93bc28f4a64436b0150894b1d994c0 : SUCCESS in 1h 38m 36s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/d32f90a838de4dcfbb3f97e2414a7d13 : SUCCESS in 2h 04m 34s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6e0f96e73f8443138855076471c75c03 : SUCCESS in 1h 29m 45s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/750bb22ad5794b908e59410491c9784d : SUCCESS in 1h 54m 28s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/5522870737f74439b26b58013eca1364 : SUCCESS in 2h 04m 03s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/ea40f7813aa74c6a8a00c0578a9e672b : SUCCESS in 2h 18m 14s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/47b7cf869c05430b8de0676051150b90 : SUCCESS in 2h 57m 44s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/59330d6bcd03467b86e77c90989dc813 : SUCCESS in 2h 07m 37s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/387fc4ff23fd438cbf201044eb876c79 : FAILURE in 3h 02m 50s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/88c05724379d4fe18d909c92fb5b4182 : SUCCESS in 1h 53m 28s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/a484f29b9ecb439b91ae68280fec1f44 : SUCCESS in 2h 04m 30s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/62858ada247947a3aff4d0533429edab : SUCCESS in 2h 05m 43s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/4068e82fcbb8479f821c0a78e429139d : SUCCESS in 28m 08s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/5f0c30cc2edb44bda7b621a6e1eeab31 : SUCCESS in 30m 55s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/f27329cf349349b98442090ffddd5b84 : TIMED_OUT in 2h 32m 32s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/4481f6adfe2d4dc9aeeca0577708dec1 : FAILURE in 49m 04s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/7b4b909caaf24829bcc44ed58a13c408 : TIMED_OUT in 2h 31m 52s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/cba569ad5cb5406c9d2382d1b9e11b9c : SUCCESS in 1h 03m 33s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/0030c8aa8f014738add74836ada5a051 : SUCCESS in 1h 28m 54s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/61bbc60305504592a5acd5a81e9e19f4 : SUCCESS in 1h 32m 45s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/120dfa9aaadf490fbbb3511fa3a6efe3 : FAILURE in 1h 42m 04s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/cdcd34a35d494e2e82f24f605313770a : SUCCESS in 2h 19m 52s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/aad614ebd5e0493f905204cf11cdbfba : SUCCESS in 2h 16m 38s (non-voting)\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/4235c18471e04adf997165e957fb99c9 : POST_FAILURE in 1h 38m 19s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/56e4af8c0dde40dcaedea8b23c379d93 : SUCCESS in 1h 56m 00s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/7459cf332a934dc8bb7c52fbd8f7b45d : SUCCESS in 1h 31m 02s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/4e123db73121419daf40cb3e40f42c78 : SUCCESS in 3m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb7c5eef2dc8446b9aa1e9d5cee157d4 : SUCCESS in 3m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6849db3155c0477d9b052dd379f93ca4 : SUCCESS in 3m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b37da94e2dcc49e08f4476b72dcaf429 : SUCCESS in 3m 58s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2b13ffb10f0c4dedafcec2be276c8e5d : SUCCESS in 4m 19s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2528a1f6041c46c3aab8357289705fae : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c10395583a14f1cab181e041c8d3eb3 : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":5},{"id":"8f039bc31783400bcb991b987df890313ca0d51b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-25 12:12:21.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f7e5eaeb5204586b1d8a84f53ec2117\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/22f12564c1824a56b191b9137f57968f : SUCCESS in 1h 18m 14s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/3365d175969148408468589077478326 : SUCCESS in 1h 16m 53s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/bff3d8ac2a944360a7bd5dcedff3b093 : SUCCESS in 1h 16m 33s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/39caedda5f5848f5b233e5467c8a3901 : SUCCESS in 1h 19m 24s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/fc3cd1a0cf814dd1b30523c47349f251 : FAILURE in 1h 24m 26s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/38741ae7c393441a819457b4a51e226e : SUCCESS in 1h 44m 29s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f58459b1576c4ad5bb01b3527d48ac62 : SUCCESS in 3m 16s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f6d0a72f13304739942e3de8595ee776 : SUCCESS in 3m 50s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/caf4f2e5ffc24411a7782714d0b54843 : SUCCESS in 5m 59s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f31be6fa45b1d788f69fd947bf7141351fda5bd1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-13 12:23:33.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"637fc4c31e5aeeb410e4e0aeb15db0b11114e6f3","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-13 12:23:45.000000000","message":"Hashtag added: kolla_monitoring_refresh","accounts_in_message":[],"_revision_number":6},{"id":"3b10699cc2344f2538f2fe70ff388b598d4bf38b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 15:25:23.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27ae7aa21c7646edb80331ec05b6ba63\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/1481130d5b844b00ae4bbe5cdccb81fc : SUCCESS in 1h 03m 36s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/16625251038c420cb722ef563a51fd07 : SUCCESS in 1h 31m 28s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/bdb41fe88ffb4649a03a0a3c7159132a : SUCCESS in 26m 12s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/5fa2aa3e6b864d1aac3acb22cc5b9312 : SUCCESS in 1h 00m 27s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/fbd64442093e46cd9186e588c36507a7 : SUCCESS in 46m 56s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d2150d6a3e29419886ca81cea8ef199e : SUCCESS in 1h 27m 24s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/db51064294bf46a2bc4ffa805eaacc2c : SUCCESS in 34m 13s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/96458b76845d4794b05a123dda96a7ae : SUCCESS in 1h 24m 09s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/01bbed35c1a04e1d93dbc5a17b80bb99 : SUCCESS in 1h 27m 17s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/743aa00a166445659fcedb7b04bc42a7 : SUCCESS in 1h 50m 09s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/f0d6ab43cf954f43b47dd30f4d2348d0 : SUCCESS in 1h 34m 46s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/ac17f5446be144d59f25ff7d2ae97aa8 : SUCCESS in 2h 05m 33s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/29c670af1161404faf3ade11b27102d4 : SUCCESS in 1h 29m 05s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/4a37ad2d613f404b8d34331d3a9a8e80 : SUCCESS in 2h 02m 17s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/4336a57c0be04771a5d1a8dadc0fb5b2 : SUCCESS in 2h 02m 09s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/93806905ff934ed396dd8ce636e6f00d : SUCCESS in 2h 18m 36s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/8d2e98d89c8b48bb91654f3ca7c55ee3 : SUCCESS in 2h 58m 41s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/65dd8a1fc42a4fa48b9d3ed9514ce666 : FAILURE in 1h 40m 48s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/9eb0dd93285e461ba39c289ed480edde : SUCCESS in 2h 56m 21s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/252a8ffb9f8e4912abc8d39cbef637a1 : SUCCESS in 1h 47m 02s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/a9623f35ca6746e0864765ffcc856b8b : SUCCESS in 1h 58m 17s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/c2887adb74e4485aa239493fc7596edd : SUCCESS in 1h 56m 22s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/81df4610beb44e9fa22752bc8f47a9bc : SUCCESS in 26m 18s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/70f0fa19b0d34dc7acb8e28a9e15ad6c : SUCCESS in 29m 50s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/b4225f962f7e4bba8f82026f99e00902 : SUCCESS in 34m 35s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/defc159a8d70444a8d7b154ef467276f : FAILURE in 1h 08m 07s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/9688a4077db644e0bc234009c4af9581 : SUCCESS in 1h 20m 36s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/c3455d8ae62c42018e7291e1b0e66e63 : SUCCESS in 49m 19s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/ac0fea86e4194dfa9e82d7b05610c4fc : SUCCESS in 1h 24m 53s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/99b26d46b9894832abf3702f1038c850 : SUCCESS in 1h 25m 22s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/b126297dd78a44149362312345097edd : FAILURE in 1h 24m 07s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/0f5d3bc2d76d45708508c8f5bb44b289 : FAILURE in 1h 52m 44s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/3cacf80daee14a0a99d5c8029ccd4c0d : SUCCESS in 2h 19m 41s (non-voting)\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/e7e5decde21c44d3bd6c67a97fbb6262 : FAILURE in 2h 26m 10s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/ae27b49e9d8b41c88c5e7ddf32ca12db : SUCCESS in 1h 31m 03s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/b683a55f4b5d4c258eb8d760d1628e23 : SUCCESS in 1h 20m 35s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/c30fd1fc782f44279cb2edd08772aafa : SUCCESS in 2m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5d5487f0bc9474eab7c2ef174491b3b : SUCCESS in 3m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8e7792c7b6554592a2d305775f0ef540 : SUCCESS in 4m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a5814de0ff7347aab9fe4494c57f97fe : SUCCESS in 3m 24s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/75b8cfc874e14b9a992688f9fd0af350 : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0684f296c50043fb8ebb49f55fc84d80 : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4ec32fb6cb040249febe9b37800acaa : SUCCESS in 6m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d294a67d5a7a422dade1df7b37a1a530 : SUCCESS in 9m 48s","accounts_in_message":[],"_revision_number":6},{"id":"297633a645c00e9685214b23f12cb59c3f4fae12","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 20:08:33.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/743acb0b1f7b4c27b1d075c23cf4a307\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/1596702ce4f5417a9d06b470166bc655 : SUCCESS in 1h 29m 35s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/0ef26f81aedb49dcba2e686f34ed69dc : SUCCESS in 1h 16m 48s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/d44608d47d0e462aa7fe87d44c571a21 : SUCCESS in 2h 07m 18s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/27acc98a7aeb4dd9ac97a00b83a3b650 : SUCCESS in 1h 59m 07s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/d7d164735e624edfa78035263419139d : FAILURE in 2h 16m 19s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/3e2ac1e885fd4945b3c68edb6c775932 : SUCCESS in 1h 28m 14s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/37548cd2d7c84b3ab336481bffbee957 : SUCCESS in 4m 04s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bd1439132e0c4c1b82ba329f02d2a29e : SUCCESS in 3m 47s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/1937b644005d4e24b02f277722a74a57 : SUCCESS in 7m 10s (non-voting)","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"9b2e65df2fbc0ae0ae850c4b6487820ca076523d","revisions":{"c978a15147dde3dd770de7455758751f0e3df01f":{"kind":"REWORK","_number":1,"created":"2026-04-24 07:16:52.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/36/986036/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/36/986036/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/986036/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/986036/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/986036/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/36/986036/1"}}},"commit":{"parents":[{"commit":"4a7e5ee0dd0fe3ea31a3e6d59a51fe0ca72bda2c","subject":"CI: sync Debian scenario jobs with kolla-ansible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4a7e5ee0dd0fe3ea31a3e6d59a51fe0ca72bda2c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-24 07:12:08.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-24 07:16:44.000000000","tz":120},"subject":"WIP: Add vector","message":"WIP: Add vector\n\nChange-Id: I681644e317e6341654895d7d47c79f379253a5e2\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c978a15147dde3dd770de7455758751f0e3df01f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c978a15147dde3dd770de7455758751f0e3df01f"}]},"branch":"refs/heads/master"},"6e1fd532b0da7505915307a14557aa5c7333ad10":{"kind":"REWORK","_number":2,"created":"2026-04-24 10:24:12.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/36/986036/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/36/986036/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/986036/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/986036/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/986036/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/36/986036/2"}}},"commit":{"parents":[{"commit":"4a7e5ee0dd0fe3ea31a3e6d59a51fe0ca72bda2c","subject":"CI: sync Debian scenario jobs with kolla-ansible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4a7e5ee0dd0fe3ea31a3e6d59a51fe0ca72bda2c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-24 07:12:08.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-24 10:24:12.000000000","tz":0},"subject":"WIP: Add vector","message":"WIP: Add vector\n\nChange-Id: I681644e317e6341654895d7d47c79f379253a5e2\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6e1fd532b0da7505915307a14557aa5c7333ad10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6e1fd532b0da7505915307a14557aa5c7333ad10"}]},"branch":"refs/heads/master"},"84224f811bf2d01b2ad0e9e11e4c6ca6a4ff3e7d":{"kind":"REWORK","_number":3,"created":"2026-04-24 15:54:14.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/36/986036/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/36/986036/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/986036/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/986036/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/986036/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/36/986036/3"}}},"commit":{"parents":[{"commit":"4a7e5ee0dd0fe3ea31a3e6d59a51fe0ca72bda2c","subject":"CI: sync Debian scenario jobs with kolla-ansible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4a7e5ee0dd0fe3ea31a3e6d59a51fe0ca72bda2c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-24 07:12:08.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-24 15:54:11.000000000","tz":0},"subject":"WIP: Add vector","message":"WIP: Add vector\n\nChange-Id: I681644e317e6341654895d7d47c79f379253a5e2\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/84224f811bf2d01b2ad0e9e11e4c6ca6a4ff3e7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/84224f811bf2d01b2ad0e9e11e4c6ca6a4ff3e7d"}]},"branch":"refs/heads/master"},"4891cbfedb251ae228784eb332202c30c1116c25":{"kind":"REWORK","_number":4,"created":"2026-04-24 15:55:01.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/36/986036/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/36/986036/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/986036/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/986036/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/986036/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/36/986036/4"}}},"commit":{"parents":[{"commit":"4a7e5ee0dd0fe3ea31a3e6d59a51fe0ca72bda2c","subject":"CI: sync Debian scenario jobs with kolla-ansible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4a7e5ee0dd0fe3ea31a3e6d59a51fe0ca72bda2c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-24 07:12:08.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-24 15:55:01.000000000","tz":0},"subject":"WIP: Add vector","message":"WIP: Add vector\n\nChange-Id: I681644e317e6341654895d7d47c79f379253a5e2\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4891cbfedb251ae228784eb332202c30c1116c25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4891cbfedb251ae228784eb332202c30c1116c25"}]},"branch":"refs/heads/master"},"17e0ee5b3547e4adc1dd784c2fd3d0ba3f53300d":{"kind":"REWORK","_number":5,"created":"2026-04-24 18:02:29.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/36/986036/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/36/986036/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/986036/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/986036/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/986036/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/36/986036/5"}}},"commit":{"parents":[{"commit":"4a7e5ee0dd0fe3ea31a3e6d59a51fe0ca72bda2c","subject":"CI: sync Debian scenario jobs with kolla-ansible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4a7e5ee0dd0fe3ea31a3e6d59a51fe0ca72bda2c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-24 07:12:08.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-24 18:02:28.000000000","tz":0},"subject":"WIP: Add vector","message":"WIP: Add vector\n\nChange-Id: I681644e317e6341654895d7d47c79f379253a5e2\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/17e0ee5b3547e4adc1dd784c2fd3d0ba3f53300d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/17e0ee5b3547e4adc1dd784c2fd3d0ba3f53300d"}]},"branch":"refs/heads/master"},"9b2e65df2fbc0ae0ae850c4b6487820ca076523d":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-05-13 12:23:33.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/36/986036/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/36/986036/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/986036/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/986036/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/986036/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/36/986036/6"}}},"commit":{"parents":[{"commit":"a7c184013662781384bb433ec063d6f93543bdf1","subject":"Update master for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a7c184013662781384bb433ec063d6f93543bdf1"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-24 07:12:08.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-05-13 12:23:33.000000000","tz":0},"subject":"WIP: Add vector","message":"WIP: Add vector\n\nChange-Id: I681644e317e6341654895d7d47c79f379253a5e2\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9b2e65df2fbc0ae0ae850c4b6487820ca076523d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9b2e65df2fbc0ae0ae850c4b6487820ca076523d"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"17e0ee5b3547e4adc1dd784c2fd3d0ba3f53300d","theirs":"a7c184013662781384bb433ec063d6f93543bdf1","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
