)]}'
{"id":"openstack%2Fkolla~986149","triplet_id":"openstack%2Fkolla~master~I770b9e62c66f8cc63475d51f80eecacc8415b91f","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-04-27 10:14:02.000000000","reason":"Change was submitted"}},"hashtags":["kolla_drop_zun_kuryr"],"change_id":"I770b9e62c66f8cc63475d51f80eecacc8415b91f","subject":"Drop zun and kuryr","status":"MERGED","created":"2026-04-24 17:12:32.000000000","updated":"2026-04-27 10:15:43.000000000","submitted":"2026-04-27 10:14:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"986149","meta_rev_id":"61d75ea3ee6b40617f6a0f03bef81d05030d9f1e","_number":986149,"virtual_id_number":986149,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-27 10:14:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-04-27 06:06:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2026-04-27 06:18:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2026-04-27 08:07:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2026-04-27 08:07:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":-1,"date":"2026-04-27 06:06:37.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2026-04-27 05:48:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-24 19:41:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-27 06:06:37.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2026-04-27 06:18:36.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-04-27 08:07:05.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"b254b7fadf8639e37c13db8fc963f83418774dc4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-24 17:12:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b31846e4a0c6f23356a29bd07639cb7c08fe36d7","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-24 17:12:42.000000000","message":"Hashtag added: kolla_drop_zun_kuryr","accounts_in_message":[],"_revision_number":1},{"id":"9a9feae5f47d39ff8bc6216c8187ed64b582f3b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-24 19:41:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44ab307b5e7c43d1b16149d2a3f50208\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/47350971455440c0bf24d7bb7ab571fe : SUCCESS in 3m 23s\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/3c1fc25a9ca0416caecd2e086e2ec5b8 : SUCCESS in 1h 22m 42s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/16ae1750f40b4cafafa4691bc7a769d2 : SUCCESS in 1h 55m 28s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/c56eabc3e24744e9b6f4319593cf5e60 : SUCCESS in 27m 39s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/edc56dfa0b4a419ea4073949fd661e32 : SUCCESS in 1h 14m 18s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/75a25b8dd7964df5b9653bb9ba32ab99 : SUCCESS in 1h 03m 09s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1b04f23d112a428b974c96e48b0b0ac6 : SUCCESS in 57m 51s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/51ad9112f4784cc69b19833ea18132f7 : SUCCESS in 25m 17s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/e15530b3e0dd4d6b9b350d48d94fd46e : SUCCESS in 1h 33m 39s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/3d55609eb29e41668a005f6728f42547 : SUCCESS in 1h 24m 31s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/58a4073c9e4f4154ba95300ecd481e20 : SUCCESS in 2h 19m 14s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/1766ae17024d42b992d9875dec193232 : SUCCESS in 2h 02m 28s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/9cbe2add9ea04cc0b987df8c3c594415 : SUCCESS in 2h 09m 11s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5bdc7b9171224d17b5ab2812ce3d1594 : SUCCESS in 2h 15m 43s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/c5d6c8e4334e49cba733818e470df991 : SUCCESS in 2h 28m 06s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/cc6a6dd68f714508b389c71970d88c99 : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a96032655cd4e978f57aa7b914eb519 : SUCCESS in 5m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e1998fe29c9e4c728482adc0adcd0144 : SUCCESS in 3m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/15bdab33335a4387b94f80b237da7eec : SUCCESS in 3m 36s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b4aaeb79c7f341de9704c49518fb94bf : SUCCESS in 5m 47s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/140af66475ad4a93a777981fafbd905c : SUCCESS in 3m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad991301a63e4b51af87e35057c286b9 : SUCCESS in 8m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0d6d2aa628d54b2491a12178fb2c8358 : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":1},{"id":"21ea4f27442a4846d8567812b96baf6e2d8c6e9f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-25 10:34:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e0af1ff0d4d4edc96e8947efa565127\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/f4057a3b7a924373bb37f7a198805d2a : SUCCESS in 1h 22m 10s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/1c80594259a640d5819e4d668150acda : SUCCESS in 1h 23m 57s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/bb45744da51c410f857b27698bf7974a : SUCCESS in 1h 31m 35s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/2c3f09a5e31c4a7ea4fde1548bc4031a : SUCCESS in 1h 21m 20s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/7e9535ab3271461d97d3d4317991fcd9 : FAILURE in 1h 29m 46s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/898498c2d93348f4aeac49a68d322cd7 : SUCCESS in 1h 36m 08s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e3c08fefcc8f4d87b1e28e1fbfe0540f : SUCCESS in 4m 26s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/41dc3a2bc12d41dcb842ec80417699e5 : SUCCESS in 4m 17s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/8cd5be30e7c845f2a4839bfd40d9db24 : SUCCESS in 6m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bf2dc17d5396202bcaf17a5df89bf92f721dfb7c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-27 05:48:34.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"f49de5479996e0f7c4912bb76bc2f4907abfc08a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-27 06:06:37.000000000","message":"Patch Set 1: Backport-Candidate-1 Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c4bff674826aad61f284f59e935888b6f4981f59","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-27 06:18:36.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"4d8ea9e8366d8c821afb5bce9bf5693dfed6fa8d","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-04-27 08:07:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d9584c65accf74f01e8b3a7866f5aeb084caf9c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 08:07:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"569c0c97d976a7810bca636c9b7d33159f9cb5ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 10:14:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c77c0984f20443bc8ef73cf6010fb1f0\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e572ef0c014141d59e040f4aa24091c4 : SUCCESS in 2m 02s\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/b9ba39cc73aa435c82509d3df71c1b5e : SUCCESS in 1h 05m 28s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/3b37b77f53034e90bbd6a1f97d348965 : SUCCESS in 26m 05s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/1a25027457a84ed3b8882c7f460a132e : SUCCESS in 1h 13m 18s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/36e088712965436ea41f339dca19a2ec : SUCCESS in 36m 28s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fb9ac6840c594cb0b36c09504248a940 : SUCCESS in 1h 33m 37s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/096423b8c2a24f09a1d4935dff910743 : SUCCESS in 27m 06s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/818411b94fd04969b4b9c1b0cad2f984 : SUCCESS in 1h 24m 40s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/2ee5fc9207e742a09e67b20c12d3d36b : SUCCESS in 1h 42m 12s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/2cf65531600246b8b58334f15dd99dd7 : SUCCESS in 1h 35m 21s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/245c425da79f4838a455472500249539 : SUCCESS in 2h 05m 54s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7573be52c8d44c64a11a744951a921eb : SUCCESS in 1h 27m 47s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/8ce79aedb9ac4620b5a4fc852909bcba : SUCCESS in 1h 54m 55s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/c52836ffec4644e5a6468c912cea7ae2 : SUCCESS in 2m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d625a1647c0247c5bbffc8cfdb2d0c2d : SUCCESS in 3m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2435b78a9492427084960056879d5e26 : SUCCESS in 2m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/543ea90d7c5649f1b07e8dad9bfd81ef : SUCCESS in 3m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fc08a12cbd04311b37be1f94f67a4bc : SUCCESS in 6m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/64d9539d777f4ccfbe58a558f5bc3157 : SUCCESS in 3m 04s","accounts_in_message":[],"_revision_number":1},{"id":"ad0ae2e58283f2c848ec2fc0d3e527ff41da3cd5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 10:14:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"61d75ea3ee6b40617f6a0f03bef81d05030d9f1e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 10:15:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7584a3bb308644509296b6414a08d0fc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0a6f1958ec441738d23dbe2fd009847 : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8cdaf9131883431c9349a00d8f34ed1f : SUCCESS in 1m 26s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"77b92c33f979ad5ae4321a1b3604ce90a99415a1","revisions":{"77b92c33f979ad5ae4321a1b3604ce90a99415a1":{"kind":"REWORK","_number":1,"created":"2026-04-24 17:12:32.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/49/986149/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/49/986149/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/986149/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/986149/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/986149/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/49/986149/1"}}},"commit":{"parents":[{"commit":"90cfc97e5bea5a4800e46d4af64ca352f561f11c","subject":"Remove reference to deprecated app.wsgi in Aodh image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/90cfc97e5bea5a4800e46d4af64ca352f561f11c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-24 17:11:48.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-24 17:12:29.000000000","tz":120},"subject":"Drop zun and kuryr","message":"Drop zun and kuryr\n\nZun is broken in 2026.1, it has been released broken and there are\nno reviews on the patch to fix it [1].\n\n[1]: https://review.opendev.org/c/openstack/zun/+/982362\n\nChange-Id: I770b9e62c66f8cc63475d51f80eecacc8415b91f\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/77b92c33f979ad5ae4321a1b3604ce90a99415a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/77b92c33f979ad5ae4321a1b3604ce90a99415a1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-1"],"failing_atoms":["label:Backport-Candidate\u003d1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
