)]}'
{"id":"openstack%2Fkolla~986377","triplet_id":"openstack%2Fkolla~master~Id63c035fafaa68a9004c65161c5a3a3ad5544f02","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"37203":{"account":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"last_update":"2026-05-06 06:00:17.000000000","reason":"Change was submitted"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2026-05-04 08:13:54.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_13252\u003e by clicking the attention icon","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-05-05 17:26:48.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-05-06 06:00:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id63c035fafaa68a9004c65161c5a3a3ad5544f02","subject":"set_configs: Add support for creating directories and setting perms","status":"MERGED","created":"2026-04-27 13:36:44.000000000","updated":"2026-05-06 06:01:28.000000000","submitted":"2026-05-06 06:00:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":15,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"986377","meta_rev_id":"800c8ccd130688f1818e674fe6707c6ba54be6ac","_number":986377,"virtual_id_number":986377,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-06 06:00:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2026-05-06 03:54:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2026-05-06 03:54:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},{"value":-1,"date":"2026-05-05 10:34:27.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":-1,"date":"2026-05-06 03:54:58.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2026-05-05 14:35:47.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-27 16:48:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-28 09:14:34.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-04-29 06:13:13.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2026-04-29 06:25:34.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2026-04-29 17:30:37.000000000","updated_by":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"reviewer":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"state":"REVIEWER"}],"messages":[{"id":"7780ae0265017d2a32e58621286b522a47b854c9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-27 13:36:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"987262e32cab4f7dcc1d6e238d3b7ffc061680e9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-27 13:52:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4cae27fc0f0e7c563a0c44e0b986fa395e779d6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 16:48:00.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/97015473e7014e2a9358770d4be62edc\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/6981e2aa14ae47ad88224947a801407e : SUCCESS in 1h 06m 10s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/072763cbb7464840986babfc01fccc21 : SUCCESS in 1h 18m 47s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/fd895ac1602348cc84d92105d4ad49fc : SUCCESS in 25m 20s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/04b0b6554c1745dc9d24cb1c1f52cc13 : SUCCESS in 1h 06m 59s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/31fcef9a127b4729830f41f1a5ac5475 : SUCCESS in 1h 05m 12s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/07e9d7b4f33441a5969829bbe9599836 : SUCCESS in 1h 26m 20s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a5b45b187e2b4ddd9f99f082039b4de6 : SUCCESS in 32m 18s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/0162162aacdd4bd28bcd04eab55b8f2a : SUCCESS in 1h 24m 17s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/03aa4ee90c77452d8b785a30b3a879da : SUCCESS in 1h 26m 55s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/1f404e7af3dd403c87f8584f45bf4930 : SUCCESS in 1h 47m 30s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/809726db3fe34516ba5a59d662655021 : SUCCESS in 1h 38m 04s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/3dbd7c11b3c540cabd2a5a7d9fda7769 : SUCCESS in 1h 59m 54s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5080d0e8f1014c7693bbbb088522c32f : SUCCESS in 1h 33m 49s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/4c2dcdff451142b2b444eb6b74ec9b3b : SUCCESS in 2h 02m 44s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/8ff57115d2b14ec89d76eb5c12146859 : SUCCESS in 2h 00m 38s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/b1db07195c204647850f6abb1422b7e5 : FAILURE in 8m 23s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/5be6ea847eb0471487c8053944aa9deb : SUCCESS in 2h 53m 59s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/a2c5e0adec8945cea02c361e634f8a3c : SUCCESS in 2h 06m 23s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/3efb1699fc7e4497a86d25f794901d6e : FAILURE in 1h 30m 09s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/39b76707ec9b4ecfb8f9653672e2b4d5 : SUCCESS in 1h 57m 58s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/91155204db014b94802fe75e0e15400a : SUCCESS in 2h 03m 43s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/419a9c86071744c69114440192ef5336 : SUCCESS in 1h 56m 40s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/a85f4bec700e4dd1af33e3bb0a036c23 : SUCCESS in 26m 42s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/982dda8aad51416d88987b9f594f28d7 : SUCCESS in 30m 40s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/6ea22de23f304b6391823972f874ba2b : SUCCESS in 35m 38s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/104bfd02352447978dd78b9bcda9c535 : FAILURE in 1h 02m 23s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/afbd3346760a423099f883ffe0ea5473 : SUCCESS in 1h 26m 26s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/a0bafd5764bc42e49bce2ffc8ea46ff8 : SUCCESS in 1h 41m 45s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/23f4342b6206457887dac4ef73de0828 : SUCCESS in 1h 26m 23s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/50a16ad56e7f435cba645f60c3cae70f : SUCCESS in 1h 31m 06s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/7106aa960d934534a51d45660c5742f8 : FAILURE in 1h 44m 25s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/d571ae35476e4135adf8c2befca51f76 : SUCCESS in 2h 21m 35s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/9b134df28199491ea488674a1d6e7b86 : SUCCESS in 2h 28m 32s (non-voting)\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/6d58f582edfc42e9926627e9f4781d51 : SUCCESS in 1h 55m 35s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/74ee481ba3664d1ca1b68e46a976c47e : SUCCESS in 1h 06m 15s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/56eb2961884f48c3851b7ce6da537bf3 : SUCCESS in 1h 52m 06s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/305543c7de0a4627be95b4b4adb45119 : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35182e0984204b8097e09cabe80631d2 : FAILURE in 3m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/69b5b86a4a5940bda67de9e13568efa4 : SUCCESS in 3m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1cd1582eeb0441e5888f82d6dec1c72c : SUCCESS in 2m 04s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/39ab610c98304692948852730c891d7d : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f52f3771ed484c5f993b4850a4f95a75 : SUCCESS in 3m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3185c2bc00d4e5a8380705cb33f2984 : SUCCESS in 4m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b2bf56da67ed4c68a2bb20d6206cb683 : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":2},{"id":"b23182b484dc2c7b8fe5d890cd71cff3a0e9d418","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 18:48:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d813b5098fe8478c9b951c659ffe59e0\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/d03943e5b62b4d369841972c94dbcca4 : SUCCESS in 1h 35m 20s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/e7ee7b4995f544dd890d043b7a1c89af : SUCCESS in 1h 58m 32s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/842f4b894ca441939116bc3aa60b76a0 : SUCCESS in 1h 37m 28s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/5c33949f2a844b048cd0432bf76d96c3 : SUCCESS in 1h 32m 25s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/6e83d78c843d43c3a197dcec2008ebaf : FAILURE in 2h 18m 24s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/0e59feed48b545229ba1d8cfcc40a38e : TIMED_OUT in 3h 02m 49s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bea21eab7c4548d09a88765fb66a2c38 : SUCCESS in 3m 42s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/38090f4161b44f16a96dcc48f2bfc56a : SUCCESS in 4m 06s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/9a5b7fe84d7c407fb57d01208f4d06dd : SUCCESS in 7m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c752b9d563875d266eb02aefdf9bf2a480d6954c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-27 19:24:05.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"506d01bf6fabe90337e1db388feb4a3dfaddebd8","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-27 19:24:08.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":3},{"id":"e58fb2d617bb366cddee58e62f352bd1844772ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-27 19:24:46.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"02d50988f5a3584e880449ea0ba559f40add110c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 22:14:28.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/58547f384e864654952c6a3ed8dbc49b\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/f36c303aacd143a8af7cfb5e1f12fb4c : SUCCESS in 1h 15m 52s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/f400d01c523a4c3f928178f8d1a45fba : SUCCESS in 1h 20m 36s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/2cb9c6fc3bd8484aaa46bc99792d87b9 : SUCCESS in 26m 39s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/991fb5513328459894c430d2bb09ed12 : SUCCESS in 1h 14m 00s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/24534e9693cb47368583ce83c944c6e7 : SUCCESS in 48m 28s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8b9d41ec182d4a18a691b6e5e5bfcbca : FAILURE in 56m 55s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/51b5ad28fd2a4bdba4486b8f08cdc3ad : SUCCESS in 25m 26s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/1112028997744494a49824369d6a8623 : SUCCESS in 1h 21m 13s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/76c1bf0608c94ad8b526cda9582884f3 : SUCCESS in 1h 21m 31s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/eb5718a3a93e4046a0fac8dcd78c657a : SUCCESS in 1h 45m 31s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/327fce34bf76420cb52ed5afba9dfa3d : SUCCESS in 1h 30m 09s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/6f797323741142bf85c3c2c8329d070f : SUCCESS in 1h 59m 53s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f989690579914d12923be64367059a03 : SUCCESS in 1h 24m 12s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/f8e248cb9f2b470bbf69aca92b9845a5 : SUCCESS in 1h 57m 27s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/9b396db8e9f84dfb881852e387328730 : SUCCESS in 2h 03m 09s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/1e9090becb614240982db0d5b23ccb1b : SUCCESS in 2h 08m 24s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/c7f8e0c87df444a685bcf315bad8b721 : SUCCESS in 2h 49m 08s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/05d5376947f643eb949a4addaf92fab9 : SUCCESS in 1h 57m 06s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/4880d4ec549146e786f214ffa0e03ddf : SUCCESS in 2h 37m 05s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/a51898ea5c5241b7be32b9f9dd445f77 : SUCCESS in 1h 52m 24s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/6444aee096644e078fcbae7ad4982e50 : SUCCESS in 1h 57m 21s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/ba11f83408c3441fbc1fefbf32e82574 : SUCCESS in 1h 57m 26s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/14904d8b29d447c091b87d65079879f5 : SUCCESS in 27m 31s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/2417221b32fa4a3d9476b334b392205e : SUCCESS in 33m 03s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/7a69a52aa6ab41ce9bbbd6304a3b3e19 : SUCCESS in 18m 57s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/73ffc24016194c959d3fdad4b87c2c5e : FAILURE in 1h 04m 22s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/9fd34f77188241d09766661b27372167 : SUCCESS in 1h 07m 05s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/96c7e05657e341929396c5fb5b6397f2 : SUCCESS in 1h 39m 46s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/2d1682c766694c81ae48fc32524d8935 : SUCCESS in 1h 26m 30s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/d99b8d589b194168b73a84fa14635701 : SUCCESS in 1h 30m 34s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/2e410635c58b4ba4bccf19ac458ea6ea : FAILURE in 1h 41m 09s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/c545766b86ba43749958fd014195e183 : SUCCESS in 2h 28m 32s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/bdf0dc53ad26437c9d5f718ebe3cd52a : SUCCESS in 2h 26m 36s (non-voting)\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/e35e968ecb6a498ab36b72d85aca2152 : SUCCESS in 57m 43s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/21a189fad1c34bdcb4e32aaa58010229 : SUCCESS in 1h 17m 28s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/085af24e4ad84f7b8ca694433313524a : SUCCESS in 1h 36m 11s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/70b9e6c9c6234b2b95986590a5631dbe : FAILURE in 1m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7de1f34da36b48c0917470639e954e4f : FAILURE in 2m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/33a5861468d64b37b220eda53de7fb9a : FAILURE in 2m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e3c4fdc384834208bde448cfff286a6c : FAILURE in 2m 30s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/abb951a401c946368c3a549f998d0e8a : FAILURE in 4m 30s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d2688b326dc44e2c98d8ae3d52962d51 : FAILURE in 2m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73f919c1fe8b45a28db97fa54f1ea489 : SUCCESS in 6m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ef12e71269fc4f46b8e3d105f8e9d16a : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":4},{"id":"be2734dd505087baf0df396d73226441b5e8c21a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 01:54:36.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c39d465865f64786ace0c065f769b44b\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/00f7fdf0154d4255894e2045adc04f64 : SUCCESS in 1h 31m 22s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/8bce6270c5d742a4a985391eade3da42 : TIMED_OUT in 3h 15m 27s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/2321500b74a6437989651c434a0de2bf : TIMED_OUT in 3h 07m 41s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/9ccc86cae32b43faaa2700479d11c4e6 : SUCCESS in 1h 12m 42s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/712c576c16084e1bbf9a10b035534f72 : TIMED_OUT in 2h 31m 55s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/011082434be3471cbd243f0788d8b206 : TIMED_OUT in 3h 02m 12s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/889a6ab8a4bd47f1b6d8a32cc8158010 : FAILURE in 4m 14s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/23c177f9df0a45ffa862289dd2f3200e : FAILURE in 3m 11s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/89276464c4c64d52aa349f0bc924f2c1 : FAILURE in 5m 32s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f79f7ff125f7fb43c2ea0318495ffe37102f9c5b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-28 07:07:47.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2259d2014c9ba5c96e5aa5edac335e91e5371096","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-28 07:08:25.000000000","message":"Patch Set 4: Review-Priority+1","accounts_in_message":[],"_revision_number":4},{"id":"5c20bdd8b63903259222f5c5ec0dd02549c6cd68","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-28 07:09:14.000000000","message":"Patch Set 4: Review-Priority+2","accounts_in_message":[],"_revision_number":4},{"id":"a852c36704714cbfa5f834279c1de401d5fccaee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-28 07:31:05.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"49f8bf943e163c96d9a46ea11ecfeb3cd25aa76d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-28 07:32:47.000000000","message":"Patch Set 5: Review-Priority+1","accounts_in_message":[],"_revision_number":5},{"id":"068b233e951b26f6d88a66f17c70627fccab02b1","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-28 09:14:34.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"f3c7d468344b1cb8ae1fc4e2fea60542c26bb69f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 10:29:48.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/26e76fc089a94ac59ea02ba6a200f307\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/7792daa05b1d4f67966b74110c529927 : SUCCESS in 1h 09m 26s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/498395227e6f47e28665c250c72bbc3c : SUCCESS in 1h 17m 03s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/38cae733c7aa498e9d0094a46793dcbe : FAILURE in 28m 21s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/ec3a64027c054c26823b2629faeab76e : SUCCESS in 1h 22m 53s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/de9f05157bcd489d95267799e4f37ea3 : SUCCESS in 32m 40s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0fe5b5ef9ee34737ba6d9ebfc460d861 : SUCCESS in 1h 17m 42s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d6b29349c4d14d208a6ad8270f8cf760 : SUCCESS in 16m 07s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/6bdcc512067e4461b8062dc5594ea88f : SUCCESS in 1h 22m 16s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/a13411690bec4b94bb382deb54b4a9ba : SUCCESS in 1h 25m 01s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/9634b0b5f9954c36b81442d2eebf53c4 : SUCCESS in 1h 44m 54s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/7fcb40be5df040c7ac35df5401f58f8a : SUCCESS in 1h 35m 12s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/11008746923c4d2da1bd474d0e806575 : SUCCESS in 1h 59m 02s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d3ae15d63df3414591f3caa1475c0d62 : SUCCESS in 1h 30m 26s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/608acf2df2914e22b4d8bb7f83843236 : SUCCESS in 1h 58m 47s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/2c59f177aec64050a93377012a59335c : SUCCESS in 2h 04m 21s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/e31a2b782e4541dda1a998ccf6467220 : SUCCESS in 2h 08m 01s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/4e272cdbe3ca4fb6911dd6bdd529747a : SUCCESS in 2h 53m 41s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/54614279998441368befcd49cd0b3cbe : SUCCESS in 2h 10m 00s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/f07dd6367fb14cf99c7093d74725c672 : SUCCESS in 2h 39m 58s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/e481b93ea3c94960910d415d15fe8a09 : SUCCESS in 1h 49m 24s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/be693c75688a4c529721961238d88ed7 : SUCCESS in 1h 56m 45s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/4edeae46ebb74911a996698df301765e : SUCCESS in 1h 55m 06s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/ed6abfb8423c43d4bb427ca5d2d82d5c : SUCCESS in 28m 12s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/9cadef10d5a741dfac50d86f752a4de9 : SUCCESS in 31m 07s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/490cc99400e34730890bf68472fefcba : SUCCESS in 39m 57s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/4be35fbf2cd24478b06ab0a781253f8f : FAILURE in 1h 14m 37s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/198f9f3872e64faa80740dc752075793 : SUCCESS in 1h 02m 01s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/90a7e1da63474ff2b06a5b7a1149ba9f : SUCCESS in 1h 48m 46s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/c8f1342a7b37459d9615547656642da6 : SUCCESS in 1h 29m 42s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/71b792d7312a49a5b4f2e39cc89781c8 : SUCCESS in 1h 32m 48s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/b5f2d6bbd011414f8ed86d28379d1e22 : FAILURE in 2h 12m 09s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/86d6bd701ce848168c82c97afbc29cf8 : TIMED_OUT in 2h 33m 03s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/c120c7aedf924e66b48fe35c28046834 : FAILURE in 1h 43m 27s (non-voting)\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/744cc0cd86d44101b754aa2d9dea1142 : SUCCESS in 2h 11m 22s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/71bf8ff14b2e49439a056324460574cc : SUCCESS in 2h 03m 14s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/2b7f1e1dc5d448be9c214cd39ab50ee0 : SUCCESS in 1h 54m 48s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/088690db01574092a00ab62cef642f92 : SUCCESS in 2m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9acf613d41a46fc9f91d35a53948eb8 : FAILURE in 2m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/65f9adbdd840470d9209b3d36ef652cf : SUCCESS in 3m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/33ac1ffe5e4a48ae80223e8d6d02186c : SUCCESS in 2m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3cef22d1dd9a44b4b374df7334e17bdd : SUCCESS in 3m 52s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/77ab60959b944fbb86d8d7be6fd78469 : SUCCESS in 3m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ec0a99f46a14bd492d49c03aea6238f : SUCCESS in 5m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dc6af635f33445b7a1e895189269cc1e : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":5},{"id":"4721c752a653fee94938032ac646fcafba7f52bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-28 11:41:35.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"50c3733b0eedc9929cb3c4e145224860f467cda2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-28 11:41:40.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":7},{"id":"b7dac01d133d4d3328fd874ec620ebd53f4fe1e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 14:51:27.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5a1f0a5ca5f1465aa182e41af0b01e5d\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/60f13bea921447edb473a7476195b63c : SUCCESS in 1h 06m 44s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/d116d13b92484479806d245cf14f703f : SUCCESS in 1h 29m 18s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/33e6ede863f6406aa31bb6687a043a2e : SUCCESS in 30m 25s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/22d074b2dceb451f876d802bb2fb21b4 : SUCCESS in 1h 01m 20s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/8364552449154a3bbd14e77a2970c191 : SUCCESS in 1h 04m 06s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/efeea2ad5c274b9ca04e6b3dbca45a5d : SUCCESS in 1h 03m 59s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8ed4316937034bbbafba774969921f70 : SUCCESS in 28m 13s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/07903997af6d4ebcb56e699bcd2bce9e : SUCCESS in 1h 33m 25s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/ea6f3efb60434387a561503f3eddcb6a : SUCCESS in 1h 22m 30s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/ec47c2f41a6946b096553081a77ccf01 : SUCCESS in 1h 40m 50s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/b9f466c8236d418abb2f8183891c2be9 : SUCCESS in 1h 34m 23s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/7efe9b17c3bb4591b9b8cf55b082d21f : SUCCESS in 2h 07m 50s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/89dc9643add5461d8932bd6536665475 : SUCCESS in 1h 27m 52s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/e4b6fa4201cd4208bc05e8d71531a42b : FAILURE in 47m 03s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/1f5808dd62894f15ab84e0999f539144 : SUCCESS in 1h 59m 48s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/8858d3e011cc47729715f7fce8a0f6be : SUCCESS in 2h 13m 17s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/a11613500ad94c09be21f61d0ce5a93d : SUCCESS in 2h 52m 37s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/a663a41ceb2c44788513e1384cad1bc7 : SUCCESS in 2h 02m 10s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/d3f1be00926d4dd3a7e646ccfc50832a : TIMED_OUT in 3h 02m 34s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/d0d745ce248e446188167c9bdd946747 : SUCCESS in 1h 08m 43s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/7cea3314822b40faa83f64db3b47d042 : SUCCESS in 1h 19m 39s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/7cb1f979b4c949168c836011bf3ee599 : SUCCESS in 2h 04m 16s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/13bc9391728a4e9387d93665b1dc9f52 : SUCCESS in 33m 03s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/ebb15c095ee947079b9bde3029c3ed7e : SUCCESS in 28m 24s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/460a68e34e134a5cad8d240f3665e05d : SUCCESS in 34m 57s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/cdcc4844bed640259c6242708daad7d4 : FAILURE in 45m 48s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/81afa2b0212449a9864483c9fb952347 : SUCCESS in 1h 44m 03s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/3dd17c809a754d059cc876395110828c : SUCCESS in 1h 37m 47s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/f35d4bbd49e74e06b79fea0d7e46fdac : SUCCESS in 1h 26m 06s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/d7d73a5107cb42cf8c6fe3ca71d3bacd : SUCCESS in 2h 05m 04s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/86fe4d8cfe044b01accec8f81a6acd58 : FAILURE in 2h 20m 22s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/8f0d1c04a3194ca1b0041dc5c2c21a0d : SUCCESS in 1h 29m 07s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/be1bf081ecc244ada3b0a98ff68cf75a : TIMED_OUT in 2h 34m 11s (non-voting)\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/0b4002cca2664b51a94a28361d7206e2 : SUCCESS in 1h 58m 13s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/db4d99cefa994697b80e33b5a8843291 : SUCCESS in 1h 38m 11s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/6d07b7d947e44159ae0e7d946031d390 : SUCCESS in 1h 37m 06s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/8c6c7d142e394961addd191e6e426eb8 : SUCCESS in 3m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/663e2a6537c34e3389d47be40a17ecda : SUCCESS in 3m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4ed4da055bd940d5af2f6543f0a1c349 : SUCCESS in 3m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/97c603b1edbd4dedb8b8e0daf2011d31 : SUCCESS in 3m 27s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a65fbe95217945ba877b054479642604 : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/686191c41aac4fbf98f35badb0da4bdf : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b0cd5662f394a09ac480728cdd0d4ae : SUCCESS in 6m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f0d2ff656d344b69a0e2c787b5bd98a5 : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":7},{"id":"80426395e638f3698795cfa06d2e37aba29a68ca","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-28 15:32:30.000000000","message":"Patch Set 7: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"33830440feb57e2548d26fc06b67585bd74acf56","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-28 15:41:39.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"d1ef52e0b1b594259e487cc497777a9d953565c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 18:43:15.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1e7605e5aa1745148f46e4cf01c8fdd7\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/12de22779c6b4ca5b546ce3fdd6bcb54 : SUCCESS in 1h 07m 02s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/26e7e383254f4aee8bea2db5749066ad : FAILURE in 57m 22s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/7d5fd3671b8543a1b2e79fb0951c97da : FAILURE in 26m 08s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/4f3983ee04584590b9309fc04a8a1279 : SUCCESS in 1h 02m 37s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/7774feff60ce4d00b84365febcce905b : SUCCESS in 47m 58s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/979134fc74614f93a215828792ff3b3d : SUCCESS in 1h 37m 31s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/511712eac67c43eba671278395078ada : SUCCESS in 28m 46s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/61322a81faf14b95950bea58ccd43b85 : SUCCESS in 1h 23m 06s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/b218582bea7845ba9db3fca8dcc3b955 : SUCCESS in 1h 29m 22s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/87db4527a6354385b845125392bd22db : SUCCESS in 1h 43m 26s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/8088f6045f874cb0a710734714e98ab3 : SUCCESS in 1h 00m 33s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/ffd199bd739b4726a45d57f96283aeae : SUCCESS in 2h 09m 12s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/579dc302df1749ef982d60092de696ba : SUCCESS in 1h 25m 49s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/c92aa40029164d6996e847f591c39f58 : SUCCESS in 2h 02m 58s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/088d3eb1da604d4a8b320e3ca5f1f248 : SUCCESS in 2h 41m 12s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/e6772ac895254a969afee499d1614972 : SUCCESS in 1h 33m 27s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/b4c2add6486b4f25a9390c663b3bf518 : SUCCESS in 2h 08m 29s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/e79e39da5ed44fa983f8f8a548d26645 : SUCCESS in 1h 15m 21s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/9ad4032ea6ef48adbd912aeacf8d87b9 : TIMED_OUT in 3h 04m 45s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/17fd35d6cbc54cb3af2a2b045b6a0141 : SUCCESS in 1h 57m 38s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/47bd62ad352a498096983b225d10ae2e : SUCCESS in 1h 16m 06s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/ec87446081fd471384a4bf0c0d5481fd : TIMED_OUT in 2h 33m 53s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/0c000585282845f6b89fa002d05a410b : SUCCESS in 21m 39s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/2339893d777d4bd68753bbdbdfb39bf3 : SUCCESS in 29m 25s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/9996201defaf4d77995759669a708f7a : SUCCESS in 35m 16s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/86bc5112fd454bd191f8969c957f37d3 : FAILURE in 1h 12m 49s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/a38cbdc068564a01bac904d966eda654 : SUCCESS in 1h 29m 55s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/8bc5fed3d9b34d87a9051a592f6a0393 : SUCCESS in 1h 20m 02s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/dff35ecc2dc3494c90f6c09a51f0523f : SUCCESS in 1h 56m 55s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/5dc1ad79cf174e329b502e7093c2e859 : SUCCESS in 2h 17m 42s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/f5d8662e17c0412fa5d0648bb16f46e6 : FAILURE in 2h 07m 00s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/a4e9041c0f15478fb4a8ab68c418cbf5 : SUCCESS in 1h 28m 33s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/986b42fa38164d42820c1fc7154d10b8 : TIMED_OUT in 2h 35m 37s (non-voting)\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/86e76e1014f14fb2a5ae501b29f6a709 : TIMED_OUT in 2h 34m 28s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/4b18cf78dd7e4224bcd9cd9c3bc057d3 : SUCCESS in 1h 15m 56s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/9effb1d3dcdb47ada92eda55e96ac0ae : SUCCESS in 1h 49m 49s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/4d92967c73024ff7b56c9bd704c2a2b2 : SUCCESS in 3m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35bb63e41b5e47cba601d22c670bcc0b : SUCCESS in 3m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/daacbcbe372b45c0b08bd38357072b1c : SUCCESS in 3m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/48f15ebc696746a381a5dbe378054dea : SUCCESS in 3m 15s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/19f9b45c708241d499fdbd0acd42cc09 : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0c5a71d7035349918f0141e90226f31f : SUCCESS in 2m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7da87dbf9de45018bf98a3db4ef9915 : SUCCESS in 6m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/84b8d0ca54bf4097b30744b41a7e6259 : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":7},{"id":"f88519863f078676cc3c4461115262e4885e0245","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-28 18:44:54.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"5a9e45576c91a72e3cd0a3ea91e67092ff9465e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 21:39:49.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f77bb0b407d84844a6a7d2f429738f9b\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/704afd61bed34f79bd2e6d311c364759 : SUCCESS in 1h 03m 21s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/d27c87d84dd04471a8cf5f52fae0f4de : SUCCESS in 54m 24s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/ed8d6b69c53c4d3f8608d19c1f378cc9 : SUCCESS in 12m 50s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/ecb0c1062a7648418cf543d4956b3c1c : SUCCESS in 1h 02m 52s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/29636ad1cc8c4355a85de9422c920d2c : SUCCESS in 34m 42s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5833d81cf7ea49b188cb480ac89646d6 : SUCCESS in 55m 58s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/142e0aa5ed2e42b9804f24bf88ad1ee0 : SUCCESS in 13m 49s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/66abc68e4e3f459da4a9844b1083b1d3 : SUCCESS in 1h 18m 05s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/6d535ac6a35b49429a4943b438def811 : SUCCESS in 1h 20m 32s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/7423eb92eba14ee4b04ccac0a233a974 : SUCCESS in 1h 41m 58s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/1055bca2b816477db60785896f805743 : SUCCESS in 1h 31m 42s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/f3d87e392c21491faf6aee24e80357dc : SUCCESS in 2h 01m 38s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c298ec48b5cc42f7a5fc8e82cc61fcbd : SUCCESS in 1h 26m 51s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/1c9db595426a494da3ec7c510f2a7879 : SUCCESS in 1h 53m 47s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/4b43126e4dc746a38e6319ec4fc7111b : SUCCESS in 1h 58m 20s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/b9978baa5d8a4f5d801faddc70437fa1 : SUCCESS in 2h 16m 03s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/12aa5df1bcd94d9cb30f7df56e9859ec : SUCCESS in 2h 53m 01s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/f67fa1a81fa04087ae091598578f0317 : SUCCESS in 2h 06m 14s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/bc35c4b30c804b1c96ff212d436d5d9e : SUCCESS in 2h 42m 50s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/fdb5ba3d2eef4818ae614bf855735dd8 : SUCCESS in 2h 03m 27s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/d83033c7f0ab4d70bd69f3b61b7fd877 : SUCCESS in 1h 58m 21s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/aa8567d204cc46eba79cfb60eab1e9b8 : SUCCESS in 2h 02m 15s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/497af511e7a24dd8bc05e8320112e923 : SUCCESS in 29m 51s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/ce1bca76b820426cae7d0af4d5a31ae7 : SUCCESS in 33m 12s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/ddeac674e1934445acc6865610256e23 : FAILURE in 37m 54s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/6062fd7a76254e2bafe87d4d369a0f4e : FAILURE in 1h 10m 44s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/cc68143cf0264b53bf59051f60d28a78 : SUCCESS in 1h 33m 46s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/7723d1c814ec48fb9f2950cc2b7a112d : SUCCESS in 52m 02s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/aa15139903ec46ebb7f5a7e22c72ca03 : SUCCESS in 1h 25m 27s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/5ddc0be2b96846daa79a9c6ce46ae647 : SUCCESS in 1h 29m 52s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/9dc4236361e44ad4a68fe77e2d7651da : FAILURE in 1h 39m 12s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/2a48c891db8b43e29a1e1b2084f0c256 : SUCCESS in 2h 21m 05s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/2cb13d4f391e4d38a9a266097d6bf6f6 : FAILURE in 1h 43m 02s (non-voting)\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/74a03b393e2446a29d8ced1ca166e2ea : SUCCESS in 1h 53m 27s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/b4c2cb9fdad1419cb9fc801cfe72c9bb : SUCCESS in 1h 32m 29s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/77b557d1e82748aeb1be82b6a78010e7 : SUCCESS in 1h 51m 23s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/5bf54af285cb4c35b28cc67416d20eae : SUCCESS in 2m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e8dac9fa1354091baa496ca808df056 : SUCCESS in 2m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/278d00e56e2e4109ab9f34f1143f7373 : SUCCESS in 2m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0f2d2e2fbc7c461f8de31f4fa92a52c0 : SUCCESS in 2m 26s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3dc4d91cba5447cf8acb02ab635db877 : SUCCESS in 3m 36s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8b35f45942f24962929b2858e26f9627 : SUCCESS in 3m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e546ff7ac3484e2ebc2db008c33ff5d5 : SUCCESS in 3m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/88d9d67557594046a12ed47f102e40cd : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":7},{"id":"175915c602557b4b33af3870e7b692dc22c08b89","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 04:16:14.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6ab4655484c4503ae454ae20c5f0409\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/1a3f61251abb4bfe9ce857ebdfef4f1d : SUCCESS in 1h 20m 16s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/98d73225bbf644b388d9db5403878a09 : SUCCESS in 1h 56m 57s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/f3efe5d81cfe4528b2b9da1bb309c41a : TIMED_OUT in 3h 09m 18s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/47285572f6c84a2bb0a6aaff8cf07108 : SUCCESS in 1h 57m 36s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/f06f2c2b19eb41e2aaf1c0bf51fd8194 : TIMED_OUT in 2h 32m 31s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/71154147dc1341b3aff6c93885673e81 : TIMED_OUT in 3h 03m 11s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/98ca2343e7af4608a548b2d2613044fd : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a8bfbd3462134e1891bbaf41e42f415f : SUCCESS in 3m 46s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/2d212df32e6a42c6bff109fe3d61f04b : SUCCESS in 7m 44s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"bd4f9dbab52e8ae75f66724debd3dc1c3ca60274","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-29 06:13:13.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"696cc563daee80c2e8daac2c582924fcfdbdfc2f","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-29 06:25:34.000000000","message":"Patch Set 7: Backport-Candidate-1 Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"a3a33f7cb58121cac27cb1d050710e7349a1dde9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 08:48:48.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7541e81eda054aaebed44a585c3afd1b\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/e01307f111fe41f9b02ae73998f47ec7 : SUCCESS in 1h 15m 40s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/eb6487f485d44e8e95fc82a2a5e71dd6 : SUCCESS in 1h 20m 39s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/5c9667288e564dcb898d87bdd25096cb : SUCCESS in 25m 26s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/cd11aa7fc968401487c2995dddc4b69d : SUCCESS in 1h 03m 47s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/746bd56870bd489cb642901ba2cf887b : SUCCESS in 1h 04m 02s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/da28e8c954ed407dbbc492622c34e546 : SUCCESS in 1h 27m 26s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/94f1af4564a04e92bbca8d9496436e16 : SUCCESS in 20m 04s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/1e04d3b97c4f4e68adf3927d542ad0b0 : SUCCESS in 1h 27m 42s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/8fc0c604cd9344678963e22d85e96882 : SUCCESS in 1h 31m 44s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/53361d2af28d41dcb986e9765a26b985 : SUCCESS in 1h 52m 24s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/4ac5b436f3b94e488b13741abb72fdbb : SUCCESS in 1h 41m 50s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/62065ed5bc384bd68f513f902df38e4d : SUCCESS in 2h 12m 31s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/23e27e5b61a74d8fb8c9f00e260a8424 : SUCCESS in 1h 33m 44s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/32741cff938a4d3084aff63bf085f579 : SUCCESS in 2h 08m 27s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/c2b3d634fda34112ae16afdd9df78f1d : SUCCESS in 2h 09m 29s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/173c2c0b590a4e57808313a4a249f0f4 : SUCCESS in 2h 21m 40s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/4f92a0f84a6241bf864f4a9f77ab0caa : FAILURE in 2h 14m 37s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/1e41ba47b5ff413487a225245b9222ed : SUCCESS in 2h 11m 12s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/dfca2c264a93484c8f334bb6cae20b7a : FAILURE in 1h 41m 38s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/8bdd951d6a2b4ca6928ced436e15aaaa : SUCCESS in 1h 53m 27s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/36532de84776420cbf89f948b4c391f5 : SUCCESS in 2h 02m 05s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/2680a597fb4c48e18b4b4764d41d977b : SUCCESS in 1h 56m 39s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/c8a15ef898314e5bb88a7422545a8281 : SUCCESS in 29m 51s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/b63ca4fe4479414b87004d80e3a3bfca : SUCCESS in 36m 08s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/d8dac9ad658747cebe15def93346f159 : SUCCESS in 26m 40s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/5b62ec8b644f4ac78d69782394461b22 : FAILURE in 1h 05m 40s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/e263058867b5451e82a21f501c220895 : SUCCESS in 56m 48s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/683547f71ee84a25b823dea54df6ab09 : SUCCESS in 1h 57m 22s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/3b6e2bb0ede94208a768460eb616a4e4 : SUCCESS in 1h 35m 39s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/52377a94db11462bb77efe1091d9d14d : SUCCESS in 1h 37m 57s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/5a8f61553b3a46c18e36a4277d33bba7 : FAILURE in 1h 43m 54s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/2598c03366094bf598519caa4d5cf758 : SUCCESS in 2h 24m 30s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/d7b5ded650bc44d293c8fb71e2ba4717 : TIMED_OUT in 2h 33m 27s (non-voting)\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/2582532d51954512b8ff0865241e1146 : SUCCESS in 1h 49m 54s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/4d5e0299e4f34be28aa3b479dcba5739 : FAILURE in 2h 02m 50s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/7b53c99ece6b4d68858f4a77303ec380 : SUCCESS in 1h 51m 40s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/9ff9731c6d9d49a5b3726c6559150339 : SUCCESS in 2m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9214503e439e4758bc163a612eafa381 : SUCCESS in 4m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/73f70f8a42b241e79f5495b4b95b4deb : SUCCESS in 5m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/588d5be1010441b4a52bdd87c53ab601 : SUCCESS in 4m 33s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/61aa7a2fa33b4dd4b06eb93dc2663445 : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2c717d3be53740e2a77f07f8c745a410 : SUCCESS in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3afc39eac5ac4376881a282a0a3034be : SUCCESS in 5m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/899db618ca764b85aac716e654a08110 : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":7},{"id":"bcf320e9d31e7879853f440b256fdb06adf382d1","author":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"date":"2026-04-29 17:30:37.000000000","message":"Patch Set 7: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"b4551670ce0b2958a4960ffa741817c8ea6ab4d0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 19:09:52.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d43c56116aae4e16825303c7eb84f7f7\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/e9fa8fc1c9894a6498c4a8d9a9cdb7c8 : SUCCESS in 2h 23m 39s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/60d75138008947d198d434ef14d75f33 : SUCCESS in 1h 22m 44s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/386447287dc34bb5b8566db896af951e : TIMED_OUT in 3h 07m 36s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/428807e95eb44f5a8a9b769f86134ae7 : TIMED_OUT in 3h 28m 05s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/93d4fd7da7ed40589f9fbc15a6e63d14 : FAILURE in 2h 29m 54s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/ab97ad983c5b45fbaa50b71b0d6b45dd : SUCCESS in 2h 08m 56s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/78c59b5fd8f64083afc6014a8c9f47ce : SUCCESS in 3m 53s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8ef5960a15544feb9df3c6807bb33b3f : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/40ab6292d979408a9421be1c897ee5c4 : SUCCESS in 10m 21s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"b0d3c9c042b58f6ec5a78988156db6513cbcec01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-30 04:52:10.000000000","message":"Patch Set 8: Published edit on patch set 7.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"de30deebe444f0d88ed30ec1e1bf134104817690","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-30 04:52:14.000000000","message":"Patch Set 8: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"2afe35dd28ce443267e675b17ef1e39aec55b8d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-30 04:53:47.000000000","message":"Patch Set 9: Published edit on patch set 8.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":9},{"id":"2a57da2392861ec315ebd351f05727c80cbfc4ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-30 04:58:04.000000000","message":"Patch Set 10: Published edit on patch set 9.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":10},{"id":"f62dd2cb818a6d044c496af9281bf1b2e47e0e52","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-30 04:58:17.000000000","message":"Patch Set 10: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"7b701462a3f6b6696de30c2580c0535192e33706","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 07:57:36.000000000","message":"Patch Set 10: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cb07fedbea214ce188a0a5f5a251e3c2\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/59633a5ab200486f80688e0dc7140686 : SUCCESS in 1h 01m 38s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/8feece628a59420fa47628ecf3714db8 : SUCCESS in 1h 14m 44s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/192bd52b54bd47fe84b81d97eeb24cc6 : SUCCESS in 29m 22s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/28cd77c9189e4ba5abfd21425a2ac043 : SUCCESS in 1h 18m 28s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/83f271cd48da43eb935858467fd13b64 : SUCCESS in 1h 07m 41s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9358c9436e4b48d3a04e659fd6cdee9a : SUCCESS in 1h 21m 02s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9a9fc680a0284499abbf007a88f5786e : SUCCESS in 12m 43s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/131b2a3cff9d4b8dac6f20e653cd647e : SUCCESS in 1h 22m 20s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/9915a8449346420281f90b698d560324 : SUCCESS in 1h 23m 11s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/94438656bf954ca090ca7ce7fbaa26d8 : SUCCESS in 1h 45m 05s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/fb257c07819d4880b1a5d22615ace527 : SUCCESS in 1h 33m 51s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/3dbbef2b63094a46bc5332dd4f22fa52 : SUCCESS in 2h 06m 17s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/da97f0b8b26b454ead4d660f467b044f : SUCCESS in 1h 25m 51s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/8bffcb7098d54115b4d6f3ae6231613c : FAILURE in 2h 18m 19s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/b7b8011711134c9cb421a301350f67b3 : FAILURE in 1h 29m 44s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/a24ad06f3962456ca66fb403a9d3ac37 : SUCCESS in 2h 10m 09s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/662ede6fe8064702afed888d181428ea : SUCCESS in 2h 42m 49s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/6131deb702ef45f381358c11a4bdb8d2 : SUCCESS in 1h 56m 34s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/e74a3576bc9f429ba24d5b2078507f29 : SUCCESS in 2h 56m 19s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/20cc1f11696c47f49da34c20db430e21 : SUCCESS in 2h 01m 13s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/575fbd4d74b541648a57937a3b0b4630 : SUCCESS in 2h 05m 20s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/b56e6dc09c1b44c899b04eec528372ca : SUCCESS in 1h 53m 52s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/ab290269055640a9ad1984d7884ebff6 : SUCCESS in 29m 18s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/cccd1f3113cd465bb490d7c631bb8995 : SUCCESS in 30m 20s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/50cffb6917a54595bbba21a941d60a8e : SUCCESS in 34m 49s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/d9318eb4afad458fb00b18f708bbcb9c : FAILURE in 1h 09m 19s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/7a0d3b59cd6942efbec4c48f34fabc47 : SUCCESS in 1h 48m 16s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/e32ca40ed86a47b895ad5e2eed3b1be3 : SUCCESS in 1h 08m 00s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/5de4147b1d4c46a6a7742bf7afb194f6 : SUCCESS in 1h 24m 24s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/41b95d33f76047c9bf0069295a8927e9 : SUCCESS in 1h 26m 54s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/bccf1f2672f74739a1e62fa9f582bd92 : FAILURE in 1h 42m 04s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/60c8a27becbf40d39d149d301a1e92e3 : SUCCESS in 2h 20m 23s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/2dc0d1f96f134bd8aeeda3c6068acf8e : SUCCESS in 2h 30m 22s (non-voting)\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/bb5715b0e371412a82ae42e83e53aa87 : SUCCESS in 1h 51m 41s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/9c9faf743c87478c89aa38c86982a0f7 : SUCCESS in 1h 41m 12s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/7d69a5a3a9fd4ee0816b3fa9033f530d : SUCCESS in 2h 21m 19s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/e4cb2c98897d4f7192fe1bafa0a6b315 : SUCCESS in 2m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00597a22f10c4806b9433768b684fa00 : FAILURE in 3m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ccc7f00cc41a42b2868b844a6423e6d8 : FAILURE in 5m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/582bdcb099ce49cfb18d90de2da23193 : FAILURE in 3m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/dc0cd070d31a4288a384a35628b99e93 : FAILURE in 5m 27s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/38ebbc145b45487ab3fcc59e1d328e1b : FAILURE in 2m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ad013d731e34e469b79db8aae0f0ddc : SUCCESS in 5m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/89b4c94439b54606a82c00bb687ee5ee : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":10},{"id":"fecc76faa6020316098c16af1dbdd976add74bf3","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-30 07:58:03.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"4bbcf8f2d8d5bb0defb968b3ec045f8f5f4bdc4a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 12:45:04.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad9d7e976a9f4727ae417e3b70df3b48\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/9758e4ea77184b43b692138e70f70f39 : TIMED_OUT in 3h 11m 55s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/ef13d73a3b09431ba7125124114277ad : TIMED_OUT in 3h 02m 03s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/42fe4f6d55af497c9ced10a978550ccd : SUCCESS in 1h 31m 55s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/4dfe5a5e3e514697883b524c9eb3d597 : TIMED_OUT in 3h 06m 34s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/13b3312a8b174588ac35b67d6b60863d : TIMED_OUT in 2h 33m 43s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/48b9995bfe0a41cf9d2984d1e09c5634 : TIMED_OUT in 3h 03m 10s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d986d7b02c48483e8d90d562c0315be9 : FAILURE in 5m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ff1f5ba1595146b0b6ce01d89dfaaae2 : FAILURE in 4m 33s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/988f13f7ebed42c2842479daca835ffd : FAILURE in 9m 33s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"c34080cfba8c1c1970d2eba99354d627486a606d","author":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"date":"2026-05-04 08:26:21.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"d05514bbe32167e7b099f40823a3c4c69c522ea8","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-04 09:38:55.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"69e557c85dd4832f3913f8b153a30f355d3ed95f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-04 11:00:46.000000000","message":"Patch Set 11: Published edit on patch set 10.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":11},{"id":"50361c915db0b8203dcaa4060adcbae9175bce4b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-04 13:20:39.000000000","message":"Uploaded patch set 12.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":12},{"id":"e522366875975c5493efb426980433a45e6f51c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 16:21:07.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2a869921f3384a9fad50f46d7dac3a15\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/7d18a29bd7a848879e715f764cb694df : SUCCESS in 1h 04m 29s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/80c50dd173f046b48b2a79abdc4e2042 : SUCCESS in 1h 18m 54s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/99e70eb328334bd5804b9c62c7c1ec9c : SUCCESS in 27m 33s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/6d93e0fe356e456e8677881225b02be5 : SUCCESS in 1h 01m 26s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/47b37316520f40b9ae7d2e40afa78aa9 : SUCCESS in 43m 39s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/169e9c888d9a49c7b5ddf09c7d0f20e4 : SUCCESS in 1h 41m 17s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/36f6f30f5b2040aaa0ed9701593ca604 : SUCCESS in 39m 09s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/0adedcf306944a3798dfdbf72c837a9a : SUCCESS in 1h 23m 41s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/a1f2fe54cede466ca663a8d5c331f784 : SUCCESS in 1h 21m 35s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/ec75b89041d6439f9a70230118bb4940 : SUCCESS in 1h 45m 55s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/5ef8e75f16a7493582b9071d35884b8d : SUCCESS in 1h 31m 32s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/f6532fe5410d47c68ad13c5722225547 : SUCCESS in 2h 13m 55s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/85951928c96744ab80bdd85ae8ee35c0 : SUCCESS in 1h 41m 47s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/79c5756b1b0d40b49c44db22f6effa68 : SUCCESS in 2h 02m 33s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/6b2d9e7acbbb4f3383370eec8561a479 : FAILURE in 1h 39m 12s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/a21fe7c413894137883ac832a7738c3d : SUCCESS in 2h 10m 46s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/5271e2ec7af24f88b4ee5eada319fe6b : SUCCESS in 2h 57m 26s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/4d34c109bbd2458d9e4ec790993eaa98 : SUCCESS in 2h 08m 02s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/27c8325bcf0044d4a6ae62cba00f6d84 : SUCCESS in 2h 44m 14s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/ee883c63231944078fadf5dabbf200dc : SUCCESS in 1h 56m 20s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/493c31c3254547de8c3107c955891217 : SUCCESS in 1h 57m 00s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/2aae3ed3d1f1437faf084535902ae5d7 : SUCCESS in 2h 09m 11s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/20af18af4aa649e881cea5d8adee507d : FAILURE in 28m 28s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/232c3ee1aeb240a3af9ac2f06b226eb2 : SUCCESS in 30m 59s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/53255d05544842b1a9c44fc7253e8090 : SUCCESS in 33m 08s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/098bf7baedc54662aa499284eb9ffb64 : FAILURE in 1h 10m 09s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/ebd100ba1aeb4097a037b1f33b86cc42 : SUCCESS in 1h 36m 08s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/a0b225f607984fd698979f6206c7e5da : SUCCESS in 1h 07m 11s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/e68d13b04e344a5898d8a1784744193e : SUCCESS in 1h 26m 32s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/e29a1b2c4e844f428e0d3d1ed2604cfe : SUCCESS in 1h 33m 40s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/3446e8cae4884658bbc8a6e8b56e2b6e : FAILURE in 1h 38m 49s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/af49aba669ff46148c6ff1a61ea1f22f : SUCCESS in 2h 23m 05s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/e08cc307597a49ceb0ada7719784aa9d : TIMED_OUT in 2h 33m 41s (non-voting)\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/875ad05d48f64bd0823b8470bd7118df : SUCCESS in 1h 22m 53s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/c332e6c40ad14e1eb009d152394cf654 : SUCCESS in 1h 05m 43s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/ffe3bf401cb948bc813caf3422261dad : SUCCESS in 1h 32m 21s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/2ba75813022e4f38b6e39dc9a109b909 : SUCCESS in 2m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ba08ef30cb0429aa1f1edfd15d921ab : SUCCESS in 3m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a2611104fba74c519cf97af93211c03c : SUCCESS in 3m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8bf56f5d7e3646cf9177b7610becd376 : SUCCESS in 3m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fe058a9630744bac9847b4210514e6f5 : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3bccbea6d37041d7b8feec10f9ff5e8d : SUCCESS in 3m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5415dc755896492dafd0062b1d2566ef : SUCCESS in 6m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/28c375364a864157b867159ffad66226 : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":12},{"id":"63f8f51d1e5a0c5f07f5bf292f1076ec37e3da45","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-04 17:26:47.000000000","message":"Patch Set 12: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"d8854414aa02a9a38b2e65b186374b145aa9c864","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 17:52:52.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1a649c5eb4d48dda65a7129d5675a91\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/d5daf03462694a87a62a8c106db807e7 : SUCCESS in 1h 10m 08s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/bbf82977909445e8b70024a6404d8c0a : SUCCESS in 1h 06m 40s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/76166839f71c44a7838155a19851b14a : SUCCESS in 1h 12m 38s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/9d72ba07fde5489591d73f0a12125eef : SUCCESS in 1h 27m 43s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/ef7bb797d1054b35ac475a692c40780f : FAILURE in 1h 31m 53s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/21343683f1e045839d05e102133a139a : FAILURE in 48m 33s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/682b9194f55040f6a1f48ec0b5946f46 : SUCCESS in 4m 13s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4e62e2256aaf42e6bb157c87d7088089 : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/ed8c21f8a27b42b18b8c96aa0d78fd65 : SUCCESS in 5m 39s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"e1a91d95eb381bb3415eebabaa8f615e0eb7ed0c","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-05-04 18:53:26.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"205b4aede4f8c74de9ac914285100ea5f9c90318","author":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"date":"2026-05-04 19:32:52.000000000","message":"Patch Set 12: Code-Review+2","accounts_in_message":[],"_revision_number":12},{"id":"4118ec357d0d55d2280daf51b2b0834ba0d2418b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 20:24:33.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7a9661cbf53044fdab9b7c87d4286392\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/10c6ca2093464b59867daa7711280bb2 : SUCCESS in 1h 03m 50s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/c44b51f932fc4096913d8e762026826a : FAILURE in 58m 20s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/a95b646ac8c14695bc55272c11719478 : FAILURE in 26m 17s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/f8158b79c9344197a90d67af83d681c7 : SUCCESS in 1h 07m 33s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/446da8c60e174250a5f87ec22c694d7f : SUCCESS in 1h 01m 05s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7d35a0525cac4684831b745f201fca33 : FAILURE in 2h 03m 17s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/63c7570b0a2347bfbc973e62f38c0b10 : FAILURE in 32m 45s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/460b2ede379b4476a12266d8bbd1c5a4 : SUCCESS in 1h 22m 46s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/a6d0e93a92684b739e030193874e94f6 : FAILURE in 38m 08s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/b720ea6ad19b4b9fa25a4f190594089b : FAILURE in 31m 00s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/9dd7c84068cf42b48028752d54698d9e : SUCCESS in 1h 44m 32s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/e3567364c18f46e8be92c90141ead5ff : SUCCESS in 2h 10m 29s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/842d5953b0f7458fa3a59df2697493e3 : FAILURE in 48m 10s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/e2ad1d16ae274f968662d68a93697acc : FAILURE in 42m 37s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/fe1369a4cc8247ffacaa20cad638b487 : FAILURE in 42m 17s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/31bcb7587ce043ecbe40fa7e474527f6 : SUCCESS in 2h 15m 34s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/5b92eff9c56c4182a20274c37d7336b2 : SUCCESS in 2h 56m 06s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/031eb69e681b4336be2f56dbce75d7c3 : FAILURE in 52m 58s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/911b335fffb840039e7cdbf384d8058d : FAILURE in 43m 32s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/2efccce59b1b4959aff9bba47be88289 : FAILURE in 39m 42s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/18cf383ded21427da1782ebde301689b : SUCCESS in 2h 09m 27s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/65629d0a5e064eed88b8e4b2041009b3 : FAILURE in 1h 03m 04s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/8ac681124b4246eba11f8f84f5871a2a : TIMED_OUT in 2h 34m 10s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/1c6a5ff3434a49309dde3fb18df2d81f : SUCCESS in 29m 08s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/9ca499fecd9243d28219f61901cf96e9 : FAILURE in 21m 00s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/35cd11f54070476ab967b7526494bb8b : FAILURE in 45m 39s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/0e32475bfe0f4a70a7b7d329b2809e26 : SUCCESS in 58m 00s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/9a09b18b6f6642cfa97000e4064f15c2 : FAILURE in 44m 26s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/bef78fa260ea43c2af02ca38af1131a3 : FAILURE in 38m 50s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/7103f00475a842f789626614cac427d1 : FAILURE in 44m 24s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/9941bef6a052409e96c737868e48d4f4 : FAILURE in 38m 38s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/dbec2a30f5bb4bd19251a020b1c2b731 : SUCCESS in 2h 32m 16s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/b54c9e1cd214475e927f00de8fa62827 : FAILURE in 51m 03s (non-voting)\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/28175a35ef4c4f02be5fa7c80a7e8167 : FAILURE in 40m 51s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/f77fb1e251bd404da0e74e259b00e495 : SUCCESS in 1h 36m 43s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/43e0041bf3f741c2b8f38f143dbf9bc2 : FAILURE in 44m 00s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/1a5ef24c011948f88b83085fbce6b2ac : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a28796e742414f2992dbe6a751ad911a : SUCCESS in 3m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c386258fa1a14be5ae4f8691c8f389f8 : SUCCESS in 2m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e73633a104db4793ac188fa7852f108f : SUCCESS in 3m 15s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8d705d3e9da544049d627f8fc6dd4a05 : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5d256e6fc3354400b3f626d97a547976 : SUCCESS in 3m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7371477cd31c4cb798a67d0849e50597 : SUCCESS in 6m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/10e1604158314381909c523926ba4383 : SUCCESS in 5m 51s","accounts_in_message":[],"_revision_number":12},{"id":"849240ff36e687865a7b7122f547d2cff230dd20","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-05 04:06:14.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"989921c5b498b32989c8f4462f58732fe6da9676","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 07:15:12.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a6510dbcb454466b9c92f31a8f3ba7e3\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/53e7ca5d5a1a42b394b6e4d5cac03458 : SUCCESS in 1h 01m 51s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/7c54b84367244fb2accf9729abb5a7ee : FAILURE in 1h 14m 24s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/ab28b0cdb0e9486e905a98dee7a551cd : FAILURE in 17m 07s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/b0173de3fb62420da52f5862fb47c406 : SUCCESS in 1h 00m 56s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/ebb4e34ba24e4f79b5147e957181dd80 : SUCCESS in 37m 28s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aa2310bd27b54ccd8f9efe1ad46d4fe5 : FAILURE in 1h 33m 19s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/109e98c6ed764997a37289c491c9cce8 : FAILURE in 31m 58s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/92b898f193824be0aed77ef9623fa27a : SUCCESS in 1h 21m 21s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/ecd9f272f56840d0a251e236a3bce2a2 : FAILURE in 31m 37s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/46cfdd83e3884c428a69739bd224a279 : FAILURE in 26m 40s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/952f6a068d2343899c689d8c84c69f95 : SUCCESS in 1h 31m 28s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/a1a621090a0341a6a8f16022012182e0 : SUCCESS in 1h 57m 51s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/38950f0a61b64618a2f065609efc1106 : FAILURE in 54m 21s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/a56bddd9506f4408a54beecb77fbe5c6 : FAILURE in 58m 55s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/b3980cc0100e4669a64650db0371e2a2 : FAILURE in 47m 34s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/b55f115948164a789b694e2e3faccd54 : SUCCESS in 2h 26m 13s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/69445ae076d84e96a7c79e6193d0f59a : SUCCESS in 3h 02m 48s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/f067398282f2493b87cc909862228650 : FAILURE in 53m 45s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/d3b037b2e1554f668cffd1b34abbdc6d : FAILURE in 1h 18m 04s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/28f75ebc008049309c1bf8669642809b : FAILURE in 52m 20s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/3aec8ec91bd64d4894454e492ae4e228 : SUCCESS in 2h 10m 48s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/a5f8b822b6d34609b21b5603ddc97940 : FAILURE in 1h 00m 38s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/74110abb0eaa437f89a0dd224e8069be : FAILURE in 12m 59s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/31d1840e2d2e4028b9f1a408be23498c : SUCCESS in 31m 16s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/2aab0c9e58184c7e9763eefd46ab35f8 : FAILURE in 34m 29s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/f1303eb4439e4497a9eead9d5789381b : FAILURE in 35m 02s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/f01841a7d8e745ab950a12ab03386afc : SUCCESS in 1h 41m 17s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/64d5605fba90453d9b1d6ca419748a94 : FAILURE in 1h 06m 58s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/674081477ee04946b1d8adc949be5e34 : FAILURE in 35m 24s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/2504054d578e4f84b863608a71bedb3f : FAILURE in 42m 31s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/b64e09fa4eea41f4a12c7e04cff7e7eb : FAILURE in 38m 40s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/154e6a81d3ae483f976e700c132155fb : SUCCESS in 2h 20m 32s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/663b80abcb404485aae5763626779395 : FAILURE in 56m 24s (non-voting)\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/dd66d09ea8654a8c9bb36c647551d58b : FAILURE in 35m 58s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/0d9acb76bcbb4e199d8642e8a73468ca : SUCCESS in 2h 03m 10s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/52740c3bfb874bcab8f33eae6064f6ab : FAILURE in 53m 38s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/82b7af235cec4d81895888a522b4854e : SUCCESS in 2m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d06874b51ad147a2a0f9e178d176a40a : SUCCESS in 3m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f738b3ec2673409eb16b1fa6328e1ec2 : SUCCESS in 2m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/670d67937a5e44f8816d5ee936e04ceb : SUCCESS in 3m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bd8155ad9a604130b56d4b2b65361b87 : SUCCESS in 4m 39s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/04bc7ac4626b4e0cb8ee98d8e651a6e2 : SUCCESS in 3m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fd5cc582d2843cda5b0dec8ceaf9da5 : SUCCESS in 7m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b7ef96664f2348e5bc15a2f80f7a3fdb : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":12},{"id":"b5b61a2e8288d931c62270ecc1bc5e09e57cab8d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-05 07:49:55.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"e8c384646956f682390529f0ce670284af086b4d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-05 08:13:52.000000000","message":"Patch Set 13: Patch Set 12 was rebased\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":13},{"id":"a23037d62b3999386c8179c38cabd4170287903f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-05 08:45:58.000000000","message":"Patch Set 14: Published edit on patch set 13.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":14},{"id":"3a33c1e9881bd288c53aa100dea3dc943fcd8861","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-05 10:34:27.000000000","message":"Patch Set 15: Published edit on patch set 14.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":15},{"id":"77d8ac1bad334234b152fe3119e44a6fed93e8f3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 12:08:57.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66589582dbd04dcdbd1d0871c9816fbb\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/ba25478198cc464085fe933bde568e9a : SUCCESS in 1h 21m 59s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/08652f7bc4664079af44433367e8210d : FAILURE in 1h 14m 59s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/0802d35085b347d78da236575542bf10 : SUCCESS in 1h 16m 42s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/2129a0e0f02f4ae78885c4828191b234 : FAILURE in 1h 17m 14s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/0f30726663d841b89c3e54743075ddde : FAILURE in 1h 33m 16s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/d4445ab72e2b47ecb21e8d345fd61d6d : FAILURE in 34m 35s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d0b8da96b36a4db3a15398399d8aeb47 : SUCCESS in 4m 15s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/97c944db44a2443f8d39aa6eccbd86cb : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/213fbde6818d4eb98ed926b42dbe2315 : SUCCESS in 6m 54s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"f5f2532c7339ad725bdfa8c9440065d77ee8ad58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 13:20:48.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2e55d1c9fec54137bc451527cf51ad40\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/ababe49a82314326b6f6dc6d82bc4104 : SUCCESS in 1h 08m 14s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/dca6f2b97dcc4aedafdb1739d71b83bb : FAILURE in 59m 36s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/912e94c222f141c7af2dae46767b6a7d : FAILURE in 25m 43s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/a803f5e5d4b14b3cbb0fd530adf6da86 : SUCCESS in 1h 00m 06s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/c9ea2dd509a34d0ba659a4bf772b720b : SUCCESS in 50m 06s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/693a8f0182f749f0af8d2eea9853c7f3 : FAILURE in 1h 12m 43s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d74c485ccf514a3dadea0cd092616794 : FAILURE in 25m 30s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/55528fb1b7ed41d1b881cf7f9fb6de90 : SUCCESS in 1h 17m 40s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/a65c01dfe8224302bc11151c46e05cb7 : FAILURE in 34m 23s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/36fd9c10c0ed4cab89a67553ffd40dcc : FAILURE in 1h 55m 12s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/2a1d2b3857374cd6aaab15ccea909542 : SUCCESS in 1h 30m 29s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/f2a0f6d3f8454a12b4022a89cab5ddb5 : SUCCESS in 1h 58m 51s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d9b87b438daf420daf310be09c4d1d11 : FAILURE in 33m 24s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/bc36d38655524cad927daa25ae075336 : FAILURE in 34m 27s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/b58e749e4c874c36b516a1b9000bb3df : SUCCESS in 1h 55m 26s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/6942708851394b95813ea3690e01a838 : SUCCESS in 2h 05m 20s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/d78c0b89c0274daf8566195e5d37e9bf : SUCCESS in 2h 44m 59s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/b122af9110c34be48f4ec692626e8077 : FAILURE in 37m 54s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/48ad790a14574858816d36297547faf8 : FAILURE in 40m 37s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/cc25e57261af4799902eb1bcc8d91d9c : FAILURE in 41m 14s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/18215c1bc8524102802c893c27d4a93d : SUCCESS in 1h 55m 40s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/b9f152fece0440a5a73b2d173e5308e0 : FAILURE in 40m 52s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/287028bab99d41cdb1513b71c667b70b : SUCCESS in 27m 41s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/a29db09042e14524bec4d7ebf0ac243f : SUCCESS in 30m 03s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/d0aaf8ff561f49a28cdf5915813048e6 : FAILURE in 13m 36s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/3299f6b3e3e14710937af9d3bcdef37b : FAILURE in 37m 17s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/6a4085c7fb16441ea7236792bfe61564 : SUCCESS in 1h 20m 57s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/dfcd82cbaadf44f79ca18ad944bc270d : FAILURE in 42m 22s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/302db067e94b473792feab12687404a5 : FAILURE in 34m 50s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/47de604efafc4c45a82c8c317129af74 : SUCCESS in 1h 29m 47s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/785f3b5b45bf4bc69f777f598b3c56a0 : FAILURE in 36m 13s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/acb89925b3c74266a2ea6fa48613bfde : SUCCESS in 2h 20m 16s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/6b51857327dd431fb3a7d687c63b083c : SUCCESS in 2h 19m 44s (non-voting)\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/977831460c0c4f81991b00f2fcb3062c : FAILURE in 43m 36s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/21a8f70e237943b096eebcee38658704 : SUCCESS in 1h 29m 37s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/1f74e6af562343a294caa530ddf7a968 : SUCCESS in 1h 20m 14s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/acaa835d79ed40a19bba57bd10063fa0 : SUCCESS in 2m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0fb464304a84c2a95b1e7ca225efc1e : SUCCESS in 2m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fb838c52fec743f7bee7efcb6dd2d1b3 : SUCCESS in 3m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6c985dd7bd0940f89d6f06b2513290cc : SUCCESS in 3m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/df172a70acd74e62b118c03648a69040 : SUCCESS in 5m 55s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3407fe453b748718fd02904698fef53 : SUCCESS in 2m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8ab2bea3cfa4ff78a162ab481990b41 : SUCCESS in 3m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/07c12de4bb2f4160866b249a1df1cca2 : SUCCESS in 2m 49s","accounts_in_message":[],"_revision_number":15},{"id":"b36fd0b68c26eb93e4d7793a275f821452f8b9b7","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-05 14:35:47.000000000","message":"Patch Set 15: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"2fa2d078aa1756270ecfae0555c2d61ce3a3c1f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 17:25:53.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0cd8b235dcfc4034a998c15a1aaf310a\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/4148e1878ba942c29162d7c78ace68d4 : SUCCESS in 1h 03m 12s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/d56e4d14098047aeb16ea457d8d606d0 : FAILURE in 1h 18m 06s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/4df88fd7e2d34e19bce6de4f0b994d0f : FAILURE in 28m 36s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/1a6ad86e88cd475ba70d96aad7e93c8a : SUCCESS in 1h 01m 53s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/8dbce2586f044bd7bff43803ea273b8f : SUCCESS in 35m 07s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/37364a74cddb4b6689123aea9e105602 : SUCCESS in 55m 05s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/84cc89afbe224be78fcbaed12b461753 : FAILURE in 31m 22s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/ac157ccde08943fbbf324d69cd17fa5b : SUCCESS in 1h 21m 05s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/6502a2aae33b4c9ebe7fa19060f5a46d : FAILURE in 33m 24s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/6e27009ef0f040e4a1467106a1eede7c : FAILURE in 32m 27s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/4f2db17ef932452aa13de4ee6664ffaf : SUCCESS in 1h 42m 08s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/593db34fbe9745a2aeeae2fa48b2fcb1 : SUCCESS in 2h 00m 59s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5fb743ced07a46519f5bf11852b0a088 : FAILURE in 35m 04s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/cb0d3a6964f74a80b0bd0b3b2dc5c037 : FAILURE in 34m 05s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/3ba3b99a215a42d5bc89ae6b4cda93a8 : FAILURE in 37m 38s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/c609b05f851047d8973dfef4e4b7ab01 : SUCCESS in 2h 08m 30s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/57f9c5d45f474d3cbfd67a2b7c01d834 : SUCCESS in 2h 48m 17s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/e8b2a57f6c41430f896a2e9d9147b223 : FAILURE in 38m 20s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/cc18a88ce94a484c972fb23b66e248d9 : FAILURE in 38m 11s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/83deddfb29eb4ffebfafbda9b9accaa3 : FAILURE in 42m 10s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/8edc9ae27e984394b29cf3e55242eefa : SUCCESS in 1h 55m 41s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/66ce4229fe2d497993db1a670df89a30 : SUCCESS in 1h 49m 24s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/9d2d3a67fae34d7886573f8fdad01d0d : SUCCESS in 28m 05s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/c0c2ee55d79b4f2b91f95c5ea457a73c : SUCCESS in 30m 03s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/4c7fc03c20c34f808df728535778b2a5 : FAILURE in 19m 14s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/bf576a9992d2453ca58a6e71a4aa6d1e : FAILURE in 52m 05s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/e5e437be3df64fd1985e506235c1994c : SUCCESS in 57m 37s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/218f30ea36a04009b1c249362615fc2c : SUCCESS in 53m 24s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/1473c8d7f8b54a54a75b0d503810c5f7 : FAILURE in 35m 06s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/ed7107541f464e39a9df143239d35e25 : SUCCESS in 1h 27m 15s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/8e82eab55fe64da68bbb791dd21fab6d : FAILURE in 37m 57s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/3c2e4303d6894c60ac1710b45a83a7f5 : SUCCESS in 2h 19m 39s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/8e6df8c3462748779d5d3cbac00f0cae : FAILURE in 38m 29s (non-voting)\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/4affd367850f4d0392fd44d4545d6708 : FAILURE in 34m 05s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/e3483413db8d4eeda102f840b33bca8e : SUCCESS in 1h 40m 02s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/0d8a281cf2ea45c5ba52d0884d6d1dd5 : SUCCESS in 1h 22m 06s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/fe2ac3e2d8924e7c9c497c913c3c967a : SUCCESS in 2m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fa563b4b9cf49aa8aa47e452adb41ce : SUCCESS in 3m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5feb58b582b742899cf0e96112a5a0f6 : SUCCESS in 4m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a56f9278185c45b39fed3f6e13f47c46 : SUCCESS in 3m 44s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a8feadf4577f4dec92eb0532adb736be : SUCCESS in 6m 01s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4ca90d55de5d40af84da305c4d4dc173 : SUCCESS in 3m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/288fe057347349da9162278612ee230d : SUCCESS in 6m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/16232e6f16ad497c9dce58e0cbbf77eb : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":15},{"id":"0c76280c3102f16f25284687e2198a88842593b4","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-05 17:26:48.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"d970eb8d78a5664f9a0b39e5cdc8fca03b1934da","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 17:31:59.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5fa95454dce948798632bc3b16359c04\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/7ae205f497ed42c7aaa00a0470699c6a : SUCCESS in 1h 22m 41s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/2887628d90ea482cb3dd83c4b97a611c : FAILURE in 1h 18m 06s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/1bb84c2c728f4d178741abb4f860caea : SUCCESS in 1h 17m 00s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/28691abe19554f53868adf3ad6bc696b : FAILURE in 1h 24m 24s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/a87219fb223a480a9e32b14c9a9e2d78 : FAILURE in 1h 25m 43s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/a64cb8fd00ae4fb687d37e3dbddf536e : FAILURE in 54m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6feb74e66b4d42e3a90561c1c18f32f6 : SUCCESS in 3m 26s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ad76c2ea900642af9c5c4465a0290337 : SUCCESS in 6m 12s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/a2463cec2f124d88991b421e7680a281 : SUCCESS in 8m 23s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"9c151344cf68ae83957690d1f0c3e5950722fdf0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 20:26:35.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5b3dac7b43146048496440b6e71405b\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/20c3842069344046823d76f5ec13cac3 : SUCCESS in 1h 09m 34s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/e338d31fd25a4c74bf14010da4f6c0ac : SUCCESS in 1h 20m 34s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/9fccdc7a0af84c669a3b96a650fcc31a : SUCCESS in 24m 23s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/0a146c904ea94d4fb18c86a560683765 : SUCCESS in 1h 12m 42s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/2617a5a0c565411a93ba3cc6aedf38ae : SUCCESS in 55m 34s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/de55916a7a68495e8a78d0335f234573 : SUCCESS in 1h 16m 30s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ec3932114dd6445392e088cfe9cc2ffc : SUCCESS in 28m 07s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/4c62e0de59664737999fe31f99e0f59c : SUCCESS in 1h 27m 23s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/94dbd3663ce0470fb3c4dd8358a4635a : SUCCESS in 1h 31m 03s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/02b539e9383148909f527c59aae2edf7 : SUCCESS in 1h 48m 28s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/89dabc863daf42459e3d0d3e9de45fc3 : SUCCESS in 1h 39m 42s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/acb29f60ab144a2d8db87d082f8317ea : SUCCESS in 2h 10m 25s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2d10289313014e84bdcd95ac7dcc4ebd : SUCCESS in 1h 34m 49s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/ab81da631d6a4ce88ab0d3249b0aed1e : SUCCESS in 2h 01m 33s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/9b59106bc31348b7ac21a71f1603b52e : SUCCESS in 1h 59m 56s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/5057be61f806473986091c2a47c12216 : SUCCESS in 2h 16m 08s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/f77de49073854b12b55c19311d5d7983 : SUCCESS in 2h 51m 31s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/2e9cbd71d61c4a9bb9e6acd399e097c5 : SUCCESS in 2h 03m 21s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/2cbd24ac4f7845e9a138d28a04a0f91f : SUCCESS in 2h 58m 04s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/c86710aaa3dd4e058e1809f2ca4f4113 : SUCCESS in 2h 03m 15s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/04c8ac91f81d47ab95d3f83390cc44fe : SUCCESS in 2h 09m 03s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/f114f9cd6b3b4e6297984d00b4e5f646 : SUCCESS in 1h 56m 18s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/0ec990713b2849d3bd8330f57e70e09f : SUCCESS in 39m 49s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/213055e1b97d4ec39454fedf596f879a : SUCCESS in 38m 32s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/f718a374b50d4fa6a28942b2c9a76f86 : SUCCESS in 34m 23s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/6d1eb62f25284030bc6d8d2fdc94d0d3 : FAILURE in 1h 11m 41s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/e8c951f1633d48e0a9736f2dcacfaf21 : SUCCESS in 1h 28m 03s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/9ab2d710bb27429e8bc93fa8d4256e3b : SUCCESS in 1h 37m 40s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/6ac494f638404b0fa5bada2308421012 : SUCCESS in 1h 28m 16s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/079173909d2344b58626e81ace32d5fc : SUCCESS in 1h 30m 18s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/cce9b9e030794c39921dcb562efc9ed7 : FAILURE in 1h 39m 56s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/67b6b1e82ba440c1bab49e3347b6d314 : SUCCESS in 2h 19m 52s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/213d7286e93741b498b8483a3d2eb2b5 : SUCCESS in 2h 18m 45s (non-voting)\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/fb06cfd62a674d7f87dff07d461379f2 : SUCCESS in 1h 50m 32s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/e48c7d4aa9e747a6a7c7dd7cd5b0b81e : SUCCESS in 2h 06m 00s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/ed2653fc73014f5a90b1fd81711bd424 : SUCCESS in 1h 51m 47s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/6ddc41e1c2b346bdbfe2ffd101fd48d1 : SUCCESS in 2m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afab0e19419d41f2bcd14d4f7a7a0948 : SUCCESS in 3m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/93917fe4498347418843d953c675238f : SUCCESS in 3m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/51a3a41a371448d8be613f311860c307 : SUCCESS in 2m 53s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b4c3970ba949499dbb18a935ca06e9a7 : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c2a2abd1c86d4c038bee4007d32354a6 : SUCCESS in 3m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1854f3ae598b407fb733325cdd87ebf1 : SUCCESS in 6m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b8a605af66be4cfab946a6640a264d6c : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":15},{"id":"93b67a590f8746f9d3804c42439540cb90829d4a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-06 03:54:58.000000000","message":"Patch Set 15: Backport-Candidate-1 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":15},{"id":"1123e095ab60a97652936ccceb353fd411d9e1cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 03:55:25.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"4960879504b21ec3532e28e3d8a051373dea7055","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 06:00:17.000000000","message":"Patch Set 15: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9dcdb0bf817f471f92dbb2ee3bbc31ae\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/8b9a455a00874b0f92693191db5a138f : SUCCESS in 59m 24s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/ef64ceb0d7ed4fd188090a67bf52f482 : SUCCESS in 20m 04s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/e6302ac4a27b4ace8adec851eb5bd9d5 : SUCCESS in 1h 01m 29s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/4542e341711141bf97246ca84b2f7ec7 : SUCCESS in 32m 22s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0fdf20c27d1b4036aa7fd12fc95b82fb : SUCCESS in 1h 20m 09s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d6ea7bbf85144fd893ba680aa4c0534e : SUCCESS in 27m 56s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/48022665f4444f04bb04559fbf456411 : SUCCESS in 1h 22m 47s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/d19e21272bfc44ca8b6fa7baeb5e707f : SUCCESS in 1h 45m 04s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/c03fef3ff40e4fe48f5f84e039f3d65b : SUCCESS in 1h 36m 25s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/616f3f3907a94f5cb6f42e1377d60513 : SUCCESS in 2h 04m 02s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ec1186e17e6943eaba0e50c715a4b18d : SUCCESS in 1h 21m 38s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/dac9f577287c42c898b4953209371cf6 : SUCCESS in 1h 53m 45s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/00e3aa0f4f77490eb85927b51c802ae1 : SUCCESS in 2m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/160e098117f744bc850aaeef31ca794a : SUCCESS in 2m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/70e99c8799c841a992a3e9411b031299 : SUCCESS in 4m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c5edb7cc0d3a4a6f9df237d48eb1a90f : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/428186502b73428c9ca3d00e795e647e : SUCCESS in 6m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8f01447f360347e3b446166d0d2cc26d : SUCCESS in 2m 58s","accounts_in_message":[],"_revision_number":15},{"id":"a329af42016aa084a8b811363a871f53935d343e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 06:00:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":15},{"id":"800c8ccd130688f1818e674fe6707c6ba54be6ac","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 06:01:28.000000000","message":"Patch Set 15:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9146d4aecd340fe8496684554164680\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0db433e9d4ac4f03ab01f3cc5566cb17 : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/941d88195d2a49b982d4ec2d80bef917 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"33c0e26e8a0290f0b6d647bb67e18a4f500907fc","revisions":{"aae3b6c09aff56e2966b9f1fee37476836994426":{"kind":"REWORK","_number":1,"created":"2026-04-27 13:36:44.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/77/986377/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/77/986377/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/986377/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/986377/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/986377/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/77/986377/1"}}},"commit":{"parents":[{"commit":"afc09fc357d936f7c5c7b9ee17255623924352a2","subject":"Merge \"Update Prometheus to version 3.5.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/afc09fc357d936f7c5c7b9ee17255623924352a2"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-27 13:28:17.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-27 13:36:35.000000000","tz":120},"subject":"set_configs: Add support for creating directories and setting perms","message":"set_configs: Add support for creating directories and setting perms\n\nChange-Id: Id63c035fafaa68a9004c65161c5a3a3ad5544f02\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aae3b6c09aff56e2966b9f1fee37476836994426"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aae3b6c09aff56e2966b9f1fee37476836994426"}]},"branch":"refs/heads/master"},"60c81c1a62fb392a67013086e6dbade21881446f":{"kind":"REWORK","_number":2,"created":"2026-04-27 13:52:13.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/77/986377/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/77/986377/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/986377/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/986377/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/986377/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/77/986377/2"}}},"commit":{"parents":[{"commit":"afc09fc357d936f7c5c7b9ee17255623924352a2","subject":"Merge \"Update Prometheus to version 3.5.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/afc09fc357d936f7c5c7b9ee17255623924352a2"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-27 13:28:17.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-27 13:52:07.000000000","tz":120},"subject":"set_configs: Add support for creating directories and setting perms","message":"set_configs: Add support for creating directories and setting perms\n\nChange-Id: Id63c035fafaa68a9004c65161c5a3a3ad5544f02\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/60c81c1a62fb392a67013086e6dbade21881446f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/60c81c1a62fb392a67013086e6dbade21881446f"}]},"branch":"refs/heads/master"},"656db2e4d5154e32521a9acf8e3ebeb396081f86":{"kind":"REWORK","_number":3,"created":"2026-04-27 19:24:05.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/77/986377/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/77/986377/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/986377/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/986377/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/986377/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/77/986377/3"}}},"commit":{"parents":[{"commit":"afc09fc357d936f7c5c7b9ee17255623924352a2","subject":"Merge \"Update Prometheus to version 3.5.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/afc09fc357d936f7c5c7b9ee17255623924352a2"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-27 13:28:17.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-27 19:24:05.000000000","tz":0},"subject":"set_configs: Add support for creating directories and setting perms","message":"set_configs: Add support for creating directories and setting perms\n\nChange-Id: Id63c035fafaa68a9004c65161c5a3a3ad5544f02\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/656db2e4d5154e32521a9acf8e3ebeb396081f86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/656db2e4d5154e32521a9acf8e3ebeb396081f86"}]},"branch":"refs/heads/master"},"4e1121901a5f883f10acb85069e328142540bd9a":{"kind":"REWORK","_number":4,"created":"2026-04-27 19:24:46.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/77/986377/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/77/986377/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/986377/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/986377/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/986377/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/77/986377/4"}}},"commit":{"parents":[{"commit":"afc09fc357d936f7c5c7b9ee17255623924352a2","subject":"Merge \"Update Prometheus to version 3.5.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/afc09fc357d936f7c5c7b9ee17255623924352a2"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-27 13:28:17.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-27 19:24:46.000000000","tz":0},"subject":"set_configs: Add support for creating directories and setting perms","message":"set_configs: Add support for creating directories and setting perms\n\nChange-Id: Id63c035fafaa68a9004c65161c5a3a3ad5544f02\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4e1121901a5f883f10acb85069e328142540bd9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4e1121901a5f883f10acb85069e328142540bd9a"}]},"branch":"refs/heads/master"},"fbd6c78a710ef81a81d013f72dfdb53d65380074":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-04-28 07:31:05.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/77/986377/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/77/986377/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/986377/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/986377/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/986377/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/77/986377/5"}}},"commit":{"parents":[{"commit":"a1ce6387d7525dbb6828e8d7c3488e65ba8c37cb","subject":"keystone: Remove wsgi file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a1ce6387d7525dbb6828e8d7c3488e65ba8c37cb"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-27 13:28:17.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-28 07:31:05.000000000","tz":0},"subject":"set_configs: Add support for creating directories and setting perms","message":"set_configs: Add support for creating directories and setting perms\n\nChange-Id: Id63c035fafaa68a9004c65161c5a3a3ad5544f02\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fbd6c78a710ef81a81d013f72dfdb53d65380074"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fbd6c78a710ef81a81d013f72dfdb53d65380074"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"4e1121901a5f883f10acb85069e328142540bd9a","theirs":"a1ce6387d7525dbb6828e8d7c3488e65ba8c37cb","contains_conflicts":false}},"ac4d7a4a79af1000dc9c9106294dba81fa09b66d":{"kind":"REWORK","_number":6,"created":"2026-04-28 11:41:35.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/77/986377/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/77/986377/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/986377/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/986377/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/986377/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/77/986377/6"}}},"commit":{"parents":[{"commit":"a1ce6387d7525dbb6828e8d7c3488e65ba8c37cb","subject":"keystone: Remove wsgi file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a1ce6387d7525dbb6828e8d7c3488e65ba8c37cb"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-27 13:28:17.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-28 11:41:35.000000000","tz":0},"subject":"set_configs: Add support for creating directories and setting perms","message":"set_configs: Add support for creating directories and setting perms\n\nChange-Id: Id63c035fafaa68a9004c65161c5a3a3ad5544f02\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac4d7a4a79af1000dc9c9106294dba81fa09b66d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac4d7a4a79af1000dc9c9106294dba81fa09b66d"}]},"branch":"refs/heads/master"},"efa2fdfc89490c507dd659ccdace99dd1cdfc742":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-04-28 11:41:40.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/77/986377/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/77/986377/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/986377/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/986377/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/986377/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/77/986377/7"}}},"commit":{"parents":[{"commit":"78146c684d0f35f51000b178817c3eb7f4c12329","subject":"keystone: Remove wsgi file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/78146c684d0f35f51000b178817c3eb7f4c12329"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-27 13:28:17.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-28 11:41:40.000000000","tz":0},"subject":"set_configs: Add support for creating directories and setting perms","message":"set_configs: Add support for creating directories and setting perms\n\nChange-Id: Id63c035fafaa68a9004c65161c5a3a3ad5544f02\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/efa2fdfc89490c507dd659ccdace99dd1cdfc742"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/efa2fdfc89490c507dd659ccdace99dd1cdfc742"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"ac4d7a4a79af1000dc9c9106294dba81fa09b66d","theirs":"78146c684d0f35f51000b178817c3eb7f4c12329","contains_conflicts":false}},"f907a29637934072b45d2754779e37b944ff6fcf":{"kind":"REWORK","_number":8,"created":"2026-04-30 04:52:10.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/77/986377/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/77/986377/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/986377/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/986377/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/986377/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/77/986377/8"}}},"commit":{"parents":[{"commit":"78146c684d0f35f51000b178817c3eb7f4c12329","subject":"keystone: Remove wsgi file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/78146c684d0f35f51000b178817c3eb7f4c12329"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-27 13:28:17.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-30 04:52:08.000000000","tz":0},"subject":"set_configs: Add support for creating directories and setting perms","message":"set_configs: Add support for creating directories and setting perms\n\nChange-Id: Id63c035fafaa68a9004c65161c5a3a3ad5544f02\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f907a29637934072b45d2754779e37b944ff6fcf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f907a29637934072b45d2754779e37b944ff6fcf"}]},"branch":"refs/heads/master"},"2bc1da04fb5684513003cc6e5d12ff15e142d746":{"kind":"REWORK","_number":9,"created":"2026-04-30 04:53:47.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/77/986377/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/77/986377/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/986377/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/986377/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/986377/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/77/986377/9"}}},"commit":{"parents":[{"commit":"78146c684d0f35f51000b178817c3eb7f4c12329","subject":"keystone: Remove wsgi file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/78146c684d0f35f51000b178817c3eb7f4c12329"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-27 13:28:17.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-30 04:53:46.000000000","tz":0},"subject":"set_configs: Add support for creating directories and setting perms","message":"set_configs: Add support for creating directories and setting perms\n\nChange-Id: Id63c035fafaa68a9004c65161c5a3a3ad5544f02\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2bc1da04fb5684513003cc6e5d12ff15e142d746"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2bc1da04fb5684513003cc6e5d12ff15e142d746"}]},"branch":"refs/heads/master"},"93e50d69f62c0dbda13f2270d76848351d452805":{"kind":"REWORK","_number":10,"created":"2026-04-30 04:58:04.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/77/986377/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/77/986377/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/986377/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/986377/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/986377/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/77/986377/10"}}},"commit":{"parents":[{"commit":"78146c684d0f35f51000b178817c3eb7f4c12329","subject":"keystone: Remove wsgi file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/78146c684d0f35f51000b178817c3eb7f4c12329"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-27 13:28:17.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-30 04:58:04.000000000","tz":0},"subject":"set_configs: Add support for creating directories and setting perms","message":"set_configs: Add support for creating directories and setting perms\n\nChange-Id: Id63c035fafaa68a9004c65161c5a3a3ad5544f02\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/93e50d69f62c0dbda13f2270d76848351d452805"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/93e50d69f62c0dbda13f2270d76848351d452805"}]},"branch":"refs/heads/master"},"f1f77e732f4167c8eb4e16cf805757398b765bbd":{"kind":"REWORK","_number":11,"created":"2026-05-04 11:00:46.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/77/986377/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/77/986377/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/986377/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/986377/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/986377/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/77/986377/11"}}},"commit":{"parents":[{"commit":"78146c684d0f35f51000b178817c3eb7f4c12329","subject":"keystone: Remove wsgi file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/78146c684d0f35f51000b178817c3eb7f4c12329"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-27 13:28:17.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-05-04 11:00:46.000000000","tz":0},"subject":"set_configs: Add support for creating directories and setting perms","message":"set_configs: Add support for creating directories and setting perms\n\nChange-Id: Id63c035fafaa68a9004c65161c5a3a3ad5544f02\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f1f77e732f4167c8eb4e16cf805757398b765bbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f1f77e732f4167c8eb4e16cf805757398b765bbd"}]},"branch":"refs/heads/master"},"a460d9986ae92354d0b2e5049aedbaa5ac4bf1c3":{"kind":"REWORK","_number":12,"created":"2026-05-04 13:20:39.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/77/986377/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/77/986377/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/986377/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/986377/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/986377/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/77/986377/12"}}},"commit":{"parents":[{"commit":"78146c684d0f35f51000b178817c3eb7f4c12329","subject":"keystone: Remove wsgi file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/78146c684d0f35f51000b178817c3eb7f4c12329"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-27 13:28:17.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-05-04 13:20:33.000000000","tz":120},"subject":"set_configs: Add support for creating directories and setting perms","message":"set_configs: Add support for creating directories and setting perms\n\nChange-Id: Id63c035fafaa68a9004c65161c5a3a3ad5544f02\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a460d9986ae92354d0b2e5049aedbaa5ac4bf1c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a460d9986ae92354d0b2e5049aedbaa5ac4bf1c3"}]},"branch":"refs/heads/master"},"197c95211c9391efa7a52e4a22c7cb1de3bf3a56":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2026-05-05 08:13:52.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/77/986377/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/77/986377/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/986377/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/986377/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/986377/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/77/986377/13"}}},"commit":{"parents":[{"commit":"1627e5da6d6ac754be3e0afe61eeddf6e38f93b6","subject":"Merge \"Remove Babel from openstack-base\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1627e5da6d6ac754be3e0afe61eeddf6e38f93b6"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-27 13:28:17.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-05-05 08:13:52.000000000","tz":0},"subject":"set_configs: Add support for creating directories and setting perms","message":"set_configs: Add support for creating directories and setting perms\n\nChange-Id: Id63c035fafaa68a9004c65161c5a3a3ad5544f02\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/197c95211c9391efa7a52e4a22c7cb1de3bf3a56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/197c95211c9391efa7a52e4a22c7cb1de3bf3a56"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"a460d9986ae92354d0b2e5049aedbaa5ac4bf1c3","theirs":"1627e5da6d6ac754be3e0afe61eeddf6e38f93b6","contains_conflicts":false}},"30b74b9be3a85e061cbce3e62dca8ad386eea8f8":{"kind":"REWORK","_number":14,"created":"2026-05-05 08:45:58.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/77/986377/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/77/986377/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/986377/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/986377/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/986377/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/77/986377/14"}}},"commit":{"parents":[{"commit":"1627e5da6d6ac754be3e0afe61eeddf6e38f93b6","subject":"Merge \"Remove Babel from openstack-base\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1627e5da6d6ac754be3e0afe61eeddf6e38f93b6"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-27 13:28:17.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-05-05 08:45:58.000000000","tz":0},"subject":"set_configs: Add support for creating directories and setting perms","message":"set_configs: Add support for creating directories and setting perms\n\nChange-Id: Id63c035fafaa68a9004c65161c5a3a3ad5544f02\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/30b74b9be3a85e061cbce3e62dca8ad386eea8f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/30b74b9be3a85e061cbce3e62dca8ad386eea8f8"}]},"branch":"refs/heads/master"},"33c0e26e8a0290f0b6d647bb67e18a4f500907fc":{"kind":"REWORK","_number":15,"created":"2026-05-05 10:34:27.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/77/986377/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/77/986377/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/986377/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/986377/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/986377/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/77/986377/15"}}},"commit":{"parents":[{"commit":"1627e5da6d6ac754be3e0afe61eeddf6e38f93b6","subject":"Merge \"Remove Babel from openstack-base\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1627e5da6d6ac754be3e0afe61eeddf6e38f93b6"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-27 13:28:17.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-05-05 10:34:26.000000000","tz":0},"subject":"set_configs: Add support for creating directories and setting perms","message":"set_configs: Add support for creating directories and setting perms\n\nChange-Id: Id63c035fafaa68a9004c65161c5a3a3ad5544f02\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/33c0e26e8a0290f0b6d647bb67e18a4f500907fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/33c0e26e8a0290f0b6d647bb67e18a4f500907fc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-1"],"failing_atoms":["label:Backport-Candidate\u003d1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
