)]}'
{"id":"openstack%2Fkolla~986430","triplet_id":"openstack%2Fkolla~master~I9f4243a40d0907639324308d84d8a860c1e54941","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-04-29 08:16:50.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-04-29 08:28:29.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"37306":{"account":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"last_update":"2026-05-04 09:48:46.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-04-29 07:03:11.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"I9f4243a40d0907639324308d84d8a860c1e54941","subject":"Update Prometheus to version 3.5.3","status":"MERGED","created":"2026-04-27 17:08:24.000000000","updated":"2026-05-04 09:49:42.000000000","submitted":"2026-05-04 09:48:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"986430","meta_rev_id":"a53ac74e9f2c9df407d389068a466cb9c4767899","_number":986430,"virtual_id_number":986430,"owner":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-04 09:48:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2026-04-29 08:16:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2026-04-29 08:28:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2026-05-04 07:41:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-04-29 07:03:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2026-05-04 07:41:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2026-04-29 08:16:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2026-04-29 08:16:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-27 18:53:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-27 19:44:43.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2026-04-28 09:42:36.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-04-28 10:13:45.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2026-04-28 20:16:07.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2026-05-04 07:41:37.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"958730a5baa07573903bb4bd015783a01c1015a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2026-04-27 17:08:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"64d63b3c0dbb6318c1355cef2fb6a10d213cad40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 18:53:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f3ef97e58a9048a898d3ab13ca19bf54\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/7415f8413a574f41b3afe138ef452869 : SUCCESS in 1h 20m 50s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/960405ee0a02434f8279d9cdccf51331 : FAILURE in 1h 39m 52s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/1effd15c72c640659b6d195b52d28d42 : SUCCESS in 26m 09s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/e99735b1e00c4563be0b89b0e83e4576 : SUCCESS in 1h 02m 19s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/4d215a0d4f9f481591ca6f80ccca2809 : SUCCESS in 48m 51s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/18ba6e15e4884e4b9fbba981ff9e8cd4 : SUCCESS in 1h 33m 23s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5d15f69af19a414baebf6c3ae2cda317 : SUCCESS in 27m 33s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/16bf4a4e02bb4724b6d3243863b97e3c : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d56de70b1c1476ebeff43210d04b8d5 : SUCCESS in 4m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/33c0eb88ef284eddb6cba40ba4cfdabd : SUCCESS in 3m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/29943813d3ec4ad7b8beb0291cace057 : SUCCESS in 3m 33s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1abdc6b196af436586f72fbb6710762e : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/55e18f177e06437e9159d9b29e918415 : SUCCESS in 3m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0c887ba72f743da9d5929ffc590baa9 : SUCCESS in 6m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/09b2c7c84bc24df5b24e28ca91118f7a : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":1},{"id":"91c2f0e8f74e6f5f884a0ee0772d57411cb97bc4","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2026-04-27 19:21:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"28b66b1b1441c4d4ef02186f570751540c30c4a7","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-27 19:44:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"526be0ac3dc262a033217b559dec74a2e2e08eb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 20:44:26.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8d95e9b87e5943fa82dccc5f2c614dce\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/f08823b7e61c434d966812a13e75cc47 : SUCCESS in 1h 03m 02s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/bb2cbd3679134335aee739af5fa6077d : SUCCESS in 1h 01m 38s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/84ad26f763b041dc9e9e265ff0990e57 : SUCCESS in 19m 56s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/9a3740712b9048cd928593871f41f948 : SUCCESS in 1h 03m 15s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/c244e8b6f3974ed18867ba60cbe09066 : SUCCESS in 34m 02s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/70f62b99fc474f9391031f8325e66a87 : SUCCESS in 1h 18m 24s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ee555a0dfaa44767a43427546a60fd11 : SUCCESS in 27m 31s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/c7267a0300644143bc7f798419e09a52 : FAILURE in 2m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d9b8fd88a2a49ff865af9c584116635 : FAILURE in 1m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/de88d5434df240f1ac44b38a0454d2f0 : FAILURE in 2m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/790b403f5215419ba28c86118e4299b0 : FAILURE in 2m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/37afd74a49da446bb1f261a2f32b60bc : FAILURE in 4m 29s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e1f0026169e46d0abefb0acb6306aeb : FAILURE in 2m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57789c6e8d3e476aafb4cdd1e27b0ee3 : SUCCESS in 6m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9b225355e65e46d490bf9f81209b013c : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":1},{"id":"f7619dde39281385b215c22860b1db7647c95ba3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 22:58:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8cf6a11b8804e65a2f2fcaee68bf8f3\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/077b1ba179e2491098eac6ba99822a3d : SUCCESS in 1h 19m 37s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/283978f5723f481194d8ed976c787d8c : TIMED_OUT in 3h 17m 54s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/d45673d4a13c42bb82717d651069f016 : TIMED_OUT in 3h 18m 38s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/cf6205effde7490f9fa6c86cdbc5cf3d : TIMED_OUT in 3h 15m 01s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c1bbfd6e0bfb437f9f928bed7d780fc5 : SUCCESS in 4m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d2aa7a43f0cf44b8bf51098519b133d6 : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/4a8997a8e2ee4ca082dae7037564dbd1 : SUCCESS in 9m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4f295577cbdd188a5063224bcd7e712d444dee67","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2026-04-28 06:17:11.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"796a47461cbdb640c23b53664b7d9f8872ad11fc","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2026-04-28 06:17:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0b2198c3c026e28aada190163cd0e52f478774e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 07:52:07.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f9e32d9db96b45bab2eff66cc52c82ac\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/7f101623f8cf4acb9a842368ade6ada6 : SUCCESS in 1h 11m 27s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/a4934d8a49f040bf8f73715518251206 : SUCCESS in 1h 28m 12s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/066b5d6fd2dc463c988f3687dcb47d32 : SUCCESS in 11m 25s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/415a41d06548487ea0701493841f3618 : SUCCESS in 1h 00m 46s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/d21366f7ddaa4b53b12e1d1bcdebc5ed : SUCCESS in 36m 28s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7f27a1e3870a4aa6a2062d2f37f88203 : SUCCESS in 1h 19m 10s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f61d9f9aefcc4b1ea0cb56f310917d59 : SUCCESS in 31m 08s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/dc004caaa7954312a20b0f42817a50ee : FAILURE in 3m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b1dffcf37294de181c5df6edfd833d2 : FAILURE in 1m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/607e7d6b03864fe98fc59fd5e401b6fb : FAILURE in 2m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9d924fd8e81549e1ad64d3ae8fe72391 : FAILURE in 2m 26s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0eaaf03eb6664d89b543d0de402a4675 : FAILURE in 4m 39s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f615c59bc9bc4dc3bbd7572c69c33f9e : FAILURE in 1m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90a2fd560dd74936a145e15e49afd200 : SUCCESS in 7m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/080f8f5c7bc84adc9d4f95a2b906b473 : SUCCESS in 3m 28s","accounts_in_message":[],"_revision_number":2},{"id":"762d198bab50ab2bc9635099199829d05a3fa739","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2026-04-28 08:48:13.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"73a17186a1d7b103415bf37c980c9b92f198c736","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-28 09:29:28.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f1d1dc032a7b83b9bea2594eac26a55f7267d4e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 10:11:30.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63034188fb214aa1a2a40d0dfbe1712d\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/ebec48b9a20049c499c9134cce033d76 : FAILURE in 1h 06m 58s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/95492c408e6a4ef18b342cdffa65fa44 : SUCCESS in 54m 31s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/10a1b568c15747578c6449868a30acc0 : SUCCESS in 28m 03s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/ddda39d0d2e04cd29fa81e11af8294aa : SUCCESS in 1h 05m 28s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/82fa6f79ded943609288a996a152acae : SUCCESS in 1h 04m 25s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ea05517efdf54286aac063c14c8c462b : SUCCESS in 1h 02m 03s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/40767c603b6945158716693860ccce74 : SUCCESS in 28m 51s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/8384fade7b56455ca16047a66f24af96 : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4efac0b2d02746a99814ebdada7e1bc4 : SUCCESS in 3m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/baa6644d4eae42e19c5bcd582508b9ab : SUCCESS in 3m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0db62a6b765a42e5a43ed77cb8fc3cb2 : SUCCESS in 3m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/93e02eb68ebc4b32a724281b51341304 : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/36a39506157c416fb21c6a7d4ed30522 : SUCCESS in 2m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d50969ea47940909776abe126c41da7 : SUCCESS in 7m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fdbb40a43af945789d41846a92d57fc8 : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":3},{"id":"bbd6b9f11e2ece6aed6c6326eda09b6e770deef6","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-28 10:12:51.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"a1d643ad6d9ea587b90728ff9ebdffa43b22e075","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-28 10:13:45.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"e719201fb3c0d6b3c22b0104b11c3aa155b9691d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-28 10:13:52.000000000","message":"Patch Set 3: Backport-Candidate+1","accounts_in_message":[],"_revision_number":3},{"id":"dbb76087facec7ce51d3cbb290392c0ce1574c3f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 17:32:02.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1298718b361a4c2d8e8ac4d1db7cc20d\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/250f6a37de6d453dae2a905beb48553a : TIMED_OUT in 3h 04m 06s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/1849d90295f5478db555b18eb01c84e0 : SUCCESS in 2h 42m 35s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/e6d306d04b88401992adb6054679a73d : SUCCESS in 2h 38m 34s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/92ed4a5c840f4a2a8e546b2ffa9cfb52 : SUCCESS in 1h 49m 55s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3d31a4222b1342ab9e63fc399badb2da : SUCCESS in 3m 56s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/17454e77f67841a2ac0ed6d4fbdb56cd : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/10aa47f7b2564b319b8247aee7cd3f4c : SUCCESS in 9m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7dfa9453374d74e5274907ee16eec3218d59e348","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-04-28 20:16:07.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b82e7da140dd783a75cd0c48fc406351cbdb43f4","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-04-28 20:16:56.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"e8bdf347c3de8729446f1646b15494838ed5ab2d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2026-04-29 06:44:17.000000000","message":"Uploaded patch set 4.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"aa90ff62e096327c52e6a023b221677f39da01dd","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2026-04-29 06:44:34.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"ea191f47aa14726da858d16593f9ce8b13483115","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-04-29 07:03:11.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"e233214d3e762660a3011522f3c8f74863ca555d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-04-29 07:04:44.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ffb6fc102707d138b1bda137d414c6bb2f4550a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 08:09:29.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/635a5698fe974fe5a2905ef0e019eca9\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/f8d29e62517c4aaabf21719345c6c0b2 : SUCCESS in 1h 15m 58s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/81e244d2ede74b0db49413bb0b5a38ec : SUCCESS in 1h 20m 23s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/7ca364958cc6472b92c00b53b03b7c0a : SUCCESS in 26m 26s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/b3c91bb542d84091918b7c1e0fc2a0f3 : SUCCESS in 1h 19m 31s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/077b5c10fb2b4c9d884a6889e271a688 : SUCCESS in 1h 15m 47s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/852d4c786f594b20aab57c2b3282b9dd : SUCCESS in 1h 22m 42s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c138c198536549b9be7b590a2c2ac8d0 : SUCCESS in 29m 30s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/efac9e197a814d46a9d897c93ebd0098 : SUCCESS in 1m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75f9a00f22e3453abcb2467224043c90 : SUCCESS in 2m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ada76bb339654650976854ba5cc74d10 : SUCCESS in 1m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cce06f9b64764ed58431f57029e3b1bd : SUCCESS in 3m 12s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/03e0cd57f8f44fde8ab293578d1690e0 : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8391b11650db411a991b44abfd1eb3ec : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8245b530ba04ddbb6ac3672ae540083 : SUCCESS in 7m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a2d44fe41b0d464d8f1b5eaf8306e6a2 : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":4},{"id":"1b0b2501d4baa031711efd233aad4efe4f866f70","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-29 08:16:50.000000000","message":"Patch Set 4: Backport-Candidate+1 Code-Review+2 Review-Priority+2","accounts_in_message":[],"_revision_number":4},{"id":"1c9e9040385dd69766e3ec5a991f9b63c3caa4fc","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-29 08:28:29.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"c213add95b3258a532df64af8ea3efce63aefd1c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 15:12:59.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d77fdf38779d4882ae422fa6b2e20619\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/1f6e66c458024ff3ac309cc3d6334c24 : TIMED_OUT in 3h 11m 31s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/ed4154943d104905951e7c0407f408fb : TIMED_OUT in 3h 28m 37s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/1d21da27cba7408a8db8ff59a83d93be : TIMED_OUT in 3h 12m 23s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/9ad1f40e35fb42d69e822625a037b843 : TIMED_OUT in 3h 16m 15s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4d1ceab0a81041878cd2bd90f854d7b2 : SUCCESS in 4m 06s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bd8b5c2bf7ba41a28adf5ee9d0d4e756 : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/12794d8c886a4ea2bcda06b007af2e43 : SUCCESS in 8m 37s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7c971e07fb3533a1b2810f21f869ee4d75e222d5","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-05-04 07:41:37.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"f84a523c49649d8e9d2dfa3fda5392c1adfbebb0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 07:42:10.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"db5f542a4d504cf2def0b563b1cf16554a14be5f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 09:48:46.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31bc4f4b4c0d491d8d1d3722e0fc9437\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/f58e16c01ca54804ac1eaabfc2307497 : SUCCESS in 1h 18m 59s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/7eadcf5da54d46fb975f2d58ddc8257f : SUCCESS in 24m 38s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/5bd599e07ca7480f9c450eff0210e987 : SUCCESS in 1h 31m 14s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/3dbaf56d1d4a42488da76c3694ba92e7 : SUCCESS in 48m 12s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4def2d86a549444b9a3af3753abfd474 : SUCCESS in 1h 23m 42s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/030efeef349143fa9da0359ac0fe78a5 : SUCCESS in 29m 05s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/24128d2d90a04113b270f6ecf82c8541 : SUCCESS in 2m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ddb8755bd064ddbbbb502292440a066 : SUCCESS in 3m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/35a43cf77cd749f284b4b5e185e1e30c : SUCCESS in 2m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/de495bd49d29427b9a38c3ef909c4106 : SUCCESS in 3m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db4e3dff4c44412ca1a01c46d87bd825 : SUCCESS in 6m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6f31481db406496286d7a76535349909 : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":4},{"id":"e47d90edd81e67b0127749f415e3ab40bee94f0c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 09:48:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"a53ac74e9f2c9df407d389068a466cb9c4767899","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 09:49:42.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f7dabc8e35647f7ae63a1f6759fbe38\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5122fff9675b4bf58b024f9d766370d6 : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/38a00d53fb914b3cacb99f2c30cf6c96 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"655991d734af3d44c44e1fb4e47b7a850807258f","revisions":{"3cf7656f699bc0ef3855b40e4fe22082cfce7ac0":{"kind":"REWORK","_number":1,"created":"2026-04-27 17:08:24.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/30/986430/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/30/986430/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/986430/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/986430/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/986430/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/30/986430/1"}}},"commit":{"parents":[{"commit":"afc09fc357d936f7c5c7b9ee17255623924352a2","subject":"Merge \"Update Prometheus to version 3.5.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/afc09fc357d936f7c5c7b9ee17255623924352a2"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-04-27 17:04:17.000000000","tz":120},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-04-27 17:08:22.000000000","tz":120},"subject":"Update Prometheus to version 3.5.3","message":"Update Prometheus to version 3.5.3\n\nThis release fixes mutiple security issues.\n\nChange-Id: I9f4243a40d0907639324308d84d8a860c1e54941\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3cf7656f699bc0ef3855b40e4fe22082cfce7ac0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3cf7656f699bc0ef3855b40e4fe22082cfce7ac0"}]},"branch":"refs/heads/master"},"3516e71770f93b5cb81fd384dd9964c1344cd5c7":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-04-28 06:17:11.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/30/986430/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/30/986430/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/986430/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/986430/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/986430/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/30/986430/2"}}},"commit":{"parents":[{"commit":"afc09fc357d936f7c5c7b9ee17255623924352a2","subject":"Merge \"Update Prometheus to version 3.5.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/afc09fc357d936f7c5c7b9ee17255623924352a2"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-04-27 17:04:17.000000000","tz":120},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-04-28 06:17:08.000000000","tz":120},"subject":"Update Prometheus to version 3.5.3","message":"Update Prometheus to version 3.5.3\n\nThis release fixes mutiple security issues [1].\n\n[1] https://github.com/prometheus/prometheus/releases/tag/v3.11.3\n\nChange-Id: I9f4243a40d0907639324308d84d8a860c1e54941\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3516e71770f93b5cb81fd384dd9964c1344cd5c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3516e71770f93b5cb81fd384dd9964c1344cd5c7"}]},"branch":"refs/heads/master"},"732fe433f7cad022466b3128566bac2130af899b":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-04-28 08:48:13.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/30/986430/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/30/986430/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/986430/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/986430/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/986430/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/30/986430/3"}}},"commit":{"parents":[{"commit":"c61312fd9ce82f11e48890caaea1761c2b0a9191","subject":"Remove upper-constraints from Kolla linters.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c61312fd9ce82f11e48890caaea1761c2b0a9191"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-04-27 17:04:17.000000000","tz":120},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-04-28 08:48:10.000000000","tz":120},"subject":"Update Prometheus to version 3.5.3","message":"Update Prometheus to version 3.5.3\n\nThis release fixes mutiple security issues [1].\n\n[1] https://github.com/prometheus/prometheus/releases/tag/v3.11.3\n\nChange-Id: I9f4243a40d0907639324308d84d8a860c1e54941\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/732fe433f7cad022466b3128566bac2130af899b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/732fe433f7cad022466b3128566bac2130af899b"}]},"branch":"refs/heads/master"},"655991d734af3d44c44e1fb4e47b7a850807258f":{"kind":"REWORK","_number":4,"created":"2026-04-29 06:44:17.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/30/986430/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/30/986430/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/986430/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/986430/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/986430/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/30/986430/4"}}},"commit":{"parents":[{"commit":"c61312fd9ce82f11e48890caaea1761c2b0a9191","subject":"Remove upper-constraints from Kolla linters.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c61312fd9ce82f11e48890caaea1761c2b0a9191"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-04-27 17:04:17.000000000","tz":120},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-04-29 06:44:14.000000000","tz":120},"subject":"Update Prometheus to version 3.5.3","message":"Update Prometheus to version 3.5.3\n\nThis release fixes multiple security issues [1].\n\n[1] https://github.com/prometheus/prometheus/releases/tag/v3.5.3\n\nChange-Id: I9f4243a40d0907639324308d84d8a860c1e54941\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/655991d734af3d44c44e1fb4e47b7a850807258f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/655991d734af3d44c44e1fb4e47b7a850807258f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d1"],"failing_atoms":["label:Backport-Candidate\u003d-1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
