)]}'
{"id":"openstack%2Fkolla~986488","triplet_id":"openstack%2Fkolla~master~I39ab300240b0178f3697ff2dcdf1fab2c3b828bf","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2026-04-28 14:28:03.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-04-28 15:55:06.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-04-28 11:16:41.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}}},"hashtags":[],"change_id":"I39ab300240b0178f3697ff2dcdf1fab2c3b828bf","subject":"Drop mod_wsgi","status":"MERGED","created":"2026-04-28 07:23:19.000000000","updated":"2026-04-29 02:27:22.000000000","submitted":"2026-04-28 15:55:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"986488","meta_rev_id":"a7a850bf640db5116a9baebe08948136a0f04ee0","_number":986488,"virtual_id_number":986488,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-28 15:55:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-04-28 14:28:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2026-04-28 11:16:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-04-28 11:20:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2026-04-28 11:36:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":-1,"date":"2026-04-28 14:28:03.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":-1,"date":"2026-04-28 11:36:48.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2026-04-28 11:36:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-28 09:14:54.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-04-28 09:53:44.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2026-04-28 10:26:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-28 11:01:36.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"abe756f4a05e28d2e9d1dde42be36eb8348cf9de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-28 07:23:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"52b143470657b3af8766d67908a910315a4b1ce9","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-28 07:32:53.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"0aeefe56474eecaf83c30da4223973362a8db627","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-28 09:14:54.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ab577d072e0cb52a191f169aa5a9aed674ac7483","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-28 09:53:44.000000000","message":"Patch Set 1: Backport-Candidate-1 Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c286262bc772a1db3de0f4057cd5cc6d10229b31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 10:26:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38107a05a8064eaaaa68b738e2c7672e\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/9271f4a256f24f7e8df66187e93d2827 : SUCCESS in 1h 12m 30s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/00ee41422f554b2999b98fdf0796f078 : SUCCESS in 1h 23m 48s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/b06afee13af84165965a6e352ec5d3c2 : SUCCESS in 24m 52s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/2001316f0c0844868bbd141919e87a55 : SUCCESS in 1h 09m 53s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/17727f18d4774cc6b236b68ea3f1f0e5 : SUCCESS in 48m 44s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b9090d1d7b0c47ce940a9de59ac9a0c3 : SUCCESS in 1h 17m 33s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/759d3c0be24e46938787591c771363f4 : SUCCESS in 15m 43s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/d44a38be2ce84b6d86528128cda23e36 : SUCCESS in 1h 23m 32s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/78ccf25834864bd9816c73215fbcdf71 : SUCCESS in 1h 26m 00s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/8f64092a14344ea59b492561b90aa6f4 : SUCCESS in 1h 45m 26s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/f91b825d367c4ff89cdff7d37451690c : SUCCESS in 1h 38m 36s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/1b5e79b161f64fa0922cba371dea6b06 : SUCCESS in 2h 00m 29s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b7c819eda81740f88808eec633718139 : SUCCESS in 1h 27m 10s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/82939086fe024e178310b5a35cee2ef2 : SUCCESS in 1h 58m 35s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/18a83d0437344ab188319c9b71c6d90a : FAILURE in 1h 49m 59s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/aaf447de52484583909db96b1bc94abe : SUCCESS in 2h 12m 43s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/80c91992ca7a4faebb9d9812e5959354 : SUCCESS in 2h 49m 13s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/426809ba9f5f4dd5a5726297f23fec4f : SUCCESS in 2h 04m 25s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/9d92b40a51d9438c8ec07e75be8fff31 : SUCCESS in 3h 01m 43s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/e31bfc4a5e174705970e8eeffc788130 : SUCCESS in 1h 49m 05s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/4f04d79415f54da0aca3ea2ac6b9cfaa : TIMED_OUT in 2h 32m 48s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/6e8bd919f6b9475a811059907435f407 : SUCCESS in 2h 30m 52s (non-voting)\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/b77fda460b2748f4b3df1d9c258eb842 : FAILURE in 46m 14s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/2481676cbe79471e912e0c3894af9737 : SUCCESS in 1h 44m 43s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/bc63083aef8b404baa40a152cae2c3be : SUCCESS in 1h 49m 27s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/58a4188ce1ce4938a80ade44cb46289e : SUCCESS in 1h 59m 34s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/79c176ab7df446018e36fdaaf7bed8af : SUCCESS in 1h 33m 10s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/f5cb28706bc746f7b618d2c5b356fe6c : FAILURE in 2h 16m 02s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/2c38f4ee2fb3437a8a80b71fcfd65257 : SUCCESS in 2h 23m 13s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/64688dd58cc14a9fa217e909fee95967 : TIMED_OUT in 2h 33m 36s (non-voting)\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/5e338654cc7f424c9596314ecd892247 : SUCCESS in 1h 52m 17s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/e6cb6db907b148e783a4197d7d2ef668 : SUCCESS in 1h 42m 54s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/ff0cdc8aa1af4c8cbde125f0a36e3ddd : SUCCESS in 2h 10m 44s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/3fe4cb69feab403394eabeb3cd3164d5 : SUCCESS in 3m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c410de46346c49d892128454bf02566d : SUCCESS in 3m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/468f5d9267c84047852d0f97f53ee906 : SUCCESS in 4m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0caf6405410540a2a0b71f2b46f8bbfd : SUCCESS in 3m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9364f711f9ab4ee0b60116b407a57a2f : SUCCESS in 5m 57s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/529c22796244492c88b0d355311427c1 : SUCCESS in 3m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b7dea45099c42038d3c8eab35753406 : SUCCESS in 7m 10s","accounts_in_message":[],"_revision_number":1},{"id":"82ec125786f12a026d3a4a4c8ea6a2f81f737c9b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-28 11:01:18.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"93a8aa82522d3e2791105749924c1786cd11b732","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-04-28 11:01:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"394023e64b06635d896ca62d8d83a5c9dd06939f","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-28 11:16:41.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"409b0f8ed1099b3c60c793e05901f7ec18b91a76","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-04-28 11:20:24.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"53699941e509063899230219c59503611cb784ab","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-28 11:36:48.000000000","message":"Patch Set 2: Backport-Candidate-1 Review-Priority+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5b1580687d8f074e1dc85bdf8c4f2095896a8803","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 13:50:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93913c5e07464919b48327843ffba927\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/2beeb207874f408dbd3753bdfd85643d : SUCCESS in 1h 12m 55s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/bc029e3df17548819a4c59b49557b2e7 : SUCCESS in 1h 02m 16s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/1b6ad38b777e45bbafb9b102c506e692 : SUCCESS in 24m 42s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/9d0be270897844228d790bac74692ea7 : SUCCESS in 1h 04m 28s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/792658ad4cb744519aa7612604e944de : SUCCESS in 54m 19s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b8b9f05615db4dcaa3ebaf940b6a273f : SUCCESS in 1h 23m 01s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f70c87b054174f828263fa311f94e536 : SUCCESS in 33m 32s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/cc142fd6328847868cfb566f2eee55ce : SUCCESS in 1h 31m 01s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/8f56ce98d57049f2aba95b450d39d320 : SUCCESS in 1h 31m 10s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/5382003d962240bcb97692bf7eeb1ae1 : SUCCESS in 1h 45m 08s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/98d794a06fd847c1bf13a5bb50c048bf : SUCCESS in 1h 41m 28s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/dc3b571b05984bf783293b19c90de2a3 : SUCCESS in 1h 58m 55s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4b3c16dc1ae44a28918e1db32bd3f2dd : SUCCESS in 1h 23m 24s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/d0f07c03ddf84418b97c53bebe98609b : SUCCESS in 1h 57m 01s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/53794dff821246a78861b8e62d7fb21d : SUCCESS in 2h 00m 46s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/e71e51941a474fd092cba8a7241ac10b : SUCCESS in 2h 16m 16s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/226252f3701643948c2ae12dcc067cd0 : FAILURE in 2h 02m 15s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/161cb54a4f704745a0cfb5ab1d005045 : SUCCESS in 2h 00m 46s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/27fc8899f3d94a0ba460a01d3d71a489 : SUCCESS in 2h 47m 05s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/9ed26e8637b0471582df0736fbd50d5e : SUCCESS in 1h 49m 06s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/bb26021404bc41b687c879db04487225 : SUCCESS in 2h 04m 11s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/d0ed7fcfd68f4fdfa5b59d0d1a9f6963 : SUCCESS in 1h 55m 19s (non-voting)\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/d0699c01e39b49b987ea2fa13357342f : FAILURE in 53m 32s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/44f8a55a5ef349c4b1bd224a678b5e5e : SUCCESS in 1h 02m 48s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/3ad8d754bb514b989f8ee6c5e0c09c64 : SUCCESS in 1h 01m 27s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/6b0d5db99b374e9c9d18aef2877bbbae : SUCCESS in 1h 33m 54s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/1c8f82a955944c21b6f50e79bfa9f6d1 : SUCCESS in 1h 35m 18s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/2faa816b80944a708abd8edc8dc1a503 : FAILURE in 1h 45m 30s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/cbfea6119ec84e459771fb6c2b71ef1d : SUCCESS in 2h 32m 43s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/967fc72c7a904ebe8790820c6a8ed593 : FAILURE in 1h 42m 11s (non-voting)\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/15ca13f9d8884e21a222cbff74822b4e : SUCCESS in 58m 01s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/e6603986c6304d38a479cd9080187164 : SUCCESS in 1h 04m 26s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/c12a75cb8450443daeb85de6dc692d79 : SUCCESS in 2h 08m 36s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/ab1f117554434feaac7a3d3e6676d87a : SUCCESS in 2m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d0ed617ff344a0b9a986beb8cd7a02b : SUCCESS in 3m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9f9c7702d31541c9b5d3cfd9a8caf69b : SUCCESS in 3m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/04936a7e69a0475ba4d2840e3ba8839f : SUCCESS in 2m 12s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ec3763f4c0824d1795d1cbd4af163e74 : SUCCESS in 3m 27s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b11e7dfe6c044fb19dbe4d10847a59b4 : SUCCESS in 2m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b91bc01fff7c4f169782b255982b7dfe : SUCCESS in 6m 04s","accounts_in_message":[],"_revision_number":2},{"id":"e05c530b707a73210fcaf2a54c802327123caca6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 13:54:39.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a012460e0ba83cc1f1da40fd4f8183655f0b0358","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-28 14:28:03.000000000","message":"Patch Set 2: Backport-Candidate-1 Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2735f4f336d04028331514bc87712a8e9c1caf72","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 15:55:06.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5579a3f934224cb2b1898b64e24f86a5\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/46ba3fdef3474b75a688d38397d6508d : SUCCESS in 1h 06m 04s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/90980f8d1f2946368e1b772522b8e705 : SUCCESS in 22m 29s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/443202f7841e4a6c9a0d3c36c7868586 : SUCCESS in 1h 01m 38s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/8abad7a0c463459eb77b1c805eea3b46 : SUCCESS in 34m 12s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9d759db886d84fc5b9f8eb77cf232664 : SUCCESS in 53m 23s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e10b23fbc7b3434da1c94c4e2be17ac9 : SUCCESS in 23m 07s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/b3fffa55a2dd4e2ab382ef2453f72b20 : SUCCESS in 1h 22m 38s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/bf7a8e9cd49e4215b668d73a071fbd9a : SUCCESS in 1h 42m 12s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/bea6c07cb15446e5bbb6ede5a051315e : SUCCESS in 1h 36m 28s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/9626dd1a05434436a0349fdd11837f62 : SUCCESS in 1h 59m 48s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/02669d8d80974b7c8249e69e28f415bc : SUCCESS in 1h 30m 07s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/eec9915f696e47e69f3d3c4dac0144f4 : SUCCESS in 1h 55m 55s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/35f241adff684692b3851d40082486cc : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edb6fcb4e0484b3dbbcc4161fd04b5ca : SUCCESS in 5m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/45cebfaa6c6d4fd7a79bca393c0a246d : SUCCESS in 4m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/60ca381561524268a547344e17970952 : SUCCESS in 3m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a29b643457974acfbdad1208393b8ad3 : SUCCESS in 6m 48s","accounts_in_message":[],"_revision_number":2},{"id":"43d153690273d93a281842b78c4b4cfcdbc47f57","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 15:55:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"35e2cfe9cfd2c77833f393b98e0f9a7ada897c42","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 15:56:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2b0dfe2bbac41eaaaa772b5c7a92ad3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26231548f44149dfa0f6185c45c5be5e : SUCCESS in 46s","accounts_in_message":[],"_revision_number":2},{"id":"a7a850bf640db5116a9baebe08948136a0f04ee0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 02:27:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db5e739633da43a39964fac5e5bb21dd\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/8aaf7814d5024fed9147d8cb4f5413db : TIMED_OUT in 3h 08m 59s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/f604d8357cb84af7be9071d9a1ce3b09 : SUCCESS in 2h 16m 11s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/42c69f1c25bf461293d65419623ff897 : SUCCESS in 2h 13m 41s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/1455006bbc47423690a4301fdb385bdd : SUCCESS in 1h 33m 51s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/3fd0711304ce4630bb7728c2dacbc817 : TIMED_OUT in 2h 31m 52s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/addfb94342124889931454883670934b : TIMED_OUT in 3h 02m 13s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/88dd23f0323c4910a87fdcfbb49df1f3 : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8dda3099fca946fe837689217f715411 : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/6a56b2d4e84645fe8757ba07144f5e83 : SUCCESS in 8m 25s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ae6b0eb1e6d96d294a89d412a3917f27d1758af4","revisions":{"958baccf5b3abebee64c4b208d7727a5f3f72cf2":{"kind":"REWORK","_number":1,"created":"2026-04-28 07:23:19.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/88/986488/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/88/986488/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/986488/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/986488/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/986488/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/88/986488/1"}}},"commit":{"parents":[{"commit":"c61312fd9ce82f11e48890caaea1761c2b0a9191","subject":"Remove upper-constraints from Kolla linters.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c61312fd9ce82f11e48890caaea1761c2b0a9191"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-28 07:22:51.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-28 07:23:17.000000000","tz":120},"subject":"Drop mod_wsgi","message":"Drop mod_wsgi\n\nIt\u0027s not needed since [1].\n\n[1]: https://review.opendev.org/c/openstack/kolla-ansible/+/982414\n\nChange-Id: I39ab300240b0178f3697ff2dcdf1fab2c3b828bf\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/958baccf5b3abebee64c4b208d7727a5f3f72cf2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/958baccf5b3abebee64c4b208d7727a5f3f72cf2"}]},"branch":"refs/heads/master"},"ae6b0eb1e6d96d294a89d412a3917f27d1758af4":{"kind":"REWORK","_number":2,"created":"2026-04-28 11:01:18.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/88/986488/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/88/986488/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/986488/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/986488/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/986488/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/88/986488/2"}}},"commit":{"parents":[{"commit":"c61312fd9ce82f11e48890caaea1761c2b0a9191","subject":"Remove upper-constraints from Kolla linters.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c61312fd9ce82f11e48890caaea1761c2b0a9191"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-28 07:22:51.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-28 11:01:17.000000000","tz":0},"subject":"Drop mod_wsgi","message":"Drop mod_wsgi\n\nIt\u0027s not needed since [1].\n\n[1]: https://review.opendev.org/c/openstack/kolla-ansible/+/982414\n\nChange-Id: I39ab300240b0178f3697ff2dcdf1fab2c3b828bf\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ae6b0eb1e6d96d294a89d412a3917f27d1758af4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ae6b0eb1e6d96d294a89d412a3917f27d1758af4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-1"],"failing_atoms":["label:Backport-Candidate\u003d1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
