)]}'
{"id":"openstack%2Fkolla~986638","triplet_id":"openstack%2Fkolla~master~I21fc99f42b0fbfa45ff2d9366ea603866e6682d4","project":"openstack/kolla","branch":"master","topic":"no-babel","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-04-29 22:25:19.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-05-04 15:06:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I21fc99f42b0fbfa45ff2d9366ea603866e6682d4","subject":"Remove Babel from openstack-base","status":"MERGED","created":"2026-04-29 12:28:06.000000000","updated":"2026-05-04 15:07:35.000000000","submitted":"2026-05-04 15:06:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"986638-no-babel","meta_rev_id":"2bf5dcf4cdfb2206ed3acdca32465c0fb378eedb","_number":986638,"virtual_id_number":986638,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-04 15:06:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"all":[{"value":2,"date":"2026-05-04 10:05:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},{"value":1,"date":"2026-05-04 09:57:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-05-04 11:22:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-05-04 11:22:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2026-05-04 11:22:36.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-29 15:18:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-29 18:54:03.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-05-04 10:05:50.000000000","updated_by":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"reviewer":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"state":"REVIEWER"},{"updated":"2026-05-04 11:22:36.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"92088ee50d82d6ac109af094f74874ea2695ca51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-04-29 12:28:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"24af881232ba2710f7e8a3fa30f7c2a98279e8c9","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-04-29 12:48:48.000000000","message":"Topic set to no-babel","accounts_in_message":[],"_revision_number":1},{"id":"53be9e15223dd23a5dafe883c001fff66c26c76b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 15:18:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dc0ba4f1fd4647b797d50311de6399d6\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/16e261408e204c9a817f693cc27eb863 : SUCCESS in 1h 06m 35s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/6379808c963a4d3ea0c615cf4cde5a4d : SUCCESS in 1h 26m 03s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/1433ca3bba4743189d7ccb5c5e0d4f0e : FAILURE in 47m 35s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/1a927540b4c44f969906437d5413194e : SUCCESS in 1h 21m 59s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/71a9950ce817489685f7ae8bfe446d10 : SUCCESS in 45m 28s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bb2209c102124476adddfe25758a3379 : SUCCESS in 1h 31m 48s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/712213d85c5c4a3b803a9a4c6f8984a4 : SUCCESS in 30m 38s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/e2be649b5b3e489982906d235dde538f : SUCCESS in 1h 21m 18s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/e55dfe7309584adda4200bcaeec9023f : SUCCESS in 1h 23m 22s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/b8cb32a134ba418484a74b924355ff7d : SUCCESS in 1h 45m 31s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/45f0da38f4de491f87a4640a2181cb4f : SUCCESS in 1h 38m 45s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/6184eef391484daca780ac6b31625b7d : SUCCESS in 2h 07m 54s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d03ecf77c7654837b5d5e8bde3a87609 : SUCCESS in 1h 32m 22s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/e29dcd936762483685ffae5278bfbfad : SUCCESS in 2h 02m 55s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/88d9873bb58640d5b45f605e1761f580 : FAILURE in 2h 28m 24s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/6636d2889587469f8e5bdc077a1c068b : SUCCESS in 2h 08m 45s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/48d506040d8148e993e68488d01430ab : SUCCESS in 2h 48m 53s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/7634dd43febe41e8844610a766a26373 : SUCCESS in 2h 02m 17s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/e69dfe91c7634e1592ed000372d1709d : SUCCESS in 2h 48m 52s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/9dd71661e08749fdadeae6dcc04b06fe : SUCCESS in 1h 49m 33s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/10c10fa8c14c4c9d99dada3d76eb0a0c : SUCCESS in 2h 02m 11s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/7aaa066ead3d4d289121bef8a8cfb573 : SUCCESS in 1h 58m 07s (non-voting)\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/d1d3286c95da4ebfb333181db414d0e1 : FAILURE in 57m 24s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/9b6119e71614484b854d81198a024d3c : SUCCESS in 55m 56s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/db3bf271e64449ddb2adbf793d2b1227 : SUCCESS in 1h 37m 10s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/f23026be6ab44072a1b992826b3d7e01 : SUCCESS in 1h 34m 52s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/f961dbfaee1841a69b403da9d9101f28 : SUCCESS in 1h 34m 54s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/9071e60756a44788a6d4975d58a2ebc3 : FAILURE in 1h 43m 36s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/2b90d5239f9b468bbd5a434959cd43fa : SUCCESS in 2h 24m 03s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/50aff5057f284237b9a2260a5b934e6d : FAILURE in 1h 48m 59s (non-voting)\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/008840e874304eecba00a77d80307965 : TIMED_OUT in 2h 34m 36s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/9d13b75c5d6d4217aabb3e9c558c42d4 : SUCCESS in 2h 13m 01s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/799c26f9671f47399d2ccc3d64af3e3c : SUCCESS in 1h 47m 31s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/a7a74cde5dcf40cfbd0cdaabbc869752 : SUCCESS in 3m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a2477dd8a4f491c959ccfca414940ac : SUCCESS in 2m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/13e8d95d07e145819963896270e8b8de : SUCCESS in 3m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/226ca4fd488149ae8da9c8da9aea3692 : SUCCESS in 6m 05s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b9f4ca10f7ca4af7b59391bf0f10cb56 : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/72f3ec060f454000b0ee35dbed386dd4 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77c6bfc9aabe40a88db9e42474f3e51a : SUCCESS in 6m 38s","accounts_in_message":[],"_revision_number":1},{"id":"c7d79c32660217daf04c43da3d6252ac947ae975","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-04-29 18:34:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b0c0d1ea61842b95d10739a8f319e85b54554fd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 21:23:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c54c1e42dea649cb8335c485d726a0f6\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/eeb8ce005da54383ac6c35d0858557d9 : SUCCESS in 1h 08m 56s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/9d9798c75c404d11b7e6003d84eba9e4 : SUCCESS in 1h 02m 20s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/22ec3d22f04e4fd2ba200a05e6da7dd6 : SUCCESS in 12m 01s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/8d4a5239aa464aa19b3591010d24744b : SUCCESS in 1h 00m 11s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/d66bf74fc4334da1848245cd574b6fc4 : SUCCESS in 1h 06m 09s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/49e41b48d00e4f01abb28926c1ca8ab8 : SUCCESS in 52m 18s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7e2015eac0f54b55bb8f54ef7263b35c : SUCCESS in 29m 19s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/ad0ac27de3954c9b8e47151f6a2d037a : SUCCESS in 1h 22m 15s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/32aa61ce70fe4aee9946fcfdf306b035 : SUCCESS in 1h 32m 02s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/a7e4f5ff079a4995b0b34f08d90de0bf : SUCCESS in 1h 38m 35s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/433349fd3f614b09a451a4cd7f995ee0 : SUCCESS in 1h 31m 09s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/4aac30e65410430ebf528e1b9448cff4 : SUCCESS in 2h 03m 34s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6925a3aa7bd9420098233c8c58e11bb3 : SUCCESS in 1h 25m 42s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/0cbcf31c3d76435c8e52f398d020104e : SUCCESS in 1h 57m 34s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/c4ebc5e3141944fbb5e3c2a57a4adf5c : SUCCESS in 1h 52m 56s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/8e4adfce066f49a09f23afcc7c384640 : SUCCESS in 2h 09m 46s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/ffb32ea48f1044debd90e1cc3f628820 : SUCCESS in 2h 47m 41s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/92254823100c4ab9a96643034644ae5e : SUCCESS in 1h 58m 47s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/8bc942a6c763448c93146da3bc686405 : SUCCESS in 2h 43m 02s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/bfbdac7f96d64fe5b01c3ef305073f0a : SUCCESS in 1h 56m 05s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/1367b407f3c44af8a44631e179939f22 : SUCCESS in 1h 57m 33s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/41775d83e1a84bb491a36364a5ca3734 : SUCCESS in 1h 55m 44s (non-voting)\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/f2cb9b32b4294bb3b00d9c29621e793c : FAILURE in 38m 31s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/35bd23e211ed412490cc902a87cc665a : SUCCESS in 1h 42m 37s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/e055891443054655ade485e5a58e6607 : SUCCESS in 1h 36m 44s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/746ce3db30e4462089014052523b9960 : SUCCESS in 1h 30m 47s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/4a4684a804ae4d62adee55b9f786709c : SUCCESS in 1h 30m 42s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/cd53b23e2fe44031bf9c04955acc9cbe : FAILURE in 1h 39m 07s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/36ac4b8e86904670ae2e8ba67aa3f4bc : SUCCESS in 2h 16m 34s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/81537ac2d69f44a99979eb04c119840a : SUCCESS in 2h 22m 10s (non-voting)\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/60ca9a8fd9ef4cfb99eb4ebe1f6721f2 : FAILURE in 1h 37m 31s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/5ac295df350e446c8939c3b1747e5ebf : SUCCESS in 2h 06m 40s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/6ff59f4dfdfa4c5eb5e9525daa3c3f8d : SUCCESS in 1h 52m 18s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/3cfa750cd12e4d44a53e0a17fa181ab6 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/263960fd585e44e19ac4a9d353ef4f27 : SUCCESS in 3m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f3c74f5be57b4227854809834e1b74a0 : SUCCESS in 3m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d8bb3cd7963849ccb922a18efa630f31 : SUCCESS in 2m 09s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5d1d15530eac4f92a673c996480d9954 : SUCCESS in 5m 46s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e6f71421bcb4f3b8b776fc1faa4aba9 : SUCCESS in 2m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84a840f1c1534bffbaa15abe283386a5 : SUCCESS in 4m 10s","accounts_in_message":[],"_revision_number":1},{"id":"048015830b91a41bb37f54760500c969e26a6027","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-29 22:25:19.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"4037888e8af7edaf5c09a3384f149715185ad46c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 10:01:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f06cdca2e424f21bfd7bf138e1ce44e\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/90cf8c67c6d24ce9ac00a5bf4f0dc8df : TIMED_OUT in 3h 06m 43s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/59634e4504794a6e9f0c9c3f82047037 : SUCCESS in 1h 08m 02s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/17d30bf977a04cc5a666d25b2b5e6688 : SUCCESS in 1h 16m 17s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/4a8328bb5ff4456d9d31e6a2d47a2955 : SUCCESS in 1h 25m 54s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/9397bf52e4144117bdb6854ddf4f327b : TIMED_OUT in 2h 31m 58s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/f4038c8e780e456b81db044ade4325d2 : SUCCESS in 2h 31m 44s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/95a41bb0d9b740e6b71026a73eef5b70 : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4377163dfdca4c72876b68a7d0c15fda : SUCCESS in 4m 27s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/c6eaf01107064ec88c666d991bdd308f : SUCCESS in 5m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7afbdd6fee973bd93feeef8323f966882cfabf95","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-05-04 09:57:14.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"2fdc0b2072a24d776e9c2ccf196dfb4f7cef20d9","author":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"date":"2026-05-04 10:05:50.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"62438457b93b9662966764003c4ca9aa94fff3f1","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-05-04 11:22:36.000000000","message":"Patch Set 2: Backport-Candidate-1 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d713201a4f07ffe5a8cd9498fa9324f4d9f86ac3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 12:51:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f235d9403ea4c1da1c7f98e6ee4bbf7\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/554c63b5062a4524abc55ea92fec08c5 : SUCCESS in 1h 03m 30s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/4718d876e5db4c17abbd80ac4124e655 : SUCCESS in 1h 20m 39s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/e3eedc30336c4027b65404498e6f9cd3 : SUCCESS in 28m 52s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/fe78138bfe874ea0b238b1c1fddc02eb : SUCCESS in 1h 08m 48s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/dd6acecb256c4f64a9496af0bb55465d : SUCCESS in 46m 36s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d94dded9f2c14dddab6f2e1b8e189916 : SUCCESS in 1h 22m 15s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e6447d117113446e90d44d930569b0a0 : SUCCESS in 27m 08s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/c67fdad34a644be685c8e9f1da5b6061 : SUCCESS in 1h 21m 59s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/45d034fdc7214f49813543d25b6fecff : SUCCESS in 1h 23m 34s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/3f2668c25e824ad086d69f3db5b51935 : SUCCESS in 1h 45m 31s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/9456e61b858944909e65bbc979f71349 : SUCCESS in 1h 33m 16s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/f7f1da58116545208f91ee0f87647081 : SUCCESS in 1h 59m 35s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3ed7834c2bbd404a88f7f47a8a09b273 : SUCCESS in 1h 25m 54s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/6eedfe524d81496cbd80b0254dccd791 : SUCCESS in 1h 58m 07s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/d213c33a24b34f60ada5760edbd904ff : SUCCESS in 2h 13m 52s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/1be61187d1074371bb8b14d56e034f27 : SUCCESS in 2h 09m 09s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/a9dad57dbe9549f0b52344dae2a7fcb6 : SUCCESS in 2h 52m 37s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/062dee87c35043109ed77aeb8041305a : SUCCESS in 1h 59m 20s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/eed0b9a9186443ff8c996385d1ea0d37 : SUCCESS in 2h 44m 02s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/e67712c44390431a9dd464e2a07f873e : SUCCESS in 1h 47m 44s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/a1309f9ea33548adb533682725ae1287 : SUCCESS in 2h 06m 06s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/5f5b03dd6a454254b15912a63d1aed56 : SUCCESS in 2h 02m 34s (non-voting)\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/7a91e25f69954d54845c97668300b69f : FAILURE in 1h 07m 10s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/f24ba5cce51947a894c27cb869cd8325 : SUCCESS in 1h 28m 24s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/8213f6af0aa142309327c9e6186b91f8 : SUCCESS in 1h 44m 02s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/347f954f35ec4cb3969067fd40f833c6 : SUCCESS in 1h 47m 06s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/5fd4e029264f45119acccb288a126aac : SUCCESS in 1h 29m 43s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/1ef0bdd060b44d0884c1dd3be245fc81 : FAILURE in 1h 40m 57s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/cd1224ebf22343f593932318576c997a : SUCCESS in 2h 33m 22s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/f179fb3d1e0b43baa169b0d8376ddc1e : FAILURE in 1h 34m 50s (non-voting)\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/987c1cc887f34da5b1c1383b6b0f28eb : SUCCESS in 1h 27m 41s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/54710950800b4f139d2a09df27a2de20 : SUCCESS in 1h 33m 27s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/0cbd9dcbd9a747b78dbccc05703378a1 : SUCCESS in 1h 57m 20s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/2fdbc3411a714025a9f133df04b830dc : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97b38124036c468fa49d10b512518801 : SUCCESS in 3m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0dcc106a025a4add8f92380bad86ea09 : SUCCESS in 3m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/28fc52e8be2a46289c7edd93d77f5fe2 : SUCCESS in 3m 31s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fc9fe5229bbd44fa905f7310268f0068 : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ae55ae128ff74ff3baf71e9aaec998d0 : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7234866b5ed7482f809ff5676b082c72 : SUCCESS in 5m 28s","accounts_in_message":[],"_revision_number":2},{"id":"bd3e53f0236472ea3cb6066df9e45356b3665116","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 12:52:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"791245b27022ef9048d55c47bde7a28417975670","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 14:39:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82e6499c51ac4aed8847bb494871784e\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/c54e9449013640d781f705a30dd2af67 : SUCCESS in 1h 14m 41s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/da0f95c16efd43039a29a25f592cbc28 : SUCCESS in 1h 24m 28s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/60550f9b88f14c7fba0a68709213ee89 : SUCCESS in 1h 18m 45s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/ec13b289bfe2488c97d2c693f8c78cec : SUCCESS in 1h 21m 33s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/80e78a73ef1443369b0c0ef69c3619d3 : FAILURE in 1h 49m 56s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/0e34b49183b34c87a3c509c629ea1653 : SUCCESS in 1h 37m 24s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9cf723057d354ac3b8416ef7c719b7f0 : SUCCESS in 4m 01s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6f02c5ec16894a44b1191a0ece351b61 : SUCCESS in 4m 27s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/843092babc6e47eb807e3b84c903a832 : SUCCESS in 7m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2794986f466c7b196036f71089df6eb2947d6098","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 15:06:36.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/601bbb1f69b14e509096f58d45c42e39\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/5d789c4f0ea84f9b9f88fcec16a04d73 : SUCCESS in 1h 23m 34s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/e979248c3f4b4bfdb0187914bb449db5 : SUCCESS in 26m 06s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/fe0cdc39f9f641f48b56b8daa3f19014 : SUCCESS in 1h 13m 32s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/042ae52067b64cce934db0b1f901e8f1 : SUCCESS in 1h 02m 47s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/86ea975d627849b3a5a94d37e6b3997b : SUCCESS in 1h 06m 42s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/df0f2e4ac516477ead20984d631167a2 : SUCCESS in 30m 16s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/6cc3926c176d4c50909043157690a5d7 : SUCCESS in 1h 31m 13s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/ab2ed175b32a41648e16ac63e135e676 : SUCCESS in 1h 55m 52s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/10daeca312ae4cba971bfeb35572cda0 : SUCCESS in 1h 41m 48s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/90969107eec94062b7b29e408ca96b17 : SUCCESS in 2h 01m 47s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ff5e13756c8247b490e45c4e4f8e7a73 : SUCCESS in 1h 37m 23s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/77051c50da974fd18f6f1d1a0dd0f38f : SUCCESS in 2h 13m 38s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/7a78e10e1dfe43c0bb3a1850773fdeae : SUCCESS in 3m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b957062928bc45efae623121f96987ed : SUCCESS in 3m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7e472581ec704a34b4897e4b4d9718f9 : SUCCESS in 3m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/eec7b10b46604c9689482232cdcebfde : SUCCESS in 3m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1074d2ed572a4f1fa6aef89e0266dc37 : SUCCESS in 6m 28s","accounts_in_message":[],"_revision_number":2},{"id":"c981641e69fdacce198832fbc9416069aed0dc85","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 15:06:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"2bf5dcf4cdfb2206ed3acdca32465c0fb378eedb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 15:07:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/333127dd92ca4aea9ed62f13d2b26ec9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4624cd4817a74e54af8f3ce4079fceec : SUCCESS in 45s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"45ef3acc79d3f6603d648b094732369c41954dfc","revisions":{"a8d89cbcb4227f50cd91499e4dfaf5a07ae878fe":{"kind":"REWORK","_number":1,"created":"2026-04-29 12:28:06.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/38/986638/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/986638/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/986638/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/986638/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/986638/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/986638/1"}}},"commit":{"parents":[{"commit":"78146c684d0f35f51000b178817c3eb7f4c12329","subject":"keystone: Remove wsgi file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/78146c684d0f35f51000b178817c3eb7f4c12329"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-04-29 12:27:47.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-04-29 12:27:47.000000000","tz":120},"subject":"Remove Babel from openstack-base","message":"Remove Babel from openstack-base\n\nThe Babel dependency was removed from most OpenStack projects years ago.\n\nChange-Id: I21fc99f42b0fbfa45ff2d9366ea603866e6682d4\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a8d89cbcb4227f50cd91499e4dfaf5a07ae878fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a8d89cbcb4227f50cd91499e4dfaf5a07ae878fe"}]},"branch":"refs/heads/master"},"45ef3acc79d3f6603d648b094732369c41954dfc":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-05-04 09:57:14.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/38/986638/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/986638/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/986638/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/986638/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/986638/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/986638/2"}}},"commit":{"parents":[{"commit":"fe4910f14f2134877af671a41fd499f7ef52e8c5","subject":"Merge \"Update Prometheus to version 3.5.3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fe4910f14f2134877af671a41fd499f7ef52e8c5"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-04-29 12:27:47.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-05-04 09:57:02.000000000","tz":120},"subject":"Remove Babel from openstack-base","message":"Remove Babel from openstack-base\n\nThe Babel dependency was removed from most OpenStack projects years ago.\n\nChange-Id: I21fc99f42b0fbfa45ff2d9366ea603866e6682d4\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/45ef3acc79d3f6603d648b094732369c41954dfc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/45ef3acc79d3f6603d648b094732369c41954dfc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-1"],"failing_atoms":["label:Backport-Candidate\u003d1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
