)]}'
{"id":"openstack%2Fkolla~987109","triplet_id":"openstack%2Fkolla~master~I07bcebfdc62ae16a87221ee1e66c2f38e36fc0c2","project":"openstack/kolla","branch":"master","topic":"ironic-vnc","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-07-31 07:11:39.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"4571":{"account":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"last_update":"2026-09-22 20:16:27.000000000","reason":"\u003cGERRIT_ACCOUNT_4571\u003e replied on the change","reason_account":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"}}},"hashtags":[],"change_id":"I07bcebfdc62ae16a87221ee1e66c2f38e36fc0c2","subject":"Add ironic-vnc-container container image","status":"NEW","created":"2026-05-03 22:43:33.000000000","updated":"2026-09-22 20:16:27.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":10,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"e2ac21ecd29a54820b1b168ca7b7b73b446c6dc4","_number":987109,"virtual_id_number":987109,"owner":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-05-05 03:17:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":-1,"date":"2026-07-31 07:11:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"CC":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},{"_account_id":39457,"name":"Murray Groves","display_name":"Murray Groves","email":"murray@stackhpc.com","username":"MurrayGroves"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-04 01:30:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-05-04 01:56:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-04 05:04:12.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2026-05-04 07:01:39.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"CC"},{"updated":"2026-07-24 13:05:54.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"CC"},{"updated":"2026-07-31 07:11:39.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2026-07-31 15:31:07.000000000","updated_by":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"reviewer":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"state":"CC"},{"updated":"2026-09-22 12:49:32.000000000","updated_by":{"_account_id":39457,"name":"Murray Groves","display_name":"Murray Groves","email":"murray@stackhpc.com","username":"MurrayGroves"},"reviewer":{"_account_id":39457,"name":"Murray Groves","display_name":"Murray Groves","email":"murray@stackhpc.com","username":"MurrayGroves"},"state":"CC"},{"updated":"2026-09-22 13:13:25.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"}],"messages":[{"id":"311cbf981da293f9550c1d88cc3f0a285dd52fd5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2026-05-03 22:43:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c8d3660d9190b7e58cdea50c1b0cfeb7f5f6b38b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2026-05-03 23:29:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"927ec72a722b632e6fe214c47cf2668517666742","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 01:30:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9ad77e46124485dbceabc7cf8573cd7\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/ea74cea2113847a6aade26b0682693e5 : FAILURE in 1h 07m 02s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/9a799c59a76f4c54950c7b1dba0f48f0 : FAILURE in 1h 07m 37s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/8d517809445c4f0fa95353e4e58073be : FAILURE in 1h 59m 00s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/f804f623d04b423bb12f7e59044672cd : FAILURE in 1h 06m 26s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/72eae1762fc4430ab6059ac687d5726a : FAILURE in 1h 18m 22s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/c60726104e14402486397afef72d26bb : FAILURE in 36m 04s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/dadf88596cae4754ae0bb32050717bed : SUCCESS in 3m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0b4638af03b242ecbd45dd07cecb9506 : SUCCESS in 3m 34s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/bd6f82dbc4f648729ffa676ea824bd84 : SUCCESS in 6m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9b32dcdf33000ba356119643ad57075787ad91a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2026-05-04 01:32:24.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c9e8a8659cf8f4e89b02a6e01c94fc21b5c1abbf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 01:43:17.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6291a6451b54421b90c79b9939cd896\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/22a157061f0f4068b537d8de78c0e365 : FAILURE in 4m 57s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/3c7976f32b2b4b2b9d2f93c5c19ad817 : FAILURE in 5m 09s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/ca721287212a41fda217f31de8964d5e : FAILURE in 5m 53s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/38a1c045f40742d785f900f0ec4228bd : FAILURE in 3m 57s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/8f33ff6d15d14af284a728f42e8ace24 : FAILURE in 9m 31s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/1284ae3337ca4732befea4866659a4d6 : FAILURE in 8m 35s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/70dd76a651d7483f9e573b53a324ead4 : SUCCESS in 3m 28s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f06c30efc171447596856bb6c5e8cc8e : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/99866bc722e940ba888cf7d1cf612185 : SUCCESS in 6m 31s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"69a5a65f484ad574a4fedd5d94fed23e74640eca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 01:56:41.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7892a17d2a084b5c98ef2c601831c267\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/6dd15bbf523a45938cedc84e0ab330f9 : FAILURE in 4m 32s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/d95c265194a94deea764ea1ee11a481d : FAILURE in 3m 36s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/925571724adb4ca5a16c1e870bfce944 : FAILURE in 3m 15s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/04a8885ec72f4fb3a1e9fdb5fa6487fa : FAILURE in 3m 32s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/60adfe3ab44a474eb9ca3e7dc20481eb : FAILURE in 3m 16s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c545902a60f543ea9b00dee3d0acc4ed : FAILURE in 5m 11s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/47392d1bba734b6bac72358643916cf6 : FAILURE in 2m 33s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/e779acb543f441fb83d6ae97eedb4a45 : FAILURE in 6m 11s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/030b287185c9498ab2d41e8906c927ac : FAILURE in 7m 42s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/da97a506e9404c9b80a97e9519f68fcf : FAILURE in 6m 35s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/f90d8154a658434aadc9432415511706 : FAILURE in 5m 51s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/7cda5051e9f94bc1a5bc14c831c202f4 : FAILURE in 8m 34s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9d12b6272b3e480e9ba9868c202f2251 : FAILURE in 6m 42s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/a7f606cf016d4fe9a177a1063f3e9b07 : FAILURE in 6m 08s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/39090985ef1d42479f28d8979522a99a : FAILURE in 9m 25s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/de2b07e8d85d4d5c9d975e9c12a409f4 : FAILURE in 9m 20s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/c69aa9c4a8444af39f6231b435476519 : FAILURE in 9m 32s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/8af6a34a0ae244d896e0e89b7a4432b7 : FAILURE in 8m 38s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/98b80e385554457ab18844bca1ba01fe : FAILURE in 7m 08s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/4880aca349cc495aaec241464304c789 : FAILURE in 7m 56s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/b8a9bd3390a74e2d836e7d65bbddac4c : FAILURE in 7m 36s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/30cc3dc8a1f648d1b7ffc423067f87bb : FAILURE in 8m 15s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/50cd0214f59f42f98ffeb6c220094e0a : FAILURE in 7m 16s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/ca9337250def4ebeb2ea5497c5c7eb22 : FAILURE in 9m 09s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/ab9d1fbed7544d72886d1c09f4538199 : FAILURE in 8m 16s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/dfa866d6da3f4bca83e243e607309d5f : FAILURE in 8m 20s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/81627e85d5c14927a2c96bcdaf987239 : FAILURE in 12m 46s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/307c8936dfa346d980b03f7d2a6a271d : FAILURE in 7m 55s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/8d27312305c146d3bb2da2ebbc7f1fd6 : FAILURE in 6m 47s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/e20c4b4e5c694a44a97b36f6185eedce : FAILURE in 6m 58s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/2a930a734ab343d3b3f4e6583764a816 : FAILURE in 8m 15s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/a8677a80bda7496084422da51e5cd5ae : FAILURE in 8m 51s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/2b25e2dc6b444a36ac642357b36173a8 : FAILURE in 8m 04s (non-voting)\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/e4c3d07fc3e54202b0f93f39863415f0 : FAILURE in 6m 54s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/a6dfb77894564142a6259013df4ed684 : FAILURE in 5m 10s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/7235469feea64a6583595dac984b5d58 : FAILURE in 8m 39s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/d0b703bb0ea0456b89d0b7d38f8e4a8d : SUCCESS in 2m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44463279243b4542a0b002eab5ced2f8 : FAILURE in 3m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/be9040bdea2a49fe86d1a89bf669b40c : SUCCESS in 3m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c2d15c0585854a598162a5f71bcbbc1f : SUCCESS in 3m 27s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a5c00e8ed22b44cd924b7d16dcbd0685 : SUCCESS in 5m 46s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a53c437db4644cf8b519904686e3f6e2 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08d4e4c989c546c0b1872b3dea7e85aa : SUCCESS in 5m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c3e03a170aa54af59a07f29747133a6e : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":3},{"id":"7c242d8db86275c2d1f623d2ca7dbe91969194c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2026-05-04 03:55:51.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"7b4a7ce93c7dce7924522219a2b7869853aa861a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-04 05:04:12.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"adfffca7215d13d1587c5b733f8561cb8db750f4","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2026-05-04 06:44:06.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2a594b17536ebc4512ec646bb74379a67ef71e7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 06:48:47.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3a24f0834a124e279b8d8ebe3f8ea255\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/0b8c867b180448348b385d3e693f3b69 : FAILURE in 1h 02m 34s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/ea1af5bf979d483faaa1e976968af14b : FAILURE in 1h 03m 29s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/70448132be184dd7bcad1f86d4cccfb4 : FAILURE in 26m 51s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/33357a41054b48e293b9694d706beafe : FAILURE in 1h 10m 13s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/a6cba568738640249ae10d5e06953b3e : FAILURE in 1h 01m 27s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b0b4d31f19594183bf4d2f09634290bf : SUCCESS in 1h 00m 44s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f910a10a782a447391e3492db050e545 : SUCCESS in 15m 01s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/c7f38968f99c416ca83439377faee6c1 : SUCCESS in 1h 26m 09s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/055c1ab76665434ba745a33d37f941bb : SUCCESS in 1h 26m 10s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/14e84cfcb90347bea850f6d5f2f7a1d6 : SUCCESS in 1h 46m 53s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/b8207d8f36f5433cbe783ad058320ed9 : SUCCESS in 1h 36m 52s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/11199cee45d3475e8b6e875d0eb6b7c1 : SUCCESS in 2h 10m 18s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ac703e56994241068d8f6d35d87d0488 : SUCCESS in 1h 30m 52s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/d3fda98ea93e424f928d7076956d0ba8 : SUCCESS in 2h 06m 17s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/ef74ca4192844407968313994d0a8e34 : SUCCESS in 2h 05m 40s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/44d837004e144a45b5b9966e7963bbda : SUCCESS in 2h 07m 46s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/77a5887548004160a7fba8b5d7ae0e16 : SUCCESS in 2h 50m 11s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/4eca9ed7854042a28307d7680da32f87 : SUCCESS in 2h 01m 24s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/dcf51349b96249f793b5645cc1a5f8d9 : SUCCESS in 2h 51m 25s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/8ae0fe4d104f4a2bb2d1ecf493ab2baa : FAILURE in 41m 42s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/8ba0758d9e1d48b6afe9b01db4a4c3db : FAILURE in 48m 37s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/6f4e42e4f6774a36a077dcb54deb5d13 : SUCCESS in 1h 52m 16s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/27e947a0a7ec4c82b0307d394abd6bfb : SUCCESS in 28m 41s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/23887c3eb7a843a29a82eef3edef589c : FAILURE in 29m 30s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/5ae607e9b69048229eb3c231d5ac11a7 : FAILURE in 22m 39s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/e1839334e1cf4c99934888a3ea3f4591 : FAILURE in 1h 07m 25s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/4b085da1ee8e43bfb47cbf1324b682bb : SUCCESS in 1h 15m 13s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/c9d4dc9ef11b4b87b4d70ce59481cdbc : SUCCESS in 57m 17s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/1473ca03a198443383e69c1b496e3054 : SUCCESS in 1h 30m 14s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/baf887f04f83425c9d717298916cf6e9 : SUCCESS in 1h 28m 22s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/4b214216bf264d34a348c1d4e4877a26 : FAILURE in 1h 40m 00s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/3d6cf808a834449591fe90743438983d : SUCCESS in 2h 15m 09s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/f440344ad26942b0a3218d30d3797f27 : SUCCESS in 2h 31m 25s (non-voting)\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/2f20cca8b7be43fdbd52c1a6f052ec47 : SUCCESS in 1h 49m 30s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/b37d307784b744a1b58545e8c70adf36 : SUCCESS in 1h 06m 16s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/e3afeb493827419597c337d10ec4478d : SUCCESS in 57m 42s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/1d5e4cd307554940877a86a9077a8252 : SUCCESS in 1m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5fedef998cd4d5380d3c086c114f410 : FAILURE in 2m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3c353acaa45746aea18bd9b9fc357206 : SUCCESS in 2m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5de2862192c9493bb6ca6690802d3b8f : SUCCESS in 1m 53s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/eeb53348bce5469da0fad0593d175a4a : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ebde69ed905b4be091ab96e27d2a8cc6 : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afdfb0395056484594f29360692c9018 : SUCCESS in 6m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/96c9314aa2824ed992b85903aff63826 : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":4},{"id":"3d5344aa36987c3b9652e2d256a1756e229520a5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 06:56:02.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f210ba1978d452084089387a0917a9f\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/6dd658ec092741faba9fe10e667c26b5 : FAILURE in 1h 08m 21s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/a4d1fefcacfb4804aa99de3545070db6 : FAILURE in 1h 17m 19s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/090e776ea688497e8310e66ae52c9a92 : FAILURE in 1h 13m 14s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/e2a8bf5967984a53910558468a474291 : SUCCESS in 1h 06m 27s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/38a002a84f4c46daa4dc8a2f6522c67e : FAILURE in 1h 40m 53s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/4a2069b541374d028a5fc2be09fe7503 : SUCCESS in 1h 29m 17s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/832df926b106468a860340465763c47d : SUCCESS in 3m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f7d1d00627fa4b37a25f3c35feae8579 : SUCCESS in 3m 47s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/988b4172779a46bc880073da94c41a3f : SUCCESS in 7m 28s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"0d808bee70dbfa7540d7e1a60ce2a0ea75c9ff76","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2026-05-04 20:13:48.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"11694bf5b4cf565ede14859b443cecf550241e20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 03:17:59.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b25f03c22d784612a435e2f757402ece\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/c0bdf2ccf474419c87acf6f222b38fdd : SUCCESS in 1h 07m 13s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/4b30be7a54244c1cb1bf938a2e9a1a30 : TIMED_OUT in 6h 59m 06s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/1f0426fabe3a40eabae5124fdf589ab2 : FAILURE in 59m 57s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/64e4e7d3f9a140c4bcd42f364c8a2051 : SUCCESS in 1h 02m 13s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/693be4c2c9bf4a0d83db6e39bd90802e : SUCCESS in 36m 50s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4d508292d2794496ae83e10e4bcca3d9 : FAILURE in 1h 46m 01s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e8f6b496e71a436ba0e743eb017769f4 : FAILURE in 42m 56s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/97009d29901a4f2583bcd622209a0173 : SUCCESS in 1h 20m 03s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/8698c6bb41864469b228dcf9e0ef2d36 : DISK_FULL in 41m 12s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/babeaf2cdaa5446486015f9edb66866e : FAILURE in 1h 11m 59s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/bc483caaf8b5437b805bb0e0f14c392b : SUCCESS in 1h 30m 27s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/19c31df5a56c4f18a1b1fe93c15958dc : SUCCESS in 1h 54m 58s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8023f98563544616afa21ec062ed111f : FAILURE in 33m 53s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/875e3a85c41242b0aaaa21bcef5f679e : FAILURE in 41m 01s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/5b7fbe3e4dfe48b39e949f79e81c3775 : TIMED_OUT in 3h 03m 56s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/9c6d058b3bda4c47945b8c244b6aab18 : SUCCESS in 2h 25m 23s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/362f402e4617412f9940d5ca6348588a : TIMED_OUT in 3h 04m 00s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/d15152a8efb444848e17fbff6c2b0b3d : FAILURE in 45m 10s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/8f3ced73ccd642cca336260389f6b6ef : FAILURE in 44m 21s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/c64c759458984ec6b89311fbe4d02c9e : TIMED_OUT in 2h 34m 07s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/e7527838ceb14511a12a9dc1921afdd1 : SUCCESS in 1h 59m 42s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/00ee7cff3a334930a10baac72c4dad9c : FAILURE in 49m 56s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/f3e89349aaa64c3da8a908e385c3a649 : FAILURE in 17m 23s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/0a339ab6c9a1400abdaa5893d6bf58cf : SUCCESS in 28m 55s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/f1179d349d0049959c7fd43759a13c02 : FAILURE in 17m 13s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/7956b1492b7241ef995c39f4621d3793 : FAILURE in 34m 04s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/51bf030d942b466186d01c1655b163fe : SUCCESS in 1h 26m 00s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/2fd2ee36d6c949ec86acb3fad2e40b5e : FAILURE in 38m 29s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/e81a831002ea47299c7f2ffb626fc075 : FAILURE in 42m 17s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/7d4559e4cfc943cd9d7c87e1cc3ace30 : FAILURE in 39m 58s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/2362bcded7cb42b68d928b1d7c04734d : FAILURE in 45m 16s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/fec9a3d96bcd4ff88851ecac0dafe0f1 : SUCCESS in 2h 20m 16s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/c53846149b1145a88fe86432f8e34889 : FAILURE in 43m 45s (non-voting)\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/11dd96c17cc6448fab74091dbb6ff4c0 : FAILURE in 42m 16s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/526d599a883f48f69a4e9dde390e4886 : SUCCESS in 2h 10m 51s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/1babdffe52db4dc0aa59d5d436ad5522 : FAILURE in 31m 14s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/acd673c1f6104954b0c5ebfbbd3b3ea1 : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1174be73d13407d89508da1cc16806a : FAILURE in 3m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5054c991e7ad489fb18c5fe422565ef9 : SUCCESS in 3m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dde7b7e787114cc5ba4166a86b98376a : SUCCESS in 3m 17s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7b89f6f2896b43fd8131a4177738f00e : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d2b56e1d0e684967a53fd560ce5242d2 : SUCCESS in 3m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ccb76309eff44ebbb05f2c70c5a7144 : SUCCESS in 6m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ecc4dd6680654fbabaf049aaa27012e0 : SUCCESS in 2m 25s","accounts_in_message":[],"_revision_number":5},{"id":"7fe482d0d04bd26ac823d83fac4a4753b56b8be8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 03:42:15.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c595f26fd69c4cd9bc8abc2f71c52f64\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/3b7f6769e61b40a49c81bc029ef46d2d : SUCCESS in 2h 06m 59s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/2e61c5fb238d46249d73417b2a3b8249 : TIMED_OUT in 7h 26m 18s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/8f798f98d9f04c16bc156dc2de3e7331 : SUCCESS in 2h 13m 49s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/9ee062af30f84c6a9ccbde6309d99c54 : FAILURE in 1h 50m 18s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/2104b2feb9774663a8eb8b74b4843d92 : FAILURE in 2h 00m 43s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/7b57865f648541e289628a61f5ce403c : TIMED_OUT in 3h 19m 34s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/556742b234824121a7b5901455d4f406 : SUCCESS in 4m 15s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3e300b4ea7ea4effa0a4c61d2f4a79bf : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/902b7dfde1b74675b460479f8abcc037 : SUCCESS in 6m 51s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"0cd6f38e5ce9bc2ff0e4aff54abf44e0de60b1a7","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2026-05-06 00:31:24.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"cdcd7de0148f54163ece7095a2a6540e8e5a1832","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 02:29:03.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a5cead267cc4b11b09a92096d2a0a0e\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/415db1f39f324ec19e2a215817f40e80 : SUCCESS in 1h 18m 08s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/632ff564bcdc41acb0fb1d380919702e : FAILURE in 1h 23m 07s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/d7a585a15a414c60ab16d194f1a4e73b : SUCCESS in 1h 18m 31s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/ab7a2566393b4f2cbe350a9128b251c4 : SUCCESS in 1h 26m 04s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/a4ecd19a97a540e691bc5bad9713900b : FAILURE in 1h 35m 53s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/bde0d44d90764c3b9a12e6c7953d5ffe : SUCCESS in 1h 56m 06s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/42e3fc41a4dd4d55bcc6165f0d6a725f : SUCCESS in 3m 16s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0ce55b4a693b4e259b4e95cda3d3821e : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/faa2fdd8ede64bc9bff47768890689a6 : SUCCESS in 7m 29s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c52c03cf4a1c1d82aa61f6e9d72b4727279a7a2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 03:28:20.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4d7603772934494eae6f8aa2791a2f4b\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/ab6046b7972045eabca053e7552db226 : SUCCESS in 1h 13m 15s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/b112d784569941d1aee20962cbf3f3b1 : FAILURE in 54m 30s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/3963c1a54f7142dab6d3445ae24baeb3 : FAILURE in 24m 39s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/574bfd3392d14a2bb7703f9f50ff9889 : SUCCESS in 1h 01m 59s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/c74c2f1db4d14ac885bcfe4fc812504b : SUCCESS in 44m 59s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/176f188be94e40bf9e2e273e819e4ae3 : SUCCESS in 1h 30m 26s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0a7901bda615484ea8c27f3c91eb637b : SUCCESS in 30m 21s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/1ae749331f714257b6d6ac1ebaf6caa7 : SUCCESS in 1h 24m 31s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/d0111ef49db04229a01f41f4ca0add59 : SUCCESS in 1h 26m 16s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/0b1be036176a4de8b6198b99bec4f9aa : SUCCESS in 1h 49m 24s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/c9571752b96048289f3a20291ec27362 : SUCCESS in 1h 33m 08s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/3d10ffeb76cd458ba1edc5722f7d6c83 : SUCCESS in 2h 03m 42s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c15b170567c94a068703f62681d9190d : SUCCESS in 1h 28m 58s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/288f15b1587046b6a77ac13ef134682f : SUCCESS in 1h 57m 44s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/3bc607dc3dfa4ce99a60f120b8bb6528 : SUCCESS in 2h 04m 27s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/e010acbe34cc47ffb8b1134fb48eb4b6 : SUCCESS in 2h 12m 19s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/02a6457716cf4910abb532efc0e3b7a1 : SUCCESS in 2h 55m 28s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/630bfb2d08a34cbe9f5a5e462c28db17 : SUCCESS in 1h 57m 45s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/60e64ece782f4af39159f681f239d79a : SUCCESS in 2h 46m 18s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/6a883344a15c45f6a1da763816a62488 : FAILURE in 41m 58s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/3956f2da0bc04b9a90aa980423281f68 : SUCCESS in 1h 57m 34s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/1eb367602a8b4c0a8e17548aca18e2db : SUCCESS in 1h 58m 05s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/3a8f78a93c6446eca959d652d6583160 : SUCCESS in 26m 09s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/745311c4579444608c038c8f79c491c9 : SUCCESS in 31m 13s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/ac6e3eb2108a465a88c0967c63aa75d2 : FAILURE in 21m 21s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/c5f1e248b13d46e5a6712fb8c85e6d98 : FAILURE in 1h 06m 30s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/0aad913230d2443c8bd301f4defe77c8 : SUCCESS in 1h 41m 13s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/41c40c9ab14f4fa6857b9c1418fcfd50 : SUCCESS in 1h 37m 02s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/5756d0533323452e812ec99b5d22980f : SUCCESS in 1h 25m 38s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/65464212ac49450aae09e777cb68fa8a : SUCCESS in 1h 27m 54s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/d8271ecff0a640b09ced9acef31908b5 : FAILURE in 1h 42m 27s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/b62e0d0e31da4e158d6009f49fc7789a : SUCCESS in 2h 19m 48s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/011e0d7435fc476cbca21aacbae676e2 : SUCCESS in 2h 23m 24s (non-voting)\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/4e80be325c464743918828c02083d47c : SUCCESS in 58m 02s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/989da7038fb74118a72a569d098c224a : SUCCESS in 1h 13m 50s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/7347356493284e6bba5998057d0a19b6 : SUCCESS in 1h 08m 40s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/2d9498183af648feb29f6bdc4d8dbc9c : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c147df9acfe4b73a60242af5944a364 : FAILURE in 3m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e98287a4dbda4f8e8d15b51cd174b40b : SUCCESS in 3m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/500e8f719eb34e389a14f961b3728812 : SUCCESS in 2m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/65be65448c3e4e5ea649fad48e11fe0c : SUCCESS in 4m 52s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/517911831bf149509112ccf0e722e7ef : SUCCESS in 2m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c88f87b871094549bb1a75a40c1ba202 : SUCCESS in 5m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a538ca7d52374bc58f329da5d73938bf : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":5},{"id":"10cda929c986c019a4f8fb1a535c54a856b0b2b8","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-07-31 07:11:39.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"1c70020e0aa1ce32b1bb0274c66ecedb118962c8","author":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"date":"2026-08-05 12:42:11.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"8da22db041a5a1636f310e7031f539ee2940988c","author":{"_account_id":39457,"name":"Murray Groves","display_name":"Murray Groves","email":"murray@stackhpc.com","username":"MurrayGroves"},"date":"2026-09-22 12:49:32.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7ab8bf8c417686fc5cd7091467973fa355b6ee18","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-09-22 12:58:06.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"3b1d952784d7be28e22570ad58785db6782e1b3e","author":{"_account_id":39457,"name":"Murray Groves","display_name":"Murray Groves","email":"murray@stackhpc.com","username":"MurrayGroves"},"date":"2026-09-22 13:12:56.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e2ac21ecd29a54820b1b168ca7b7b73b446c6dc4","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2026-09-22 20:16:27.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"5d7c3154dfc99340b74d34e8d879e69456e0c4d0","revisions":{"0cee1a177e21b8db9a8671e8f9851039f08ae97f":{"kind":"REWORK","_number":1,"created":"2026-05-03 22:43:33.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/09/987109/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/09/987109/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/987109/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/987109/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/987109/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/09/987109/1"}}},"commit":{"parents":[{"commit":"a962fb3d2427ceed31ba58757c4e675e3f5dea75","subject":"Add ironic-novncproxy container image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a962fb3d2427ceed31ba58757c4e675e3f5dea75"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2026-05-03 22:38:20.000000000","tz":0},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2026-05-03 22:38:51.000000000","tz":720},"subject":"Add ironic-vnc-container container image","message":"Add ironic-vnc-container container image\n\nAdd a new container image for the Ironic VNC container, introduced in\nthe 2025.1 (Epoxy) release, which provides a headless Firefox and\nx11vnc environment for Ironic graphical console drivers. Enable the\nmozillateam PPA on Ubuntu to provide the firefox-esr package.\n\nChange-Id: I07bcebfdc62ae16a87221ee1e66c2f38e36fc0c2\nSigned-off-by: Steve Baker \u003csbaker@redhat.com\u003e\nAssisted-By: Opencode - Claude Sonnet 4.6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0cee1a177e21b8db9a8671e8f9851039f08ae97f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0cee1a177e21b8db9a8671e8f9851039f08ae97f"}]},"branch":"refs/heads/master"},"6c625220beadbf48eecb88b236bf2c58d357af97":{"kind":"REWORK","_number":2,"created":"2026-05-03 23:29:06.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/09/987109/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/09/987109/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/987109/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/987109/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/987109/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/09/987109/2"}}},"commit":{"parents":[{"commit":"a962fb3d2427ceed31ba58757c4e675e3f5dea75","subject":"Add ironic-novncproxy container image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a962fb3d2427ceed31ba58757c4e675e3f5dea75"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2026-05-03 22:38:20.000000000","tz":0},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2026-05-03 23:27:24.000000000","tz":720},"subject":"Add ironic-vnc-container container image","message":"Add ironic-vnc-container container image\n\nAdd a new container image for the Ironic VNC container, introduced in\nthe 2025.1 (Epoxy) release, which provides a headless Firefox and\nx11vnc environment for Ironic graphical console drivers. Enable the\nmozillateam PPA on Ubuntu to provide the firefox-esr package.\n\nChange-Id: I07bcebfdc62ae16a87221ee1e66c2f38e36fc0c2\nSigned-off-by: Steve Baker \u003csbaker@redhat.com\u003e\nAssisted-By: Opencode - Claude Sonnet 4.6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6c625220beadbf48eecb88b236bf2c58d357af97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6c625220beadbf48eecb88b236bf2c58d357af97"}]},"branch":"refs/heads/master"},"453e7f36380602dddf4eb5d92fcc62aaad7560ff":{"kind":"REWORK","_number":3,"created":"2026-05-04 01:32:24.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/09/987109/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/09/987109/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/987109/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/987109/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/987109/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/09/987109/3"}}},"commit":{"parents":[{"commit":"a962fb3d2427ceed31ba58757c4e675e3f5dea75","subject":"Add ironic-novncproxy container image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a962fb3d2427ceed31ba58757c4e675e3f5dea75"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2026-05-03 22:38:20.000000000","tz":0},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2026-05-04 00:07:49.000000000","tz":720},"subject":"Add ironic-vnc-container container image","message":"Add ironic-vnc-container container image\n\nAdd a new container image for the Ironic VNC container, introduced in\nthe 2025.1 (Epoxy) release, which provides a headless Firefox and\nx11vnc environment for Ironic graphical console drivers. Enable the\nmozillateam PPA on Ubuntu to provide the firefox-esr package.\n\nChange-Id: I07bcebfdc62ae16a87221ee1e66c2f38e36fc0c2\nSigned-off-by: Steve Baker \u003csbaker@redhat.com\u003e\nAssisted-By: Opencode - Claude Sonnet 4.6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/453e7f36380602dddf4eb5d92fcc62aaad7560ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/453e7f36380602dddf4eb5d92fcc62aaad7560ff"}]},"branch":"refs/heads/master"},"80e326647fcab8644c5983f83d83bba7751dbb8c":{"kind":"REWORK","_number":4,"created":"2026-05-04 03:55:51.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/09/987109/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/09/987109/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/987109/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/987109/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/987109/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/09/987109/4"}}},"commit":{"parents":[{"commit":"a962fb3d2427ceed31ba58757c4e675e3f5dea75","subject":"Add ironic-novncproxy container image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a962fb3d2427ceed31ba58757c4e675e3f5dea75"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2026-05-03 22:38:20.000000000","tz":0},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2026-05-04 03:55:38.000000000","tz":720},"subject":"Add ironic-vnc-container container image","message":"Add ironic-vnc-container container image\n\nAdd a new container image for the Ironic VNC container, introduced in\nthe 2025.1 (Epoxy) release, which provides a headless Firefox and\nx11vnc environment for Ironic graphical console drivers. Enable the\nmozillateam PPA on Ubuntu to provide the firefox-esr package.\n\nChange-Id: I07bcebfdc62ae16a87221ee1e66c2f38e36fc0c2\nSigned-off-by: Steve Baker \u003csbaker@redhat.com\u003e\nAssisted-By: Opencode - Claude Sonnet 4.6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/80e326647fcab8644c5983f83d83bba7751dbb8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/80e326647fcab8644c5983f83d83bba7751dbb8c"}]},"branch":"refs/heads/master"},"5d7c3154dfc99340b74d34e8d879e69456e0c4d0":{"kind":"REWORK","_number":5,"created":"2026-05-04 20:13:48.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/09/987109/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/09/987109/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/987109/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/987109/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/987109/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/09/987109/5"}}},"commit":{"parents":[{"commit":"a962fb3d2427ceed31ba58757c4e675e3f5dea75","subject":"Add ironic-novncproxy container image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a962fb3d2427ceed31ba58757c4e675e3f5dea75"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2026-05-03 22:38:20.000000000","tz":0},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2026-05-04 20:12:53.000000000","tz":720},"subject":"Add ironic-vnc-container container image","message":"Add ironic-vnc-container container image\n\nAdd a new container image for the Ironic VNC container, introduced in\nthe 2025.1 (Epoxy) release, which provides a headless Firefox and\nx11vnc environment for Ironic graphical console drivers. Enable the\nmozillateam PPA on Ubuntu to provide the firefox-esr package.\n\nRHEL 10 and family no longer packages x11vnc and xvfb.  These will need\nto be installed from source in a later change.\n\nChange-Id: I07bcebfdc62ae16a87221ee1e66c2f38e36fc0c2\nSigned-off-by: Steve Baker \u003csbaker@redhat.com\u003e\nAssisted-By: Opencode - Claude Sonnet 4.6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5d7c3154dfc99340b74d34e8d879e69456e0c4d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5d7c3154dfc99340b74d34e8d879e69456e0c4d0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
