)]}'
{"id":"openstack%2Fkolla~987794","triplet_id":"openstack%2Fkolla~master~Ie28d1cc615cc09f013438285b4dcbc2dbab0f4b5","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-05-08 11:57:10.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}},"hashtags":[],"change_id":"Ie28d1cc615cc09f013438285b4dcbc2dbab0f4b5","subject":"Remove influxdb from repos.yaml","status":"MERGED","created":"2026-05-08 05:53:39.000000000","updated":"2026-05-08 14:05:41.000000000","submitted":"2026-05-08 14:04:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"987794","meta_rev_id":"b8d0e24bb47e575ff86306081de8a0b4f7a79f97","_number":987794,"virtual_id_number":987794,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":37313,"name":"Alma MAILLET-CONTOZ","display_name":"Alma MAILLET-CONTOZ","email":"alma.maillet-contoz@infomaniak.com","username":"nicolasmc","status":"Intern SRE at Infomaniak"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-08 14:04:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"recommended":{"_account_id":37313,"name":"Alma MAILLET-CONTOZ","display_name":"Alma MAILLET-CONTOZ","email":"alma.maillet-contoz@infomaniak.com","username":"nicolasmc","status":"Intern SRE at Infomaniak"},"all":[{"value":1,"date":"2026-05-08 09:23:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37313,"name":"Alma MAILLET-CONTOZ","display_name":"Alma MAILLET-CONTOZ","email":"alma.maillet-contoz@infomaniak.com","username":"nicolasmc","status":"Intern SRE at Infomaniak"},{"value":2,"date":"2026-05-08 08:25:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2026-05-08 11:57:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2026-05-08 08:27:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-05-08 08:54:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":37313,"name":"Alma MAILLET-CONTOZ","display_name":"Alma MAILLET-CONTOZ","email":"alma.maillet-contoz@infomaniak.com","username":"nicolasmc","status":"Intern SRE at Infomaniak"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2026-05-08 11:57:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":37313,"name":"Alma MAILLET-CONTOZ","display_name":"Alma MAILLET-CONTOZ","email":"alma.maillet-contoz@infomaniak.com","username":"nicolasmc","status":"Intern SRE at Infomaniak"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2026-05-08 11:57:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":37313,"name":"Alma MAILLET-CONTOZ","display_name":"Alma MAILLET-CONTOZ","email":"alma.maillet-contoz@infomaniak.com","username":"nicolasmc","status":"Intern SRE at Infomaniak"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2026-05-08 11:57:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},{"_account_id":37313,"name":"Alma MAILLET-CONTOZ","display_name":"Alma MAILLET-CONTOZ","email":"alma.maillet-contoz@infomaniak.com","username":"nicolasmc","status":"Intern SRE at Infomaniak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-08 07:18:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-08 07:19:33.000000000","updated_by":{"_account_id":37313,"name":"Alma MAILLET-CONTOZ","display_name":"Alma MAILLET-CONTOZ","email":"alma.maillet-contoz@infomaniak.com","username":"nicolasmc","status":"Intern SRE at Infomaniak"},"reviewer":{"_account_id":37313,"name":"Alma MAILLET-CONTOZ","display_name":"Alma MAILLET-CONTOZ","email":"alma.maillet-contoz@infomaniak.com","username":"nicolasmc","status":"Intern SRE at Infomaniak"},"state":"CC"},{"updated":"2026-05-08 07:19:54.000000000","updated_by":{"_account_id":37313,"name":"Alma MAILLET-CONTOZ","display_name":"Alma MAILLET-CONTOZ","email":"alma.maillet-contoz@infomaniak.com","username":"nicolasmc","status":"Intern SRE at Infomaniak"},"reviewer":{"_account_id":37313,"name":"Alma MAILLET-CONTOZ","display_name":"Alma MAILLET-CONTOZ","email":"alma.maillet-contoz@infomaniak.com","username":"nicolasmc","status":"Intern SRE at Infomaniak"},"state":"REVIEWER"},{"updated":"2026-05-08 08:21:09.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"CC"},{"updated":"2026-05-08 08:25:15.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2026-05-08 08:27:11.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-05-08 08:54:38.000000000","updated_by":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"reviewer":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"822d5e2f215d5448ff22a82ad5cd59460e042dbc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-08 05:53:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0c34ca3b02db978b985c4a5bfa136235bb0d59ba","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-08 05:53:52.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"f992c8e7a8029ac2239a5e94566769c5eb0ea9fc","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-08 06:17:25.000000000","message":"Patch Set 1: Backport-Candidate+1","accounts_in_message":[],"_revision_number":1},{"id":"834790a51220e8edaa61846220a4729fdae5c2df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-08 07:18:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7d0e7c374ac4128919a1fe9261a8cee\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/18709f5aedad4b848964015805d50d11 : SUCCESS in 1h 09m 37s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/174798c7d8db46409f522fd9f6204931 : SUCCESS in 1h 20m 10s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/e70edc5ab77c4e70a2b9d1119e31b843 : SUCCESS in 24m 59s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/ce4f6dcefd8343a9bdc50e7ce116646d : SUCCESS in 59m 42s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/0d8ddfe006e64b7fb41dd535d7eccc3b : SUCCESS in 48m 51s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3d476d27200746acadf53e9469f25d95 : SUCCESS in 1h 13m 23s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e5604516f2314cab879b1768a5c6fc89 : SUCCESS in 26m 34s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/6215be853e0d4779b7895c1cc7d2c421 : SUCCESS in 2m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35053ff99195414b9dfd41cb949afdb7 : SUCCESS in 4m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/44a91fff5669439f90df8eada646ee57 : SUCCESS in 2m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0f2503f9e9d24acc8d446630ad324a1d : SUCCESS in 3m 28s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5852df49d37041acb7dc836e1b5c2e66 : SUCCESS in 7m 38s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d8a60d264803498990f0fd0dd951a7e1 : SUCCESS in 2m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0134a3fa169d4754ba2d8f8069b99701 : SUCCESS in 6m 14s","accounts_in_message":[],"_revision_number":1},{"id":"0fcff9c0693d17cf437fae98590c4a4123c88be9","author":{"_account_id":37313,"name":"Alma MAILLET-CONTOZ","display_name":"Alma MAILLET-CONTOZ","email":"alma.maillet-contoz@infomaniak.com","username":"nicolasmc","status":"Intern SRE at Infomaniak"},"date":"2026-05-08 07:19:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fbf8dd835addbb721b7f712ae7665d0376cef675","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-08 08:19:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2209e36c962747729d393885aef0eb7c\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/628d2f9711bb4ce68f0b200d9659ef61 : SUCCESS in 1h 20m 20s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/0a19780ca24c466eb606bda70574d255 : SUCCESS in 1h 29m 21s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/fe3583ceeb644c479efe202f867495e6 : POST_FAILURE in 2h 23m 59s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/e0f86a9680dc42ff9efa39c5029e28b9 : SUCCESS in 2h 14m 53s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ad25b00fbd634d5cbbdd58b1c68a2f11 : SUCCESS in 3m 19s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/31a0932baa414f979780f46e68b8e0d2 : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f7867106749a423e9f68a6ca74ba0dbb : SUCCESS in 7m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e16b44d3693dee67c4501fa94dd52c99c0442177","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-05-08 08:21:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9dd6c53c747086eea001b9f490d8484b10d424f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-08 08:23:50.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"aee95eef6ae63d06db7998710c7844401e54d30d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-08 08:24:38.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":3},{"id":"47c5790d99504fd89f3db79172aa41a79faa21ec","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2026-05-08 08:25:15.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"d34b1c802b16047877a50cd5daf3218c806428e6","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-05-08 08:27:11.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"1a961e89614e998303cfb741f7eefc0708f0f8fd","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2026-05-08 08:54:38.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"f9a8dac01a03f89b66494785810ac9bff76b380d","author":{"_account_id":37313,"name":"Alma MAILLET-CONTOZ","display_name":"Alma MAILLET-CONTOZ","email":"alma.maillet-contoz@infomaniak.com","username":"nicolasmc","status":"Intern SRE at Infomaniak"},"date":"2026-05-08 09:23:32.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d94bef8302225b9ce64bf00c5b5b4ae99acc2b48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-08 11:21:50.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f035fb334ff741bfb47888e7c5df3add\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/032959b866dc4b0fa1f8c5402082047f : SUCCESS in 1h 18m 40s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/350c1ea464d34845ae7a3a67d9b11688 : SUCCESS in 1h 11m 21s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/590a2c99190746f1a8b1a4327b20ce0c : SUCCESS in 27m 05s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/45f90564d76c4d8ba37a3495ebd09f15 : SUCCESS in 1h 15m 53s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/1cefc5e40e724d4398e9585855e44071 : SUCCESS in 1h 01m 20s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/af4bc25543b846c6925b3eeb42d3458d : SUCCESS in 1h 05m 26s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7027427eeaa14ccfb11031b201a55cef : SUCCESS in 26m 50s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/6fff3ef8dc54478cbf948682e1ba9a37 : SUCCESS in 1h 24m 04s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/81b77f34207747e18b9c84e418b637f1 : SUCCESS in 1h 27m 51s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/440f41cf8ad34d12bb33b2e73534d5a8 : SUCCESS in 1h 52m 32s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/7980b36c2556488bab50adb071c35866 : SUCCESS in 1h 30m 31s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/b807b7f57c3b429ba57fb59c40e62cf1 : SUCCESS in 2h 04m 43s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9e5040a4e2be4192a52ba9b90a32e8fb : SUCCESS in 1h 28m 28s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/c212925a514d458ebb4782d5d5ca5fd9 : SUCCESS in 1h 59m 59s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/ac2ddc3275f5485f8a629a0aad0c0524 : SUCCESS in 1h 56m 26s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/295e221d2fae454b88afd9636df8920f : SUCCESS in 2h 07m 02s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/ea684decba8d4adaa8d279bd4ad452ae : SUCCESS in 2h 55m 32s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/39079d37bb7d480bb171686f6852e530 : SUCCESS in 2h 00m 00s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/556ea09f2bac46e1aabcba3b535379ac : FAILURE in 1h 28m 05s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/07d4d4b2a25d40b7b479835a761e6673 : FAILURE in 1h 06m 43s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/621751e39dba4c5c949b566fb5b57a4c : FAILURE in 1h 07m 31s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/c89499db7f7d49efbd7d4bbe75717a0c : FAILURE in 55m 31s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/148c4b436cb84775b42b289dec5144ed : SUCCESS in 26m 32s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/2296ed9e646c4b859bfba4cdf85a67fc : SUCCESS in 29m 41s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/75d7ba0620aa4624ade59029c46e0b7e : SUCCESS in 30m 03s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/2ddd1237fd17451aa030064745640f8b : FAILURE in 57m 18s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/ff6a4ae2071948048738b9ecbd3dd67a : SUCCESS in 1h 22m 21s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/832ad81e42014290a7bcad95d4b495fb : SUCCESS in 1h 53m 45s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/99723dd6789c4c1bab2cffcb4ee3e128 : SUCCESS in 1h 24m 38s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/750f3e6cb258437182b54d5d703ad22a : SUCCESS in 1h 32m 30s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/ae0e17109c6a484a9ccf4594fd5476ea : FAILURE in 1h 38m 02s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/0a45db4683c1402f99287602307bd390 : SUCCESS in 2h 19m 21s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/d9a122e5e5b0462e96ff764078de926e : FAILURE in 1h 41m 09s (non-voting)\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/40cf0c14c16b4bb0a210934272e1c2ab : SUCCESS in 2h 09m 12s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/4a0886e97d194374b310fc917127bb6b : SUCCESS in 2h 10m 05s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/ba77113acedc401cb297d59d4eb64e1f : SUCCESS in 1h 04m 26s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/28de39c11646414091d335cff54497ac : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97d416fe2ca14ddd9ad0126f317c07d2 : SUCCESS in 4m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/41b444541a0d48c49d21d242f0750425 : SUCCESS in 5m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d11a358c8fa14fdb91d4c79ab10d753c : SUCCESS in 3m 50s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f4e68a5f243e4607b079309ec7eda374 : SUCCESS in 6m 21s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3221b26d29724c31b925a872bd50fc0c : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23bf7ad2f61849039aa5e11ba0a89ba1 : SUCCESS in 6m 26s","accounts_in_message":[],"_revision_number":3},{"id":"5565362a780e6b22dc9bddd8e2abe70efd6f07a3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-08 11:46:54.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b7436aaf1164c63845845c8474aa456\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/62fbce560aee47f3a1f4d25c9abb5e3e : SUCCESS in 1h 25m 44s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/bf53749838fc4db8a7262fb4f0fe184c : SUCCESS in 1h 31m 33s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/30b7f125aa864ca0a833b01a1851c2d4 : SUCCESS in 1h 36m 35s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/970371c5792b4ebea0f6cea82c98c087 : SUCCESS in 1h 36m 48s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/8a082bdefbc1412ebce7dad9a87cb8f6 : FAILURE in 2h 12m 57s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/26029d27cc184594bb2773229843bc91 : SUCCESS in 2h 08m 53s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4f81454264c9456389e4bdac140a3f27 : SUCCESS in 3m 25s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7b873718d76e4e878a138340385627a4 : SUCCESS in 4m 01s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/48bcef4aac944f67b995aecb1dd1e6fa : SUCCESS in 8m 10s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"49dc97c77632f368b9895643b36ecbf87120edfc","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-08 11:57:10.000000000","message":"Patch Set 3: Backport-Candidate+1 Code-Review+2 Review-Priority+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9f36bd357b999b3a655eb603940ac02cff86a1b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-08 11:57:25.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9eb7a80f84832202ab7202cada8b7a30b925ae11","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-08 14:04:22.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0e19216b9b44f188445f0f11e2673ed\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/cec58df31ebf4993b259c0d8adea74bf : SUCCESS in 1h 31m 52s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/99bb727eead7432b84b17137c0cc9d56 : SUCCESS in 26m 14s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/1e23b079de984aaf82d8fcea5c9bf2b7 : SUCCESS in 1h 00m 10s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/32462e35f09d46b29dc38bde22edf801 : SUCCESS in 45m 47s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8de6235c126a41f09f3c79651e16cb73 : SUCCESS in 1h 20m 00s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7b7b6bf1e53b4efe94a34b7f387f60c5 : SUCCESS in 29m 26s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/2319cdfebdf04dc59fc96a1663caceb2 : SUCCESS in 1h 23m 47s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/6e62f40557524b32b655d7f8949f36f3 : SUCCESS in 1h 51m 08s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/e6773b73e38447219a8b7f95e302753e : SUCCESS in 1h 38m 59s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/46bdc3c92b3643ed83e4e8f3d6687862 : SUCCESS in 2h 04m 16s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/79dbbbdfdd6a41cf94490cfb74fa2455 : SUCCESS in 1h 28m 09s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/f356f7d5b3ce48f9a6f12abda145f611 : SUCCESS in 2h 05m 57s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/c640ec1540374011ad79fd98594935f5 : SUCCESS in 2m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97388ef5dbf54dcfa3a86dd975bc7ce6 : SUCCESS in 2m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/14c8e064ef8b485e84987b975624bbdd : SUCCESS in 3m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/aaf3fbba4e2f420c972f5d89321a7d52 : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b2a7a9a89ac4e269cd54234e2e8f90d : SUCCESS in 8m 24s","accounts_in_message":[],"_revision_number":3},{"id":"59b942102d2d99119300a38bb28d81a581bd59ea","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-08 14:04:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"b8d0e24bb47e575ff86306081de8a0b4f7a79f97","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-08 14:05:41.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/888175378e5c4cc5986821f8af7bde41\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98dabb4610fe4c50ad4d6bac22c4685b : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"83b8161b12bb1a32b8b872c46533cfa6a2fe9d7f","revisions":{"0043a2b54b8baf6b43b4475bb073eaaa3c925554":{"kind":"REWORK","_number":1,"created":"2026-05-08 05:53:39.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/94/987794/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/94/987794/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/987794/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/987794/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/987794/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/94/987794/1"}}},"commit":{"parents":[{"commit":"610eef2089c186c6e698bb1781501ee8c696f7d5","subject":"Drop apache installation outside of httpd container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/610eef2089c186c6e698bb1781501ee8c696f7d5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-05-08 05:52:59.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-05-08 05:53:36.000000000","tz":120},"subject":"Remove influxdb from repos.yaml","message":"Remove influxdb from repos.yaml\n\nIt was ommitted in the original influxdb removal patch [1].\n\n[1]: https://review.opendev.org/c/openstack/kolla/+/973051\n\nChange-Id: Ie28d1cc615cc09f013438285b4dcbc2dbab0f4b5\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0043a2b54b8baf6b43b4475bb073eaaa3c925554"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0043a2b54b8baf6b43b4475bb073eaaa3c925554"}]},"branch":"refs/heads/master"},"82b213ccc74d6a30d0b230860eb01fe7d710fb36":{"kind":"REWORK","_number":2,"created":"2026-05-08 08:23:50.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/94/987794/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/94/987794/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/987794/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/987794/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/987794/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/94/987794/2"}}},"commit":{"parents":[{"commit":"610eef2089c186c6e698bb1781501ee8c696f7d5","subject":"Drop apache installation outside of httpd container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/610eef2089c186c6e698bb1781501ee8c696f7d5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-05-08 05:52:59.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-05-08 08:23:39.000000000","tz":0},"subject":"Remove influxdb from repos.yaml","message":"Remove influxdb from repos.yaml\n\nIt was ommitted in the original influxdb removal patch [1].\n\n[1]: https://review.opendev.org/c/openstack/kolla/+/973051\n\nChange-Id: Ie28d1cc615cc09f013438285b4dcbc2dbab0f4b5\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/82b213ccc74d6a30d0b230860eb01fe7d710fb36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/82b213ccc74d6a30d0b230860eb01fe7d710fb36"}]},"branch":"refs/heads/master"},"83b8161b12bb1a32b8b872c46533cfa6a2fe9d7f":{"kind":"REWORK","_number":3,"created":"2026-05-08 08:24:38.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/94/987794/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/94/987794/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/987794/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/987794/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/987794/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/94/987794/3"}}},"commit":{"parents":[{"commit":"610eef2089c186c6e698bb1781501ee8c696f7d5","subject":"Drop apache installation outside of httpd container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/610eef2089c186c6e698bb1781501ee8c696f7d5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-05-08 05:52:59.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-05-08 08:24:25.000000000","tz":0},"subject":"Remove influxdb from repos.yaml","message":"Remove influxdb from repos.yaml\n\nIt was ommitted in the original influxdb removal patch [1].\n\n[1]: https://review.opendev.org/c/openstack/kolla/+/973051\n\nChange-Id: Ie28d1cc615cc09f013438285b4dcbc2dbab0f4b5\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/83b8161b12bb1a32b8b872c46533cfa6a2fe9d7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/83b8161b12bb1a32b8b872c46533cfa6a2fe9d7f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d1"],"failing_atoms":["label:Backport-Candidate\u003d-1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
