)]}'
{"id":"openstack%2Fkolla~987796","triplet_id":"openstack%2Fkolla~master~Id25e39cee9fd64d147301a39528aa8525bd682c2","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-05-08 11:12:28.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}},"hashtags":[],"change_id":"Id25e39cee9fd64d147301a39528aa8525bd682c2","subject":"ironic-pxe: Remove httpd setup call","status":"MERGED","created":"2026-05-08 06:17:05.000000000","updated":"2026-05-08 15:36:37.000000000","submitted":"2026-05-08 15:35:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"987796","meta_rev_id":"0e0837e4a1bffbe47996fa5d6540fd4cf2b4c5b9","_number":987796,"virtual_id_number":987796,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-08 15:35:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2026-05-08 13:23:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2026-05-08 08:22:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2026-05-08 08:27:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-05-08 08:41:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2026-05-08 13:23:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2026-05-08 06:17:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2026-05-08 06:17:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-08 08:18:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-08 08:22:55.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-05-08 08:27:46.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2026-05-08 08:41:24.000000000","updated_by":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"reviewer":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"224e592d1b0f7ef13da9de7ce0d88fadbec61c2f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-08 06:17:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ad4942e0fe97c40ea50da1463aa3965a71d7b999","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-08 06:17:15.000000000","message":"Patch Set 1: Backport-Candidate+1 Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"4f1a010fcedc65d543d99956f04662fc6f3f7528","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-08 08:18:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f6cef97f13804452834f6da0789d73e5\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/83bb69cb2b74466a92b2b8729be1483d : SUCCESS in 1h 10m 07s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/9768961d425746ba89b357d232fb02ab : SUCCESS in 1h 04m 30s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/d9493928a04e4b4da76b1709be3f8033 : SUCCESS in 26m 52s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/214e943c755a429da0b723a70dac02ed : FAILURE in 14m 31s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/5ae3dd7d8a8745b9b7c328e046580c8b : SUCCESS in 52m 28s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/919e0be6631945a083478d6dba28b0d5 : SUCCESS in 58m 13s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ab0380c81f9c4e87a49df19eb848c8ec : FAILURE in 28m 21s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/a20759fdc707473ea9064b82ad414378 : SUCCESS in 1h 15m 15s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/284166885c464d768cf7915e4d34316e : SUCCESS in 1h 21m 13s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/f08e19b728d24654889706947c027d42 : SUCCESS in 1h 41m 23s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/517035ca692e4e34a4d9c5ecc9638764 : SUCCESS in 1h 33m 26s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/3396db7191a34d7ab337f6c0943fea08 : SUCCESS in 1h 59m 33s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8e5773cd283740dcb4d378708928b3fc : SUCCESS in 1h 24m 33s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/5b3f6b8687c043bd8a735595c25c7f75 : SUCCESS in 1h 58m 10s\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/8fb95e4a8bcd433eac48d81250ccdca2 : SUCCESS in 1h 46m 29s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/60728e2a12be4468b9264164c035c6fa : SUCCESS in 1h 58m 22s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/d5c6fe6145614f3daca6f5ed2f12c568 : FAILURE in 39m 08s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/f22e0db4764e42778babc70bf491cb45 : SUCCESS in 2m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9581a6ca44964cbc84ea01a96aee988f : SUCCESS in 3m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0a251d49d148417f8de1f307e84bfc2c : SUCCESS in 2m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ba56beb4f24346259ea79d94bb8f1d2c : SUCCESS in 3m 49s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b5da829da9484877a40b0761a831e463 : SUCCESS in 3m 11s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cbf92af7a2dd4b0893d10d9af26e1249 : SUCCESS in 2m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52cc22147aca4358a2436e17ee324e39 : SUCCESS in 6m 19s","accounts_in_message":[],"_revision_number":1},{"id":"35cd2d1c49dc6ca37340ed9b6c49a886367accff","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-08 08:22:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4ba57dc0cc08fcc211834f4d2c493c84ecc6dffc","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-05-08 08:22:55.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ff10a7466f73f2522190214dc0fb0565d7a53558","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-05-08 08:27:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1e4f1be0f1c184900ea8c2f202e6875bc9f87163","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2026-05-08 08:41:24.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ef423c84708c80405cfc2b2b6f3dfef05aadcf66","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2026-05-08 08:46:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0dc2660c65659ca47c0d4f68fd9c5e18984dbd91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-08 10:40:05.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/57757bf4c0334ac3a5130c1fb3c2ae4d\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/ee19be8e607f4b5b81b0b9c4a3b49122 : SUCCESS in 1h 04m 02s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/6d7acbd5810d413cb90e076b1cc573d4 : SUCCESS in 1h 29m 57s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/39a11e0aff584941805cbdd7072254f5 : SUCCESS in 29m 25s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/091e8f4ee0454b79beb9c7d5140553f5 : SUCCESS in 1h 14m 57s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/67a86bc4b95144a0b0b504dbbd24a6ea : SUCCESS in 43m 54s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/07385fee6016421a87ba80bf2d3d6d3d : SUCCESS in 56m 47s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/52a00bc04794471fa97ff76f99c54e4c : SUCCESS in 29m 55s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/e6c403d396444f1cac0428cb9aaf04c0 : FAILURE in 1h 19m 52s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/cbe217784e56419e88b59685df9b5ee6 : SUCCESS in 1h 21m 56s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/49e4af7232a14193be4e6450686d04ae : SUCCESS in 2h 04m 04s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/7c6ab998591b44ecaffcf61e46418099 : SUCCESS in 1h 45m 26s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/3f1a79d489f34fdfbc4fa33ebdf53377 : SUCCESS in 2h 16m 16s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/542604daf5014aad9dde2d712733622b : SUCCESS in 1h 42m 02s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/6dfa8da6fcbe4e2ea7bd4f28a7fd66f5 : SUCCESS in 1h 58m 59s\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/1f6de2f81f21422687dc26e628bf4703 : SUCCESS in 1h 49m 50s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/30ea7fae5bc146f69955e5a784d3aa33 : SUCCESS in 1h 55m 57s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/f50f95f570504f0d815ababf1738dc49 : SUCCESS in 1h 52m 41s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/04cd083e5d1646d0b14193caf822414b : POST_FAILURE in 3m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92f1d74555804595a693fd3544dc3c5e : SUCCESS in 3m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/30efe4051ebf4e6e8b118a14b247441f : SUCCESS in 4m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/06a164b581754db3a239174eedd789e7 : SUCCESS in 2m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4733a17b562b4b14b64c8ad9c87f0e58 : SUCCESS in 6m 02s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f1198810271d42328d6acc9d91c99be4 : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/229192ee3b3a4881aa0db36993cb1a77 : SUCCESS in 6m 00s","accounts_in_message":[],"_revision_number":1},{"id":"36339448c6911165b4e6528ca459c081aee7cb7e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-08 10:50:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/671d51b8906040829b289175a159c72d\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/b4f9b14f9d0c4f8484e19fa00096493e : SUCCESS in 1h 23m 35s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/9df2e526ebe145b6b0ae1568127295a7 : SUCCESS in 1h 24m 38s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/24f82388d3cc4159a8709686e7906f9a : SUCCESS in 1h 21m 00s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/4ab549786558463fa91d57cb4ce2a857 : SUCCESS in 1h 30m 42s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/90d6bdb645174ec5b448f5f3a7ec4f70 : FAILURE in 1h 38m 31s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/d67539b1cebd47c6929d434fb23fac35 : SUCCESS in 2h 21m 50s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c14b2a0ec16241f398e19e2b1c742acf : SUCCESS in 3m 52s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/810a277b95264de8a1e43e8076c5e9ba : SUCCESS in 4m 01s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/46d97f20395e4848aeb10002fdad6673 : SUCCESS in 7m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e5c083eb38d6235326f4d07c0c18798ee0d00687","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-08 11:12:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"16b640dc2d291a4c6a55f2d7b5b40de514457b33","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-08 13:23:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e72cbe4598ffae4a31096c310822e34af156e3cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-08 13:24:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e26b65267f5e4c7f8c950cc2f10913ba\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/e928f2c0102d4dc3b64a00e63c76ac21 : SUCCESS in 1h 00m 57s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/17f9c2cf57714d9aa6e8f8c64e9f92a2 : SUCCESS in 53m 55s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/14942367ef834f97909dada75ee9b421 : SUCCESS in 13m 40s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/ba365e2a340849a0b9eb673d83e8c526 : SUCCESS in 1h 00m 14s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/36dfd2db6937411ea2ee4a7ab518ac53 : SUCCESS in 34m 47s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9d54147de92843659a623807d263e7c9 : SUCCESS in 1h 03m 16s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d963252b72ef455fb9f783ef039da14b : SUCCESS in 31m 13s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/2f48104af6d04948b0229e35d709bf0e : SUCCESS in 1h 26m 53s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/3a6368a1006c4db7ad51c183dfcc34a9 : SUCCESS in 1h 30m 00s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/6914037243b7457aa60506539c8b93fc : SUCCESS in 1h 48m 48s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/1b2f8d9634c848b892b9e6ddc30d3892 : SUCCESS in 1h 37m 28s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/8cde182dd05a418fb7569fb64bf51dba : SUCCESS in 2h 10m 40s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/685acb81b52b44cb96d71280cc03ee3e : SUCCESS in 1h 32m 08s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/d1e7952eb4394363b80fbb088243a7ad : SUCCESS in 2h 02m 36s\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/b3ea89c8f1e849f8834aeba3915a2b76 : SUCCESS in 1h 49m 00s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/11548dd7da8a40bb8b070d0378fceb28 : SUCCESS in 2h 00m 25s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/0633ab59509d424ab621dc2a16d6c20e : SUCCESS in 1h 50m 55s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/04195481a3294357bbc81b0bf59d7453 : SUCCESS in 2m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97f474a7b2ad46de8aa329a2d5757316 : SUCCESS in 3m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/07d5e8bb18794cce89872c34dbc27216 : SUCCESS in 4m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3f7df9f4d39b49d5ae92c33e41b86a79 : SUCCESS in 3m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b5cd8e8089854d4fab55ccaca1d741e6 : SUCCESS in 4m 18s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/95c9c4971ca140c7925cb556977178d1 : SUCCESS in 2m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1512134565644df58d739e45efdac262 : SUCCESS in 7m 25s","accounts_in_message":[],"_revision_number":1},{"id":"02d2d52d67d942b5372886cb8bf891b8d39b5598","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-08 13:25:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"63cfc26adc5bf3643d9cf504f0aa674dea7408e2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-08 14:44:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83b23d5279884c418fc685c12c51aaff\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/5b1c479b8a31468e9fe5e309a6f4d508 : SUCCESS in 1h 22m 21s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/d54086d568404f43a3df341a05f023a0 : SUCCESS in 1h 58m 21s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/41b46d75860442e091572000c1c067eb : SUCCESS in 1h 21m 09s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/a8f8887116b64217868c1d6c783f66a5 : SUCCESS in 1h 18m 46s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/79baa88e657b4a5997df57a622a24272 : FAILURE in 1h 40m 52s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/30a356e1ea864f2689ee29736a273f01 : SUCCESS in 1h 50m 25s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/688eefee793e4d578dd40b249dd17460 : SUCCESS in 3m 39s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4738e2e4c37b44a38952e9cfa08708ca : SUCCESS in 3m 37s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/cf9435fcfb7a4f9fb6d726760a205688 : SUCCESS in 6m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0fb15aaa73e0a4229b8197e1b79f74fb2aa6f90f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-08 15:35:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67440fb2a3e64b799de029ed80a35862\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/714a18ecba964acb88e27792e0c29900 : SUCCESS in 1h 13m 30s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/5cf48ba24e6040d3b6d664aad8ec1985 : SUCCESS in 28m 59s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/9a9cd71143e94b7cad9060b3d0d7b88f : SUCCESS in 58m 04s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/650b25c490b94c5e96c8df0a06a602c6 : SUCCESS in 48m 49s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1a686c11d4024efaa0472a71fc60299a : SUCCESS in 1h 01m 19s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ae65b84c3ed4432993f0c1d57d349cfa : SUCCESS in 27m 27s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/203771654fae426a9e4100e4a8a9fbb6 : SUCCESS in 1h 26m 16s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/7cfb141f31934cde95c3ada65e3af3b0 : SUCCESS in 1h 51m 06s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/dd72c867e85b431aa10172a8c5358b2e : SUCCESS in 1h 33m 39s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/b076a49622c243dfaea6506ae9af9b17 : SUCCESS in 2h 08m 43s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0ecc1a02ac53458189e5720b00c45c5f : SUCCESS in 1h 24m 05s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/718c55ec752c404f8164b0eb88b1c045 : SUCCESS in 1h 55m 53s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/b1d3afc15d204d83a2a7d446f2d050d1 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecd5b95b1ee04813ab9bdcc793296982 : SUCCESS in 3m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2ec76b9d90f24b8fac01ab3a7b379b46 : SUCCESS in 3m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fbe900b9b6f54c1c93f5022e2a6a2de8 : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a25ec1a4ae764d5ba544e3c5888ce26f : SUCCESS in 6m 23s","accounts_in_message":[],"_revision_number":1},{"id":"cd49a29f4a43dd4b348ac0b7427e878324d069c4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-08 15:35:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0e0837e4a1bffbe47996fa5d6540fd4cf2b4c5b9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-08 15:36:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2fd5d553cb3441f7bd1eddb191fa4c98\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/485462993ea04bb7bc0f72d49fdbd03b : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5df2cc1252265bacef4d1dee30355d1ff5464709","revisions":{"5df2cc1252265bacef4d1dee30355d1ff5464709":{"kind":"REWORK","_number":1,"created":"2026-05-08 06:17:05.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/96/987796/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/96/987796/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/987796/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/987796/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/987796/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/96/987796/1"}}},"commit":{"parents":[{"commit":"610eef2089c186c6e698bb1781501ee8c696f7d5","subject":"Drop apache installation outside of httpd container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/610eef2089c186c6e698bb1781501ee8c696f7d5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-05-08 06:16:21.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-05-08 06:17:02.000000000","tz":120},"subject":"ironic-pxe: Remove httpd setup call","message":"ironic-pxe: Remove httpd setup call\n\nWe\u0027re not using ironic-pxe as ironic-http anymore and kolla_httpd_setup\nhas been removed from base.\n\nChange-Id: Id25e39cee9fd64d147301a39528aa8525bd682c2\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5df2cc1252265bacef4d1dee30355d1ff5464709"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5df2cc1252265bacef4d1dee30355d1ff5464709"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d1"],"failing_atoms":["label:Backport-Candidate\u003d-1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
