)]}'
{"id":"openstack%2Fkolla~987843","triplet_id":"openstack%2Fkolla~master~If4fbc24c5c339f7dee7bf67a09fc6448ede0c9fd","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"35345":{"account":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"last_update":"2026-05-13 14:32:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If4fbc24c5c339f7dee7bf67a09fc6448ede0c9fd","subject":"horizon: rebuild static assets after theme changes","status":"MERGED","created":"2026-05-08 13:19:58.000000000","updated":"2026-05-13 14:35:08.000000000","submitted":"2026-05-13 14:32:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"987843","meta_rev_id":"14d1222cc9cd3cf5f2519fd4207bc52d3f6d755f","_number":987843,"virtual_id_number":987843,"owner":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-13 14:32:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2026-05-13 12:22:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2026-05-13 12:48:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2026-05-13 12:22:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2026-05-13 12:48:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2026-05-13 12:22:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-08 15:33:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-13 12:22:57.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2026-05-13 12:48:41.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"e62dce7f8304fe51511ced01c1c2830d06d09040","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2026-05-08 13:19:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2d5e4bfd3fddc37e3e07c24c515888797f005d42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-08 15:33:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0454ad4456b740abb09d6e30d419e8f9\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/3c474b1ead1349e390f761d2975428fc : SUCCESS in 1h 06m 05s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/97f826667484478998e8d74353d8738f : SUCCESS in 1h 17m 51s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/59940aa552ec430fbe078f7abc66a73c : SUCCESS in 24m 09s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/a2ef001fd7644c8dad5ff901e1e65e9a : SUCCESS in 1h 13m 46s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/b2ed75c25e184292adfe1d26cc305541 : SUCCESS in 54m 05s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dc6f9416f2f742f290206751dbb00598 : SUCCESS in 1h 23m 04s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2fbc412521d0435999507649732734e5 : SUCCESS in 30m 37s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/2f9a5084d5dd46f3bcbeeb72d9605eef : SUCCESS in 1h 16m 48s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/10af1e48eafd489c9f569d09c6e0b54c : SUCCESS in 1h 21m 48s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/9ebd7233a4b24b67b50dc71ebe86c15a : SUCCESS in 1h 55m 50s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/e10254becaf2439093687d642fcbb4d1 : SUCCESS in 1h 31m 14s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/0666a0cc6b1a4925a97e1d2f0a6811dd : SUCCESS in 2h 06m 58s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/361bec8c374d48fb930a32887a7dc227 : SUCCESS in 1h 34m 15s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/fe2d1b000e8e424cbf6a2f54b3ac2419 : SUCCESS in 2h 11m 40s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/b83cdfc53c744b2aa08c56d08e7aa95c : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf76f1c0795e477584f9aba95815d65e : SUCCESS in 4m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c27f4ab22f1d486995676d2c0491fcc6 : SUCCESS in 5m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/94136bbfce4242d9afa723fa8a0d9249 : SUCCESS in 4m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/49ee5cc5ba3c40f5ae24049e8b044258 : SUCCESS in 5m 58s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f8c07fd5b3724312a55c04fed92027b2 : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef21535f6ece40cb84d4c585a6d7fb8e : SUCCESS in 7m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a8c214f161a44cacae55331edea6d3fc : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":1},{"id":"888007bb8b8edd2950f602819a4edcefe0b0f2ca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-08 20:43:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5283a788f0141d0a82be36153152d6b\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/29fe4e36627044e7970fdcfa6e9259f9 : SUCCESS in 1h 43m 27s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/f3a24f3496be41518ab40b496cd8261d : SUCCESS in 1h 39m 52s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/eb46f6a02d464f74a36e94a80f8c1657 : SUCCESS in 1h 41m 09s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/b02ef9677346499baac087b4b9b7c33b : SUCCESS in 1h 36m 17s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/5c9150a008864ec7a0eb838f6eadaa17 : FAILURE in 1h 35m 01s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/f265c0408cd546c4bc4b17550297ccae : SUCCESS in 1h 23m 36s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/26d907153b9f474fb4ce0380e017c7c3 : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6894bbc088d04fe9b19f13cff1f9e144 : SUCCESS in 4m 33s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d2f891d85dbf43f2b0d9304ac35abf18 : SUCCESS in 6m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"69d74d5cd2f9ade7fa438842ad7165ac420af950","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-13 12:22:57.000000000","message":"Patch Set 1: Backport-Candidate+1 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"15d0c22a810ce80b6cc468f7f7ca43c043146855","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 12:23:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dc938a1f1e7084e636d05dce18a38817d371b3b3","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-05-13 12:48:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"004717171ac379fd8c423cff1625e5d92d4bc432","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 14:32:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e0d7b2ac3dc4db497e9a3714ce22886\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/101c74219b9b4619be63a00650ea9384 : SUCCESS in 1h 26m 02s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/50180f25976c41d898f9df12a79575c1 : SUCCESS in 26m 21s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/9e3db091cd5846fbba3b9eb9a1459028 : SUCCESS in 1h 10m 44s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/460206bad5a643518a7b0285bd1a9d4c : SUCCESS in 1h 02m 31s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6b41ccff5e5048bbbfaa7ef76b7df8de : SUCCESS in 1h 15m 55s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2dcc8b586b294ca4bcaede937d29752c : SUCCESS in 31m 28s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/3a9a22f1c2834c0d8ee5ede5f005546f : SUCCESS in 1h 24m 16s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/ef2174db843e4d3789207063a01745eb : SUCCESS in 1h 48m 29s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/4a5ce6f3d4ba45949dc2c888bde124fe : SUCCESS in 1h 31m 54s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/403153b0145943da82342660e02bec2b : SUCCESS in 2h 08m 34s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3aaa827c223a4a95afbfa04837f45640 : SUCCESS in 1h 23m 57s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/6852d42c618c4d5da93f845d6af3284f : SUCCESS in 2h 01m 33s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/50c37b7b493e49cc8fe39655fa6709a1 : SUCCESS in 3m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e0d3f6f55c143afb75cc413beb35ccf : SUCCESS in 3m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c25f5a528042453a8cbdbf8a82e3d959 : SUCCESS in 4m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4550e155695748a1a62665b65378d3d6 : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da25e22402cf4451857bd6b2d41c64c2 : SUCCESS in 5m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fb2ed64a15b84d60a4bccd4d6c7c3b43 : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":1},{"id":"25001c97947f3f19fda9810fcf797c08fd9bca02","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 14:32:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"14d1222cc9cd3cf5f2519fd4207bc52d3f6d755f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 14:35:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/376c967e3b4c404aa896849a39d98e1d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35164f5a017644e6a904017d1a48a752 : SUCCESS in 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/efa0ec47650946e8a3820a1b5d92494c : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"21cacdd855bca15be846d392ec6795d667bf8ac1","revisions":{"21cacdd855bca15be846d392ec6795d667bf8ac1":{"kind":"REWORK","_number":1,"created":"2026-05-08 13:19:58.000000000","uploader":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"ref":"refs/changes/43/987843/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/43/987843/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/987843/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/987843/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/987843/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/43/987843/1"}}},"commit":{"parents":[{"commit":"610eef2089c186c6e698bb1781501ee8c696f7d5","subject":"Drop apache installation outside of httpd container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/610eef2089c186c6e698bb1781501ee8c696f7d5"}]}],"author":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2026-05-08 13:19:12.000000000","tz":120},"committer":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2026-05-08 13:19:57.000000000","tz":120},"subject":"horizon: rebuild static assets after theme changes","message":"horizon: rebuild static assets after theme changes\n\nInclude the theme directory in settings_bundle so theme-only changes\ninvalidate the cache and rebuild the static assets.\n\nCloses-Bug: #2148389\n\nChange-Id: If4fbc24c5c339f7dee7bf67a09fc6448ede0c9fd\nSigned-off-by: Grzegorz Koper \u003cgrzegorzk@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/21cacdd855bca15be846d392ec6795d667bf8ac1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/21cacdd855bca15be846d392ec6795d667bf8ac1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d1"],"failing_atoms":["label:Backport-Candidate\u003d-1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
