)]}'
{"id":"openstack%2Fkolla~987862","triplet_id":"openstack%2Fkolla~stable%2F2026.1~Ie28d1cc615cc09f013438285b4dcbc2dbab0f4b5","project":"openstack/kolla","branch":"stable/2026.1","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-05-08 16:28:39.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}},"hashtags":[],"change_id":"Ie28d1cc615cc09f013438285b4dcbc2dbab0f4b5","subject":"Remove influxdb from repos.yaml","status":"MERGED","created":"2026-05-08 14:55:15.000000000","updated":"2026-05-08 21:19:04.000000000","submitted":"2026-05-08 19:59:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"987862","cherry_pick_of_change":987794,"cherry_pick_of_patch_set":3,"meta_rev_id":"d7cba5286e271113cfb38192444202bb0c92d5c4","_number":987862,"virtual_id_number":987862,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-08 19:59:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2026-05-08 16:28:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2026-05-08 15:35:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2026-05-08 16:28:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-08 15:35:18.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-05-08 17:54:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"040c7720f06a95ae49c03fa6029073c5b6099b0d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-08 14:55:15.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"02437aed31f1eb0856135f2255167741fefbe57c","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-05-08 15:35:18.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f50d568d65d39f9f2e54f2b2a2f9dd4a2d223ae0","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-08 16:28:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3390ef611e6b4199753a62bedf56c90db61bba86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-08 17:54:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b13df99df1943c7880ed3987d2fb158\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/d1e51aa6c11e4f68af87657d8dd5c5a0 : SUCCESS in 1h 21m 44s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/5b219323887e4a979bc86d1319168ac0 : SUCCESS in 1h 36m 17s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/4904543e57134ba6aa74cad07da25fd5 : SUCCESS in 23m 49s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/77127e31cbf549409916f83d81d544b6 : SUCCESS in 1h 03m 17s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/31da0aaefe6348b8857ea09bcd20be3d : SUCCESS in 1h 12m 02s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/781e9a3fa081439491955fda7255878d : SUCCESS in 1h 22m 33s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0171adfc12f34d978213d60784ef668b : SUCCESS in 27m 02s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/1e64761780ac40bda42db587bf9ccc26 : SUCCESS in 1h 28m 45s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/61062db2376d4d798071c524ee8372c5 : SUCCESS in 1h 29m 16s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/69cd39acbf654763896d46ae9f69baab : SUCCESS in 1h 58m 02s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/f6be22f4dac9445f8d68ab3db1649f9e : SUCCESS in 1h 39m 16s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/5f7400435d914286aa4981b324d72036 : SUCCESS in 2h 10m 55s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7a8b63be89e647a984853e72ec375f88 : SUCCESS in 1h 34m 39s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/266834edae064fa0b18a74390dc2b6b1 : SUCCESS in 2h 07m 43s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/6f6a73595db540f6b84a9e44bee73d7b : SUCCESS in 2h 05m 22s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/fcaeee2f5a6f4c47a504ca662723cdc7 : SUCCESS in 2h 13m 28s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/adc7e87700e043b3824d1b836a7277c3 : SUCCESS in 2h 58m 01s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/9c2e624ad36d4955a8470a12a2fe4c35 : SUCCESS in 2h 02m 52s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/fd5a572c5ada4466bce0181e09a62e3e : FAILURE in 2h 51m 57s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/0470d868d5f340b987f9b7058b1785fb : FAILURE in 56m 54s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/4c608339bf82419b85045bb328f74aaf : FAILURE in 1h 05m 08s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/dff2dd377e9a4510890fac29d51d2386 : FAILURE in 1h 09m 17s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/e9a0b440e3ed43bda73667767ea0d3dc : SUCCESS in 28m 48s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/ccca59dce8394d988bb0b7135b825378 : SUCCESS in 31m 10s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/06ea94ac467049e0932026016b0de6ac : SUCCESS in 23m 12s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/d441f4c9981649b0824e76f382d20ff2 : FAILURE in 1h 21m 17s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/51ca8ed869344c519e8498495c5ebdbe : SUCCESS in 1h 02m 48s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/9fbd46df675f43708f11ed52cc26dc94 : SUCCESS in 1h 41m 14s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/8b2c911c782e4cf89307cbb1f2f22307 : SUCCESS in 1h 25m 59s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/6eac250609544504b0cf1f35d6e923be : SUCCESS in 1h 26m 41s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/57d342e62aa6400ea08b5bcd04f53842 : FAILURE in 1h 44m 01s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/e42b1ac4b3a646019271cd2abe442116 : SUCCESS in 2h 18m 25s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/b9b83bfe97984232a1896c7abf94d849 : SUCCESS in 2h 19m 41s (non-voting)\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/78a26e9b89274704989d73f37a7906d0 : SUCCESS in 1h 57m 32s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/f1515758360846caaa856662631d77b0 : SUCCESS in 1h 31m 19s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/fd6613048a464ec08a0b27a0b3b19174 : SUCCESS in 1h 42m 22s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/c18ce9dad74847faa8938eca744e6b4e : SUCCESS in 2m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a61ba3ee1a24004bc843480615bccd5 : SUCCESS in 4m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3b13e1f315c24afc805988485230fe16 : SUCCESS in 3m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5c67acc4aa5d4aa8924aecad89a9f14c : SUCCESS in 3m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/afc7aeeb290740ea91adea0142398669 : SUCCESS in 3m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13ff6113451c4e50b1669a0ce6086f9e : SUCCESS in 6m 36s","accounts_in_message":[],"_revision_number":1},{"id":"ae1a9e1d6de542b412f23584df6e71d8b42b3b2d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-08 17:55:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0f2d9f242aefd1227fb0b7948e7082460ba9f3ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-08 19:59:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d12aa5ff90b4655bc03de1d5ee718e5\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/8281a26fe61b4423bee70b079c710337 : SUCCESS in 54m 36s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/11b315012c454bbd9c84a240a4786147 : SUCCESS in 22m 51s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/6b17e3ca522844bb85392293f7e0e277 : SUCCESS in 1h 03m 40s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/dc6b5b2bda584e2ab1a5b2b180e2ec69 : SUCCESS in 1h 06m 18s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0df35ca7f10144c4a14ca98870f4f3ff : SUCCESS in 1h 04m 51s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d7524fc49c6d4e0bbddc662fe81fa95f : SUCCESS in 14m 06s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/3d9baa1201914f9ba20218d9ca7d111e : SUCCESS in 1h 23m 51s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/4bb75d90e0a841eca5a082a5013ced68 : SUCCESS in 1h 53m 12s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/914da7aafb86421ca2c46544792c69be : SUCCESS in 1h 32m 33s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/e9b5c3ee9b664dcc9602fa5e6b4c08cf : SUCCESS in 2h 04m 13s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9bd6504cc87d4ff993135c6d150d5611 : SUCCESS in 1h 25m 47s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/1e194ecdd9e64f559b01d2cc2acdc328 : SUCCESS in 2h 02m 25s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/d4bcc2b9956c4d98b3c39a9801804644 : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2945fd7f0904beb87850ec76c543e19 : SUCCESS in 3m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/45d7060f164f46a8bd635a6c70fe3346 : SUCCESS in 2m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ae55fae380e5485186553bc15339e261 : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0abe737cc8da43d29b221c4552fe05ee : SUCCESS in 6m 32s","accounts_in_message":[],"_revision_number":1},{"id":"3fcc8baf608fc9ae446e9cb23c079d512733564b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-08 19:59:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"24f6dd3ff3c9b38045c64f7a16f7e6d76d4e45b0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-08 20:00:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/318b31be448b4697af0c084da6d28693\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90382fd4c637484baa11c6fad66ee112 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1},{"id":"d7cba5286e271113cfb38192444202bb0c92d5c4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-08 21:19:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/679c0775aa6e40a0b826448bd9ab8990\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/b8c60c4754024f26836165daa18b3e35 : SUCCESS in 1h 32m 21s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/7fe574b7a07044c1a8f4cb2ee97aa17a : SUCCESS in 1h 27m 54s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/6e3c299146f24f04b35836799b1d0f2f : SUCCESS in 1h 26m 04s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/1ff08ce460de4a0aa8a8dab8569daffc : SUCCESS in 1h 33m 40s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/20938a74161341708c902bb81d94d33c : FAILURE in 1h 29m 58s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/1950bba083dd4a6ab278f4ba2615cb90 : SUCCESS in 1h 39m 30s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"64821bed72690ea79af7f9df9eb8fd8431a6e30f","revisions":{"64821bed72690ea79af7f9df9eb8fd8431a6e30f":{"kind":"REWORK","_number":1,"created":"2026-05-08 14:55:15.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/62/987862/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/62/987862/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/987862/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/987862/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/987862/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/62/987862/1"}}},"commit":{"parents":[{"commit":"5a6c6c3cfa7bafcbf402dd0d542ad15c2f1cf904","subject":"Update TOX_CONSTRAINTS_FILE for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5a6c6c3cfa7bafcbf402dd0d542ad15c2f1cf904"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-05-08 05:52:59.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-05-08 14:55:15.000000000","tz":0},"subject":"Remove influxdb from repos.yaml","message":"Remove influxdb from repos.yaml\n\nIt was ommitted in the original influxdb removal patch [1].\n\n[1]: https://review.opendev.org/c/openstack/kolla/+/973051\n\nChange-Id: Ie28d1cc615cc09f013438285b4dcbc2dbab0f4b5\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit 83b8161b12bb1a32b8b872c46533cfa6a2fe9d7f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/64821bed72690ea79af7f9df9eb8fd8431a6e30f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/64821bed72690ea79af7f9df9eb8fd8431a6e30f"}]},"branch":"refs/heads/stable/2026.1","conflicts":{"ours":"5a6c6c3cfa7bafcbf402dd0d542ad15c2f1cf904","theirs":"83b8161b12bb1a32b8b872c46533cfa6a2fe9d7f","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
