)]}'
{"id":"openstack%2Fkolla~988347","triplet_id":"openstack%2Fkolla~master~I3db224fbf7ffaf153fb6307d506fdad980bb429d","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-05-13 09:04:13.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-05-13 09:39:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3db224fbf7ffaf153fb6307d506fdad980bb429d","subject":"Ignore D001 Line too long errors in release notes","status":"MERGED","created":"2026-05-12 19:59:46.000000000","updated":"2026-05-13 10:03:08.000000000","submitted":"2026-05-13 09:39:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"988347","meta_rev_id":"a4895fa771bc872904f5b8fe75a4d100ad9928ae","_number":988347,"virtual_id_number":988347,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-13 09:39:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2026-05-13 09:04:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2026-05-13 09:04:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2026-05-13 09:04:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"reviewers":{"REVIEWER":[{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-12 20:10:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-12 20:54:03.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2026-05-12 20:54:03.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"state":"REVIEWER"},{"updated":"2026-05-12 20:54:03.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"state":"REVIEWER"},{"updated":"2026-05-12 20:54:03.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2026-05-12 20:54:03.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2026-05-12 20:54:03.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"}],"messages":[{"id":"6a08a9b039344c8ecd616d1e15128fce66c060e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-05-12 19:59:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"288f48a896fd1f38c484dd6138983790df8cf03b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 20:10:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/149bf68109f94ac4b5d4f344542bc19e\n\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/fec321ff741f47518ef150a3ff6d0eed : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a55ee83235045e88bec0f3925335657 : SUCCESS in 3m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12e0b3d5445c4aa78c32e4b6a9f9752c : SUCCESS in 5m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c2323b7e443848909104cfaa29ec63fc : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":1},{"id":"ee3eed4aed09ce8859f25b54ba28af433f530fa1","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-12 20:56:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"796231012d27a11ced54f9b85d9bf30e72fff3a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-05-12 21:48:47.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"2d3bdc1afb9a14d6dec9cab47bd36d862253ecc1","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-05-12 21:48:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"142da625b9b108bc231900a624203fdbb5d870b1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 21:57:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/763e170fbc7d4f279bf99d8cb8fecd41\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/31031601f5744e7ab9181d4b86d4824c : SUCCESS in 4m 03s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8ac034eb410649ec96079448ba8bd5f6 : SUCCESS in 3m 48s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f7cc1edd21fa4f1b925cefbbd39ff59a : SUCCESS in 6m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"97aef4d3c681b24bbef528c209bae4e75795efc4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 22:02:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8128a5a4c52346d59f3bc156dceb1c34\n\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/bd70bacfd07442c5946a3c1e536b0ec1 : SUCCESS in 3m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60b62a89429541e4a4b31048f18b9cb5 : SUCCESS in 4m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ac86c13813a242d4a9f2b1505484d60e : SUCCESS in 2m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/657861bd5ac7492ba820259dc8581e75 : SUCCESS in 2m 55s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7833c0300ec14cd6a04e277f0bc6f54b : SUCCESS in 6m 55s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1bd51cac6e7d46b8a649cea0a3d3cd62 : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00a506be2ad948eb862aaf2429979134 : SUCCESS in 7m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c5752dbd986b42488bcfdf70fa440afd : SUCCESS in 7m 33s","accounts_in_message":[],"_revision_number":2},{"id":"7adf5abd98ae9ebdd63f6fa58ceba58b5b770d26","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-13 08:07:24.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a186ca5a61d725b4f9057581877199684245dd6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-05-13 08:25:55.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"fc7835a406318229d4bd6401b06afa7240952bc6","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-05-13 08:27:08.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f2fc4b005b0393f1d592b7658f983db4e20d2ce5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 08:43:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bdd5141f745b46dc926ba3ae5106a29a\n\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/d55ac9a80f5d48598048d054d532b059 : SUCCESS in 3m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/383d76038f5e4a08b306528b7a799ab4 : SUCCESS in 3m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2229bffd02c344bd843311d9e1f853d9 : SUCCESS in 4m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0c364946e24d4b9c8f1c7f7a74f31f77 : SUCCESS in 3m 12s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/027ddc449a3b4002aa9fa0357417b863 : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2f1c126731164cfbb11904f1dd716084 : SUCCESS in 2m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/969c65d6f8cb4d298efcafb0af325313 : SUCCESS in 4m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f0536ed4744c405db3c6e2d686a0fcae : SUCCESS in 10m 05s","accounts_in_message":[],"_revision_number":3},{"id":"4a6328106bc85bd389213c1f829b8b5c898b5c1d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 08:56:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/449adbf9080e461791f95630cf16a774\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3a6332fd60364ff69e2b2dca9fd52302 : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7a710a6088094aafb10086f1b983fd89 : SUCCESS in 7m 36s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/528e0820425b4502b87b915ff35774de : SUCCESS in 8m 32s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"af8f658819e21be6fc7f31802c67a12532d13ae9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-13 09:03:13.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"c6457bc7bce939518245ee1348cd938ce1c1f27c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-13 09:04:13.000000000","message":"Patch Set 4: Backport-Candidate+1 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"2e1d7ff00e79b9e0f61a7e7389b98a5a56f38b87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 09:24:20.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb36e9440394414bad5850ea5bcbccf0\n\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/45c420d686444e73a37b7ba7c375a4c7 : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80c9e65408ec4a55b39acb9b288bca98 : SUCCESS in 5m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8cb12519919a40d0bc1a5176e4b04aea : SUCCESS in 2m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/51427e4d9712408fa2f68412a603ebe1 : SUCCESS in 4m 09s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/460634561bcb41d8b888c43afe3ad561 : SUCCESS in 8m 21s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c6c4ea94f8cc4a1e9b19d64a0d1e2df6 : SUCCESS in 3m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53b3a646907846918d53913f21f867ed : SUCCESS in 6m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5788c0f3a70e4105b3e218c6762d8595 : SUCCESS in 9m 13s","accounts_in_message":[],"_revision_number":4},{"id":"5936a5a96583acfe52487561d9aaafba41b4903e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 09:25:36.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"f7ab9841ef18464d754a1bfd09b075915aeb21b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 09:39:14.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a755f86cf3a4d9ab29bd8778ce5de8a\n\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/264c8320e21a4ebeb415913d02ba19bf : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32bee74d6fd3422c98d8ffef3bb7e47c : SUCCESS in 5m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a57bad4120a442ed83eed29b62e446e8 : SUCCESS in 3m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/35ef04d5f1584f5486cfc0628b32f0f4 : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd4d187785e747c3a72e02ae4363c86a : SUCCESS in 6m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9ad0d7d36cf143c6b794e5f7b4fc9ad7 : SUCCESS in 8m 19s","accounts_in_message":[],"_revision_number":4},{"id":"59e9fb71258ec9e574af01bdb90220a882ebd7f5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 09:39:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"a48ea8fd97843a77d24c76f3c8c20740e05e375a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 09:40:44.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9408c7b0b73e41ba9bce4d1873d608c7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87ab834be5e1448dae17f5901412c1d8 : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/21a312e9b56f4bf7b86ccfd7c3bb2ca5 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":4},{"id":"a4895fa771bc872904f5b8fe75a4d100ad9928ae","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 10:03:08.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/931a0573228d4b35b3701ce8f3d1f367\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/da5e2f9774e8421989d61363597f23ce : SUCCESS in 3m 39s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/268d62f25398439b9fb7e273383b7895 : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b3525bd2c82642c7b94df72af1b8a9e4 : SUCCESS in 6m 23s (non-voting)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"5c3d61abbbc3b5c0c484f5054def1ba294145eed","revisions":{"bd237df0b0c968377e0af3eb91de37f949840f4e":{"kind":"REWORK","_number":1,"created":"2026-05-12 19:59:46.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/47/988347/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/47/988347/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/988347/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/988347/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/988347/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/47/988347/1"}}},"commit":{"parents":[{"commit":"1ff4715aa6d7e2ecb84ec7a608b87b3ba95f4ea6","subject":"Merge \"ironic-pxe: Remove httpd setup call\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1ff4715aa6d7e2ecb84ec7a608b87b3ba95f4ea6"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-05-12 19:57:34.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-05-12 19:57:34.000000000","tz":120},"subject":"Fix D001 Line too long errors in release notes","message":"Fix D001 Line too long errors in release notes\n\nThe release notes job was changed [1], leading to build failures.\nResolve by fixing all occurences of `D001 Line too long` errors.\n\n[1] https://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/thread/RLOKQLQUDAVSML4C4OW2EM4OXOGRCYBA/\n\nChange-Id: I3db224fbf7ffaf153fb6307d506fdad980bb429d\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bd237df0b0c968377e0af3eb91de37f949840f4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bd237df0b0c968377e0af3eb91de37f949840f4e"}]},"branch":"refs/heads/master"},"66d63fb52daadb4aa087dd0b9e4336381ca9b809":{"kind":"REWORK","_number":2,"created":"2026-05-12 21:48:47.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/47/988347/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/47/988347/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/988347/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/988347/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/988347/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/47/988347/2"}}},"commit":{"parents":[{"commit":"1ff4715aa6d7e2ecb84ec7a608b87b3ba95f4ea6","subject":"Merge \"ironic-pxe: Remove httpd setup call\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1ff4715aa6d7e2ecb84ec7a608b87b3ba95f4ea6"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-05-12 21:47:36.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-05-12 21:47:36.000000000","tz":120},"subject":"Ignore D001 Line too long errors in release notes","message":"Ignore D001 Line too long errors in release notes\n\nThe release notes job was changed [1], leading to build failures.\nResolve by ignoring occurences of `D001 Line too long` errors.\n\n[1] https://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/thread/RLOKQLQUDAVSML4C4OW2EM4OXOGRCYBA/\n\nChange-Id: I3db224fbf7ffaf153fb6307d506fdad980bb429d\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/66d63fb52daadb4aa087dd0b9e4336381ca9b809"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/66d63fb52daadb4aa087dd0b9e4336381ca9b809"}]},"branch":"refs/heads/master"},"bc2a7ae0b790df435882adafe114424cd1cb81d2":{"kind":"REWORK","_number":3,"created":"2026-05-13 08:25:55.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/47/988347/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/47/988347/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/988347/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/988347/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/988347/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/47/988347/3"}}},"commit":{"parents":[{"commit":"1ff4715aa6d7e2ecb84ec7a608b87b3ba95f4ea6","subject":"Merge \"ironic-pxe: Remove httpd setup call\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1ff4715aa6d7e2ecb84ec7a608b87b3ba95f4ea6"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-05-13 08:25:12.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-05-13 08:25:12.000000000","tz":120},"subject":"Ignore D001 Line too long errors in release notes","message":"Ignore D001 Line too long errors in release notes\n\nThe release notes job was changed [1], leading to build failures.\nResolve by ignoring errors in existing files.\n\n[1] https://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/thread/RLOKQLQUDAVSML4C4OW2EM4OXOGRCYBA/\n\nChange-Id: I3db224fbf7ffaf153fb6307d506fdad980bb429d\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bc2a7ae0b790df435882adafe114424cd1cb81d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bc2a7ae0b790df435882adafe114424cd1cb81d2"}]},"branch":"refs/heads/master"},"5c3d61abbbc3b5c0c484f5054def1ba294145eed":{"kind":"REWORK","_number":4,"created":"2026-05-13 09:03:13.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/47/988347/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/47/988347/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/988347/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/988347/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/988347/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/47/988347/4"}}},"commit":{"parents":[{"commit":"1ff4715aa6d7e2ecb84ec7a608b87b3ba95f4ea6","subject":"Merge \"ironic-pxe: Remove httpd setup call\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1ff4715aa6d7e2ecb84ec7a608b87b3ba95f4ea6"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-05-13 08:25:12.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-05-13 09:03:10.000000000","tz":120},"subject":"Ignore D001 Line too long errors in release notes","message":"Ignore D001 Line too long errors in release notes\n\nThe release notes job was changed [1], leading to build failures.\nResolve by ignoring errors in existing files.\n\n[1] https://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/thread/RLOKQLQUDAVSML4C4OW2EM4OXOGRCYBA/\n\nChange-Id: I3db224fbf7ffaf153fb6307d506fdad980bb429d\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5c3d61abbbc3b5c0c484f5054def1ba294145eed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5c3d61abbbc3b5c0c484f5054def1ba294145eed"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d1"],"failing_atoms":["label:Backport-Candidate\u003d-1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
