)]}'
{"id":"openstack%2Fkolla~988573","triplet_id":"openstack%2Fkolla~stable%2F2025.1~If4fbc24c5c339f7dee7bf67a09fc6448ede0c9fd","project":"openstack/kolla","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"35345":{"account":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"last_update":"2026-05-22 08:09:32.000000000","reason":"Change was submitted"},"22540":{"account":{"_account_id":22540,"name":"Harry Kominos","display_name":"hkominos","email":"hkominos@gmail.com","username":"hkominos"},"last_update":"2026-05-22 08:09:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If4fbc24c5c339f7dee7bf67a09fc6448ede0c9fd","subject":"horizon: rebuild static assets after theme changes","status":"MERGED","created":"2026-05-14 07:53:57.000000000","updated":"2026-05-22 08:11:04.000000000","submitted":"2026-05-22 08:09:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"988573","cherry_pick_of_change":987843,"cherry_pick_of_patch_set":1,"meta_rev_id":"fbe0657fddb33d3fcae81e0bd1cf9c4d5e134bb6","_number":988573,"virtual_id_number":988573,"owner":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22540,"name":"Harry Kominos","display_name":"hkominos","email":"hkominos@gmail.com","username":"hkominos"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-22 08:09:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2026-05-22 05:32:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2026-05-15 22:18:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2026-05-14 08:58:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22540,"name":"Harry Kominos","display_name":"hkominos","email":"hkominos@gmail.com","username":"hkominos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2026-05-22 05:32:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22540,"name":"Harry Kominos","display_name":"hkominos","email":"hkominos@gmail.com","username":"hkominos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22540,"name":"Harry Kominos","display_name":"hkominos","email":"hkominos@gmail.com","username":"hkominos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22540,"name":"Harry Kominos","display_name":"hkominos","email":"hkominos@gmail.com","username":"hkominos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22540,"name":"Harry Kominos","display_name":"hkominos","email":"hkominos@gmail.com","username":"hkominos"},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-14 08:58:27.000000000","updated_by":{"_account_id":22540,"name":"Harry Kominos","display_name":"hkominos","email":"hkominos@gmail.com","username":"hkominos"},"reviewer":{"_account_id":22540,"name":"Harry Kominos","display_name":"hkominos","email":"hkominos@gmail.com","username":"hkominos"},"state":"REVIEWER"},{"updated":"2026-05-14 09:53:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-15 22:18:12.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-05-22 05:32:45.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"d50eb88ffe8e4cce3ca99ffa3aa9769fedcab155","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2026-05-14 07:53:57.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"08a5e009a30f764039715965ec1442d6c27f2580","author":{"_account_id":22540,"name":"Harry Kominos","display_name":"hkominos","email":"hkominos@gmail.com","username":"hkominos"},"date":"2026-05-14 08:58:27.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"936c682801744415ee4a1566472e3ec7b3647e83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 09:53:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c7ebfcbb6ab4c5a821595385e15b4e4\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/3a459ae049f64a92ae118f180f45e83f : SUCCESS in 50m 51s (non-voting)\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/60e809fb9bc041128994ea1bb1155fc3 : FAILURE in 1h 38m 46s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/ee9bffa595c645209976d068f5853278 : SUCCESS in 1h 28m 43s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/ced80d4454a44e06a99e73125b7b53bd : SUCCESS in 1h 36m 10s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/1e829a16c8a241a2bef0adad5485e4d1 : SUCCESS in 1h 05m 08s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/492fa908dc054c0a9a5f48d741ab7dbd : SUCCESS in 34m 27s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/b8d4a8f2aa4045be96471a1a77e79e50 : SUCCESS in 57m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d000a63dd06c477084bad7648f7b2db0 : SUCCESS in 5m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aee24ed92b884e3088b0b938caac765b : SUCCESS in 3m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/490df02965574bf49d108ea516db4090 : SUCCESS in 3m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db87dfb236fc44bbac1e84ce9acd3ec7 : SUCCESS in 2m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8107b93459244618a5775b990d4b1f88 : SUCCESS in 6m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1cba7794180a4d36b08ae6bff7566148 : SUCCESS in 6m 58s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/ac154a45bc0b45d5ba2040c586006410 : SUCCESS in 2m 58s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/ae25523ba56447b98148f6829515ba98 : SUCCESS in 1h 00m 29s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/2ba5e65fe4cc4bfe9616121d094578c5 : SUCCESS in 1h 28m 27s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/391c7ad715aa46f1aded6cfe8e247115 : SUCCESS in 1h 26m 43s\n- kolla-build-rocky10 https://zuul.opendev.org/t/openstack/build/b876bfa3995f481c92b8098031a30a09 : SUCCESS in 1h 11m 48s\n- kolla-build-rocky10-podman https://zuul.opendev.org/t/openstack/build/b4f372b6775a4688ab369f3100f70e20 : SUCCESS in 1h 19m 01s\n- kolla-ansible-rocky10 https://zuul.opendev.org/t/openstack/build/79ea6eb097bf4e14941fab90508d42c1 : SUCCESS in 1h 19m 43s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/048e3b7cc24c47ed99e7ded8f2f35777 : SUCCESS in 1h 04m 27s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/5f55574bca2744c4a51a573ff6e991b2 : SUCCESS in 29m 28s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/6ba73b14ca92474f96d4910c54d9a5c8 : SUCCESS in 1h 03m 10s\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/9a5d80f6d68f40f2822bf1f1e856062a : FAILURE in 1h 24m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4fc4d8e0b16f55eb613cc80b9beaebf875c4dd40","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2026-05-14 10:32:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2770eb37d5efc730adea98ce98297c0f966239d4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 18:48:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f509daf567ac45fbbc67de225345fb4e\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/0262461952694050b97ad4edd99e26dd : SUCCESS in 58m 41s (non-voting)\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/668e460832ce44e4971de86492d4a5ed : SUCCESS in 2h 21m 28s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4d03bf3addef4d6f818cfa7e45a4d195 : SUCCESS in 55m 09s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/307faccde523495593a8e2e33890eed7 : SUCCESS in 1h 42m 33s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/b372ac6dbd554f51abb5e26945f939f6 : SUCCESS in 1h 39m 08s (non-voting)\n- kolla-build-rocky10-aarch64 https://zuul.opendev.org/t/openstack/build/cdfe03304239452f8548533d4078f481 : SUCCESS in 2h 32m 55s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/934bafd2dd3241b0a602dbe4fa48112d : SUCCESS in 1h 25m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"235628223c7baff462dd5773bbcac194a1650342","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-05-15 22:18:12.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"4c4fa132fe610080fbeeeff667a53fdac46ad8bd","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-22 05:32:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"41e82be9c2f6a721f543a9d9310069b60b8f1cd9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-22 05:33:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0133179502a84b02888692b4ec8a361e2c7b85d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-22 08:09:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f296d74a718b46cc99c26e49ba0ad45c\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/68ec1d97c5904264b58bd1d2db584f2b : SUCCESS in 37m 02s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/1947c3dff2cb4a48bea818c7985af5e0 : SUCCESS in 34m 48s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/78687363ea1c4ec389680251858f278d : SUCCESS in 58m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36634845ba7f4cf3a8382e14b568365a : SUCCESS in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f032fb0a4f8e401b8bc0de5953ea1de4 : SUCCESS in 6m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/179dea6a9f7047e39a7aeb8fe80692c7 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c9c4f05eeb74b6091ed0bb4dfdd6f8e : SUCCESS in 6m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/34a7b26d3ef8443ab4e332fcb71d460b : SUCCESS in 4m 31s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/24d2a69144d74bc0adc52d2aefd057c0 : SUCCESS in 2m 30s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/e2369ebe02af4b1bbf88cf956e0747df : SUCCESS in 56m 33s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/cc1c122a130d4ae09fa8dd7f27bf048e : SUCCESS in 54m 36s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/f52ec2da89c8420fbac7f7675298fa7c : SUCCESS in 1h 24m 16s\n- kolla-build-rocky10 https://zuul.opendev.org/t/openstack/build/e003c6525b4b47f6a6e6a2161c1db1bc : SUCCESS in 44m 07s\n- kolla-build-rocky10-podman https://zuul.opendev.org/t/openstack/build/60e3454128ed45cf98a6258db83f611d : SUCCESS in 42m 31s\n- kolla-ansible-rocky10 https://zuul.opendev.org/t/openstack/build/937302fde9744d0db03b3f47805acb57 : SUCCESS in 1h 11m 40s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/9cd5313d3ba246fb99117f448caaceae : SUCCESS in 33m 25s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/aca353ea3b674666b090076ec4f29e70 : SUCCESS in 30m 42s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/182ee37f4a8c421e9064417a307af05b : SUCCESS in 1h 06m 14s","accounts_in_message":[],"_revision_number":1},{"id":"834a421406b7e7f3df5d28eca746097b05e71e65","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-22 08:09:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"fbe0657fddb33d3fcae81e0bd1cf9c4d5e134bb6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-22 08:11:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64708add98ec422aa52669c2c9f5d47d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1dbf8b66f3a3420181efeda95e56f2cc : SUCCESS in 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d74dada665684e2a881c1ca2fccb59e9 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"771cb5c8ed0b28201abc7e016f05ba64a84f3763","revisions":{"771cb5c8ed0b28201abc7e016f05ba64a84f3763":{"kind":"REWORK","_number":1,"created":"2026-05-14 07:53:57.000000000","uploader":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"ref":"refs/changes/73/988573/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/73/988573/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/988573/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/988573/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/988573/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/73/988573/1"}}},"commit":{"parents":[{"commit":"7a321268da289844fcfacb6662f43a0c15452877","subject":"Ignore D001 Line too long errors in release notes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7a321268da289844fcfacb6662f43a0c15452877"}]}],"author":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2026-05-08 13:19:12.000000000","tz":120},"committer":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2026-05-14 07:53:57.000000000","tz":0},"subject":"horizon: rebuild static assets after theme changes","message":"horizon: rebuild static assets after theme changes\n\nInclude the theme directory in settings_bundle so theme-only changes\ninvalidate the cache and rebuild the static assets.\n\nCloses-Bug: #2148389\n\nChange-Id: If4fbc24c5c339f7dee7bf67a09fc6448ede0c9fd\nSigned-off-by: Grzegorz Koper \u003cgrzegorzk@stackhpc.com\u003e\n(cherry picked from commit 21cacdd855bca15be846d392ec6795d667bf8ac1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/771cb5c8ed0b28201abc7e016f05ba64a84f3763"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/771cb5c8ed0b28201abc7e016f05ba64a84f3763"}]},"branch":"refs/heads/stable/2025.1","conflicts":{"ours":"7a321268da289844fcfacb6662f43a0c15452877","theirs":"21cacdd855bca15be846d392ec6795d667bf8ac1","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
