)]}'
{"id":"openstack%2Fkolla~988845","triplet_id":"openstack%2Fkolla~master~I9e2ba00671eb7d05de3c40ed9524f20922e081b1","project":"openstack/kolla","branch":"master","topic":"remove-networking-infoblox","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-05-27 05:17:04.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-05-27 09:59:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9e2ba00671eb7d05de3c40ed9524f20922e081b1","subject":"Remove support for networking-infoblox","status":"MERGED","created":"2026-05-16 14:58:48.000000000","updated":"2026-05-28 10:43:46.000000000","submitted":"2026-05-27 09:59:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"988845-remove-networking-infoblox","meta_rev_id":"c114a85d687ffed0d39d60ebdcfdf0df90142583","_number":988845,"virtual_id_number":988845,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-27 09:59:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2026-05-22 09:09:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2026-05-22 11:02:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2026-05-24 10:28:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":-1,"date":"2026-05-23 05:32:19.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2026-05-19 15:51:19.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-16 17:21:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-19 06:56:20.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2026-05-19 15:51:19.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2026-05-22 11:02:46.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"a6beb086aabee9f738e5e759c8e35e32eb879df8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-16 14:58:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ef3a732094c4f9aeea993f9023984efe49a44c89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-16 17:21:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c2adf690f3d6475daf7d8cda1f55a33d\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/950f64b599984fbcbf91f77326c15815 : SUCCESS in 1h 05m 27s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/0900c185f8c3483e891bac26ca4223a4 : FAILURE in 20m 51s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/d2800af6271d4d068bbf4387276e4e2c : FAILURE in 7m 28s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/0c59180fc300488b873595543368a3d1 : SUCCESS in 1h 15m 20s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/9d062ff03aa544ed9926b2570b88ee5e : SUCCESS in 1h 00m 09s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e0f29fee00a54be2bbaf8dfb3cb36994 : SUCCESS in 1h 22m 39s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/153cddb1f5b340df97cfb6d8ce8afa3f : SUCCESS in 13m 11s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/abad4ffb29a943a999bff42b0a45d452 : SUCCESS in 1h 35m 45s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/a99c9b594eda4aa1a30c4f34ebb645f9 : FAILURE in 17m 54s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/6d28b22d5088498d957fd9c577a09c88 : SUCCESS in 1h 31m 13s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/f98b3b282a19419b809905c33b678b9a : SUCCESS in 2h 13m 13s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/89cf68cbf12d43ae8ee5998ebc08f6ac : SUCCESS in 1h 31m 19s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/c9b5c6d441fd49b3959d004178547179 : FAILURE in 44m 07s\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/b22a93b48b3d4b6ead7564c7cd2e7529 : FAILURE in 15m 13s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/7e945707ddf6407f8cb9283a1be4230c : SUCCESS in 1h 34m 21s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/1223896f5ffe43d8811928fb28cc6606 : FAILURE in 24m 54s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/88ff0b85f69044c39f918adcc53c30f2 : SUCCESS in 2h 21m 03s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/d1dcd454f82140b3995ef068adcdb841 : FAILURE in 1h 40m 29s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/335c06ab1e394a0fb4044a9763688318 : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c95d3d444214599ae0050c1ee5cb910 : SUCCESS in 5m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ab9d119c7e474b00b61d801ccb51d233 : SUCCESS in 2m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ffeaa14dd9344b9591de29e23c706fdf : SUCCESS in 3m 16s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/99e4b34511aa41adb0a77d2b2558389f : SUCCESS in 4m 12s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d0c6edd2ca004c579e202fefc064f886 : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4695dc61d08d416b8208aec09e3d9cf0 : SUCCESS in 6m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/633f43a20aab455cbf80d51d116b1eb7 : SUCCESS in 7m 16s","accounts_in_message":[],"_revision_number":1},{"id":"d6f9f62f793bb683d6456bada3f8303ddd1b66eb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-16 23:55:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8e9e18d92b94049a25d1567a6aed271\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/32373de533aa49cebd4afbbc81be3c37 : SUCCESS in 2h 26m 22s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/a87f888fdc554afc92098d5cadf4c34a : FAILURE in 24m 14s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/9c261b0f51d64e0eb99f874a3d304389 : TIMED_OUT in 3h 07m 50s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/02ae978c477147f6a23138980d01fa44 : SUCCESS in 1h 59m 23s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/c946fde2a92b40539555fc3e8ea58aa8 : FAILURE in 2h 25m 11s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/62d8aca896e54911adc6783f03bc79ae : FAILURE in 28m 56s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/31aca006411d42a39cb02f938b79bf15 : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/25336775aef2403e9ac75a93b0462aff : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/996a775ffe6f48ed95e035aad1d0f41b : SUCCESS in 8m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d9398dce15b2e60fc91198f8537a216df3137359","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-05-19 06:56:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"88f65a546539a0d17fa69a61ed8ad37683e53bcf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-19 15:33:42.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"f03af702768147fd65cbcde17221cc6fe36eb35a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-19 15:33:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2bfbb500d9f1611850286e0b11ef5307b38b2d1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 15:34:02.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/kolla-ansible/+/988844 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"53c3a84c6de25094151e403126387b8d097005da","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 15:34:03.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/kolla-ansible/+/988844 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"0a08d0d824f1a19b5bd68172b1203647ddad1b45","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-19 15:34:31.000000000","message":"Topic set to remove-networking-infoblox","accounts_in_message":[],"_revision_number":2},{"id":"42711eeb59dd1ea9801078f90995f13daadfb2bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-19 15:34:34.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"943fbf5e3835b5994ed1b696493cca170b510e47","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-19 15:51:19.000000000","message":"Patch Set 3: Review-Priority+1","accounts_in_message":[],"_revision_number":3},{"id":"a7b1ec85481de851695e6ca450dac17a67c65612","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 18:51:14.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/24a8733234bb4147b286d4ea6b091fcd\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/e9662a82727f4feeb8578c15e95de035 : SUCCESS in 1h 09m 19s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/0f971ae6d2d44c89a209ab3e568ff09c : FAILURE in 23m 35s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/e5cf0b8b43494f8a937b7be0973ef469 : FAILURE in 11m 06s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/d69e0599baae483a94b2defec81758ab : SUCCESS in 1h 17m 31s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/94fbaa9f9a1e449e9b5caf3cca54ee64 : SUCCESS in 44m 36s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/33c767ccf4d944fcb97a9f23417f57a4 : SUCCESS in 1h 10m 56s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e86199c5eda54ff6aba035cd04bf7c1e : SUCCESS in 33m 48s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/47b60bd9dda74ea9aa7bf2c9c9c574d8 : SUCCESS in 1h 41m 06s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/42df62349b7d4343a443c7f32f0d87ab : FAILURE in 14m 15s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/0dcb277b09c143ea8d61427f60fa7d46 : FAILURE in 37m 06s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/6cde5896f5264229a0a8ce269c2091a9 : FAILURE in 36m 39s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fadec6e6533845fca1b8cbd25f298c76 : SUCCESS in 1h 52m 03s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/78cb91e06de14bb394f71cdc062a0649 : SUCCESS in 2h 58m 58s\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/b5c1034dc7e7402f8ba202627f3ad064 : FAILURE in 16m 47s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/f64e83654dd1469ebe0d0527fcdb4aac : SUCCESS in 2h 02m 38s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/d32dcf1688eb466b923d13f652452cb1 : FAILURE in 19m 36s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/ebf18ce56ab743f48af595ce61408c0f : FAILURE in 50m 50s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/9623009d51514a069743975f9422cc32 : TIMED_OUT in 2h 36m 04s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/1329727fbf3a4fbab23116f45b7c8ab3 : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fec20a5a8f844d458014df56758da66d : SUCCESS in 4m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/52762fd6f702480883ad01e2a5da75ad : SUCCESS in 4m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/70bbbc2224364080803960abdfc2590a : SUCCESS in 4m 16s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/51e36573340d4ea2be25986d262780ab : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c10be12fed054d45b0fdbbca4da87df3 : SUCCESS in 3m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6ee0e3699f44938917bbad1a21bbe5f : SUCCESS in 8m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0695ded86ae54368a64a9bfafc47a0c5 : SUCCESS in 9m 33s","accounts_in_message":[],"_revision_number":3},{"id":"3831669394477ee1121678d782ad71be165c5eaa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 21:57:50.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c34fbe3bf1e14d4b84a5d43d89847f83\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/bf41e90a7ee44f71a5903fc9e2c7a648 : SUCCESS in 1h 43m 09s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/542e00f4875f44f49da98fe1d88ecc9b : FAILURE in 32m 15s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/f4767e9e296944f6bc77cec3b63fe67f : SUCCESS in 2h 29m 33s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/e2034f24a89449adb11d5b4a3847bfad : SUCCESS in 2h 07m 45s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/097947e715d444ccaa65dfa9599e14e3 : FAILURE in 1h 38m 33s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/6bd8df2dfe664cafa44be36afa027132 : FAILURE in 23m 56s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9e600cd4ed614a82a59726947d7b63fd : SUCCESS in 3m 30s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/68422e0048834c90a8742ab58fa22c36 : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/4904fa6a8c9c4f91a82662541b5fc0bf : SUCCESS in 8m 51s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"052e6e271262b525d7b16d02c374e09f4bcbe4a9","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-22 05:28:22.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ac1842870b6696f32253fe717d543b527b848e25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-22 08:08:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/786a3455c2a544c6b5f994db5d4d3a89\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/341d3ad7c9cf42a5b20fc692aeac34d7 : SUCCESS in 1h 02m 44s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/96eeb8a7f1d54d23911df0c9597182a5 : SUCCESS in 1h 14m 32s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/12951eb2a0bb4cde850dc81539997d3c : SUCCESS in 20m 22s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/f832b4192886437fbab3164cbf850c09 : SUCCESS in 59m 52s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/4c452e322268448c8314b7ec09bf2ce0 : SUCCESS in 49m 04s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a8b969fd618c420f831e355307595eda : SUCCESS in 58m 19s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c6d6b02abd6c496da81e3cf21bafaccc : SUCCESS in 25m 55s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/5f01edb9fe3749f6bfa558384941d422 : SUCCESS in 1h 27m 44s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/8b1a9aad17424941b97128569f9e1cf7 : SUCCESS in 1h 24m 53s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/d6a43eef3df443509d0b1480fe757b92 : SUCCESS in 1h 30m 03s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/d084a907027e40109c3d82be509ab3bb : SUCCESS in 2h 05m 10s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cb9d5bbb97f64119b2f30040130ebe95 : SUCCESS in 1h 23m 53s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/3e5ea438894e40ee9200a2dbe8111db7 : SUCCESS in 2h 05m 30s\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/8c01f8a77ee0474bba9aec8615a2e846 : SUCCESS in 1h 24m 03s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/ab4ba25502f64f90991855b18a218af6 : SUCCESS in 1h 27m 40s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/befb87b102e54e7a914e04df31fd7a57 : FAILURE in 2h 13m 19s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/870f42e8e4a745dbaeabe6eb952d7dc5 : FAILURE in 1h 13m 25s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/fc76594856e24ff2a3fae2e680c64132 : TIMED_OUT in 2h 33m 18s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/e02f112fd2c2407ea82e0814a243928b : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/250761692e024ca6b279bba9c38aa203 : SUCCESS in 2m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8f18cb928c1d41ba8d88d2aa41a8b300 : SUCCESS in 3m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/840bbdb3ff2f496f80fc3696fcb77a2b : SUCCESS in 1m 58s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f28763d548d8422297b5c474a6ceb900 : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/461fa7fa66cb46228c4c54feb38327be : SUCCESS in 3m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a3e3468bcd14f72a116092e72e75e11 : SUCCESS in 7m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/882457de5fbe4b77be69770b40cef699 : SUCCESS in 9m 31s","accounts_in_message":[],"_revision_number":3},{"id":"c363698321034be4f25d308f84bf2322a81b715c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-22 09:09:50.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"d9f446b41a20b201584c560c12e975d3f7e8a5d9","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-05-22 11:02:46.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"65d5665d576db0b05e5fa8441f76b456aa1b9b0f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-23 00:42:11.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7a19ddd30094a6494e66938569d787f\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/c076a9bbea1c444d83d60b761c0b663e : TIMED_OUT in 3h 05m 33s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/8a196774dbe64e128274017e42803fe3 : SUCCESS in 2h 53m 54s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/bcc55ab832d44d7c94d130f6747a91ef : SUCCESS in 1h 41m 50s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/2f067c3f5d8d418987c10a775f440680 : SUCCESS in 2h 20m 23s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/9c149cbf0c6a4b249bfeab0371af1363 : FAILURE in 2h 18m 24s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/a2d44265424c413a864b5ac1f611b0d5 : NODE_FAILURE Node(set) request c157d17c5a15431aa8f10dcdf31857b6 failed in 0s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b67ba0643f2c4444bccc52c8638686e6 : SUCCESS in 4m 57s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/aea2820ec62d44a1ba5cfca06ee2b04e : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/5ceaf76405ed48aab15b6cfacc5348ce : SUCCESS in 9m 47s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ca66d738b81fb01e78f242546d988928668e29d7","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-23 05:32:19.000000000","message":"Patch Set 3: Backport-Candidate-1","accounts_in_message":[],"_revision_number":3},{"id":"8c0fec031f9bfd93bea4d20be92410387875e50f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-24 10:28:59.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"c62f851328f6dbc5de7ab19439f349217ed51280","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-24 18:31:37.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"14617768809a156d12a1ad00c540384bf39133c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-24 20:28:58.000000000","message":"Patch Set 3: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/kolla-ansible/+/988844 is needed.","accounts_in_message":[],"_revision_number":3},{"id":"4b571ea3bc39b009910ddc8b9a4c32eddb8bc347","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-27 05:17:04.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b1a73a2c28b449d5eb0e56daf1880d600f50d383","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 07:52:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05ff50dba95b487392f4705c8d75b3a9\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/004a4f8cc01c4186a068a45db37b815d : SUCCESS in 1h 08m 22s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/0dcfeff066b84ba3940cb8c1d7519b6b : SUCCESS in 1h 20m 28s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/354ea3620a824d48a15dc848c14482fa : SUCCESS in 25m 49s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/a330e1f29b3e436592970f4ad9177c2a : SUCCESS in 1h 12m 37s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/a2ec6f5e5d934ee99d852510032076dc : SUCCESS in 1h 01m 18s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/11fc5a02eb274b4baa66eb6a96f0baa5 : SUCCESS in 1h 14m 10s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/391e7b4451694047a55c5adb0776579e : SUCCESS in 30m 22s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/2b100ed9f26f4db1a7cd9f86333a0c62 : SUCCESS in 1h 15m 40s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/80d5653d603948dea4a4aa8f5c3dd7b5 : SUCCESS in 1h 30m 00s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/e1427c2d44c8487ab334f4938a02b2ce : SUCCESS in 1h 35m 46s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/22effff132f54ad291ebb775564536c6 : SUCCESS in 2h 18m 41s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/27a18f2bd5f348c5bbdbb840569b9a4d : SUCCESS in 1h 20m 55s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/6f28fe9e80b34cd8a9934c9fed02dfc5 : SUCCESS in 2h 01m 30s\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/bf686d0ac9334c148ca0543d70b3edce : SUCCESS in 1h 33m 44s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/849e3184d2614ab88f2c46324a0d8c47 : SUCCESS in 1h 27m 43s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/6cb097772d1e470295308cf7cfacb1ec : FAILURE in 1h 37m 02s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/fdbd3a82da8b4efc9998550f24b79b81 : SUCCESS in 2h 32m 06s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/31837908a9df493a9601d1d7a2f8c3ee : SUCCESS in 2h 23m 15s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/0882f5405a1546758da27dcd5a3e5aea : SUCCESS in 2m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/751a6f8373d54ea0b3a89d8584fe7a66 : SUCCESS in 3m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ea2e8570136d47f7b3cc6b798511876b : SUCCESS in 2m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/76ae9c535f4e48dd9552e4064421265e : SUCCESS in 3m 44s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a0b1e4de589d497898636767475bdc4c : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c1a9a24ff5bc49f292a8463702e59727 : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77857d8c95924152a968b4c210c05aa2 : SUCCESS in 6m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e347e808541741b08e42cf81f319f588 : SUCCESS in 9m 36s","accounts_in_message":[],"_revision_number":3},{"id":"c6cef10776c188002f445b58b0dd9248a302f94d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 07:52:41.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"13a260820c11e34a2456ec8d3acc7a37fc9f7efd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 09:59:12.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/abb6d90ecce94d48917dbffd008f0f21\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/a75a8dac9cac4ef9bce812f25c0a2040 : SUCCESS in 59m 11s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/b50463add3684d79a4a15ce8d8694524 : SUCCESS in 11m 53s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/7d60dafa4d8f43ee969b6a6a25642941 : SUCCESS in 1h 11m 10s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/d3cf12ce5b36408c901f56810f4168a0 : SUCCESS in 1h 01m 50s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1f975657c7104cfc8888705c4b2ca42b : SUCCESS in 1h 24m 01s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/497b9608fb2940c7b60350302d1d3a6a : SUCCESS in 14m 01s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/dbf5dd229159418b938076166e297cfa : SUCCESS in 1h 19m 27s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/0abb0cc01ff64303b93349584a9f06bc : SUCCESS in 1h 32m 26s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/9c84dc0e403f41bcb240db57d8821b83 : SUCCESS in 2h 05m 45s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b43029fc21cf43928c589c7fa0a8f2ee : SUCCESS in 1h 23m 32s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/88fa48b3efe941ed89ae8aa78aff5103 : SUCCESS in 2h 04m 34s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/4ba70193264a4e64aad5f1ef40d193d5 : SUCCESS in 2m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61bfd888363d465699743f67425f6ba6 : SUCCESS in 3m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6c60259a2fdf490cbc43f63489f6ef2f : SUCCESS in 2m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8341a0815920404bb76aa7081c7bafe6 : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f82bd3626c34ae7ac8cf21e72b71818 : SUCCESS in 7m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ac03cd641dfd4799a5aabf123f130010 : SUCCESS in 8m 00s","accounts_in_message":[],"_revision_number":3},{"id":"88d0b8ca241c68deceaaa5229dd840a38675128c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 09:59:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"1d6adaaa1cbb94c9ff2711e198c659bb07fe9776","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 10:00:10.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/478a10ff69ee4bd7adbcfaebddadb985\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f02427a64bf54de0984ac6923b0db298 : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/834bb089b6384609ad1dd52774d9ea8b : SUCCESS in 45s","accounts_in_message":[],"_revision_number":3},{"id":"c114a85d687ffed0d39d60ebdcfdf0df90142583","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 10:43:46.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a0babdf1a47f4e25969d08e937d93a26\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/48334f9406c04a0f8b407e2879dc1924 : TIMED_OUT in 3h 05m 58s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/295a46ac08ca4a61b37c5ca3e433e19c : SUCCESS in 1h 44m 26s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/c6fc0ff0ed43441e963a7ddf6186b02a : SUCCESS in 1h 16m 14s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/b58c2a682c2a42f5b9199ccd6d486ab8 : SUCCESS in 2h 21m 56s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/65acc96b15554001a99bb46ef6a5d4df : TIMED_OUT in 2h 32m 27s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/c0463f5f89184b109e6c05b6238ef206 : SUCCESS in 2h 45m 24s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/11e09117a5364a69918cda2a7d715c5d : SUCCESS in 4m 13s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4439b9cb478d43d2bc20b7d8a848806a : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/aac62cb727ba4f12a298247ee0ab976b : SUCCESS in 9m 33s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"701ab493bd8c53bfcc2298ec831e0985aee7c7aa","revisions":{"b8eaec595a652d18b48c02ffee31c1c69a5cdb68":{"kind":"REWORK","_number":1,"created":"2026-05-16 14:58:48.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/45/988845/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/45/988845/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/988845/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/988845/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/988845/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/45/988845/1"}}},"commit":{"parents":[{"commit":"a9bae9fe56dced112e71de51bd9f382f00d584ff","subject":"Remove tox-py310 ansible-core version in test-reqs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a9bae9fe56dced112e71de51bd9f382f00d584ff"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-16 14:58:22.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-16 14:58:22.000000000","tz":540},"subject":"Remove support for netwokring-infoblox","message":"Remove support for netwokring-infoblox\n\nThe project[1] has received no commit for 4 years. Its latest commit\nstill uses upper constraints from the old, abandoned Wallaby branch.\n\n[1] https://opendev.org/x/networking-infoblox\n\nDepends-on: https://review.opendev.org/c/openstack/kolla-ansible/+/988844\nChange-Id: I9e2ba00671eb7d05de3c40ed9524f20922e081b1\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b8eaec595a652d18b48c02ffee31c1c69a5cdb68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b8eaec595a652d18b48c02ffee31c1c69a5cdb68"}]},"branch":"refs/heads/master"},"8017b62f1a027dcd9f0b6dbd70efb4fd202fddb5":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-05-19 15:33:42.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/45/988845/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/45/988845/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/988845/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/988845/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/988845/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/45/988845/2"}}},"commit":{"parents":[{"commit":"a9bae9fe56dced112e71de51bd9f382f00d584ff","subject":"Remove tox-py310 ansible-core version in test-reqs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a9bae9fe56dced112e71de51bd9f382f00d584ff"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-16 14:58:22.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-19 15:33:42.000000000","tz":0},"subject":"Remove support for networking-infoblox","message":"Remove support for networking-infoblox\n\nThe project[1] has received no commit for 4 years. Its latest commit\nstill uses upper constraints from the old, abandoned Wallaby branch.\n\n[1] https://opendev.org/x/networking-infoblox\n\nDepends-on: https://review.opendev.org/c/openstack/kolla-ansible/+/988844\nChange-Id: I9e2ba00671eb7d05de3c40ed9524f20922e081b1\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8017b62f1a027dcd9f0b6dbd70efb4fd202fddb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8017b62f1a027dcd9f0b6dbd70efb4fd202fddb5"}]},"branch":"refs/heads/master","description":"Edit commit message"},"701ab493bd8c53bfcc2298ec831e0985aee7c7aa":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-05-19 15:34:34.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/45/988845/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/45/988845/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/988845/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/988845/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/988845/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/45/988845/3"}}},"commit":{"parents":[{"commit":"129e88848d511ebdac4e42124e1192bc5c38d4b5","subject":"Merge \"Update mission statement in docs index\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/129e88848d511ebdac4e42124e1192bc5c38d4b5"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-16 14:58:22.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-19 15:34:34.000000000","tz":0},"subject":"Remove support for networking-infoblox","message":"Remove support for networking-infoblox\n\nThe project[1] has received no commit for 4 years. Its latest commit\nstill uses upper constraints from the old, abandoned Wallaby branch.\n\n[1] https://opendev.org/x/networking-infoblox\n\nDepends-on: https://review.opendev.org/c/openstack/kolla-ansible/+/988844\nChange-Id: I9e2ba00671eb7d05de3c40ed9524f20922e081b1\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/701ab493bd8c53bfcc2298ec831e0985aee7c7aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/701ab493bd8c53bfcc2298ec831e0985aee7c7aa"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"8017b62f1a027dcd9f0b6dbd70efb4fd202fddb5","theirs":"129e88848d511ebdac4e42124e1192bc5c38d4b5","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-1"],"failing_atoms":["label:Backport-Candidate\u003d1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
