)]}'
{"id":"openstack%2Fkolla~988846","triplet_id":"openstack%2Fkolla~master~If3d0e65d345a2b102eec2d21e053c68b6096e298","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-08-17 11:10:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If3d0e65d345a2b102eec2d21e053c68b6096e298","subject":"Drop leftover of monasca support","status":"MERGED","created":"2026-05-16 15:00:45.000000000","updated":"2026-08-17 11:11:59.000000000","submitted":"2026-08-17 11:10:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"988846","meta_rev_id":"ed5154d20f1e678facb31a48ec31043ad1de1490","_number":988846,"virtual_id_number":988846,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-17 11:10:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2026-08-17 09:56:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2026-05-17 13:55:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2026-08-17 10:32:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":-1,"date":"2026-08-17 09:56:50.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-16 16:37:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-17 13:55:51.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-08-17 09:49:19.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2026-08-17 09:56:46.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"146b20b1d2f68565f6b367809735c80ea29cfdee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-16 15:00:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"754ae24a27918d7e1deea44f2113321ad9ce5ff7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-16 15:06:04.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"c9a590dd56894e85730551e7a4c6d425071ad14a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-16 16:37:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/330aed3840f048a5bbe2b22362440416\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/4b3f93eec4fd4e9fbeb28f1e2f1d7aa7 : SUCCESS in 1h 15m 59s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/a0a736e6774241a288abfcdf0c77599f : FAILURE in 19m 57s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/5f680e4ed56341afa9c3af430a629482 : FAILURE in 11m 23s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/9b78afd797784d5bb71d01aedad35a9e : SUCCESS in 59m 39s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/1fc0ab9e79654ced94e0a57f9533c869 : SUCCESS in 1h 00m 31s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/810c4c8eef3d43639aa1e934385e64b4 : SUCCESS in 1h 25m 11s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/523b73d3184b4876843e55b1140ca84c : SUCCESS in 22m 17s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/8ce1754e5c454d038fa995ade4d0adf8 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c0f33dc0f9648fa8c797ab9f4f75a7b : SUCCESS in 3m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b78d0af379f946f790b9483676bb8e5c : SUCCESS in 3m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/505a0e4a3985492ca2c16ba34ba50f4d : SUCCESS in 3m 13s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f7e82b10c2fa4725a31736e4eb445fb4 : SUCCESS in 7m 45s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f0911331d937457a81ac78d74e925548 : SUCCESS in 3m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e623c694c744a2096239e773e3300e2 : SUCCESS in 3m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/33c7c9d7250b4beb9b5fec9a76387089 : SUCCESS in 9m 33s","accounts_in_message":[],"_revision_number":2},{"id":"9942463bbd35e669e08f9c0c7467151f705c3ceb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-16 20:37:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/657c5ac844af479c9162c8ad1f3b5053\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/dedb46642b7a4a4e85842cd1d85304a6 : SUCCESS in 2h 09m 47s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/d4956ab564d54e17886b640cc2baa317 : FAILURE in 24m 06s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/de5ab0be4a2e460484881f7faa91cefa : TIMED_OUT in 3h 08m 53s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/5bb7476fd1d4474ea8edf36f8ec590fc : SUCCESS in 1h 57m 53s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/874ff17f0505478abac41ee04eb5b86f : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/33b45e78c3e6421481e0893b37417487 : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/ff6d5d6835004c9b9fea01c3844e7c02 : SUCCESS in 9m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6d2923cee08c9a0771a6bc78b46295c2e2d1d5ba","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-05-17 13:55:51.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"320e195774d8de5bdc9cb0d8ac13049fc318a456","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-17 15:28:29.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ff54cb3cbdb54cdc8fa6762a53b86a97\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/16e2a7d335d24650ae41619d0a2bbbda : SUCCESS in 1h 00m 10s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/aa1980c2bdd044c4a6376c80b84a1730 : FAILURE in 21m 18s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/3038d536a6014c65abb1344d510a736d : FAILURE in 10m 59s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/4de36d5daa4b4bc39b893b43d68075d5 : SUCCESS in 59m 42s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/d2acafd716c04c62b4fe7490bd5ba995 : SUCCESS in 43m 56s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4c9a608d2cd6433c898f89d199e86838 : SUCCESS in 1h 27m 27s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/32075e119e824a67a5e073f5a7cd5fc7 : SUCCESS in 27m 40s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/a30186d6931f49a68e7485a2ea25ee7e : SUCCESS in 2m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e1ba90be91d419c9e87fd7e7eb40bca : SUCCESS in 6m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/402ebc0b07fc44b2bdcc9483bcfc2cea : SUCCESS in 2m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/612b2ad9001c4ff69fbe452bab987366 : SUCCESS in 3m 23s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fb30e1cebc45477585ffc66132a9faf2 : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5e5ab7a9e44643b5923a3eb84d96fa2f : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3a21419c86d4a8988f872668643e41e : SUCCESS in 4m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/71380a84008f41e6b7f084362e25e4fa : SUCCESS in 9m 44s","accounts_in_message":[],"_revision_number":2},{"id":"d933614aca808115035b2f938a921e5337346f49","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-17 16:16:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65c2c5ce00354b889e8115aba274c037\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/f9434ab921f94606a025663c6f57b93f : SUCCESS in 2h 14m 24s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/2d14595fde84432193238274b8a3ace7 : FAILURE in 28m 51s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/8bbbcaa10bbf437fa556e9fb4b6f9742 : SUCCESS in 2h 13m 50s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/80778ce7984d4e89b512940d134efe3a : SUCCESS in 1h 54m 15s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1e5855ec7bb3410ca90b49c6fa95f7dd : SUCCESS in 4m 06s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7cd2f5c840fd4d9b8bd3d66aef57d973 : SUCCESS in 3m 48s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/fb0a2d3bb59741eda8eec31323e777d7 : SUCCESS in 8m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2cefb4d69f5dd67d86996b2f5ab59d68f01b8465","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-08-17 09:49:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"143fbe776baa323d7a86d58031a5e7fa91ac7a87","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-08-17 09:56:46.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"6dcc863643cf98c57555491ab310ac9a6c1847ce","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-08-17 09:56:50.000000000","message":"Patch Set 2: Backport-Candidate-1","accounts_in_message":[],"_revision_number":2},{"id":"c30df02b3b6a9b05db1c4da1155b93a4951f50d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-17 10:31:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2b09f43a7c54d7383a34f07e66046c3\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/3007869f4b3e4feabd874af6c9425855 : SUCCESS in 16m 52s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/77893c78759849178d9eb807e9f0db6f : SUCCESS in 25m 36s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/7cd953acdb374ecf8e19f2051e6aab0e : SUCCESS in 19m 18s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/9f4b4e22255a4076a2a2f9ea644e5c2f : SUCCESS in 16m 28s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/6ea7c5c408e34e7892f5f9d666b913e0 : SUCCESS in 41m 03s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/23bd7fb39c464be68afd33e016fdad8f : SUCCESS in 27m 29s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6384c116a2ad4240a2842dbccf7c9a98 : SUCCESS in 14m 50s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/a890b27e20dd4e5cb999f478b9da1fbe : SUCCESS in 2m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9836c9eff5df4632aa7e9bb00a25d33e : SUCCESS in 3m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0b54afd4b0a24bafb9508fe032dcab7e : SUCCESS in 3m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cc910421726e46a1ba8b178c65c6efd8 : SUCCESS in 3m 09s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0f9adaf7379345d5a3f27784515220ae : SUCCESS in 7m 15s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e0a5e505067e49cfba95689a74e5d661 : SUCCESS in 2m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e8cdb73a31241c99f050830583e739a : SUCCESS in 4m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2ee8cdbe772c4e418f7f68e14c3a9326 : SUCCESS in 7m 59s","accounts_in_message":[],"_revision_number":2},{"id":"d4633d81528e2ce23ed99617aa19e83e983a8ad7","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-08-17 10:32:58.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"13098d1b19186b370ca529ffd9d9c9685d8765e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-17 10:33:11.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"bdd3db0c939962333fffb9d8b5280070c0b17a0a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-17 11:01:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/674d325e2a274d1e859f02dafaa67acb\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/c8d73fdb879b41cabc1ff00062490929 : SUCCESS in 51m 00s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/fdaca64c0bd347cba9e156a23cfccf1b : SUCCESS in 49m 59s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/4f94a76789d246c2a4b28a2e7595987e : SUCCESS in 1h 00m 06s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/b3a5405b8c1546da9fa902589bb0c023 : SUCCESS in 53m 58s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/58e81c5eb39c40f7a9d82faac5f5ead2 : SUCCESS in 6m 01s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/828b039d78564e1abc8a82300d68746f : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/5f903aac30764eaca6358acd91cad2ac : SUCCESS in 9m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"aae2f438bc9f31aba9e09c0c4bd534ee209190fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-17 11:10:57.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4d900deb0c2409b9cbbdd68b0e171c5\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/2fedd6e0bbb6472d912c594a932faa85 : SUCCESS in 28m 36s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/4b51f5c9f1be41c89fee9bb91967888d : SUCCESS in 14m 59s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/fb88761f42a840fc8c3cb209abfc553f : SUCCESS in 24m 10s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/3dcd5799c6e545c1b20e97e92de06978 : SUCCESS in 37m 07s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cff702d6f9af4460b740aeb1bd03a128 : SUCCESS in 15m 32s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/df3c889bdf3e424fb89f8f073329cef9 : SUCCESS in 26m 42s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/60120e61a6c943f7813952f97e733d8f : SUCCESS in 2m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1a27dffa2044a0a8763d192461ad11a : SUCCESS in 2m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6c8b12d03a794849b43fea0930a87c41 : SUCCESS in 3m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2663dfb733b54d95add270f99b60f233 : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4c686f6382a49c0a35c6ce3f7be3335 : SUCCESS in 7m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/978168fbce3440e4adefbde09f24f394 : SUCCESS in 10m 25s","accounts_in_message":[],"_revision_number":2},{"id":"cc1245db9f6ac0c948b89445061d8e275defa440","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-17 11:10:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"ed5154d20f1e678facb31a48ec31043ad1de1490","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-17 11:11:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a7eb6ea71394226b437eaec548a4d4b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d4e757b91af4b2aab352fd204a7ffa5 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5c5b19b28769d4e9497fe1f0e5d8da0a67c43b6b","revisions":{"256f480dcaa739b0ab570a3fa9530c0ac5389375":{"kind":"REWORK","_number":1,"created":"2026-05-16 15:00:45.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/46/988846/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/988846/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/988846/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/988846/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/988846/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/988846/1"}}},"commit":{"parents":[{"commit":"a9bae9fe56dced112e71de51bd9f382f00d584ff","subject":"Remove tox-py310 ansible-core version in test-reqs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a9bae9fe56dced112e71de51bd9f382f00d584ff"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-16 15:00:11.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-16 15:00:11.000000000","tz":540},"subject":"Drop leftover of monasca support","message":"Drop leftover of monasca support\n\nMonasca support was already removed[1].\n\n[1] 52a53927f5d390886c3940e364fe9fa888e3bc98\n\nChange-Id: If3d0e65d345a2b102eec2d21e053c68b6096e298\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/256f480dcaa739b0ab570a3fa9530c0ac5389375"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/256f480dcaa739b0ab570a3fa9530c0ac5389375"}]},"branch":"refs/heads/master"},"5c5b19b28769d4e9497fe1f0e5d8da0a67c43b6b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-05-16 15:06:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/46/988846/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/988846/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/988846/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/988846/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/988846/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/988846/2"}}},"commit":{"parents":[{"commit":"129e88848d511ebdac4e42124e1192bc5c38d4b5","subject":"Merge \"Update mission statement in docs index\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/129e88848d511ebdac4e42124e1192bc5c38d4b5"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-16 15:00:11.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-16 15:06:04.000000000","tz":0},"subject":"Drop leftover of monasca support","message":"Drop leftover of monasca support\n\nMonasca support was already removed[1].\n\n[1] 52a53927f5d390886c3940e364fe9fa888e3bc98\n\nChange-Id: If3d0e65d345a2b102eec2d21e053c68b6096e298\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5c5b19b28769d4e9497fe1f0e5d8da0a67c43b6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5c5b19b28769d4e9497fe1f0e5d8da0a67c43b6b"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"256f480dcaa739b0ab570a3fa9530c0ac5389375","theirs":"129e88848d511ebdac4e42124e1192bc5c38d4b5","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-1"],"failing_atoms":["label:Backport-Candidate\u003d1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
