)]}'
{"id":"openstack%2Fkolla~988849","triplet_id":"openstack%2Fkolla~master~I5937a17b0d70f9c1db4b2617854d0180c37fe2c0","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"37203":{"account":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"last_update":"2026-07-02 08:29:15.000000000","reason":"\u003cGERRIT_ACCOUNT_37203\u003e replied on the change","reason_account":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"}},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2026-07-01 19:59:49.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"17669":{"account":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"last_update":"2026-07-02 14:09:16.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-07-02 14:09:16.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-07-01 17:10:29.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2026-07-02 14:09:16.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2026-07-02 14:09:16.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5937a17b0d70f9c1db4b2617854d0180c37fe2c0","subject":"Fix parsing of permissions with special bits","status":"MERGED","created":"2026-05-16 15:41:33.000000000","updated":"2026-07-02 14:11:09.000000000","submitted":"2026-07-02 14:09:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":24,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"988849","meta_rev_id":"b09274b9243570235b45da2f82a9f6db540a9e83","_number":988849,"virtual_id_number":988849,"owner":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-02 14:09:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"all":[{"value":2,"date":"2026-07-02 08:29:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},{"value":2,"date":"2026-07-01 19:59:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2026-07-01 17:10:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"all":[{"value":1,"date":"2026-07-02 08:29:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"all":[{"value":1,"date":"2026-07-02 08:29:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"all":[{"value":0,"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2026-07-01 16:15:36.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-16 18:49:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-26 09:08:44.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"state":"REVIEWER"},{"updated":"2026-05-26 09:08:44.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2026-05-26 09:08:44.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2026-05-26 09:08:44.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2026-05-26 09:30:34.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-06-08 08:33:41.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"5f908211b9752cd3c67dc3525555ec61ebca8e97","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-05-16 15:41:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bc5e318d7a5ed3deedfecce1edee7484a70cfcd2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-16 18:49:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/85f40de108554416a9e86c715a5f729f\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/426d57f71c14448b85376f32574250bd : SUCCESS in 1h 02m 20s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/0a9ab2a96bcc4a6fbde417bb6ec2ecbd : FAILURE in 15m 47s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/cc6654c6d0414c8bb575186838161699 : FAILURE in 9m 26s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/2e1ace9f240f4417bc2ba0ce82faf76e : SUCCESS in 1h 04m 40s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/e49f86760ee8499ab1895303ceccb262 : SUCCESS in 31m 55s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7f4029f568f34d49b8aab610614c958d : SUCCESS in 1h 36m 02s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5b0ab81f310f4e808ddd402314971d98 : SUCCESS in 30m 07s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/c0ff87c4f365412da965f0c55904ed80 : SUCCESS in 1h 41m 01s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/69c5bebc47704cc29a91b01be58eba53 : FAILURE in 11m 40s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/5cc7aacec7e44e2585d8a7f049232dab : SUCCESS in 1h 40m 48s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/1644887b94b4483697cd70194534ec56 : SUCCESS in 2h 29m 33s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/78f28acd1cbb494381c1ef378d7176ee : SUCCESS in 1h 23m 58s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/ff556a98d46c4063bf48f2924d2d433c : FAILURE in 56m 49s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/a87d4366aa10428e88af2b306bf35e59 : FAILURE in 32m 22s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/6869cce55bfb403b96bb3cc21c92de2a : SUCCESS in 2h 27m 23s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/e8c29d3d2e764d1dbdc81171b23e0df5 : TIMED_OUT in 3h 05m 53s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/c228f249f91c4026b5cf7ece7c0752e3 : SUCCESS in 2h 08m 25s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/35433b989a4c4c2bb0349c7f6b0b592a : TIMED_OUT in 3h 05m 52s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/148619fcc9014cc185838e1f94a51231 : FAILURE in 23m 55s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/8ea900ee2b4444718ec094f47f8c67ed : SUCCESS in 1h 57m 21s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/c63ae4b787ff430ba3fe5e3c022c2cde : SUCCESS in 1h 59m 16s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/67a139e6f3fe4be8ab151ca1d55399f5 : FAILURE in 17m 26s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/a171fd27a7fd41bda304a6fb3f3d1b82 : SUCCESS in 31m 14s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/b0c97f38c0eb4f9cac4b07ab98039bc8 : SUCCESS in 27m 21s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/b36644f06b314c689200572d83892f53 : FAILURE in 14m 31s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/6a5b9a678656413095da51fcbdc338dc : SUCCESS in 57m 55s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/024bb1e1dd3a4040a8f1b0022c035ddf : SUCCESS in 50m 22s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/333d2ce05c2b4d089a0a27c3a46e822a : FAILURE in 12m 52s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/d6ff4db1655f484bbee3db1dfca9b1b5 : SUCCESS in 1h 32m 48s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/39d48b549d884f3a85e366fd1c9f946c : FAILURE in 17m 16s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/1c0d49e1e6ce431aa0cd78ee884cfba7 : SUCCESS in 2h 23m 00s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/234fd46ed6404628a4ceb745a3c18437 : FAILURE in 1h 36m 29s (non-voting)\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/ae24a16866644e6186d609a9be0f3da7 : FAILURE in 9m 04s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/c3ec2db5922042e89dae1768d784d4f1 : SUCCESS in 1h 07m 15s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/e1a116136b314031b9b4db100010bc42 : SUCCESS in 1h 02m 31s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/67e78e72fa294321b56af6fd744868a3 : SUCCESS in 2m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acbf54a13144477d89e0e07d5093b709 : SUCCESS in 2m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/202c4bb382fb4e9f92fb3430b6637440 : SUCCESS in 2m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4f63d0454c1e4b9680d0be7618ebaeb8 : SUCCESS in 4m 43s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/674d6494eec24c6aad4d5593a419b27f : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3c10f847a944144a1a6b4ec2f604d29 : SUCCESS in 2m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/024185b57387486e91a31e81f914ab47 : SUCCESS in 3m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d81733224537431ba19865a710aa0338 : SUCCESS in 8m 18s","accounts_in_message":[],"_revision_number":1},{"id":"8e5e9809e7198b3afa9f74ed9007c103739ba571","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-17 00:01:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f53e53c57984927be9817e7b00a0ad5\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/03569eb15a5c4295a559e08c56a43449 : SUCCESS in 1h 15m 10s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/f34c8947800e4764a3de01329f0e8352 : FAILURE in 29m 25s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/4154f72e55b54d87a429da4074a41845 : TIMED_OUT in 3h 09m 30s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/b1aeca53733947408a10e72f909de0e1 : SUCCESS in 1h 56m 51s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/7ac727b696384b5496fb43de72d00958 : FAILURE in 2h 26m 29s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/913cdd8e5f11443e8f6bbf88560b0a00 : FAILURE in 28m 38s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2a4988ede5a04514a0064c5d4cf59929 : SUCCESS in 4m 26s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c9c9eb8cc3e844c4aadd5b9f43dc8974 : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/98d4f7dc77c942ce970b27dec2a1cb17 : SUCCESS in 8m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4c2dd7695ee5f47697584599246ac0a86ca04623","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-05-22 16:38:51.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"ae525997abd2dbfd3549398b4dcd9b4ec88e26f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-22 19:46:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09e3782ceedc46bd90d301f2177b39ce\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/6a9b6021d5b941069598a042e7fc4411 : SUCCESS in 1h 02m 22s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/86b8460a094040c996896fe54b99a483 : SUCCESS in 1h 18m 06s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/d92a5699c78c4c8bb888b13175461390 : SUCCESS in 27m 31s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/b1ac4073fb9847658dda256d49153fb3 : SUCCESS in 57m 53s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/f6604c46130243ea91e48e138b7fbe70 : SUCCESS in 55m 15s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/53e612d88a2b4946b987c8eca0da8827 : SUCCESS in 58m 46s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8613859fae4a4933bff4b014bba31669 : SUCCESS in 43m 02s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/55fd249abd7f44d4a72fe2618fdd7ffd : SUCCESS in 1h 29m 51s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/38183162e46249ef8dc8773acb7e4779 : SUCCESS in 1h 41m 56s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/892ed7df905a4769b5ccb867422dfc1d : SUCCESS in 1h 30m 27s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/16460e535f8844d099fa1b1590f01a78 : SUCCESS in 2h 14m 04s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/11caedbe24cb4530b7191054d2106d75 : SUCCESS in 1h 44m 41s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/9ef9bccdc94040c886540ef03e4f0395 : SUCCESS in 2h 07m 09s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/de6135492e6748fb9957570171692a10 : SUCCESS in 2h 30m 48s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/3734d64e74f7420f90121abc7472ddcc : FAILURE in 2h 13m 41s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/8d92e15d047f4f47849e7e97cf5ec8be : TIMED_OUT in 3h 04m 29s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/2a99bd5f8ad7435d82cebf4d4ce6049b : SUCCESS in 2h 06m 16s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/4b006dfe6d06491c9ebe60e0697b6d10 : SUCCESS in 2h 50m 41s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/03ca820b7e1b402cb32fbc3671bebcd7 : SUCCESS in 2h 02m 41s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/d589de9082fe47b2b39abc3c7c00bece : SUCCESS in 2h 04m 32s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/def364b6daed41ee92f245dfe474b91f : SUCCESS in 1h 50m 11s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/66de968418634350b8e53b91c580796c : SUCCESS in 27m 02s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/890b3e670ddb4bf593cb0393a8d3435d : SUCCESS in 33m 39s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/997d178b6fe14e8ca68d05dd5b03e077 : SUCCESS in 38m 20s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/c7bea1b5a18b4c589413462169d8d39e : FAILURE in 1h 21m 31s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/b65db1e748e64c0eb5af6f12415b5d61 : SUCCESS in 55m 51s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/ba7d62874b614ccab8c7541c5dc0bd8d : SUCCESS in 1h 17m 00s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/00bdb94fb3144b7aaa1c863a110b2b2c : SUCCESS in 1h 32m 59s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/7d6ede88eec248e5897749f846ad1ef3 : SUCCESS in 1h 27m 32s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/434b0214ba5743039988828a25abc95d : FAILURE in 1h 45m 46s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/64deb5f438894a92b3de6e90a03f4509 : TIMED_OUT in 2h 36m 41s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/703773f85bcd445a82e75e4f545c1559 : SUCCESS in 2h 24m 43s (non-voting)\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/02d5194ac59a493593cccd9c0b30dbaa : SUCCESS in 1h 53m 04s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/a6c4f5f066fb4b0ea6d79a636730e7c0 : SUCCESS in 1h 31m 55s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/a78805510dbc455ab458efb0d6bc779e : SUCCESS in 1h 52m 24s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/cbf3cd58354c47d085d9edca5aa89e31 : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c9520416410427aa6369ed1394629d1 : SUCCESS in 2m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2152acdaeed64cc38b0aecb37c0edd08 : SUCCESS in 3m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f4643ed50935493faf7a77eca57764f2 : SUCCESS in 3m 36s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/eb84d66a383b4776bd9ddf215e4f7c41 : SUCCESS in 5m 46s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/90c94da1f86e448ca2394b14ee37531e : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4c41a0fcaad4fec83f83ac786514994 : SUCCESS in 4m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/73cf7a8f0b2c479bb690348b651f32d3 : SUCCESS in 9m 17s","accounts_in_message":[],"_revision_number":2},{"id":"cf807dac2d290d8fd971e17055922a21a06bef96","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-23 02:06:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b6603b509da4b8e95badff5388205e9\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/45931d859482407184fed2ad74f5412a : SUCCESS in 1h 19m 16s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/3cede2b7a1044d859ebaf6064e33530c : SUCCESS in 1h 58m 59s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/41977bd5b21a45a7916ccad31bc63340 : SUCCESS in 1h 13m 36s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/0486d01ec0b94c92be12704e6cf81d94 : SUCCESS in 1h 25m 09s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/65bd21ebd09c422c9842699f142a4adb : TIMED_OUT in 2h 32m 48s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/cbcfc7d89b634769b76b2fefe1ec1cda : NODE_FAILURE Node(set) request a4ce982ca5a74ad9b7ada313668ffa97 failed in 0s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/241f00632b3c462e8599ffd646dee71d : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a663493d2de6496eb7a53ecb5f3cc0d2 : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/0d6e767254f74d0e9677f1ae94bdf320 : SUCCESS in 9m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c42670dc3b68330dec147b4e9d48f73211407b47","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-05-23 09:27:26.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"7ebc53f4031023f5bd6da9a97caf48c504657de6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-23 12:34:16.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7171ccaba467484cbb5896b17e1e27c0\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/425a88ed50c449b9936e0d7e643eefc1 : SUCCESS in 1h 02m 14s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/1c38b6a50bb8472389bd267166ad8007 : SUCCESS in 52m 17s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/d5e30fba75b8411ba68ce730ce6fd3fd : SUCCESS in 11m 42s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/aed514f599984d8c86fab7c2c576c98e : SUCCESS in 1h 08m 21s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/1c92e8aa880249b692deff532522073b : SUCCESS in 38m 19s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8b0eb134110a41b79ed04ec573482827 : SUCCESS in 1h 02m 23s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3c327c86b0274b6a923133c160de06b1 : SUCCESS in 17m 48s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/95cdd75d37214a2e8a0c4a11bae927bb : SUCCESS in 1h 17m 13s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/c0cfd89b6bd24b3ab1d9594033e6df63 : SUCCESS in 1h 20m 44s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/336b33cc40974a94ac7dc7ed0a484302 : SUCCESS in 1h 51m 23s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/22bd7fef0f2a418da95d7fac89e637be : FAILURE in 1h 04m 10s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/abc1e0c762f049d1a43e57f5d3c0a415 : SUCCESS in 1h 49m 18s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/7bbb63126a674f6bb14cbc6c524d4fd2 : SUCCESS in 2h 06m 34s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/ad5c86f0eb31493db0871a9d7e390ff0 : SUCCESS in 2h 00m 09s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/72c9ba1b3762462fa6a5eb0b23756bc0 : FAILURE in 2h 37m 52s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/1de81e89a9eb4e7eba544b1c0abc261d : TIMED_OUT in 3h 03m 49s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/c223ddb438d54c838147b1201ed0ee7a : SUCCESS in 2h 22m 50s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/a3d8a5613986468cb48ebb20b19b7fcb : FAILURE in 2h 56m 14s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/9d57ef92495242dfb00de170b41974fa : SUCCESS in 1h 52m 57s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/e029cfdadbd24958ad3101e73d762b07 : SUCCESS in 1h 54m 52s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/cc240d1db9d84e7baadd30f3ef857ab2 : SUCCESS in 1h 50m 04s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/e114c75356f44be789ac7ceaffb7a722 : SUCCESS in 29m 15s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/65eb93b50452469b996b374563328090 : FAILURE in 29m 11s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/1bc0252631bf4950a285202705218e44 : FAILURE in 20m 49s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/88f8bbfb393a43c7a8bc07b1160d9015 : FAILURE in 1h 10m 04s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/13be4c13481b4db396cccba8d92ab8da : SUCCESS in 55m 14s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/24608f1883454f35b69a86310d354d08 : SUCCESS in 1h 16m 50s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/9fdfd35867974c27b14f609335392e8d : SUCCESS in 1h 30m 03s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/f86d8fe007a14a8c92f03e08db98618f : SUCCESS in 1h 29m 15s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/1312d9be27b14ae1939ef64ad65fe66f : FAILURE in 42m 29s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/cac70e624b2f4322897aa0a5e6d88664 : TIMED_OUT in 2h 33m 56s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/156603390eb04511a856d21e1666df62 : SUCCESS in 2h 19m 12s (non-voting)\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/34abb51b497e4e8cbf0bc70596eca268 : SUCCESS in 56m 30s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/4c6db12eaf13476ba8f67794487138ef : SUCCESS in 1h 14m 20s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/a3b5783e1b2947f18d1f3fa1b1d20fd7 : SUCCESS in 1h 32m 44s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/204aa7149b0e4fb9985908c7611b2463 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31b257aa27c74e7591d4f2fa20483c0f : SUCCESS in 4m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7e0075417d3f447a88955467babe4f0b : SUCCESS in 3m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/22900b32145f41e485b38c2b2483e034 : SUCCESS in 3m 04s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7cf20c5aa5234033babdf037f5bbfdb4 : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/feb735e6092f45f8bbc0dcf6e246f8a4 : SUCCESS in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71dc1e31a4474bf8bda0aba543a805ef : SUCCESS in 4m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cefb1913a2e246ba960b2854bddc951b : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":3},{"id":"8346c1fb78ae276d3ed6ce72b3c742dd6b9b0a95","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-23 14:59:22.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86aa6e4ad4f746fd8d73b81fd74a4952\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/cd5723aca6a44d238377ce2ef214cd82 : SUCCESS in 1h 22m 49s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/d7cd48a3ceaa4149a8285ef946e81153 : SUCCESS in 1h 58m 15s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/b1cf8814c0bb4c1ca9ff3ce767b3bb20 : SUCCESS in 2h 14m 02s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/a0447a3ff42e4260a0584be70d2f91e2 : SUCCESS in 2h 06m 01s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/cebb7abe20714060a2e97fb0db036125 : FAILURE in 2h 14m 27s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/2b47d2f6df20453188acb359e2d090a5 : SUCCESS in 2h 30m 35s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0fec2f52663d46e582cae5d15ae3398d : SUCCESS in 3m 40s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/98f4ed1d63134f2fa574bbc197ff8b6e : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/bbedc324dee648e99bfd703d274ee797 : SUCCESS in 5m 23s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5fad65239f87412771570cb11065416449e07b0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-05-23 18:20:39.000000000","message":"Uploaded patch set 4: New patch set was added with same tree, parent tree, and commit message as Patch Set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"3470ea1d71ccbacbcfcaddb7d9ce705123069816","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-23 21:50:40.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a934bd28762c4ab7a9fde13fa7394511\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/29048b3a0dc943cf89dd5be3c4694721 : SUCCESS in 1h 06m 25s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/2eab355756974932a4eed2a114f52ca2 : SUCCESS in 1h 12m 39s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/892fb095590c4cf783d81c8d7c2f1ad4 : SUCCESS in 27m 49s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/4d829208d51b4ba6b20722f68a7512ab : SUCCESS in 1h 05m 19s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/0fd0479a7ae24034ad3e03259860aa21 : SUCCESS in 45m 50s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/04921c2a3b854a89bf90c1367c9a483a : SUCCESS in 1h 02m 46s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/016efdc043b04639a05467f307b50c55 : SUCCESS in 28m 56s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/1264cb1903ad4effa078cd2a24bdf1cf : SUCCESS in 1h 28m 08s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/8067931542d743dab8e7346d451d74ea : SUCCESS in 1h 51m 19s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/80f0f96750664a4394fd47f30cb6a225 : SUCCESS in 1h 23m 53s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/ea2960b7ff5c48569ada8f70b587f064 : SUCCESS in 2h 00m 26s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a9c11a605e7a4cad818efba8b622da17 : SUCCESS in 1h 53m 49s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/9fa9bd4b9df04489b3286905f360549d : SUCCESS in 2h 37m 58s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/e5bfc98064c74cc2b87b63c81678c3bb : SUCCESS in 2h 28m 30s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/8bdeb88b3cb6491b82623a5982fe788c : FAILURE in 5m 17s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/5ae2fa1fbbc644ccbfe9c416303b25d7 : SUCCESS in 2h 03m 41s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/5c022798ae1f440ba8455a9a05585f38 : SUCCESS in 2h 32m 47s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/0886cb65613b49c2881be0e43e4a3e9f : TIMED_OUT in 3h 03m 49s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/4129d8f7f9ff45d084926ab2e140fda4 : TIMED_OUT in 2h 32m 48s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/b2a76d04a28e40d9bdd3c85658a8aa24 : SUCCESS in 1h 13m 37s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/15346edc6e674daeb92dbe73e376f43b : TIMED_OUT in 2h 32m 55s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/9f0d197475134a2fb0813ce3fdec34f5 : SUCCESS in 32m 47s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/56001f8155cb45fa874c72ba9fe7fc4a : SUCCESS in 42m 10s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/c80ace55400a4d75bfcf0186954eee5c : SUCCESS in 33m 15s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/a944231c32764246918ea4f5c4e38aed : FAILURE in 43m 58s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/aad719df1c0642968c39144cba59bf35 : SUCCESS in 1h 01m 29s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/978de8ed91844353954a9db3e4ec416d : SUCCESS in 1h 34m 22s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/151b09e3b3ab47409eb9d244bdfdec5e : SUCCESS in 2h 02m 04s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/9f23d5bdae1541ac963236ba2c3bf77b : SUCCESS in 1h 58m 28s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/10fff7f2c9b047d88a6f077cac0ac936 : FAILURE in 2h 10m 09s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/fb1878c83e764f8aa858740fd199bed6 : SUCCESS in 1h 26m 41s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/d906bcb53d294033a0ded4540efd3be0 : TIMED_OUT in 2h 33m 09s (non-voting)\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/fe8d7eb151d549429a707a0a099ab309 : FAILURE in 2h 08m 15s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/59e98b2171e446c2a55a9a9e98cbbcfe : SUCCESS in 1h 34m 45s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/ae6034a746f94a79ba63d983e6ebe154 : SUCCESS in 1h 55m 10s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/4502b954fd2047d483e34f4a4adf4718 : SUCCESS in 3m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9351701c399480db352dbd119a4e82f : SUCCESS in 3m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/58f0977e646c48b68cdb68b805a0b8f2 : SUCCESS in 3m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e6cfbea78e9e4975a0400eaf1b87d39d : SUCCESS in 3m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b94fc30fa1844c8391f7d42115af34f3 : SUCCESS in 5m 03s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/de9644db4618467e8ba151ce0f3d6ba9 : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05b31ebe4ca940eb9e93c4910f9c01a8 : SUCCESS in 5m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/86a0685ac9024884bcf538a5d55e2805 : SUCCESS in 6m 19s","accounts_in_message":[],"_revision_number":4},{"id":"b824738983cf1ab49e687f55adfe44e8fd07ea33","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-23 22:48:16.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/736c7815e8ea462d9451f7977a18b3a1\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/74da4926caf948ba9fc5d63c2ec1f903 : SUCCESS in 1h 25m 07s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/b55dbf2d70304d698aba2e454e2689e2 : SUCCESS in 2h 10m 16s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/f83d2b22f26649909771b52d0b9af8a0 : SUCCESS in 1h 49m 24s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/ca1684c2957c492d8ce8b41d548bac6a : SUCCESS in 2h 30m 55s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/eedec280cb754093a9ad2ed1cb6ed552 : TIMED_OUT in 2h 32m 12s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/9ca668717c8b4b0ab51976a5362c99e7 : SUCCESS in 2h 37m 49s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e8b86060207c4326ad2db0945023fd7f : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8c03058337a643679f7a9ad9ecde1966 : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/c7326a5432df4f6391119b180f91d459 : SUCCESS in 8m 20s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6425f223e5088d6b238f9362862d84b0c8df3692","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-05-25 19:08:09.000000000","message":"Uploaded patch set 5: Commit message was updated.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"006a429b9dec2c5f9f1934053e1a97ec5d9c14b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-05-25 19:29:45.000000000","message":"Uploaded patch set 6: New patch set was added with same tree, parent tree, and commit message as Patch Set 5.","accounts_in_message":[],"_revision_number":6},{"id":"e6c86dd07a369c6ce5230bf8bc58e0c8c7126a9e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-05-25 20:47:48.000000000","message":"Topic ULTIMUM/master removed","accounts_in_message":[],"_revision_number":6},{"id":"466f313ac6d186248ab45109a9db48a54e3db695","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 22:39:23.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/eede8bc3c1984dd684be58205e9379a1\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/3077822715db4cf285e86b1e0507513b : SUCCESS in 1h 16m 04s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/0f3a4b6b66fd4cffbd7d56dea2b61830 : SUCCESS in 1h 23m 04s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/ecaef5e29af346bab2d7f3cbce6802d2 : SUCCESS in 24m 24s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/b532b34191874ac4acab26a35d852588 : SUCCESS in 1h 18m 02s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/587a944110e640e29f050b1330c20910 : SUCCESS in 32m 46s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/78a87acafcff4953aa40bba7299ff674 : SUCCESS in 51m 56s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0f9be147931648a4a39fd04e37cc6dd7 : SUCCESS in 21m 52s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/db063726cb9542329d246a6f9405f8f5 : SUCCESS in 1h 23m 39s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/e7b609f3ce4748a890d785107aa84679 : SUCCESS in 1h 33m 01s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/995267f7dee44097a9ba00e1b8139546 : FAILURE in 1h 34m 03s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/a3811bf2168244539b38954b0bbecfb4 : SUCCESS in 2h 07m 46s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/76be211d6b67425dbf4ed14a1831e2a9 : SUCCESS in 1h 25m 57s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/ed5a8c870df4480fa66762dbabcc3dca : SUCCESS in 2h 03m 20s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/c2c0464bdc9e4004a200af0988c231db : SUCCESS in 2h 07m 30s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/2c87b793b88e4eb7a115661eae29f96a : SUCCESS in 2h 04m 52s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/9cb92637ae044a85b2188a939190dcae : TIMED_OUT in 3h 04m 02s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/3398e5d5f8434d49973b737fd84705d7 : SUCCESS in 2h 03m 26s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/f2c15dc83a7c42449aabb553e896c831 : SUCCESS in 2h 53m 56s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/3f31baf6c0244d0c8092c5677abfbc18 : SUCCESS in 1h 55m 29s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/9f2d67e86e6a427a821790fe672677be : SUCCESS in 1h 59m 44s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/33e2dc3cb1964b7ea22d50f28d9e0c15 : SUCCESS in 1h 57m 18s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/e17d1ee62fdb4cd8a109283a9b98606e : SUCCESS in 29m 52s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/e6687d0a722e45988fba8e1e15f4d46b : SUCCESS in 30m 44s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/c04adda85c7a470c9d7cd820ecc6d5fd : SUCCESS in 35m 37s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/6e8b4df9d6f045289c421e156d0ff3cb : FAILURE in 1h 02m 31s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/fd4bfddf621e4352ae3153d7e9049a62 : SUCCESS in 56m 26s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/f342b65717764961a625e1bc6529adba : SUCCESS in 56m 12s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/4d45f4b5d6764023948b8a67186a1b6b : SUCCESS in 1h 28m 19s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/71629f3b64c9480c8f1f5b2f8f7870d6 : SUCCESS in 1h 28m 46s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/7555e8e0ced04d1ba3177f9d41092d39 : FAILURE in 1h 44m 32s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/e01c841396354258b57bd0545971e553 : SUCCESS in 2h 22m 50s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/bdcb72308dff4ef1aabe05f134ebc406 : FAILURE in 2h 10m 57s (non-voting)\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/a7d3c027dcd14bcb923fc719cdd1fdb8 : SUCCESS in 58m 24s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/67de17542941408b9c407233446714d3 : SUCCESS in 1h 03m 41s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/88f0bccbfb9441928a89ba988803d1f2 : SUCCESS in 1h 48m 28s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/97d6a83225b649df9426fecc062f5f37 : SUCCESS in 2m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f6d12d76aef452a9015bd1b2a596f30 : SUCCESS in 2m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cff1f20c1d46422797aaa40d27744dc6 : SUCCESS in 3m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3c1061ea323a440b932b53abd2fb44ad : SUCCESS in 3m 44s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/04da1b967d014b3fab470959ab864795 : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5c28725b982f45d09c6f86af978bb27c : SUCCESS in 2m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/978c63ff3467403e94b70029adaa70a6 : SUCCESS in 6m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/720c515e82d6436cb4fb450dc8b44bc0 : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":6},{"id":"2f988ef3efd51613740b1a6f9a3de1d3980bc094","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-05-25 23:25:25.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"53145d0506cb9680bbf8009f0e684df1d24fb046","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 23:32:16.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/977b2c20b19e4fa4935f1ca6fc4da74c\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/6ff944261d3c46c6a6044e0f5723b472 : SUCCESS in 1h 41m 04s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/64654dfdd95b4387be2d5116d35d04a9 : SUCCESS in 2h 12m 52s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/dc0a5303cbd84c70a5e40fb5f08d0acc : SUCCESS in 1h 38m 36s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/2f205455e673403eb2902a99e8b5748f : SUCCESS in 2h 18m 54s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/b651be808cdc497fb8d5ce5dd92584d1 : FAILURE in 2h 31m 21s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/f565a902f44a4f65a3c12579edb1e8e4 : SUCCESS in 2h 43m 08s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e612a39f9e204248971a174da8f2cf6c : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/cc7da7baf57a43799f0c8788e4f6726a : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/4e717cb3955942169ea57edab129bc5a : SUCCESS in 6m 42s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"bf7143f3888820ad6ad55e0c7cb732dc291dc9ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 02:32:37.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82391813e8d6445fadd4147859216e48\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/33959ef1a5274e97be880f79836464e7 : SUCCESS in 1h 02m 57s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/82106694dfed41b499d8d88254070c89 : SUCCESS in 1h 04m 14s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/9a17df21d05d4a04a224478b8d0cbdcc : SUCCESS in 32m 46s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/1da72e07a612453d87ce2a342e312841 : SUCCESS in 1h 23m 16s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/dfd94e79aef641fc8a771b508f951966 : SUCCESS in 1h 05m 53s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/359cb64364af43b49a4a037adb6718aa : SUCCESS in 1h 03m 54s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/95698abb91fc49c6b96841f624ddcd00 : SUCCESS in 21m 11s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/dfa5694289e5427fa6cda6badeeb1868 : SUCCESS in 1h 36m 30s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/f56e1e93b7c54c478cef4b8f171ec059 : SUCCESS in 1h 33m 20s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/01544cf8877c45c08e89d62048621b65 : SUCCESS in 1h 52m 27s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/372619941c47428abf66bb58543c653a : SUCCESS in 2h 12m 35s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4946cd325fd54331b9b332d764c3869d : SUCCESS in 1h 36m 06s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/571ab44b1e10409daef2b8064a5f5e84 : SUCCESS in 2h 22m 17s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/cd7cfae1e5634edfa6156ae9d51c5467 : SUCCESS in 2h 21m 08s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/e4a11eedd3b24dc0aabfb6aba993fbf7 : FAILURE in 2h 06m 30s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/aabe85f3cfb3465a9d083d219f13b7ed : TIMED_OUT in 3h 03m 08s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/12b5b72d0dcc4fac84ad0ce574e2acfa : SUCCESS in 2h 27m 40s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/73193f627fdb4c23bb24a67b0dcecdfe : TIMED_OUT in 3h 04m 48s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/d23757dfd8a74534a23e1a765dfe7436 : SUCCESS in 1h 48m 10s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/7cb5af8f118a4ba79ea9b3686a154794 : SUCCESS in 1h 58m 35s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/719968df800b4ba8a5d9fab382db2563 : SUCCESS in 2h 15m 39s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/bf212abd247d43d9bc09374d29a8918f : SUCCESS in 29m 14s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/b85b987d960f4cb7a2f294b78f395a05 : SUCCESS in 30m 54s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/d1c8263c02164355ab3204dc29147a6c : SUCCESS in 29m 15s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/0e76eae03ab141e193689bca03643eda : FAILURE in 1h 07m 53s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/62ab51d1ca254b828a716664cbe06f15 : SUCCESS in 1h 23m 11s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/64b9ab097cb1466a83fab0bd6397842f : SUCCESS in 1h 39m 12s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/1b34f301e49e492caf8259b8bb39fd5b : SUCCESS in 1h 27m 44s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/9759312263b74e599251365bda5b130d : SUCCESS in 1h 30m 03s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/53fc59340f3c4b0bb21f86d7e05483ea : FAILURE in 1h 43m 28s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/51bacda43af4430998353958d55d80a7 : FAILURE in 1h 19m 58s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/2b32d5724c134d168870faae5a140c97 : SUCCESS in 2h 21m 05s (non-voting)\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/8710139929bf463db4c95d2f55059fad : SUCCESS in 56m 39s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/fa9529347a924fe3be6fa5e248650813 : SUCCESS in 1h 02m 42s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/f1ea773f0a704d3bb77d5f1e5befd08c : SUCCESS in 57m 52s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/5c8139e5153d4ea9884855c174d866d1 : SUCCESS in 1m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c79af5c39f847e1aba90f9d04145446 : SUCCESS in 3m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/67b12a52090b4ec78ad0c5c79137096d : SUCCESS in 4m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b9a27a7b2bac49deaebed836a080ec1d : SUCCESS in 2m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5599697bc80d4aa8a58cf4118ca20015 : SUCCESS in 4m 39s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ca4ea7bebcb49a1af9ffd892b95ba65 : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b69480d1dc404710ba1651dd53c7717c : SUCCESS in 7m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ba1c4d74530a4183a4331a58d09a26d3 : SUCCESS in 9m 05s","accounts_in_message":[],"_revision_number":6},{"id":"994bae026cd0f5784b2a466a026363913946e533","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-05-26 09:08:44.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"85416bd31c38c5604d9038112a6b738937349cbb","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-05-26 09:30:34.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"71995b85c7f705fe414aa981a4e078a44bc1768b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-26 09:33:54.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"a06c7a72903a99e7b88cd8d451c217b5c692227b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-05-26 14:30:42.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"c7bf41e0b6ebcbc24114bb4c9f4f2a744de7f035","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-05-26 14:32:35.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"d58516b27df4e5b0b5806ef278d2b32f30c72132","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-05-26 14:33:09.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"c11c106180498119acb5baf6bd2723502e61b409","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 17:38:02.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22c140d87e6d4d908771e21a3387b7cf\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/87e3c81407464e0db56a8a5ba45ea8bb : SUCCESS in 1h 08m 41s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/8cc9390abbcf41c2b9f19bc34b4dc89f : SUCCESS in 1h 26m 32s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/61fd3b186ad845659bc563278f575b14 : SUCCESS in 27m 57s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/b16f068064994818a078c9a3e3b2effd : SUCCESS in 1h 27m 22s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/34427ccfe9824ce897582322fb88db1a : SUCCESS in 38m 45s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4a5ac729deb14316b637587a87c4aac3 : SUCCESS in 55m 57s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6a98b849a86540fe93e2cb2183b87411 : SUCCESS in 32m 47s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/91159738149e4e00b83c29599e65542c : SUCCESS in 1h 21m 27s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/2c4b18105a18425281beb46cc6e4f6b1 : SUCCESS in 1h 21m 16s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/77f56dab0a0b4a86bd97fe578773b13e : SUCCESS in 1h 31m 17s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/2eb25ba11ff54d3291bc4b999f4534c2 : SUCCESS in 2h 31m 32s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c97fd8b4d7ee4e50b0dfe92672043179 : SUCCESS in 1h 30m 04s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/93b25c379359415d9fe2141506ae6bf8 : SUCCESS in 2h 09m 26s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/2b6b0518768a4e5bb8b04764dd4efc03 : FAILURE in 2h 03m 29s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/e6782accac904d88887ad4508bfe8cb5 : SUCCESS in 2h 12m 06s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/81d0ed5000b84e089a1eecd62011e33b : FAILURE in 1h 26m 55s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/e75357de69fa482ab9b47177a0ae599a : FAILURE in 1h 44m 56s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/13bfbd6b645e409fae0cd5b23a9c9f63 : TIMED_OUT in 3h 03m 28s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/82791cbf15b3427ab014f131227a51af : SUCCESS in 2h 11m 07s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/e091c70d3c4047d0b9eb94b33058b431 : SUCCESS in 2h 01m 38s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/1890e8ee711143f49b1024abf61fa84a : SUCCESS in 1h 55m 11s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/d1dcb030f7634631a3c0b184c0b0ad49 : SUCCESS in 29m 29s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/30e931e7c4f04cf08f1f711e09e78cb9 : SUCCESS in 31m 30s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/31c902d6401d4226ab5b67bcf8eedd19 : SUCCESS in 38m 03s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/c4d44931f90744a7b98d544aa259b24a : FAILURE in 1h 08m 11s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/aabd9dee764f445091a795f821920573 : SUCCESS in 1h 34m 18s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/f0e713135f2d4deda133e174a065c863 : FAILURE in 30m 01s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/90fd2ad3b39e40228ece6e6b03b9fed6 : SUCCESS in 1h 31m 17s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/829d2708167f40b48203ea2ad4a6598f : SUCCESS in 1h 26m 44s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/f76e9d6d36664f6e8aa938ab32899a81 : FAILURE in 1h 45m 42s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/ae42f61fe319480f82ae090397895f0b : TIMED_OUT in 2h 33m 37s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/c75eefa755aa4d919618618fd7bb4726 : SUCCESS in 2h 28m 10s (non-voting)\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/bee6988d99c04226843a035dffaa269e : SUCCESS in 2h 06m 59s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/d56912ca13f7431681d6794af07b0e12 : SUCCESS in 1h 07m 28s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/24a2d1b46b3c458a97974737f13db9ec : SUCCESS in 1h 47m 52s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/d64dad63df3641e1a75181178729f240 : SUCCESS in 3m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e261ac6224043d9ad7d17f9743092ca : SUCCESS in 4m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/878ffe7e9a674ab3b20a729360fc55d9 : SUCCESS in 3m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c695a23d115742eeb4460449a5348800 : SUCCESS in 3m 51s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/417a6c6642a1468daf1fdf916026d85f : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb0b9b7944b04adb994af4f67d3e8648 : SUCCESS in 3m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe8118ad265a4e7a9fa979546f69d102 : SUCCESS in 8m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/44bdd57aea014d22a19300b22b8bdc2d : SUCCESS in 9m 54s","accounts_in_message":[],"_revision_number":8},{"id":"94d67f67c8bdf76b723f8df58878d32bd89c4204","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-05-26 17:48:04.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"05f85cada3dc1ad9105dbe374a44cb9982d6335f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 18:56:49.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2ea32aa483e4881924a33dddc79da97\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/7e5a5c11b61347a9adfb20415f2ec862 : SUCCESS in 1h 24m 09s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/ba8552c4cafb4bf9aeb637d488423874 : SUCCESS in 2h 31m 08s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/5b3fd264fc6a4d47b50d0afd96614471 : SUCCESS in 2h 08m 44s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/32e5092838184981b549ac7a0fc2f97d : SUCCESS in 1h 13m 19s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/e6062b285a0a4f04ac3955f04229ddd3 : TIMED_OUT in 2h 32m 08s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/cae8a6437d884505ac460c085bd0db03 : SUCCESS in 2h 07m 56s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5b33cbffbac047a985d54c5f3a557bf8 : SUCCESS in 4m 51s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/55bba6628aa04b9f8411a4aa91b3a0bf : SUCCESS in 4m 25s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/717ae22ad8e546d3870efe34c183be94 : SUCCESS in 9m 25s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"6ff7371c95a5de19e5c992019e6335918e6468ca","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-06-01 11:22:46.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"b8edd7ef60ae0f3feab5a21b210f1c882edbc140","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-06-04 09:38:12.000000000","message":"Patch Set 8: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"b6d7d248f52785194d23d14fefa2a2587a460b1a","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-06-08 08:33:19.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"a9cd1e9bbd38cee1a1d1fc10199e692e45e0daf6","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-15 14:26:31.000000000","message":"Patch Set 8:\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"27b9f3778741e525f86fb6979a2c2b8ba10f0ce2","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-15 14:26:35.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"a087a56c5650453d10ac4105e8cb30a97afca1d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-06-21 19:29:38.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"c2b747ced7024ac5952609e9976abff38317311f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-21 21:02:37.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1757ca285cc4ba6964d785a0aec6783\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/584674a23a1d4409bd9fd6d047c98317 : SUCCESS in 45m 48s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/9e5dda4d1f9e4c5ca699fdd3c16b38f5 : SUCCESS in 43m 45s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/ea0ee7c550954b9b9fdedd798177f079 : SUCCESS in 46m 29s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/de17e49f2b954f1a899ab3cccc18e77c : SUCCESS in 37m 55s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/b129fd9fe5d9456bbc059760f624218a : FAILURE in 1h 30m 25s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/4894631864694f43ac61094cabf29666 : FAILURE in 13m 00s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/caf6f02ea3e047c8b24e7f824ba3453a : SUCCESS in 3m 59s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/cdb5e69044fd40f0afdef38a6ed5dfa7 : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/444ed266b4484c8798a9263a0de80199 : SUCCESS in 10m 32s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"cafb3f406e823d2b9d881d9db9127f5959c411c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-21 21:15:55.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/933984673bf444329e01ed81b14e62ea\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/130ded6d6fdf42ccb3f7c8d8c1f5f341 : SUCCESS in 15m 15s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/73bbe6c94fda498d9ab99cac4daa6a27 : SUCCESS in 17m 11s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/f04b3dca9cd94087bb8d1661fc1f45f3 : SUCCESS in 12m 56s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/3702822a3b044803b7b4f477d94c51e4 : SUCCESS in 26m 43s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/38645ec602004b62a721f7bdba4a887e : SUCCESS in 56m 31s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d3c97a8cbdb64ae69c9a7b12cf9debe1 : SUCCESS in 13m 21s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/dcacfb2c300a4bcbb122d1032519dbbe : SUCCESS in 13m 47s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/3eff54636740442e8bca06bb953f8ab7 : SUCCESS in 1h 04m 59s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/1e0f7207b07547bb9a53b31b3e7f8d6a : FAILURE in 6m 22s\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/f99ff1ecd7154e42a51fbc7814bfbecd : FAILURE in 6m 35s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/07832fddf15040f0927bcd97ee0e50ce : SUCCESS in 1h 04m 15s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/3f93a1dfd1b043d790bed4d4381d8fda : SUCCESS in 1h 10m 41s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/291af4f6aa644473a74a97658a95c20a : FAILURE in 5m 49s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/4b3b9236727d4b3f8ffa46ba4fde0562 : FAILURE in 7m 02s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/cd3cf13c5db54dcda8eec2927da6eb69 : FAILURE in 8m 03s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/5979ae46f69344f1b97560a6167506d4 : SUCCESS in 1h 30m 42s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/fe3c3bfe52b844f887d25d6e0188d9a6 : SUCCESS in 1h 37m 34s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/d2317ad66e324e0d83b77670986091aa : FAILURE in 8m 50s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/ba6028e148b04ee592ba080cc69d3ab6 : FAILURE in 8m 06s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/c8ab93cf75264d3290f4e82d3c847fb5 : FAILURE in 6m 44s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/dd8bb80790b545679fcfa361a3bf302b : FAILURE in 44m 04s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/31db0f8dbaef471eae87457367b1a6eb : FAILURE in 8m 22s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/e99e58e67ec34cf0aaa97b0952fca0c2 : FAILURE in 7m 33s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/a5807284ea074e96b69ea4edc702c822 : SUCCESS in 23m 10s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/52122c4bfd064706a2591c283c55c194 : FAILURE in 8m 27s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/dede3dcb6dba4672b4ab2d24087b374d : FAILURE in 6m 09s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/c18788497a544f379e11b22be090b2b7 : SUCCESS in 37m 54s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/10646621e380428fb295ebebd015c5ee : FAILURE in 6m 24s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/18314d87c56a4c79838e8c7f4d533fb6 : FAILURE in 6m 42s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/1872665004ec4e6286e3cbc0be6c9a2d : FAILURE in 6m 47s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/18cd843277394f9d9667971793b36ee5 : FAILURE in 7m 11s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/e0dc6f38480442a48918eb7c471a758e : SUCCESS in 1h 44m 03s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/cd5e580ce9b642788252464be97c9914 : FAILURE in 8m 01s (non-voting)\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/899f471348bd4d1581b52be7666dd89b : FAILURE in 7m 44s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/f1296dfb7f804d5d90e5272e635363b5 : FAILURE in 1h 14m 32s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/a4f95131877d4317a87c107a6fbe5c65 : FAILURE in 12m 27s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/93d710f09e4c46bf9671d46a05722d92 : SUCCESS in 2m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5df0bbfb579541ad8be0853082c804fa : SUCCESS in 4m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a3f026498e6f4184ab387d95c90f99ec : SUCCESS in 4m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/579be394a4dd49faa5a642b4f06583ea : SUCCESS in 3m 08s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/04cbc0b50de0460cb003899514a5aec0 : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7c55278c439541d4a5037a493d8b654a : SUCCESS in 3m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f66d46e117e342ccb4cf4bbd1e0b0b44 : SUCCESS in 7m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/23dd9bd774ce4d058ae458bf7e581455 : SUCCESS in 6m 51s","accounts_in_message":[],"_revision_number":9},{"id":"5be0ef173d85818f2da3f83c2916e26355a7ee59","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-06-22 09:05:49.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"1cbd339c7fa7f99a24805f4f6b61e0fd3e41d43b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 10:45:08.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e378614145d43c084a8f93c93536b60\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/35452cdde8a34fdea707c68c9bfae7e7 : SUCCESS in 35m 42s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/db32053b423a468db6c09a33cf477aef : SUCCESS in 43m 24s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/47ec3e4a99884bc7a7096ab5cd691b66 : SUCCESS in 38m 39s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/441504791db444fc8d776c60bb0ea030 : SUCCESS in 45m 45s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/7d0e0d9adde64976bb21a9a10cedbab5 : FAILURE in 1h 21m 01s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/490373bfa46644aba25e8292e6069a21 : FAILURE in 10m 58s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/bcf019e248a541599d60a20f533cc77a : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ee6100e20b5b4d80be01ea5566f794bd : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/606d348e0aba4c7da5a9d36eafa81cc1 : SUCCESS in 9m 05s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"83aaa427d010f35b2cafa55603317fe60c305827","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 11:31:40.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/04dec669afcb404cb06f5dc530f1c634\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/d375f13e8abf4242b36691d1f9f04c6a : SUCCESS in 15m 20s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/0d946af8a9b348b69272bc31153e7a78 : SUCCESS in 28m 11s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/e5e0fe11574544a6b6ed76b65aee9cbc : SUCCESS in 12m 27s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/165d01d49ce441adad118f55c2f195bd : SUCCESS in 26m 29s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/d1fde9efdd434e9384b512b6d36066ea : SUCCESS in 58m 20s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1e3d4ca3445f4066a3d45f667daf9c2f : SUCCESS in 27m 59s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5f79a30ac5b7483dac5af5a67a4e3124 : SUCCESS in 25m 02s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/7eddc2405b0f44a5883c7a272e0347bd : SUCCESS in 1h 07m 54s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/8cf9adfdad694b7a8dec028e8df086b5 : FAILURE in 7m 21s\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/608e896cc09a49be9fdeec8ab8d0c157 : FAILURE in 8m 15s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/7a50b77ee13447b1bdfb3c2fa36d09c1 : SUCCESS in 1h 06m 55s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/8f18a83e8e1b406e9308fb4162647058 : SUCCESS in 1h 35m 07s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bf4645dd6c084c2faafdbde5956d1843 : FAILURE in 7m 12s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/5d5e659179fc49168273a0787e76e7e4 : FAILURE in 7m 02s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/d79cdf01ffb643c8ad6e1b83973b93c1 : FAILURE in 9m 08s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/74cd5f58fd7640ecaac857fd3fbd9987 : SUCCESS in 1h 35m 04s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/72747e63ac884d32a58ce963a220faa7 : SUCCESS in 2h 22m 41s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/9e53d519d0d94d3f9f9a4d5e4324e7c5 : FAILURE in 9m 01s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/ea472e8b6b8c49429450a5aaea4c90f1 : FAILURE in 10m 41s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/291a0be42120465487b2a833c865ed56 : FAILURE in 10m 03s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/bc57e53a653a4cc9ab6f0a96664d38cd : FAILURE in 55m 20s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/7de736774509485c9808578ea7f4268c : FAILURE in 10m 28s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/cd96776c314347f89d84fb7d420c7832 : FAILURE in 10m 36s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/9ce9062fde204a179177eee8dcbdd575 : SUCCESS in 27m 52s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/d2ccd034729f4227b7d1a2ef38173199 : FAILURE in 9m 21s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/07240ddf54f04f21b19104e46941d7eb : FAILURE in 8m 49s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/17082be6910d4d67bedf06a3bddaebd3 : SUCCESS in 41m 07s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/d19c3ee44c1e4750ae2b1608cff127fc : FAILURE in 9m 19s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/a44f60c434364779b21acb0d400d8d6f : FAILURE in 7m 23s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/e15ae0dce45c45d099df2e6f1c7f2c7c : FAILURE in 7m 13s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/0c65b26255fb497cb08dbfd4c5096289 : FAILURE in 8m 32s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/39d318ca61b34f0da4d0d25769ce044b : SUCCESS in 1h 46m 58s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/ee6032b2f682499cabfa4496a19775cf : FAILURE in 7m 58s (non-voting)\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/9ad797b6ff0c4d388359e06c3d02eead : FAILURE in 7m 30s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/f6c879efa2834c889b9c3e371c9f5ff0 : FAILURE in 1h 04m 30s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/bd28554ee939427494148ae00f833959 : FAILURE in 8m 11s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/6f7b33d408ad4dc58282f1b42f42fad3 : SUCCESS in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fbfd0a10b5444b7bb94443ed20baf76 : SUCCESS in 3m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f2e411e8e6bb46c19ab23f4f782b5597 : SUCCESS in 3m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/142ba440ce6f483c970393e6c0483a47 : SUCCESS in 2m 43s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6cebae25dab5459f9fdbb895e1f0549f : SUCCESS in 7m 35s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/64ffd97654e14817bdf57f5898e09ee2 : SUCCESS in 3m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d72c900cd2c546569e75439c2eefe984 : SUCCESS in 6m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c654d3712f004ae6a7950edeeeca13dc : SUCCESS in 7m 13s","accounts_in_message":[],"_revision_number":10},{"id":"99d527128f1bfef2eff565964d658f19e8e94a18","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-06-22 14:14:25.000000000","message":"Patch Set 10: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"086497177e11bc0cf4505013f8fe4a73c2d1ca7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 16:49:02.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e9547e541794bbabf8e438b0e2ff0ac\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/f392fd07a3ce47518f9c15e315f20d48 : SUCCESS in 21m 37s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/f3c02865905143f3aa91ebf2cee47a7f : SUCCESS in 25m 05s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/26c588081baf4c48a21001d8f6fb0603 : SUCCESS in 22m 26s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/cc4a411943574d50a6ba083569d1211b : SUCCESS in 18m 03s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/abd76b75015842c6bfd6c25c1ad71050 : SUCCESS in 35m 29s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/49c350c1d4f44023b5ee22fd41775538 : SUCCESS in 26m 17s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/bbad18ee0a1e4704b455da56ac24d2d7 : SUCCESS in 14m 44s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/e55d6da1a64d49e1a38626792c1bcfbc : SUCCESS in 1h 08m 15s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/ec19c7fc032441ee925ea11a7779c017 : SUCCESS in 1h 05m 49s\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/5b39c2b7582747e9bb34fbd1adecca8b : SUCCESS in 1h 43m 20s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/b5f83a01d41141bb81899d1e973ce642 : SUCCESS in 1h 06m 55s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/89c5276c7168455a8111065a4c00450f : SUCCESS in 1h 35m 27s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e3212b770a034990b693219795c06f1a : SUCCESS in 1h 05m 12s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/e73387c879954215bbe84b4d3b2fb1da : SUCCESS in 1h 40m 44s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/f4b57d08600742f3bd20af9f5aca132c : SUCCESS in 1h 39m 02s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/cc4941cb57a749aa93c5171d66bba607 : SUCCESS in 1h 42m 36s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/3d7b5695bc8e44d0af7d6e98cbd38173 : SUCCESS in 2h 22m 23s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/751ef8ddcc0548ba99dc65574b61c325 : SUCCESS in 1h 48m 57s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/dad981a3302a42a8bd1f2336ad9838c2 : SUCCESS in 2h 30m 32s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/3e740d3d3a8e4411ab4bffffd64ac70c : FAILURE in 44m 12s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/4928934a909b4b31a2c51524df54f559 : FAILURE in 45m 17s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/04905b493b0f4678aee551cdc698449f : FAILURE in 45m 49s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/08ebfb156dc04924a17482bb7e4aa716 : SUCCESS in 26m 54s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/dc5b78b4f4c040ad83e6551d626ceaff : SUCCESS in 25m 45s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/2dc8d98591c84a3fa457be6a1253eb6c : SUCCESS in 34m 28s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/85747c5487a44825b44fcc3a3402f1b8 : FAILURE in 54m 29s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/4eca6abd15b0453eb5b740e97867e21e : SUCCESS in 1h 25m 37s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/c97ba4f0c7c949d499e225d71512b6c8 : SUCCESS in 1h 18m 50s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/9707a508a6dc4614957979044a100674 : SUCCESS in 1h 06m 26s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/51f22ae27a1441f0b2df7ffd40093abd : SUCCESS in 1h 07m 35s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/f44ed6cf2d134a75bab41fde45420fab : FAILURE in 1h 19m 22s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/495b5dd7613e4addb994da411c3e6d08 : SUCCESS in 1h 47m 17s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/5cf85d8bf4b447f19fb251baefbe51fc : FAILURE in 1h 21m 20s (non-voting)\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/99dd3663123e48558d55d3c69250ed46 : FAILURE in 1h 17m 44s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/e2413ecf317244a3804891495cc20bb3 : SUCCESS in 46m 27s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/a7aae00f24a241879ad9fa8be6f7028c : SUCCESS in 1h 09m 04s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/c9dfe80c9fed494dad036068e6bfbc84 : SUCCESS in 2m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76061d2b1d474f5cb8376bd8157a9208 : SUCCESS in 3m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1d09f2be3fe842b08d0ea94a90a0a876 : SUCCESS in 3m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7214a1515a0b4a4cabd48b30f2457a0c : SUCCESS in 3m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/621b0bd70ac146bdbe99346c3655f690 : SUCCESS in 4m 10s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e861f433ad394d1aa5c1365a28450bbc : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fd326e323cd43f9bd5de2d2cfbbce89 : SUCCESS in 6m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f4f0d694c33c4546ba311fd20353c0fa : SUCCESS in 10m 33s","accounts_in_message":[],"_revision_number":10},{"id":"a4b81ad5a31d2077af1711200cd9343349c4b687","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 17:14:54.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df9e8400d61e4ea08b636468c30f732d\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/34412b6b2caa45f0a8fa751fed85a391 : SUCCESS in 52m 17s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/f9057ae723e4438cae2d5e9bce3bca13 : SUCCESS in 28m 15s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/12a1ae46d0ba4740b16b01a6fbf2d3ee : SUCCESS in 47m 26s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/d52e6a3108be4749b9e13280aff7cd27 : SUCCESS in 50m 56s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/995a4c74def64d5bbab84dce1a6279bd : FAILURE in 1h 50m 35s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/08d5aaab9e2f43b1b68d15dfe54626c8 : FAILURE in 1h 14m 46s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1e3ed89824274616bd3193f1fc632750 : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4897cad44c5f4d32adbe45b87490d3db : SUCCESS in 6m 02s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/8f895a429d1f4072a8d4f7cff62c102b : SUCCESS in 8m 50s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"bb2b3882d4b48b22e1d8e4158c6ab2be87d8bc73","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-06-25 16:14:25.000000000","message":"Uploaded patch set 11.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":11},{"id":"c87bb4e2b7c0d7d4a1f583e909d9614d8e9dc906","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-06-25 16:22:56.000000000","message":"Uploaded patch set 12.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":12},{"id":"ed087883c9ad3a8c99d9fe53c78856e458f3ba70","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-06-25 16:23:25.000000000","message":"Patch Set 11: Review-Priority+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":11},{"id":"43ceeb9b9fff305d5afec816561fc902938753d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-06-25 16:27:13.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":13},{"id":"09b0837b750e22ffdddc500c0739841097d787e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 19:00:04.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bebf2dd848f442f2a2e408dc56e7286f\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/0c5c1061ffb641a38f55846250b36699 : SUCCESS in 28m 09s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/39c62adf87914b5fad528da05ec36c44 : SUCCESS in 14m 02s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/81ef884d264f426da1be4c7a4a0cd7a8 : SUCCESS in 13m 54s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/dd619e0610a14a8a8e984653a1a0b10c : SUCCESS in 21m 13s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/f656fb67f5f8442dac301d07a55f1cbf : SUCCESS in 46m 25s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8a575e9c74634e1b9cd942b385f7a477 : SUCCESS in 30m 11s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/bdbf373871f742ffbb17237abc63407a : SUCCESS in 30m 36s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/2f43d35d7e984e52ba1c062a2b6b6453 : SUCCESS in 1h 04m 14s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/f3f11b3b022e4fd9b44d023d97ad3790 : SUCCESS in 1h 14m 41s\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/90a5b683f689414f83f5986f0234c6d3 : SUCCESS in 1h 38m 50s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/958ca4c2dd2d42018fa6036c77c25793 : SUCCESS in 1h 04m 38s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/f8a22665d1b1422cbe6395fcfd41bbce : SUCCESS in 1h 35m 24s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b1e55a5167c941b4855cac8c94fc7af1 : SUCCESS in 1h 07m 09s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/9c60e58944884d57961f4d137bd05d14 : SUCCESS in 1h 41m 33s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/1291872e963e4e11b787f17278927635 : SUCCESS in 1h 33m 31s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/9539df27ead04469b9a4056cd5610160 : FAILURE in 1h 14m 50s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/f0bd66842d8840d19711b4cc5854a330 : SUCCESS in 2h 29m 50s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/070f4d4a51a94f75bc491adf98be219c : SUCCESS in 1h 38m 24s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/67a60965b1b744c1b08827a75552695e : SUCCESS in 2h 30m 37s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/acbd05d7fa094ff09de6cc42e1d523ea : SUCCESS in 1h 31m 28s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/f81f2f67420844d993cd7e0297c701a8 : SUCCESS in 1h 22m 16s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/9b4192b707714f2d866a9a22b9821981 : FAILURE in 1h 06m 14s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/1c29da6cbadb4a838ae8d7eeedd6047d : SUCCESS in 25m 48s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/3050fdd69e3c4c5c9c0a56d18d1bc098 : SUCCESS in 25m 05s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/e043e894e5814e4597e0eefd29de82ba : SUCCESS in 31m 11s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/4c7a258091174e6293c0ac6be5b60eb8 : FAILURE in 54m 09s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/4a88829d2ba6414cb8cf0038dc37de18 : SUCCESS in 1h 06m 24s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/dd5192aa99e740ee89a07c714a7a0992 : SUCCESS in 1h 29m 11s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/52af0c763d28451284c130e80d73860f : SUCCESS in 1h 05m 39s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/f479f23a118a4bccafc6f4b1a9b81f8b : SUCCESS in 1h 08m 26s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/f5aa53b854fb462e9c462fbc5ff21ef7 : FAILURE in 1h 23m 12s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/7505dda7777b4fb89087e0fd245e3a5e : SUCCESS in 1h 49m 06s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/5dc2009bfc1840829fda17fd85cf45b0 : FAILURE in 1h 26m 28s (non-voting)\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/9c2a77ed4662436f8ba75ac03cb8696e : SUCCESS in 44m 17s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/ceea32c7f78e40d585fa2b9ddaa228a9 : SUCCESS in 47m 23s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/fc53c2937c1d4e5f99fed67e3174c799 : SUCCESS in 41m 37s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/afe36a26037b41c496bc7a7966152345 : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/979e7aaf1ae8434aacb2bd58e7a0f024 : SUCCESS in 4m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7ed7d3d60571425095452955cb71304e : SUCCESS in 2m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6d752bf7cdf74aa5b3f2fbb7e7b7ea62 : SUCCESS in 4m 21s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/71afdfb262e848508eb9bfb76883273f : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7846a7b52eb4471f995390e3c04e69fa : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c9b632a0ec74a8fb165f6c09574c3ae : SUCCESS in 8m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d3e42961311441dbb08aa6ae0902c5d6 : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":13},{"id":"14c22f1e9e8283ab48861a5a8d9d306408c5a5d6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 23:51:10.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a60bcf25d12e43e1932dcc99392b6986\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/62dc1f659bf0449a889507e60740805a : SUCCESS in 28m 23s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/3673a1d47608424d96906a5f7af574de : SUCCESS in 34m 25s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/e5c03a23c4f94bcd87d5ab6a7e0cdc46 : SUCCESS in 29m 56s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/de3deca4864f408e85c05f896df7f70c : SUCCESS in 32m 26s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/82026711cc8f4933b32a5924ed1307c9 : FAILURE in 1h 02m 54s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/38535805f170479ab2a4a298d7b061e1 : FAILURE in 1h 31m 54s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/842fe1801e5a4e8dafb84519ee1487dd : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b947702c5772494fb0af54f97ee11fdd : SUCCESS in 4m 09s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/4f86091e8afd417bbd3b0405ef832097 : SUCCESS in 6m 34s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"823337b2ad2ff6c08fe340ee82ec29429730a804","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-06-26 08:29:05.000000000","message":"Patch Set 13: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"5feded2a4270177d7b6e66d13b5d9eaf2657b64c","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-06-26 17:45:42.000000000","message":"Patch Set 13: Code-Review+1","accounts_in_message":[],"_revision_number":13},{"id":"579614b7935df94b55dbbaa48512d55bd41a8e40","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-06-29 13:02:20.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"cad9584d4c7b91ed8922048bcef75f0899e80589","author":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"date":"2026-07-01 08:57:39.000000000","message":"Patch Set 13: Code-Review+2","accounts_in_message":[],"_revision_number":13},{"id":"f82bb9fe190beea4d09f1d2a10ca0c68c5e6a73b","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-07-01 11:21:28.000000000","message":"Patch Set 13:\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"f2e818b0203b5dfcf037b8c36a6a55d871a650ce","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2026-07-01 15:51:19.000000000","message":"Patch Set 13: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"d2b4e405df3b4ba3cb5b6655289118a0c134150c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-07-01 16:14:07.000000000","message":"Uploaded patch set 14.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":14},{"id":"673cf5cec84d860be84332f3b4402f253cb365bd","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-07-01 16:15:36.000000000","message":"Patch Set 14: Review-Priority+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":14},{"id":"f679ca85c9fcff717d0635ac20742cdc6c02a1e9","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-07-01 17:10:29.000000000","message":"Patch Set 14: Code-Review+1","accounts_in_message":[],"_revision_number":14},{"id":"1205bb253a61e668bb937bed6e1a54b86d4ddec2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-01 17:59:45.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4db88cae46f42758c0600585deaf030\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/851996afa841408e8430ff69d8dbf1ea : SUCCESS in 50m 32s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/94b4ac0b7d50416da779564eb1fa3025 : SUCCESS in 46m 26s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/5c789b1a730e4f1c888e59baf86df7be : SUCCESS in 47m 51s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/e3498626137341e6bc0a66a72c77625b : SUCCESS in 48m 46s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/3827f332e0154d14985062af5507e63e : FAILURE in 1h 31m 18s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/af21139a31934d87bf92d893e010e3f4 : FAILURE in 1h 25m 38s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/04e9f4ca170f40c389c212f7ab07f313 : SUCCESS in 4m 27s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/16d34f83d33f48d2b8fff8e447bfd1f3 : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/1eaaa1ce6b1a4810b916ba31d5066575 : SUCCESS in 8m 02s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"3b9410db4270ab80d7995508872cb2d171bd113d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-01 18:26:55.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc93400083e64113ba5760853bb30a76\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/1f7efc6d49f846aea1c0d13d619b6625 : SUCCESS in 22m 31s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/b81abdb06b3a4c9886d9b7a1fee22a48 : SUCCESS in 25m 44s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/01b7e1188d034ce7856f1b13ee3b4f7a : SUCCESS in 28m 39s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/01e065cf4dcb482e811c0ec01a1582a3 : SUCCESS in 14m 38s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/c1c6c16df99a43098774da8571b4b508 : SUCCESS in 59m 25s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/100332c71e824c289bd5c5e10c7fe928 : SUCCESS in 25m 38s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/49918ffc88b94cfaaf70490e018d0352 : SUCCESS in 36m 22s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/63a685d924b64da8abbef5f7f71a216e : SUCCESS in 1h 10m 33s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/0d11b8d657a148dd96b821e54dffbd08 : SUCCESS in 1h 08m 13s\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/2c33deef4ab1431aa91feb6e393cb6a6 : SUCCESS in 1h 35m 58s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/968753c16013454fbbebd9fd057cb240 : SUCCESS in 1h 02m 59s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/88fc74bee36443368d19c6f32cbe7a30 : SUCCESS in 1h 27m 52s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a08d7e8f9b0c4376bb2e50e8c53542dc : SUCCESS in 1h 05m 22s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/fdd45b3ef2e246f4a57f5d965c52067e : SUCCESS in 1h 34m 49s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/5122e314462544388404add40622e238 : SUCCESS in 1h 29m 06s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/d6c711b9398048b89527720e233304ec : SUCCESS in 1h 23m 33s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/41795031390d4f54a90a7272e48d7437 : FAILURE in 23m 39s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/a4cf88f365e242c29bc7ce4d65d3ed02 : SUCCESS in 1h 32m 20s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/8159a94717fd4fd09aafda2608ffa15e : SUCCESS in 2h 09m 53s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/37094ddcac4341b083f74a606537fa1e : SUCCESS in 1h 16m 53s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/d3dd7f9518a54aa38f1d55ecea7e0e8e : SUCCESS in 1h 17m 26s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/f069c97c682a4705bfdc280e5377ffbe : SUCCESS in 1h 19m 02s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/878960b349f44f63b0eb5a9316886f6d : SUCCESS in 23m 28s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/d6d7dbbfaf8a4182aaad916d86d98fab : SUCCESS in 23m 43s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/746c7940147e4885ae229a6a1662b9e3 : SUCCESS in 29m 16s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/aaca0e1cd19043449af6aeb04d6ec569 : FAILURE in 38m 52s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/3f923f6cd9124eeebad9f83a40cc16e3 : SUCCESS in 39m 24s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/9bcbe880220545ecbd245f99a0946ecc : SUCCESS in 1h 18m 35s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/9878065dae8a43e6beb9fd78d1b9cb43 : SUCCESS in 1h 10m 12s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/e54de918319540be8a5f1b6a9a723f58 : SUCCESS in 1h 10m 19s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/511e0dfeaa8444dd81f2c993986b6796 : FAILURE in 1h 20m 36s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/002ead604e504f6880e557960db1398e : SUCCESS in 1h 39m 14s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/7da9c0ab8f1a4787927caf3629f2583f : FAILURE in 1h 20m 34s (non-voting)\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/3ddd7eeef3a9442d97ab5121bb4baf99 : FAILURE in 1h 10m 26s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/cf5292e0992e45c2b02744d684b7d86a : FAILURE in 58m 33s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/d22c95578280424fa1988ba11680940d : SUCCESS in 1h 14m 37s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/4239c2fe57c344dbbf5d4e46d2ee4c3b : SUCCESS in 3m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3621567d404a4f17b71453c532d5bab4 : SUCCESS in 3m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dc2a3b413b554ecd98329a22202d6c20 : SUCCESS in 3m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/28ef73183f84433284c20599020e06a1 : SUCCESS in 2m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/43be104384f34efd82b6e567009df6c7 : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b96d859bfb9d459cb5e170342a86e6c3 : SUCCESS in 3m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24cbdbf1d1c14c468163bedaf335e681 : SUCCESS in 6m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b447a358ad794756b6d0129ee9546d64 : SUCCESS in 7m 03s","accounts_in_message":[],"_revision_number":14},{"id":"b331a89ddff490b27e51b3efce54b8d393adfac3","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-07-01 19:59:49.000000000","message":"Patch Set 14: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"3093f2af9cb975bd5a8e591c6d6903484069f31c","author":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"date":"2026-07-02 08:29:15.000000000","message":"Patch Set 14: Backport-Candidate+1 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":14},{"id":"e8af6a7209f58a8f70addd9cf5a2fab3c58ce64b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-02 08:29:48.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"77d8e2c67e5d546fd7ded29d9eae074fa9a5e2b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-02 09:36:37.000000000","message":"Patch Set 14: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/92437d5daa664f7e84368ef69136ba78\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/87a82b31394a43f08f86315616e453cd : SUCCESS in 25m 41s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/e5c23098eec5461aacb5aba1a6cd4c5f : SUCCESS in 30m 14s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/77fee989c4124ff7acc5666950d960ec : SUCCESS in 26m 27s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/a021bff18efe4c59a1c214c9f648f240 : SUCCESS in 1h 03m 27s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c0c394b531234f36a04ac3555a82e3a5 : SUCCESS in 29m 24s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2113aa6d06254277b86ce257fb91bdcc : SUCCESS in 34m 11s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/d73d5c514fca4baba1b0b7baf702cecd : SUCCESS in 1h 02m 01s\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/a82cace4392a4cf2bba41b8e93cb2f99 : FAILURE in 47m 20s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/804d2a68d087464b84a17f0d3671f30f : SUCCESS in 1h 05m 05s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/b594b3693bb846d8b26abd219da60fac : FAILURE in 48m 05s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e1a087907ee141e7af69eee8bc6fb7a6 : SUCCESS in 1h 05m 31s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/710f1c89e9ce460ea0d6334cdeeacf5d : FAILURE in 46m 33s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/b0479f8a8f8e4018a67521f7a8b2a95e : SUCCESS in 2m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e4ee53d35eb46908a89af680792ea8d : SUCCESS in 5m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3e971c2ded88433f8079ca9f68fe10ef : SUCCESS in 3m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9d1c7475fc2f42acab27b29a728f1407 : SUCCESS in 3m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/748bc1b0318146b3b8953fa95492b3db : SUCCESS in 6m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/355d00b2b5f24d4180e01bc1f7bb047f : SUCCESS in 8m 10s","accounts_in_message":[],"_revision_number":14},{"id":"d1865dd114720ec21b3055d13d1b126b05d38080","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-07-02 09:46:16.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"efc267c48835976802eeb0bf81ce3107a64974a4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-02 11:54:07.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/985a62da820c419aac339fc5132a9310\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/3db9d3fe77c24321a6b11e531c7d6374 : SUCCESS in 44m 55s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/67cca00dbc794089af8f80506fb81ae6 : SUCCESS in 45m 36s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/10606e5455f343e29079cd1a9b7ae8e1 : SUCCESS in 46m 46s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/ed7bae6beb0b4f8aac0385cf39778631 : SUCCESS in 50m 19s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/2a02a5ff1bb24b41920e312eca5c9af9 : FAILURE in 2h 06m 23s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/cb45d9daf8a44db8ad2d0b8d1677db7f : FAILURE in 1h 17m 22s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/37580cae4fa84b75aa02c93d40e6bf64 : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5fb2475f061b43d384a3525d1adcc122 : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/707945cbe37a4444b75ae811b6bc377c : SUCCESS in 6m 27s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"93b34c5a06372a9539c6d530b38ecfe8a498be93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-02 12:26:55.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2108683839d0428ba90da404b4cff5d4\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/1367cdda46a349f897125c7c6a327eca : SUCCESS in 53m 28s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/d03466cb46be4957a5e72ba58af1a3bf : SUCCESS in 23m 25s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/4cd0d296eed04c2db1a83335518a9f54 : SUCCESS in 27m 25s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/d79882200cff495ca5423021a3eb3df0 : SUCCESS in 16m 24s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/decdc3e8276d4e28a54b7caafb3fe926 : SUCCESS in 34m 35s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/10d46b15f9f8413598a0865f0ec6d835 : SUCCESS in 28m 43s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/aa821507c9a941fc8a65950cf7f1c4ac : SUCCESS in 32m 43s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/b3803524ebb94339ada03c9db1b8ee20 : FAILURE in 17m 16s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/b5894f5998e540a7a94e40522123e88a : SUCCESS in 1h 01m 21s\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/3389e7c24b464fd087f88a1bdb8498b3 : SUCCESS in 1h 40m 15s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/70b08f6af55646b884e681bd95225d29 : SUCCESS in 1h 01m 35s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/f3fc7c731780408e9f80961494108b1f : SUCCESS in 1h 42m 24s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/75c91747790643b5b06ff41c32985bae : SUCCESS in 1h 02m 46s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/ddaca05eb5ff43b6938a67a1a15cbe1f : SUCCESS in 1h 43m 57s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/2d55a4c309594dd4bf6b3d0dcce38fea : SUCCESS in 1h 44m 00s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/f9e07239cf134a4fa831ec89722f878a : FAILURE in 1h 31m 02s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/68f51feb714e47e6a7cca0d71ade5e18 : SUCCESS in 2h 38m 51s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/d201a08c2fed4036af2dd121dbbf26b9 : FAILURE in 1h 29m 31s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/00a5c5a2658c4164935b120d9b020f96 : FAILURE in 1h 28m 00s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/1b00456476f54b57a4fb7c0861eeb9e6 : SUCCESS in 1h 32m 11s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/8986fceeb261405b8ee90c7b846cedef : SUCCESS in 1h 28m 41s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/a19b08ec3894455dbc4abbb88da59e1c : SUCCESS in 1h 23m 09s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/4b6a0d348a7c4cca9256a63f224c134e : SUCCESS in 24m 56s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/4a91801886294077895113524f37969b : SUCCESS in 27m 34s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/cc6e3b7e46984ca39ee41b28b2ab86a6 : SUCCESS in 35m 39s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/bf261a3c10f24183914ff72f75fd3677 : FAILURE in 51m 06s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/0dc40a25e54548a09c9826654c281648 : SUCCESS in 1h 24m 33s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/2a02959efd6f4821b7734907cff0681f : SUCCESS in 40m 21s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/8336fa13c1bc409ab629e25941d4f270 : SUCCESS in 1h 02m 57s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/627ea03b04ea4f8bad614f03567aeba7 : SUCCESS in 1h 30m 11s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/07b411bb22174e568a6c55f4e0643d19 : FAILURE in 1h 19m 56s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/c884d7363a76419d8ada91c387b244eb : SUCCESS in 1h 44m 05s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/99d327e8d2234df9b8a53cd1997a745d : FAILURE in 1h 23m 03s (non-voting)\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/d5e34c2fcda045e6abf13861334c73b7 : FAILURE in 1h 10m 48s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/ace7e5fdb08e433186cc60855c4ee64d : SUCCESS in 48m 16s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/8f35b2bf9a0d4e2587c9bb6d14c02e4c : SUCCESS in 1h 08m 38s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/c4e71536972546fea988bec6e8e3c439 : SUCCESS in 2m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51f6684b63c740ae8caa39f88ff9e4e7 : SUCCESS in 3m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2f171f875c7049a28a9fce6142d804fd : SUCCESS in 3m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/96f0e645d9b44055b7023776f57d033c : SUCCESS in 3m 26s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/526cb68d79864b00acb602d9c2c4d902 : SUCCESS in 3m 18s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/62e4ce759fd04f8aa17840296c4c3dcc : SUCCESS in 2m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75de515e03cc433ba613c2c22d973102 : SUCCESS in 6m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1544548bafee4b74a8ac20806940e788 : SUCCESS in 7m 16s","accounts_in_message":[],"_revision_number":14},{"id":"81313307be7f81123b7e1ba8187d36c45ab291db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-02 12:27:26.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"6e08ec5f332291c325f0b0344d7b35474f803e50","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-02 14:09:16.000000000","message":"Patch Set 14: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/485d51272ddb47aab799a9a89e35bbdf\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/1ba130ba7fd84df0abbb6100422b7cda : SUCCESS in 14m 38s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/3bb8f86bc85d4c66866cd7f26805ecf6 : SUCCESS in 13m 38s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/4a281e0127744549af7b856e3521fa80 : SUCCESS in 14m 55s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/f9b7930a8841420e9f31c78b13ebe5db : SUCCESS in 33m 07s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f1f9112d032840158ebfa143502cfafb : SUCCESS in 14m 58s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/65aa18d0665d4ac0a7ff8be3456f0974 : SUCCESS in 14m 46s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/129a2b89f8b94d81b8dcc0c18956ad78 : SUCCESS in 1h 01m 32s\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/b7d218f813dd4ed79a57083566632620 : SUCCESS in 1h 41m 13s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/003093a1279a4c24be1d67d55d4f8fc5 : SUCCESS in 1h 03m 11s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/62ece887a93b43888b82109daa3e34e5 : SUCCESS in 1h 34m 33s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5c08e9e3ebde41e6bd81614272bdb16b : SUCCESS in 1h 07m 34s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/287adf5f9b1f4053a76d86df8ad5b29f : SUCCESS in 1h 40m 39s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/dba0aa48cf204415bc42e879e1fbcdd5 : SUCCESS in 1m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0f7bd60974e4ed59815d7730db5ad5d : SUCCESS in 2m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0b3602e9aca744009a79e39ae97ad068 : SUCCESS in 2m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2fb826fb74fb462f9ef869b4cdc8743a : SUCCESS in 2m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6dc95fb3d78f4513b69c892995b52fa3 : SUCCESS in 6m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/47778c61ca2e4cb9b2f7b7e80a4829eb : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":14},{"id":"adb7cdfdb8a3737a04e1ca8c7e7987320aff2a47","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-02 14:09:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":14},{"id":"b09274b9243570235b45da2f82a9f6db540a9e83","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-02 14:11:09.000000000","message":"Patch Set 14:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b682439dd0c048d795474c574d106de9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bfafd686f7e4722b8e51e139e16a0ab : SUCCESS in 1m 15s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd267bc720bc4b22b07484aad496dc03 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"505428fd51f3801c39850437f908fcf106a8fc94","revisions":{"debd1a5978f297c65cb290110f05d7b0fd12a632":{"kind":"REWORK","_number":1,"created":"2026-05-16 15:41:33.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/49/988849/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/49/988849/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/988849/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/988849/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/988849/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/49/988849/1"}}},"commit":{"parents":[{"commit":"129e88848d511ebdac4e42124e1192bc5c38d4b5","subject":"Merge \"Update mission statement in docs index\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/129e88848d511ebdac4e42124e1192bc5c38d4b5"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-05-16 13:05:48.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-05-16 15:41:11.000000000","tz":120},"subject":"Fix parsing of permissions with special bits","message":"Fix parsing of permissions with special bits\n\nkolla_set_configs currently fails to parse permissions\nsuch as 2775 because only legacy values like 0755 are\nconverted to Python octal notation.\n\nAccept standard UNIX permission notation for\npermissions with special mode bits such as SGID, SUID,\nand sticky bit.\n\nChange-Id: I5937a17b0d70f9c1db4b2617854d0180c37fe2c0\nSigned-off-by: Michal Arbet \u003cmichal.arbet@ultimum.io\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/debd1a5978f297c65cb290110f05d7b0fd12a632"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/debd1a5978f297c65cb290110f05d7b0fd12a632"}]},"branch":"refs/heads/master"},"c64ca1618581e69c130b0be9974feefd012a89bc":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-05-22 16:38:51.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/49/988849/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/49/988849/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/988849/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/988849/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/988849/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/49/988849/2"}}},"commit":{"parents":[{"commit":"e1e228ab45674b2aaf2f48ef76c4cb2cf25a4f7e","subject":"Revert \"set_configs: Add support for creating directories and setting perms\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e1e228ab45674b2aaf2f48ef76c4cb2cf25a4f7e"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-05-16 13:05:48.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-05-22 16:38:35.000000000","tz":120},"subject":"Fix parsing of permissions with special bits","message":"Fix parsing of permissions with special bits\n\nkolla_set_configs currently fails to parse permissions\nsuch as 2775 because only legacy values like 0755 are\nconverted to Python octal notation.\n\nAccept standard UNIX permission notation for\npermissions with special mode bits such as SGID, SUID,\nand sticky bit.\n\nChange-Id: I5937a17b0d70f9c1db4b2617854d0180c37fe2c0\nSigned-off-by: Michal Arbet \u003cmichal.arbet@ultimum.io\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c64ca1618581e69c130b0be9974feefd012a89bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c64ca1618581e69c130b0be9974feefd012a89bc"}]},"branch":"refs/heads/master"},"a905687c352656d49f790b2f69d26667c3035c69":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-05-23 09:27:26.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/49/988849/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/49/988849/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/988849/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/988849/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/988849/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/49/988849/3"}}},"commit":{"parents":[{"commit":"2a43eeebc8968a5683a5964ae585035658884080","subject":"CI: add aarch64 rocky10 publish job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2a43eeebc8968a5683a5964ae585035658884080"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-05-16 13:05:48.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-05-23 09:27:16.000000000","tz":120},"subject":"Fix parsing of permissions with special bits","message":"Fix parsing of permissions with special bits\n\nkolla_set_configs currently fails to parse permissions\nsuch as 2775 because only legacy values like 0755 are\nconverted to Python octal notation.\n\nAccept standard UNIX permission notation for\npermissions with special mode bits such as SGID, SUID,\nand sticky bit.\n\nChange-Id: I5937a17b0d70f9c1db4b2617854d0180c37fe2c0\nSigned-off-by: Michal Arbet \u003cmichal.arbet@ultimum.io\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a905687c352656d49f790b2f69d26667c3035c69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a905687c352656d49f790b2f69d26667c3035c69"}]},"branch":"refs/heads/master"},"fa037c1f33f30796f244e67e5e81836536347a15":{"kind":"NO_CHANGE","_number":4,"created":"2026-05-23 18:20:39.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/49/988849/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/49/988849/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/988849/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/988849/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/988849/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/49/988849/4"}}},"commit":{"parents":[{"commit":"2a43eeebc8968a5683a5964ae585035658884080","subject":"CI: add aarch64 rocky10 publish job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2a43eeebc8968a5683a5964ae585035658884080"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-05-16 13:05:48.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-05-23 18:19:40.000000000","tz":120},"subject":"Fix parsing of permissions with special bits","message":"Fix parsing of permissions with special bits\n\nkolla_set_configs currently fails to parse permissions\nsuch as 2775 because only legacy values like 0755 are\nconverted to Python octal notation.\n\nAccept standard UNIX permission notation for\npermissions with special mode bits such as SGID, SUID,\nand sticky bit.\n\nChange-Id: I5937a17b0d70f9c1db4b2617854d0180c37fe2c0\nSigned-off-by: Michal Arbet \u003cmichal.arbet@ultimum.io\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fa037c1f33f30796f244e67e5e81836536347a15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fa037c1f33f30796f244e67e5e81836536347a15"}]},"branch":"refs/heads/master"},"4ce2725d6644efd347e569a75e85a2ea51aa6f40":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2026-05-25 19:08:09.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/49/988849/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/49/988849/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/988849/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/988849/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/988849/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/49/988849/5"}}},"commit":{"parents":[{"commit":"2a43eeebc8968a5683a5964ae585035658884080","subject":"CI: add aarch64 rocky10 publish job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2a43eeebc8968a5683a5964ae585035658884080"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-05-16 13:05:48.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-05-25 19:07:16.000000000","tz":120},"subject":"Fix parsing of permissions with special bits","message":"Fix parsing of permissions with special bits\n\nkolla_set_configs currently fails to parse permissions\nsuch as 2775 because only legacy values like 0755 are\nconverted to Python octal notation.\n\nAccept standard UNIX permission notation for\npermissions with special mode bits such as SGID, SUID,\nand sticky bit.\n\nChange-Id: I5937a17b0d70f9c1db4b2617854d0180c37fe2c0\nNeeded-By: https://review.opendev.org/c/openstack/kolla-ansible/+/989850\nSigned-off-by: Michal Arbet \u003cmichal.arbet@ultimum.io\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4ce2725d6644efd347e569a75e85a2ea51aa6f40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4ce2725d6644efd347e569a75e85a2ea51aa6f40"}]},"branch":"refs/heads/master"},"032c42a36824e7c2e6e77d6e13405e81384779cc":{"kind":"NO_CHANGE","_number":6,"created":"2026-05-25 19:29:45.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/49/988849/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/49/988849/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/988849/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/988849/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/988849/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/49/988849/6"}}},"commit":{"parents":[{"commit":"2a43eeebc8968a5683a5964ae585035658884080","subject":"CI: add aarch64 rocky10 publish job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2a43eeebc8968a5683a5964ae585035658884080"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-05-16 13:05:48.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-05-25 19:29:04.000000000","tz":120},"subject":"Fix parsing of permissions with special bits","message":"Fix parsing of permissions with special bits\n\nkolla_set_configs currently fails to parse permissions\nsuch as 2775 because only legacy values like 0755 are\nconverted to Python octal notation.\n\nAccept standard UNIX permission notation for\npermissions with special mode bits such as SGID, SUID,\nand sticky bit.\n\nChange-Id: I5937a17b0d70f9c1db4b2617854d0180c37fe2c0\nNeeded-By: https://review.opendev.org/c/openstack/kolla-ansible/+/989850\nSigned-off-by: Michal Arbet \u003cmichal.arbet@ultimum.io\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/032c42a36824e7c2e6e77d6e13405e81384779cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/032c42a36824e7c2e6e77d6e13405e81384779cc"}]},"branch":"refs/heads/master"},"354825dc19310d7d96e4c328ef265f99f8aa27bd":{"kind":"REWORK","_number":7,"created":"2026-05-26 14:30:42.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/49/988849/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/49/988849/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/988849/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/988849/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/988849/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/49/988849/7"}}},"commit":{"parents":[{"commit":"2a43eeebc8968a5683a5964ae585035658884080","subject":"CI: add aarch64 rocky10 publish job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2a43eeebc8968a5683a5964ae585035658884080"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-05-16 13:05:48.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-05-26 14:28:22.000000000","tz":120},"subject":"Fix parsing of permissions with special bits","message":"Fix parsing of permissions with special bits\n\nkolla_set_configs currently fails to parse permissions\nsuch as 2775 because only legacy values like 0755 are\nconverted to Python octal notation.\n\nAccept standard UNIX permission notation for\npermissions with special mode bits such as SGID, SUID,\nand sticky bit.\n\nChange-Id: I5937a17b0d70f9c1db4b2617854d0180c37fe2c0\nNeeded-By: https://review.opendev.org/c/openstack/kolla-ansible/+/989850\nSigned-off-by: Michal Arbet \u003cmichal.arbet@ultimum.io\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/354825dc19310d7d96e4c328ef265f99f8aa27bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/354825dc19310d7d96e4c328ef265f99f8aa27bd"}]},"branch":"refs/heads/master"},"76374b752f4030199e96ae20d7b4af934daf7039":{"kind":"REWORK","_number":8,"created":"2026-05-26 14:32:35.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/49/988849/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/49/988849/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/988849/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/988849/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/988849/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/49/988849/8"}}},"commit":{"parents":[{"commit":"2a43eeebc8968a5683a5964ae585035658884080","subject":"CI: add aarch64 rocky10 publish job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2a43eeebc8968a5683a5964ae585035658884080"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-05-16 13:05:48.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-05-26 14:32:25.000000000","tz":120},"subject":"Fix parsing of permissions with special bits","message":"Fix parsing of permissions with special bits\n\nkolla_set_configs currently fails to parse permissions\nsuch as 2775 because only legacy values like 0755 are\nconverted to Python octal notation.\n\nAccept standard UNIX permission notation for\npermissions with special mode bits such as SGID, SUID,\nand sticky bit.\n\nChange-Id: I5937a17b0d70f9c1db4b2617854d0180c37fe2c0\nNeeded-By: https://review.opendev.org/c/openstack/kolla-ansible/+/989850\nSigned-off-by: Michal Arbet \u003cmichal.arbet@ultimum.io\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/76374b752f4030199e96ae20d7b4af934daf7039"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/76374b752f4030199e96ae20d7b4af934daf7039"}]},"branch":"refs/heads/master"},"0f3c67de1b86cfa69f38e0a996e0ea227844fe77":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2026-06-21 19:29:38.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/49/988849/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/49/988849/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/988849/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/988849/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/988849/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/49/988849/9"}}},"commit":{"parents":[{"commit":"2a59f5662f26d314e992da0182bb024bbad1f5d8","subject":"Merge \"Update Prometheus to version 3.5.4\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2a59f5662f26d314e992da0182bb024bbad1f5d8"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-05-16 13:05:48.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-06-21 18:58:34.000000000","tz":120},"subject":"Fix parsing of permissions with special bits","message":"Fix parsing of permissions with special bits\n\nkolla_set_configs currently fails to parse permissions\nsuch as 2775 because only legacy values like 0755 are\nconverted to Python octal notation.\n\nAccept standard UNIX permission notation for\npermissions with special mode bits such as SGID, SUID,\nand sticky bit.\n\nChange-Id: I5937a17b0d70f9c1db4b2617854d0180c37fe2c0\nNeeded-By: https://review.opendev.org/c/openstack/kolla-ansible/+/989850\nSigned-off-by: Michal Arbet \u003cmichal.arbet@ultimum.io\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0f3c67de1b86cfa69f38e0a996e0ea227844fe77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0f3c67de1b86cfa69f38e0a996e0ea227844fe77"}]},"branch":"refs/heads/master"},"0cf899d14b3b86990da592b86e9623b12b97d6d3":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2026-06-22 09:05:49.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/49/988849/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/49/988849/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/988849/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/988849/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/988849/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/49/988849/10"}}},"commit":{"parents":[{"commit":"8108d3b8c8c1e714f7d9e62311344346ca9f3a4c","subject":"Merge \"repos: Merge user-provided with default repos.yaml\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8108d3b8c8c1e714f7d9e62311344346ca9f3a4c"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-05-16 13:05:48.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-06-22 09:04:47.000000000","tz":120},"subject":"Fix parsing of permissions with special bits","message":"Fix parsing of permissions with special bits\n\nkolla_set_configs currently fails to parse permissions\nsuch as 2775 because only legacy values like 0755 are\nconverted to Python octal notation.\n\nAccept standard UNIX permission notation for\npermissions with special mode bits such as SGID, SUID,\nand sticky bit.\n\nChange-Id: I5937a17b0d70f9c1db4b2617854d0180c37fe2c0\nNeeded-By: https://review.opendev.org/c/openstack/kolla-ansible/+/989850\nSigned-off-by: Michal Arbet \u003cmichal.arbet@ultimum.io\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0cf899d14b3b86990da592b86e9623b12b97d6d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0cf899d14b3b86990da592b86e9623b12b97d6d3"}]},"branch":"refs/heads/master"},"87f86e76583c95e31d8ef1430285f322e0a6e1ae":{"kind":"REWORK","_number":11,"created":"2026-06-25 16:14:25.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/49/988849/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/49/988849/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/988849/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/988849/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/988849/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/49/988849/11"}}},"commit":{"parents":[{"commit":"8108d3b8c8c1e714f7d9e62311344346ca9f3a4c","subject":"Merge \"repos: Merge user-provided with default repos.yaml\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8108d3b8c8c1e714f7d9e62311344346ca9f3a4c"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-05-16 13:05:48.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-06-25 16:14:22.000000000","tz":120},"subject":"Fix parsing of permissions with special bits","message":"Fix parsing of permissions with special bits\n\nkolla_set_configs currently fails to parse permissions\nsuch as 2775 because only legacy values like 0755 are\nconverted to Python octal notation.\n\nAccept standard UNIX permission notation for\npermissions with special mode bits such as SGID, SUID,\nand sticky bit.\n\nChange-Id: I5937a17b0d70f9c1db4b2617854d0180c37fe2c0\nCloses-Bug: #2154269\nNeeded-By: https://review.opendev.org/c/openstack/kolla-ansible/+/989850\nSigned-off-by: Michal Arbet \u003cmichal.arbet@ultimum.io\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/87f86e76583c95e31d8ef1430285f322e0a6e1ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/87f86e76583c95e31d8ef1430285f322e0a6e1ae"}]},"branch":"refs/heads/master"},"8ba278955322d41fdb665ae30d56aae8cfa3361b":{"kind":"REWORK","_number":12,"created":"2026-06-25 16:22:56.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/49/988849/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/49/988849/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/988849/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/988849/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/988849/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/49/988849/12"}}},"commit":{"parents":[{"commit":"8108d3b8c8c1e714f7d9e62311344346ca9f3a4c","subject":"Merge \"repos: Merge user-provided with default repos.yaml\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8108d3b8c8c1e714f7d9e62311344346ca9f3a4c"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-05-16 13:05:48.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-06-25 16:22:48.000000000","tz":120},"subject":"Fix parsing of permissions with special bits","message":"Fix parsing of permissions with special bits\n\nkolla_set_configs currently fails to parse permissions\nsuch as 2775 because only legacy values like 0755 are\nconverted to Python octal notation.\n\nAccept standard UNIX permission notation for\npermissions with special mode bits such as SGID, SUID,\nand sticky bit.\n\nChange-Id: I5937a17b0d70f9c1db4b2617854d0180c37fe2c0\nCloses-Bug: #2154269\nNeeded-By: https://review.opendev.org/c/openstack/kolla-ansible/+/989850\nSigned-off-by: Michal Arbet \u003cmichal.arbet@ultimum.io\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8ba278955322d41fdb665ae30d56aae8cfa3361b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8ba278955322d41fdb665ae30d56aae8cfa3361b"}]},"branch":"refs/heads/master"},"5995a08ab50fcd17fbcf2633a968d2b205e8da8e":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2026-06-25 16:27:13.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/49/988849/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/49/988849/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/988849/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/988849/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/988849/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/49/988849/13"}}},"commit":{"parents":[{"commit":"108660d64406067b5d066c612f5de882d45b0e24","subject":"prometheus-mtail: Switch to a new fork","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/108660d64406067b5d066c612f5de882d45b0e24"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-05-16 13:05:48.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-06-25 16:27:02.000000000","tz":120},"subject":"Fix parsing of permissions with special bits","message":"Fix parsing of permissions with special bits\n\nkolla_set_configs currently fails to parse permissions\nsuch as 2775 because only legacy values like 0755 are\nconverted to Python octal notation.\n\nAccept standard UNIX permission notation for\npermissions with special mode bits such as SGID, SUID,\nand sticky bit.\n\nChange-Id: I5937a17b0d70f9c1db4b2617854d0180c37fe2c0\nCloses-Bug: #2154269\nNeeded-By: https://review.opendev.org/c/openstack/kolla-ansible/+/989850\nSigned-off-by: Michal Arbet \u003cmichal.arbet@ultimum.io\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5995a08ab50fcd17fbcf2633a968d2b205e8da8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5995a08ab50fcd17fbcf2633a968d2b205e8da8e"}]},"branch":"refs/heads/master"},"505428fd51f3801c39850437f908fcf106a8fc94":{"kind":"REWORK","_number":14,"created":"2026-07-01 16:14:07.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/49/988849/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/49/988849/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/988849/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/988849/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/988849/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/49/988849/14"}}},"commit":{"parents":[{"commit":"108660d64406067b5d066c612f5de882d45b0e24","subject":"prometheus-mtail: Switch to a new fork","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/108660d64406067b5d066c612f5de882d45b0e24"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-05-16 13:05:48.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-07-01 16:11:49.000000000","tz":120},"subject":"Fix parsing of permissions with special bits","message":"Fix parsing of permissions with special bits\n\nkolla_set_configs currently fails to parse permissions\nsuch as 2775 because only legacy values like 0755 are\nconverted to Python octal notation.\n\nAccept standard UNIX permission notation for\npermissions with special mode bits such as SGID, SUID,\nand sticky bit.\n\nChange-Id: I5937a17b0d70f9c1db4b2617854d0180c37fe2c0\nCloses-Bug: #2154269\nNeeded-By: https://review.opendev.org/c/openstack/kolla-ansible/+/989850\nSigned-off-by: Michal Arbet \u003cmichal.arbet@ultimum.io\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/505428fd51f3801c39850437f908fcf106a8fc94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/505428fd51f3801c39850437f908fcf106a8fc94"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d1"],"failing_atoms":["label:Backport-Candidate\u003d-1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
