)]}'
{"id":"openstack%2Fkolla~992036","triplet_id":"openstack%2Fkolla~master~I2c755fae4a7832d76e11e6112e2289efc7f24437","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-06-08 16:00:11.000000000","reason":"Change was abandoned"},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2026-06-08 12:27:11.000000000","reason":"\u003cGERRIT_ACCOUNT_27339\u003e replied on the change","reason_account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}}},"hashtags":[],"change_id":"I2c755fae4a7832d76e11e6112e2289efc7f24437","subject":"proxysql: Pin to 3.0.8","status":"ABANDONED","created":"2026-06-08 08:50:04.000000000","updated":"2026-06-08 16:00:11.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0af9cf93561324a802caea440da5724140be3f09","_number":992036,"virtual_id_number":992036,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"rejected":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":-2,"date":"2026-06-08 15:13:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":2,"date":"2026-06-08 14:53:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2026-06-08 13:26:05.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":2,"date":"2026-06-08 14:54:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":-1,"date":"2026-06-08 15:13:31.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":1,"date":"2026-06-08 14:54:53.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2026-06-08 14:53:47.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2026-06-08 13:31:47.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-08 09:21:31.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2026-06-08 10:27:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-08 13:26:05.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-06-08 14:54:53.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"70a96dc7d37571551148de0dba087ef7deb50ecb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-08 08:50:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d6b99bff00961033ca36095fcd3b5729998f3644","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-06-08 09:21:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a7a5f96fb13bc4d8dca2ed221420b38a8bc4b7c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 10:27:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/28475de324934e7aa9bc17982a5a2108\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/1fb846ca020e40b9852138f4e990498b : SUCCESS in 1h 11m 03s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/1d32363a8c9d4473887e473b7cc8aca5 : FAILURE in 1h 10m 19s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/1d67c7818cf74201884b4473fc942bf0 : FAILURE in 26m 52s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/27e4dc963ceb40bba07cb3f77230ceef : SUCCESS in 1h 16m 25s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/d0b291d136704b3099e4d6298bd74c9f : SUCCESS in 1h 00m 14s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/577becaca18b4201a6edc93a07ee63ce : FAILURE in 1h 06m 15s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a8fc47c2e8cb4ac58d3901bcffa071a1 : FAILURE in 28m 37s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/c3061519ea9b464493988aa2e7dd66ce : SUCCESS in 1h 25m 10s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/5fbf85feacb24ea88843467820605c88 : FAILURE in 33m 30s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/1b2bf3e074b14140a2150ab65960926a : SUCCESS in 1h 18m 12s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/1caf9623bcb148928ed8c8675a00dcd8 : FAILURE in 30m 38s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/92b47d32745b4ecd8269bb7f57509fd1 : FAILURE in 34m 49s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/c056b2b3ab434d55b5efd8bd0167fae7 : FAILURE in 31m 45s\n- kolla-ansible-debian-trixie-cells https://zuul.opendev.org/t/openstack/build/e731439d987a40899840e2f48a47f669 : FAILURE in 34m 08s (non-voting)\n- kolla-ansible-rocky-10-cells https://zuul.opendev.org/t/openstack/build/e5845ef9b5a14ae8acbbfd989854a681 : SUCCESS in 1h 35m 44s (non-voting)\n- kolla-ansible-ubuntu-noble-cells https://zuul.opendev.org/t/openstack/build/1ea8101e72e5404b8e8f5dfa994be89a : FAILURE in 32m 20s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/175d4f3009b04e71b3b7f7f1a4f8a534 : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6819ddb73b94166ad3c68015b68d071 : SUCCESS in 4m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d9d4a3a7e6364a1a80a6a0cbe096fe1c : SUCCESS in 5m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a8078a2e5f434576b31a89af0fb68e6f : SUCCESS in 3m 26s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1e85bb10bd554cb68c0e73528b66ca6f : SUCCESS in 8m 39s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/718b4205683a4104b398794f1d6abb1e : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f35fbf5999a34f86b65aae9bccf1bae1 : SUCCESS in 7m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/02585eea66aa4b34a0424a9a68261ef4 : SUCCESS in 9m 22s","accounts_in_message":[],"_revision_number":1},{"id":"fe03e825019c6a09418c3f3a02aba751f301141d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-08 10:29:35.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"2fafe615661020346e4de765b069ae51130fbcd7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-08 10:31:08.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"3dc24ac6025e9081e575e4e5259f10d202dc9191","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-08 11:14:02.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"49b6a05e5cb787c19aaf33adc1bbdf431989a643","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-08 11:14:06.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":4},{"id":"17ea776f86c2762591d04a8ec09b941fbea90aa9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-08 11:21:00.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"0e2f9f901cbc521a54d6d2bb8f264188b8b98021","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-08 11:21:05.000000000","message":"Patch Set 5: Review-Priority+2","accounts_in_message":[],"_revision_number":5},{"id":"183399991e372db933df3b36cfff071e228ed6d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-08 12:23:38.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":6},{"id":"1b9c7e65c74c73ec53e2d4d7a34ec61b4c100a89","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-08 12:23:46.000000000","message":"Patch Set 7: Commit message was updated.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":7},{"id":"83732c81287272d1e2cddc2390c491c6e08ba1d1","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-06-08 12:27:11.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"7570d91595c3a1a85cf1267cee4d0b37824b90e6","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-06-08 13:26:05.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"14661e54a0fd849c1e99d207e3f57260bb0b54d1","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-08 13:31:47.000000000","message":"Patch Set 7: Review-Priority+2 Workflow-1","accounts_in_message":[],"_revision_number":7},{"id":"abb61aae16aaa181464fcaea4f22b9153b5b5d7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 14:28:53.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1bd56b88d3dd4b2da08cc8de6a177d55\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/a3217b77a0d54d4c9326f56918536c59 : SUCCESS in 1h 14m 37s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/83cb0e954a1c4ae39aaf00e391185d9d : SUCCESS in 1h 20m 05s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/d0c45f9cbb734f338fc2c03b32adb0b7 : SUCCESS in 13m 58s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/eecaacde645a4f4cb3bf4814ca369dfb : SUCCESS in 1h 20m 14s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/6e2510d9a7df4ac8bf3d216937ddf16a : SUCCESS in 49m 24s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b06b0c19d3eb41d3b5bf0ab67ca27549 : SUCCESS in 1h 25m 23s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5aca619407084df6b27b7176e8a82ee8 : SUCCESS in 28m 43s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/1d84f0f0c2a24d49a1916a84713c68be : SUCCESS in 1h 21m 32s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/35f12268c82c45cfa62427d11882b42c : SUCCESS in 1h 22m 16s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/468ddf24470c495e86b7d3bbbaa207eb : SUCCESS in 1h 21m 20s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/47fe152ec5cc4917bfb7db2b8505b288 : SUCCESS in 1h 52m 13s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3350a07157a04f45a22be034fd51dba8 : SUCCESS in 1h 24m 19s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/a4d1dc25cb1946379c1f418c151cf570 : SUCCESS in 2h 02m 51s\n- kolla-ansible-debian-trixie-cells https://zuul.opendev.org/t/openstack/build/853cc6901be24079a9608d000e2c7fb7 : SUCCESS in 1h 36m 17s (non-voting)\n- kolla-ansible-rocky-10-cells https://zuul.opendev.org/t/openstack/build/f9fb096c56634759bd77e9360f2099ab : SUCCESS in 1h 31m 18s (non-voting)\n- kolla-ansible-ubuntu-noble-cells https://zuul.opendev.org/t/openstack/build/5a73e34adf284fafb8ebf3704e88dd0f : SUCCESS in 1h 39m 53s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/4a9e11d585cc47708a1bd25a0b3bb707 : SUCCESS in 3m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ab8d7a91c3443519c41442b0766082c : SUCCESS in 3m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7c67faca6f9a425588a9f7779a127470 : SUCCESS in 3m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7e525fb314754f119581a47c88c6758b : SUCCESS in 2m 35s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ed39fd28b01a4e678b8f86672d51a4ee : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7c69afc5b8264ed69f0d982aa9dcbac7 : SUCCESS in 2m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fba10f228dce4a109d7471ceeef33f2e : SUCCESS in 6m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/16ba849dd61b4fccb3ec19bed12382b4 : SUCCESS in 9m 46s","accounts_in_message":[],"_revision_number":7},{"id":"4969007bf97b817b39336bbcd9e63d8171cef2ef","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-08 14:53:47.000000000","message":"Patch Set 7: Backport-Candidate+1 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"69cb2f586a92bf06c14547403838ec1e2da62640","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 14:54:51.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"7dd5549566951a6a126ba4cb4952c0ef3ba226e2","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-06-08 14:54:53.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"ce5ce62d1f8b3dc536a0b01ac65c4c47f3e47cd1","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-08 15:13:31.000000000","message":"Patch Set 7: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"aa9de1a2712faa9443d41827b54ce9e5af7fef00","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 15:13:35.000000000","message":"Patch Set 7: Verified-2\n\nThis change is unable to merge due to a missing merge requirement.","accounts_in_message":[],"_revision_number":7},{"id":"0af9cf93561324a802caea440da5724140be3f09","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-08 16:00:11.000000000","message":"Abandoned\n\nnot required","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"55ed28581e2f31718028f39ef980bd82ea09b2e6","revisions":{"4c52db67bf59195b6705938348079ef5bcf61086":{"kind":"REWORK","_number":1,"created":"2026-06-08 08:50:04.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/36/992036/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/36/992036/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/992036/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/992036/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/992036/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/36/992036/1"}}},"commit":{"parents":[{"commit":"bf4ee2deaa53c2dc585f3d270e036579350b1b52","subject":"Merge \"Implement container image build for kerbside.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bf4ee2deaa53c2dc585f3d270e036579350b1b52"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-08 08:47:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-08 08:50:02.000000000","tz":60},"subject":"proxysql: Pin to 3.0.8","message":"proxysql: Pin to 3.0.8\n\nIt seems [1] has broken our ProxySQL config.\n\n[1]: https://github.com/sysown/proxysql/commit/37e72ea3f98cb83b2ea832166c14905de643e620\n\nChange-Id: I2c755fae4a7832d76e11e6112e2289efc7f24437\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4c52db67bf59195b6705938348079ef5bcf61086"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4c52db67bf59195b6705938348079ef5bcf61086"}]},"branch":"refs/heads/master"},"c1470356d8153a0fd49ab47f666b2ca4c19a99e0":{"kind":"REWORK","_number":2,"created":"2026-06-08 10:29:35.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/36/992036/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/36/992036/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/992036/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/992036/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/992036/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/36/992036/2"}}},"commit":{"parents":[{"commit":"bf4ee2deaa53c2dc585f3d270e036579350b1b52","subject":"Merge \"Implement container image build for kerbside.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bf4ee2deaa53c2dc585f3d270e036579350b1b52"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-08 08:47:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-08 10:08:16.000000000","tz":0},"subject":"proxysql: Pin to 3.0.8","message":"proxysql: Pin to 3.0.8\n\nIt seems [1] has broken our ProxySQL config.\n\n[1]: https://github.com/sysown/proxysql/commit/37e72ea3f98cb83b2ea832166c14905de643e620\n\nChange-Id: I2c755fae4a7832d76e11e6112e2289efc7f24437\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c1470356d8153a0fd49ab47f666b2ca4c19a99e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c1470356d8153a0fd49ab47f666b2ca4c19a99e0"}]},"branch":"refs/heads/master"},"d89ff349500c9e9276a060bc816e3c715b81b290":{"kind":"REWORK","_number":3,"created":"2026-06-08 10:31:08.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/36/992036/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/36/992036/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/992036/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/992036/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/992036/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/36/992036/3"}}},"commit":{"parents":[{"commit":"bf4ee2deaa53c2dc585f3d270e036579350b1b52","subject":"Merge \"Implement container image build for kerbside.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bf4ee2deaa53c2dc585f3d270e036579350b1b52"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-08 08:47:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-08 10:31:08.000000000","tz":0},"subject":"proxysql: Pin to 3.0.8","message":"proxysql: Pin to 3.0.8\n\nIt seems [1] has broken our ProxySQL config.\n\n[1]: https://github.com/sysown/proxysql/commit/37e72ea3f98cb83b2ea832166c14905de643e620\n\nChange-Id: I2c755fae4a7832d76e11e6112e2289efc7f24437\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d89ff349500c9e9276a060bc816e3c715b81b290"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d89ff349500c9e9276a060bc816e3c715b81b290"}]},"branch":"refs/heads/master"},"08f08d46cc338c642ba74d78573ba4a2cdf27fe2":{"kind":"REWORK","_number":4,"created":"2026-06-08 11:14:02.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/36/992036/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/36/992036/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/992036/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/992036/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/992036/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/36/992036/4"}}},"commit":{"parents":[{"commit":"bf4ee2deaa53c2dc585f3d270e036579350b1b52","subject":"Merge \"Implement container image build for kerbside.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bf4ee2deaa53c2dc585f3d270e036579350b1b52"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-08 08:47:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-08 11:13:59.000000000","tz":0},"subject":"proxysql: Pin to 3.0.8","message":"proxysql: Pin to 3.0.8\n\nIt seems [1] has broken our ProxySQL config.\n\n[1]: https://github.com/sysown/proxysql/commit/37e72ea3f98cb83b2ea832166c14905de643e620\n\nChange-Id: I2c755fae4a7832d76e11e6112e2289efc7f24437\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/08f08d46cc338c642ba74d78573ba4a2cdf27fe2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/08f08d46cc338c642ba74d78573ba4a2cdf27fe2"}]},"branch":"refs/heads/master"},"d87755cca55f8a8e9cefa9aa9738aa7e7fe760c9":{"kind":"REWORK","_number":5,"created":"2026-06-08 11:21:00.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/36/992036/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/36/992036/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/992036/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/992036/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/992036/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/36/992036/5"}}},"commit":{"parents":[{"commit":"bf4ee2deaa53c2dc585f3d270e036579350b1b52","subject":"Merge \"Implement container image build for kerbside.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bf4ee2deaa53c2dc585f3d270e036579350b1b52"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-08 08:47:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-08 11:20:58.000000000","tz":0},"subject":"proxysql: Pin to 3.0.8","message":"proxysql: Pin to 3.0.8\n\nIt seems [1] has broken our ProxySQL config.\n\n[1]: https://github.com/sysown/proxysql/commit/37e72ea3f98cb83b2ea832166c14905de643e620\n\nChange-Id: I2c755fae4a7832d76e11e6112e2289efc7f24437\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d87755cca55f8a8e9cefa9aa9738aa7e7fe760c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d87755cca55f8a8e9cefa9aa9738aa7e7fe760c9"}]},"branch":"refs/heads/master"},"fc2e1d55ca5d3ba8d7134c092c67f4c12b1d05a5":{"kind":"REWORK","_number":6,"created":"2026-06-08 12:23:38.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/36/992036/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/36/992036/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/992036/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/992036/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/992036/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/36/992036/6"}}},"commit":{"parents":[{"commit":"bf4ee2deaa53c2dc585f3d270e036579350b1b52","subject":"Merge \"Implement container image build for kerbside.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bf4ee2deaa53c2dc585f3d270e036579350b1b52"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-08 08:47:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-08 12:23:38.000000000","tz":0},"subject":"proxysql: Pin to 3.0.8","message":"proxysql: Pin to 3.0.8\n\nIt seems [1] has broken our ProxySQL config.\n\n[1]: https://github.com/sysown/proxysql/commit/37e72ea3f98cb83b2ea832166c14905de643e620\n\nChange-Id: I2c755fae4a7832d76e11e6112e2289efc7f24437\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fc2e1d55ca5d3ba8d7134c092c67f4c12b1d05a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fc2e1d55ca5d3ba8d7134c092c67f4c12b1d05a5"}]},"branch":"refs/heads/master"},"55ed28581e2f31718028f39ef980bd82ea09b2e6":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2026-06-08 12:23:46.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/36/992036/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/36/992036/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/992036/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/992036/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/992036/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/36/992036/7"}}},"commit":{"parents":[{"commit":"bf4ee2deaa53c2dc585f3d270e036579350b1b52","subject":"Merge \"Implement container image build for kerbside.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bf4ee2deaa53c2dc585f3d270e036579350b1b52"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-08 08:47:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-08 12:23:45.000000000","tz":0},"subject":"proxysql: Pin to 3.0.8","message":"proxysql: Pin to 3.0.8\n\nIt seems [1] has broken our ProxySQL deployment.\n\n[1]: https://github.com/sysown/proxysql/commit/37e72ea3f98cb83b2ea832166c14905de643e620\n\nChange-Id: I2c755fae4a7832d76e11e6112e2289efc7f24437\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/55ed28581e2f31718028f39ef980bd82ea09b2e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/55ed28581e2f31718028f39ef980bd82ea09b2e6"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Verified\u003dMIN"],"failing_atoms":["label:Verified\u003dMAX"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d1"],"failing_atoms":["label:Backport-Candidate\u003d-1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"failing_atoms":[],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
