)]}'
{"id":"openstack%2Fkolla~992223","triplet_id":"openstack%2Fkolla~master~I171b8b4442af3373ec482598d57914b28bc070e0","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I171b8b4442af3373ec482598d57914b28bc070e0","subject":"DNM: Test proxysql","status":"ABANDONED","created":"2026-06-08 12:35:15.000000000","updated":"2026-06-11 07:41:38.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":false,"meta_rev_id":"664c6537bdc03a2a3ca70ba296511701531d336d","_number":992223,"virtual_id_number":992223,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-06-08 14:39:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2026-06-08 14:39:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"04a4366bfe37aa38f4e3c8516d91841a720cf49f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-08 12:35:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"518a90268c34ae49448f0e8fbb7ccfda15089d84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 14:39:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8e1d07e43712406a8dcaf4a2d884a05e\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/2983b3be93ff4664aeb20aa8ebf0721a : SUCCESS in 1h 09m 32s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/587608247c404756b15bc189844af708 : SUCCESS in 1h 14m 24s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/3999b52e3f404e32b161f030b60d3a5c : SUCCESS in 21m 17s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/d13797a567f64aeba7da2ca260d1de0a : SUCCESS in 1h 19m 45s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/3a6e9e5be05849d7afd0e499e0b908a0 : SUCCESS in 36m 17s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b463de7f3dc048669be3d31ff438ea95 : SUCCESS in 1h 33m 47s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5f4d93c596b14f0d9a2edc2407d3f020 : SUCCESS in 30m 33s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/490122d1cd1146afa8ceb7e72d186f8a : SUCCESS in 1h 20m 34s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/5af683682a8d47d5a13748712b126d61 : SUCCESS in 1h 22m 58s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/ec4898a4b5ac4121ad89b19bafe0217f : SUCCESS in 1h 29m 42s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/fd8f5f8fcca74bd9a550b50fba36a99f : SUCCESS in 2h 01m 27s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/33973cd8d6a64381a0af86f0e3ad4ea5 : SUCCESS in 1h 58m 45s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/1afd3c64823b48cb99a770ac69f6a69c : SUCCESS in 1h 22m 57s\n- kolla-ansible-debian-trixie-cells https://zuul.opendev.org/t/openstack/build/584af72781af47eeb741cc3885081e54 : FAILURE in 1h 14m 17s (non-voting)\n- kolla-ansible-rocky-10-cells https://zuul.opendev.org/t/openstack/build/53f9545b539248e8be3ec0e785c4c975 : SUCCESS in 1h 37m 22s (non-voting)\n- kolla-ansible-ubuntu-noble-cells https://zuul.opendev.org/t/openstack/build/6062b071b47a4206a471661ab2446cf0 : FAILURE in 1h 04m 43s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/616c67b781444d49897f59fc0c10ab95 : FAILURE in 29m 28s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/fe6eb9ecea924fbb8bbec46d7438b52e : SUCCESS in 19m 58s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/066126ffce27434e9d5e180203d794b5 : FAILURE in 35m 03s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/bac02e8004d54310b4f764e2795204f0 : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/795323cb48704c20a1f723e699e1e60c : SUCCESS in 3m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d09610b5b1b344fbb9a1f676a36f4056 : SUCCESS in 3m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0b93472b801742ec913dd0ac75f26c58 : SUCCESS in 4m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8aac44f4de084d97a1d055f1c04ae978 : SUCCESS in 7m 52s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f8fe15401e314d1086b944b3cf218522 : SUCCESS in 2m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3111dba32e3543b5876831e0d61a4e38 : SUCCESS in 6m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/12f1c6e4ea9d4bba8a4b67522aa00fa9 : SUCCESS in 9m 04s","accounts_in_message":[],"_revision_number":1},{"id":"d3b69c01e9b20b7b2d4c92fac8f7d75758eb63ff","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 11:52:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1664b866b5e14422a91b0147087d1471\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/1308b9fef3d74a4e8cb9abb0a8261959 : SUCCESS in 1h 54m 55s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/042f0885416e4fb79d09dca665d5152c : SUCCESS in 1h 19m 34s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/61c1c1a1feed4f87a773e539bdcf5376 : SUCCESS in 1h 36m 41s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/6333af512b3b46edbc5be0bba6b0af83 : SUCCESS in 1h 42m 32s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/d18a6a104fc54cac933364bfaaa9a290 : FAILURE in 1h 40m 41s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/8ad3c58c294949b48de0e703da60eb4a : FAILURE in 31m 55s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9a5b2c5ef2064416ae4cd4fec4f0cddb : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7303fdfb193345f9be819c6f4098832b : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/38bcd195052a404b9a5979584843f995 : SUCCESS in 8m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"664c6537bdc03a2a3ca70ba296511701531d336d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-11 07:41:38.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ad01171978fbdf5e39a9eed53b133119292ca65a","revisions":{"ad01171978fbdf5e39a9eed53b133119292ca65a":{"kind":"REWORK","_number":1,"created":"2026-06-08 12:35:15.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/23/992223/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/992223/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/992223/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/992223/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/992223/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/992223/1"}}},"commit":{"parents":[{"commit":"bf4ee2deaa53c2dc585f3d270e036579350b1b52","subject":"Merge \"Implement container image build for kerbside.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bf4ee2deaa53c2dc585f3d270e036579350b1b52"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-08 12:35:01.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-08 12:35:13.000000000","tz":60},"subject":"DNM: Test proxysql","message":"DNM: Test proxysql\n\nChange-Id: I171b8b4442af3373ec482598d57914b28bc070e0\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ad01171978fbdf5e39a9eed53b133119292ca65a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ad01171978fbdf5e39a9eed53b133119292ca65a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
