)]}'
{"id":"openstack%2Fkolla~992394","triplet_id":"openstack%2Fkolla~stable%2F2026.1~I1c67f6b0f4780b7caf83ccf95a20861d9fefe20b","project":"openstack/kolla","branch":"stable/2026.1","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-06-11 16:49:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1c67f6b0f4780b7caf83ccf95a20861d9fefe20b","subject":"keepalived: Install mariadb client package","status":"MERGED","created":"2026-06-09 06:40:16.000000000","updated":"2026-06-16 15:33:50.000000000","submitted":"2026-06-11 16:49:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"992394","cherry_pick_of_change":992239,"cherry_pick_of_patch_set":4,"meta_rev_id":"d6f7a05e47edf6cc26cd0bdb17ffc19794ea7131","_number":992394,"virtual_id_number":992394,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-11 16:49:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"all":[{"value":2,"date":"2026-06-10 15:06:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},{"value":2,"date":"2026-06-09 07:24:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2026-06-10 23:26:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2026-06-09 10:17:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2026-06-11 12:20:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2026-06-10 23:26:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},{"value":1,"date":"2026-06-09 07:24:25.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"}],"CC":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-09 07:24:25.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2026-06-09 08:47:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-09 10:17:22.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-06-10 15:06:24.000000000","updated_by":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"reviewer":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"state":"REVIEWER"},{"updated":"2026-06-11 09:40:10.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2026-06-11 12:20:51.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"}],"messages":[{"id":"9c2b741b0197485965c3390fdc3613046a1aee1f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-09 06:40:16.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"591531c6e9b79a89bc09d960426c8e441608654f","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-06-09 07:24:25.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"36c28aae5858d8376ae32eebc25731b51ac4ee50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-09 08:47:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/be7546ef5de14c27bd2cbf2bf0fa0433\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/f4ff7d9a1fed47a0a1f8cc9f67981801 : FAILURE in 1h 02m 10s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/e9c84098d643442a943273c98c9e8e4f : SUCCESS in 1h 21m 31s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/c5e1391b13bc4826883a2abaf948255f : SUCCESS in 26m 52s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/8bc8dce4475546f58bf479fdf15ff87c : FAILURE in 1h 04m 45s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/76c14f144736484b9047ad64d80795ae : SUCCESS in 42m 38s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/11fef93d74b94ab28c419a98e6f6cdf4 : SUCCESS in 1h 18m 18s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3e30c7e2d66a41f1a07a2db3017a22eb : SUCCESS in 28m 43s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/6e900ce3c33240cfa09a3336e84af6e6 : FAILURE in 28m 33s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/e5b4f37960ae41e2a5c75a024be9569e : SUCCESS in 1h 24m 57s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/b2ab17a758f74f6296b517af88a50e45 : SUCCESS in 1h 46m 11s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/e14417521c924a61bfb8275102cf2432 : SUCCESS in 1h 23m 09s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/dc75f28d21534156938eb778b3b89321 : SUCCESS in 1h 48m 07s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d819940aed654a2c9450ced583ec2f7a : SUCCESS in 1h 27m 35s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/0e519ba5e7714300bf0d72d7c5b8155c : SUCCESS in 2h 06m 03s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/1b45d4984caf45d3971c20a939176bb0 : SUCCESS in 3m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0637feb96c764a2898b4c694ec611f87 : SUCCESS in 4m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2b35699286c946609583e729a202a631 : SUCCESS in 2m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0b9c73e5d70c419ebceb082db48227dc : SUCCESS in 3m 16s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0b24cd0d0872476585f6ce98b3721158 : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6bf0c46216bb469fa9372f6dc9e8649a : SUCCESS in 3m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d79834d02564b12844fecb3aa35b558 : SUCCESS in 6m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fc3f9e3697a742b1aef3b377ef452034 : SUCCESS in 11m 53s","accounts_in_message":[],"_revision_number":1},{"id":"86864c42084b119b79b9fd2fe5e3ea71c4cd9a9a","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-06-09 10:17:22.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"462e8900fe9be76a94bc4d13119c62e621f6ea9d","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-06-09 15:09:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dd5e651e0fa35768dfb6f7e0e921bf0ca64516bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-09 16:46:17.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c299ee1b57674153a0585703e483f5aa\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/74fcbf51e3f843f9b20114a56a2d4062 : SUCCESS in 1h 01m 57s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/737177e56039468480525ef161f76e70 : SUCCESS in 58m 07s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/6469145bd33b4d80be874e0a92041635 : SUCCESS in 22m 04s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/cc57a014fbda412295338011087ab436 : SUCCESS in 1h 01m 37s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/6cc1d47b6f3846f2aeb5aae0d838d80e : SUCCESS in 58m 39s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/798c5cfacd624f238a3f11fb38584422 : SUCCESS in 1h 29m 11s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e4d0e25c37d540c5a3e336c3b1f8a8c9 : SUCCESS in 13m 03s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/e14d6d83c00343e6b59484f3bcba409c : FAILURE in 49m 49s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/f1526151a4d94e31b8459e5106aa8285 : FAILURE in 41m 33s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/b9df26b007344d2eaf7446becb8f6044 : FAILURE in 1h 15m 05s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/05369c5593454ea99bd1fbf006e4b3f9 : FAILURE in 53m 39s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/c1354e4928d4444bacbbf6e6d4c72973 : FAILURE in 1h 35m 22s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/afc565191c184f87864906e00c82d16b : FAILURE in 54m 14s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/d243a655de114548b15ca0cd0c5a0c87 : FAILURE in 1h 35m 39s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/d06b39b6fa324e79ae0cfcd52069d8cc : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2642873e500046149c37841eb407794e : SUCCESS in 4m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/abe75c75fa0e4a099d5ffe276b5c4e77 : SUCCESS in 2m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2f223726fd7f4c0da2e1240b3b0c18e0 : SUCCESS in 4m 59s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/60ad09f028a740e6ad6d358c2a4e6628 : SUCCESS in 5m 45s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ddaa8f6c34474c1fb8f15bd40b22a56a : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c7f9881706049c28858655922ef1c68 : SUCCESS in 7m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/762766614ace4acb854cf2f81cd98e30 : SUCCESS in 10m 40s","accounts_in_message":[],"_revision_number":1},{"id":"07e5afc27a7d978492620aa0bc29091fbcfe1bc8","author":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"date":"2026-06-10 15:06:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ef8118c5a856cd14492dc7566035e27409731986","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-10 16:37:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c6965cb9f841d4116cfce271657c2fd3d25ed712","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 17:16:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/463380abbca74052a39c16b1b94e06cf\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/14d1b982a2354997b77c49abb5e50f7d : TIMED_OUT in 3h 15m 06s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/fba617e48dda44f8be20d3fd2f551483 : SUCCESS in 2h 45m 38s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/eedd430987344e5d91133dae9e41cb3b : SUCCESS in 1h 25m 57s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/ea7460aca46d4f02889f6c1cfdfa29e3 : SUCCESS in 2h 15m 51s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/3936e311741c48cea07a6a9c6a47f732 : NODE_FAILURE Node(set) request 409df93443e54c44b68435dfcf734262 failed in 0s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/9fb660e703ad4d92ac5b86a52954a2d0 : NODE_FAILURE Node(set) request d6fc2e9863a4400caa6352679442610f failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ef26021dd1169a8c4d92d4ad9b4ef12c7b834ebf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 18:59:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/838c1c7811984056abd07117fe25cae1\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/ec45ad436729494cac35087df011535d : SUCCESS in 1h 02m 54s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/c46aa6a430494ccebffb0c780d809aed : SUCCESS in 1h 08m 52s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/2ba941c8d9c14cc69f85ea874764c7d4 : SUCCESS in 20m 41s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/54b8b21c7b144aebb78c6a55062d2606 : SUCCESS in 1h 01m 58s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/7d25c28195214cf080ad7bcb6304ce85 : SUCCESS in 58m 22s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c6966f32eb6445e984d4f021b4c19cd7 : SUCCESS in 1h 00m 31s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e260dde421784ec0970c333a9967314d : SUCCESS in 13m 40s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/e62ab1db952f413a9d5f4ff6736f6c4d : SUCCESS in 1h 26m 12s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/ca57ee02c03b4c6c8182bd919262db75 : SUCCESS in 1h 35m 11s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/79d2a713378741a9bcf36ecc442a6106 : SUCCESS in 2h 03m 30s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/3f78ce5a8f1742d8a882c1f3505f947a : SUCCESS in 1h 29m 25s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/cf729e923d304577a7e530a8bb953085 : SUCCESS in 2h 02m 16s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2a09334875214380b1d51621fe7ca792 : SUCCESS in 1h 32m 23s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/a5491db44c2a441d9b45eaf2af485483 : SUCCESS in 2h 07m 25s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/d0c67b2c736d4e069f82fa3189bf1c88 : SUCCESS in 3m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d2f2026b8324894b13d90b8a2483580 : SUCCESS in 5m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/02a937012b104ff4bfb892f12a3f584d : SUCCESS in 5m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fa2258de419e4f1db5de38580b23bd6b : SUCCESS in 3m 35s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/129fc3fa78834cb99fdc5e771edcb9a2 : SUCCESS in 7m 04s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/81736523e51841829f1eb1bce0b8158a : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd4303725fc3493c91d8441eac1ed9b2 : SUCCESS in 6m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ef2a266a00384076829fbba744698f66 : SUCCESS in 9m 43s","accounts_in_message":[],"_revision_number":1},{"id":"07b6624a8c877588702a9a48369a216837adca39","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-10 23:26:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d2d24cf5c7730eb6556dbe5908188f9f6075f0b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 23:26:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"32388f83e0a6d92b3784872d4e04e22941976f3e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-11 01:32:30.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6047593a22254306a65383b58036385e\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/24c2034ae8bc4b41b1af30a9750ad509 : SUCCESS in 52m 51s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/46d8f6a4e94c457a8ec38fc5b3a303ab : SUCCESS in 26m 52s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/d765af22b8c94701a3bf96ffc645ccda : SUCCESS in 1h 10m 49s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/f1ef92b498224ea599b13f529a19d481 : SUCCESS in 57m 23s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ef67086ab68d4f688d99a2ec3d99171c : SUCCESS in 51m 43s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/28b69a34f5224a78934c3a93c12e4379 : SUCCESS in 27m 10s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/8248701197784f15864e381fa14111bc : SUCCESS in 1h 28m 53s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/713e1bf2c0064c77a31434226b828bba : SUCCESS in 1h 53m 25s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/618998a9b4234fc88414ec63907cbb6e : SUCCESS in 1h 20m 41s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/0f754075784f4f44bb82874f558379fb : SUCCESS in 1h 46m 08s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/11cb24c0e7434cd5885c10e4bd18de0c : FAILURE in 1h 50m 03s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/98dbf48dfa4d46ffb10b6d546c05bd8c : SUCCESS in 2h 05m 21s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/9b16c44b35cc4e729b52a33a4affe71c : SUCCESS in 2m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7318a7ae091e47f79645d3da74a94f7f : SUCCESS in 3m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e0c45f526cc242ef900367e4ad30ae62 : SUCCESS in 2m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0663005c00854f81ba528e4b4f5c70da : SUCCESS in 2m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93f1406e58404749a94abae22a47eb96 : SUCCESS in 5m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0b884cfc48a64c02bac4edd378d1c6e9 : SUCCESS in 5m 45s","accounts_in_message":[],"_revision_number":1},{"id":"917cfc280f3fa68291941f45b8080cbcba930386","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-11 07:30:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fbb1d5a47fa2af0a85c6e2abe327072b0b09bf8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-11 09:36:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e18c6840cd3a49a4a217e680aff481b2\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/964fea911916412bba9f7c5e1dee9de6 : SUCCESS in 1h 16m 36s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/0ea01894bebe4260826fc7ac1c395a12 : SUCCESS in 1h 07m 39s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/9ef30c382bf94fc9868ecc58790fd855 : SUCCESS in 28m 10s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/e6834a5159c548c5a6d28ce66bc4ea20 : SUCCESS in 1h 24m 06s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/aad58ae751354e6f8e75eb53fd00c3f1 : SUCCESS in 1h 05m 26s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/68dbbdfe5f36472dbe99952fbc3653cb : FAILURE in 1h 12m 12s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f0c1962e17e74174b6d1e0bfcf90bb2a : SUCCESS in 16m 45s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/e893b17af8094d1680492f417a53c0f0 : SUCCESS in 1h 27m 28s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/f595b572ef344c6c80491fab443e3dab : SUCCESS in 1h 27m 07s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/21c4596331b345dcb3c728980304f8f4 : SUCCESS in 1h 51m 46s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/4981d5a0d3874172822a1c3a7ef040cb : SUCCESS in 1h 24m 25s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/2b42b9fd431742d1bdabcdf660f64914 : SUCCESS in 1h 55m 38s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6fd06f8f6ead4438a161f5e542e6a95a : SUCCESS in 1h 29m 37s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/27348d4dcaf24f75b928dcda726d6ea6 : SUCCESS in 2h 05m 58s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/d2530a71c951471e9c068ce5a12191e7 : SUCCESS in 3m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28bc718bf6f84afb890029ddaf5c870e : SUCCESS in 5m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cff60a1ba490427588e38a38ea150961 : SUCCESS in 2m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a378fecc0f024541918a86ee67781575 : SUCCESS in 2m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a55403d077d84ca39b4839f087940df6 : SUCCESS in 4m 04s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2f29bd680f734d4fbef1a18687763994 : SUCCESS in 3m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f02262a2a3a449348f4a9ae418e0210b : SUCCESS in 7m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0bebb200e842439abb56b95c4ea37b78 : SUCCESS in 10m 54s","accounts_in_message":[],"_revision_number":1},{"id":"df14a69ae51c70cb48478602c6e9275eceb7dedf","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-06-11 09:40:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"eeb0b1630a9f3312627ba77e220880f38b8878f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-11 11:45:31.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7ddac87d08fd4b5783ed5a9bb61894f6\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/7eec5bfd8b724b28bfa6baa9754e5481 : FAILURE in 1h 17m 07s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/0e14fa809e104ddaabbe6ad057ea14f0 : SUCCESS in 1h 03m 40s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/fbca3a123a6246f4b52281bf818c7071 : SUCCESS in 26m 41s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/0c4f6a2ef78d4353a355cbe8aa30fc2a : SUCCESS in 1h 03m 08s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/8680cfdad1bb4768b2ad7d6548790beb : FAILURE in 1h 00m 26s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/592d4e6490364549959c826343c97b2e : SUCCESS in 59m 52s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/115b8c03e0954fb6ae323eb54273e82b : SUCCESS in 20m 44s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/b4431c37aca0429b9429544fcf3e22aa : SUCCESS in 1h 22m 22s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/0a3a1b7e8a304193b698069a3c20784b : SUCCESS in 1h 23m 56s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/5ab8e3c5883043bcb32bea3727601181 : FAILURE in 57m 36s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/d2039b493c4b4c7bbb560eeb0a380606 : SUCCESS in 1h 20m 54s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/8f28c24082c3477c95748c3572af2b4e : SUCCESS in 1h 51m 55s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b54ab910eebb4f38865d0e54a958a16d : SUCCESS in 1h 27m 59s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/2d3b8f4c7d1940e884deae500c969865 : SUCCESS in 2h 04m 04s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/4fd801553db7475d9bdd74d871a71b3e : SUCCESS in 2m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebce63c6437846d09d545dbbca3054d7 : SUCCESS in 3m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/027484cbe31142668f8dfdc87dfdb25e : SUCCESS in 3m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e8b1d5621f1e455fa67e85d1f67cfdd8 : SUCCESS in 5m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/323a17787d2e44c9aafd176aed5345b5 : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0fca845de8114b189b2c0162e16deaa1 : SUCCESS in 3m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb8a549d07f64effad95be12923a85cf : SUCCESS in 4m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8ef02ea1bec14511a10b218452067e55 : SUCCESS in 9m 49s","accounts_in_message":[],"_revision_number":1},{"id":"0def54bed58cd733108fe037ea9be8f90ad4dd78","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-06-11 12:20:51.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b37a4d6085111be3926c6c31e8837f9df97ecd21","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-06-11 12:50:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"628f84ad337dfccb92a3aeb9d028027447420786","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-11 14:51:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d6fed65c8fe4b2ba27f8aa8d1f20c42\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/11e8f6cadcf14be4945d014f631c3d40 : SUCCESS in 1h 05m 39s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/8808d04bccaa45c087b7b409e4486990 : SUCCESS in 1h 01m 12s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/b9b2f9f9f93c48cda9852473365e37a1 : SUCCESS in 13m 29s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/8caa9ecba4014f4887790a68774151d1 : SUCCESS in 1h 10m 54s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/6e12409f64bc42c48aab4df12db0c4cb : SUCCESS in 50m 55s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5b414ad30bde41b3b184e65bc5a8fed9 : SUCCESS in 1h 01m 43s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/88b293441e1246c8a750b69f5337df9a : SUCCESS in 24m 34s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/94546cc8423b4f68be99e3a2b7dd6882 : SUCCESS in 1h 19m 19s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/7ab67c834dc24cc7949594b0665236f8 : SUCCESS in 1h 33m 42s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/cd8dfef3bf8944c99e4bc6274940bdde : SUCCESS in 1h 59m 28s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/a4592b71d4504ba6ac40bad972bee88d : SUCCESS in 1h 18m 04s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/c7f46dcc8dc64d6492a5f8e8e1095c7a : SUCCESS in 1h 48m 51s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/44d0acea1cb34d28af587274bd309f81 : SUCCESS in 1h 26m 26s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/08bf56555a37482ca2a7206d065024db : SUCCESS in 1h 59m 36s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/8010638a57f24a3bb2236dfb66fcb4e6 : SUCCESS in 2m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46bda79b408c43e78e7a83e5061e1f76 : SUCCESS in 3m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/87552f05701f483997b1354beaeb31d0 : SUCCESS in 2m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9853b79281c6486283d7f13a3d0a2782 : SUCCESS in 2m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ef6a91847fd84f48bdc490397f180d09 : SUCCESS in 3m 25s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b8fd4ab0de694def833bd630114d6ef4 : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce17bc55484749abab6d836168a6775e : SUCCESS in 6m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b8a44f2e379040d0ad787995f1456d77 : SUCCESS in 6m 01s","accounts_in_message":[],"_revision_number":1},{"id":"8c14ddf495ee13a506d879ec7d064e59fefb49b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-11 14:52:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"faffc3984b585111a739a721a827449ca575b063","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-11 16:49:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9bd1aa966e594ba98a2df82c2f07dde9\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/e61020ab10e64db1b2ddda34e25a6328 : SUCCESS in 59m 02s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/dc98e7056dee477fa3e7812bbf2382ee : SUCCESS in 28m 57s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/f903bf2e0d6e4cf19d5f7f7f4a602733 : SUCCESS in 1h 00m 16s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/0f408104619043bd98f23c017828711b : SUCCESS in 1h 00m 31s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/55fe95906294468f982459ed923c446c : SUCCESS in 1h 08m 27s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ad6911c5222948c18344024c8886596d : SUCCESS in 32m 50s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/0594751b521a4d17ab739e95d630cfad : SUCCESS in 1h 21m 27s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/bd84d82e1b4e4bd086f8d3b8fe5edafb : SUCCESS in 1h 47m 45s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/c8817a30abbf417396b1964b216add22 : SUCCESS in 1h 16m 24s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/fc9243010e044af0b79859f71b90f97e : SUCCESS in 1h 46m 10s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aa922349f2d44775a0f6014b9f784c3b : SUCCESS in 1h 25m 25s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/fcce4082cd7b476e9e0d4f065bfc1c1a : SUCCESS in 1h 56m 25s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/903a86aceb3c463cb4a5dcc791c2b915 : SUCCESS in 2m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e350ff47b7a34f25b2461529c2ff3067 : SUCCESS in 4m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4e63177632bf41d19da7cf8b1fd047f5 : SUCCESS in 2m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ca127dbd2d5d418c8ce9f1956c8ffd05 : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80957ed08efd4f909a9e6ce8b1663e9d : SUCCESS in 6m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0fcdedbc967f4c61a13aa4f2f94699d2 : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":1},{"id":"57c23b282d6d0eb7b19e8a14283a0bc5ac577279","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-11 16:49:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0c68ff0a9eb7414d9ceb4c0152274d880f61ed2f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-11 16:50:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6354f0e8132f4692b91c31e4a85882c9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43e545c483174eaab683cb991aaf39a1 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":1},{"id":"d6f7a05e47edf6cc26cd0bdb17ffc19794ea7131","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 15:33:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/817398661a2640418caec69a053e8576\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/fc71bc5ca38b4ec8aadf7ef5ecea9bad : POST_FAILURE in 2h 32m 32s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/4a73519d2d39479faa7dfc715b0499b5 : POST_FAILURE in 1h 24m 23s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/144522f147ff43089a4b8ac3af2d9b64 : SUCCESS in 1h 28m 20s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/684a8ff6e7b74d93abf783f7510295ad : SUCCESS in 2h 27m 15s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/01c794ef79a54d3db40b068b52fc326b : FAILURE in 2h 03m 20s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/4b49f73e5f694d578ac697059c890301 : FAILURE in 1h 42m 11s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f65e33a5ab82c66723e4a10e27193f17a00faa12","revisions":{"f65e33a5ab82c66723e4a10e27193f17a00faa12":{"kind":"REWORK","_number":1,"created":"2026-06-09 06:40:16.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/94/992394/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/94/992394/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/992394/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/992394/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/992394/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/94/992394/1"}}},"commit":{"parents":[{"commit":"2c3ea11e133c6c378413d6325fdd53ad792692de","subject":"Merge \"horizon: rebuild static assets after theme changes\" into stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2c3ea11e133c6c378413d6325fdd53ad792692de"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-08 13:08:39.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-09 06:40:16.000000000","tz":0},"subject":"keepalived: Install mariadb client package","message":"keepalived: Install mariadb client package\n\nNeeded-By: https://review.opendev.org/c/openstack/kolla-ansible/+/992020\n\nRelated-Bug: #2155865\n\nChange-Id: I1c67f6b0f4780b7caf83ccf95a20861d9fefe20b\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit 7477b2244ee525ffd6effeb4229608991c639fc8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f65e33a5ab82c66723e4a10e27193f17a00faa12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f65e33a5ab82c66723e4a10e27193f17a00faa12"}]},"branch":"refs/heads/stable/2026.1","conflicts":{"base":"bf4ee2deaa53c2dc585f3d270e036579350b1b52","ours":"2c3ea11e133c6c378413d6325fdd53ad792692de","theirs":"7477b2244ee525ffd6effeb4229608991c639fc8","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
