)]}'
{"id":"openstack%2Fkolla~992395","triplet_id":"openstack%2Fkolla~stable%2F2025.2~I1c67f6b0f4780b7caf83ccf95a20861d9fefe20b","project":"openstack/kolla","branch":"stable/2025.2","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-06-09 08:05:22.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}},"hashtags":[],"change_id":"I1c67f6b0f4780b7caf83ccf95a20861d9fefe20b","subject":"keepalived: Install mariadb client package","status":"MERGED","created":"2026-06-09 06:40:32.000000000","updated":"2026-06-09 14:45:49.000000000","submitted":"2026-06-09 09:13:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"992395","cherry_pick_of_change":992239,"cherry_pick_of_patch_set":4,"meta_rev_id":"871910ed2ced1a93d5965ab23e01f202731a7689","_number":992395,"virtual_id_number":992395,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-09 09:13:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-06-09 07:24:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2026-06-09 08:05:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2026-06-09 08:05:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-06-09 07:24:33.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-09 07:24:33.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2026-06-09 07:59:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1d6cfdc072e32b518df1baca7ddb68454b6dbd4f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-09 06:40:32.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"5d6b9b7dd517cebd7f9ccb250725be7a6ac1a056","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-06-09 07:24:33.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"74038bd7ba74a1388bc75e0774859d0363e28eb4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-09 07:59:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e380350abe041ae9c9b38c58eb4bd36\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/8d85c78abeb749c9b8ef9f51f7efc7df : SUCCESS in 43m 17s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/edaa5446d7f149ca86975f529d12aa2d : SUCCESS in 49m 48s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/4d62e38507f14b3da274eb18b9485338 : SUCCESS in 32m 20s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/d6f08d877af040a889c08161c3ca4c19 : SUCCESS in 50m 02s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/de781305c8894c6cbe18063821299888 : SUCCESS in 51m 06s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/301bd1a426ee43d18f552d16a1a97e24 : SUCCESS in 1h 11m 39s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/661166d1cafe445ab52232dc3ccf587b : SUCCESS in 27m 28s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/9a19b4c56d064daebe1ca91698668def : SUCCESS in 3m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48886eacb78a4e12ad178bf3fe2feecd : SUCCESS in 2m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e907f7a4d48743d5b572940617a514f0 : SUCCESS in 4m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/693ff6b146364a5cb0efe54e64e5d0c1 : SUCCESS in 3m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ee773017126f4af6b6dae3d69d54f9b3 : SUCCESS in 5m 47s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/490ac6a454c547078f5afe2b3ee252a2 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30897f8ea57f42bcaa8b319dac9effce : SUCCESS in 6m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e12124018fbe4ebe9e46147e07143877 : SUCCESS in 7m 26s","accounts_in_message":[],"_revision_number":1},{"id":"17976e330f7c30ac76391e202dc47ad51b12206a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-09 08:05:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bfd7b72a3071a8f1569580278992f3fa27d40599","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-09 08:05:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"30cb7d2586b087906eb77586e9cc74db78976548","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-09 09:13:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af18b8324fad4c39a04096e6ffca5686\n\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/ff3e68291b21438e9330d40bf4a4c676 : SUCCESS in 42m 46s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/a7f6ed8fe1a24c13b6373572097096b4 : SUCCESS in 29m 21s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/3504daab2e484cfaa3dabed5b97d7d6e : SUCCESS in 38m 06s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/e34bf9acf4254a8d9ca296cb5227274c : SUCCESS in 59m 06s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/170de359fad84a99a1edecc2aa258314 : SUCCESS in 41m 36s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e363af211d9a42eca062bdaab7d73750 : SUCCESS in 20m 36s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/82bd5e7eb00c47ae98f78234c9a12610 : SUCCESS in 2m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f61ae125c0e4b76906519c1e911c01c : SUCCESS in 3m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b911dccf8d844b43b712d2bacc69a5fd : SUCCESS in 3m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d3aa6101ed964cf9a86f799d73638b25 : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54d2c4c251aa4209a80ea2e0c889c976 : SUCCESS in 3m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1840453922934a41a0f3938176c5ce5d : SUCCESS in 7m 01s","accounts_in_message":[],"_revision_number":1},{"id":"6cf22f064ca454d12ef78da24f6d3bf37ff3e658","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-09 09:13:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b46b5ad435dd82578456eef67662865f369b97e3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-09 09:14:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1bdd020486d840e293eea06d9e69fb2f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2711b93b7f73492b9ca8e8a892a83aba : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1},{"id":"871910ed2ced1a93d5965ab23e01f202731a7689","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-09 14:45:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a7e463a8af34de1ae1bc774ff6586f4\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/155030893f0346509c42d7895788ebea : SUCCESS in 1h 26m 38s (non-voting)\n- kolla-build-debian-bookworm-arm64 https://zuul.opendev.org/t/openstack/build/7f4e4377dec6420c85deebfe644fbc81 : SUCCESS in 1h 02m 39s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/01ce55c3c9294a79883c83d90f17c96c : SUCCESS in 1h 04m 00s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/7c1c94da8a494c78ba0ab6d7f05f487e : TIMED_OUT in 3h 02m 11s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2fd002690077ee4d2f0a3922b45d624d4dfa193d","revisions":{"2fd002690077ee4d2f0a3922b45d624d4dfa193d":{"kind":"REWORK","_number":1,"created":"2026-06-09 06:40:32.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/95/992395/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/95/992395/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/992395/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/992395/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/992395/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/95/992395/1"}}},"commit":{"parents":[{"commit":"9f23df96a1b14a1b3ce2bcad6344b94aa7c840ac","subject":"Merge \"Update Prometheus to version 3.5.3\" into stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9f23df96a1b14a1b3ce2bcad6344b94aa7c840ac"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-08 13:08:39.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-09 06:40:32.000000000","tz":0},"subject":"keepalived: Install mariadb client package","message":"keepalived: Install mariadb client package\n\nNeeded-By: https://review.opendev.org/c/openstack/kolla-ansible/+/992020\n\nRelated-Bug: #2155865\n\nChange-Id: I1c67f6b0f4780b7caf83ccf95a20861d9fefe20b\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit 7477b2244ee525ffd6effeb4229608991c639fc8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2fd002690077ee4d2f0a3922b45d624d4dfa193d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2fd002690077ee4d2f0a3922b45d624d4dfa193d"}]},"branch":"refs/heads/stable/2025.2","conflicts":{"base":"bf4ee2deaa53c2dc585f3d270e036579350b1b52","ours":"9f23df96a1b14a1b3ce2bcad6344b94aa7c840ac","theirs":"7477b2244ee525ffd6effeb4229608991c639fc8","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
