)]}'
{"id":"openstack%2Fkolla~992396","triplet_id":"openstack%2Fkolla~stable%2F2025.1~I1c67f6b0f4780b7caf83ccf95a20861d9fefe20b","project":"openstack/kolla","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-06-25 15:12:16.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}},"hashtags":[],"change_id":"I1c67f6b0f4780b7caf83ccf95a20861d9fefe20b","subject":"keepalived: Install mariadb client package","status":"MERGED","created":"2026-06-09 06:40:50.000000000","updated":"2026-06-25 18:53:58.000000000","submitted":"2026-06-25 18:52:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"992396","cherry_pick_of_change":992239,"cherry_pick_of_patch_set":4,"meta_rev_id":"966e94ec351caa029615cfeba5cca7e5542098e1","_number":992396,"virtual_id_number":992396,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-25 18:52:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"all":[{"value":2,"date":"2026-06-10 15:06:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},{"value":2,"date":"2026-06-09 07:24:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2026-06-09 10:17:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2026-06-11 13:31:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2026-06-25 17:13:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},{"value":1,"date":"2026-06-09 07:24:43.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-09 07:24:43.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2026-06-09 08:19:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-09 10:17:28.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-06-10 15:06:33.000000000","updated_by":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"reviewer":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"state":"REVIEWER"},{"updated":"2026-06-11 13:31:55.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"}],"messages":[{"id":"2eaa79f38b616bd79055daaf3b1b8aad3d6562fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-09 06:40:50.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"55d818075adfdf253ccdb4485e647ba0e6d2b04a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-06-09 07:24:43.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"509e144a45e38d0c5703ce181e6df58cc0da63b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-09 08:19:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d62c2c3b5bc6441cb3dcef5c6f84cb0c\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/746efdf78e3c473083540788bb4f8a96 : SUCCESS in 1h 04m 03s (non-voting)\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/2f1cbfc3b33f4d179887faeb2e3be6c5 : FAILURE in 52m 47s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/1988cc981d834d448abe53ac603d7b64 : SUCCESS in 1h 29m 50s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/daeffdf81a4d4dfb99589592ce7bd962 : SUCCESS in 56m 51s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/326d1f0287b141d4a5ac1dc5a1a569b9 : SUCCESS in 50m 17s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/49ec4ff57bcf4e639e032534f35d23e2 : SUCCESS in 23m 50s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/8a9339a2d50948869ddb1f98baceaff6 : FAILURE in 10m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cc0847dde8741c88b060d7f3f77fed2 : SUCCESS in 3m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c8e67c2507e74a18b29209c2cb8d55b6 : SUCCESS in 3m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7f69204f869c4a298b0e4e33d56a4a3a : SUCCESS in 3m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/162a6f32d83c470e9c740fe13d205162 : SUCCESS in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b90dfc4a86d54a55baced03433e405fa : SUCCESS in 7m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1282b8d48a774ede88be946746365b50 : SUCCESS in 8m 55s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/a85262945a9149bea148b71f43681971 : SUCCESS in 3m 05s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/2d4ad2f07fe4448c9deea07e2fdfb7c8 : SUCCESS in 58m 36s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/065725186ed2469ebbde8b90ecc9c83d : SUCCESS in 1h 21m 59s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/141d75a53aa2476ca89d51030c703f1b : SUCCESS in 57m 54s\n- kolla-build-rocky10 https://zuul.opendev.org/t/openstack/build/c499187b58d74f99bd8fa2947995de72 : FAILURE in 50m 43s\n- kolla-build-rocky10-podman https://zuul.opendev.org/t/openstack/build/d5f1cdddf8c5455aaf0bcaac832e8220 : FAILURE in 52m 45s\n- kolla-ansible-rocky10 https://zuul.opendev.org/t/openstack/build/a1f0a6e99cec4834bc8130cb9fdc1fe9 : SUCCESS in 58m 56s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d6954d0e6d284a47b9d4ccca4be6f53b : SUCCESS in 42m 59s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/400b8f49d1034b16a4a6f15561d9af2b : SUCCESS in 27m 04s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/0c5259b286e74fdc8a8d4e3eeb2135e0 : SUCCESS in 1h 05m 28s\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/18ad9b1378ec4fa6a4389798cb0a8fc8 : FAILURE in 1h 28m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d48cb1ea549ec83a6c152ce69b88932ab93d985c","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-06-09 10:17:28.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"fc784a8300ba0cb0eb3cc5629c6abd0af527521d","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-06-09 15:12:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c572ce765d81743c07e9163f5269f534f48c609e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-09 16:28:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cae04fb828614f648d46125ebf641346\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/c7d567442170455d97b2e2f6564dbeac : SUCCESS in 1h 06m 47s (non-voting)\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/8a7e44d0f57f449a8ae4bfe03fba9082 : TIMED_OUT in 3h 12m 56s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ac6f0378eb8a4e9b98eac1ac59874f5e : SUCCESS in 1h 42m 21s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f7782bea607647c8b890bfe217e04bcd : FAILURE in 12m 38s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/a6c117c803cd4a3ab18772108508fbe5 : SUCCESS in 1h 07m 55s (non-voting)\n- kolla-build-rocky10-aarch64 https://zuul.opendev.org/t/openstack/build/964ba42415c4493bab3f222082574c02 : SUCCESS in 1h 33m 42s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/eaf4f94262d24f64b2ff249a5a977ae4 : SUCCESS in 1h 04m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"940eaaee56610715fa2853746b3e8a807437fbb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-09 17:18:07.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/87da12e1226c495daa810bdd790af49d\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/c6abc0969c254c6c8dddeed33edc7219 : SUCCESS in 46m 38s (non-voting)\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/61492e7bef6749a0a7e3ed177faef6b2 : FAILURE in 40m 39s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/a4591503f6064a90aa0978be6adfcc33 : SUCCESS in 1h 45m 29s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/ee3d9417b4e84f3388318dcf3e4be4ed : SUCCESS in 1h 30m 17s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/35e2fbdc76574cc7b625e9a73f51b0d3 : SUCCESS in 48m 29s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/73b4b66ee3db48b18590bf3a8648f51c : SUCCESS in 36m 10s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/1418e33bdff14c04bbfb6bd951ee58d5 : FAILURE in 6m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7fa561f1e0e4b068fd5d2b8cb6ea412 : SUCCESS in 2m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e438482dd000409c860b3d95d5869ea4 : SUCCESS in 2m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/44a0f52100d54f979acf08d63c4ff16e : SUCCESS in 4m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c524304459ee47cb9bbdcaf7e90b6f9a : SUCCESS in 3m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f3512541fd84d619b469096822fc97f : SUCCESS in 6m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3b6af406bf1f4320a33180a116bae6f7 : SUCCESS in 9m 04s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/9a0c7d1d4f1446d3a213694aac89eae2 : SUCCESS in 2m 19s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/0bc402fabc1b49c198ae5e7fb36d3f8c : SUCCESS in 1h 57m 12s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/73dc19050974455e9fd158406fec20e2 : SUCCESS in 1h 16m 15s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/10571846f0134ccf872dd9e70386bdff : SUCCESS in 1h 28m 53s\n- kolla-build-rocky10 https://zuul.opendev.org/t/openstack/build/84d6e8e17c074e799b53c047e100f422 : FAILURE in 39m 37s\n- kolla-build-rocky10-podman https://zuul.opendev.org/t/openstack/build/dcbcc509012049e8b5bf589af4d2e3d7 : FAILURE in 1h 10m 56s\n- kolla-ansible-rocky10 https://zuul.opendev.org/t/openstack/build/565d9b4c859c40709f8034eade0107a1 : SUCCESS in 45m 39s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/20465fda300946d4bfcec2cf8e1f32e0 : SUCCESS in 50m 21s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/37287b2daa0149aabbe520a6b2463a78 : SUCCESS in 27m 29s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/ccac1320d54f4d10af69fa45390c7522 : SUCCESS in 1h 05m 06s\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/7293b35f6e3a4c9ea4ca58da277d89aa : FAILURE in 1h 30m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"541c458a00f12b269417d91e47873620927aa212","author":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"date":"2026-06-10 15:06:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1e1d4bdaffd7946a4129431914f7745da00c7ad1","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-10 16:37:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2c175d6649be96543f82e8d01a152feba033d545","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 17:19:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e718b468dc9469a9390be46500b09ff\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/2fc179b9f58d431a9142038558eadb6e : NODE_FAILURE Node(set) request ba3bdba4b57f43b38b6a45959d6225b8 failed in 0s (non-voting)\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/6e0e93d706e8481f8ae9c73eded4b797 : NODE_FAILURE Node(set) request 543a27d88a5c4da38af24108a4b7420e failed in 0s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c1db17be5613455694809ee472bc7e96 : NODE_FAILURE Node(set) request e0d8f2ae055d4dfda30950fea100c0b2 failed in 0s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/130ddfa14d824b5fa36b01c1f7119cae : NODE_FAILURE Node(set) request d5f5c1115528424186084b53c8ac1be0 failed in 0s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/1fb32e7a940a44f8b77e01b69097075b : NODE_FAILURE Node(set) request c930f9b51ea3486fb544c5d6ed200c15 failed in 0s (non-voting)\n- kolla-build-rocky10-aarch64 https://zuul.opendev.org/t/openstack/build/4cd1e0fbb75047ba8552f637a2c540b5 : NODE_FAILURE Node(set) request 3ef354b0bbc04ff7b67f306973c2151b failed in 0s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/07dfe5a0905341fa8377ea4817b2d550 : NODE_FAILURE Node(set) request efbd39f3a340433bb0601491453c2b9a failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"52d59c27ee04c27658e465cc5d465d6bbfb2aa4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 18:37:21.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c19b6cae7131427d8d54d56d24fe5d9b\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/e277c4f5e4524e22b0cc9b72a7c65e6f : SUCCESS in 51m 14s (non-voting)\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/17768f0e46fe477791d2fe294579530c : FAILURE in 52m 49s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/c957054a9ce6434b8e91e8116a436172 : SUCCESS in 1h 39m 22s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/19e3753df426428286ac66de24d584b4 : SUCCESS in 59m 10s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/92578cd8760c41d5818e4dc01a2dd307 : SUCCESS in 36m 34s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/ce8495477570416bb9a204240ae52c84 : SUCCESS in 29m 07s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/724dbdacb7d244ce998aeb0b3496a718 : FAILURE in 8m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/517e335ecf24459a8b6a58c2414581ea : SUCCESS in 3m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/96b707761f454a6a901849cc72f6e593 : SUCCESS in 3m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0c7e9be0820f4c7b8d8f7ffbc9a3c44f : SUCCESS in 4m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/690b08c44888460a88eb2bbffcd0ffa8 : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e7592cd41aa49918bc6171dc099f8d9 : SUCCESS in 5m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4db4e5e387d64511bbfe1a1dbf408a24 : SUCCESS in 9m 41s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/29a42e1134b4472c9a1a8591e93b0e33 : SUCCESS in 2m 09s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/8331ad0cb3c0412cba6e3f954590d68e : SUCCESS in 53m 34s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/6559e26f82954c808f797af88b64eb37 : SUCCESS in 1h 17m 15s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/dad0a8e610854cd48b399e02dc0fbfa3 : SUCCESS in 1h 29m 29s\n- kolla-build-rocky10 https://zuul.opendev.org/t/openstack/build/f2ae51b4a3774b81b5182b6e296aa17e : FAILURE in 52m 02s\n- kolla-build-rocky10-podman https://zuul.opendev.org/t/openstack/build/51187120cb614e518eba3e183db53f9e : FAILURE in 51m 24s\n- kolla-ansible-rocky10 https://zuul.opendev.org/t/openstack/build/9adbf796e0d048e6b4ee943958407ff7 : SUCCESS in 47m 21s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b3f53c2cbc5444e19fc46338e28f41f2 : SUCCESS in 52m 32s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/a465d656863c438287955e3fd63ef153 : SUCCESS in 29m 27s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/38c751609c5a42729a6e66b9a3e87cea : SUCCESS in 1h 05m 40s\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/d6bc8feb876846488d139f030bf6ef6e : FAILURE in 1h 35m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"698744e1df129af66f9eff176cf53f20f92d473f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-10 23:27:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f2d0fd26e3a67ba9e3f63145ff7c055de7425b9f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 23:28:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/664e85617bb94513a4c42808713358d7\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/01706726025f4c0f87b110920ab0430b : NODE_FAILURE Node(set) request 9c6e85c84cf547fab4552d1a4a471644 failed in 0s (non-voting)\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/df6306b876984d699d091296251f551f : NODE_FAILURE Node(set) request 6c7df4f1c9b7429db179130286c27759 failed in 0s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/51f0e7672a6a4e95b2212a612cdb08b3 : NODE_FAILURE Node(set) request 480a5c1ab5a144b7969b7cea6a1e9e3b failed in 0s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f9a7f599f0de40b48249e91bc549a77a : NODE_FAILURE Node(set) request 550722f8922743a1b638a84dfad75c7f failed in 0s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/ebace572b56a49d09b8ba560adc42d45 : NODE_FAILURE Node(set) request 3a2fffcac0b84f2096c40f365092a47f failed in 0s (non-voting)\n- kolla-build-rocky10-aarch64 https://zuul.opendev.org/t/openstack/build/6c9fe25c07e5484a9f82b2d01005738d : NODE_FAILURE Node(set) request a43b4e8f70424394afdfa46a24695622 failed in 0s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/ff8e99a471ad4232af42b1af0ad38748 : NODE_FAILURE Node(set) request 9550c8b30d4349529295ce73c27e50e9 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4bf8a6abc6220c4647c11382526a80fd09766cb4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-11 01:00:37.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/56a55d8dddc84c2583e61d5e0ed68c7f\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/33b0e3d433634538a23cad432f9bc2a5 : SUCCESS in 1h 05m 04s (non-voting)\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/4c0c88c17f7844ada313461d89d15f4d : FAILURE in 49m 08s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/9dd69f0563a84cd4ac9d4155f28ab604 : SUCCESS in 56m 44s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/5eb4b7057200492299017d22bf615c5f : SUCCESS in 1h 11m 41s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/9bd138e994464e99bd5c97e47b4e4f4a : SUCCESS in 52m 20s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/22ea83bd6eae42ffa8943a5edd202bdd : SUCCESS in 31m 24s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/163d55660a5a46ea84afcd707e04716e : FAILURE in 5m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5dfbc2acb3574ca3b254222ec472d115 : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7ba2bb80e4834f60bca747a9bff78d6c : SUCCESS in 3m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6133e658e78e495794eb87bff11bfd6e : SUCCESS in 3m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/99516c096b984babb8ce29355f25e42f : SUCCESS in 3m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6583431b3ec44808ac7d0d3b939fbff : SUCCESS in 4m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5a9a702094ab458bb4fd3f96bbd0d530 : SUCCESS in 8m 18s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/dd3f8071678e4d6da9ac847d6e6c6168 : SUCCESS in 1m 48s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/7ea7a1b4ab234b119471275b4864d25f : SUCCESS in 49m 30s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/e882bbc26525493c9d0a2fded61101f9 : SUCCESS in 55m 39s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/ce93b740ed2847c1b93491682c76559b : SUCCESS in 56m 17s\n- kolla-build-rocky10 https://zuul.opendev.org/t/openstack/build/bd283d336d8e424689202abfc86791de : FAILURE in 39m 44s\n- kolla-build-rocky10-podman https://zuul.opendev.org/t/openstack/build/29372057c02e4541b3ad66ee3c9734c6 : FAILURE in 48m 09s\n- kolla-ansible-rocky10 https://zuul.opendev.org/t/openstack/build/b35106e2228045b681df4cd2dbfbcb49 : SUCCESS in 46m 12s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/330c772aa39346af859a8cc60f4956b0 : SUCCESS in 58m 15s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/4f73e7e6aa51406195becee3cf734c60 : SUCCESS in 22m 15s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/e40ac93fe79b41d1b511f8cece419c4d : SUCCESS in 1h 03m 28s\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/7755323ca5c843f8a5e74810cb6dcd27 : FAILURE in 1h 27m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"13180af809cb82f1227311182fddb2887783820b","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-06-11 13:31:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0b70c22f9f54802c3e7eec0be733c3281f83ef12","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-25 15:12:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3854fee9e3179ac63e5f15b94a07b69b8e7bd3e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 16:52:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5347deb0478e4e798e6b9a6874fef0f1\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/d873f10a3bd64d788f210fd29fcccfa8 : SUCCESS in 1h 05m 36s (non-voting)\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/e2bb0fd535c54a2ba0e3684f93b9470e : SUCCESS in 43m 11s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/cb5d5023be204dd382b54f0a0afc2134 : SUCCESS in 1h 29m 47s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/d435a573d8b349ad8815daadde99ec36 : SUCCESS in 48m 14s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/9ed147b157904756a4e38bb0245eb5d2 : SUCCESS in 52m 21s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/37b4e1dc8aba4f6eabf535b57f406a3a : SUCCESS in 31m 28s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/708019180c1c4fa8889a2bad666daab0 : SUCCESS in 59m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7547340d1bf54492aeb09d128fe3c4b3 : SUCCESS in 4m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/42166d3849eb46fd986ffb114e692053 : SUCCESS in 3m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/de13b97b282d43b08e9f2e381ccf0c10 : SUCCESS in 4m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ec5da95ba3cf4df695a14141b76f2fe9 : SUCCESS in 3m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bda66a914aee4572a12229e009a24fe5 : SUCCESS in 8m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2e072b8b1ce34d98b39b2e63501d403b : SUCCESS in 8m 45s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/439de594ef9b4945ad5eb579a99b2e23 : SUCCESS in 2m 19s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/79d0c884c0a548128b3b5a9628f9fafb : SUCCESS in 1h 02m 56s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/a640275d405f4408823042f0eef2d858 : SUCCESS in 1h 20m 05s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/a94b8b3c4f674057bc5a334f3719b16b : SUCCESS in 1h 32m 44s\n- kolla-build-rocky10 https://zuul.opendev.org/t/openstack/build/2c352f493b9c44b594f354685b7816af : SUCCESS in 54m 23s\n- kolla-build-rocky10-podman https://zuul.opendev.org/t/openstack/build/7108fd3b8d5a44d5aaeced9d4acfe1e2 : SUCCESS in 41m 53s\n- kolla-ansible-rocky10 https://zuul.opendev.org/t/openstack/build/2e99eeefd8cb437d9415049d8b29a840 : SUCCESS in 1h 22m 19s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/cb48db68c6f942518ff3733c3436168b : SUCCESS in 53m 53s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/9e7ecc59b48f4419bae3b78c640359db : SUCCESS in 37m 33s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/aad3f8d1261e4355a64b6824cd9f343d : SUCCESS in 1h 16m 06s\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/b69436b5cb1e4a928147eb293d1aa03c : FAILURE in 1h 28m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0256bec28d6ab69baa1422191bb00497d5b691f2","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-25 17:13:36.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"09370fe4584c3da6f9dadad853cfe89c5bdb4a36","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 17:13:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c0cdfb6fe9c9d6b24de229e117943b06aa26ee18","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 17:24:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/389f2f4f0cb347268b2739f7ffc32fcb\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/26aa9adc630e4bd8bb6e7043f47517d0 : SUCCESS in 53m 36s (non-voting)\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/6da4723c91084f80a32e95925ec3f82a : SUCCESS in 1h 55m 45s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ac5016772a0448d4adf88e9d252cfd0f : SUCCESS in 1h 05m 07s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ea1c1147f107449897f81a883dbf1b9a : SUCCESS in 1h 05m 30s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/235caa9c428f4df5ba17d6f97ddb0ff5 : SUCCESS in 1h 45m 34s (non-voting)\n- kolla-build-rocky10-aarch64 https://zuul.opendev.org/t/openstack/build/bf39b3bd2a164265b568022d0649656d : SUCCESS in 53m 30s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/8590bc65e4d94cbdb75bd0fcb00a67b7 : SUCCESS in 1h 11m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5298c1f9c57352ba13fc17ed43ba5cae8d6bf98b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 18:52:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2050134771f24dada72bd1995faf318d\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/89577bb5a88843afa0c9aa494159c943 : SUCCESS in 44m 31s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/1b14c4210fd645b8974fd15812a02478 : SUCCESS in 18m 14s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/11920747e15d4579868acfe81c700c18 : SUCCESS in 58m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01da85ad73a2456b9d5a872c88cbb028 : SUCCESS in 3m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2a10a24ce17d4bfb88473f73d731304a : SUCCESS in 3m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/af74cf955c9641ebae713f1d458b7da8 : SUCCESS in 2m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fefea969561f4ce9b386877abbf9dd9b : SUCCESS in 8m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fec43a5aee6c4f8ba98ea536d224f4d3 : SUCCESS in 9m 21s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/0561da0412bc4d5da70becf361b59aa0 : SUCCESS in 4m 23s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/9f71927b68cb417dbd78b40de6cac843 : SUCCESS in 58m 29s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/0f41983449604bfc8f4d88a1bc46953b : SUCCESS in 1h 31m 32s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/75b2408a8abc46f9b14b8704a7517925 : SUCCESS in 46m 20s\n- kolla-build-rocky10 https://zuul.opendev.org/t/openstack/build/6f6fdaa0bcbd43e49bc96d69d98b27b3 : SUCCESS in 49m 54s\n- kolla-build-rocky10-podman https://zuul.opendev.org/t/openstack/build/697df9317b5b4f0c9a9682ac59158481 : SUCCESS in 1h 16m 21s\n- kolla-ansible-rocky10 https://zuul.opendev.org/t/openstack/build/81b67df6f07948eda3aa026a93145a27 : SUCCESS in 1h 16m 31s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/633d04c1977c4197b2f445d6b01594b6 : SUCCESS in 38m 04s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/68a991dc94364ddd913d10ac986c8d4c : SUCCESS in 27m 34s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/7982e4355d6d420193c792dd4197df91 : SUCCESS in 1h 06m 36s","accounts_in_message":[],"_revision_number":1},{"id":"0540e589e21dd2f82ff0e6976bbc9bdda611cb7e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 18:52:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"966e94ec351caa029615cfeba5cca7e5542098e1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 18:53:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f0854a17a0d4676a3a0e0376c503ba6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55b10623c1774d77954c7e0fd7990144 : SUCCESS in 1m 28s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b70d53e53ed7fddd5a027500100ca98643bba44e","revisions":{"b70d53e53ed7fddd5a027500100ca98643bba44e":{"kind":"REWORK","_number":1,"created":"2026-06-09 06:40:50.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/96/992396/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/96/992396/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/992396/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/992396/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/992396/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/96/992396/1"}}},"commit":{"parents":[{"commit":"b94865f90dc0e4130e028b25f7c6e1356ebaae4c","subject":"stable-only: fix zuul config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b94865f90dc0e4130e028b25f7c6e1356ebaae4c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-08 13:08:39.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-09 06:40:50.000000000","tz":0},"subject":"keepalived: Install mariadb client package","message":"keepalived: Install mariadb client package\n\nNeeded-By: https://review.opendev.org/c/openstack/kolla-ansible/+/992020\n\nRelated-Bug: #2155865\n\nChange-Id: I1c67f6b0f4780b7caf83ccf95a20861d9fefe20b\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit 7477b2244ee525ffd6effeb4229608991c639fc8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b70d53e53ed7fddd5a027500100ca98643bba44e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b70d53e53ed7fddd5a027500100ca98643bba44e"}]},"branch":"refs/heads/stable/2025.1","conflicts":{"base":"bf4ee2deaa53c2dc585f3d270e036579350b1b52","ours":"b94865f90dc0e4130e028b25f7c6e1356ebaae4c","theirs":"7477b2244ee525ffd6effeb4229608991c639fc8","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
