)]}'
{"id":"openstack%2Fkolla~993188","triplet_id":"openstack%2Fkolla~master~I91b42fe530f87af7d78bb1a543b29a4b0d50872e","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-07-23 12:16:08.000000000","reason":"Change was submitted"},"35366":{"account":{"_account_id":35366,"name":"Michael Sherman","email":"shermanm@uchicago.edu","username":"shermanm"},"last_update":"2026-07-23 12:16:08.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-07-23 10:05:22.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}}},"hashtags":[],"change_id":"I91b42fe530f87af7d78bb1a543b29a4b0d50872e","subject":"Rework tar normalization to a shared function","status":"MERGED","created":"2026-06-12 21:31:07.000000000","updated":"2026-07-23 12:17:03.000000000","submitted":"2026-07-23 12:16:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"993188","meta_rev_id":"3d1ffdb5b9efa53c80eff9614210dc3f04e11f59","_number":993188,"virtual_id_number":993188,"owner":{"_account_id":35366,"name":"Michael Sherman","email":"shermanm@uchicago.edu","username":"shermanm"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-23 12:16:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2026-07-23 09:50:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2026-07-23 10:05:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2026-07-23 09:50:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":-1,"date":"2026-07-23 09:50:23.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2026-07-23 09:50:23.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-12 22:55:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-23 16:09:37.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2026-07-23 09:13:54.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"b7632c5eef460af3e4ac26bd4174164e9f6ffb00","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35366,"name":"Michael Sherman","email":"shermanm@uchicago.edu","username":"shermanm"},"date":"2026-06-12 21:31:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cd95086b841b2b506b351c55a7b1c1040417a0bb","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":35366,"name":"Michael Sherman","email":"shermanm@uchicago.edu","username":"shermanm"},"date":"2026-06-12 22:00:28.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":1},{"id":"c5acfda694e6cf2681e94529e256154dfc907a49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-12 22:55:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8492942d00fc4ad7ab1b3f8c8c138c3a\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/385d97a75bbe479f9e402252ac1a0870 : SUCCESS in 1h 03m 05s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/2f0efe423e4b4c179a51279f6f472e00 : SUCCESS in 1h 18m 57s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/158a0f88ee4a40cf85906fb2b03c2329 : SUCCESS in 26m 26s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/d2b5bb215fd4450689c3d2a5f68ad186 : SUCCESS in 1h 11m 46s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/dd395e3382464d528f6c0103a28d9af6 : SUCCESS in 35m 09s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1c7ad52562cd4c8ca5adb653313e4c03 : SUCCESS in 58m 02s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/391a634d62c646c5bec7d7808b85f8ad : SUCCESS in 14m 09s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/45f1167c622e43d380413cfdcc6eee8a : SUCCESS in 2m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6fa58420a5c54600a6e2af31f7b6966f : SUCCESS in 2m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1afbbb72a78940faa58462b320f442ed : SUCCESS in 2m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d0cd1a00c8d547b28f9b8054eb56720e : SUCCESS in 2m 51s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a26882a5b0ee4c7b9856ca2e0f8fcebd : SUCCESS in 4m 00s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b71ec97270a846c39fb093dc8b7c297f : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2339465b32c4fe89220091ea1a65d77 : SUCCESS in 6m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f3ae06ce3b464cb9a192f615847c39f4 : SUCCESS in 8m 51s","accounts_in_message":[],"_revision_number":1},{"id":"7a25bc6a291dca1dc547b325d5d29509a938c0ca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 23:44:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ceec8fbf2ba3451087b59cb11b9b7161\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/59a2e3aec1e54e47a742a4396ea303ab : POST_FAILURE in 2h 18m 04s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/91d02219ddb94f818a5096fdd002d8c7 : SUCCESS in 1h 21m 12s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/6f72cf2190d0476fb00dc78f24998731 : SUCCESS in 1h 21m 18s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/4954a4071fca4999ab0e1378d8632628 : SUCCESS in 2h 08m 38s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4baf5a4d68c94452a77ee4756dcdd5ea : POST_FAILURE in 4m 13s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e3f7d4869c094d77bc318957d20fce7d : SUCCESS in 4m 05s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/e5319063ab2b43fdba987bb8d2685621 : SUCCESS in 10m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1fd47cf5fa0fed49c23440c147c687cf55cce45c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-23 16:09:37.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"b2e1c4a818cef5c2cf0fd58dc7fb490551adc171","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-07-23 06:16:58.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nThe following files contain Git conflicts:\n* kolla/image/tasks.py\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"cc40430a0e745828c5f91271961e41c839571f49","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-07-23 06:17:55.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":3},{"id":"c42c0c501fddd7f07c66ba795d578925bea0d549","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-07-23 06:18:11.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":3},{"id":"0f272f95085dbb29b5cf50e9e309825e24fa2004","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-07-23 06:19:42.000000000","message":"Patch Set 4: Commit message was updated.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":4},{"id":"82ddddc350160f354d1c665f90bafcebd93edaae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-07-23 06:19:57.000000000","message":"Patch Set 5: Commit message was updated.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":5},{"id":"5edd93180302a11f4dc903a371f77ad9c25f5e96","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-07-23 06:20:29.000000000","message":"Patch Set 6: Commit message was updated.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":6},{"id":"e18f0732e33f992c3c07684c5f9c87a7edc533ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-23 07:32:15.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7f4c8589b5c41edaad4ebab0da48614\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/78012f2f8a534440a3d78b9dd6d1b632 : SUCCESS in 59m 26s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/9d5b587c60d44e8e811d8cd41be7ab59 : SUCCESS in 21m 33s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/4cdb1318eed14f75a6a0f449aa3073e0 : SUCCESS in 25m 40s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/0486c3e201fc481e924659411aeea8d8 : SUCCESS in 16m 32s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/3b12beff7bc645e38202199d44eb1387 : SUCCESS in 42m 46s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0690e21ff17d411ea32dcd755515aed1 : SUCCESS in 33m 23s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6fe3bc9d4b294399a9f2d4b64be89969 : SUCCESS in 27m 44s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/3df7ead74e7a48209ba375ef6eb084dd : SUCCESS in 2m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ca3e966d8474979852dcc8bd4ca1122 : SUCCESS in 3m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d60869998fc548ec8d9698e1988e7e12 : SUCCESS in 2m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7482d674a5874c3d9ef7ca5c3ed183bc : SUCCESS in 3m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6ed2b43dd3324f028f588b8ac44ee059 : SUCCESS in 3m 28s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/20767b5229484c62889050d540ed9674 : SUCCESS in 3m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/884709bd18b8429fad3abbb2066ef16a : SUCCESS in 5m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/768ceaf34ef94ea690e73b8383144bb3 : SUCCESS in 9m 34s","accounts_in_message":[],"_revision_number":6},{"id":"801d530cef45eb31a8df17ec24e3e60761193091","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-23 07:38:23.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95bbbfd89b0248739f8210389dd2c708\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/7093a05de4b54508bac762d7b866a721 : SUCCESS in 44m 26s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/b4ee49a2b3e1497db5f37ffe9e4fd515 : SUCCESS in 31m 28s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/cc198be2b3964008950a5061e941d700 : SUCCESS in 41m 52s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/8a7e7c3952c0458f93baf2f3183389c0 : SUCCESS in 31m 46s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/40b9f0c18a0246df9a22522ce2c88070 : SUCCESS in 4m 15s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f679b8eadcaf4236920d80ed7db1e4f2 : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d9f751ce08fb477080784feca9283ff8 : SUCCESS in 7m 44s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"ff087f03b80ec0dbd8fb1707de711ab074cd8643","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-07-23 09:13:54.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"d4be9d4012e51e96286459fb4377a5cc9e7a9bf9","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-07-23 09:49:58.000000000","message":"Patch Set 6: Code-Review+2 Review-Priority+1 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"443a34d8fd17a60b78ac255ebf27554c36fbcf0b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-07-23 09:50:04.000000000","message":"Patch Set 6: -Workflow","accounts_in_message":[],"_revision_number":6},{"id":"0c32f78a33f7f9e9e3c75065503ccf50169cd214","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-07-23 09:50:18.000000000","message":"Patch Set 7: Published edit on patch set 6.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"da13b5fe5f4c292114449f9b34831ac1a136db79","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-07-23 09:50:23.000000000","message":"Patch Set 7: Backport-Candidate-1 Code-Review+2 Review-Priority+1 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"fd97589f821b4e0dfe9016918a081603d341db8d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-07-23 10:05:22.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"4c9a46edaab26305307d8e817cde5e246d625d60","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-23 10:50:20.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb66855859864625933d0a5989222735\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/20e0c4c2999e4c5c83aa856163d46453 : SUCCESS in 46m 06s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/5c10e12625004cbf9be77c5b07dff7ac : SUCCESS in 44m 02s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/0504ed589fff48068d17f22798a968eb : SUCCESS in 48m 34s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/75de155cc22e4e25b112696b89eaeeba : SUCCESS in 57m 29s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/30a230b36aa1407882b90186f99d616e : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d8abf8deaa4b4981878d32c1bac2c58d : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/c4f55e24354e42e088c350aa635cc16a : SUCCESS in 7m 50s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"5e54b232cc2a3a2e5e1845240a08b1792959d62c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-23 11:16:02.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41922686d34b48d1bc3ce64f4fd658ae\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/9567174455054e6b871c0d489104e9e0 : SUCCESS in 1h 05m 47s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/f46ef5e4710341328bcbd99dac2d0441 : SUCCESS in 13m 19s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/4d9b1ed0fb6b479abd0c6af3c8c5d014 : SUCCESS in 22m 23s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/56fcbcee926941b5a963fa7d74fa7ac8 : SUCCESS in 19m 37s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/9ab76660190a48b2952b7f75eef12119 : SUCCESS in 30m 54s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/052b5c26788a4609b07eeda5f99d31cb : SUCCESS in 26m 49s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7549e715207642c8a8566e836f1e7a9c : SUCCESS in 14m 23s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/44fe7f863884476ebb33e2bb39439a1f : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c14088bd8274a6e83616fe1b0a3a5a6 : SUCCESS in 3m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/02fbf37f77014a8288453ff8b8db203a : SUCCESS in 2m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1b666567ca8045cf946007d79c11a1bd : SUCCESS in 3m 36s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/01a386f20cc0413ebab79690dd76a519 : SUCCESS in 7m 48s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b41b163285754987801095f1e112af8f : SUCCESS in 2m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5da52bddf334c3e861f7d0858f8abbd : SUCCESS in 9m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2da9d13cd8bc43d6b534360ee0792722 : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":7},{"id":"d01eaf4f62a2fe6b0b599fdbe9b76d2f58938fa7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-23 11:16:19.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"0218a8919e1226c74c8e60dd6124327082514427","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-23 12:16:08.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66626a367a164884a31c5a473a565ec6\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/edd5466883a04396b8215e72618afe3e : SUCCESS in 22m 19s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/8dbd22433df5462890109e7f97785315 : SUCCESS in 28m 44s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/7923dcb0326549b3946ca3bbf52d738d : SUCCESS in 22m 25s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/f16b9a8297c84025af0348b0a7e46742 : SUCCESS in 52m 37s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e95d104353994c3191de1e5399187227 : SUCCESS in 37m 46s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/176f164b88b34a4d92e11c17f9a6ab71 : SUCCESS in 26m 51s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/8f0e1eff0f1643d48f8ddbf57fba7b26 : SUCCESS in 2m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35b2ef2bdb0148e58459b0d00a717308 : SUCCESS in 3m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/508e2b14303b48c98fdbdfc4de027c29 : SUCCESS in 3m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6bba0d2c82c441f3a118ab629df3b8fc : SUCCESS in 3m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00ac992a2a4d47188305416ce35a8a86 : SUCCESS in 6m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b0c1b1c45bb7416cb025018f014498a3 : SUCCESS in 8m 45s","accounts_in_message":[],"_revision_number":7},{"id":"a440b2c0b92e945bc2eba2c924deea32b50fc5d0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-23 12:16:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"3d1ffdb5b9efa53c80eff9614210dc3f04e11f59","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-23 12:17:03.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e00ac9c8bd8841d6a7e3ce2f3d46a6fa\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4f4c0074f8d4b42bc8e184a864d5956 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"789c02ae79dabcaa5c141baf99532f464472b0c8","revisions":{"ccc06df2a2c146513c696ace2e5bf13d84d6be89":{"kind":"REWORK","_number":1,"created":"2026-06-12 21:31:07.000000000","uploader":{"_account_id":35366,"name":"Michael Sherman","email":"shermanm@uchicago.edu","username":"shermanm"},"ref":"refs/changes/88/993188/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/88/993188/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/993188/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/993188/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/993188/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/88/993188/1"}}},"commit":{"parents":[{"commit":"4e85395936cf17d6912b68fc51ca6c18e90156f7","subject":"Merge \"CI: Switch to use Debian Trixie upgrade job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4e85395936cf17d6912b68fc51ca6c18e90156f7"}]}],"author":{"name":"Michael Sherman","email":"shermanm@uchicago.edu","date":"2026-06-11 22:04:47.000000000","tz":-300},"committer":{"name":"Michael Sherman","email":"shermanm@uchicago.edu","date":"2026-06-12 21:30:45.000000000","tz":-300},"subject":"Make build context archives reproducible","message":"Make build context archives reproducible\n\nThe plugins, additions and local directory source archives embedded\nfile modification times from the source checkout, so the archive\nbytes changed on every build even when the content did not. This\ninvalidated the container engine\u0027s layer cache for ADD instructions\nconsuming them, including the empty plugins-archive and\nadditions-archive added to every image.\n\nThis change pins the mtime of all archive members to a fixed integer\nvalue, and combines the two existing `reset_userinfo` definitions and\nnew mtime fix into a single `normalize_tarinfo` filter.\n\nThe chosen mtime value is the unix epoch for two reasons: it\u0027s a\nconvention for reproducible builds, and pinning to an integer prevents\npython\u0027s tarfile from emitting extended PAX mtime headers.\n\nThis does not make archives of git-sourced plugins reproducible,\nsince the .git directory content itself differs between clones.\n\nPartial-Bug: #1839319\nChange-Id: I91b42fe530f87af7d78bb1a543b29a4b0d50872e\nSigned-off-by: Michael Sherman \u003cshermanm@uchicago.edu\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ccc06df2a2c146513c696ace2e5bf13d84d6be89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ccc06df2a2c146513c696ace2e5bf13d84d6be89"}]},"branch":"refs/heads/master"},"579255d0e3af64b4151453f918ee4224aad17fd3":{"kind":"REWORK","_number":2,"created":"2026-07-23 06:16:58.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/88/993188/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/88/993188/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/993188/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/993188/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/993188/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/88/993188/2"}}},"commit":{"parents":[{"commit":"ac1b60432590506c3b3fd113959148278596ef72","subject":"Merge \"CI: run copr periodic job only on master\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac1b60432590506c3b3fd113959148278596ef72"}]}],"author":{"name":"Michael Sherman","email":"shermanm@uchicago.edu","date":"2026-06-11 22:04:47.000000000","tz":-300},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-23 06:16:58.000000000","tz":0},"subject":"Make build context archives reproducible","message":"Make build context archives reproducible\n\nThe plugins, additions and local directory source archives embedded\nfile modification times from the source checkout, so the archive\nbytes changed on every build even when the content did not. This\ninvalidated the container engine\u0027s layer cache for ADD instructions\nconsuming them, including the empty plugins-archive and\nadditions-archive added to every image.\n\nThis change pins the mtime of all archive members to a fixed integer\nvalue, and combines the two existing `reset_userinfo` definitions and\nnew mtime fix into a single `normalize_tarinfo` filter.\n\nThe chosen mtime value is the unix epoch for two reasons: it\u0027s a\nconvention for reproducible builds, and pinning to an integer prevents\npython\u0027s tarfile from emitting extended PAX mtime headers.\n\nThis does not make archives of git-sourced plugins reproducible,\nsince the .git directory content itself differs between clones.\n\nPartial-Bug: #1839319\nChange-Id: I91b42fe530f87af7d78bb1a543b29a4b0d50872e\nSigned-off-by: Michael Sherman \u003cshermanm@uchicago.edu\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/579255d0e3af64b4151453f918ee4224aad17fd3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/579255d0e3af64b4151453f918ee4224aad17fd3"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"4e85395936cf17d6912b68fc51ca6c18e90156f7","ours":"ccc06df2a2c146513c696ace2e5bf13d84d6be89","theirs":"ac1b60432590506c3b3fd113959148278596ef72","merge_strategy":"recursive","contains_conflicts":true}},"ad22005afe3d395bf1e40cde7c3f9f09ec21633c":{"kind":"REWORK","_number":3,"created":"2026-07-23 06:17:55.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/88/993188/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/88/993188/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/993188/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/993188/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/993188/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/88/993188/3"}}},"commit":{"parents":[{"commit":"ac1b60432590506c3b3fd113959148278596ef72","subject":"Merge \"CI: run copr periodic job only on master\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac1b60432590506c3b3fd113959148278596ef72"}]}],"author":{"name":"Michael Sherman","email":"shermanm@uchicago.edu","date":"2026-06-11 22:04:47.000000000","tz":-300},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-23 06:17:55.000000000","tz":0},"subject":"Make build context archives reproducible","message":"Make build context archives reproducible\n\nThe plugins, additions and local directory source archives embedded\nfile modification times from the source checkout, so the archive\nbytes changed on every build even when the content did not. This\ninvalidated the container engine\u0027s layer cache for ADD instructions\nconsuming them, including the empty plugins-archive and\nadditions-archive added to every image.\n\nThis change pins the mtime of all archive members to a fixed integer\nvalue, and combines the two existing `reset_userinfo` definitions and\nnew mtime fix into a single `normalize_tarinfo` filter.\n\nThe chosen mtime value is the unix epoch for two reasons: it\u0027s a\nconvention for reproducible builds, and pinning to an integer prevents\npython\u0027s tarfile from emitting extended PAX mtime headers.\n\nThis does not make archives of git-sourced plugins reproducible,\nsince the .git directory content itself differs between clones.\n\nPartial-Bug: #1839319\nChange-Id: I91b42fe530f87af7d78bb1a543b29a4b0d50872e\nSigned-off-by: Michael Sherman \u003cshermanm@uchicago.edu\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ad22005afe3d395bf1e40cde7c3f9f09ec21633c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ad22005afe3d395bf1e40cde7c3f9f09ec21633c"}]},"branch":"refs/heads/master"},"8ee8a0a275e1d70cf86643ec04c0d57f0645bbfc":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2026-07-23 06:19:42.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/88/993188/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/88/993188/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/993188/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/993188/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/993188/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/88/993188/4"}}},"commit":{"parents":[{"commit":"ac1b60432590506c3b3fd113959148278596ef72","subject":"Merge \"CI: run copr periodic job only on master\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac1b60432590506c3b3fd113959148278596ef72"}]}],"author":{"name":"Michael Sherman","email":"shermanm@uchicago.edu","date":"2026-06-11 22:04:47.000000000","tz":-300},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-23 06:19:42.000000000","tz":0},"subject":"Rework tar normalization to a function","message":"Rework tar normalization to a function\n\nChange-Id: I91b42fe530f87af7d78bb1a543b29a4b0d50872e\nSigned-off-by: Michael Sherman \u003cshermanm@uchicago.edu\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8ee8a0a275e1d70cf86643ec04c0d57f0645bbfc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8ee8a0a275e1d70cf86643ec04c0d57f0645bbfc"}]},"branch":"refs/heads/master","description":"Edit commit message"},"289a4e9b37c2df5273e6870066871aca9ed63503":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2026-07-23 06:19:57.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/88/993188/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/88/993188/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/993188/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/993188/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/993188/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/88/993188/5"}}},"commit":{"parents":[{"commit":"ac1b60432590506c3b3fd113959148278596ef72","subject":"Merge \"CI: run copr periodic job only on master\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac1b60432590506c3b3fd113959148278596ef72"}]}],"author":{"name":"Michael Sherman","email":"shermanm@uchicago.edu","date":"2026-06-11 22:04:47.000000000","tz":-300},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-23 06:19:57.000000000","tz":0},"subject":"Rework tar normalization to a shared function","message":"Rework tar normalization to a shared function\n\nChange-Id: I91b42fe530f87af7d78bb1a543b29a4b0d50872e\nSigned-off-by: Michael Sherman \u003cshermanm@uchicago.edu\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/289a4e9b37c2df5273e6870066871aca9ed63503"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/289a4e9b37c2df5273e6870066871aca9ed63503"}]},"branch":"refs/heads/master","description":"Edit commit message"},"43d7e868258eb4597e396441188ccf6375cd219b":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2026-07-23 06:20:29.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/88/993188/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/88/993188/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/993188/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/993188/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/993188/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/88/993188/6"}}},"commit":{"parents":[{"commit":"ac1b60432590506c3b3fd113959148278596ef72","subject":"Merge \"CI: run copr periodic job only on master\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac1b60432590506c3b3fd113959148278596ef72"}]}],"author":{"name":"Michael Sherman","email":"shermanm@uchicago.edu","date":"2026-06-11 22:04:47.000000000","tz":-300},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-23 06:20:29.000000000","tz":0},"subject":"Rework tar normalization to a shared function","message":"Rework tar normalization to a shared function\n\nAdd missing tests\n\nChange-Id: I91b42fe530f87af7d78bb1a543b29a4b0d50872e\nSigned-off-by: Michael Sherman \u003cshermanm@uchicago.edu\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/43d7e868258eb4597e396441188ccf6375cd219b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/43d7e868258eb4597e396441188ccf6375cd219b"}]},"branch":"refs/heads/master","description":"Edit commit message"},"789c02ae79dabcaa5c141baf99532f464472b0c8":{"kind":"REWORK","_number":7,"created":"2026-07-23 09:50:18.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/88/993188/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/88/993188/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/993188/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/993188/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/993188/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/88/993188/7"}}},"commit":{"parents":[{"commit":"ac1b60432590506c3b3fd113959148278596ef72","subject":"Merge \"CI: run copr periodic job only on master\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac1b60432590506c3b3fd113959148278596ef72"}]}],"author":{"name":"Michael Sherman","email":"shermanm@uchicago.edu","date":"2026-06-11 22:04:47.000000000","tz":-300},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-23 09:50:16.000000000","tz":0},"subject":"Rework tar normalization to a shared function","message":"Rework tar normalization to a shared function\n\nAdd missing tests\n\nChange-Id: I91b42fe530f87af7d78bb1a543b29a4b0d50872e\nSigned-off-by: Michael Sherman \u003cshermanm@uchicago.edu\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/789c02ae79dabcaa5c141baf99532f464472b0c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/789c02ae79dabcaa5c141baf99532f464472b0c8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-1"],"failing_atoms":["label:Backport-Candidate\u003d1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
