)]}'
{"id":"openstack%2Fkolla~993712","triplet_id":"openstack%2Fkolla~master~Ie5b9970a4575b6258dc238059d3f33ddfe155789","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-08-05 14:40:56.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-08-05 18:13:12.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2026-08-05 10:39:27.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"Ie5b9970a4575b6258dc238059d3f33ddfe155789","subject":"Add cleanup_repos to revert build_only overrides","status":"MERGED","created":"2026-06-17 05:25:29.000000000","updated":"2026-08-05 18:14:25.000000000","submitted":"2026-08-05 18:13:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":58,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"993712","meta_rev_id":"ccfc7d443ba03c6f47803caa0cb6a3361e45767d","_number":993712,"virtual_id_number":993712,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-05 18:13:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"recommended":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"all":[{"value":1,"date":"2026-08-05 12:41:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},{"value":2,"date":"2026-08-05 14:40:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-08-05 10:39:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},{"value":1,"date":"2026-08-05 14:40:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},{"value":-1,"date":"2026-08-05 08:19:34.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},{"value":1,"date":"2026-08-05 08:19:34.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-17 06:34:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-17 20:27:00.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-06-24 13:41:04.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2026-08-05 12:41:10.000000000","updated_by":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"reviewer":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"state":"REVIEWER"}],"messages":[{"id":"6eff1b394f6da8fc6088f901139eda33ec885f86","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-17 05:25:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a75d629727a85deb24586eb0ee559b05bbc05790","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 06:34:08.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b4cb1881b45544f0b12c6fb80903f3d8\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/8648e31cc2df438c8a643586eaeb3ef5 : SUCCESS in 1h 02m 34s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/ebf2bf49609b451e96abb098f06e6c9f : FAILURE in 32m 19s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/3eb4496055a74bbe87e2ede08f8e3a6d : FAILURE in 17m 44s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/3bac16832ca24fa28726cdc6db738361 : SUCCESS in 54m 22s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/eb00317b751a4678a4c916375074d3a9 : SUCCESS in 37m 42s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8902919e737a42c1adc7ce62d7155262 : FAILURE in 26m 50s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fd23ab1a8d1544d3a60f35e040a5f936 : FAILURE in 14m 26s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/875e3e0f7949475ca0f6cd1188c0faf8 : SUCCESS in 2m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7090a403f67144aaab7353a76ced0a4e : FAILURE in 2m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2e56e929ba504ddeb137bfc2fd4c2afb : SUCCESS in 4m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/76ec0418789b4938bda07f1b4af74972 : SUCCESS in 3m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6e1c96de40064c61bf4bebc11c6e2ec7 : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ab1a8d4444354ddc8faf7c7a220bdcc0 : SUCCESS in 3m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2ccc5f8435a4b48b02a76e90517f98d : FAILURE in 5m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a84372bd5ba4b1ab2ad164c02b6ef43 : SUCCESS in 9m 25s\n\nWarning:\n  Comments left for invalid file kolla/common/config.py","accounts_in_message":[],"_revision_number":1},{"id":"8ad67164adc1a57ad968e4c62e8825b7651fedc4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-17 09:30:59.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"f89064e3187a091d1ba02ebbe7b2c5080e4fcab3","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-17 09:46:32.000000000","message":"Patch Set 2: Backport-Candidate-1 Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"f435be8392fcb74f46d352cbb0a6da940b7cc4a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 10:49:21.000000000","message":"Patch Set 2: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c15810326ddc4167a9537e51b60f98fc\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/f4799276f0da486fab4af16742316396 : SUCCESS in 1h 03m 56s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/d5dc1a64a2b44d61a3db55ca7af92967 : FAILURE in 27m 44s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/0ad576dd5b5f4487aadbb34751d677d7 : FAILURE in 17m 57s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/aaf3862ccb094dcfad690ffd200135c2 : SUCCESS in 1h 16m 20s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/1071bcf7521f4892be15b810cf778113 : SUCCESS in 34m 32s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1c55233752ba483da3c46da0c64fee0c : FAILURE in 36m 36s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fdd65b565d664d748f9da65ef610178f : FAILURE in 19m 40s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/76df0f2a5bda4d7a9d2f0858a550a08e : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5426be595b2e4cb2bd8a08afff0cd282 : FAILURE in 3m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fd4535bbe64a426cbe95cbcad68e1870 : SUCCESS in 3m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c07943761d7448889364d48a77f2ba9e : SUCCESS in 3m 35s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/89a79dc546ec4f97af25f656a7c6b138 : SUCCESS in 7m 56s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3fd1274ca00144a88846d64a8e667c8f : SUCCESS in 3m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23029479ff1a49caa6a14efdb14a7c13 : SUCCESS in 5m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a8cb66e7b7bc4ad996b12384bc181d5f : SUCCESS in 6m 05s","accounts_in_message":[],"_revision_number":2},{"id":"821a7e52c95bcd6c6574a771d6f960ae73eb5ee2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-17 16:16:40.000000000","message":"Uploaded patch set 3.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"e25a3d533fc0064a4adefbdabe3741ec73d0a716","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 18:02:01.000000000","message":"Patch Set 3: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a245d90082404cebb943e1bf23f3f3a9\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/8d3fb1fe58f342649e3e3b0425534bde : SUCCESS in 1h 16m 05s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/581fbf72d75a4c98b0e6654ba76b9ccf : SUCCESS in 1h 36m 06s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/ed94271b127041c398f7faeefa48375c : FAILURE in 24m 42s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/5f025633236d47fdab029d5f3b43fc0a : SUCCESS in 1h 15m 17s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/5c15e98720c748aa8b4ac8db60bd21e8 : SUCCESS in 42m 56s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2faca5e67ff9485791619c00f83a415e : SUCCESS in 1h 10m 03s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f3aca438474d4e8ca064b0a8ffb1c172 : SUCCESS in 32m 17s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/103490e795b44862b3669d2d2523da39 : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f643a508bbe4460a964c32920a519ca7 : FAILURE in 4m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e1cb95e316a948ab9f7a301649236bde : SUCCESS in 2m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fd8c4ea9c421453a8fd25a1f125c5b04 : SUCCESS in 3m 51s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e6db725f57b6482a9b2259e6e2e048ae : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a6f767e008c24caba56ca4a7f1c253ab : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/418ce4c962e0479caea0d3ad285c9b56 : SUCCESS in 8m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cae7160877ce4d6bb76e52910256f9fd : SUCCESS in 11m 45s","accounts_in_message":[],"_revision_number":3},{"id":"6b4f7f88bffa106e3bdb8db5e52e32c6f3328a6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-17 19:23:36.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"909939abbed5b437fc0ea3aadb1774275ebe0b5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 20:29:05.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f5d1e9089c944ca0b0fc4663d01236f6\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/64181f10179b42c4a1a5cecead14561e : SUCCESS in 59m 36s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/abc4e33a775c494f9cc7865994de0697 : SUCCESS in 58m 48s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/38465a3e0aab47ed9f9c3694dd77a7b9 : SUCCESS in 18m 46s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/83afd209fda34d9aa0336a30d5152d3f : SUCCESS in 57m 46s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/bfae7aa041b4448e8dc98798cc32c38c : SUCCESS in 27m 07s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/63d8b3ee59fe46c9bae568e8aa33e1b1 : SUCCESS in 1h 01m 05s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c8da063b20044da7b36434d768028132 : SUCCESS in 33m 31s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/9400ff381168426395699e340b3f3663 : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e2a956f9f204c37a194476264f9d830 : FAILURE in 4m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2b9770124e594d21bdaa4962c235c4bb : SUCCESS in 2m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7d91161d5dfe4a54bd09986f7332359d : SUCCESS in 3m 16s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3f50c9cf63704cd687b7c9b9974a5f78 : SUCCESS in 5m 50s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/de428a3aacd24ae6bb68dac054e28b6a : SUCCESS in 2m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/437c6b2990eb42cfb970e2005c01e3ff : SUCCESS in 6m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/321da4a405a4486bb6ddd4cb88aecadb : SUCCESS in 9m 02s","accounts_in_message":[],"_revision_number":4},{"id":"9993babe8f31a48ca99e4e7f60a6be27ad63b84b","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-18 09:53:32.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":4},{"id":"5da911ac75cb368b1702d4f5b5e209cf4d3de99a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-19 07:42:47.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4376bda4f31343e299835fd4984d3555\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/c23a376c02f145b98b2307bade3c51c5 : TIMED_OUT in 3h 07m 01s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/10247491930a4e0da36ac3e0d690ce99 : SUCCESS in 2h 19m 46s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/55d25a902c9d459c9e9b0316b09b8fe1 : SUCCESS in 1h 58m 23s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/d426370a97664c8d920d6be6b023e744 : TIMED_OUT in 3h 09m 10s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8bd954f548614723aa4daa58075e1398 : SUCCESS in 3m 35s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c30ca907265f4bfab5c65fbd9f8b79ca : SUCCESS in 4m 24s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f35b6e38e079418d9dd3b5bcb99debca : SUCCESS in 6m 35s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c2027322fe3286c207ce7d3c173fe4888f0b7d93","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-22 06:19:13.000000000","message":"Uploaded patch set 5.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"9d8b3de8c1e98e7ddb68eaccaae8c0700ee574e8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-22 06:19:21.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":6},{"id":"b92c0f491a7dd431cf7a8f4c4279c76dff25f13b","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-22 06:19:23.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":6},{"id":"99c7f399aa09023fffa096a3101b4c0a7f1b764f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 06:53:50.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd40085fd6e348c399f11cbe219d89fa\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/963a943cf6f148e080dc3016023ba38e : SUCCESS in 22m 32s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/4a4695a5d4cf4a6c80af83af68315d5d : SUCCESS in 21m 17s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/1e99cc3bf92d4191b4c56ecb8b3e28a6 : SUCCESS in 21m 37s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/f815d7603ac741febfe04cefd7b43ecb : SUCCESS in 14m 25s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/00da212b5d2c4eb3b9da3c443abbf32f : SUCCESS in 25m 14s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7113c739f85b4166892169786ec94e5a : SUCCESS in 12m 42s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4b9ee3da4b164aa197c1fb7265e0e497 : SUCCESS in 28m 37s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/fe8bd40aa0ee479a9c19704d83301be1 : SUCCESS in 1m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8f1b4b14c7b4661a46781db83ad4748 : SUCCESS in 2m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/87bbd2861da24511a75551d02643fb27 : SUCCESS in 3m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/eae55d8c7bf94993b5b196adb00a6484 : SUCCESS in 3m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e6bb955819cb453892af606e00145671 : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0616da52950a4de5ab3685184c15a77b : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b369e839c3264e3d9f844c4a511cbdd3 : SUCCESS in 6m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7bc60f3a3d2844fea3b9fae0f6c5408c : SUCCESS in 9m 24s","accounts_in_message":[],"_revision_number":6},{"id":"80bc4b47a59f201720b9d92a57826b2ea76e4f1d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-22 06:55:44.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nThe following files contain Git conflicts:\n* kolla/template/methods.py\n* kolla/tests/test_methods.py\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"2ed306ef3667f091b409e54e5018a6889b9aa477","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-22 06:59:39.000000000","message":"Uploaded patch set 8.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":8},{"id":"9cb0ca5e482460f5110e1d6d9113f3802b40ab17","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-22 07:00:59.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":8},{"id":"40af08f11144413bc0d3022be3cc73eea4bc381a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 07:49:15.000000000","message":"Patch Set 8: Verified-1\n\n(8 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dc249091f23c43949470a64ad3775377\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/00ec9053315e4a0192b122c8021a090c : SUCCESS in 26m 46s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/510bb8f7f84e48ef80166604343cac60 : SUCCESS in 24m 20s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/b636df37115c46d888936ef85e6c0ed9 : SUCCESS in 24m 08s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/34d3e67abeba45ed8d36e65f9f70408c : SUCCESS in 26m 12s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/8c36f220134c44c5a736463c4a1648ee : SUCCESS in 47m 22s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4e59a1a54324426680acce8fce463788 : SUCCESS in 19m 28s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/81d3c509354a494ba331249999866eeb : SUCCESS in 24m 56s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/dafca864b2a5448791f407bb34a593e9 : SUCCESS in 3m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/944e51198bc0451e99e11852730639a1 : FAILURE in 2m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/732ef4339ab64c90b33a1f64e0740266 : SUCCESS in 3m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4cedcbfdbe784f4baf173af1f4763cbc : SUCCESS in 3m 28s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/57142070f2d14454acdd754f17f97888 : SUCCESS in 7m 25s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/269f2db826424898a1bf78ad78d10056 : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aebe59a3c27c4362b49f8482253cb11d : SUCCESS in 6m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/85c1dff4e6224c29b3bd2124e41de305 : SUCCESS in 7m 14s","accounts_in_message":[],"_revision_number":8},{"id":"972fd4dba44c01b223ffb01efbd54beae8852519","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 08:38:00.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87d8ad0a229c4ff88275e54a9ae3b6a1\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/401ad731f7364c40833e7e1e6df1f957 : SUCCESS in 35m 27s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/15e9c50106d94337bdb4892b576b17ea : SUCCESS in 35m 14s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/900b4dcc3dd2427eac980c84b57214d9 : SUCCESS in 40m 12s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/97cb9b3941d146da853aae3f10515b8d : SUCCESS in 37m 42s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/436219a33d3740f884d3d106a8e2963f : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bb22cf8454644bb0ad7e7be3a838c1b3 : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/37cc48b743ed4adca513e3d842ad4dae : SUCCESS in 8m 30s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"f4ed83675bd24bd4f5ec3a3622872729c5ceae71","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-24 09:28:53.000000000","message":"Patch Set 9: Patch Set 8 was rebased\n\nThe following files contain Git conflicts:\n* kolla/tests/test_methods.py\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"0c737157517e783b5af409bdf082f71e1d5e478d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-24 09:29:27.000000000","message":"Patch Set 10: Published edit on patch set 9.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":10},{"id":"2f0672dc2969d678e2831a583385b3a788fa891d","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-24 09:29:40.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":10},{"id":"a7ef3f72298f664598785eb7a0da8a01ab863abd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 10:26:07.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18cad3e7d3684919ae1c37f3780f5d57\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/930c9f5f939444409389e517497f2437 : SUCCESS in 23m 58s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/36166844709b402ab35af3dd9542549a : SUCCESS in 21m 59s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/f78e63a5408e4f46ac117d8441d78b2e : SUCCESS in 21m 14s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/4ac1250c42cf4bcdba1d7da3030261ab : SUCCESS in 15m 54s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/2c70390ed3774ff482992d87f7838cf7 : SUCCESS in 50m 34s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ab4bd520bca94e258f75eba0450141a2 : SUCCESS in 27m 00s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4614a736d7d045749f1139023c698b69 : SUCCESS in 29m 47s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/9134e24c16614e94ac7fd7fde8556e4b : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de3a7fbfce3043a5a3502f52802fae2a : SUCCESS in 6m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/abd9ff9b9ac549759ea8d431ceefee4c : SUCCESS in 3m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6fe94bba5aea4f1fa670fedee6094374 : SUCCESS in 5m 28s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/56a9b74513f749aabcd7d888a0226cfe : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a1502f507a4b4754894988d8d84306e8 : SUCCESS in 6m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c594026ec9f74e8e886801823ff7433b : SUCCESS in 7m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ea7cc36940cc4de1ab4a3b21abb85863 : SUCCESS in 9m 42s","accounts_in_message":[],"_revision_number":10},{"id":"532323592e56860297b9b3a70ff439b66e466018","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 13:29:11.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dcac68d910624d92a13069c76caaab3c\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/2afcbaa1b6b84c5f977549be26dc396a : SUCCESS in 59m 35s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/0d534653361c42bf899ff57390afe5b0 : SUCCESS in 30m 15s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/19d46fec0dfb44f7b59f8e567e627a82 : SUCCESS in 56m 21s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/d70479b6220c442abe12514898993013 : SUCCESS in 58m 36s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5efd4414f723475081a35cc109aa9ba4 : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6dbbd9fb1940404fba03a85b11bb2ea1 : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/485febb18f1a447897255de216d2d65c : SUCCESS in 8m 12s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"18273196ed1d7f30145e32ab51c668c098f0cce8","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-06-24 13:41:04.000000000","message":"Patch Set 10: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"c1ae79a36bc979da298061fc5272696e948444bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-25 15:15:04.000000000","message":"Patch Set 11: Patch Set 10 was rebased\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":11},{"id":"37eed1048dd50c3ebee4234f0c6dca6f82bb0b99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 16:07:12.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ae2b0adbe8943c1a3df4e2f83d9c2b2\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/9ac743777edb43c1957c849401c1a17a : SUCCESS in 24m 30s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/e74e7c95d00d49979243ee6d4c766606 : SUCCESS in 26m 01s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/55eaedc407ae4569bcbb126bb1c73509 : SUCCESS in 13m 51s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/b5df10d0ebed4e7e98dafb483574c2f0 : SUCCESS in 15m 07s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/a8873edae8824c609ed9a94d4f713dd4 : SUCCESS in 42m 56s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/74516eba9c7d4c9ea9f97b4c2be9879f : SUCCESS in 27m 26s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b5ef1670d02f46479086e4af9bcde929 : SUCCESS in 15m 28s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/e8fd93eefd2b4becbb3215de0225393d : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5eccedf7b7054608b20aff74c9edb263 : SUCCESS in 5m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3422bd55d85949298e9717bab0bf557d : SUCCESS in 5m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dbb2daf174944f54bd50fd6e25082b6a : SUCCESS in 4m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b8e8e1c1ae664dac94c9f3af77556935 : SUCCESS in 8m 14s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c22625a8a6a34ce9910d4c0293f50f3c : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e1457e301234dc4a004598bb0c986ef : SUCCESS in 7m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/26d3ab2405ac49fb86a5ea433a8aede5 : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":11},{"id":"65d165c375cfa9e7a6333c74d9605510f92fdd4e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 19:39:32.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ad2af7d7ec544879179418c0abe7835\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/7cfcf6178e7343f19fae873fedb66e08 : SUCCESS in 43m 38s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/c98f03eb3a4a4f3884e5f0710a187964 : SUCCESS in 40m 14s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/8a3887279c574d019f64c552ca7924a7 : SUCCESS in 42m 35s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/7b8f72f91cbf422eb15add5e4220802b : SUCCESS in 29m 16s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7c4fd3737aed45f9b2897a7a29629ecf : SUCCESS in 5m 12s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/60dc0fa1a1364751a39b79d3981fa6c2 : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/dcd522547339471e89c556c498ca448a : SUCCESS in 8m 43s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"750829075f06bc1eb6270c9c8ca269b61f94852c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-26 06:03:59.000000000","message":"Patch Set 12: Patch Set 11 was rebased\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":12},{"id":"4e61ef020c550b2f543849f654957539f6ad6199","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 06:43:02.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c2a61ecc1e84fb4b80119213312e241\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/40d8d4ac75ed4372bc2350ce304660d3 : SUCCESS in 20m 20s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/08b5c0e993214a0d96571bbffa1362e4 : SUCCESS in 13m 29s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/4ce8aae912cc4135ac1cf30997ea99a8 : SUCCESS in 21m 53s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/b041dfd0bf414c4693ad4609d773bf17 : SUCCESS in 12m 56s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/0b2f06d7a0dc4be79adb72a06f2cc808 : SUCCESS in 25m 17s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e4acd65535734acabafe77a9a087212c : SUCCESS in 27m 45s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b15fe9c4548949d097f0eb8f9392f809 : SUCCESS in 31m 16s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/c468f4255458408abb0f198a54989b13 : SUCCESS in 2m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a46eb20e2bb14c14a6992385a43e597f : SUCCESS in 3m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/44b3d6240ee24fa4b4cf59bf942830d4 : SUCCESS in 3m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bd3e342609c54731b29a8aeef409f0c0 : SUCCESS in 3m 02s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/475c7fa64c0c4985a48298c53440a019 : SUCCESS in 6m 12s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/45fcbb8f384048d88a96fae0a6df23ab : SUCCESS in 3m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60b418cf81934d48a75b97bba84c21df : SUCCESS in 6m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/34f1953aa34e46c49eaeabf8c598f331 : SUCCESS in 10m 03s","accounts_in_message":[],"_revision_number":12},{"id":"cd52929ba5b85735e83b4034d5e5bacdbbb95098","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 08:58:27.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cae86f480ab54e51beab1ae9db504d05\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/3a81c0dbd7154289833242c6545c9a08 : SUCCESS in 51m 32s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/639044aff85a4be7a68ab0c261e1d249 : SUCCESS in 51m 26s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/7db0639f23d640fdb4115a51604ac2e3 : SUCCESS in 37m 02s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/3315cc2f88284558ae2f7360da2fcd8c : SUCCESS in 31m 49s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/173af6da3329401d93b96182d5cda435 : SUCCESS in 6m 02s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/329baf2f14d742f98dd557960f2125fc : SUCCESS in 3m 59s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/4a2c0611dfbf4866be41097b3fa682bc : SUCCESS in 8m 01s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"8f5448f20c0ca70fc10d7e41e8ec2f4c6e71b275","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-07-13 05:41:27.000000000","message":"Patch Set 13: Patch Set 12 was rebased\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":13},{"id":"34f73399b7f6abdddb19e4fef9aadf74752a2292","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-07-13 06:17:12.000000000","message":"Uploaded patch set 14.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":14},{"id":"dd9ef070856e6be9c1104b8b1cb80ec8ae307437","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-07-13 06:17:17.000000000","message":"Patch Set 13: Backport-Candidate-1 Review-Priority+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"34d04a98cda422887ba487813fdc66643fa7de00","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-07-13 07:12:13.000000000","message":"Uploaded patch set 15.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":15},{"id":"39d2c3a7a155051215db53e76facc18d1e39e1bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-13 08:22:05.000000000","message":"Patch Set 15: Verified-1\n\n(6 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a3d21caa7fde44b19def052b566dd133\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/6f270f8d621e42e4a6c06b6221ba0e4c : SUCCESS in 1h 06m 47s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/285feb1f7f654ae9a96bb935f316ebb5 : SUCCESS in 22m 53s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/51f4040d063d492c96b48001ba3b8bcc : SUCCESS in 16m 54s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/d8f4ae4fc6534548a5e9d2693f9628a4 : SUCCESS in 15m 05s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/7008777d22394d8eaf32cbc05ba48709 : SUCCESS in 42m 43s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/389d49bf55174865b87f476a85649b2f : SUCCESS in 21m 09s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7a675f63db2243b7a51ab0fe336b2422 : SUCCESS in 24m 29s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/4adbe22bfe0a4478898c710240067867 : FAILURE in 5m 07s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/94ecf9ff678b458e9161cfbdeccee133 : FAILURE in 3m 55s\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/21916f79c64e436c882e7b93bbca14be : FAILURE in 3m 44s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/e3b6a4424ba04c7eae13d1b40f689231 : FAILURE in 3m 53s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/7324f251ced84d4c9c951cc1d3930bb0 : FAILURE in 3m 50s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/28a67f8d84e143d9a0e1aebabd2790b8 : FAILURE in 3m 49s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/205caa7445684d7f880e0c0aafa53709 : FAILURE in 4m 13s\n- kolla-ansible-debian-trixie-bifrost https://zuul.opendev.org/t/openstack/build/707d863e279c4eada1f44ca21f656b3b : FAILURE in 3m 10s (non-voting)\n- kolla-ansible-rocky-10-bifrost https://zuul.opendev.org/t/openstack/build/d8f87b092dc74467b6d4d3ce6b565392 : FAILURE in 4m 09s (non-voting)\n- kolla-ansible-ubuntu-noble-bifrost https://zuul.opendev.org/t/openstack/build/79a4a77d832b4789941efdae21adb56f : FAILURE in 4m 12s (non-voting)\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/b40a4fa5b585440f9c54bac9d6153aa8 : FAILURE in 5m 13s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/f9cd3f99fd9040c6ad27bd6619ed1b00 : FAILURE in 5m 02s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/59516325f5f7496f884a23ee8e7d3d81 : FAILURE in 5m 21s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/f70088efc934457096a859c03400f463 : FAILURE in 5m 21s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/a9c0e09edea64236b30526deb33b1c62 : FAILURE in 5m 10s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/a79462aedcb042a6a92e9266d1f12188 : FAILURE in 3m 35s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/391860e4bf8043dfb10d835553ca648a : FAILURE in 3m 49s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/b631c1a96e9d4ef29c5af17898b84393 : FAILURE in 3m 25s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/a2e7588f5afd45a1ad900a737cd5701f : FAILURE in 4m 22s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/815f9dc8405e466fa8b580ad245ad624 : FAILURE in 4m 51s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/603c1dd474de4be9b877fd4ee8e2790c : FAILURE in 5m 04s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/c2b8fba1bd4149f08c3e02c135e15d62 : FAILURE in 4m 49s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/322e265d59d049f8b58a5a1d355e5713 : FAILURE in 4m 50s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/50b4256b65fd4e5cac19a88a0ca8d145 : FAILURE in 4m 35s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/467515cc3f4b4ac4bec5ba24795a26e8 : FAILURE in 3m 44s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/19e12204b45d4092bfca2e106d809fc1 : FAILURE in 4m 06s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/781bb4a5116f430cb0f29fb900bba12a : FAILURE in 4m 25s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/258e58c08156407db02d0aa2e1907569 : FAILURE in 4m 17s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/830e85a00e7e49689e09c2af56acb2aa : FAILURE in 4m 18s (non-voting)\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/3ac1a6cffd0c4927b3ecb7e030bd3bea : FAILURE in 4m 30s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/79a4d73ab19249889dc92a43b6ae43c2 : FAILURE in 2m 46s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/7aa4df5b7b914e149b65436d7576a137 : FAILURE in 4m 06s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/ae050f6983f543dea2d38dc9f69f88b2 : SUCCESS in 3m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/857253924b524398913d4f7a2d19ff23 : FAILURE in 2m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c7f04985826f4eca8715043ff359e8cb : FAILURE in 2m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9414a975ecdc4e1bad98138cde173a2b : FAILURE in 3m 55s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ac9320a5703a4b66b330a0d5e5b37dff : FAILURE in 6m 01s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a4410fe986a4466a9bc35a8f4d8df20e : FAILURE in 3m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9dc1688c3e2436c9f7b314fe38e8ffa : SUCCESS in 6m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f30a4397fac49f5aaebddda47c39777 : FAILURE in 2m 06s","accounts_in_message":[],"_revision_number":15},{"id":"51f225954993cd6a6475b700d8362f9d311d4341","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-13 08:25:34.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95a0c73af4194c1694206b132cc3eca4\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/79c4405b135e40dbaccecd3ea1c6d93c : SUCCESS in 51m 05s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/30dbf4fdfb034e57a79d806fdf98f36c : SUCCESS in 1h 08m 00s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/89c0eba66f6040e4b4984ef832726f3c : SUCCESS in 1h 10m 07s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/00835e3f97b549778b932bff249753a4 : SUCCESS in 1h 09m 34s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/72e1ae1ca9464aab8cfae3fc4ec56257 : FAILURE in 5m 57s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/a9a3c4b411e940ce9452123e29af25ae : FAILURE in 5m 15s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/18ac62bdea1a48afad51dc0f83abf860 : FAILURE in 5m 24s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/39cf100e0edc45a891a4cc469bfe4b01 : FAILURE in 6m 33s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/16b9548fbe154d46b81017f22a775c1a : FAILURE in 10m 01s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"853efc3d7a0da4a1c373fc6cc42663366fe1374a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-07-14 05:21:17.000000000","message":"Patch Set 16: Patch Set 15 was rebased\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":16},{"id":"de3de7bed6b72932a0f1eece988e509720c09dfd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-14 08:11:54.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71602426e95e448b8f0cd9b8829dbd3d\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/eb23a34d36b1438488c2418aacae6625 : SUCCESS in 46m 19s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/a6f6a6bf41e942b58da57584ee7aaf0f : SUCCESS in 46m 03s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/e71f214e81d34179ad9ab3927ad27f2f : SUCCESS in 45m 57s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/854a3a54bdbc40868046e2cc39be2a4a : SUCCESS in 41m 29s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/b232178fb764478fab7aeed9a04f5dd0 : FAILURE in 1h 30m 59s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/f60516f894754169a1e913583fecce12 : FAILURE in 1h 19m 25s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1cbb5d2073974d3c9a61129c6f217867 : FAILURE in 4m 57s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0125fce5b86b487ebb81827a5f98736f : FAILURE in 5m 16s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/e66bc215376344c1b36fa9beddda38b1 : FAILURE in 9m 02s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"1d959e512e856db370fe18eee431de7f80d54244","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-14 08:26:34.000000000","message":"Patch Set 16: Verified-1\n\n(6 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b5e31d2eed8a4d7f8b0467b62d0f705c\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/97583d7691884647a5573e48e58d7f71 : SUCCESS in 58m 14s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/17375aabf95846f28d3d1515675b333f : SUCCESS in 23m 27s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/bd5e95c096aa400aa4549b59eee0ead1 : SUCCESS in 14m 02s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/d6b6b5d7122e42f19cff58129b219864 : SUCCESS in 25m 02s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/6f6d176739e841c58963fad4038dc322 : SUCCESS in 43m 41s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bd34c1d16ec6474da90afee7b4080584 : SUCCESS in 25m 52s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a11b21e4ae1c4872bb5b7df0c16a22d5 : SUCCESS in 28m 32s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/c08b43b519734159bc96f04c9e9d2e3b : TIMED_OUT in 2h 33m 46s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/963d5d4ae0f44074beed52305430ddee : SUCCESS in 58m 48s\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/b143c3c81c8b497fb81e67571a529f26 : FAILURE in 1h 30m 52s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/a75f16e3fd8f4d6ea49b627c9ef44e76 : SUCCESS in 1h 01m 54s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/d46240ca058c4dfdb2fc2b44f1c6fde9 : SUCCESS in 1h 28m 04s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9687df053c5f49cfbdb3c9d4ea764219 : SUCCESS in 1h 01m 51s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/c95f2617e8154d11a95dc3795ebae4de : SUCCESS in 1h 33m 27s\n- kolla-ansible-debian-trixie-bifrost https://zuul.opendev.org/t/openstack/build/54fb355863cc4f4bbfe757e1073e548f : SUCCESS in 24m 06s (non-voting)\n- kolla-ansible-rocky-10-bifrost https://zuul.opendev.org/t/openstack/build/79939c38630a483ab56057bb80fb45c9 : SUCCESS in 20m 13s (non-voting)\n- kolla-ansible-ubuntu-noble-bifrost https://zuul.opendev.org/t/openstack/build/f13845ca466b40d1a095557c8a2476c8 : SUCCESS in 15m 01s (non-voting)\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/69dd7333a48643b3b5cb4b11860b983e : FAILURE in 1h 06m 43s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/44087a1b9a0b411db7aef35d8a9adcbf : SUCCESS in 1h 33m 06s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/1ea49d9caae84b16ae33e56198cbe1f1 : SUCCESS in 2h 11m 13s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/ee56dbe594dc4b8589b6331fcd63a3ae : TIMED_OUT in 3h 02m 32s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/d1b9619dbc6e4c3ebe44bbd45ca08f12 : TIMED_OUT in 3h 02m 34s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/9b51bf2f31844525b5e7939d3918d387 : FAILURE in 51m 16s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/9aaaf7fb23464684aecbc85a7c4828e6 : FAILURE in 54m 12s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/831c8d433e8445c7a040a20427f445a1 : FAILURE in 56m 36s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/5cafe54ac2de47fc819dfeff5f79ebe3 : SUCCESS in 22m 32s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/cdcece7b386642978937b90c3080c988 : SUCCESS in 22m 58s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/8f688bc5ddd4459fbd6904365ce27d2d : SUCCESS in 29m 25s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/27e641a67fff4f3cbcec1ce05108169f : FAILURE in 44m 36s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/1684796c7c3140c4a6e71aff00fa9a1a : SUCCESS in 43m 26s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/b2a83de2904a43b6bfb773dae05a34a0 : SUCCESS in 37m 06s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/dbdb53c40d404ddb890ec58ff255f87f : FAILURE in 42m 45s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/9bce17df89f44cc3b651086f174d2fe4 : FAILURE in 46m 26s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/749ccb8bb0ec4c86a70eb8cca75f3d88 : FAILURE in 1h 00m 45s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/74ce8a5b98ae4920a02444587853578f : FAILURE in 1h 03m 51s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/1020b1e4494f48ff8ab112b8591f7fa5 : FAILURE in 1h 02m 42s (non-voting)\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/d1ba1b1d19d44a918f2c5966af9257cc : FAILURE in 1h 19m 41s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/0a7ad3d061664127809f51338887be0d : SUCCESS in 48m 49s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/d254fef3be7c4c6eb9730da95c0fa50e : SUCCESS in 1h 27m 41s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/f67de97b21734540ba04244f66f50952 : SUCCESS in 2m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60a6cc125f4246238f771b171951cee3 : FAILURE in 2m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fd5a1a6570a84a488b6c484fda82e725 : FAILURE in 2m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6173e6989a2242708bab4fe4dd5a6139 : FAILURE in 2m 47s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9d64208a12fa4625adb50cf82e90d4f2 : FAILURE in 4m 14s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb7dcf60ccb241d591908fdc3a775e77 : FAILURE in 2m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20a26b603a304c73bada66d3c632955c : SUCCESS in 6m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7cac0a3036794c1ca7c75413d5cb7705 : FAILURE in 2m 46s","accounts_in_message":[],"_revision_number":16},{"id":"4d30cfed1b8a66a0d7f2da6bd21e5799e8fea79b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-07-14 08:55:22.000000000","message":"Patch Set 16: Backport-Candidate-1 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"90dc4f226c87d83ff0cb89c01795a984c99b915a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-07-14 08:56:04.000000000","message":"Patch Set 17: Published edit on patch set 16.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":17},{"id":"168aec3b20c30d4af1f5acfd70c332076fe5d3ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-14 11:31:37.000000000","message":"Patch Set 17: Verified-1\n\n(6 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/aa250b3090da408aae9efb70dbb7aa3d\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/46eea4750b1449d0a5afd535bc2d10f7 : SUCCESS in 59m 32s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/4b0952276f544c7681ceb573391ed649 : SUCCESS in 28m 39s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/542966abe4b84720949bfbff4d315a5c : SUCCESS in 26m 24s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/57f8c858f1fb49f8912b2473dab16ee0 : FAILURE in 29m 14s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/a805a8c3e4d7440fa4e92f77ba3db3de : FAILURE in 55m 22s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f7d38a2e79d9490ea16a5bb877e9718e : SUCCESS in 27m 20s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e9f45741e8ab47d5b9b8c6c4829b368f : SUCCESS in 27m 24s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/39f5a331b5f343f59fb91a3a666b7215 : TIMED_OUT in 2h 33m 49s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/d058300e54844ef6826e4499c149f737 : FAILURE in 44m 55s\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/808c712e019640b4a844d3b355173c8d : FAILURE in 1h 31m 54s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/eba9c8b4b31041c99c523ddda9b74ef2 : FAILURE in 51m 09s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/6fd2c5b68d4e4ae2b78deaca5185ac6c : SUCCESS in 1h 30m 14s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cce7df8c8bce464abdebe761c947de02 : SUCCESS in 1h 03m 59s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/f4ef4f4739f94a468c4e12597e161e0d : SUCCESS in 1h 42m 43s\n- kolla-ansible-debian-trixie-bifrost https://zuul.opendev.org/t/openstack/build/5c060b4f514a4c35ab5e780483a27a7e : SUCCESS in 30m 15s (non-voting)\n- kolla-ansible-rocky-10-bifrost https://zuul.opendev.org/t/openstack/build/d44f8e4141174299acaf60b6baebf813 : SUCCESS in 27m 56s (non-voting)\n- kolla-ansible-ubuntu-noble-bifrost https://zuul.opendev.org/t/openstack/build/40039142d1274529971aa094b2450667 : SUCCESS in 32m 04s (non-voting)\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/2b1e940d95e241ca9987a0664f896571 : FAILURE in 1h 06m 54s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/d5c07eec5086473b906a9f06596d1df4 : SUCCESS in 1h 28m 30s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/88f7bef485c7441d90dda2b32ae5d9ff : SUCCESS in 2h 14m 40s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/a401dee8d2464c5292258c062e4c8470 : SUCCESS in 1h 33m 44s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/f43098ea8738494c96ffbb406d762edc : FAILURE in 55m 14s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/38c3c119d29842119343bad10902231b : FAILURE in 55m 28s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/b9dcab525b2f47ec8574eb28796d1f71 : FAILURE in 56m 17s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/51a385f5a3c24da188491f8d9688b480 : FAILURE in 56m 41s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/5a3a627ef3634c40a173ab2b87b0b904 : SUCCESS in 27m 23s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/cba14ca6cfb74131b5ac5c83f0ed3648 : SUCCESS in 27m 10s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/d54c51c24c414c47a3336ae886747e68 : SUCCESS in 30m 02s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/19915021f2924845aa1945ab0f4bc149 : FAILURE in 54m 08s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/dd2171a5fd6f46ea81ae127592a7cb45 : SUCCESS in 1h 22m 27s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/3085f3cff2104d789e06922fd4f0de22 : SUCCESS in 58m 09s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/1dfcb6f09e1f410aa22d9eac49c11989 : FAILURE in 43m 26s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/a2960ae66eee4e628a66f998f22280d8 : FAILURE in 47m 43s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/7544325743a24ed6a111366023e1704f : FAILURE in 1h 05m 35s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/3bf65bbdd2044201998cffe02b738496 : FAILURE in 1h 07m 38s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/32a4b5ce523f4e789cdb83fa48eacb1a : FAILURE in 1h 08m 48s (non-voting)\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/c120d09e3be244eba3d2cf8e4b236bf2 : FAILURE in 1h 10m 37s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/07c01c2a10cf47ce84e5b7c7513879bd : SUCCESS in 46m 43s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/2080eb849d454d859d399123a33687dd : SUCCESS in 1h 21m 52s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/49b69d6e002c4234abd335d42f55a644 : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5453f3b0a7364512857f45b3759fdcbb : FAILURE in 2m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/44163c45d0244f2f94ba0a5c9a3821a7 : FAILURE in 3m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b90452c651a7477497e0241e198c6fe6 : FAILURE in 2m 30s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/49ddc6e6d6e44a779bc68eba08f4c902 : FAILURE in 4m 03s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b7e86f62b7f04122ba1cfc91a076f3c4 : FAILURE in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfc08fe956d84845ab82c0a02592768c : SUCCESS in 8m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2da3b520d39f40ca96818e9e2d39edc1 : FAILURE in 2m 38s","accounts_in_message":[],"_revision_number":17},{"id":"ab760ecc3dea0a63d525114bbcf6421732313c3f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-14 11:35:36.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c19e39edb324b858d1cb677c9cd3586\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/a1440095127b463580b98c70a2656bee : SUCCESS in 44m 58s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/3822bb33f9d944d78c4720e7098687e2 : SUCCESS in 44m 17s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/dbcff8ae9d6a4552a9a4af99e2c3572e : FAILURE in 42m 34s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/0a60ef30423b42efb7a07ca75f847d58 : SUCCESS in 44m 13s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/015874ffd350443da08e0a4a44034bc2 : FAILURE in 1h 42m 11s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/ae758cb8aa1041c8a226d3af0727ce02 : FAILURE in 1h 33m 41s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c16bcaddaa8e4b40bcaec199a16bdec2 : FAILURE in 4m 41s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/139e56dbf74a4349bd41805923271ab8 : FAILURE in 4m 55s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/87a1e0f1721d42c293c174b2e5e24392 : FAILURE in 9m 36s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"950df703b1d7e0f4801ac662a2a063a8a249af38","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-07-15 13:35:17.000000000","message":"Patch Set 17: Backport-Candidate-1 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"950d6577b63bcdf7cce331ee3470acb84940f91c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-15 16:12:23.000000000","message":"Patch Set 17:\n\n(6 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9fbb661108e24ffe8db3e028290f02ac\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/54e3f43bb79749288eae0fbd19043a3f : SUCCESS in 53m 39s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/acb8e46b1e084e879236ae6ddc09452e : SUCCESS in 14m 23s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/5aca7352d2664f6e8534bcde0382f2bd : SUCCESS in 29m 45s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/57a61fc2f3ea4bc9bb67fbcc9a2bb211 : FAILURE in 32m 39s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/9b70c8590a8148f8b0c3704903b187d3 : FAILURE in 36m 43s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a56ac64b54814bc095747ee882dea766 : SUCCESS in 32m 46s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/089693f392394af38f94e06625880977 : SUCCESS in 40m 13s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/62d5a2ae67de430a9e33a1e9cc6edfa2 : TIMED_OUT in 2h 33m 43s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/0a091e7584034679963c981a45f1b000 : FAILURE in 51m 27s\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/368c4f09ae7e4a00a4f41813153e0582 : SUCCESS in 1h 37m 32s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/4e8e004d6ea8499bbd6828ecef1fc94b : SUCCESS in 1h 19m 14s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/67c6e85708e14e4ab89248f2ff2a3c57 : FAILURE in 1h 37m 46s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e65edc33b2254058bb1315f3ce9115ea : SUCCESS in 1h 22m 05s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/0d2a4043c7214878a322b228f953f7b7 : SUCCESS in 1h 37m 29s\n- kolla-ansible-debian-trixie-bifrost https://zuul.opendev.org/t/openstack/build/7969e08549034f36a8b856198d883bfb : SUCCESS in 27m 37s (non-voting)\n- kolla-ansible-rocky-10-bifrost https://zuul.opendev.org/t/openstack/build/2943a48f3d4d48ad946d7512a32439dd : SUCCESS in 15m 39s (non-voting)\n- kolla-ansible-ubuntu-noble-bifrost https://zuul.opendev.org/t/openstack/build/74e4fb2958304b04871e1e89b8aa08d3 : SUCCESS in 35m 34s (non-voting)\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/d43fd72df41e424dbfa3bf78c8e3baeb : FAILURE in 1h 16m 00s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/72c3ed1849224cd7920fa432b0c13247 : SUCCESS in 1h 38m 06s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/c4345b8cca894b4d95f309098cf252c6 : FAILURE in 1h 15m 46s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/775d4932b6b441be93c9a704e5b37a36 : SUCCESS in 1h 35m 36s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/c26780ab5f2847c1aafdda3a2520987c : SUCCESS in 2h 21m 12s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/1eb788bfb638445598290dc89e1c6264 : SUCCESS in 1h 21m 24s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/d711dcedc25d4398afce6b3a43f0a1d9 : SUCCESS in 54m 38s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/53f12278de0c49a5889fdb6933cf224f : SUCCESS in 1h 27m 45s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/38aeac921f644396b1aa7a1d0812562e : SUCCESS in 43m 16s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/02438322d76a4701a06520ddd343ad20 : SUCCESS in 15m 56s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/2a843d1a08b94426b2a58d536d168ba8 : SUCCESS in 30m 23s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/0b49ccafee794baabfce0e64ccb27613 : FAILURE in 54m 32s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/458be367e7f8400793f60cbd94275001 : SUCCESS in 1h 24m 58s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/9c1d9cdfdb2c4876b7141157022aea14 : SUCCESS in 1h 22m 49s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/1408bbfc159a44cdbb54b48cd5084f78 : SUCCESS in 45m 22s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/76390476deec4545adff6cb4d453486c : SUCCESS in 46m 06s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/8b4b5d1eb4a84dc6a4354a010439d92c : SUCCESS in 1h 15m 03s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/42f09e99bd9647f7b975940bfb2f5aa5 : SUCCESS in 1h 38m 23s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/29fb7830f6e24aa7b77a6d1ed49b7ded : SUCCESS in 1h 17m 29s (non-voting)\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/19a0bf997b3f413ca82da17b7d864272 : FAILURE in 1h 09m 45s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/1ef7113a60d94a338aef6ae85f6084ab : FAILURE in 1h 05m 10s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/48a4518e44f145e385a2d33a3228d2de : SUCCESS in 1h 48m 38s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/78a5ad4f76b64f25b93b61e1c383b56c : SUCCESS in 2m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72f91dcdb2a147d89bfd168f5fa950ea : FAILURE in 3m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/53af851abb914c60b24a180e26ca255d : FAILURE in 3m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/784d350efb694f5a9def69d3ba1ff3c1 : FAILURE in 2m 23s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6afa62ee16a149158ce4ab683c2d4960 : FAILURE in 7m 22s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/de26ddc484334ce79054aaeb77dc0c5c : FAILURE in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e63e4668c904957bbc8863e16074688 : SUCCESS in 7m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a8343773ccf34612beed0b6be11f7698 : FAILURE in 1m 59s","accounts_in_message":[],"_revision_number":17},{"id":"86688ba79e340bc966d686702c8fe228fb6331fc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-15 19:36:22.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b56a6ed95bf54848993f6bcdacded748\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/8f87f573b6d04eb0928cd081034a9276 : SUCCESS in 44m 22s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/206d173f520b464ea8cbf19a046f81f9 : SUCCESS in 44m 51s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/99829121f82f4f379cadf9b9adcafec6 : FAILURE in 30m 30s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/7107bc6ff36449dca9ca4a500f4d3cdd : SUCCESS in 1h 01m 58s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/ead26a1adc894cb5b92e202f6b9f6567 : FAILURE in 1h 14m 14s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/90cbf5183db741688c9df93e0d120a3d : FAILURE in 59m 43s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e9149b51b0ff40d286d616cc062a00ac : FAILURE in 3m 53s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/393a7e7632cc45249688779094b66e18 : FAILURE in 4m 45s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/886d0c9d3c9447c8a2577b637d876e9c : FAILURE in 9m 10s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"9cb71540af7564d20a39f8a5cf61f1b0d97140cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-07-17 08:23:00.000000000","message":"Patch Set 18: Patch Set 17 was rebased\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":18},{"id":"e6805d24d45dfbfd82349d89b81838438578d149","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-17 11:01:15.000000000","message":"Patch Set 18: Verified-1\n\n(6 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/64cc0bc054d44ca8a3bd62390f3e5044\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/834b9975181344abb38fe279d0fadba8 : FAILURE in 55m 44s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/e5285507fa3248a281657cb0e5313e67 : FAILURE in 13m 44s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/4cc1da63a2474b5caf4badfaeac64dc9 : FAILURE in 34m 48s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/fc8c373712eb4751930431670f490ff0 : FAILURE in 17m 50s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/def22a6809dd4befba22a0b43c749b4c : FAILURE in 40m 04s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c421405e6ec04d7394effdc38741a63c : FAILURE in 23m 57s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d8e734f555e04746bd4338534b14d59a : FAILURE in 21m 05s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/1407da2b11b54b6787e018cd4ed1154f : TIMED_OUT in 2h 34m 35s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/f76ee182aa8646b780fe5bc3958f4520 : SUCCESS in 1h 01m 24s\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/28f6ad0170d948629e56f865d2d734f6 : FAILURE in 1h 45m 20s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/63ae45116faf48a8ab20491d71d5439b : FAILURE in 49m 59s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/d9b447833dd140369a4bb3e51f40b780 : SUCCESS in 1h 33m 08s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d57db48c06d94030a7079f9b15713690 : SUCCESS in 1h 06m 26s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/2f9a1a158f65458795b19270ae75f3a6 : SUCCESS in 1h 38m 21s\n- kolla-ansible-debian-trixie-bifrost https://zuul.opendev.org/t/openstack/build/78045aa6e2a34228960ae958b8f04459 : SUCCESS in 13m 46s (non-voting)\n- kolla-ansible-rocky-10-bifrost https://zuul.opendev.org/t/openstack/build/c1d8d93201d64f98af822f7054788f06 : SUCCESS in 15m 04s (non-voting)\n- kolla-ansible-ubuntu-noble-bifrost https://zuul.opendev.org/t/openstack/build/b373b9fd6cb44c40953ce894fed4264b : SUCCESS in 36m 48s (non-voting)\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/67440fbabddc42ca9bb80627f36ef0ee : FAILURE in 39m 48s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/a9675ecc6df24df388ac64d31ed43f4e : SUCCESS in 1h 34m 54s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/69a9b91427944d679680dfcbe6a899c9 : FAILURE in 2h 03m 23s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/6aa090514cb44fe880a25ade5d740104 : FAILURE in 1h 34m 50s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/2b18828ae9df4edbabb0b087e1102496 : FAILURE in 1h 46m 49s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/a3c46db757fe4aa8be6d7d1d00d762e8 : FAILURE in 20m 40s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/6054b9f6d3d74a90a714c8e24930a355 : FAILURE in 16m 19s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/93975605493248ec994047043bbc10b5 : FAILURE in 21m 22s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/4a9e503a15d84fd4aa2c0ac960f6e91b : SUCCESS in 24m 41s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/2f26f6c30d964e6f92d738a795a7afde : SUCCESS in 24m 52s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/6f3b7cd61d8f49b2b7b6ccda72560c05 : SUCCESS in 20m 23s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/6eead3e5dfde4ca9a795cd1383987bc5 : FAILURE in 55m 29s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/d130b5dd5ff84b2f80799d08dd6e5c2f : SUCCESS in 1h 31m 35s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/4b0504b23c1b46f795c1b7f20b38315d : SUCCESS in 1h 24m 08s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/026b678cd36c4cebb7c1479284ca31b9 : SUCCESS in 1h 03m 40s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/744ca7a2ce1c46f492ce2a67a7c6fc1b : SUCCESS in 1h 11m 00s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/15bb9aebd52042dfafc521306ec67c30 : FAILURE in 1h 11m 51s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/9ab33a0b111f4950ac58e1b255c5bba6 : FAILURE in 1h 08m 05s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/b50d8e7712c64f0bbbb66b4ca5bc697a : SUCCESS in 2h 18m 06s (non-voting)\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/474a9c10f5604136920b9499dc1c088c : FAILURE in 1h 15m 46s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/c7fed207f5d244959cc2f3cf3ca87e34 : FAILURE in 1h 08m 27s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/1f540993892c40b1ab54e9090d4f3ce4 : SUCCESS in 1h 36m 50s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/6cf9c273989b4ea9a8df33820424e812 : SUCCESS in 2m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7055212e569149f79227f4571c22eecb : FAILURE in 3m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dcbb4b6f8b574c58ab2d1df64d055ebc : FAILURE in 3m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c9867f6b536a461bacf833e26eddf259 : FAILURE in 4m 58s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1211d7fb718b4cbe8435401944a33ecf : FAILURE in 5m 44s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/24382ce1b4fb4ad0a4eac6f5592fbb69 : FAILURE in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59ce878e315642639324591e6e8a228c : SUCCESS in 5m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/84c4277776f44437b9bc8c9b557767fb : FAILURE in 3m 04s","accounts_in_message":[],"_revision_number":18},{"id":"abdfe54ee4ad800fbb653139946f71bbf1cf6d2d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-07-17 11:03:35.000000000","message":"Patch Set 19: Published edit on patch set 18.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":19},{"id":"22aee0699a58482a92350a7e788648b942d048a3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-17 13:18:08.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0da796ed41e54451bcc7a538b0d2dca1\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/5e70868051894ee08efff413222913f9 : FAILURE in 31m 22s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/cb74a51c5fcf48fc9101429a09fd7963 : FAILURE in 41m 09s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/7f79e8a19ded4a1ba8ecc2d68abfc72a : FAILURE in 45m 48s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/c459e12724da4f3ab7a898e735eddd27 : FAILURE in 42m 55s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/156e7743d05c43a38eef543dcf963683 : FAILURE in 56m 29s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/364cdd69678c469e8ff1eb79489dd434 : FAILURE in 55m 11s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/42b26c7c7d4d4fd699217e65615877d6 : SUCCESS in 4m 18s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/58a2bf8a4fff494a8c8e65b99464b41f : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f0b5d0dbe66d48daadd22be01d82b09c : SUCCESS in 6m 18s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"2acef928ab6844c30710463c8d1a6953ae2c1e4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-17 18:07:25.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c6000373f0b045f5b702bbedd6b53e74\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/a2579ef99904463bb4b54803450e0b9f : FAILURE in 53m 56s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/49219bf28b4944029b4fee026840517f : FAILURE in 23m 24s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/64fe0932d8cf44e69e5dcd4fa68da8ce : FAILURE in 32m 35s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/7d93476b3d0b4164aac492efe1d5147d : FAILURE in 27m 54s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/8e858d0209584b7cb5edfe211cbde2d3 : FAILURE in 55m 07s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d5ee05af7fce413e8a6bc188afe05165 : FAILURE in 28m 26s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4cf7504ec7aa4da5a3e0c2d9a3dfbcef : FAILURE in 27m 54s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/ccd6d37f9bd9489a9837ab05659cdf2c : TIMED_OUT in 2h 33m 45s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/b4ef39c1e8d344459ece04b87fc8c25f : FAILURE in 34m 29s\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/d35771ce596a417eb55016c78d293e29 : SUCCESS in 1h 04m 58s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/f7a57a50c78e460299f3bfc2ba07b8f4 : SUCCESS in 1h 07m 32s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/da28b926766949b3a0a9463069bc74cc : RETRY_LIMIT in 3m 36s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/22e60aa297054705bc4bfd94a256a544 : SUCCESS in 45m 05s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/b7583093717b4de5b7307575f0b5065c : SUCCESS in 1h 08m 16s\n- kolla-ansible-debian-trixie-bifrost https://zuul.opendev.org/t/openstack/build/7464fe6f522b4ab7b6b6f8e301916fd8 : SUCCESS in 27m 02s (non-voting)\n- kolla-ansible-rocky-10-bifrost https://zuul.opendev.org/t/openstack/build/0a16c9456f70481fb2861b1d200d7ccb : SUCCESS in 20m 28s (non-voting)\n- kolla-ansible-ubuntu-noble-bifrost https://zuul.opendev.org/t/openstack/build/976c6065c34445389042208263937419 : SUCCESS in 30m 17s (non-voting)\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/3392eb1a38d246aabfa8a92b0f32cf38 : FAILURE in 55m 49s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/81a4f4fca1e7462b8a5ac09ccd97a84b : SUCCESS in 1h 16m 22s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/ec38378d6fd442ed9ca4f8db3f971c61 : SUCCESS in 1h 38m 12s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/593234c799724921b29bbb6d65387e7f : SUCCESS in 58m 50s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/d88ff9eb39264d1d93097bd57c550397 : SUCCESS in 1h 50m 13s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/854ca66f23514510b637d9a6c48a7f21 : FAILURE in 16m 43s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/9f328259da5b4bb99b1d3a6b63f4abaf : RETRY_LIMIT in 3m 46s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/fe270debd2f14f1a8e6a5b5cdb92175c : FAILURE in 12m 12s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/0cc77d56d9354bacb92c4b3e421388a6 : SUCCESS in 16m 44s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/8604b5dbedf048c2b7a73f1cd8aa32d7 : RETRY_LIMIT in 4m 21s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/b4dd65b6813f4264b0d3e36eb9563b68 : SUCCESS in 29m 00s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/c55d7a0d2d8246f78e672416eafe9834 : FAILURE in 1h 00m 18s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/0c81285ab96b413e88583a73b899d27c : SUCCESS in 1h 10m 21s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/8c44f2c00556471b89d25b4614435d12 : SUCCESS in 1h 31m 33s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/a234aaf4c73347cab518e35973cc6a9f : SUCCESS in 46m 14s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/c2ac0c0689ac44d384d96b739065d1bc : SUCCESS in 42m 52s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/66bb4b1a2ac046378db122fe4c4a8c5b : FAILURE in 54m 32s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/58cd5951b1db486884004c717f05c9e0 : RETRY_LIMIT in 6m 29s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/2ec569479b49424ba5472f5ae8b7dfec : SUCCESS in 1h 26m 09s (non-voting)\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/59578b775c1b4ed5aac20b7edc515d2f : FAILURE in 1h 16m 37s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/9213cba02d61400b93bc367f4906a3a5 : FAILURE in 59m 22s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/cf57c6d702964abab48cc962452d4e4d : SUCCESS in 1h 28m 21s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/b24aed093c3f4a23941609adb8cb4b54 : SUCCESS in 2m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8426e6425f6e4ebb89ff74e8316ad19d : FAILURE in 4m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/88a35f5c0b704edb9b1b622bb0476d9d : SUCCESS in 3m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/43374414ae9d4f9fa3f321cbfb47741a : SUCCESS in 3m 04s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/48042331bb304ba7981491c74b921275 : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc0167094f044c32958cda2da9d421b5 : SUCCESS in 3m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ffa42334de24af5b71d97b4ac52464d : SUCCESS in 6m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/643bcf57b9ab4f1a91ed66bf75a556ae : FAILURE in 3m 14s","accounts_in_message":[],"_revision_number":19},{"id":"978ba3b000146ad9e4e54f025e08e979e7346889","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-07-20 06:24:45.000000000","message":"Patch Set 20: Patch Set 19 was rebased\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":20},{"id":"9992acab34691066cb6de5acb1e74be295b0f0b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 09:04:27.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a6d1fa68c35d46d48dfae61de883e7da\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/00692c5c3d6a4813a1cddcff1c11de71 : SUCCESS in 59m 40s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/1d13e4e039334c99aee1feb666b8e557 : SUCCESS in 25m 12s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/e9b6ee735bb644a4a0674d90a91f6edc : SUCCESS in 27m 46s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/7b968d82419f48248265fda91c8ffc64 : FAILURE in 19m 44s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/2a940cc986b942f3ac2c391aaf60516c : FAILURE in 1h 06m 36s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b453e579f4274cdaae36e3876abb47c2 : SUCCESS in 24m 50s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/dfef77040a7c4a55af111b32f8d6d5ad : SUCCESS in 31m 38s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/78ce3c7db99e4094ab5375c0f11f16d6 : TIMED_OUT in 2h 34m 43s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/f7eed9143a984977854aecfa9e277ddb : SUCCESS in 1h 02m 03s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/576b6f7e18044cafa66ffb92f881db0e : SUCCESS in 1h 38m 31s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/8b310351d76749f084ab684998d8ba41 : FAILURE in 56m 48s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/99a664373e574274b2f8d628c9e3eb7a : SUCCESS in 1h 30m 08s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5865e8e0e0b04afa939d4336d6317188 : SUCCESS in 1h 04m 17s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/439bfd57b5904d0383fd06fa6dc5d6b8 : SUCCESS in 1h 35m 58s\n- kolla-ansible-debian-trixie-bifrost https://zuul.opendev.org/t/openstack/build/7f78c3149b41451eae6ce2b7dff3f6d4 : SUCCESS in 26m 32s (non-voting)\n- kolla-ansible-rocky-10-bifrost https://zuul.opendev.org/t/openstack/build/166bf21f8f2845d5b7e4ccd570bb737f : SUCCESS in 30m 51s (non-voting)\n- kolla-ansible-ubuntu-noble-bifrost https://zuul.opendev.org/t/openstack/build/afc48c9839134a9cb605a5c928c03755 : SUCCESS in 20m 21s (non-voting)\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/47900136167049f7ab399b3fbf76dae7 : SUCCESS in 1h 28m 13s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/4a99ed9dc19d4a98b39a41fdbfa1309b : SUCCESS in 1h 32m 42s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/4ff5481d4be348bb904d8ed99a706641 : FAILURE in 1h 17m 38s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/3f13ae0281d443818df1a5be1560c86a : SUCCESS in 1h 34m 04s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/4376a63e70194d0d8eb1719ab81731a5 : SUCCESS in 2h 17m 08s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/f99e0ff085c4467798ee8864417c00bf : SUCCESS in 1h 15m 21s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/676dd943c8444b9c8dee3f44f1ae197a : SUCCESS in 1h 21m 28s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/61dc93e5e52441a097d16ed1497eec10 : SUCCESS in 1h 22m 01s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/8ceaddd477174d8fbeeff39f5a02b507 : SUCCESS in 23m 26s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/119cb3209ea944608c13c6533815549f : SUCCESS in 27m 18s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/0d6ef74e98c7413e92e3a9c1b1707d86 : SUCCESS in 34m 54s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/28a31f602bcb4659833a8cfa80901390 : FAILURE in 28m 08s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/59e50319f5a943fea5f31e8834c75aa0 : SUCCESS in 57m 47s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/0525471a76a44c01839078b5be47df3b : SUCCESS in 1h 20m 54s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/2d1b729954ca450da0b79323780cdac6 : SUCCESS in 1h 02m 08s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/b8226d9c3eea47ffad2e13292c377d2d : SUCCESS in 1h 07m 40s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/294f64c3bef445bc91928d503f952ab7 : SUCCESS in 1h 49m 44s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/3f206e31567e4dc8ae74c131125063fb : FAILURE in 1h 20m 22s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/8ba4bf822bb741bba049edbdff8654bf : SUCCESS in 1h 49m 24s (non-voting)\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/bbfeb063d20a4cc487362452b7979ae4 : SUCCESS in 40m 49s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/ce32f79518ae4f84986afcd4e48ecd61 : FAILURE in 1h 02m 26s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/aef66bfa94b2433d8cad84bb6c974b7c : SUCCESS in 43m 53s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/5a7270bbb9a8429baf39963cb1e38e75 : SUCCESS in 2m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd88c91bf9fc441587d5207c81ef9f13 : FAILURE in 3m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4cac4eaa7f7247409d9bd5faf272fe26 : SUCCESS in 3m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/375a4f1c97444292b1db72b1afc552bd : SUCCESS in 3m 36s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/735dfdd78cbc465bb0f7fc788cb66a4d : SUCCESS in 7m 13s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/39801fd557f04cf0bcf2bf9c5cd09fc7 : SUCCESS in 3m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f2c0ba0975d4e58b9a5f00b54fa2a72 : SUCCESS in 6m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d9a057fe6fa34ac699771eb41b76c899 : FAILURE in 4m 18s","accounts_in_message":[],"_revision_number":20},{"id":"79acbae760cf97185a2b3a59dc1f5418d61d38f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-07-20 09:28:24.000000000","message":"Patch Set 21: Published edit on patch set 20.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":21},{"id":"e50a9a149fa5708531429a9442dc4a942809f20d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-07-20 11:54:49.000000000","message":"Patch Set 22: Published edit on patch set 21.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":22},{"id":"64d4126b0b2569227991c0ba5dd59d7d0033be8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 17:34:34.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c23f05d376bf460ea3097daa450d25ee\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/ffd8e92d6d3242d59d58d196377ed23c : SUCCESS in 1h 03m 14s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/9d47fa3416314c7683e2618f4474db3e : SUCCESS in 31m 43s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/11049ffedd2046ecbdeaf543cd5a73d4 : SUCCESS in 31m 48s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/1e3938454ba649ee85b2057e065187be : SUCCESS in 26m 40s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/4984be4340c142668a1279975e55de3b : SUCCESS in 1h 14m 51s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f1888227f2dc4c3c95232ecdf01b2d86 : SUCCESS in 26m 36s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/45f45517c81e4767898e77435d50f408 : SUCCESS in 31m 11s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/29cff7abc9aa44a095d5d9f7d8c808e8 : TIMED_OUT in 2h 33m 21s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/a9ce1fed088642f9a913e677f11be9f4 : SUCCESS in 1h 11m 55s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/762d514199484558aca7bf3e91ce1f8b : SUCCESS in 1h 33m 48s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/02de268cbdcb4ae4a8b4ebcf39b7d6b0 : FAILURE in 52m 47s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/b3b1f91a9ac04e4a94af1f83b0439444 : SUCCESS in 1h 34m 14s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2d27cee5a14f424596b005965841e7b2 : SUCCESS in 1h 05m 21s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/6dc621004d384ad3b56b7c971d261a55 : FAILURE in 1h 35m 07s\n- kolla-ansible-debian-trixie-bifrost https://zuul.opendev.org/t/openstack/build/3c095c91c67849ba9f7c1038f1bf7009 : SUCCESS in 28m 08s (non-voting)\n- kolla-ansible-rocky-10-bifrost https://zuul.opendev.org/t/openstack/build/07e88506c8c3413b99362d0cd8a5ade4 : SUCCESS in 32m 38s (non-voting)\n- kolla-ansible-ubuntu-noble-bifrost https://zuul.opendev.org/t/openstack/build/041f2e7a7cf6482490844469da45b59c : SUCCESS in 37m 23s (non-voting)\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/51d4464e152b4a55a4893967a249ec91 : SUCCESS in 1h 31m 20s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/e1c32209c28242b0ab9dda9f3cb212cf : SUCCESS in 1h 17m 41s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/4140f62f1c87490cbc4e58099045a90e : SUCCESS in 1h 39m 42s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/41f466d9141d4df78f80384f7c47d7e4 : SUCCESS in 59m 56s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/3b4192d439504cff83558c495c462d52 : SUCCESS in 2h 01m 51s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/58b900961e244e05b369ab8167062d98 : FAILURE in 47m 56s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/4e7e76abf58b49289220f33cfdf3ba1c : SUCCESS in 1h 04m 58s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/586ee7e608c84a6ba878a8f30386c83f : SUCCESS in 1h 37m 02s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/68242a57ced84fe7bdf32822bc9bb2c7 : SUCCESS in 17m 10s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/6fb2ba39da1c4201a14b7c38d0a696ca : SUCCESS in 15m 58s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/4770c2f932744506966c822bfe45aa59 : SUCCESS in 30m 28s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/5c5ad4e39d684fab92203c47b03059b3 : FAILURE in 50m 58s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/678068d9585e49008456adfcacad3ee6 : SUCCESS in 1h 22m 49s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/663efeb985254ac78eda3614499f7715 : SUCCESS in 1h 24m 58s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/08c879d253e54b6bb72d6b279c91f9c9 : SUCCESS in 44m 03s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/13f9d591e8ce40aeb8ec7ed849190aaa : SUCCESS in 48m 33s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/17e54e32b8b244f9a3f9fb76adef4d3d : SUCCESS in 1h 19m 08s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/1e17e2518e9249569eda5ced5b3504cd : SUCCESS in 1h 10m 48s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/5f7a84e50b414aab9048ef9f8c418c32 : SUCCESS in 1h 31m 22s (non-voting)\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/3e216d82c06d4685be5197c430000dce : FAILURE in 1h 22m 20s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/7b1629765e6541fb882fa8526e372ce3 : FAILURE in 1h 12m 56s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/6baccd650097497ea4a24397f14af2e5 : SUCCESS in 1h 44m 41s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/fa864b6d470f4a8bb2fabf11643ecb3d : SUCCESS in 3m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8f8fb421e174c5c99a99585fac64350 : SUCCESS in 3m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8ee3a270ab8647be8c8122a7871935e9 : SUCCESS in 4m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/77c5ff24b0c040f7b972791233725229 : SUCCESS in 3m 36s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5353106fb69f49de96787ecfbaac76be : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dfe105d426254bd0a7b41d5b56701ef8 : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fd84632e83f44b6998976e3093155d1 : SUCCESS in 6m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c3553b905a148f88b3c3dbfa67eb210 : FAILURE in 3m 23s","accounts_in_message":[],"_revision_number":22},{"id":"427399f295c15d89c68c7451a6af737875b53add","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 19:13:05.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea61902205c5464cb1ada1f7436abc89\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/deacfdd66f05419d975020c8592e8966 : SUCCESS in 43m 04s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/9fa603bfaeb645d982a6feafb7932444 : SUCCESS in 32m 09s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/47a7afe7af064fde8f108f95ab90b329 : SUCCESS in 26m 22s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/dc6d541a4fef418a845955acd6730a0f : SUCCESS in 36m 07s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/0e5449330ca4486dbf77288d471576fb : FAILURE in 1h 32m 44s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/f7307e558cae4752933055fd9d0584bf : FAILURE in 1h 12m 44s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0208a571880e42d7b2279dc029579b1e : SUCCESS in 3m 18s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f47d7aef08394105bd3a64e26271432e : SUCCESS in 3m 35s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/927e61a5d3324360977bc8a462a43866 : SUCCESS in 6m 41s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"16b401825bdc66399fb422cd7cfeefdf0e3fc7af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-07-24 08:13:06.000000000","message":"Patch Set 23: Patch Set 22 was rebased\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":23},{"id":"e10d32ac912c5137de29076174b807e92388eadc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-24 10:47:58.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4860a93f2a244051ba497ba33bce4b28\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/a0550755d5584ca59db4daef2c6100cf : SUCCESS in 45m 51s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/56fe8dd5f38b4a23880b80380143d925 : SUCCESS in 29m 45s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/90ea8783538c45eaa8e83abb31fb05ef : SUCCESS in 27m 55s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/d67b5d221b164d2db2eb071cd87cfb53 : SUCCESS in 27m 06s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/34e4efb3a4ca473cb1162d2de88939e6 : SUCCESS in 1h 14m 19s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eff57789cbcd4eec8eaaf028aba43d8e : SUCCESS in 27m 39s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8b56e358ccf94bb6b6a1e141fdfa6b01 : SUCCESS in 39m 32s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/2794bd5911054760b455b7244d0bec04 : TIMED_OUT in 2h 33m 25s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/cbd78acf717f46bd913af947e376ba7a : SUCCESS in 40m 13s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/169ced67ebb54a53a484095c737f4c08 : SUCCESS in 1h 20m 38s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/d9b38afa3b23421b8e86d87ba50e96b8 : SUCCESS in 43m 15s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/4911a4dbb28141e2a3395829de6bd03f : SUCCESS in 1h 09m 44s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8fbe35162aba41d883dcb20120936f82 : SUCCESS in 1h 03m 11s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/d38300da11f545ea83f591b2727c75c7 : SUCCESS in 1h 38m 50s\n- kolla-ansible-debian-trixie-bifrost https://zuul.opendev.org/t/openstack/build/34efbde30d0f4e9f95ba68b0fe0f44ce : SUCCESS in 29m 51s (non-voting)\n- kolla-ansible-rocky-10-bifrost https://zuul.opendev.org/t/openstack/build/8105696d31b3468f8972e16c569d839d : SUCCESS in 31m 12s (non-voting)\n- kolla-ansible-ubuntu-noble-bifrost https://zuul.opendev.org/t/openstack/build/695dfb82190f4281801c1571508e8ecd : SUCCESS in 27m 09s (non-voting)\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/2fb4b6fb6d304c2e9e5ee012da437795 : SUCCESS in 58m 53s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/e0794815d9434fcc860d17d68fc4660d : SUCCESS in 1h 01m 58s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/dacc52d0bca54f089c8fefed8303e61c : SUCCESS in 2h 04m 29s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/0b85c584aea5406fbd244a6cf8adfcb0 : FAILURE in 46m 40s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/2ae5b0a1f48b4a489d9069442cf2a3ee : SUCCESS in 2h 02m 38s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/eb6227db93654da28c9bd09fc5dd9940 : SUCCESS in 1h 15m 33s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/64775ab431dd4d2a8c05ecffa297744b : SUCCESS in 55m 02s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/39cee7bd692f4aac80f1ad9f0f19ce24 : SUCCESS in 52m 21s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/42e9e068d1834cf6a34dfb930dd00fd1 : SUCCESS in 17m 35s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/9ba39d3e244e47538fa658c0cfd39026 : SUCCESS in 15m 38s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/652125467f4b48d0be3b6deea5f868c1 : SUCCESS in 26m 29s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/858334d38d054e67acf62c5c4a06c8a4 : FAILURE in 50m 59s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/da3132125fa446f681964388ad737fbb : SUCCESS in 1h 27m 19s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/f52192a4a8794916b19015080b733080 : FAILURE in 1h 20m 44s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/38937914d2a54ef6ac6a605c0867cbaf : SUCCESS in 1h 01m 21s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/032e5182e50e4f1c9fc894c83da77c1a : SUCCESS in 42m 34s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/bef6d5af0ad345e99f779b8d61f8dc67 : SUCCESS in 1h 29m 26s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/1f07bf4cd2564fa48ddddcaaaeddc858 : SUCCESS in 1h 25m 47s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/51f35f17f2184b7f9d595d9da3e8a3ef : SUCCESS in 1h 23m 44s (non-voting)\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/1115cfe8aff74bb3ac60acedb1965666 : FAILURE in 1h 13m 44s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/09de5ed9614244afa6c2858b72521470 : FAILURE in 1h 07m 25s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/422e1dc1e5584b49935b3f21d634e257 : SUCCESS in 1h 33m 15s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/125ee29c03d34a29bb67596ef0179acc : SUCCESS in 2m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7845971072be474ba5305d9be460cf56 : SUCCESS in 3m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/034c2334b6a54c51a80b5a0e5b56c8b8 : SUCCESS in 3m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/648321cdee48407783d344932e5c63d4 : SUCCESS in 3m 13s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0d02d51603ba4673a7773192e3b00bda : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5ddc9f4cbdf54a71a03b511116dc6f19 : SUCCESS in 3m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15385aabf3814e09ac66a5c828a947f2 : SUCCESS in 8m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1880a92a297c481cb8e2469354a211f9 : FAILURE in 2m 46s","accounts_in_message":[],"_revision_number":23},{"id":"d0729a1cbc33137aee12f5b57dafe8a944b3d89f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-24 11:10:44.000000000","message":"Patch Set 23:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b8d588ae19d4608b0226e873797e8f4\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/4ebdc6e03e3347098338be634e0df05e : SUCCESS in 29m 32s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/3629d4fb89ca40a89393995935901c49 : SUCCESS in 38m 21s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/6246d7fd44d34d089804478cc0f92ba9 : SUCCESS in 48m 19s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/c476882baf364003a1b5b3c518a989d6 : SUCCESS in 33m 41s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/00de1875991c4441adddd81417a417cc : FAILURE in 1h 41m 05s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/9af670c6f1724f09b7729d05e3eb436f : FAILURE in 1h 24m 49s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f3506f5f65434c36b6eaa85877b66c72 : SUCCESS in 4m 11s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/158a7fc39fed4d8e8c46a176114480da : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/34937db8296f47a2bb3778563c5c3d25 : SUCCESS in 9m 27s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"cf459aec0b1680ce785043cc22e3d3ff738ffd14","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-07-27 14:21:23.000000000","message":"Patch Set 24: Published edit on patch set 23.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":24},{"id":"c0ce098d1e62ae745192a1c6e0e96428cf6ddc39","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-27 18:41:41.000000000","message":"Patch Set 24:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b07485d5a5540b2a5a2a8350774bd19\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/48d7dcdda24646f7b326886e920286f4 : SUCCESS in 34m 56s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/22f5275de33f4beb8663867d6df0c845 : SUCCESS in 45m 39s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/d57d4787c07045358d35a85f87dee0c2 : SUCCESS in 37m 39s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/9800974094c74b1f98aca7ed62648860 : SUCCESS in 42m 21s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/b0bae77700a2408bb6a8fcddbdc29c3f : FAILURE in 1h 32m 35s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/b4fb06877ad34b118632b8b1ed3f6224 : FAILURE in 1h 34m 18s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f5b1674569284a4d8bfb99a002e139e1 : SUCCESS in 4m 11s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/442dc789fc594a889f810bff1ea299fe : SUCCESS in 3m 29s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f17b23d50cde4b91b4067df3666d4580 : SUCCESS in 10m 02s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"0f0904ffe13fbe7446cb36af01920eebe02a56a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-27 20:06:40.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06f25d63080c44c8adb2383d90458d8a\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/cb5ed1a85d2843a38b999e33e3e9831b : SUCCESS in 50m 20s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/eedfd54a962546a1817b4e9c4cae061e : SUCCESS in 24m 42s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/a9c2e3ce66044658b071966f5f9e2344 : SUCCESS in 29m 16s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/e8bc128531fd4728aecd5de64b75a766 : SUCCESS in 22m 17s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/b7fa9008b56944feb451dcace6d5d7f5 : SUCCESS in 1h 13m 15s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/625f01d5627c4d8880a375e4c275f929 : SUCCESS in 25m 09s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8bdd7117f1384e588fdfadf4b2db712a : SUCCESS in 32m 06s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/d820f50c63d948c4975d6127438ffb5b : TIMED_OUT in 2h 34m 13s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/9fadd12f5dcb4abf98da5f42b7fde7cf : SUCCESS in 1h 00m 45s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/eccba3c32b734bb08e67bde49472d415 : SUCCESS in 1h 31m 50s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/cdb378b680dc44e1b141d95c1f49bd10 : SUCCESS in 1h 02m 34s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/1e64448c33cb4aac872821282ca7cbd7 : SUCCESS in 1h 29m 32s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e910137dac1f4e56860c24c93d7973d0 : SUCCESS in 1h 10m 09s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/357de854c29945f3af49dc8e3409c518 : SUCCESS in 1h 44m 31s\n- kolla-ansible-debian-trixie-bifrost https://zuul.opendev.org/t/openstack/build/6d1428a8976640f49c29d4db9ac91ea2 : SUCCESS in 31m 18s (non-voting)\n- kolla-ansible-rocky-10-bifrost https://zuul.opendev.org/t/openstack/build/208aad32954047d8b008a5f90833e016 : SUCCESS in 24m 22s (non-voting)\n- kolla-ansible-ubuntu-noble-bifrost https://zuul.opendev.org/t/openstack/build/8492337b5a544e6db3db42e16550acbb : SUCCESS in 37m 16s (non-voting)\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/da0488fab1464f06ba1e95a39e7be8bd : SUCCESS in 1h 28m 25s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/9796efd362c349ab8201bc201a070270 : SUCCESS in 1h 30m 04s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/56e1b1556d434363be3688868a85c4fb : SUCCESS in 2h 31m 35s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/b90dbe4ba4d2485b887ccd67e4eed303 : SUCCESS in 1h 42m 36s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/af14df134e8d4bc8a84df3bb20d07393 : SUCCESS in 2h 35m 39s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/e568d84b48f04420ba43ef9da212b730 : SUCCESS in 1h 19m 45s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/f4277c1888b6498cb2c27b1f5955a090 : SUCCESS in 1h 20m 31s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/e77c9b4caf6d428584bf67117299acd8 : SUCCESS in 1h 23m 42s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/cd711e412e6745d59d67daf30a1071e6 : SUCCESS in 23m 44s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/a884b6e55eaf46ca999245765ebde93a : SUCCESS in 25m 12s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/ef2f720b48fe4543ba06f640018a0ff6 : SUCCESS in 31m 37s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/4c9e2e5a98be4763af9f4c7dbbf786cd : FAILURE in 50m 56s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/42b3888390f84edc98efb625f01d6207 : SUCCESS in 1h 11m 09s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/bfe593cc920442e393c6554fd71555de : SUCCESS in 1h 25m 43s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/f67ea1100f4546fd85350f313dfe5eaa : SUCCESS in 1h 01m 25s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/842e6fec765c46b0bf82a641d6000753 : SUCCESS in 1h 24m 20s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/07b8f382fff948e0b69bd26304cbb1a9 : SUCCESS in 1h 40m 47s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/2a62c1d4ea1949b6901bfcaa8a921795 : SUCCESS in 1h 22m 22s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/c155320c4dcb4691a272e502ae42b247 : FAILURE in 1h 23m 45s (non-voting)\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/f82684a3ddad4f1283bc2293c7f6ad4a : FAILURE in 1h 07m 39s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/c2ba8b30866041ae9aa7ea63e764a5c3 : FAILURE in 59m 14s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/515315968f57463aab1db442309c39ea : SUCCESS in 1h 31m 09s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/5351cf0171cc42c586cc571813f0bea0 : SUCCESS in 3m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f8ba442432f4a9fb7c487ac362bf213 : SUCCESS in 3m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/76b74a8fdc934662b66464d9f9fe43ca : SUCCESS in 2m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f2ef8bf672b94609afd0158edf207ab1 : SUCCESS in 4m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ca813dad6b4544dbbdaa6670a3b43e6b : SUCCESS in 6m 52s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cd79892b879747fe9e9a2fda25af3981 : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9239eefb19ba4db297b4ec299964a29e : SUCCESS in 6m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/69e16bf1a3e24893a5541b4e61dcb8e0 : SUCCESS in 6m 06s","accounts_in_message":[],"_revision_number":24},{"id":"acabc650b30d541a86a454cd0581aecc076896aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-07-28 07:22:38.000000000","message":"Patch Set 25: Patch Set 24 was rebased\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":25},{"id":"1a12d79b0d99ce437618b8be0ef5a3440aed62e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-28 10:11:24.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e22cdf8d24ac4e2fab6f48aa928889b3\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/0086d36adf8c43498d0f54c797f48e5b : SUCCESS in 48m 34s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/995cb952904f40d8ba01b64139156924 : SUCCESS in 26m 07s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/a582e8f4ba31449a9c2f3b5750aa5147 : SUCCESS in 25m 11s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/335b357030fa47628f77dfa7b6d8479a : SUCCESS in 30m 35s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/406d6ab6f88a4b15967451820eb7dc03 : SUCCESS in 1h 05m 24s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f12f5eaf8bcb4aceb8e406e28b5db4c3 : SUCCESS in 21m 32s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/35beeb3f4e7247f190acf83a0fe87367 : SUCCESS in 36m 38s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/a8380cf3319846a984d3c127f9be3bf8 : TIMED_OUT in 2h 33m 35s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/ab9795bd1a104b33b0bee73f76e398bb : SUCCESS in 1h 03m 31s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/e44c16329ee34566bc40a6146d878304 : SUCCESS in 1h 32m 40s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/f1a36e8370324592a054aec1a357fd15 : RETRY_LIMIT in 4m 57s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/57a361482028468aa99dbb66a6164625 : RETRY_LIMIT in 5m 09s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9915cae738074129bd954412b009c08c : SUCCESS in 1h 04m 52s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/14abc620a13c43599c1b983725a0b9eb : SUCCESS in 1h 40m 56s\n- kolla-ansible-debian-trixie-bifrost https://zuul.opendev.org/t/openstack/build/8d78cdd7050f4ed4bd78d2f6d5062f53 : SUCCESS in 27m 11s (non-voting)\n- kolla-ansible-rocky-10-bifrost https://zuul.opendev.org/t/openstack/build/eef41001d86b4269bb52212ea467ce35 : RETRY_LIMIT in 5m 07s (non-voting)\n- kolla-ansible-ubuntu-noble-bifrost https://zuul.opendev.org/t/openstack/build/31b6138235c64e45a520e6d23c315780 : SUCCESS in 30m 23s (non-voting)\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/c4e1af6bb9534d52a00a6b2d63d4e89f : SUCCESS in 1h 38m 17s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/997cfa3de171445c929fcdb6dbcf9933 : RETRY_LIMIT in 5m 48s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/ad89acc400af44a48e0cc86c0ed24dc6 : RETRY_LIMIT in 7m 35s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/53d0daf48489428fa4b9d8d6c5231798 : FAILURE in 1h 16m 17s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/5621889dae784d2aa9c12a2e6b65bbb3 : FAILURE in 2h 44m 41s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/e58f26001cf34ef682d976c20bdef477 : SUCCESS in 1h 17m 28s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/c67bcd723b634953874a4122fab6e233 : RETRY_LIMIT in 5m 56s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/f774e4bfe56647b78803e51908ee8672 : FAILURE in 1h 05m 39s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/e6aea512be0647d8992c83e5c29a7085 : SUCCESS in 22m 57s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/1ef1a576c4364bbc9fb3d301cb26c850 : RETRY_LIMIT in 5m 53s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/41f32d941927431d87f9924f5c7709f6 : SUCCESS in 27m 35s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/7f91949cb1d3477cb32bc14e4360f3b0 : FAILURE in 41m 55s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/e8c8505744344593964a6e9586dbbe1a : RETRY_LIMIT in 7m 00s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/62aefc09c8b34804a71cf8e67745da0a : SUCCESS in 1h 20m 52s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/ac5ee58886474a0a92270496858193a5 : SUCCESS in 1h 04m 57s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/2432e37a2aa04221bbd303020b5224a6 : SUCCESS in 1h 07m 56s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/d530d82aec6d41fb97c4dbf00ca329b9 : SUCCESS in 1h 42m 18s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/01523c1800414901bf4fe96aadde62ed : RETRY_LIMIT in 6m 36s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/5447e82590ad4006a7e4045b251b3b72 : SUCCESS in 1h 09m 22s (non-voting)\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/ceec822d26104faf947195c58f6d7db2 : FAILURE in 1h 19m 13s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/b5c7d3f8fcbe44beb85858e28acd5492 : RETRY_LIMIT in 5m 09s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/ba6a7f66b9274c56a4bcd268bda271b8 : SUCCESS in 1h 34m 53s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/f73c9d9177ad4293bf044e3152b651e0 : SUCCESS in 3m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a759a58bb8054506a0d2365783761f08 : SUCCESS in 4m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fe761d8ff69c442db0c2ea8e88b58d6d : SUCCESS in 3m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cc0f35dee95b4dcd987e855d39171bb9 : SUCCESS in 3m 29s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/24daca215a7a4baa8ee82c4edf472a57 : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/be5240219b2c4ff8908f9d8931cfc7e9 : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b028541faa6a418d925d3d0b401c1009 : SUCCESS in 7m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/147f816bf69b4a2e8ef8763048049cd6 : SUCCESS in 11m 37s","accounts_in_message":[],"_revision_number":25},{"id":"ef460f615332d222a365f79b37ef73baf68f2567","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-28 11:08:24.000000000","message":"Patch Set 25:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d499f6ed96b34d0c87d0b2dc5b16237b\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/daf9043fff3647d8a8b84bb668c08430 : SUCCESS in 26m 07s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/e0d37405af6046af9f053b1020e58859 : SUCCESS in 46m 30s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/4814832e9e7a4c74af6cfab3132ae4dd : SUCCESS in 41m 43s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/c28c69e6de724a91a1b6e073647810c8 : NODE_FAILURE Node(set) request 0507ffb32e474103b56d725675582366 failed in 0s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/ae9d8981e65948c7a27400527420c5bb : FAILURE in 1h 10m 20s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/ceedaaf2ae024a2ca82595ad39037fab : NODE_FAILURE Node(set) request 7e5a29798ab04eaaaded2e9df6d8442e failed in 0s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2e755d5a15e04b10ad82af852cf7228d : NODE_FAILURE Node(set) request f0dc8de702ec4649b23e6538a1b22aad failed in 0s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0dc9604212b84dadbdabf561dde2241d : NODE_FAILURE Node(set) request ac6a85b4de424660a42d0d7567ff4a4a failed in 0s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f18a935c6b2e48779c094a37fc977e35 : SUCCESS in 5m 45s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"459d7b4df35bacc8c3d274db2ea07f3ff7671068","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-07-28 11:30:45.000000000","message":"Patch Set 25: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":25},{"id":"f790ded09b037c7cf609c9667e5f25f108073cee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-07-31 17:19:50.000000000","message":"Uploaded patch set 26.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":26},{"id":"aff3c39cd3297ed2ae1d96c9e253d3ccbe337b92","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-07-31 17:20:17.000000000","message":"Patch Set 26: Backport-Candidate-1 Review-Priority+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":26},{"id":"fd9d5684f548904d2e5d293e9a921923d992f995","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 19:02:04.000000000","message":"Patch Set 26:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/653aa430f1b04a21bdf58d39c150937b\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/2f0245f2d19545a89236272d3476d0c8 : SUCCESS in 50m 18s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/43874cfcbec54249a3821ba214d4ee7f : SUCCESS in 49m 38s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/105475e41efc4ebca271a8e3288f1424 : SUCCESS in 38m 05s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/dca9c388f78845ad9787284a80f0fb1f : SUCCESS in 54m 52s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/0afc40c4aa9745a89bc35890dd357d57 : FAILURE in 1h 24m 10s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/94c2d81030ac414ba83cda86807a3af5 : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1f72988f629d487f868f8bb76cacf0fb : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/068cf3588d004ad2a093b423c89c0fe7 : SUCCESS in 10m 33s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"f8d6997df6a0e30148fd983043f943878e237f17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 19:05:52.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b82e49bb3fb48cda17df3bf70e2ef9c\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/35c2a1ad8e494fd7bc77ae1b21ef5752 : SUCCESS in 57m 09s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/666d3990f5f843efb13997b5d5ed7578 : SUCCESS in 12m 37s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/c636c2e5b587478fad5e2609bfbdee8f : SUCCESS in 29m 12s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/a55dee2c53054b41bba3866ddb6cb671 : SUCCESS in 16m 53s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/16d0beca356044599e173ba159faa2d3 : SUCCESS in 1h 09m 10s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/67153aeef77b4ffd8d629f05d9fdceec : SUCCESS in 24m 10s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5d7d826530954d698dd21ae51aa8b160 : SUCCESS in 37m 16s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/24d6e80d33ad43eda4ba313dda7bba98 : SUCCESS in 59m 54s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/e8889ef78d954b8faf8becb0ecd91995 : SUCCESS in 1h 31m 31s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/dd55513710ba440ba75ffdffe1726d10 : SUCCESS in 1h 01m 40s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/267423a6d2e649f89d77a024e580436b : SUCCESS in 1h 34m 40s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/818fb3c40a4e410db3f47582e033fd56 : SUCCESS in 1h 04m 24s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/0a4c0a0c2aff44a2baaab8862441b44c : SUCCESS in 1h 36m 51s\n- kolla-ansible-debian-trixie-bifrost https://zuul.opendev.org/t/openstack/build/69dc1b53b794444aa077160eea2c0ae3 : SUCCESS in 19m 33s (non-voting)\n- kolla-ansible-rocky-10-bifrost https://zuul.opendev.org/t/openstack/build/917ca30d5d7f4f719a0fdc733d6ac7ae : SUCCESS in 19m 05s (non-voting)\n- kolla-ansible-ubuntu-noble-bifrost https://zuul.opendev.org/t/openstack/build/baa52a5c13b84dd6a69597378cc27b69 : SUCCESS in 30m 15s (non-voting)\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/a44d1c5a176d437c9c67b57f179f062d : FAILURE in 1h 07m 52s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/09be14b3dc93443683e6b73a51eb30c2 : SUCCESS in 1h 28m 02s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/498a2c1a0e004189aa48cd9759b0ed4f : FAILURE in 1h 28m 47s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/c6592c4badb241ca8f0d0ddd95a9658b : SUCCESS in 1h 36m 00s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/90a5bdc33c3f474dad66e13454cdec4b : FAILURE in 1h 13m 55s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/ae3facf31da94c9191f61df52d4c17d2 : SUCCESS in 1h 17m 38s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/c22e0bd94a0849bf8a6ce7a39d57adb2 : SUCCESS in 1h 13m 54s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/cf59a2faf7164ad792f32e766efa0cff : SUCCESS in 1h 32m 35s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/a62a75a3b174479e831222fabbe0799c : SUCCESS in 23m 58s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/24be4403422f4c47af1b11603016a2ac : SUCCESS in 28m 22s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/ca8c08de648b4a1593125ee62393f747 : SUCCESS in 40m 15s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/b924b51a7627475b87a8c116db5100e7 : FAILURE in 52m 31s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/d6074eca14f54051998cca2a134912f5 : SUCCESS in 49m 38s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/45f9c92214e5413fa4215726a2744179 : SUCCESS in 1h 31m 48s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/36ba16c8beb048c0886e7dd136ac6a95 : SUCCESS in 1h 00m 29s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/ce539e2ff1ba40ad8a5384d72ce71f11 : SUCCESS in 1h 07m 14s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/f0c5a457c5104ef5b05073a08d789919 : FAILURE in 1h 20m 13s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/4531b39baccf4d0ebf410a3c6213fd4e : SUCCESS in 1h 38m 10s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/7cf363a3a0604537a21c0c1e765fc2cb : SUCCESS in 1h 44m 02s (non-voting)\n- kolla-ansible-debian-trixie-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/06ae333a680544e0b16dd3fd70790c54 : SUCCESS in 38m 51s\n- kolla-ansible-debian-trixie-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/5286972bf0cd4012bc8c9876799f050f : SUCCESS in 1h 07m 48s\n- kolla-ansible-rocky-10-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/567fbe0ecb2944f2bfabe0e91eeae015 : SUCCESS in 42m 19s\n- kolla-ansible-rocky-10-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/a3e6d059c79d410da2542ccd3c947b3b : SUCCESS in 31m 30s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/2f59755c61c74b4894b1d56fcb37071e : SUCCESS in 42m 13s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/fe285fe4cd2b4aeea5b9e9afaab69fde : SUCCESS in 57m 05s\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/ef2f89fc93d64386bbd01fb87971fda2 : FAILURE in 59m 02s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/14eb2543eb534f8198a9ae6f7e2e4d16 : FAILURE in 1h 11m 46s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/8851913ee8d04999960a543ed78d3f36 : SUCCESS in 1h 33m 41s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/e6fe20fc5c7e4833849a0d538e2fb342 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c735f55cb296451f80eafd62bdef3695 : SUCCESS in 5m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0cd4e92073884e5aae2f1c073985f790 : SUCCESS in 2m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1daaf860305d42728b741a3f37ea2fd2 : SUCCESS in 3m 15s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8db6fc563f9e4f55808e7d2492315a65 : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b59fc98a332c476983efc8a5614f1def : SUCCESS in 2m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76b64a22a22548ab8e550f40cf4fa3f4 : SUCCESS in 6m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8464866b044d427d8382e3d2049b4d22 : SUCCESS in 10m 38s","accounts_in_message":[],"_revision_number":26},{"id":"48c63d3f2bcf7cbe84e327bc17ff4585ef6175de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-08-05 08:13:23.000000000","message":"Uploaded patch set 27.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":27},{"id":"018e6e5fcb304be40079c348b367b4fa38fbd7dc","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-08-05 08:19:34.000000000","message":"Patch Set 27: Backport-Candidate-1 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":27},{"id":"dad4d27acfceea56bfdfc0ef3b04f3f40507b88b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 09:21:56.000000000","message":"Patch Set 27:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/949db610e7034ec091bc4e1d24c8ba03\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/07b5a95832e543b48bff2c807c69b8ac : SUCCESS in 41m 42s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/2490468387154737bd3d5133df5354b8 : SUCCESS in 40m 52s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/6e13743e254643239ab98edbe5bc0291 : SUCCESS in 43m 17s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/092be95c53684530a474301c297e413d : SUCCESS in 43m 26s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/dc935875ca134c1cbf09c72163109ebd : FAILURE in 1h 07m 03s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/caa980de68164f59a65137be2fe21b02 : SUCCESS in 8m 29s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7e52ffd631504ea1aca1721dd1af2a05 : SUCCESS in 10m 08s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/8320cde7ad2a4178869c462aa52575c8 : SUCCESS in 10m 39s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"6d7addac55c55e2109c1a0d4dab5c32f5af092a8","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-08-05 10:39:27.000000000","message":"Patch Set 27: Code-Review+2","accounts_in_message":[],"_revision_number":27},{"id":"41126c107160e1146c69cd6fe9126f844b448718","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 10:46:28.000000000","message":"Patch Set 27: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f4766c5972134c32b1f74884838c23bf\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/4509ba97613a4f63ab33c282d184553d : SUCCESS in 54m 17s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/9488f52f6c774b3582ed1fa3f542fca0 : SUCCESS in 27m 03s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/9e6970a97ab24626b9ba5c4ec7dc20a3 : SUCCESS in 17m 27s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/2a42f15b76444a60901e2c1b1a70aeee : SUCCESS in 17m 56s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/681ae35057ba42c489306d1a05382f5b : SUCCESS in 41m 36s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f10eeaad8beb486e9250df81949df30c : SUCCESS in 16m 15s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/86ae574058a04b9ca0f28c7629128778 : SUCCESS in 31m 53s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/8a556e8d4afc4198b7aae4078d04264b : SUCCESS in 1h 01m 17s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/3cb68de5ac5b4940b94ed752de3bd4f1 : SUCCESS in 1h 26m 03s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/8fbaefd3b8d14c0d9b1d0ccb786a36c6 : SUCCESS in 1h 04m 59s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/0fa0843dc0494d9785daa59b4ef9c78f : SUCCESS in 1h 32m 55s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/980f5ead87df4e18ab4c9f87e2e695b5 : TIMED_OUT in 2h 31m 32s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/2ee74f5c570440b092afd9edbd8295c1 : SUCCESS in 1h 31m 10s\n- kolla-ansible-debian-trixie-bifrost https://zuul.opendev.org/t/openstack/build/0493cb89abe1490f9225dba3e178154d : SUCCESS in 16m 26s (non-voting)\n- kolla-ansible-rocky-10-bifrost https://zuul.opendev.org/t/openstack/build/7073358b2f93416ba8b610097fe71a20 : SUCCESS in 17m 07s (non-voting)\n- kolla-ansible-ubuntu-noble-bifrost https://zuul.opendev.org/t/openstack/build/2d7d727989ef462ba0105314c80dc64c : SUCCESS in 17m 02s (non-voting)\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/da7c83712a7e41919da6e9796c6e1915 : SUCCESS in 1h 27m 15s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/2a25cae09bd44bc19e4122de276256bc : SUCCESS in 1h 35m 36s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/1c61516b40504bb788f3bfc8733c702e : SUCCESS in 2h 10m 33s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/81e5e52563704f90b95c2f9f9caa8b47 : SUCCESS in 1h 31m 43s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/2edc84d782954c888559f13a3175e90e : SUCCESS in 2h 11m 41s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/1bde21251e63473f98e6937a58d0e86a : SUCCESS in 1h 18m 03s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/61f943c17b7e44818c3706f4b2ba0b11 : FAILURE in 8m 00s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/a0fd39e778ca49f5b93bf72e33a67d7f : SUCCESS in 1h 23m 15s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/f705b546c4b04602a8d2a74fa9843f24 : FAILURE in 8m 20s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/39134786a37047e4bac160cb391bac53 : SUCCESS in 22m 19s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/7bfaa2645b5c4e5cad83f8d3b2e057f1 : SUCCESS in 18m 14s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/eaed6a21d0bd4ba6a69d4930b95361c6 : FAILURE in 48m 06s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/2a7433c729cf4cb1a25f53ed21386dd8 : SUCCESS in 1h 08m 55s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/5f8d687c1b18443d8843f8aa11ac0ddd : SUCCESS in 1h 16m 02s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/b645a7f2101b4d4d8a37981a44d5c39a : FAILURE in 7m 39s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/a00825f224884c3ea228a29a84817fae : FAILURE in 7m 28s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/40e86aa052174044a43f9998c345cf9f : FAILURE in 8m 02s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/508b9b9793d345e9a3a0e684706156c6 : SUCCESS in 1h 40m 31s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/4863d4aea4be4e9287d75c9e4e1b0c74 : FAILURE in 1h 09m 47s (non-voting)\n- kolla-ansible-debian-trixie-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/3a6007fa426a46f19bd70b0228b5898a : SUCCESS in 20m 37s\n- kolla-ansible-debian-trixie-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/365625a6f8054071a915381ed29bbc93 : SUCCESS in 52m 29s\n- kolla-ansible-rocky-10-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/af0797bf58f246c195f78a202b26b9aa : SUCCESS in 41m 29s\n- kolla-ansible-rocky-10-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/09c47724ebef4c70acc8a0d89d6286a5 : SUCCESS in 29m 49s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/dff2748bafc04e6483958169cadeee3e : SUCCESS in 19m 59s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/26e8b43c62554d7e90fbef2ced6020d4 : SUCCESS in 55m 42s\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/662137b5378b43acbe137cb35d3b5ee4 : SUCCESS in 43m 19s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/daeb4a51517844c5b5c2676a858161c9 : FAILURE in 59m 14s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/6d4c8d17b7ef4210a44f3f2f39769aef : SUCCESS in 40m 39s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/efec5de9560446e1a9e8500a280a7951 : SUCCESS in 3m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8608f3b896840458bcffcc71e8a64ff : SUCCESS in 4m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4d44f3e9db1e4ef185c001ae100febb8 : SUCCESS in 2m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f52304642dec4b4abfe20694719d41fa : SUCCESS in 2m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/27c4b594cec24ca7b266d094f94ffd53 : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8461cc90812e4b8a81532e14d088762c : SUCCESS in 2m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e2b1b9a564b460695e466a5ac33bf93 : SUCCESS in 7m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bb82f1e4f65b439e818b5b952fe03da4 : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":27},{"id":"6837f90324fed5638bf4dfc8f274d9b5d6b0fe8c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-08-05 12:34:51.000000000","message":"Patch Set 27:\n\n(1 comment)","accounts_in_message":[],"_revision_number":27},{"id":"93f5d7b216bc3505fad343faa99b29b2260a210d","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2026-08-05 12:41:10.000000000","message":"Patch Set 27: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":27},{"id":"3b97bddf1a67a114131423ec8f2fbed8aba87a29","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 13:35:08.000000000","message":"Patch Set 27:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ee0393a55514e30bbf0e401d911ea55\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/a4f5477e61e6488c97f9e84f8b754c0e : SUCCESS in 35m 15s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/b2eb6c99cae44c5dbedb515ba149a39f : SUCCESS in 28m 21s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/f3691d24eed341228d90a8c59dae8c33 : SUCCESS in 27m 15s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/6928fa6191264442902e175d80b55da6 : SUCCESS in 32m 44s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/abee19f01eec4ae5929119950028d859 : FAILURE in 54m 59s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ba617632c1b84a6b8738f8ed8e9e9f06 : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/694a7dbaad3d4c799a1922dd9913d2d6 : SUCCESS in 4m 12s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/5581c710d3ed4ecb983a92bbf64e07f6 : SUCCESS in 7m 14s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"6f986034b1d7b45933da210f763e870bcc9e228b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-08-05 14:40:56.000000000","message":"Patch Set 27: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":27},{"id":"ed66b61d3a272a5681b935d0171855491a90db0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 15:38:56.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9823cd3b979b439da6f0679e3ea3e085\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/41b0e807b06947a6ba1556e95579b3e1 : SUCCESS in 1h 15m 53s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/a49fe31b5c3d4f22acbbdff8f39e9e75 : SUCCESS in 26m 29s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/1661dbc76e3e4a1cbbf0890373b0a2b8 : SUCCESS in 20m 04s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/8e02159f78584eb0a87532fd0f530d1c : SUCCESS in 25m 20s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/4c5969746ffb4537bebd33bd5370169a : SUCCESS in 58m 16s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bb79c23b5cab49be9d1009131e2f10ec : SUCCESS in 21m 44s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/153f225cf3ab44e58b6618240a649d43 : SUCCESS in 31m 10s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/86784878a27b4aeeab7ec94a36af5e44 : SUCCESS in 1h 02m 37s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/5e4399129fd94ab9aff2985eef22618e : SUCCESS in 1h 33m 20s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/ed31552cdde04c43b9f9309028e62d8e : SUCCESS in 1h 29m 25s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/39635d7ae6154034a1dad84511619ae9 : SUCCESS in 1h 32m 41s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5a5d2cdca3834cad967129c01cf5c299 : SUCCESS in 1h 26m 16s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/38d33d750ec54c21bc19110a7d451afd : SUCCESS in 1h 41m 20s\n- kolla-ansible-debian-trixie-bifrost https://zuul.opendev.org/t/openstack/build/fe37d4bfbb2c4d289c3d06a366819cf5 : SUCCESS in 29m 17s (non-voting)\n- kolla-ansible-rocky-10-bifrost https://zuul.opendev.org/t/openstack/build/a1bcd2c5dd364197ae88d93aeb376c0a : SUCCESS in 25m 20s (non-voting)\n- kolla-ansible-ubuntu-noble-bifrost https://zuul.opendev.org/t/openstack/build/de732d682882430784d01be79b2ee2ea : SUCCESS in 26m 28s (non-voting)\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/0c5f95dfc3c24c7a9462d70cc7f74cdc : SUCCESS in 1h 32m 33s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/9b9afd3cddab4b3b84f0478d17c2888c : SUCCESS in 1h 41m 21s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/19d4ef745e5540c5b13a3c2a374d83c9 : SUCCESS in 2h 53m 51s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/ec350aa6dabe461ba68c3afc4b663dc7 : TIMED_OUT in 3h 02m 00s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/5eb8a56efa2d42fe8f91cb08e1a99ef2 : FAILURE in 1h 19m 39s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/d2d482787e9c4d5da8430d08e8dd6831 : SUCCESS in 1h 24m 49s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/b4bf63c10e7e479c8191e51da23ed2fc : SUCCESS in 1h 23m 23s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/d01dcb56e1054ebfba2569407d4c22a8 : SUCCESS in 1h 27m 30s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/ed103c8e8f964854b26d79eed066f2d7 : SUCCESS in 21m 51s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/451cbd6ac9724dd78d6ef7df0a19dc2b : SUCCESS in 23m 13s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/2a028635db834769bb324556b821b2ca : SUCCESS in 29m 44s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/b3d060af096d43bba8291a58408a2c0d : FAILURE in 42m 40s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/fb7314475bbb40329e5f9ba0ce529ddd : SUCCESS in 47m 22s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/f111a2c9e0d24940bf710ad1a54062db : SUCCESS in 1h 21m 50s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/2f715e076e2f419a8df3a0d35a8de8cc : SUCCESS in 1h 04m 17s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/82b14115a65644159e9997f19ff53de7 : SUCCESS in 1h 10m 04s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/84378d90194a40e8ad894a17c3d11152 : SUCCESS in 1h 44m 26s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/4819aad10a2845fc8b921c3296012fc0 : SUCCESS in 1h 54m 08s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/c72f7d87d00944459ca36d257d0f117b : TIMED_OUT in 2h 37m 39s (non-voting)\n- kolla-ansible-debian-trixie-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/040e50f3a2a941f980efd226d52a6b1e : SUCCESS in 42m 17s\n- kolla-ansible-debian-trixie-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/4101fb14e1f145bb8213442c3768cfd4 : SUCCESS in 54m 18s\n- kolla-ansible-rocky-10-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/464acc6e8ee747b6a38df40f73e09d6b : SUCCESS in 24m 42s\n- kolla-ansible-rocky-10-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/86b1f9d1cc0c4afd9cbd363533fd0688 : SUCCESS in 31m 58s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/7446a56ffd2e43efa03d892cbaa51c8c : SUCCESS in 40m 53s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/3274378ff79d4cb5a4c4d02e2853f9db : SUCCESS in 58m 24s\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/a40662de5da741e882bb0a77a956a977 : FAILURE in 1h 22m 08s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/8471d59c323b42278da4af53f1a9fbbb : SUCCESS in 56m 20s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/4c9044bd388a438084949ef4aea388d9 : TIMED_OUT in 2h 31m 35s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/d90701ec306e42b9818a60f15951d4dc : SUCCESS in 3m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa57dbea15344716a0f97b25681c61e7 : SUCCESS in 3m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5abb3ba92ebc49ee8f7d751268cb6664 : SUCCESS in 3m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/875d1f9a351049e9a6d9302965e0f4cf : SUCCESS in 3m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/605eaa4489bd4d6cad99d89ce1bf59e7 : SUCCESS in 4m 07s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a9c1a64746f84c74bac4eba18fe588fa : SUCCESS in 2m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df538285a3604f2c9d9840f01414d168 : SUCCESS in 4m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0a39d8b22f7d463692c73120ee97e821 : SUCCESS in 6m 16s","accounts_in_message":[],"_revision_number":27},{"id":"8ceeca83761457e05835092ae2482a09bb30c4a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 15:39:29.000000000","message":"Patch Set 27: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":27},{"id":"7ba72c751574a4c50d83e91811135acbbca022d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 18:13:12.000000000","message":"Patch Set 27: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7976070662b64b8cb073e6bae6a0a8a1\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/5e14f13bc57244219de1a3352ec17f2d : SUCCESS in 18m 44s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/ba76b59f23ae430d96a8f27e60f4431c : SUCCESS in 28m 46s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/dab844f037494900be539eb20ba94400 : SUCCESS in 23m 43s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/f2103e2eba7d4963a1afb98015f89af6 : SUCCESS in 56m 06s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/86b7289175274fa8b5f7b96fee622b4c : SUCCESS in 25m 44s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8921b7e2d09a40d5b665bc32be376c58 : SUCCESS in 29m 28s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/e9028a87025745678fd6199a6e2079a2 : SUCCESS in 1h 03m 10s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/87694d548c5e466e9db0601744f55b39 : SUCCESS in 1h 32m 30s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/f67ef49d4ded44a5af7bbda42242196a : SUCCESS in 1h 04m 29s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/8e3406ea5f0544e0bb540228027385d2 : SUCCESS in 1h 32m 32s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bfe4d8f6f3094f04a23a357378b96789 : SUCCESS in 1h 07m 19s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/d65d3ccededa47aca347a94c50f84bba : SUCCESS in 1h 38m 29s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/f16dcf15d8984bf39ab5f6e2dcf359cb : SUCCESS in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a53d38473034ce4843434b764e7d159 : SUCCESS in 4m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f48d78c73a8b43d88cf364b1f62aa3aa : SUCCESS in 3m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/26d19c71c4294622bf4c0269a32edf74 : SUCCESS in 3m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bfde8cefc44455eb3a0744253934241 : SUCCESS in 9m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7131c8ee738b49c2a6c1051db1e48ac5 : SUCCESS in 9m 49s","accounts_in_message":[],"_revision_number":27},{"id":"9200bbca593f88449669767550e2e03aa2c69aa9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 18:13:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":27},{"id":"ccfc7d443ba03c6f47803caa0cb6a3361e45767d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 18:14:25.000000000","message":"Patch Set 27:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a010bf3ba8e14c6fb78bbab1e16b54ad\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad7a87c70c6743c882d9a138ff9b2969 : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9e9ccf185e3845a6ac0801efd8e291b8 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":27}],"current_revision_number":27,"current_revision":"5aabec47f057d9cb32aab520d666e51bd9ee37c6","revisions":{"1423290a81b86bbc681578c27199a308e6d784c6":{"kind":"REWORK","_number":1,"created":"2026-06-17 05:25:29.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/12/993712/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/993712/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/993712/1"}}},"commit":{"parents":[{"commit":"09e30c1ae48c9c2debc1abe762a056e96e392856","subject":"Move PIP_* env variables to ARG and config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/09e30c1ae48c9c2debc1abe762a056e96e392856"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-17 05:24:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-17 05:24:16.000000000","tz":120},"subject":"Add cleanup_repos to revert build_only overrides","message":"Add cleanup_repos to revert build_only overrides\n\nChange-Id: Ie5b9970a4575b6258dc238059d3f33ddfe155789\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1423290a81b86bbc681578c27199a308e6d784c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1423290a81b86bbc681578c27199a308e6d784c6"}]},"branch":"refs/heads/master"},"e36f642ee073361eb7a99dfc0999f024aebd5ed4":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-06-17 09:30:59.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/12/993712/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/993712/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/993712/2"}}},"commit":{"parents":[{"commit":"aa62b19125bf04478979730ea86ca597e2ac7108","subject":"Move PIP_* env variables to ARG and config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aa62b19125bf04478979730ea86ca597e2ac7108"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-17 05:24:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-17 09:30:59.000000000","tz":0},"subject":"Add cleanup_repos to revert build_only overrides","message":"Add cleanup_repos to revert build_only overrides\n\nChange-Id: Ie5b9970a4575b6258dc238059d3f33ddfe155789\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e36f642ee073361eb7a99dfc0999f024aebd5ed4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e36f642ee073361eb7a99dfc0999f024aebd5ed4"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"09e30c1ae48c9c2debc1abe762a056e96e392856","ours":"1423290a81b86bbc681578c27199a308e6d784c6","theirs":"aa62b19125bf04478979730ea86ca597e2ac7108","merge_strategy":"recursive","contains_conflicts":false}},"51ba4186e184478893473780e3051bf42d50200d":{"kind":"REWORK","_number":3,"created":"2026-06-17 16:16:40.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/12/993712/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/993712/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/993712/3"}}},"commit":{"parents":[{"commit":"aa62b19125bf04478979730ea86ca597e2ac7108","subject":"Move PIP_* env variables to ARG and config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aa62b19125bf04478979730ea86ca597e2ac7108"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-17 05:24:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-17 16:16:31.000000000","tz":120},"subject":"Add cleanup_repos to revert build_only overrides","message":"Add cleanup_repos to revert build_only overrides\n\nRepos marked build_only: true are backed up before handle_repos()\noverwrites them and restored at the end of each image build, so CI\nmirror URLs are not baked into the final image.\n\nThe cleanup RUN was injected into every Dockerfile, not just images\nthat called handle_repos() for those repos. Child images with no\nbackup fell through to \"rm -f\", deleting the inherited repo file\n(e.g. debian.sources in bifrost-base, breaking cpio install).\n\nFix by adding a {repo}.enabled sentinel that handle_repos creates\nwhen it processes a build_only repo, making the cleanup conditional\non that sentinel.\n\nChange-Id: Ie5b9970a4575b6258dc238059d3f33ddfe155789\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/51ba4186e184478893473780e3051bf42d50200d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/51ba4186e184478893473780e3051bf42d50200d"}]},"branch":"refs/heads/master"},"8fd5365a62ad0db43bad8873e7dd44e938576075":{"kind":"REWORK","_number":4,"created":"2026-06-17 19:23:36.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/12/993712/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/993712/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/993712/4"}}},"commit":{"parents":[{"commit":"aa62b19125bf04478979730ea86ca597e2ac7108","subject":"Move PIP_* env variables to ARG and config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aa62b19125bf04478979730ea86ca597e2ac7108"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-17 05:24:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-17 19:23:36.000000000","tz":0},"subject":"Add cleanup_repos to revert build_only overrides","message":"Add cleanup_repos to revert build_only overrides\n\nRepos marked build_only: true are backed up before handle_repos()\noverwrites them and restored at the end of each image build, so CI\nmirror URLs are not baked into the final image.\n\nThe cleanup RUN was injected into every Dockerfile, not just images\nthat called handle_repos() for those repos. Child images with no\nbackup fell through to \"rm -f\", deleting the inherited repo file\n(e.g. debian.sources in bifrost-base, breaking cpio install).\n\nFix by adding a {repo}.enabled sentinel that handle_repos creates\nwhen it processes a build_only repo, making the cleanup conditional\non that sentinel.\n\nChange-Id: Ie5b9970a4575b6258dc238059d3f33ddfe155789\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8fd5365a62ad0db43bad8873e7dd44e938576075"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8fd5365a62ad0db43bad8873e7dd44e938576075"}]},"branch":"refs/heads/master"},"429356cdcf2f14fc5c668e17c146d915ed2d9337":{"kind":"REWORK","_number":5,"created":"2026-06-22 06:19:13.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/12/993712/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/993712/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/993712/5"}}},"commit":{"parents":[{"commit":"aa62b19125bf04478979730ea86ca597e2ac7108","subject":"Move PIP_* env variables to ARG and config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aa62b19125bf04478979730ea86ca597e2ac7108"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-17 05:24:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-22 06:19:03.000000000","tz":120},"subject":"Add cleanup_repos to revert build_only overrides","message":"Add cleanup_repos to revert build_only overrides\n\nRepos marked build_only: true are backed up before handle_repos()\noverwrites them and restored at the end of each image build, so CI\nmirror URLs are not baked into the final image.\n\nThe cleanup RUN was injected into every Dockerfile, not just images\nthat called handle_repos() for those repos. Child images with no\nbackup fell through to \"rm -f\", deleting the inherited repo file\n(e.g. debian.sources in bifrost-base, breaking cpio install).\n\nFix by adding a {repo}.enabled sentinel that handle_repos creates\nwhen it processes a build_only repo, making the cleanup conditional\non that sentinel.\n\nChange-Id: Ie5b9970a4575b6258dc238059d3f33ddfe155789\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/429356cdcf2f14fc5c668e17c146d915ed2d9337"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/429356cdcf2f14fc5c668e17c146d915ed2d9337"}]},"branch":"refs/heads/master"},"0f5a25436452afbaae00d1168c0fda363e75354f":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-06-22 06:19:21.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/12/993712/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/993712/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/993712/6"}}},"commit":{"parents":[{"commit":"21211dd8ea3461796b86b996fef8656369fda21b","subject":"Move PIP_* env variables to ARG and config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/21211dd8ea3461796b86b996fef8656369fda21b"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-17 05:24:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-22 06:19:21.000000000","tz":0},"subject":"Add cleanup_repos to revert build_only overrides","message":"Add cleanup_repos to revert build_only overrides\n\nRepos marked build_only: true are backed up before handle_repos()\noverwrites them and restored at the end of each image build, so CI\nmirror URLs are not baked into the final image.\n\nThe cleanup RUN was injected into every Dockerfile, not just images\nthat called handle_repos() for those repos. Child images with no\nbackup fell through to \"rm -f\", deleting the inherited repo file\n(e.g. debian.sources in bifrost-base, breaking cpio install).\n\nFix by adding a {repo}.enabled sentinel that handle_repos creates\nwhen it processes a build_only repo, making the cleanup conditional\non that sentinel.\n\nChange-Id: Ie5b9970a4575b6258dc238059d3f33ddfe155789\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0f5a25436452afbaae00d1168c0fda363e75354f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0f5a25436452afbaae00d1168c0fda363e75354f"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"aa62b19125bf04478979730ea86ca597e2ac7108","ours":"429356cdcf2f14fc5c668e17c146d915ed2d9337","theirs":"21211dd8ea3461796b86b996fef8656369fda21b","merge_strategy":"recursive","contains_conflicts":false}},"f2dcaac7c55fa6a33fd62b6b025545de59fd12ae":{"kind":"REWORK","_number":7,"created":"2026-06-22 06:55:44.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/12/993712/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/993712/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/993712/7"}}},"commit":{"parents":[{"commit":"5027943a9cad4c90a721d43c0d136cc437598521","subject":"Move PIP_* env variables to ARG and config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5027943a9cad4c90a721d43c0d136cc437598521"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-17 05:24:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-22 06:55:44.000000000","tz":0},"subject":"Add cleanup_repos to revert build_only overrides","message":"Add cleanup_repos to revert build_only overrides\n\nRepos marked build_only: true are backed up before handle_repos()\noverwrites them and restored at the end of each image build, so CI\nmirror URLs are not baked into the final image.\n\nThe cleanup RUN was injected into every Dockerfile, not just images\nthat called handle_repos() for those repos. Child images with no\nbackup fell through to \"rm -f\", deleting the inherited repo file\n(e.g. debian.sources in bifrost-base, breaking cpio install).\n\nFix by adding a {repo}.enabled sentinel that handle_repos creates\nwhen it processes a build_only repo, making the cleanup conditional\non that sentinel.\n\nChange-Id: Ie5b9970a4575b6258dc238059d3f33ddfe155789\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f2dcaac7c55fa6a33fd62b6b025545de59fd12ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f2dcaac7c55fa6a33fd62b6b025545de59fd12ae"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"21211dd8ea3461796b86b996fef8656369fda21b","ours":"0f5a25436452afbaae00d1168c0fda363e75354f","theirs":"5027943a9cad4c90a721d43c0d136cc437598521","merge_strategy":"recursive","contains_conflicts":true}},"88722dfd63d68e687e0cde2422c2c2fe9c3c4818":{"kind":"REWORK","_number":8,"created":"2026-06-22 06:59:39.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/12/993712/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/993712/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/993712/8"}}},"commit":{"parents":[{"commit":"5027943a9cad4c90a721d43c0d136cc437598521","subject":"Move PIP_* env variables to ARG and config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5027943a9cad4c90a721d43c0d136cc437598521"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-17 05:24:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-22 06:59:29.000000000","tz":120},"subject":"Add cleanup_repos to revert build_only overrides","message":"Add cleanup_repos to revert build_only overrides\n\nRepos marked build_only: true are backed up before handle_repos()\noverwrites them and restored at the end of each image build, so CI\nmirror URLs are not baked into the final image.\n\nThe cleanup RUN was injected into every Dockerfile, not just images\nthat called handle_repos() for those repos. Child images with no\nbackup fell through to \"rm -f\", deleting the inherited repo file\n(e.g. debian.sources in bifrost-base, breaking cpio install).\n\nFix by adding a {repo}.enabled sentinel that handle_repos creates\nwhen it processes a build_only repo, making the cleanup conditional\non that sentinel.\n\nChange-Id: Ie5b9970a4575b6258dc238059d3f33ddfe155789\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/88722dfd63d68e687e0cde2422c2c2fe9c3c4818"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/88722dfd63d68e687e0cde2422c2c2fe9c3c4818"}]},"branch":"refs/heads/master"},"8c3d9e52f229853d88c0a5b53ead311ad5eedae3":{"kind":"REWORK","_number":9,"created":"2026-06-24 09:28:53.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/12/993712/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/993712/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/993712/9"}}},"commit":{"parents":[{"commit":"f4f8a448725944528a9da9bbcd5de12e7bcc83f2","subject":"Move PIP_* env variables to ARG and config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f4f8a448725944528a9da9bbcd5de12e7bcc83f2"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-17 05:24:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-24 09:28:53.000000000","tz":0},"subject":"Add cleanup_repos to revert build_only overrides","message":"Add cleanup_repos to revert build_only overrides\n\nRepos marked build_only: true are backed up before handle_repos()\noverwrites them and restored at the end of each image build, so CI\nmirror URLs are not baked into the final image.\n\nThe cleanup RUN was injected into every Dockerfile, not just images\nthat called handle_repos() for those repos. Child images with no\nbackup fell through to \"rm -f\", deleting the inherited repo file\n(e.g. debian.sources in bifrost-base, breaking cpio install).\n\nFix by adding a {repo}.enabled sentinel that handle_repos creates\nwhen it processes a build_only repo, making the cleanup conditional\non that sentinel.\n\nChange-Id: Ie5b9970a4575b6258dc238059d3f33ddfe155789\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8c3d9e52f229853d88c0a5b53ead311ad5eedae3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8c3d9e52f229853d88c0a5b53ead311ad5eedae3"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"5027943a9cad4c90a721d43c0d136cc437598521","ours":"88722dfd63d68e687e0cde2422c2c2fe9c3c4818","theirs":"f4f8a448725944528a9da9bbcd5de12e7bcc83f2","merge_strategy":"recursive","contains_conflicts":true}},"a01c7e547e24726d568e4412db190a5ebf9d1b9f":{"kind":"REWORK","_number":10,"created":"2026-06-24 09:29:27.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/12/993712/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/993712/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/993712/10"}}},"commit":{"parents":[{"commit":"f4f8a448725944528a9da9bbcd5de12e7bcc83f2","subject":"Move PIP_* env variables to ARG and config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f4f8a448725944528a9da9bbcd5de12e7bcc83f2"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-17 05:24:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-24 09:29:27.000000000","tz":0},"subject":"Add cleanup_repos to revert build_only overrides","message":"Add cleanup_repos to revert build_only overrides\n\nRepos marked build_only: true are backed up before handle_repos()\noverwrites them and restored at the end of each image build, so CI\nmirror URLs are not baked into the final image.\n\nThe cleanup RUN was injected into every Dockerfile, not just images\nthat called handle_repos() for those repos. Child images with no\nbackup fell through to \"rm -f\", deleting the inherited repo file\n(e.g. debian.sources in bifrost-base, breaking cpio install).\n\nFix by adding a {repo}.enabled sentinel that handle_repos creates\nwhen it processes a build_only repo, making the cleanup conditional\non that sentinel.\n\nChange-Id: Ie5b9970a4575b6258dc238059d3f33ddfe155789\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a01c7e547e24726d568e4412db190a5ebf9d1b9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a01c7e547e24726d568e4412db190a5ebf9d1b9f"}]},"branch":"refs/heads/master"},"2421a9be6379c0e6ca6ff0a0e4a525e1a4b8e634":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2026-06-25 15:15:04.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/12/993712/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/993712/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/993712/11"}}},"commit":{"parents":[{"commit":"544418220e9fae6f8b9399b56bd905cd015f8da0","subject":"Move PIP_* env variables to ARG and config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/544418220e9fae6f8b9399b56bd905cd015f8da0"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-17 05:24:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-25 15:15:04.000000000","tz":0},"subject":"Add cleanup_repos to revert build_only overrides","message":"Add cleanup_repos to revert build_only overrides\n\nRepos marked build_only: true are backed up before handle_repos()\noverwrites them and restored at the end of each image build, so CI\nmirror URLs are not baked into the final image.\n\nThe cleanup RUN was injected into every Dockerfile, not just images\nthat called handle_repos() for those repos. Child images with no\nbackup fell through to \"rm -f\", deleting the inherited repo file\n(e.g. debian.sources in bifrost-base, breaking cpio install).\n\nFix by adding a {repo}.enabled sentinel that handle_repos creates\nwhen it processes a build_only repo, making the cleanup conditional\non that sentinel.\n\nChange-Id: Ie5b9970a4575b6258dc238059d3f33ddfe155789\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2421a9be6379c0e6ca6ff0a0e4a525e1a4b8e634"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2421a9be6379c0e6ca6ff0a0e4a525e1a4b8e634"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"f4f8a448725944528a9da9bbcd5de12e7bcc83f2","ours":"a01c7e547e24726d568e4412db190a5ebf9d1b9f","theirs":"544418220e9fae6f8b9399b56bd905cd015f8da0","merge_strategy":"recursive","contains_conflicts":false}},"c3071cc3b02aefcc0b09aea53c6d722f3da240d8":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2026-06-26 06:03:59.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/12/993712/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/993712/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/993712/12"}}},"commit":{"parents":[{"commit":"8139e42ce5f5567329627844d45499e891074806","subject":"Move PIP_* env variables to ARG and config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8139e42ce5f5567329627844d45499e891074806"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-17 05:24:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-26 06:03:59.000000000","tz":0},"subject":"Add cleanup_repos to revert build_only overrides","message":"Add cleanup_repos to revert build_only overrides\n\nRepos marked build_only: true are backed up before handle_repos()\noverwrites them and restored at the end of each image build, so CI\nmirror URLs are not baked into the final image.\n\nThe cleanup RUN was injected into every Dockerfile, not just images\nthat called handle_repos() for those repos. Child images with no\nbackup fell through to \"rm -f\", deleting the inherited repo file\n(e.g. debian.sources in bifrost-base, breaking cpio install).\n\nFix by adding a {repo}.enabled sentinel that handle_repos creates\nwhen it processes a build_only repo, making the cleanup conditional\non that sentinel.\n\nChange-Id: Ie5b9970a4575b6258dc238059d3f33ddfe155789\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c3071cc3b02aefcc0b09aea53c6d722f3da240d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c3071cc3b02aefcc0b09aea53c6d722f3da240d8"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"544418220e9fae6f8b9399b56bd905cd015f8da0","ours":"2421a9be6379c0e6ca6ff0a0e4a525e1a4b8e634","theirs":"8139e42ce5f5567329627844d45499e891074806","merge_strategy":"recursive","contains_conflicts":false}},"103f8911e276e3ddc6f2960016960284ebd4cae9":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2026-07-13 05:41:27.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/12/993712/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/993712/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/993712/13"}}},"commit":{"parents":[{"commit":"62164cb688191b4ad4efe0a9adf2e1816183346c","subject":"Move PIP_* env variables to ARG and config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/62164cb688191b4ad4efe0a9adf2e1816183346c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-17 05:24:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-13 05:41:27.000000000","tz":0},"subject":"Add cleanup_repos to revert build_only overrides","message":"Add cleanup_repos to revert build_only overrides\n\nRepos marked build_only: true are backed up before handle_repos()\noverwrites them and restored at the end of each image build, so CI\nmirror URLs are not baked into the final image.\n\nThe cleanup RUN was injected into every Dockerfile, not just images\nthat called handle_repos() for those repos. Child images with no\nbackup fell through to \"rm -f\", deleting the inherited repo file\n(e.g. debian.sources in bifrost-base, breaking cpio install).\n\nFix by adding a {repo}.enabled sentinel that handle_repos creates\nwhen it processes a build_only repo, making the cleanup conditional\non that sentinel.\n\nChange-Id: Ie5b9970a4575b6258dc238059d3f33ddfe155789\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/103f8911e276e3ddc6f2960016960284ebd4cae9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/103f8911e276e3ddc6f2960016960284ebd4cae9"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"8139e42ce5f5567329627844d45499e891074806","ours":"c3071cc3b02aefcc0b09aea53c6d722f3da240d8","theirs":"62164cb688191b4ad4efe0a9adf2e1816183346c","merge_strategy":"recursive","contains_conflicts":false}},"56fbbeba653b74823152cf2457e0902eedab6d31":{"kind":"REWORK","_number":14,"created":"2026-07-13 06:17:12.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/12/993712/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/993712/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/993712/14"}}},"commit":{"parents":[{"commit":"62164cb688191b4ad4efe0a9adf2e1816183346c","subject":"Move PIP_* env variables to ARG and config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/62164cb688191b4ad4efe0a9adf2e1816183346c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-17 05:24:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-13 06:16:33.000000000","tz":120},"subject":"Add cleanup_repos to revert build_only overrides","message":"Add cleanup_repos to revert build_only overrides\n\nRepos marked build_only: true are backed up before handle_repos()\noverwrites them and restored at the end of each image build, so CI\nmirror URLs are not baked into the final image.\n\nThe cleanup RUN was injected into every Dockerfile, not just images\nthat called handle_repos() for those repos. Child images with no\nbackup fell through to \"rm -f\", deleting the inherited repo file\n(e.g. debian.sources in bifrost-base, breaking cpio install).\n\nFix by adding a {repo}.enabled sentinel that handle_repos creates\nwhen it processes a build_only repo, making the cleanup conditional\non that sentinel.\n\nChange-Id: Ie5b9970a4575b6258dc238059d3f33ddfe155789\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/56fbbeba653b74823152cf2457e0902eedab6d31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/56fbbeba653b74823152cf2457e0902eedab6d31"}]},"branch":"refs/heads/master"},"84583995d1fc4070804eb76c39f77880020023fc":{"kind":"REWORK","_number":15,"created":"2026-07-13 07:12:13.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/12/993712/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/993712/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/993712/15"}}},"commit":{"parents":[{"commit":"62164cb688191b4ad4efe0a9adf2e1816183346c","subject":"Move PIP_* env variables to ARG and config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/62164cb688191b4ad4efe0a9adf2e1816183346c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-17 05:24:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-13 07:12:03.000000000","tz":120},"subject":"Add cleanup_repos to revert build_only overrides","message":"Add cleanup_repos to revert build_only overrides\n\nRepos marked build_only: true are backed up before handle_repos()\noverwrites them and restored at the end of each image build, so CI\nmirror URLs are not baked into the final image.\n\nThe cleanup RUN was injected into every Dockerfile, not just images\nthat called handle_repos() for those repos. Child images with no\nbackup fell through to \"rm -f\", deleting the inherited repo file\n(e.g. debian.sources in bifrost-base, breaking cpio install).\n\nFix by adding a {repo}.enabled sentinel that handle_repos creates\nwhen it processes a build_only repo, making the cleanup conditional\non that sentinel.\n\nChange-Id: Ie5b9970a4575b6258dc238059d3f33ddfe155789\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/84583995d1fc4070804eb76c39f77880020023fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/84583995d1fc4070804eb76c39f77880020023fc"}]},"branch":"refs/heads/master"},"6689f208c89765c923bd0adc08448e73586256eb":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2026-07-14 05:21:17.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/12/993712/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/993712/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/993712/16"}}},"commit":{"parents":[{"commit":"ce5510218093d276ac3a1207427f9d08219e8b88","subject":"Move PIP_* env variables to ARG and config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ce5510218093d276ac3a1207427f9d08219e8b88"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-17 05:24:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-14 05:21:17.000000000","tz":0},"subject":"Add cleanup_repos to revert build_only overrides","message":"Add cleanup_repos to revert build_only overrides\n\nRepos marked build_only: true are backed up before handle_repos()\noverwrites them and restored at the end of each image build, so CI\nmirror URLs are not baked into the final image.\n\nThe cleanup RUN was injected into every Dockerfile, not just images\nthat called handle_repos() for those repos. Child images with no\nbackup fell through to \"rm -f\", deleting the inherited repo file\n(e.g. debian.sources in bifrost-base, breaking cpio install).\n\nFix by adding a {repo}.enabled sentinel that handle_repos creates\nwhen it processes a build_only repo, making the cleanup conditional\non that sentinel.\n\nChange-Id: Ie5b9970a4575b6258dc238059d3f33ddfe155789\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6689f208c89765c923bd0adc08448e73586256eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6689f208c89765c923bd0adc08448e73586256eb"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"62164cb688191b4ad4efe0a9adf2e1816183346c","ours":"84583995d1fc4070804eb76c39f77880020023fc","theirs":"ce5510218093d276ac3a1207427f9d08219e8b88","merge_strategy":"recursive","contains_conflicts":false}},"6ecf7ee85615b1bc28de496f97bbc9b3cfb102dc":{"kind":"REWORK","_number":17,"created":"2026-07-14 08:56:04.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/12/993712/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/993712/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/993712/17"}}},"commit":{"parents":[{"commit":"ce5510218093d276ac3a1207427f9d08219e8b88","subject":"Move PIP_* env variables to ARG and config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ce5510218093d276ac3a1207427f9d08219e8b88"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-17 05:24:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-14 08:56:04.000000000","tz":0},"subject":"Add cleanup_repos to revert build_only overrides","message":"Add cleanup_repos to revert build_only overrides\n\nRepos marked build_only: true are backed up before handle_repos()\noverwrites them and restored at the end of each image build, so CI\nmirror URLs are not baked into the final image.\n\nThe cleanup RUN was injected into every Dockerfile, not just images\nthat called handle_repos() for those repos. Child images with no\nbackup fell through to \"rm -f\", deleting the inherited repo file\n(e.g. debian.sources in bifrost-base, breaking cpio install).\n\nFix by adding a {repo}.enabled sentinel that handle_repos creates\nwhen it processes a build_only repo, making the cleanup conditional\non that sentinel.\n\nChange-Id: Ie5b9970a4575b6258dc238059d3f33ddfe155789\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6ecf7ee85615b1bc28de496f97bbc9b3cfb102dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6ecf7ee85615b1bc28de496f97bbc9b3cfb102dc"}]},"branch":"refs/heads/master"},"7ef51cf6e7217255b38b9ca9d4d87144ba72e302":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2026-07-17 08:23:00.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/12/993712/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/993712/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/993712/18"}}},"commit":{"parents":[{"commit":"a1f942a55c235465c0309d1a32ee7974c7fb0a60","subject":"Move PIP_* env variables to ARG and config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a1f942a55c235465c0309d1a32ee7974c7fb0a60"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-17 05:24:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-17 08:23:00.000000000","tz":0},"subject":"Add cleanup_repos to revert build_only overrides","message":"Add cleanup_repos to revert build_only overrides\n\nRepos marked build_only: true are backed up before handle_repos()\noverwrites them and restored at the end of each image build, so CI\nmirror URLs are not baked into the final image.\n\nThe cleanup RUN was injected into every Dockerfile, not just images\nthat called handle_repos() for those repos. Child images with no\nbackup fell through to \"rm -f\", deleting the inherited repo file\n(e.g. debian.sources in bifrost-base, breaking cpio install).\n\nFix by adding a {repo}.enabled sentinel that handle_repos creates\nwhen it processes a build_only repo, making the cleanup conditional\non that sentinel.\n\nChange-Id: Ie5b9970a4575b6258dc238059d3f33ddfe155789\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7ef51cf6e7217255b38b9ca9d4d87144ba72e302"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7ef51cf6e7217255b38b9ca9d4d87144ba72e302"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"ce5510218093d276ac3a1207427f9d08219e8b88","ours":"6ecf7ee85615b1bc28de496f97bbc9b3cfb102dc","theirs":"a1f942a55c235465c0309d1a32ee7974c7fb0a60","merge_strategy":"recursive","contains_conflicts":false}},"fc138e92ccf6f4172f7dbf07b5264295574965bd":{"kind":"REWORK","_number":19,"created":"2026-07-17 11:03:35.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/12/993712/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/993712/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/993712/19"}}},"commit":{"parents":[{"commit":"a1f942a55c235465c0309d1a32ee7974c7fb0a60","subject":"Move PIP_* env variables to ARG and config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a1f942a55c235465c0309d1a32ee7974c7fb0a60"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-17 05:24:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-17 11:03:34.000000000","tz":0},"subject":"Add cleanup_repos to revert build_only overrides","message":"Add cleanup_repos to revert build_only overrides\n\nRepos marked build_only: true are backed up before handle_repos()\noverwrites them and restored at the end of each image build, so CI\nmirror URLs are not baked into the final image.\n\nThe cleanup RUN was injected into every Dockerfile, not just images\nthat called handle_repos() for those repos. Child images with no\nbackup fell through to \"rm -f\", deleting the inherited repo file\n(e.g. debian.sources in bifrost-base, breaking cpio install).\n\nFix by adding a {repo}.enabled sentinel that handle_repos creates\nwhen it processes a build_only repo, making the cleanup conditional\non that sentinel.\n\nChange-Id: Ie5b9970a4575b6258dc238059d3f33ddfe155789\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fc138e92ccf6f4172f7dbf07b5264295574965bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fc138e92ccf6f4172f7dbf07b5264295574965bd"}]},"branch":"refs/heads/master"},"29678bb0909118bae292af9b61f8b1704f9a5f35":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2026-07-20 06:24:45.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/12/993712/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/993712/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/993712/20"}}},"commit":{"parents":[{"commit":"dd8ae3b90e3c111375f3dfbfcd9e95a75dede6c9","subject":"Move PIP_* env variables to ARG and config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dd8ae3b90e3c111375f3dfbfcd9e95a75dede6c9"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-17 05:24:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-20 06:24:45.000000000","tz":0},"subject":"Add cleanup_repos to revert build_only overrides","message":"Add cleanup_repos to revert build_only overrides\n\nRepos marked build_only: true are backed up before handle_repos()\noverwrites them and restored at the end of each image build, so CI\nmirror URLs are not baked into the final image.\n\nThe cleanup RUN was injected into every Dockerfile, not just images\nthat called handle_repos() for those repos. Child images with no\nbackup fell through to \"rm -f\", deleting the inherited repo file\n(e.g. debian.sources in bifrost-base, breaking cpio install).\n\nFix by adding a {repo}.enabled sentinel that handle_repos creates\nwhen it processes a build_only repo, making the cleanup conditional\non that sentinel.\n\nChange-Id: Ie5b9970a4575b6258dc238059d3f33ddfe155789\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/29678bb0909118bae292af9b61f8b1704f9a5f35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/29678bb0909118bae292af9b61f8b1704f9a5f35"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"a1f942a55c235465c0309d1a32ee7974c7fb0a60","ours":"fc138e92ccf6f4172f7dbf07b5264295574965bd","theirs":"dd8ae3b90e3c111375f3dfbfcd9e95a75dede6c9","merge_strategy":"recursive","contains_conflicts":false}},"5718fe567ca60544f77e0db179efbe41ba451be3":{"kind":"REWORK","_number":21,"created":"2026-07-20 09:28:24.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/12/993712/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/993712/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/993712/21"}}},"commit":{"parents":[{"commit":"dd8ae3b90e3c111375f3dfbfcd9e95a75dede6c9","subject":"Move PIP_* env variables to ARG and config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dd8ae3b90e3c111375f3dfbfcd9e95a75dede6c9"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-17 05:24:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-20 09:28:24.000000000","tz":0},"subject":"Add cleanup_repos to revert build_only overrides","message":"Add cleanup_repos to revert build_only overrides\n\nRepos marked build_only: true are backed up before handle_repos()\noverwrites them and restored at the end of each image build, so CI\nmirror URLs are not baked into the final image.\n\nThe cleanup RUN was injected into every Dockerfile, not just images\nthat called handle_repos() for those repos. Child images with no\nbackup fell through to \"rm -f\", deleting the inherited repo file\n(e.g. debian.sources in bifrost-base, breaking cpio install).\n\nFix by adding a {repo}.enabled sentinel that handle_repos creates\nwhen it processes a build_only repo, making the cleanup conditional\non that sentinel.\n\nChange-Id: Ie5b9970a4575b6258dc238059d3f33ddfe155789\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5718fe567ca60544f77e0db179efbe41ba451be3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5718fe567ca60544f77e0db179efbe41ba451be3"}]},"branch":"refs/heads/master"},"7fac0d392951c34bc9c0a2277d22952899a319cd":{"kind":"REWORK","_number":22,"created":"2026-07-20 11:54:49.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/12/993712/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/993712/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/993712/22"}}},"commit":{"parents":[{"commit":"dd8ae3b90e3c111375f3dfbfcd9e95a75dede6c9","subject":"Move PIP_* env variables to ARG and config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dd8ae3b90e3c111375f3dfbfcd9e95a75dede6c9"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-17 05:24:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-20 11:54:49.000000000","tz":0},"subject":"Add cleanup_repos to revert build_only overrides","message":"Add cleanup_repos to revert build_only overrides\n\nRepos marked build_only: true are backed up before handle_repos()\noverwrites them and restored at the end of each image build, so CI\nmirror URLs are not baked into the final image.\n\nThe cleanup RUN was injected into every Dockerfile, not just images\nthat called handle_repos() for those repos. Child images with no\nbackup fell through to \"rm -f\", deleting the inherited repo file\n(e.g. debian.sources in bifrost-base, breaking cpio install).\n\nFix by adding a {repo}.enabled sentinel that handle_repos creates\nwhen it processes a build_only repo, making the cleanup conditional\non that sentinel.\n\nChange-Id: Ie5b9970a4575b6258dc238059d3f33ddfe155789\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7fac0d392951c34bc9c0a2277d22952899a319cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7fac0d392951c34bc9c0a2277d22952899a319cd"}]},"branch":"refs/heads/master"},"99261e0360c04f9f09a8de2c507fa403dd70ee38":{"kind":"TRIVIAL_REBASE","_number":23,"created":"2026-07-24 08:13:06.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/12/993712/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/993712/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/993712/23"}}},"commit":{"parents":[{"commit":"2cde168448dff4be91d4441562a116df3b109889","subject":"Move PIP_* env variables to ARG and config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2cde168448dff4be91d4441562a116df3b109889"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-17 05:24:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-24 08:13:06.000000000","tz":0},"subject":"Add cleanup_repos to revert build_only overrides","message":"Add cleanup_repos to revert build_only overrides\n\nRepos marked build_only: true are backed up before handle_repos()\noverwrites them and restored at the end of each image build, so CI\nmirror URLs are not baked into the final image.\n\nThe cleanup RUN was injected into every Dockerfile, not just images\nthat called handle_repos() for those repos. Child images with no\nbackup fell through to \"rm -f\", deleting the inherited repo file\n(e.g. debian.sources in bifrost-base, breaking cpio install).\n\nFix by adding a {repo}.enabled sentinel that handle_repos creates\nwhen it processes a build_only repo, making the cleanup conditional\non that sentinel.\n\nChange-Id: Ie5b9970a4575b6258dc238059d3f33ddfe155789\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/99261e0360c04f9f09a8de2c507fa403dd70ee38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/99261e0360c04f9f09a8de2c507fa403dd70ee38"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"dd8ae3b90e3c111375f3dfbfcd9e95a75dede6c9","ours":"7fac0d392951c34bc9c0a2277d22952899a319cd","theirs":"2cde168448dff4be91d4441562a116df3b109889","merge_strategy":"recursive","contains_conflicts":false}},"06a1e8203d1df041f89cf227fb6dc2fa44190d27":{"kind":"REWORK","_number":24,"created":"2026-07-27 14:21:23.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/12/993712/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/993712/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/993712/24"}}},"commit":{"parents":[{"commit":"2cde168448dff4be91d4441562a116df3b109889","subject":"Move PIP_* env variables to ARG and config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2cde168448dff4be91d4441562a116df3b109889"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-17 05:24:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-27 14:21:23.000000000","tz":0},"subject":"Add cleanup_repos to revert build_only overrides","message":"Add cleanup_repos to revert build_only overrides\n\nRepos marked build_only: true are backed up before handle_repos()\noverwrites them and restored at the end of each image build, so CI\nmirror URLs are not baked into the final image.\n\nThe cleanup RUN was injected into every Dockerfile, not just images\nthat called handle_repos() for those repos. Child images with no\nbackup fell through to \"rm -f\", deleting the inherited repo file\n(e.g. debian.sources in bifrost-base, breaking cpio install).\n\nFix by adding a {repo}.enabled sentinel that handle_repos creates\nwhen it processes a build_only repo, making the cleanup conditional\non that sentinel.\n\nChange-Id: Ie5b9970a4575b6258dc238059d3f33ddfe155789\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/06a1e8203d1df041f89cf227fb6dc2fa44190d27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/06a1e8203d1df041f89cf227fb6dc2fa44190d27"}]},"branch":"refs/heads/master"},"d7aa592e4dc6beda7f88066ba8ab0b141ac88959":{"kind":"TRIVIAL_REBASE","_number":25,"created":"2026-07-28 07:22:38.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/12/993712/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/993712/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/993712/25"}}},"commit":{"parents":[{"commit":"de3e2aacc49a02f57bef650a756cb5d37937f561","subject":"Move PIP_* env variables to ARG and config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/de3e2aacc49a02f57bef650a756cb5d37937f561"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-17 05:24:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-28 07:22:38.000000000","tz":0},"subject":"Add cleanup_repos to revert build_only overrides","message":"Add cleanup_repos to revert build_only overrides\n\nRepos marked build_only: true are backed up before handle_repos()\noverwrites them and restored at the end of each image build, so CI\nmirror URLs are not baked into the final image.\n\nThe cleanup RUN was injected into every Dockerfile, not just images\nthat called handle_repos() for those repos. Child images with no\nbackup fell through to \"rm -f\", deleting the inherited repo file\n(e.g. debian.sources in bifrost-base, breaking cpio install).\n\nFix by adding a {repo}.enabled sentinel that handle_repos creates\nwhen it processes a build_only repo, making the cleanup conditional\non that sentinel.\n\nChange-Id: Ie5b9970a4575b6258dc238059d3f33ddfe155789\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d7aa592e4dc6beda7f88066ba8ab0b141ac88959"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d7aa592e4dc6beda7f88066ba8ab0b141ac88959"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"2cde168448dff4be91d4441562a116df3b109889","ours":"06a1e8203d1df041f89cf227fb6dc2fa44190d27","theirs":"de3e2aacc49a02f57bef650a756cb5d37937f561","merge_strategy":"recursive","contains_conflicts":false}},"0638e95d9ea8a13b22c9353a96445fc37345dc42":{"kind":"REWORK","_number":26,"created":"2026-07-31 17:19:50.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/12/993712/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/993712/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/993712/26"}}},"commit":{"parents":[{"commit":"8e912a9db7f013a1abfc1e70a829e435d50622db","subject":"Merge \"ironic-pxe: don\u0027t fail if symlinks already exist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8e912a9db7f013a1abfc1e70a829e435d50622db"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-17 05:24:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-31 17:19:39.000000000","tz":120},"subject":"Add cleanup_repos to revert build_only overrides","message":"Add cleanup_repos to revert build_only overrides\n\nRepos marked build_only: true are backed up before handle_repos()\noverwrites them and restored at the end of each image build, so CI\nmirror URLs are not baked into the final image.\n\nThe cleanup RUN was injected into every Dockerfile, not just images\nthat called handle_repos() for those repos. Child images with no\nbackup fell through to \"rm -f\", deleting the inherited repo file\n(e.g. debian.sources in bifrost-base, breaking cpio install).\n\nFix by adding a {repo}.enabled sentinel that handle_repos creates\nwhen it processes a build_only repo, making the cleanup conditional\non that sentinel.\n\nChange-Id: Ie5b9970a4575b6258dc238059d3f33ddfe155789\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0638e95d9ea8a13b22c9353a96445fc37345dc42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0638e95d9ea8a13b22c9353a96445fc37345dc42"}]},"branch":"refs/heads/master"},"5aabec47f057d9cb32aab520d666e51bd9ee37c6":{"kind":"REWORK","_number":27,"created":"2026-08-05 08:13:23.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/12/993712/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/993712/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/993712/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/993712/27"}}},"commit":{"parents":[{"commit":"8e912a9db7f013a1abfc1e70a829e435d50622db","subject":"Merge \"ironic-pxe: don\u0027t fail if symlinks already exist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8e912a9db7f013a1abfc1e70a829e435d50622db"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-06-17 05:24:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-08-05 08:13:09.000000000","tz":120},"subject":"Add cleanup_repos to revert build_only overrides","message":"Add cleanup_repos to revert build_only overrides\n\nRepos marked build_only: true are backed up before handle_repos()\noverwrites them and restored at the end of each image build, so CI\nmirror URLs are not baked into the final image.\n\nThe cleanup RUN was injected into every Dockerfile, not just images\nthat called handle_repos() for those repos. Child images with no\nbackup fell through to \"rm -f\", deleting the inherited repo file\n(e.g. debian.sources in bifrost-base, breaking cpio install).\n\nFix by adding a {repo}.enabled sentinel that handle_repos creates\nwhen it processes a build_only repo, making the cleanup conditional\non that sentinel.\n\nChange-Id: Ie5b9970a4575b6258dc238059d3f33ddfe155789\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5aabec47f057d9cb32aab520d666e51bd9ee37c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5aabec47f057d9cb32aab520d666e51bd9ee37c6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-1"],"failing_atoms":["label:Backport-Candidate\u003d1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
