)]}'
{"id":"openstack%2Fkolla~993788","triplet_id":"openstack%2Fkolla~master~I78f89dd7a4379f842153feb476f5bee824c12ffc","project":"openstack/kolla","branch":"master","topic":"kolla-yaml-config","attention_set":{"37306":{"account":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"last_update":"2026-07-17 09:08:03.000000000","reason":"\u003cGERRIT_ACCOUNT_27339\u003e replied on the change","reason_account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}}},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-06-17 20:24:35.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}}},"hashtags":[],"change_id":"I78f89dd7a4379f842153feb476f5bee824c12ffc","subject":"Migrate internal state storage to YAML","status":"NEW","created":"2026-06-17 16:02:56.000000000","updated":"2026-07-17 09:08:03.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":3,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"b0dd34c3f3ca5f2dcc09350234001d1df1784d6a","_number":993788,"virtual_id_number":993788,"owner":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-07-10 19:53:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":-1,"date":"2026-07-17 09:08:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-17 16:03:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-17 20:21:55.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-07-17 09:08:03.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"}],"messages":[{"id":"e9843c30716221ec774df70fcf74a9e4441b5477","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2026-06-17 16:02:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cb23eaf6bf6d467a14e762989496cd982263c8b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 16:03:09.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla for 993788,1","accounts_in_message":[],"_revision_number":1},{"id":"7826fafaa4cafd4189e8d21bb95545f43f292e49","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 16:03:12.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla for 993788,1","accounts_in_message":[],"_revision_number":1},{"id":"30db00b6f739218e55ef0e37ed88c938d2946351","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2026-06-17 18:46:41.000000000","message":"Topic set to kolla-yaml-config","accounts_in_message":[],"_revision_number":1},{"id":"3f5d66d56fb0871e1099e242b24226c451dadfe2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2026-06-17 19:05:46.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"d7fc8e60055125ca2df309284dbc3d5bba1f4443","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-06-17 20:24:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"855dd392a3c00a03fa5edcfd04a958adad87a681","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 22:04:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72e976acc7234e749538f0c025dac0aa\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/25d6e3366dc04e7b94325a95959a7b72 : SUCCESS in 1h 01m 30s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/042c95a552a04a5992fc6add0e9eaf4d : SUCCESS in 1h 19m 31s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/34f30da52ae147d0aef01b7e078de63d : SUCCESS in 24m 50s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/e86aa4367ecd43aa98bb77ff208430da : SUCCESS in 1h 11m 12s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/acf19ae7ef6d46a7b30c7cc18f98950c : SUCCESS in 56m 58s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/07fe66363cca4442b1bdaf21a1af16f1 : SUCCESS in 1h 30m 10s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/192f539726824c2d9307e408d04c7efa : SUCCESS in 35m 09s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/debaa9cb846441ada9327e16f71cdc53 : SUCCESS in 1h 17m 44s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/b8252c318cff4bbdadef4ea146970dd4 : SUCCESS in 1h 21m 16s\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/8d7e3c9a0e09472aa1f0f1ce5c4d6743 : SUCCESS in 2h 02m 38s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/d4bfdb13f9e14de19264b7fa0f37a6f0 : SUCCESS in 1h 20m 46s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/ef5272d5107242cc97f59c1eb1044a63 : SUCCESS in 1h 50m 55s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/219e4544910040259396a06036500864 : SUCCESS in 1h 23m 13s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/a7fd85c9e44d4eac88d41fd3b31e4861 : SUCCESS in 2h 03m 57s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/d758795b16054e3ba540089aebbb8093 : FAILURE in 1h 30m 08s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/b5447611510940199a1de369022852b9 : SUCCESS in 1h 51m 36s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/5b24461f8b7b4398b1106391715d22cb : SUCCESS in 2h 38m 48s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/ca2d8da897554381be32114f0667acb1 : SUCCESS in 2h 02m 41s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/418bf32476584596853e38d10f6db3c2 : SUCCESS in 2h 56m 19s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/5600e0eca8654af2a0e32a5485ebb7fe : FAILURE in 1h 07m 13s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/f31f9f581f1f41a6b8aa0295787f68da : FAILURE in 1h 02m 03s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/7e6a39aa422f43fbaeb32a8515382a7b : FAILURE in 1h 10m 23s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/5e50aa866143448fae31006eb88dff4d : SUCCESS in 27m 32s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/952af64d70c3414ca506fbd91bc68c35 : SUCCESS in 26m 41s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/bb004e2365ba4214a9698bf4fd993878 : SUCCESS in 18m 22s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/96f10524504349ecb4ca345f341e28b9 : FAILURE in 1h 12m 19s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/a7859476804a42cdb6bcd915baf57e60 : SUCCESS in 1h 19m 01s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/b28d0f9bed9e406396eeb49214a3c5f5 : SUCCESS in 1h 32m 28s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/78255dc60c67417fb0d2803cf17d77f0 : SUCCESS in 1h 26m 05s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/bd5d670c96c84d3ba21d5db298da2df5 : SUCCESS in 1h 27m 50s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/ad000773af4c439495993806ebad1b3a : FAILURE in 1h 43m 30s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/261a271632044918a89cdaf054d57461 : FAILURE in 1h 06m 30s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/d2551f316f3647219d612925610c52c2 : FAILURE in 1h 43m 19s (non-voting)\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/79e3d31f6470467f815a12800cd74929 : FAILURE in 1h 39m 13s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/6d93c46a4cc745468ee726eb45a6353a : SUCCESS in 1h 36m 52s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/fdf16279e57c47b3abb4e440fe6a86dd : SUCCESS in 1h 47m 08s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/f001ec70a96941c6bb734718ed937960 : SUCCESS in 3m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc1c08ad59fa4799b701ffeabf21f7f9 : SUCCESS in 3m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3d59c3396a6b4c57abc0d44cae925c33 : SUCCESS in 3m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7221773347fe46e494f8e5e0464ed12a : SUCCESS in 4m 43s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e86a9e932ca34c86b8f705a0325d5de0 : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/07ca58dfaf3344fe9c8ad5bd4be4d7ef : SUCCESS in 3m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/865647c18fb6497daf668f5fcaf97a0f : SUCCESS in 7m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b86c7e2c72ff48e2b55ba34709c31566 : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":2},{"id":"a8a896ccb4df2621ebf5b60be0b7fb910c5c800d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-19 22:33:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79ef7c54fa214f9683a122751cf6f3da\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/5b9a0e69713749698c024961669b8241 : TIMED_OUT in 3h 11m 37s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/6308a202ac3a4f8697d68fe08d1887fc : TIMED_OUT in 3h 12m 16s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/4cc264399eef45238702ac9530c13585 : SUCCESS in 2h 10m 48s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/6c14bc2b10924adc90248beea13c68b8 : SUCCESS in 1h 25m 48s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/a6026dcf1dcd4cd3b424068ddc85de8e : FAILURE in 2h 30m 41s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/f1dcff4d98d64f2c88e04a40040266b3 : FAILURE in 1h 49m 42s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d0aaca65b8cf4886a7cd3a4dd0ae5ab5 : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5793d162a0224b05bda7af99c9fb8366 : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/51b25f7a11e4415cb5e8932a90089b47 : SUCCESS in 5m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cb21d0706aba2d37e7704ac4e21399f6f781d813","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2026-07-10 14:38:50.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"6cc19fd84e47f80e40a70cb92afee94a5b1369f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 17:15:28.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/936ea5a4493742f98a6a22e16612a67d\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/8f5fccd594a742a49a2f0147c7dcae1c : SUCCESS in 57m 03s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/9047f2b292fa41a7a47a2f668befe612 : SUCCESS in 31m 19s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/29683850f1a945539d555af58d2e1e17 : SUCCESS in 12m 36s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/4d1475cf9a6c4d5aa0b61129e52af574 : SUCCESS in 25m 51s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/0b5a37c8aaee4fcdb235e1a72167af19 : SUCCESS in 28m 12s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a4bb1eb982fd42f690eadd43cc8a0633 : SUCCESS in 27m 13s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b71bf8344dc5445a9a513d70909cb4ef : SUCCESS in 31m 06s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/ed77e997b53a4be5997a5820d7d1d1e6 : TIMED_OUT in 2h 33m 50s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/ac86cd9cf1a942a480657714def296d6 : SUCCESS in 1h 08m 56s\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/386ab73898d9492ea40b01a24a633d27 : FAILURE in 24m 11s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/b3774bd41daa49a4b556813cf1ce86bd : SUCCESS in 1h 08m 11s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/eab216922219478dbe2ccd28805ffce3 : SUCCESS in 1h 28m 52s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/16348769179744ac81bb3db79b0f9d7a : SUCCESS in 1h 12m 15s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/6a9f8d6e7357413ebb982e79075b7613 : SUCCESS in 1h 35m 37s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/3c07a3f03e9c49038f98eacbbaaff128 : FAILURE in 1h 40m 27s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/a297fb64a2f5486e86ad8ee89a6964e2 : SUCCESS in 1h 34m 44s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/248512d1cdf04fb68ec9f37a8cb59466 : FAILURE in 1h 08m 21s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/14daab08f0b749b49ac5e9398d79c496 : SUCCESS in 1h 44m 21s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/7e8bd86870404d68b7796fff349f76ad : FAILURE in 1h 16m 59s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/675d68cd9c6640d28846acb83826836f : SUCCESS in 1h 19m 43s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/7c2bcd5a15d74a158bc2898edfbd6199 : SUCCESS in 1h 19m 34s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/f45cbca77d444591b028e1fcd09a2da7 : SUCCESS in 1h 23m 34s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/73e2b1c29f5a4a14bdff1fed929afcf5 : SUCCESS in 26m 10s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/770effb354994dbcb37c7bb362189578 : SUCCESS in 23m 31s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/96324f3aed274b8ab62ff9c560195825 : SUCCESS in 27m 36s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/08e18e24b7af43029461aa27ee138ed7 : FAILURE in 59m 47s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/3279950e8f25450293155142adb5e1f1 : SUCCESS in 43m 49s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/9aea1d259b2148819270d20dd156d39e : SUCCESS in 1h 18m 29s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/7f04fff4a29e4fe5bd3d9fdc8f6aa98f : SUCCESS in 1h 04m 53s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/3505d74d2d764706b54f1092a7f93400 : SUCCESS in 1h 05m 04s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/f3c28f701b024e05947b808c696c3f1e : SUCCESS in 1h 44m 32s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/1478b8d3d8f14204a78f243cdbbbc674 : SUCCESS in 1h 46m 36s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/f321ab76b92748f6a1b7b26652cef8f8 : SUCCESS in 1h 44m 07s (non-voting)\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/402b0e3b0ea14b69b01aa514097ff46f : FAILURE in 1h 21m 27s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/7c93e8ca13954e20af975accb9dcd1ba : FAILURE in 56m 44s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/c3ac36acebbf4801a62b6296a9a952d9 : SUCCESS in 1h 15m 33s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/f6164814300a4dd990aba2ee15cd1c57 : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed329e4d1671461092fb64721568eb08 : SUCCESS in 4m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0db7b7e3b6834d339eb7ba642b504057 : SUCCESS in 3m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fb7f673eed174e99b9ad0b28070d44e4 : SUCCESS in 3m 15s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/27b5e99f9997467596bbf7f162def954 : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/095506abe460401f9507c2a0b4930ffb : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c08b5a16b711430c8012586b7255725f : SUCCESS in 8m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a35524f43de94da3a731837d539cc3fb : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":3},{"id":"faa3ae05d655cafdd3911bf9474c3d7e9831a292","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2026-07-10 17:16:55.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"61779b1f22c79111a0efd10658494abe730b2c9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 19:53:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/abca61c68c0b45a29d9daa3af0ebf947\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/47977e71b0f74e0189b13fea626ece73 : SUCCESS in 57m 58s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/682be493fed847b0a6c50c6f37d47de2 : SUCCESS in 21m 09s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/851d61c9c81347a18a514b8ea421c862 : SUCCESS in 19m 52s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/118b02c3b28f49d4b5197f669242de50 : SUCCESS in 25m 08s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/2d5e187e744e4ae8b551e3fae65eae06 : SUCCESS in 27m 26s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f0e3d8e4199a453782bc95d6c62db23b : SUCCESS in 24m 40s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/127bb1e84210489d9ca9a715e911ec42 : SUCCESS in 23m 41s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/770d9490e554406e82735eb216ec9bdd : TIMED_OUT in 2h 33m 50s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/40b709ada3e94131b5cb81b19800cb70 : SUCCESS in 1h 12m 38s\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/01ec61b6793249c8bcddebb9849ea0c2 : SUCCESS in 1h 42m 33s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/30f5a0cc29df4c9ab2371902410db7bf : SUCCESS in 1h 03m 05s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/63ae771de1bf47baad269b950cb03e33 : SUCCESS in 1h 26m 02s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/029db353f6b74f8fbe6f903005636a7f : SUCCESS in 1h 17m 02s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/2a884a1de9cd459e8b61256bc8ceb2aa : SUCCESS in 1h 34m 56s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/eaa9917fd2e64c22ba6969a8e69dee20 : SUCCESS in 1h 30m 34s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/049180dbf5b442718c8da09e8090edf3 : SUCCESS in 1h 33m 01s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/6be7d4fef9334ff6b54816c5f5b173fc : SUCCESS in 2h 22m 35s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/e22c6ba8b824437491b52569ff36a95f : SUCCESS in 1h 45m 13s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/c10db17a640f43d1970d5c025c8893bb : SUCCESS in 2h 15m 19s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/e29670d164f142efa8d4a217463edc91 : SUCCESS in 1h 34m 03s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/18b5482980304022920776cde32bcd94 : SUCCESS in 1h 19m 35s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/dd4fd589a1ad4c9fb80ee3bbc6aa05ec : SUCCESS in 1h 24m 30s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/3cb0ef658fb54cc4b3c986922b565621 : SUCCESS in 22m 41s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/2575646d34ba4cac8f48c8565af70f25 : SUCCESS in 23m 37s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/f50ff71f765648d7b7739ff1a8e6bad0 : SUCCESS in 23m 05s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/847e50eb2d6c4da28574ffc05364a042 : FAILURE in 5m 52s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/d0879a7a6415409cb4422d25b3ac8dda : SUCCESS in 1h 02m 41s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/15eb35b0573c4b03b5eb198ca010d6ea : FAILURE in 9m 05s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/7aa3052bedfc42a395bac965f669ab31 : SUCCESS in 1h 05m 30s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/1f45fd0f33224b93adb4b87505cf65d4 : SUCCESS in 1h 09m 12s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/c34f1a3a2f234be8911984c4e1d2b986 : FAILURE in 1h 11m 29s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/80fd94d2664a4160892a1356c0b2d292 : FAILURE in 1h 14m 54s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/6fd00bf7df8943bab11a41f0aec840cb : SUCCESS in 1h 40m 41s (non-voting)\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/206051a8df58452b836bfd0a9e3b53b6 : FAILURE in 1h 17m 54s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/a4f4a4bd2b184347891ab70b2a24ea93 : SUCCESS in 47m 26s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/d506f1a587ce4594b4bc31d82822c0cd : SUCCESS in 44m 03s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/f03a0820caa84ddab0177b165ee8e6ef : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e0f39880f0c45ef8d23de6664b86f67 : SUCCESS in 3m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/965759303f324306a6a9cfe211b99c5c : SUCCESS in 3m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/45718c2db5114d2b987b5892ada2b280 : SUCCESS in 3m 13s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/cb9084613e104f7793d5da131476e791 : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6c4e90b8d34a459bad9671d45b42ec2f : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2bc5e7658d78420faf94314bb86fd1ab : SUCCESS in 6m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b0916c32cab6450783457457b94c896c : SUCCESS in 9m 06s","accounts_in_message":[],"_revision_number":3},{"id":"19d119b5883399d0a5823f23191676a8c84d0954","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 23:39:20.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35ce941af5aa47d79b88923b044660be\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/bf847893fee844e7a3dec84338e09ad9 : SUCCESS in 1h 25m 20s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/65de108b986d48e59aeedbfce73c918d : SUCCESS in 35m 42s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/c015a2eefd1448c99c0b7b9fde13bc11 : SUCCESS in 49m 58s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/5793b6def15240659589450fdec24cdc : SUCCESS in 36m 25s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/8170d236f80649a8909365fdea1ed4d9 : FAILURE in 2h 06m 00s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/306c1c034354453aa573f5d55e2792f3 : FAILURE in 1h 31m 48s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ec26914a53824669be54fb261f46c931 : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8a592c17253f4f68a44ac5499dd2c271 : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/5eb7cc2aa96f4e319be18a2d434182a2 : SUCCESS in 9m 10s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b0dd34c3f3ca5f2dcc09350234001d1df1784d6a","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-07-17 09:08:03.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"fffd72f78163363cbaaa75192ab491e705fbb184","revisions":{"b4905fd1b644004169ab2054b1933dde75193d44":{"kind":"REWORK","_number":1,"created":"2026-06-17 16:02:56.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/88/993788/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/88/993788/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/993788/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/993788/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/993788/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/88/993788/1"}}},"commit":{"parents":[{"commit":"c2a33b186be3c5991c85fdc0fa5172f52e55900a","subject":"CI: Switch arm64 publish jobs to periodic-weekly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c2a33b186be3c5991c85fdc0fa5172f52e55900a"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-06-17 16:00:55.000000000","tz":120},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-06-17 16:02:40.000000000","tz":120},"subject":"Migrate internal state storage to YAML","message":"Migrate internal state storage to YAML\n\nThis commit migrates the internal configuration state tracking\n(KOLLA_DEFAULTS_STATE) from JSON to YAML.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/982275\n\nChange-Id: I78f89dd7a4379f842153feb476f5bee824c12ffc\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b4905fd1b644004169ab2054b1933dde75193d44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b4905fd1b644004169ab2054b1933dde75193d44"}]},"branch":"refs/heads/master"},"83ff47ae7b19e3418cf707fc2d11e882f36e92c0":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-06-17 19:05:46.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/88/993788/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/88/993788/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/993788/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/993788/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/993788/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/88/993788/2"}}},"commit":{"parents":[{"commit":"d6910f01deb3fa41b2a7662d1d5eb4a7d1b5c50f","subject":"Add YAML support to kolla_set_configs engine","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d6910f01deb3fa41b2a7662d1d5eb4a7d1b5c50f"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-06-17 16:00:55.000000000","tz":120},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-06-17 19:04:25.000000000","tz":120},"subject":"Migrate internal state storage to YAML","message":"Migrate internal state storage to YAML\n\nThis commit migrates the internal configuration state tracking\n(KOLLA_DEFAULTS_STATE) from JSON to YAML.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/982275\n\nChange-Id: I78f89dd7a4379f842153feb476f5bee824c12ffc\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/83ff47ae7b19e3418cf707fc2d11e882f36e92c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/83ff47ae7b19e3418cf707fc2d11e882f36e92c0"}]},"branch":"refs/heads/master"},"fffd72f78163363cbaaa75192ab491e705fbb184":{"kind":"REWORK","_number":3,"created":"2026-07-10 14:38:50.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/88/993788/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/88/993788/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/993788/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/993788/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/993788/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/88/993788/3"}}},"commit":{"parents":[{"commit":"6b3724a59c94dd7818821f5ea9c9d50b2a880d09","subject":"CI: Fix file triggers after docker/ move","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6b3724a59c94dd7818821f5ea9c9d50b2a880d09"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-06-17 16:00:55.000000000","tz":120},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-07-10 14:38:47.000000000","tz":120},"subject":"Migrate internal state storage to YAML","message":"Migrate internal state storage to YAML\n\nThis commit migrates the internal configuration state tracking\n(KOLLA_DEFAULTS_STATE) from JSON to YAML.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/982275\n\nChange-Id: I78f89dd7a4379f842153feb476f5bee824c12ffc\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fffd72f78163363cbaaa75192ab491e705fbb184"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fffd72f78163363cbaaa75192ab491e705fbb184"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
