)]}'
{"id":"openstack%2Fkolla~994144","triplet_id":"openstack%2Fkolla~stable%2F2025.2~I1ec8f15e14360422aa316779bd19195fd1f4f7a1","project":"openstack/kolla","branch":"stable/2025.2","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-06-24 08:06:05.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-06-23 08:48:45.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-06-24 09:10:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1ec8f15e14360422aa316779bd19195fd1f4f7a1","subject":"Update Prometheus to version 3.5.4","status":"MERGED","created":"2026-06-19 16:09:47.000000000","updated":"2026-06-24 09:11:59.000000000","submitted":"2026-06-24 09:10:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"994144","cherry_pick_of_change":993943,"cherry_pick_of_patch_set":1,"meta_rev_id":"9e76900100b6cc76a21f2ac3ac88b766d2b55e65","_number":994144,"virtual_id_number":994144,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-24 09:10:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2026-06-24 08:06:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2026-06-23 08:48:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2026-06-24 08:06:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-19 16:31:59.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-06-19 17:03:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-22 15:22:48.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"f27fa5bcd797b7ce52be26a9ee72a2e86c99cfde","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-06-19 16:09:47.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"aca31be35210f64413185611d79620b125f72e19","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-06-19 16:31:59.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0e81afbdc938fd082b544b14a54f4965afadd99b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-19 17:03:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5d73e4b89ac43f286b2ca4cb479d2fe\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/d99a598f7ddd472a8982712fc44fe454 : SUCCESS in 37m 56s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/cd897567f3474ffda11edf4e3e273a4c : SUCCESS in 47m 58s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/6018e159f86644359ba06e8ad97c4787 : SUCCESS in 30m 00s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/90f83a15a88248d99b3325216dde0816 : SUCCESS in 47m 53s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/30f9f5b71ba54adfadcdf297f28813b7 : SUCCESS in 47m 17s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/91635e4b313746debd0f2aaa2eba7e9f : SUCCESS in 44m 14s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/021a7ca061a44c45b640921224751380 : SUCCESS in 29m 53s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/e11861e901b541e9bb3475c13defd655 : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/676a784e9d2c41a884dbf82e6dc32fe6 : SUCCESS in 3m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/add336858e2d44928a70b35cc0e8798c : SUCCESS in 6m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2ccc2b75597a41f58ddeca6941c0230f : SUCCESS in 3m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b094ed4f20c049158c15a9b4cde945e7 : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/11f96942807a4a37a6cb861ad56a57b6 : SUCCESS in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f31b788ecba2477ea4bb827f16f823c3 : SUCCESS in 5m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1ac0d39956d6464abd64965fdffa9e77 : SUCCESS in 10m 09s","accounts_in_message":[],"_revision_number":1},{"id":"176256d4bfbcd711dd3c2995ccdeb071be980bcd","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-06-19 18:23:48.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"411b2e5e3eda355ffe987bff58f83afff3dc9c1f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-19 21:16:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a032bb897642494b9b650b5d707347d1\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/b27c580b5e0a4f5bb8e7fe33ed4f7391 : SUCCESS in 2h 01m 06s (non-voting)\n- kolla-build-debian-bookworm-arm64 https://zuul.opendev.org/t/openstack/build/0a57b8fbf1564956910babcd9c4ecfec : SUCCESS in 1h 13m 53s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/3fe46877b7c94f8981c12fed589f1c79 : SUCCESS in 1h 00m 03s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/f30a0664ce65487abac66f6347719689 : TIMED_OUT in 3h 04m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"23962922b0d2be1eb130d15468cfecd1bd12a13b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-06-22 07:01:46.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n* Workflow-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"4d790b1b969a74468addfc5937bef987127a7428","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 08:05:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4bdb5d209e5462c91629ca72723d3b4\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/34306a8a11a54754b0ea20d7813bcac7 : SUCCESS in 46m 45s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/733d32f7b48a47f0b2cf97974e96c659 : SUCCESS in 48m 05s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/8e040cd9725c4614914486d88f4d1c8a : SUCCESS in 17m 06s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/6a9acc5d390e4b6694bb513bde61e095 : SUCCESS in 48m 05s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/5cc95df525a645dd971f1abf1a0cd439 : SUCCESS in 1h 01m 23s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c6d46b8e61a541eea871f6662d7809f7 : SUCCESS in 38m 02s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c497c4b1a38b42ef937e2ab7ec77e7ea : SUCCESS in 32m 22s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/6a3651d9cf94443bbf83f1137d618704 : SUCCESS in 2m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ed3ab05738f4e0ea47d09288b49d499 : SUCCESS in 5m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9111564f03004f529645cc72765da5ef : SUCCESS in 4m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/74718359db634704bd5227442c5f8f21 : SUCCESS in 3m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/604bd70ef1be4b068daf9742adde5b41 : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/954c28db8ded420194ac6410c4ddbb40 : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69be26089a104b38a845ee7fc95fd673 : SUCCESS in 7m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/116c803b0afa46809d27cf420cbfc699 : SUCCESS in 8m 32s","accounts_in_message":[],"_revision_number":2},{"id":"c1a5c923bb0c897a7b5508cad84a6e1cd3534730","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 09:19:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9c4899d9f0843208fbe100574c7b4d1\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/f59e0ed47198420fa90767c1f52cf35a : SUCCESS in 1h 23m 13s (non-voting)\n- kolla-build-debian-bookworm-arm64 https://zuul.opendev.org/t/openstack/build/b00b9dbed97d401e8a5159d435215415 : SUCCESS in 1h 10m 05s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/93d214991e474e47bcb1bde432b12c2f : SUCCESS in 1h 19m 58s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/1b90f1012f66481c8c2c02713bdf1b59 : SUCCESS in 1h 19m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d58425418dcf9363be0c47c8d51394e690854b75","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-06-22 09:55:54.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"b5f72d12a018ede6775ffd132893eed6cbac9b9c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-22 15:22:48.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3e27e0bb2a86a273312d0662a5c78afee6bbfc2d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-06-22 18:01:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e52d46801dedfbb7ef1b67282feba59e21f2da8f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-22 20:13:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7d711166a631de4de3740ba0857cb360cce95656","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-06-22 21:19:08.000000000","message":"Patch Set 3: Cherry Picked from branch master.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"f862e891c425b6808df91b77a85a4765e6e6d1e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 22:25:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f636335ba2544a2fb1301c75dd1b2dd0\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/0d9bc3d3eb2f4c05a4c6a49c84469538 : SUCCESS in 41m 35s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/44f13bc35ae9462dbfc059721e7f2431 : SUCCESS in 41m 11s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/bc97174cb8d44d90bac3e5d461aaf527 : SUCCESS in 32m 01s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/529bc6fa6eee45e1ac905d5d3cc354c0 : SUCCESS in 46m 27s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/db8ed0c75589499ba384a629fd71b0a7 : SUCCESS in 59m 06s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2505a549c1444925bb297d70bd7a5d94 : SUCCESS in 36m 23s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ec2519d10d3c43e39744d80175db1847 : SUCCESS in 21m 44s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/bc6070d0a4cf4e4f98f762f9538b09fb : SUCCESS in 2m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8292f20d04a34264b5319f163098974f : SUCCESS in 2m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b1e99b4941f64a0f97bfe08c80423dc7 : SUCCESS in 5m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5dac263ccab542979e39f091a8b3a12b : SUCCESS in 2m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dae3f48e0fff48c6a860e475a28a82fe : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/95ed91c65584450fa95150e524253bdf : SUCCESS in 3m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46a7bbe03aae4e5a91d77f515ca79259 : SUCCESS in 3m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9e78047fddba48b9a0c2ac446f451fcb : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":3},{"id":"5b078a8f36cd91deb06020b62b5589b292476e45","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-23 01:18:20.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9589e535117644e592f80d0705a098dc\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/8656ce64c4ac43aba16f465f0ceb7d9a : SUCCESS in 2h 14m 43s (non-voting)\n- kolla-build-debian-bookworm-arm64 https://zuul.opendev.org/t/openstack/build/7e9e6eafd21649e9973c8347304c9bfc : SUCCESS in 1h 52m 17s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/dc7f0aff457d4ae3a2a7165187bf37c9 : SUCCESS in 1h 25m 31s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/24ebd58a709e422ab871ad2f37df4f2d : TIMED_OUT in 3h 03m 54s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6ebc8a76947cac9087a6c1c75578e4a06df728cd","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-06-23 08:48:45.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"032ab040dae7a9ba46c5a9b57dfb66e389be3c32","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-24 08:06:05.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"13c754ebd5a99e680585b03deee8adcf3ede2752","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 08:06:21.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f00140f59d8901b87d50ea89b1fa8d77d607deae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 09:10:01.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59660fad65c14a90a4deaf5ef7739a3b\n\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/6168b51e78f74554abd4cd2af6cccb34 : SUCCESS in 47m 43s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/1039b9cc463f4d6d92d9f13b42713282 : SUCCESS in 36m 53s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/e8925cb723ca4a7ea77ff6cb172a2e8e : SUCCESS in 45m 24s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/a9f5cd2c9726436da5033f9b939f1fa7 : SUCCESS in 1h 02m 01s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d339023edbb94ab4bca9cbd97232f92b : SUCCESS in 50m 19s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/462d4bc2671a4443915688f9f060dea7 : SUCCESS in 23m 05s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/583baeec594440deadd24a795be018d0 : SUCCESS in 2m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30a6296fec4845119ddb6a1168a6c750 : SUCCESS in 4m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/088528c10505462a9e0a58e22e1ea342 : SUCCESS in 4m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4840ed16a9574e9da49726ec3a4b33e3 : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebd6806fb8c344e68d15ff74a1bbc3b7 : SUCCESS in 7m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/067ad7d5f63a4871863c1ae95600a62f : SUCCESS in 9m 43s","accounts_in_message":[],"_revision_number":3},{"id":"6f769f0d001b7cbd06f6c9eb6b1b38fd574fd1bf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 09:10:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"9e76900100b6cc76a21f2ac3ac88b766d2b55e65","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 09:11:59.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f4c4aeb30ca4cdf86643f4955b3d9ba\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c62bfd7e3a5a462b80f0ccc4590dbef3 : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/799b96a5fe5347a1adcc26967ef85d44 : SUCCESS in 1m 40s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e7b1997055293d725dc62e20049954d2d59aae28","revisions":{"8527a38513b919448a64ea3000509e3abbce081e":{"kind":"REWORK","_number":1,"created":"2026-06-19 16:09:47.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/44/994144/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/994144/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/994144/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/994144/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/994144/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/994144/1"}}},"commit":{"parents":[{"commit":"11afc796b3a4a4ca2e83072c829fdd4ea9aaa505","subject":"CI: Switch arm64 publish jobs to periodic-weekly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/11afc796b3a4a4ca2e83072c829fdd4ea9aaa505"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-06-18 14:46:40.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-06-19 16:09:47.000000000","tz":0},"subject":"Update Prometheus to version 3.5.4","message":"Update Prometheus to version 3.5.4\n\nThis release fixes multiple security issues [1].\n\n[1] https://github.com/prometheus/prometheus/releases/tag/v3.5.4\n\nChange-Id: I1ec8f15e14360422aa316779bd19195fd1f4f7a1\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n(cherry picked from commit e4114abcdaa604d243da48ea0187fd5673191eb6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8527a38513b919448a64ea3000509e3abbce081e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8527a38513b919448a64ea3000509e3abbce081e"}]},"branch":"refs/heads/stable/2025.2","conflicts":{"base":"a437f0bd5f406bf1f82dfdf05c4dac16c02573ff","ours":"11afc796b3a4a4ca2e83072c829fdd4ea9aaa505","theirs":"e4114abcdaa604d243da48ea0187fd5673191eb6","merge_strategy":"recursive","contains_conflicts":false}},"d3e379532487a1e725a611ef76d138a33f9a0171":{"kind":"REWORK","_number":2,"created":"2026-06-22 07:01:46.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/44/994144/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/994144/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/994144/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/994144/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/994144/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/994144/2"}}},"commit":{"parents":[{"commit":"11afc796b3a4a4ca2e83072c829fdd4ea9aaa505","subject":"CI: Switch arm64 publish jobs to periodic-weekly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/11afc796b3a4a4ca2e83072c829fdd4ea9aaa505"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-06-22 07:01:25.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-06-22 07:01:25.000000000","tz":120},"subject":"Update Prometheus to version 3.5.4","message":"Update Prometheus to version 3.5.4\n\nThis release fixes multiple security issues [1].\n\n[1] https://github.com/prometheus/prometheus/releases/tag/v3.5.4\n\nChange-Id: I1ec8f15e14360422aa316779bd19195fd1f4f7a1\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n(cherry picked from commit e4114abcdaa604d243da48ea0187fd5673191eb6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d3e379532487a1e725a611ef76d138a33f9a0171"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d3e379532487a1e725a611ef76d138a33f9a0171"}]},"branch":"refs/heads/stable/2025.2"},"e7b1997055293d725dc62e20049954d2d59aae28":{"kind":"REWORK","_number":3,"created":"2026-06-22 21:19:08.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/44/994144/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/994144/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/994144/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/994144/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/994144/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/994144/3"}}},"commit":{"parents":[{"commit":"11afc796b3a4a4ca2e83072c829fdd4ea9aaa505","subject":"CI: Switch arm64 publish jobs to periodic-weekly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/11afc796b3a4a4ca2e83072c829fdd4ea9aaa505"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-06-18 14:46:40.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-06-22 21:19:08.000000000","tz":0},"subject":"Update Prometheus to version 3.5.4","message":"Update Prometheus to version 3.5.4\n\nThis release fixes multiple security issues [1].\n\n[1] https://github.com/prometheus/prometheus/releases/tag/v3.5.4\n\nChange-Id: I1ec8f15e14360422aa316779bd19195fd1f4f7a1\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n(cherry picked from commit e4114abcdaa604d243da48ea0187fd5673191eb6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e7b1997055293d725dc62e20049954d2d59aae28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e7b1997055293d725dc62e20049954d2d59aae28"}]},"branch":"refs/heads/stable/2025.2","conflicts":{"base":"a437f0bd5f406bf1f82dfdf05c4dac16c02573ff","ours":"11afc796b3a4a4ca2e83072c829fdd4ea9aaa505","theirs":"e4114abcdaa604d243da48ea0187fd5673191eb6","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
