)]}'
{"id":"openstack%2Fkolla~995004","triplet_id":"openstack%2Fkolla~unmaintained%2F2024.1~I1e3d920da5b2e23b6137955ba51acbaa64205df0","project":"openstack/kolla","branch":"unmaintained/2024.1","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2026-06-26 10:45:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1e3d920da5b2e23b6137955ba51acbaa64205df0","subject":"Cleanup zuul jobs","status":"MERGED","created":"2026-06-26 07:06:34.000000000","updated":"2026-06-26 10:47:26.000000000","submitted":"2026-06-26 10:45:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"995004","meta_rev_id":"1be7662d61d79c8fba7a068619a910a534ad0e33","_number":995004,"virtual_id_number":995004,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-26 10:45:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2026-06-26 10:24:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2026-06-26 10:24:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-26 07:19:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-26 10:24:14.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"2689918111962d72c517416a0e9b1dd4ff279265","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-06-26 07:06:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b62303ef5dcc6652d30dc88aa753254866527b58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 07:19:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8aeee34a180443cc8caf39c3c08350b4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba44ab269e934f35808260f2991e529d : SUCCESS in 3m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/96fdf5009e3c4fe88339d311fa050c7b : SUCCESS in 4m 24s (non-voting)\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/52758bbab12c4a7cbcae8a488c30c079 : SUCCESS in 2m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ef36ec22f45a4de2ae1b12644f78ef51 : SUCCESS in 1m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5b88da9cd9a4adcb6e873a9b86d6551 : SUCCESS in 5m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/db6db7af895c4c4881ce1f21a2d2ff5c : SUCCESS in 3m 36s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/f3f07078daf5404bad9965ec96decdc7 : SUCCESS in 4m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ab256f7d40bc42dea1001cb7ccf2f583 : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":1},{"id":"b4483b6f2bed0eb411ae3ca3f070379e8b155b1d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 07:30:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2f3591774ef4d71b6e3a2aa3283e194\n\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bf65e930de494bf2a0a3334a872bb8c9 : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":1},{"id":"581982abe2e6d99d2b03d3b069a602b1bd6db94b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-06-26 10:12:59.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"e93641356f6b674ccb536224607402dd149ce627","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-26 10:24:14.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"648b8e83b6befa79e323a800294a07edb7c7ef82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 10:29:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d48c875c6c944b23919b52049cde403c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e11fdc570bba4417b21127225d3ee579 : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/974644070f05451dbe7047e4b01f9287 : SUCCESS in 2m 38s (non-voting)\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3cd16dc9f83d440baa8077fc9e3f43fa : SUCCESS in 2m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c1356b2977e404dbbbcce67f58df19e : SUCCESS in 3m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97f203a9565d48fdb9a2b947c28ab837 : SUCCESS in 4m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1dd0f104e3ac4cd6ba84f9cbcdef40a7 : SUCCESS in 9m 33s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/3a02c26d0afe46048d311008d396d611 : SUCCESS in 2m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7df9666c29574d79ad0f95ec978ec793 : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":2},{"id":"917c95ac335566368ba238fd5957cc853cb8abf8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 10:29:37.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"12ce675e0fbcca371b4f7fc9158b84190ecaf237","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 10:45:35.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5754546f009642afa5ac72ecf195ffd7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b1adfdbd135448ab73151a52c6de422 : SUCCESS in 5m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ddf2a1591f694ef7add743979cb5f896 : SUCCESS in 4m 10s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/f6d1cc7c7e604c0e8131d8c86da5309a : SUCCESS in 2m 37s","accounts_in_message":[],"_revision_number":2},{"id":"5deae997f2d6f933e0f09467347078245bfc3eb2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 10:45:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"1be7662d61d79c8fba7a068619a910a534ad0e33","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 10:47:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db95d0816e9d4bbb9bd08c7ff77ffe27\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ec4c7a8cf6d479292706309e8ca859d : SUCCESS in 46s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ee46746c64e1f42e276f6ee8d72996d2be575f81","revisions":{"7ba0cfa0b4af0d1477b0949d2a2853e7817f2021":{"kind":"REWORK","_number":1,"created":"2026-06-26 07:06:34.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/04/995004/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/04/995004/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/995004/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/995004/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/995004/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/04/995004/1"}}},"commit":{"parents":[{"commit":"641ef6917bd4737cec8e9c6970df7b1058615bfb","subject":"stable-only: fix zuul config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/641ef6917bd4737cec8e9c6970df7b1058615bfb"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-06-26 06:59:50.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-06-26 07:05:42.000000000","tz":120},"subject":"Cleanup zuul jobs","message":"Cleanup zuul jobs\n\nCentos, Debian and Rocky builds are broken, let us just continue with\nUbuntu testing unless someone wants to invest time into fixing on of the\nother distros.\n\nAlso stop running arm64 jobs completely.\n\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: I1e3d920da5b2e23b6137955ba51acbaa64205df0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7ba0cfa0b4af0d1477b0949d2a2853e7817f2021"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7ba0cfa0b4af0d1477b0949d2a2853e7817f2021"}]},"branch":"refs/heads/unmaintained/2024.1"},"ee46746c64e1f42e276f6ee8d72996d2be575f81":{"kind":"REWORK","_number":2,"created":"2026-06-26 10:12:59.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/04/995004/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/04/995004/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/995004/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/995004/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/995004/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/04/995004/2"}}},"commit":{"parents":[{"commit":"641ef6917bd4737cec8e9c6970df7b1058615bfb","subject":"stable-only: fix zuul config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/641ef6917bd4737cec8e9c6970df7b1058615bfb"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-06-26 06:59:50.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-06-26 10:12:40.000000000","tz":120},"subject":"Cleanup zuul jobs","message":"Cleanup zuul jobs\n\nCentos, Debian and Rocky builds are broken, let us just continue with\nUbuntu testing unless someone wants to invest time into fixing on of the\nother distros.\n\nAlso stop running arm64 jobs completely.\n\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: I1e3d920da5b2e23b6137955ba51acbaa64205df0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ee46746c64e1f42e276f6ee8d72996d2be575f81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ee46746c64e1f42e276f6ee8d72996d2be575f81"}]},"branch":"refs/heads/unmaintained/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
