)]}'
{"id":"openstack%2Fkolla~995011","triplet_id":"openstack%2Fkolla~stable%2F2025.1~Ia35c87428841e57a04b42541021d6b7daa3cd43a","project":"openstack/kolla","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"39213":{"account":{"_account_id":39213,"name":"Jed Preist","display_name":"Jed Preist","email":"jed@stackhpc.com","username":"jed-preist"},"last_update":"2026-07-01 15:07:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia35c87428841e57a04b42541021d6b7daa3cd43a","subject":"[2025.2 and older] Create Ironic temp directory","status":"MERGED","created":"2026-06-26 08:39:22.000000000","updated":"2026-07-01 15:08:40.000000000","submitted":"2026-07-01 15:07:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"995011","cherry_pick_of_change":993769,"cherry_pick_of_patch_set":6,"meta_rev_id":"0e96c6d04b4349b74383da23f33398f0202a9599","_number":995011,"virtual_id_number":995011,"owner":{"_account_id":39213,"name":"Jed Preist","display_name":"Jed Preist","email":"jed@stackhpc.com","username":"jed-preist"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-01 15:07:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2026-06-29 09:52:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2026-06-26 17:45:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2026-07-01 13:26:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-26 10:19:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-26 17:45:09.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-06-29 09:52:16.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"}],"messages":[{"id":"952b6f9290ef3d57df20c4b9fb19ead3d90feaf5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39213,"name":"Jed Preist","display_name":"Jed Preist","email":"jed@stackhpc.com","username":"jed-preist"},"date":"2026-06-26 08:39:22.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2025.2.","accounts_in_message":[],"_revision_number":1},{"id":"7bea501a5dd7f0bfc6bcec520ce9c73a7f54f02a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 10:19:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/407ad0ed27774bb2900f345f1d232def\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/a31290da26644aada6ce3a577b443ba7 : SUCCESS in 41m 07s (non-voting)\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/84afd0b4483f4f2faa7357c54aa66e9d : SUCCESS in 59m 06s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/072fe5de33664991add648ac189a0d5a : SUCCESS in 1h 34m 20s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/a25b291c3d5c431082c6912ab4427f2f : SUCCESS in 46m 06s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/fae77fc8727f478e80cae3998785f2c6 : SUCCESS in 46m 40s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/6e9c9101c41b435f8f0e3ff0db99070c : SUCCESS in 25m 02s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/dcb4d355f2b84702b9aa7dcca14badaa : SUCCESS in 53m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9aa34ef058c42f99adb68f09010cdc4 : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8de31786837e40e28e4dcf9499f950c7 : SUCCESS in 4m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b502a482ddad443e8a9a793bd2e4a395 : SUCCESS in 3m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a2bc0e25b6984f1390e4b18c0b423751 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66f6660019de48d28c4a79146fb96ad6 : SUCCESS in 7m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/222fbc145e394943b0735e0a86a7a7bb : SUCCESS in 8m 58s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/af5abe9209214a3fa17c189c58af1a21 : SUCCESS in 2m 58s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/312dfb646ac54e099973a892913725ae : SUCCESS in 1h 06m 42s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/e7f8c0d0896e4fa68ed73c33458d2748 : SUCCESS in 1h 17m 33s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/5291d1b3d72f44fb943ca5a413ea8c89 : SUCCESS in 1h 18m 33s\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/deb2b8130ce34db4be25b465ddbbd796 : FAILURE in 1h 23m 01s (non-voting)\n- kolla-build-rocky10 https://zuul.opendev.org/t/openstack/build/f6f281865ee445a3b55e7fe6a55c2220 : SUCCESS in 1h 02m 22s\n- kolla-build-rocky10-podman https://zuul.opendev.org/t/openstack/build/48e15c7b0ac64ae9a5c75d3f45342e49 : SUCCESS in 53m 43s\n- kolla-ansible-rocky10 https://zuul.opendev.org/t/openstack/build/68d65015e676436794467278e4e0b139 : SUCCESS in 1h 30m 34s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d31c89a19b884778adacc2fb32417e2e : SUCCESS in 55m 33s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/bb45f10fa2e948e99df702340c806cd7 : SUCCESS in 32m 35s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/32ac60474a3e40cf800bc5ad50c56410 : SUCCESS in 1h 02m 38s\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/d590a946b83d410a98cede1441da2679 : FAILURE in 44m 29s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/d01fc5718b88401ca47d3f9c4feb862c : FAILURE in 9m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"85e36e9857093f7119fea37f074b461c104d7dcf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 10:56:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a14cb6c0a9c045eea683bf62371f3cd1\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/5d0f3b1e89d842328db3f9b8aad552cd : SUCCESS in 1h 42m 57s (non-voting)\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/c9b6b337267149babf020d79a6faf0a3 : SUCCESS in 1h 07m 10s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c819f1efabe147d3a662d87533d2c376 : SUCCESS in 1h 11m 49s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/92e5ba214fe94682aa487e19309a5663 : SUCCESS in 2h 04m 04s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/c031569a57f64a77ab0f537c5fca5f3e : SUCCESS in 55m 34s (non-voting)\n- kolla-build-rocky10-aarch64 https://zuul.opendev.org/t/openstack/build/af00f3c620294cc4bbf4d1db854a34d5 : SUCCESS in 1h 29m 00s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/ad23a85e0f1c4705af1e43f5ffd2c204 : SUCCESS in 55m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c224e87b1a6e924ecaa5256505251b9da6e964c4","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-06-26 17:45:09.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"709cb9cafe417234948187e4b736de4f5cce69c0","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2026-06-29 09:52:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d119bf33fb9b181c5ee5f981b9b4b9b2c88ca9eb","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2026-07-01 13:26:56.000000000","message":"Patch Set 1: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9a2a2f8f79e68c9fe2def49b491b29380333e27f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-01 13:27:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e219d9e3eab43c44353002c07a826510b7e11119","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-01 15:07:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c5aeefb920b491b8d9bc2567ba1d502\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b9ffda1c5a6441aca4606a3f9f3f3494 : SUCCESS in 49m 41s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/90023dd0f9534b2caa08581d5434a57c : SUCCESS in 33m 29s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/bac47e7533b7455fb6b208269368e236 : SUCCESS in 55m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/887d1c398c0047d4a016b8f2d5db1aa9 : SUCCESS in 5m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be408ee9a0b84f6791a3a16afb722cae : SUCCESS in 4m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/79a8fdb7bd2c4e5e8121b36095d2aca0 : SUCCESS in 3m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/112f87e8a2f24e12b93a8feec7dab7dd : SUCCESS in 3m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3373c900e2f74264932d6eff8c0fe9ce : SUCCESS in 9m 42s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/94007b1493754639afb0d09ced1faf83 : SUCCESS in 4m 34s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/6a6d6979306c40c987bf82a29fc6afb4 : SUCCESS in 43m 54s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/b76569998ed4429284cdcb769676bc2f : SUCCESS in 1h 25m 41s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/8948ca61065b451b8fff5801343e698e : SUCCESS in 1h 30m 40s\n- kolla-build-rocky10 https://zuul.opendev.org/t/openstack/build/1c172b42c41d4ba0abbb5a850664fa32 : SUCCESS in 55m 47s\n- kolla-build-rocky10-podman https://zuul.opendev.org/t/openstack/build/59c40aa98c5e41e3951137f85e4e48e2 : SUCCESS in 42m 00s\n- kolla-ansible-rocky10 https://zuul.opendev.org/t/openstack/build/8099aee31f994a50a1e99c0f58135423 : SUCCESS in 1h 16m 37s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b814e356640e418da291aa5d777b15fa : SUCCESS in 57m 32s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/662f66f8ea0f4126a9deb32186fd5833 : SUCCESS in 30m 18s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/b9103b8c1fc24f5cb732126646cffe8a : SUCCESS in 1h 03m 48s","accounts_in_message":[],"_revision_number":1},{"id":"b7d809fbe660982a6c1e1cff6a952ee5ba5f5135","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-01 15:07:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0e96c6d04b4349b74383da23f33398f0202a9599","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-01 15:08:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2f5d94d3e5f47038748cf62ef17709e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/071fdb18bc764c619008946ebd1a6f76 : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2d75040e46224761af3b677071fd2bb8 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"69e6bb41698837b9d3def849c07823543c62f0be","revisions":{"69e6bb41698837b9d3def849c07823543c62f0be":{"kind":"REWORK","_number":1,"created":"2026-06-26 08:39:22.000000000","uploader":{"_account_id":39213,"name":"Jed Preist","display_name":"Jed Preist","email":"jed@stackhpc.com","username":"jed-preist"},"ref":"refs/changes/11/995011/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/11/995011/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/995011/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/995011/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/995011/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/11/995011/1"}}},"commit":{"parents":[{"commit":"9b05368ea341c49d2aa129a90302fc1728a74260","subject":"Merge \"keepalived: Install mariadb client package\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9b05368ea341c49d2aa129a90302fc1728a74260"}]}],"author":{"name":"Jed Preist","email":"jed@stackhpc.com","date":"2026-06-15 14:14:24.000000000","tz":60},"committer":{"name":"Jed Preist","email":"jed@stackhpc.com","date":"2026-06-26 08:39:22.000000000","tz":0},"subject":"[2025.2 and older] Create Ironic temp directory","message":"[2025.2 and older] Create Ironic temp directory\n\nCreates a temp directory at /var/lib/ironic/tmp inside the\nIronic container.\n\nChange-Id: Ia35c87428841e57a04b42541021d6b7daa3cd43a\nSigned-off-by: Jed Preist \u003cjed@stackhpc.com\u003e\n(cherry picked from commit 5144ee7b977dbf32544778cf0431812279a35294)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/69e6bb41698837b9d3def849c07823543c62f0be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/69e6bb41698837b9d3def849c07823543c62f0be"}]},"branch":"refs/heads/stable/2025.1","conflicts":{"base":"11afc796b3a4a4ca2e83072c829fdd4ea9aaa505","ours":"9b05368ea341c49d2aa129a90302fc1728a74260","theirs":"5144ee7b977dbf32544778cf0431812279a35294","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
