)]}'
{"id":"openstack%2Fkolla~995832","triplet_id":"openstack%2Fkolla~master~Ia25d10a6ff3608e2a38d1a88bf229fe3f96d19ee","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2026-07-04 08:11:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia25d10a6ff3608e2a38d1a88bf229fe3f96d19ee","subject":"Remove redundant permission parsing tests","status":"MERGED","created":"2026-07-02 14:25:42.000000000","updated":"2026-07-04 08:12:19.000000000","submitted":"2026-07-04 08:11:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"995832","meta_rev_id":"b29bff38510b7b947aafa49d469cdd0881d6d04a","_number":995832,"virtual_id_number":995832,"owner":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-04 08:11:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-07-04 07:36:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2026-07-02 20:15:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-07-04 07:36:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":-1,"date":"2026-07-04 07:36:27.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-07-04 07:36:27.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-02 15:34:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-02 20:15:35.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-07-04 07:36:27.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"1663f6e973b4d095e75d1fa65a54a2306d1e4acd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-07-02 14:25:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"84370a2f5af85fee5a2fe3a71aadcc6cfd7f59fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-02 15:34:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a898a76f67734d67a0aa4e4e8a0aaa08\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/5c8e0d126c504651897acc33fe071166 : SUCCESS in 54m 17s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/dba33fe320e447c29b39ca540343e0b6 : SUCCESS in 28m 40s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/1783701ce44c48218c29597f96d2437e : SUCCESS in 29m 32s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/1a2d93350d5f4a54b7162cc279429076 : SUCCESS in 31m 27s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/d21cc115a5b94b0883acbb4065892ba6 : SUCCESS in 1h 01m 33s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/246d3c4afbfb435392df5e6c0fdb4838 : SUCCESS in 26m 29s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/67f7a97b0e1f4b0e9bec043b68614a16 : SUCCESS in 31m 41s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/55a7383012c94dad94854947199a6e22 : SUCCESS in 2m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02fcebeb32bb409d8ee74d3cd86db33c : SUCCESS in 5m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bb8693a0e69746bbbb09c90874d13968 : SUCCESS in 3m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/933d64ab7a654d3eb8c8a2c6c412d68c : SUCCESS in 3m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6e84730477b24e358cc203009567bae7 : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5ec014111a484b20b9ef4a0796a65d3d : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64f8b472a464422b956efa30b840878c : SUCCESS in 7m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5bf3055eaae249419e7b99c39e6504d6 : SUCCESS in 9m 19s","accounts_in_message":[],"_revision_number":1},{"id":"176bba642f9327148ae84d3561dcf2da91fcaa93","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-02 15:52:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fcefa3de8b8d4f97bca7e5cef1ec06e6\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/8230e527dbe94a378c970deefd9b589d : SUCCESS in 30m 02s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/6d2876a008864b7eb4ceb007bbf5d61d : SUCCESS in 38m 01s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/6b8e20c9adf0470db0db1512ce8484d0 : SUCCESS in 1h 07m 15s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/40739cf2f9fb4c78812c4ee3157e3c2a : SUCCESS in 42m 26s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9ad947260b3d460d8eabd1ef3345c8a8 : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a716b93800a94d8fb6987611093a4625 : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/3d9755a72d814d40b57b9734d24b9b09 : SUCCESS in 6m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"24d5378e02d09c87a81c00e2bd396adb924e819c","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-07-02 20:15:35.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"342e5eb7e023eda648749026dc6445d71e97ce58","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-07-04 07:36:27.000000000","message":"Patch Set 1: Backport-Candidate-1 Code-Review+2 Review-Priority+1 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"950eb14b42de74adc03212b1329614c71884ec41","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-04 07:36:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dbb22df68de4862c40c7ed8d55fa65882f7e2322","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-04 08:11:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e31da69d647c45e2ae68bd6995529269\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/8d2c029f480d443f842dbb748dcd705f : SUCCESS in 13m 49s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/9167a2615f24497f9313636b63863296 : SUCCESS in 17m 29s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/c2dd7bcc19664c5d83a4b685514c36de : SUCCESS in 15m 01s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/337dae96b7af4d7fa51619abb582905f : SUCCESS in 33m 35s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3953d64608b34ee9b40b2b96ae531c71 : SUCCESS in 21m 31s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ee416f423b654f4abcf5f6c8c5bb8d36 : SUCCESS in 29m 00s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/4df3828182cf4a33af1a0d26539f7b08 : SUCCESS in 1m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1102cc86f15342c7875b4dff677ae92f : SUCCESS in 3m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/667faefa4fac4949966b26a76f2c381d : SUCCESS in 2m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f952ef2bfada456684b27616b1082b89 : SUCCESS in 2m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fa5e6e31dc1459db76950b413d0a7ac : SUCCESS in 3m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a89d131064b744c38928f98c52da1c18 : SUCCESS in 5m 54s","accounts_in_message":[],"_revision_number":1},{"id":"c83c1f0af214f577965e6bfd622c75f221f2d6a3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-04 08:11:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b29bff38510b7b947aafa49d469cdd0881d6d04a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-04 08:12:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a4b44a5271d49cdb3803e6e5ce56f15\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9481af5fada841ae9d52754dcbec3152 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8c0fe8de026016dd26bd6c23d8079f3e597b81a7","revisions":{"8c0fe8de026016dd26bd6c23d8079f3e597b81a7":{"kind":"REWORK","_number":1,"created":"2026-07-02 14:25:42.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/32/995832/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/32/995832/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/32/995832/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/32/995832/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/32/995832/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/32/995832/1"}}},"commit":{"parents":[{"commit":"575baf05c4964d46f0c8f13bcdf193ce5e863c82","subject":"Merge \"Fix parsing of permissions with special bits\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/575baf05c4964d46f0c8f13bcdf193ce5e863c82"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-07-02 14:24:16.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-07-02 14:25:33.000000000","tz":120},"subject":"Remove redundant permission parsing tests","message":"Remove redundant permission parsing tests\n\nThe removed Python octal notation test is already covered by\ntest_handle_permissions_accepts_valid_permission_notations, which\nincludes 0o2775.\n\nChange-Id: Ia25d10a6ff3608e2a38d1a88bf229fe3f96d19ee\nSigned-off-by: Michal Arbet \u003cmichal.arbet@ultimum.io\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8c0fe8de026016dd26bd6c23d8079f3e597b81a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8c0fe8de026016dd26bd6c23d8079f3e597b81a7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-1"],"failing_atoms":["label:Backport-Candidate\u003d1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
