)]}'
{"id":"openstack%2Fkolla~996108","triplet_id":"openstack%2Fkolla~master~I8de87eff00b226e8a85eca317af944e9b6c97b3f","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-09-01 10:29:58.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-09-01 07:47:59.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"37306":{"account":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"last_update":"2026-09-01 10:29:58.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2026-09-01 07:50:08.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-08-13 10:37:59.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"I8de87eff00b226e8a85eca317af944e9b6c97b3f","subject":"Upgrade Prometheus Server to 3.13.x LTS","status":"MERGED","created":"2026-07-06 15:11:15.000000000","updated":"2026-09-01 10:31:00.000000000","submitted":"2026-09-01 10:29:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"996108","meta_rev_id":"80f87f9998258329837835646c607a9d102d2701","_number":996108,"virtual_id_number":996108,"owner":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-01 10:29:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2026-09-01 07:47:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-09-01 07:50:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-09-01 07:50:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2026-09-01 07:50:08.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-06 16:14:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-13 08:40:52.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-07-15 12:50:12.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2026-08-13 13:16:39.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2026-08-13 13:16:39.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"9243990db26b2836b77ddca2ec239e55617378c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2026-07-06 15:11:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8af86d7d510c484db403e3cea725fb215fb56f3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-06 16:14:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ade54d7e2164a0d971819013cf41b37\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/1a9bbf0d408e44ad8cc964f58facf4b9 : SUCCESS in 45m 44s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/f770c129a0fd4c57b87a03853539e76e : SUCCESS in 33m 00s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/26268c4d4ae64370883e1a063aee7b09 : SUCCESS in 30m 56s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/29788cf2874c4caeb4c867e15c90a1ec : SUCCESS in 26m 39s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/f6c6c39221db4ac18981475901ab5013 : SUCCESS in 1h 00m 24s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/27cc5130eaba41388fa09f366a9eaf41 : SUCCESS in 31m 21s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b5a00cd4d94e48ab9304b1ad1a0f2458 : SUCCESS in 28m 38s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/1095745e118445b98fcb03f5c6a9f781 : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da00a5dd3b224979b0a1cafeaf284cb7 : SUCCESS in 4m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/84712d1342184c74b19008f0fd83d0e1 : SUCCESS in 4m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/774c3c2fc6344c6ba03591ad4ae630ca : SUCCESS in 5m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d9aae5f3fc764dd9bb43cc31da618a5c : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7438ff18b20d46c681fab5caed27da9f : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b26d186315de4ea88495496edd93dfe1 : SUCCESS in 6m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/44005f6262e94371aa77fc68201e855d : SUCCESS in 8m 40s","accounts_in_message":[],"_revision_number":1},{"id":"884ee638a70a1f97240b7c661f6b41222997044e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-06 17:30:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a6c4ac1a460426e8561b7dd9f578b94\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/4db843b0d0ac42488a964cb980915b13 : SUCCESS in 28m 22s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/65b6ba7934cb4322bd77b0b30e050119 : SUCCESS in 44m 10s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/4517e112e47d40a79b5472ebcf42dbc2 : SUCCESS in 1h 07m 25s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/f868ddbe746e4a73b0caf8e40c6145f3 : SUCCESS in 44m 44s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/50d204b33f0240d6b0f67334e0b68995 : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ca89c20cecb141b188d07a9972537a53 : SUCCESS in 4m 17s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/94bd7ced1cee4e61a9e10d9b7fba960f : SUCCESS in 7m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"abe220678611a86804361876662f2a995bee9907","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-07-13 08:40:52.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"957bf306cf376f3d723489e1a90443254f4279fb","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-07-15 12:50:12.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4402a63cd794e2b25a549dcacb8a390d549848b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2026-07-15 18:39:12.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"83fea2ba49db6024e99f7d14a06c7ab0162f5f1d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-15 19:26:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c440cbceaf2404b92f4b4653456022f\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/4cd1ad8629524aadb1ce32ac0c4ec112 : SUCCESS in 34m 35s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/208e56d8edc94c98bd199fa413016040 : SUCCESS in 35m 40s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/5d8bbf5fa3be4ceb890b8177f820e45c : SUCCESS in 38m 39s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/82e34aaac4874d63a2048c9c739d1559 : SUCCESS in 36m 01s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/bff442c6af4c47ec8a4e09cfe7032b49 : SUCCESS in 4m 13s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/776a3b376b3f4982997e9987eec8502a : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d44dbaaece4a47cbbd13ba75155189f8 : SUCCESS in 7m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fa1050836a00df78a289ae43bc55021b495f7062","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-15 19:46:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d008546e8df460aa64fb265b43b305c\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/1dd7d0def37145a99cd236e564a02785 : SUCCESS in 54m 34s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/dd8c5811beea4ca487614386b3a22d23 : SUCCESS in 12m 40s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/060e6d2b46494547bd4d75db3a47a398 : SUCCESS in 22m 12s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/19aa584325c649548b14471cf944e076 : SUCCESS in 26m 20s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/b80843b5b7844df7981deaf0aab0b06c : SUCCESS in 1h 01m 08s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6e1a7391c44d4244b5620c4e3b6e2d32 : SUCCESS in 26m 56s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/37adcf689ed14251ae08258713fc1140 : SUCCESS in 32m 54s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/e7a81ac7068141388694a195f20226cb : SUCCESS in 3m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9dc6297cc074990b556e1e45aab30ec : SUCCESS in 2m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/99322f76d1014420a6e4c04f5095d34e : SUCCESS in 2m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5c31bab4013f42828695d8b9b72dafab : SUCCESS in 3m 15s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/65cacc7a7c5844ae9ed3449956c7f482 : SUCCESS in 7m 28s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/21fbc53f6aa54c1fa5368b29c7e62aba : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef8cfb6423b0405dba732d683a238162 : SUCCESS in 3m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9c10d79e7eb640e881e99fbbbbfe7abd : SUCCESS in 9m 32s","accounts_in_message":[],"_revision_number":2},{"id":"22e3d0760396b4ca5dadb554f60d863f53ce71ee","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-07-16 05:59:43.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"e5a9a09f80a480716f0a34fb75c6d1cee4f899b0","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-08-04 06:01:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bb8cdec91cd56b79c537e80d596d2454697c5524","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-08-04 06:01:11.000000000","message":"Patch Set 2: Code-Review-1","accounts_in_message":[],"_revision_number":2},{"id":"b5e4476c0a4b961c18b21b4887214567ffe185c6","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-08-04 08:28:28.000000000","message":"Patch Set 2: -Code-Review","accounts_in_message":[],"_revision_number":2},{"id":"1aa1855e73eb141ba568f8542a05ccfe855d5a2c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2026-08-04 17:32:43.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"5f6c90495f4a5bdc2557a970f5f17413b285267b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 18:36:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/086077cdde084880a4f0f68af59d4074\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/55020aced1b848218725aba3442b3347 : SUCCESS in 1h 02m 17s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/61e04784c6e4492db8f787a19e7e3fd4 : SUCCESS in 35m 28s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/c502ca4034724c1ca7edc92b6862d180 : SUCCESS in 28m 27s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/551dc70d41384dc79daca59cda49abe3 : SUCCESS in 24m 19s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/03c07a2f688e4e0bb0f9e8b381812a16 : SUCCESS in 42m 32s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a26796d29e3b48fa98ba91f42418850b : SUCCESS in 22m 24s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/525983e7b6d24d2fb2952c98ad9a8b13 : SUCCESS in 27m 05s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/e2370180edcd435abd5bed504158d571 : SUCCESS in 3m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3168f1a57b5d4c68a13271c3fed46f4f : SUCCESS in 4m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/24b9789aef9e4517bb796c857363765d : SUCCESS in 3m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a9a83d6b63c34607ab5914327b83da9d : SUCCESS in 5m 05s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d8213db5e17a40f7a4630fdaf5b5f9ea : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0fbb77569803412baa158e4b8f8b22ca : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7283e7c8a2d04ad785347d03a2271eb1 : SUCCESS in 6m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/95960e4191a44c36a13a88efd4e198e0 : SUCCESS in 9m 00s","accounts_in_message":[],"_revision_number":3},{"id":"6c727056638ede4892bb76ae929cf30b9adc856b","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-08-04 19:21:43.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"d4a6214c9066497b81beaad6cb8ee5d157f6db9c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 19:22:38.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd42f1c5126f4ff5b76f1d0c31e62c51\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/5c6037c77ee34cc0b5e75903852af6f5 : SUCCESS in 42m 01s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/7d11bdaf4acd49c5b23568faebcb270b : SUCCESS in 40m 17s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/442394cfe21e477d93b8e68db3d1b038 : SUCCESS in 35m 22s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/6568384c55724ef9803db832151b02c9 : SUCCESS in 39m 29s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/655cafff756c46269cc176942a238eb7 : SUCCESS in 3m 33s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/25097086d625485a97fe433d66e99819 : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/1356c65d2fb64efd8695e6aabcd38d5e : SUCCESS in 10m 00s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e0126dff835a2f398cf4c666e319ef4de618ec9e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-08-13 10:37:59.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2a32b280f975f98bd6852683765e3371f1440075","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2026-09-01 06:39:26.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"952102e76920ee1c5535bac6cf1b0268ab86abaf","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-09-01 07:47:59.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"199607a97d02ebf3b0abc08a463e3db26a1cb327","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-09-01 07:50:08.000000000","message":"Patch Set 4: Backport-Candidate-1 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"7b7cbc29b5231b5c951b2f51fed0aa932808ab60","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 08:10:27.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5a46d18644b41bebfb9904c2103b2cd\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/3a81ae4acd414228907571763db2f4d6 : SUCCESS in 35m 41s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/d3d63fcd66484c05bb660e82e9c8dd0b : SUCCESS in 52m 11s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/1a1884b245d74f75a836e71122a0973f : SUCCESS in 44m 34s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/a7a15e204d544dc09be10524907329a1 : SUCCESS in 51m 19s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/47e159a3afe54f47b226e9578f82b1eb : FAILURE in 1h 27m 21s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/690610789ca44210b624acb29248d856 : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4326adb052cb43d288e3a42376454c23 : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/2d67211621604c6d844fe9a935400543 : NODE_FAILURE Node(set) request 2cecc43d63f7493fa916bef673ca1b4d failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c3ac3e2e24dd67e57051758b5dc05d167de16136","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 08:37:20.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a1b3248b7084f138cc6bd4a398657d8\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/b1ba0bc168bc45dabcf2a267ed28b50f : SUCCESS in 24m 34s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/f608d85ba05f487088d519bc2d6eb23e : SUCCESS in 15m 04s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/a475bd4e923e4f9daeefd3de6003d937 : SUCCESS in 34m 49s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/333629f2d9dd4754aaec79ab6430e606 : SUCCESS in 18m 56s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/ed3d9ab473654788a275007fb34b0f28 : SUCCESS in 40m 42s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c2aa5238c42e4f1bb88029c5727c655f : SUCCESS in 31m 25s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/17eef7d0528e4d4facfc81ae57019c0c : SUCCESS in 24m 24s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/5a20bb03ecaf451bbcf97d4fc54426a4 : SUCCESS in 1h 20m 56s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/9a1f282a3e54441bb18f8244073cbd42 : SUCCESS in 1h 42m 58s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/b3974353f82141339b076d2c4c843d34 : SUCCESS in 1h 05m 29s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/00c0443d577d41989cd2ef9dc6ee45e7 : SUCCESS in 1h 23m 30s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2faeeaad32d7438698c5b91d332b0bdd : SUCCESS in 1h 25m 48s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/dbd52f61342447c7b455713e800ada17 : SUCCESS in 1h 55m 11s\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/1bed483389064d0bb97b6df2a61d1cdd : SUCCESS in 42m 14s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/498d89962d46429184d785d116112487 : SUCCESS in 23m 05s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/f5ac5c3ce76e45d4846ca4b315aca47e : SUCCESS in 15m 25s\n- kolla-ansible-debian-trixie-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/a74d5410fc7a4635b92ecc9e5fd7bd86 : SUCCESS in 38m 41s\n- kolla-ansible-debian-trixie-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/65fb8e9979cb44228e4389d0790f2b1f : SUCCESS in 29m 23s\n- kolla-ansible-rocky-10-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/e1165a3d54cb4739af8fdb3892847bca : SUCCESS in 20m 49s\n- kolla-ansible-rocky-10-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/0ab6c7768f0a40c48330b03cee0d789f : SUCCESS in 28m 12s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/e9d9f16e655f49f287c76ce2eb589dbb : SUCCESS in 41m 47s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/c8cb00504aca4ab68a8f68008797ca84 : SUCCESS in 57m 43s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/28d10008c2a140d192e8d91b581b7dd7 : SUCCESS in 2m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6614038683404377aa0553474f845966 : SUCCESS in 5m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/05d805aee284472ab878129298771bb1 : SUCCESS in 3m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5aef22418515495db3db436cf06ee227 : SUCCESS in 2m 31s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6ac5e2fad3d4479eb415d22d9e62e021 : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/147cc87dc0b741e28ae6a0cf304e599b : SUCCESS in 3m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84aeca471111440283b8a5eed2696948 : SUCCESS in 6m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3a45d531bbe84a6fa39b603a8217ec92 : SUCCESS in 6m 47s","accounts_in_message":[],"_revision_number":4},{"id":"67dd35bd965c757caa9a1fe004b2569b1cc35510","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 08:38:24.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"51cbb2c0e8e7658e88b391820e44d7f1cebe6c6b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 10:29:58.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0b0a3a1cda841be85f07005a4b7cefe\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/8aa9f938d6844461b0f9d160dec5954f : SUCCESS in 24m 15s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/04828bc7141e4a5d94d05942e051a409 : SUCCESS in 22m 38s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/dde681123ac24da19c558859b3e11e29 : SUCCESS in 29m 01s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/8adac3eb272245f9aefa692b9a73629e : SUCCESS in 1h 12m 50s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8d256c72d81f4c63a468b97dc7e9a09f : SUCCESS in 17m 24s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6969c6e5520346d58557df53a43026b5 : SUCCESS in 17m 58s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/56d6383c4c9144d1a079bb92d9569c7f : SUCCESS in 1h 12m 16s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/ec457d41e783453694b7eb791d3d17d5 : SUCCESS in 1h 50m 35s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/c9b74c6b561649a79bfe59f1f61ccd40 : SUCCESS in 1h 00m 25s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/4e6e14ff9b0a4bea990f3d686c32994f : SUCCESS in 1h 26m 32s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3163650c034846ecb6453c60b1269e8e : SUCCESS in 1h 11m 30s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/014c85a2059d4eac9bbe9258a30612f4 : SUCCESS in 1h 44m 00s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/d7852cafe37e4697859a5a10914bddc0 : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/592445051da74e94af269f8fa516ca0e : SUCCESS in 3m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ec71cebea2e14b8091fe047ed1e05938 : SUCCESS in 3m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2818cacab64b4a67a04985db7e27130e : SUCCESS in 2m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/302e395eab5341c18ee45ea007fbd45d : SUCCESS in 6m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/88af485d032444158e62f8c8960a1c45 : SUCCESS in 10m 04s","accounts_in_message":[],"_revision_number":4},{"id":"b10ef356935f4130b8e4a74495419c9566b921e3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 10:29:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"80f87f9998258329837835646c607a9d102d2701","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 10:31:00.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/536196fea272493abecc5041ccf350ae\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62f60b0ef4b8446b9305645e6e19a4d5 : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/55c4d6a512ce4001ae25cfa9c97d4521 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"d257a353e99792b329dc864a140a5636cbecf631","revisions":{"ce8b1ad9d5ea6f17b0ba21d663a6102a9241f98c":{"kind":"REWORK","_number":1,"created":"2026-07-06 15:11:15.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/08/996108/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/08/996108/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/996108/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/996108/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/996108/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/08/996108/1"}}},"commit":{"parents":[{"commit":"3c88fb847c87755a831ee955419aff94befd2562","subject":"Merge \"kolla-build: normalize mtime in archive creation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c88fb847c87755a831ee955419aff94befd2562"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-07-06 15:10:35.000000000","tz":120},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-07-06 15:11:06.000000000","tz":120},"subject":"Upgrade Prometheus Server to LTS version 3.13.0","message":"Upgrade Prometheus Server to LTS version 3.13.0\n\nChange-Id: I8de87eff00b226e8a85eca317af944e9b6c97b3f\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ce8b1ad9d5ea6f17b0ba21d663a6102a9241f98c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ce8b1ad9d5ea6f17b0ba21d663a6102a9241f98c"}]},"branch":"refs/heads/master"},"8a5882ad9636255f3ce2d895a5295f985c5ddc52":{"kind":"REWORK","_number":2,"created":"2026-07-15 18:39:12.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/08/996108/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/08/996108/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/996108/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/996108/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/996108/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/08/996108/2"}}},"commit":{"parents":[{"commit":"3c88fb847c87755a831ee955419aff94befd2562","subject":"Merge \"kolla-build: normalize mtime in archive creation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c88fb847c87755a831ee955419aff94befd2562"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-07-06 15:10:35.000000000","tz":120},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-07-15 18:38:53.000000000","tz":0},"subject":"Upgrade Prometheus Server to LTS version 3.13.1","message":"Upgrade Prometheus Server to LTS version 3.13.1\n\nChange-Id: I8de87eff00b226e8a85eca317af944e9b6c97b3f\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8a5882ad9636255f3ce2d895a5295f985c5ddc52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8a5882ad9636255f3ce2d895a5295f985c5ddc52"}]},"branch":"refs/heads/master"},"5aa04550e6ae96adb2533e5936db3041e01ba1a7":{"kind":"REWORK","_number":3,"created":"2026-08-04 17:32:43.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/08/996108/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/08/996108/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/996108/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/996108/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/996108/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/08/996108/3"}}},"commit":{"parents":[{"commit":"8e912a9db7f013a1abfc1e70a829e435d50622db","subject":"Merge \"ironic-pxe: don\u0027t fail if symlinks already exist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8e912a9db7f013a1abfc1e70a829e435d50622db"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-07-06 15:10:35.000000000","tz":120},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-08-04 17:32:40.000000000","tz":120},"subject":"Upgrade Prometheus Server to LTS version 3.13.2","message":"Upgrade Prometheus Server to LTS version 3.13.2\n\nChange-Id: I8de87eff00b226e8a85eca317af944e9b6c97b3f\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5aa04550e6ae96adb2533e5936db3041e01ba1a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5aa04550e6ae96adb2533e5936db3041e01ba1a7"}]},"branch":"refs/heads/master"},"d257a353e99792b329dc864a140a5636cbecf631":{"kind":"REWORK","_number":4,"created":"2026-09-01 06:39:26.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/08/996108/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/08/996108/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/996108/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/996108/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/996108/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/08/996108/4"}}},"commit":{"parents":[{"commit":"201438c78025e1364cd0e874560de14065bee428","subject":"Merge \"letsencrypt: Add ACME dns-01 challenge support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/201438c78025e1364cd0e874560de14065bee428"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-07-06 15:10:35.000000000","tz":120},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-09-01 06:39:23.000000000","tz":120},"subject":"Upgrade Prometheus Server to 3.13.x LTS","message":"Upgrade Prometheus Server to 3.13.x LTS\n\nUpgrade Prometheus Server from the 3.5.x LTS release to the\n3.13.x LTS release.\n\nChange-Id: I8de87eff00b226e8a85eca317af944e9b6c97b3f\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d257a353e99792b329dc864a140a5636cbecf631"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d257a353e99792b329dc864a140a5636cbecf631"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-1"],"failing_atoms":["label:Backport-Candidate\u003d1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
