)]}'
{"id":"openstack%2Fkolla~997944","triplet_id":"openstack%2Fkolla~master~I5973943fe83bbb00cfc8623c0cd4d4b98b369315","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-09-11 13:48:44.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-09-10 12:21:32.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2026-09-10 11:20:49.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":["kolla_drop_centos10"],"change_id":"I5973943fe83bbb00cfc8623c0cd4d4b98b369315","subject":"Drop support for CentOS Stream 10","status":"MERGED","created":"2026-07-20 09:17:33.000000000","updated":"2026-09-11 13:52:24.000000000","submitted":"2026-09-11 13:48:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"997944","meta_rev_id":"d1d35e0ce8bd40020bb0e6d3e9c7ad06363a87e5","_number":997944,"virtual_id_number":997944,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-11 13:48:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2026-09-10 13:44:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2026-09-10 12:21:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-09-10 11:20:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2026-09-10 13:44:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":-1,"date":"2026-09-10 13:44:48.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2026-09-10 11:59:33.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-20 09:51:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-31 21:01:36.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-08-03 13:23:53.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2026-08-03 13:39:05.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"d929317e7a62bd9e08937dcbea74ba8992fd945b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-07-20 09:17:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a7c6844240bd7766865c652fd9614d859069fc7c","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-07-20 09:17:50.000000000","message":"Hashtag added: kolla_drop_centos10","accounts_in_message":[],"_revision_number":1},{"id":"e9a96061c2917c62ce7b6a40492100eb16aae16e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 09:51:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4cba248bba4040f385bb27a30e7cb544\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/fc35e0e38a7d4097b902f275b28bf396 : FAILURE in 4m 03s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/97f9150605cd4d99accf9f5055d70bdd : FAILURE in 5m 36s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/3bf98f9897b24e0397aa52756106d910 : FAILURE in 2m 59s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/84b05956994e4a72b77fa0fe1a5ff008 : FAILURE in 5m 29s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/254ec1afc56145868a5722909ed36cd2 : FAILURE in 3m 51s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8a5d8d8f33c347f3a081dc290a6fcd17 : FAILURE in 6m 35s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/8acb76316ffd4a18ac306b92526af3a0 : FAILURE in 11m 30s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/6e87d194e06348a38cdcbd4c970635ed : FAILURE in 11m 18s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/203eed2537df4b0f965c3b2498d6b608 : FAILURE in 7m 10s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/7f86ad2f45f346af8840a4725c50c579 : FAILURE in 10m 23s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f9e47ea9ad224315a2936c3bab70ad20 : FAILURE in 11m 31s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/38ea56762b9845bbaa276cc1e122e988 : FAILURE in 11m 12s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/068abb266041401990831de323514097 : FAILURE in 12m 11s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/7487b113699241d08d489a1aed2adef9 : FAILURE in 9m 31s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/442671abe0704e44b70fc240c276524d : FAILURE in 9m 09s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/fa77d65542f24bfe8b04a3ebacb370ba : FAILURE in 9m 40s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/cc0a803db18144f9b9ca127fef830f56 : FAILURE in 13m 10s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/7a60f19dcaad4b5a834291fdddda9621 : FAILURE in 8m 09s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/7eb46eea9a464b1197b06fb05d07070f : FAILURE in 8m 45s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/76150a772c444d6fb3c7937f3c6cf6f9 : FAILURE in 8m 53s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/b68906bd525d49d580f32dc03f76e223 : FAILURE in 9m 27s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/4051420da21941ca9c69f22258196380 : FAILURE in 8m 14s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/0c7fc4a74751457e9ec194553e11af6a : FAILURE in 12m 36s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/2088c36608984512b1acae0d866a3097 : FAILURE in 9m 16s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/39de3ea849564436b51b9888c930e1d3 : FAILURE in 7m 52s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/e193d1fa4d95417cb1e291efbbe866fc : FAILURE in 9m 33s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/c1ee43587c554b7c96891b67c888b863 : FAILURE in 7m 56s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/60880451c93d4f6780915389bf5a845d : FAILURE in 8m 02s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/b8981f4844264f6d925108497a1b6113 : FAILURE in 9m 04s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/b266a5037814415f877f8c05ce532b2e : FAILURE in 9m 21s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/645ac40262b745dda0f400bbd7bc045e : FAILURE in 8m 17s (non-voting)\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/415a76261fd44453b894222aaed358f6 : FAILURE in 8m 02s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/8768ff991e4940ecb6f888f057bea926 : FAILURE in 6m 58s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/f7938185358d4a1d859cd588d9539cb8 : FAILURE in 6m 05s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/61dc577c572c4fee9f4e35bb1a9e7b9c : SUCCESS in 2m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1045854439d0445d8c40ac8c7cfb2681 : SUCCESS in 4m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4144a40401394c32b06669fdbc8cc279 : FAILURE in 2m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f0f6c1a8b9aa488db9767ac812176965 : FAILURE in 3m 08s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/17794ecb4e16434aad4fdbd82ed1b702 : FAILURE in 5m 43s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5373c1f90b8c46da86e084a635186bdb : FAILURE in 3m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d780bf58d064ac29cce1587a08e3637 : SUCCESS in 5m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9da14a8803604ba1a015e130f6875421 : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":1},{"id":"c7ae6878b429ce83bdc28e4581a7a0df767daf98","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 11:35:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2543418283ab4cb5880f4cf43b5fb15d\n\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/e571cfb5f902428a8a2dec7b63130229 : FAILURE in 5m 08s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/b1100168adfa45f4ac5cca9cfe261fe5 : FAILURE in 5m 26s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/e6cbfb8d61ad4590b5588da26cee5d8c : FAILURE in 5m 20s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/4fdd7ced48c84507b77e1d55ca1b78ab : FAILURE in 13m 01s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/749c2a5a85d046878588b773b95f3270 : FAILURE in 4m 21s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4c80f248b7d5477a9155e63d3fa7feab : FAILURE in 3m 43s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/4a7956cd241740c0b1351fce8054559b : FAILURE in 8m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b232f32949500dc5b71d8eb3f667010aba058b68","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-07-20 11:38:03.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"557e807785ebb40bedcb9cacb3bd860be662c24e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 13:32:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4c9dd4f70b5d44bebecd59ca688ce4cc\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/5ca86aea070441cda86c13aff4a963ca : SUCCESS in 26m 49s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/eed8f4631d9645f59ce1ba94077d57dc : SUCCESS in 37m 18s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/c18f534c94ec44a3966c2c1f1ce9025c : SUCCESS in 28m 02s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/3273c5d9e1aa4d1093d6973e2f68a512 : SUCCESS in 50m 29s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c11b4ea60e6f433d8ceddc4062eeeae3 : SUCCESS in 24m 34s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fedb8e17efc9409688261baa6b5edd5e : SUCCESS in 27m 12s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/41005a48909d41678ea7d950d1cf1b69 : SUCCESS in 58m 06s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/00fffbdded72434cae6bed1b0f7daeed : SUCCESS in 1h 34m 31s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/8e10db6fb8b74b48b9f27ad137a34477 : SUCCESS in 1h 02m 51s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/10d2c8d98ee9422c9b4a20dcd5f81e99 : SUCCESS in 1h 26m 42s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2d09d08921744c109a2ee3f93ee7de9f : SUCCESS in 1h 06m 03s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/1fe6a3ef69704daab9a8d2549950e848 : SUCCESS in 1h 15m 54s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/77a03588d9b64347a34743094575fb58 : SUCCESS in 57m 44s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/df393dc2b5d047cdab24e0018a8ee6bd : SUCCESS in 1h 30m 42s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/56a3a5b8eb3647fb8aafc0fd291d5288 : SUCCESS in 1h 38m 26s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/1bf47ad2cb0044f58805d4157c01a9a9 : SUCCESS in 1h 06m 45s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/01be47ecc3ad4e8ba50cea3d42731351 : FAILURE in 1h 38m 34s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/ed7acd8d82224304ad5477d09052ea94 : SUCCESS in 1h 26m 44s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/10a4f8fa79da417fb12cc090bc8dce83 : SUCCESS in 1h 01m 27s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/529f006ef8f14edc8f52a8382628c447 : SUCCESS in 1h 00m 00s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/85700a54cbda4aabaf693c92520e205d : SUCCESS in 19m 16s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/5ae776b17031405f87024fe25356f6e9 : SUCCESS in 17m 26s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/cdbbad89afcc4bc5a548e95a6e21b931 : SUCCESS in 30m 58s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/a3c147fcd43846c18ce68bdbe718a71c : FAILURE in 51m 11s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/bf1ee7e812c94816a3c96d8d62da0994 : SUCCESS in 1h 15m 23s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/8d6e18084ea84cdc8e67e2909751e79a : SUCCESS in 1h 34m 03s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/d33709fb4edc4b9890cffc2b2a085596 : SUCCESS in 45m 12s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/5f468918d0004606a993b11a57a65157 : SUCCESS in 1h 11m 05s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/ee0ede51d18c476f97d98b49e100b648 : SUCCESS in 1h 12m 01s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/54ae32965a3a4b8ba72e7a55bb7723f5 : SUCCESS in 1h 20m 39s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/07b67c6aea0248c78e93d0bfb5bb9fc1 : SUCCESS in 1h 10m 00s (non-voting)\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/e3085567e3aa423f928decfdd6734922 : FAILURE in 1h 08m 19s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/79ef05a19ba446f7910367c020616f76 : FAILURE in 1h 08m 57s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/639ebb2fe0e74908930fcb7d18a19a58 : SUCCESS in 1h 24m 38s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/b759467c98bd4702b25089482d6a338f : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d3254b9a8c0420b8ededeeeec752310 : SUCCESS in 4m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ff81aa99cdd24d9d99c7c38e50882624 : FAILURE in 3m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fbe4f5966e4949f784db04a911ef0028 : FAILURE in 3m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d77a419de65b4080bb2aadf6c62de2e4 : FAILURE in 5m 55s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3dedfd4a7561409d94a82fb2c1867859 : FAILURE in 3m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf334bdbae9b415d84680ef1490e9e4b : SUCCESS in 7m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/014a5abc997f46599fdbcdab99fe4430 : SUCCESS in 7m 40s","accounts_in_message":[],"_revision_number":2},{"id":"a0551e40514d149ca3efe10a3af586e78fa1fc42","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 16:18:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/11b1c3fddcdd459488351a943da8e78b\n\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/9bdf355b2d6a44ecb1f5398cd1dd6ed9 : SUCCESS in 45m 48s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/32b17814fa50450296f345fdd0e8dd3a : SUCCESS in 40m 20s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/1b9ef6d025744c8a93f8b167c5208a9d : SUCCESS in 38m 55s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/32fc2d0851bc4f50abc2fd53dc550538 : FAILURE in 1h 34m 11s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6ce72c6187ea4f3cbc6c0169684268e9 : FAILURE in 3m 51s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6751a8c53163439780efd3b6ee7b56c7 : FAILURE in 5m 10s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/64ad01699d93405a9f74a2ca784d5011 : FAILURE in 10m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"02e9e118b28f10775b11fa08bd00eaea50b19ba9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-07-27 15:08:27.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nThe following files contain Git conflicts:\n* doc/source/contributor/versions.rst\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"7eb3a3c767b08ec4a0175e5611fae271722479e0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-07-27 15:08:50.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"8d3635bf8c750a38585c832d2e24b173591c8a44","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-07-27 15:08:54.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":4},{"id":"5cc2a70bcee2645a9b17c038c92d65f2c85563a9","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-07-27 15:08:59.000000000","message":"Patch Set 4: Review-Priority+1","accounts_in_message":[],"_revision_number":4},{"id":"03028559b298ca74bcfa8c4a03765e42586a8dcf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-27 22:17:13.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f9fe1ca4a2d4aec905ee287297c8a13\n\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/30825479885c47e78d1fae49dd11d3aa : SUCCESS in 47m 16s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/de82262843cc482cac59c8512d83075c : SUCCESS in 47m 46s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/6bba7c8478c148869ea01e6ddd2847a4 : SUCCESS in 50m 12s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/7a3189ca94ee417db7e93b07c9788eda : FAILURE in 1h 26m 28s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/197788ca478e48209a1af6d667d3b744 : FAILURE in 4m 08s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/eead24c4fc9942af86b7a43541c5c07a : FAILURE in 4m 27s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/877573cde94f4e85a615d41f723ad51e : FAILURE in 8m 56s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5b50b7dd467556ce4d4af4ff8aba7af98817c35a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-27 23:06:18.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7afa00713b004cf0a74327f3fd6fc69b\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/af6ab8e2c00d4d5b9fcaf2b59746b803 : SUCCESS in 23m 57s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/896f674dbafd49ed891ae140b663913f : SUCCESS in 34m 08s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/66b5eab770034743b968879e6a2ac830 : SUCCESS in 21m 00s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/d8b1c3f0bae846acbb8bbe85b6411bde : SUCCESS in 37m 35s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3576d72362fc4027bb5d5420f6d0cb45 : SUCCESS in 30m 43s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f534167714c54e6d8d318ffeb627bd85 : SUCCESS in 36m 44s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/211b0d5b85f044e9a7a181dc6793471d : SUCCESS in 38m 30s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/da232fde773642fb884f9b5d4d612d1c : SUCCESS in 1h 12m 38s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/f5653edc3d0b4215a3f7db68fd68dada : SUCCESS in 43m 57s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/dde1e8d5ca0e40b8b2cb683726ccfc6d : SUCCESS in 1h 17m 53s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a568a2fd049f4eeb978a7b7f022d8919 : SUCCESS in 41m 09s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/7370ac43bb164659b6eddd141462ab87 : SUCCESS in 1h 11m 40s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/6ef8e7b953e644bc9f0487b0126d5945 : SUCCESS in 57m 09s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/9b71bd302c094a66b733a8e7bf0015b8 : SUCCESS in 1h 10m 56s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/fbb8a780d0eb42cf9de24eff484ce13a : SUCCESS in 1h 50m 58s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/21a9011bc6624706b4d397f3ae180f4f : SUCCESS in 1h 00m 47s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/35bf2a66b41e4ec9afe3d06571604245 : SUCCESS in 1h 48m 03s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/547ed35787bc40fbb58f5579dbabfd84 : SUCCESS in 49m 49s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/4857db64fb1a40f1accd3e5cc97a0b6d : SUCCESS in 1h 04m 54s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/2ac3d3ba39564ade987733cd1574fa3b : SUCCESS in 1h 29m 17s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/cc661bf448e245508ad383254a9d02ca : SUCCESS in 16m 05s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/3afd6e9b39cc4997b2cf2e4bb805bd57 : SUCCESS in 16m 33s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/53d73407c24d4c59a7bc84ef6dd58a83 : SUCCESS in 32m 06s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/b6bc9cc15e394492b7c8a79fe496ab22 : FAILURE in 49m 37s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/18d29c79f71842bd972ea289cee287a1 : SUCCESS in 1h 05m 52s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/9902a13b4ac342f1a007baf03e45ddb7 : SUCCESS in 1h 17m 16s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/08f398c99d7749dda240271406d16aa0 : SUCCESS in 45m 48s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/d275020a014848e0bcfaf97d57038e49 : SUCCESS in 49m 07s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/50dd64cab7b04e43851add1f05e91e5a : SUCCESS in 1h 16m 11s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/03b3f9570fc4489f8cf2f9ef17d20033 : SUCCESS in 1h 20m 02s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/d35abb84c401450b930dbbdc9694266e : SUCCESS in 1h 07m 48s (non-voting)\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/54e99a9ea972492e8b8c9ea682a5d004 : FAILURE in 1h 18m 51s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/813aaa4ebb2c4a049f762d7061bd3679 : FAILURE in 57m 42s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/c59d58bd507647cf91831d66fc98d7c2 : SUCCESS in 1h 25m 38s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/ca0583380a93428783548754ec3ed6f6 : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5470f81b47fd49d7ad37e30413ae36ec : SUCCESS in 4m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/702e9c18acb443cabd17324e833349c6 : FAILURE in 2m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f5c26553891c4d19bd41a115f665456e : FAILURE in 3m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/49ec23c3ea46466295004bb1bd693192 : FAILURE in 5m 58s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5b9654d0ea5343418250e1d963d721a1 : FAILURE in 3m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b418a14a258341e4b8cd22ebc0ba2460 : SUCCESS in 6m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c72ffa087af643c8b3fa0acd2264e6a5 : SUCCESS in 11m 12s","accounts_in_message":[],"_revision_number":4},{"id":"54eedc7663ba5c5240c68812114c9d8465e90664","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-07-28 06:43:02.000000000","message":"Uploaded patch set 5.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"3af0ab337044d3f81e9f3cc8f7e5192a42314fff","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-28 07:26:12.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8190a8e81dd347af8326b5cc6369c95b\n\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/4b2c3ce719e84421b8accd72dc214a92 : SUCCESS in 28m 48s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/a848bf17a26f42779952da803222202d : SUCCESS in 30m 33s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/d2430fb373054bb3ba48eac360058790 : SUCCESS in 30m 04s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/b27cd0410a134654904db8f9b8716d24 : NODE_FAILURE Node(set) request 79cd082b789f41dab1fe4a273f520597 failed in 0s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/27352d4acf57427da911581f438d2101 : SUCCESS in 4m 39s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e10ab38c3a894ca693cb436d141727b4 : NODE_FAILURE Node(set) request 82a1a81275694002b30d3138cdbee1ee failed in 0s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f12281eea43a464a8a70298d31ec1ce4 : SUCCESS in 7m 49s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"212b96e415bba8deff75d01fb2d2fade8184fce6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-28 08:58:13.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d6449c7b8d6d4bc092e790b5083c8b4b\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/af444697f1634f29a208490eb18b7a10 : SUCCESS in 28m 42s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/fde927a9d81f45a287f43db0d969cb39 : SUCCESS in 22m 02s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/f61c2873b501456d88fcbb8e18bcc9ac : SUCCESS in 27m 29s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/89929408e6e644f69bc557ac2f74f733 : SUCCESS in 41m 00s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5bb08f09d7424d65a7b860537cdc8ae3 : SUCCESS in 24m 02s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b2553ca0dd7745d8b68a8f84c05ea240 : SUCCESS in 29m 19s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/980637c247d1481090f2ee688a946bf6 : SUCCESS in 37m 45s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/384bdd021cce4ba585d22090e86f5eb0 : SUCCESS in 1h 32m 06s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/7b37b2feed93436f8901c5868ca3d362 : RETRY_LIMIT in 5m 29s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/df5b867a1c7b4c2c975f8c6db2dd2a29 : RETRY_LIMIT in 5m 17s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4a96c284dec249bdb81a5c12edc63588 : SUCCESS in 40m 31s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/10d67240bc474ed3a47bda5feacf5712 : SUCCESS in 1h 39m 24s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/e02fa7b8c61c4423ad87ab3ac9a2c1c8 : SUCCESS in 1h 28m 01s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/2cbb8dd4b6a04fa6b6521cb98f358b22 : RETRY_LIMIT in 7m 58s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/d86054d20b4247cd964459471cd6fb7b : RETRY_LIMIT in 7m 58s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/53a2c842f81b4955a2913727ffbb4fa9 : SUCCESS in 1h 15m 23s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/5c0bd0bb9598476087fabdc25f67a28b : FAILURE in 1h 52m 00s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/2eda1a679bc4403f913426432a924386 : SUCCESS in 50m 00s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/684b968cfa9346019cdce839e8301cdd : RETRY_LIMIT in 5m 18s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/5d8a6791c2594438a3c96a583b9bd39d : SUCCESS in 53m 00s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/b2c92e13f42c4f1d8214b1c7fd6f4234 : FAILURE in 26m 49s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/e98e8f57163840988276a55082093690 : RETRY_LIMIT in 6m 59s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/7b8d14719ad14092ae6ff3d291b5e525 : SUCCESS in 31m 12s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/a753bcb1cebd4e1fb0f586e891ca6d23 : FAILURE in 51m 41s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/568c324be740460f90ad019b507fa187 : SUCCESS in 1h 25m 31s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/35356c0033bf47b7ac54bf04143b6b74 : SUCCESS in 1h 19m 07s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/f4e558c4fa124d84af7ad34625dced50 : SUCCESS in 1h 02m 05s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/e9b830f24a064721b2d3e8cc6be7697d : SUCCESS in 1h 15m 08s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/a75df7095f174de699029f33eef7c127 : SUCCESS in 1h 03m 43s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/7e642e30a0474af89ec380856a67d381 : RETRY_LIMIT in 6m 54s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/47592150c7c4458e8bb490c21340180b : FAILURE in 52m 58s (non-voting)\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/ce1099c455da4187aefc8b66ffbad15b : FAILURE in 1h 12m 11s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/aa61007d4ef3471ab2d4fcdbd07b0ba2 : RETRY_LIMIT in 5m 18s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/6c5c72b3a7c4406d839c257d858a30f4 : SUCCESS in 1h 24m 59s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/cda2a860731d4b21b53f9370214f3196 : SUCCESS in 3m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d53bdca6438943319dab0bbe8f8137cb : SUCCESS in 3m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d7ebb87592b24d449c7f78ce03199bdb : SUCCESS in 2m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6ebc8e98938c49b28d44f7e436c497ea : SUCCESS in 3m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e3f45501f2754b67a4b3939740554c8f : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/90a15bb2c9df412ab71c342c7be8016c : SUCCESS in 3m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e28b85502b0241b3967d4cbf7c038309 : SUCCESS in 7m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9bbd42372513407dbf3af443b4ca5b6b : SUCCESS in 9m 33s","accounts_in_message":[],"_revision_number":5},{"id":"3e73866d31ac4dc168887cfe418cac817288fc5b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-07-31 07:08:09.000000000","message":"Patch Set 5: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"88a66d045249fed5db00ab9aebc44ad39371987e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-07-31 08:47:39.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"df05e3c9162893449551fc3305889b999c9736c9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 11:17:05.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a7147e46d41432cb564bf4010b316b0\n\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/1b50df9c58dd4970b9c0378c510a54c3 : SUCCESS in 55m 13s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/946fa20d656045a1a5684e844a420951 : SUCCESS in 1h 03m 34s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/f07a5f84808547778c6c46e3c06a9de0 : SUCCESS in 1h 00m 04s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/64eb9c26febc42cea56b4396a46ebe27 : FAILURE in 2h 27m 10s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b33623a7c41e45118a306a26b17243b1 : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/936770b23f1a47cea37f824634476d1a : SUCCESS in 5m 40s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/2b5a4b5c46aa4ef8bc3a2bd2c6b18aaf : SUCCESS in 9m 00s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"b5147af63d808589ccbe1d2e5aa5b3485d6108fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 11:21:34.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a2eeba9ce8846389ea51c0144ce5c07\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/d5b5e417652144af822be1b2808a224e : SUCCESS in 18m 57s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/04db5cc539a94e88a12dcd425cdf8480 : SUCCESS in 20m 21s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/9df64653625c43d9973b11dd490c5ea2 : SUCCESS in 16m 20s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/3d588c2859ee4aacb734e29c44137f39 : SUCCESS in 40m 12s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/155ee33e5a3e45ff8eeb8a8aa5316db0 : SUCCESS in 24m 57s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9794b9542ec743098923bcc2be6ea78a : SUCCESS in 27m 15s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/4ed45d05db8c43848d68f926a78cd022 : SUCCESS in 1h 04m 01s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/37770045569a4dacad89056d50dc7eed : SUCCESS in 1h 37m 14s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/aadacfe6b1aa4d1698542e901237184d : SUCCESS in 1h 01m 07s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/e1ccf6e1146046e0930137f16275e906 : SUCCESS in 1h 27m 08s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a60c48d9bc8742fe8b3d0005b174dc54 : SUCCESS in 1h 05m 12s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/e0585461a3f1458e9c082165b6666a29 : SUCCESS in 1h 33m 47s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/e75fdaebbf104c93a2b32dfdebe14286 : SUCCESS in 1h 26m 30s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/331b291edc38489e8bcdefd38fa605a0 : SUCCESS in 1h 36m 22s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/45052f8b388240d4b541820970743cef : SUCCESS in 2h 07m 43s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/f3a3e070a9644502a5cb9c37175573e5 : SUCCESS in 1h 33m 52s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/552a5154cd1e4806bc7c58e8ac67fbcf : FAILURE in 1h 32m 11s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/6b308cfad07e4f9188a478ba271b5010 : SUCCESS in 1h 29m 34s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/7d36d312056141cd9484346af043dab3 : SUCCESS in 1h 14m 03s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/1b15706d1aed4021aeb0b147dcc9a00d : TIMED_OUT in 2h 31m 50s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/005484e338d0435e850af859a166f254 : SUCCESS in 23m 44s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/22f015a58eba44df92f4f349f3f77c10 : SUCCESS in 29m 14s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/57371845926340e48e69146ee63d85a2 : SUCCESS in 21m 50s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/95a40c125f86422b85127dd04897ff74 : FAILURE in 51m 01s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/3098fa663b1641998b3b07ebcc4ea6a2 : SUCCESS in 1h 25m 32s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/e75716d19662468a87a7d61c1e4b69cc : SUCCESS in 1h 29m 23s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/181c9daca60248029c9d5b323b747a7b : SUCCESS in 1h 03m 49s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/56fcb3f322f6467c9e5b2c6204e7a5a9 : SUCCESS in 1h 03m 30s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/93bcfe48362e4b18809e9028d81d9cfc : SUCCESS in 1h 32m 11s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/b626c313616046f3944992100a0d97ad : SUCCESS in 1h 37m 09s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/3b792c294c4140c9a02948451c960fe6 : SUCCESS in 1h 50m 25s (non-voting)\n- kolla-ansible-debian-trixie-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/2c10a9007b5841fca6f4ef2bd6732409 : SUCCESS in 34m 26s\n- kolla-ansible-debian-trixie-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/d928a5c0fe1746fe8c2e552b647e785f : SUCCESS in 29m 07s\n- kolla-ansible-rocky-10-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/248517fa68464a4bab9df06740c3ccff : SUCCESS in 21m 53s\n- kolla-ansible-rocky-10-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/2b77fefa699f4ef58ceaba4d87510fc5 : SUCCESS in 28m 54s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/ea03cac9b8164c0993331b3523ac1b5a : SUCCESS in 20m 47s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/6794faacf505429c9e075d8aa64ef852 : SUCCESS in 46m 32s\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/ef741a6a225a4474a8eaaa03933c3a86 : SUCCESS in 40m 25s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/7f6a3ed520d9412599fd0c7aca63322c : SUCCESS in 48m 54s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/dd9ad5cb458e47498924308eaae73399 : SUCCESS in 1h 14m 09s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/01562583146c4cbf9c99aec08b12f9bd : SUCCESS in 2m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2785a29e27c04efe9cbce9b4f4b67a3b : SUCCESS in 2m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ec8dd9e6013540d8b4916ad1deb74166 : SUCCESS in 1m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1637cd0fc60a4c49b14d0894c571f6c6 : SUCCESS in 3m 17s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d907a0e35c0c4b98aa43623ef5f63bb5 : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7cff1c9887524ebbaf718df887c9e00f : SUCCESS in 3m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d499a073969a402a844fae7589973f98 : SUCCESS in 4m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1af53d17592b4ccf8fc7fa8603281ad8 : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":6},{"id":"ea007ee261cbab6c5817b26d96078931a12859ea","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-07-31 21:01:36.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"44b510707118543e8175b20f763760e3da3985a5","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-08-03 13:21:49.000000000","message":"Patch Set 6: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"8b09d5260b305a4067cbffab4cb5dbe44c2d4f48","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-08-03 13:23:53.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"1c99e8dfc0f120cfe29c89db7fcd42f054c04a66","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-08-03 13:36:00.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"cf78e45ddd0be917833ec3206f2c426c584634e7","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-08-03 13:37:00.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"08fe9643d82a4419b3acd7f24fb579925b189860","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-08-03 13:39:05.000000000","message":"Patch Set 6: Code-Review-1","accounts_in_message":[],"_revision_number":6},{"id":"48131a7e9fed96744dd80ba7df7d9e696277fb8c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-09-09 14:07:10.000000000","message":"Uploaded patch set 7.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"f60525d1a3d975ebb625b879b6f6ad0944ff63c9","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-09-09 14:27:40.000000000","message":"Patch Set 7: Review-Priority+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"ad873e4dab17fd2d5bd9730e3eff24bdba8252af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-09-09 14:29:55.000000000","message":"Patch Set 8: Published edit on patch set 7.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":8},{"id":"cf7d9a488a678b19813fca22421cd216246f8ece","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-09-09 14:50:45.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"d316eaeb61e2c649a12829263c49ce5c93d9cba6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 16:50:49.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7fea54280caf494396d9c209cc5e94cf\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/ea7f48826d424128b509f6f2a7006089 : SUCCESS in 35m 52s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/7a354b7641754069b97cde5f08eadfd5 : SUCCESS in 40m 14s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/b8af41c162a6400092512bcccb5c757a : SUCCESS in 32m 48s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/5884657c103449df812613f5ca18d55f : SUCCESS in 1h 05m 05s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9aedc144505140979888b2afb9756086 : SUCCESS in 36m 30s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c6438ed4624d47b997851711076c8c9e : SUCCESS in 37m 36s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/346c297dd12c436f9240c07a1d51fa38 : SUCCESS in 1h 17m 37s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/f8cfd866703a46ac95545fd2f58e1f43 : SUCCESS in 1h 39m 54s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/4c834a40d4ac45e69f364d7d3d184186 : SUCCESS in 50m 39s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/cdbe6e08a6b243fc870d75fc38ff04ce : SUCCESS in 1h 16m 01s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cc51fdbe59a3427ea6392f1405495fee : SUCCESS in 46m 04s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/3d39d0063ccb4e7abce4d8ad060a392e : FAILURE in 20m 33s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/e25871c5fe6343978bd831d6a5c3016b : SUCCESS in 1h 56m 34s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/5e321d1976df4e59b72f8eb1db12ad3b : SUCCESS in 1h 31m 34s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/c19b331ee5ee4930b738c31a7d518119 : SUCCESS in 1h 55m 34s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/11421f12d73149409a0e948e96932d54 : FAILURE in 55m 11s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/4a38bf697c4f49009ab33c9098bddf0d : FAILURE in 53m 27s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/e5eae5b596ca44bba6690b9fa3a26fee : SUCCESS in 1h 22m 53s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/9bae01c2290648ae8adfdadd74288a36 : SUCCESS in 1h 19m 14s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/0d6059aaa9ee422a871bdf47f91f3e38 : SUCCESS in 1h 28m 45s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/30c2273b764840289bc267bf2fbb74f1 : SUCCESS in 50m 52s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/5f49f48712d6456cb05aba7fed72146a : SUCCESS in 24m 57s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/f87a692e7db74d4f9e8d7dad59fcd914 : SUCCESS in 35m 05s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/fae8886fc9294427b88241b41b63a311 : FAILURE in 54m 23s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/445bad949b354b3a9000b3b741ac1552 : SUCCESS in 1h 18m 26s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/47e1e019c6c04730ab16e52e8e9c165b : SUCCESS in 1h 24m 53s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/b33114536645431ea96dc7ffcdaa4f77 : SUCCESS in 48m 38s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/c993871e57ae41e886c2bf7aa640ee04 : SUCCESS in 1h 18m 17s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/1ca916cce22c4729b6ae720de8883ed2 : SUCCESS in 1h 12m 11s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/c6cbbf63126145048bc201de68a21163 : SUCCESS in 1h 42m 34s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/aa8ab4eaea034902bb0edfa11908a052 : SUCCESS in 2h 00m 05s (non-voting)\n- kolla-ansible-debian-trixie-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/e73361ca25b7468fbd81867a46db8df2 : SUCCESS in 44m 09s\n- kolla-ansible-debian-trixie-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/998d574cb14b4401be128b4b471a2a36 : SUCCESS in 1h 01m 03s\n- kolla-ansible-rocky-10-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/acd18e7b3e7e4da5b42136df9e0fb309 : SUCCESS in 38m 01s\n- kolla-ansible-rocky-10-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/7b75c2e8a9ce4a7b9897121cac3fba14 : SUCCESS in 54m 10s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/5366e868761f427d94f0c359687804cc : SUCCESS in 41m 58s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/91df2a4831324a54aaa19e234022b8d5 : SUCCESS in 59m 48s\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/f9657a8f70c344de82ae5f0e8479377c : FAILURE in 55m 14s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/4114d8738d8b43808226739164b8e172 : FAILURE in 1h 15m 56s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/04e185ae325f4381bba13e69893284ff : SUCCESS in 1h 27m 39s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/0f742562d49f42ee8f9fec33ac0090aa : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cd6fb7838c84fc6a9b25d02b8ae35cb : SUCCESS in 5m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/50159861b5f44b249fdb822815828b64 : SUCCESS in 3m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a35b6ef8f1c44de4ac36b916c583efef : SUCCESS in 4m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fc38f6a7ab174d90b8167e68c95164c1 : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e83d2c69d3634037b63f7f2ff612faf4 : SUCCESS in 2m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ef492eedc26438e8786d546e9757ae3 : SUCCESS in 7m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ff949dbac1b64eb08eadd7e45abd633e : FAILURE in 3m 32s","accounts_in_message":[],"_revision_number":8},{"id":"2841bedec3167c50eea719b5fad42eef7529a2f1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 18:22:46.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05861d0935c64a9e9e66fb4b8f40a043\n\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/6af567b1ec574c79a5d4fe529eb3dc90 : SUCCESS in 45m 07s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/b20444bbef134f92ab14cc99855ed5a3 : SUCCESS in 52m 03s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/94cc647addb4439ca37c466c09701765 : SUCCESS in 47m 50s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/d25df80bd0fc412bb9099ec702b9da6a : FAILURE in 1h 27m 23s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/aafecde6f33a49398c824663ad69ea57 : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3af4700ee2a24aadbdadd97925fb4fbc : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/135cbc76e0f74940aaac4cc7e1809804 : SUCCESS in 11m 06s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"79349efb4c1c8cc6dc4529e569696d8c22e97de3","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-09-10 08:57:28.000000000","message":"Patch Set 8: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"3c8eb4a3e8a0ceb3e6e51d3bd1a8df4f94ce147b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 11:15:55.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4f72e5906a1847d092ad289676fd93fe\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/ff3a69fa77e540b1b4067239185e1e2a : SUCCESS in 31m 29s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/139a76bfe1c34b5fa583bd67efe780e6 : SUCCESS in 32m 25s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/a1f0d6985d8c42a7a6b27528197d9d60 : SUCCESS in 30m 04s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/72aba574461849789b8027180182ee0c : SUCCESS in 1h 06m 04s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1855d15d72c0422c85eb5fd05ea5d7b9 : SUCCESS in 33m 14s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0a10d630fd514507b004d388cf5758ff : SUCCESS in 22m 01s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/980af9c385e14f978e656c71543f99be : SUCCESS in 1h 22m 03s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/39151cbcbae4453fbdda31875beb5020 : SUCCESS in 2h 02m 35s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/85ff5750bd1244ac99d36a4abe42a9fc : SUCCESS in 1h 07m 38s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/2acf39568f52425a93589ca66b273514 : SUCCESS in 1h 31m 33s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4bb2764672b148c1a5eea0027190cedc : SUCCESS in 1h 23m 14s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/4c152b1e030a4c2dbe530b92bb0ce208 : SUCCESS in 1h 59m 35s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/f083d4f3ba27428393a9a01510245a3e : SUCCESS in 1h 47m 49s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/6e669fa87f734d3fa320c73fe0c642b5 : SUCCESS in 1h 37m 06s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/cc5c66af22484e50a23fb7f842a4fcd2 : SUCCESS in 2h 13m 36s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/8bc09b081c554ca4a56fb6437968ccd7 : FAILURE in 1h 24m 32s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/82971c0e7fe34c6b850beb7f82d07ffd : FAILURE in 1h 12m 26s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/dd923e2e409a4ebb84e286b9f3f525e9 : SUCCESS in 1h 25m 42s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/7f81b174b61646da9be25a10ba4a7fac : SUCCESS in 1h 19m 40s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/106df524672447be9c324c0aa957adf4 : SUCCESS in 1h 34m 28s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/010b0e7f7b974538b72baea684d240df : SUCCESS in 57m 48s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/67d31f87d6c44aa5935858b10268e7f0 : SUCCESS in 29m 25s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/723b6af480a04e60809febd34a86bf37 : SUCCESS in 32m 41s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/f51d08c2388f40428a189754bad3533a : FAILURE in 51m 43s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/a266f1bfef3a4b06bc1e1aabccbc36d3 : SUCCESS in 1h 31m 46s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/84bfe0f9c12e4cb7b65377d85aaccfc7 : SUCCESS in 1h 22m 00s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/04c10cf3ce2246c8a51c5107fce9ab5a : SUCCESS in 1h 10m 13s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/2ff8c1fa947b4c9bb3d089c077bbc0c9 : SUCCESS in 1h 20m 03s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/0c683ec6ab8c46589f07e95674a5b11a : SUCCESS in 1h 57m 02s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/731a75ee4b0b4414912af376be46b816 : SUCCESS in 1h 39m 53s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/8b2e5fbe7c5345fdb66c89bfb4402873 : SUCCESS in 1h 50m 24s (non-voting)\n- kolla-ansible-debian-trixie-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/97e9e85034a64f669714c0e4561f7de7 : SUCCESS in 25m 22s\n- kolla-ansible-debian-trixie-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/b8e0f3c4e00845adadb42d53818865e7 : SUCCESS in 56m 29s\n- kolla-ansible-rocky-10-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/d826aec26c5242c9b4992ec6604fd088 : SUCCESS in 33m 10s\n- kolla-ansible-rocky-10-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/8de10185135a45fc94520c8ebce38a6a : SUCCESS in 51m 26s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/0fcf20f597594eadb0a7ac3b11de9f9c : SUCCESS in 47m 27s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/7b3456a04ec34026b18143bb0bfb3acb : SUCCESS in 32m 57s\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/bec49481d8a74ebb977edd1988280bc8 : FAILURE in 1h 11m 34s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/a33100214a904a13a0876364f9a2e020 : SUCCESS in 52m 20s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/3f5c142825794ff69c53ba5618036c88 : SUCCESS in 44m 59s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/48bc286766e84073b73506d0dd14e8a5 : SUCCESS in 3m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c630725e43f043ba9a5098f481887873 : SUCCESS in 5m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dd3914cad7134384984c9334fea25a95 : SUCCESS in 2m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9db4c77fb4c34b058ccc1f55256f0993 : SUCCESS in 3m 26s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/696580e760d54ef281922fba55eca921 : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4d6a7fd433e24a988836326d3cb3ddf0 : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61db79dd3a374e6f84726ffdfe5be51b : SUCCESS in 3m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f16e2a568b834a699b74ed2ce41bab04 : FAILURE in 3m 40s","accounts_in_message":[],"_revision_number":8},{"id":"7ad0b8f5daafcde50bc103fcf35cb57b72112163","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-09-10 11:20:18.000000000","message":"Patch Set 9: Published edit on patch set 8.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"34b880f20289b4860528976a52d2097928e3aa50","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-09-10 11:20:49.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"c715d71f840aef04bc732a06a551620111c0ecc6","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-09-10 11:59:33.000000000","message":"Patch Set 9: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"fa0c3c6a2b90953761790f47581ccbd490d13cd7","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-09-10 12:21:32.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"4a3bb7db21a24aaf3e09cf149cdac1c4f828bacc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 13:02:57.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c70f9679571e432eaf945983161c536b\n\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/7e8e4c2e46ec45e4a995f88d10d22231 : SUCCESS in 40m 18s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/f26765aa08ef47d3b880b553f07c6388 : SUCCESS in 47m 49s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/67014519e1e549fbb8d17c4e726a0bc7 : SUCCESS in 43m 02s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/e91ee28998ae401d85fce3fc8ec2c5bc : FAILURE in 59m 09s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3aee6bd4029e4367a21e73e385310cb8 : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ffc38f5f3e984de987080fb587d64b1d : SUCCESS in 4m 17s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/19f4b9cde97f47eea16981c6d1f9f6ac : SUCCESS in 6m 46s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"8c5c3aaaeae4ceece250eb2d720d1109d89fede3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 13:43:05.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bced9eaa112c48c1b32afe1f9796be4a\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/feec10d22afe45f1abdc0a338f7978b0 : SUCCESS in 26m 00s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/9fd99486f08244cea357fb6a7821d081 : SUCCESS in 24m 11s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/8282cf0a169b401a864257bb6829e90d : SUCCESS in 34m 39s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/405e502543774e2daef5ec868b10d074 : SUCCESS in 45m 13s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5a0ec001282245f19b4b5163b2c33a6e : SUCCESS in 20m 19s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d1e721a35a2f4a2c833b5a68d8e3fd24 : SUCCESS in 34m 51s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/0b862699073f4ec1b27f37e99ac25849 : SUCCESS in 1h 17m 15s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/85a41a0eb4d54a6396505cff7e9ad593 : SUCCESS in 2h 08m 57s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/2cbaac0ebafc405c965384ea6041ac78 : SUCCESS in 1h 03m 11s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/7b05bee70b024f7a9fc3dbcd32f45760 : SUCCESS in 1h 29m 57s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3e8fc1f08b4746408924b5ebc8db5d02 : SUCCESS in 1h 41m 17s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/5db17da40f104688b61d96df840bb8ae : SUCCESS in 1h 53m 56s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/a5eed2efab51498f8d06c6db145bf858 : FAILURE in 1h 31m 45s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/8ec3598ab4fa4ad0a2aec7d477a5cbf0 : SUCCESS in 1h 32m 39s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/78b7766c181e4e1c9da7da2404eb3741 : SUCCESS in 2h 19m 00s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/3407076c0f7a4923b63c62cecd1cac52 : FAILURE in 1h 20m 07s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/1c0ec2115d684959b5e5d64974c357fc : FAILURE in 1h 15m 52s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/f7d196b2642a49fb97f0fb7f77db0fd2 : SUCCESS in 1h 42m 54s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/0888be23d9a24e5ba0e1d5c152e5ba60 : SUCCESS in 1h 18m 25s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/cd13945317254f3fa68fcb4d6fc99496 : SUCCESS in 1h 38m 27s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/8162196143b44e408c9e6b19822215c5 : SUCCESS in 51m 35s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/06c6326a496041e2877b1f640e205845 : SUCCESS in 29m 08s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/698224e10aa643f58c866089564dbc31 : SUCCESS in 19m 57s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/aada06c2f22843ca9d5988c31ac21580 : FAILURE in 29m 44s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/6e0d18d8e58a48cbbb105e001f542bcd : SUCCESS in 50m 31s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/beba930848fb4503abd61bbef66a9d7b : SUCCESS in 1h 24m 38s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/2903712cd9f54c82971556f12a07913c : SUCCESS in 1h 20m 00s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/e2e017781d4d4833936c7ed0aad19df4 : SUCCESS in 1h 16m 20s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/cb7745fdcca64f49bd9cf820f491d295 : SUCCESS in 1h 57m 36s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/06c3a9d4344f4379ace85ada858d5a40 : SUCCESS in 1h 41m 55s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/821ba95166384d618f24bc2f4eb72982 : SUCCESS in 1h 56m 15s (non-voting)\n- kolla-ansible-debian-trixie-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/058e8bdb9c504528a2660ba927fff303 : SUCCESS in 44m 09s\n- kolla-ansible-debian-trixie-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/a40c4d8beeed4b7b8fb1429a0ddead3b : SUCCESS in 33m 06s\n- kolla-ansible-rocky-10-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/78303f7898154f61a30487dcaaab7b83 : SUCCESS in 39m 38s\n- kolla-ansible-rocky-10-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/9401305261f84fe3a3fe656ffd59b8e1 : SUCCESS in 36m 46s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/eaa187bcb6c44db9a14940fd6ef5bc88 : SUCCESS in 24m 57s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/b296752c1f424093a359ba837f2124dd : SUCCESS in 1h 01m 35s\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/7d1677f942594777995204605b38e280 : FAILURE in 58m 11s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/955aa6ad321f4e29a2d609891da868d6 : SUCCESS in 51m 30s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/938d9f7f88434c03a224b603df51377d : SUCCESS in 1h 34m 53s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/306ba2559b0f4d9cb66f5a9ee4a27a7b : SUCCESS in 2m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69c67945530b4849b88517c9a0780109 : SUCCESS in 5m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/35aca018781c41c0a28e89c1ed3c419c : SUCCESS in 3m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8217da4ca3034023ae2c9b74c8ca79ce : SUCCESS in 2m 41s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/69f5a89e98694336aa6536b9d110cb8d : SUCCESS in 5m 57s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fa3458c0b6944dceaa8cc4a40906b3ef : SUCCESS in 3m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62ee8e015fcb49d6a05f26d3e9831daa : SUCCESS in 5m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c6d07287016649b0b5dc60b24f69e8f5 : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":9},{"id":"c7fd515825bc92de49ea1a75e7fc04a0fb6ceadc","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-09-10 13:44:48.000000000","message":"Patch Set 9: Backport-Candidate-1 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"70db6c6276d1136de7457fdd15b1f3227679595f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 13:48:01.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"fd9fdc7b01e27fcf238b89c537d34f0cf7d6fede","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 15:46:19.000000000","message":"Patch Set 9: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8acece81cd3049de938f391acc8ab3e3\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/9eb79bfa08d0492b9aac834101b14c4b : SUCCESS in 34m 37s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/49cee74a5927467e86794282e55d025e : SUCCESS in 30m 50s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/106d2ad2bc8d4959bf52cc3c0e9bbdcc : SUCCESS in 31m 57s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/eca1b18331c4455e9d729d6cabfdc8c2 : SUCCESS in 44m 51s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2e1f61eca1964a4185da1d4a4cacb5a7 : SUCCESS in 29m 05s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/db96d8a1cf0f45b5988ff7e7613d6752 : SUCCESS in 35m 28s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/a6b7842b7e684b4da4704846845f55e8 : SUCCESS in 1h 15m 46s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/107d1ae23194487e9f3de8e3e1766500 : SUCCESS in 1h 56m 54s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/b14189c38106440c80711656511b8b16 : SUCCESS in 1h 11m 00s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/a6e51131697e4b5ba07a5820b14644f8 : SUCCESS in 1h 34m 02s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/abffb176854f4c86b0f36fadcf5c04c4 : FAILURE in 24m 57s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/47162e4081c0452f8aec0e2e19cd7f85 : SUCCESS in 1h 53m 34s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/14425562eec142fb903af31b47dfddc2 : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fded73074a6c4dbab16c561202e4434b : SUCCESS in 6m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/aba3fea535bb46268f7a9bebabe56158 : SUCCESS in 3m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ae30ca9e2944415f8f3df24f12ce559d : SUCCESS in 3m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/356494a6ab0c40aaa35862a72ee023d1 : SUCCESS in 7m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fc34489556384f099105084b0b6c87b7 : SUCCESS in 11m 19s","accounts_in_message":[],"_revision_number":9},{"id":"1904e17ec7728e13e2aeb987105c332176edd6d2","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-09-11 07:17:15.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"c6820b0748ab14eb25bb189d3887bf524588ba51","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 08:52:47.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1de903d4d31843a28331f4696d2dd8bf\n\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/0004584d58724b8c84beb5dffc9f998e : SUCCESS in 56m 35s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/d685d67a43da4c688f7f6a432a04bdd6 : SUCCESS in 56m 06s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/86b58732f987419fb57cea48f5046f27 : SUCCESS in 58m 02s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/f01fbc34e78a4ccfa83b23db798b0a6e : FAILURE in 1h 31m 26s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0136313107014d66aaaf9f3b41fa3912 : SUCCESS in 4m 40s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c57fd9a31b29454cb0f24006f0a2d087 : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/0b765a4c0b624c68a84d33dae6e2ebc6 : SUCCESS in 8m 38s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"60cabc67801589e6aed29c8b8da64cab50b56ad0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 10:25:42.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59b061f4d57f433c950859800a7409f3\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/0b4ce468bed648a98b8fe968a27369e7 : SUCCESS in 16m 15s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/adead655003f457a9f49bde9d07e17dd : SUCCESS in 17m 23s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/8901a4da7ca64ebc87937ac0d80b8b79 : SUCCESS in 20m 16s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/baf4be13d4964319b541ac72006beb3d : SUCCESS in 42m 13s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fb24657fcc054de993e2a5d8dd6e3626 : SUCCESS in 25m 26s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fef073cfe575483ab8b50f7658179fe8 : SUCCESS in 30m 49s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/ef51b286ec4242feade0a15dfff2ecc3 : SUCCESS in 1h 25m 34s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/23a970b3189a495bb42fff69d16e0102 : SUCCESS in 1h 46m 38s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/de2c4c28e7bb4ca19c79baaba517fe82 : SUCCESS in 1h 05m 09s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/ffaa5bffc08443859ed752bb6acc9106 : SUCCESS in 1h 30m 23s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2ff97e428e5e4b16a5fe8caa3abdc760 : SUCCESS in 1h 19m 01s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/9c419b78cc4d4979a1334b772c87a30a : SUCCESS in 1h 52m 30s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/e7a5cabf37074835b34c170206304b39 : SUCCESS in 1h 48m 23s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/3ed8b91e6f3946c8a64b4423ea47d7af : SUCCESS in 1h 33m 07s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/b7656c2a6fe0467bb91e1d017d86e934 : TIMED_OUT in 3h 03m 53s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/956c52d8d5c1469c89f31ecca0db169b : FAILURE in 1h 23m 14s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/8ad0a1ae646348bfbffd25021d55b1bf : FAILURE in 1h 27m 38s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/a9de90c3a1ca4c53b73ad9dd4d344a59 : SUCCESS in 1h 37m 36s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/434d91c43918445db20979d33b6ddc6c : SUCCESS in 1h 19m 33s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/1d5f337676944075833b55dbfd7504c7 : SUCCESS in 1h 55m 49s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/c734b8e0dcbd4f31855190e57a5738fc : SUCCESS in 46m 14s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/b4c4e5eb59dd4477aebdef77e5e55863 : SUCCESS in 29m 21s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/629e1c9bc24048efbf3d200d3887375a : SUCCESS in 53m 00s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/fbf7dda9c6c44febb1e98e3ba8e6097f : FAILURE in 56m 36s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/b88d1a9f71e6441d8e87176a4c680c3b : SUCCESS in 1h 09m 15s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/0ce88e1b91bc4137ae868617c725dd94 : SUCCESS in 1h 30m 02s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/ed3ea4db69c9491fb916fb6bf7821225 : SUCCESS in 1h 28m 34s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/43ec14b157ea419d98529a48b7f342cb : SUCCESS in 1h 36m 37s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/41919e0f7d7b4c8fba189782a4ea3893 : FAILURE in 17m 46s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/b1e9512efeb149ce8972685419c98b45 : SUCCESS in 1h 42m 51s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/5fdf34887e4b4b5f8bc1b469cfbd4880 : SUCCESS in 2h 03m 59s (non-voting)\n- kolla-ansible-debian-trixie-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/0f78500156d946cb9d6346267005541b : SUCCESS in 35m 00s\n- kolla-ansible-debian-trixie-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/a2d4305635d34e5286069e50e9368542 : SUCCESS in 49m 54s\n- kolla-ansible-rocky-10-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/3c036527f32e45438195be837abc742b : SUCCESS in 25m 23s\n- kolla-ansible-rocky-10-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/8320cc81e56d444ba8fdbdcafb7152aa : SUCCESS in 51m 10s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/ff6940f185a24124abc7397dcb2cbeea : SUCCESS in 58m 21s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/a18a53b33f99477199a4ad0c5ace9b8d : SUCCESS in 52m 35s\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/ed0d8cf434a94188a9eb8d443184069d : FAILURE in 53m 42s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/9eebd870deb648f28c45efebb6f91c38 : SUCCESS in 52m 26s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/6892507e0fa64d3c8c5e3bacf15dfde6 : SUCCESS in 1h 35m 27s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/70e1c6f886c340ae97d99ebebc8a5e22 : SUCCESS in 3m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84027062b0154e1b9fed7dfbb6c9d8b5 : SUCCESS in 5m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/151b8c5cf66348e5b6be82d3440fad41 : SUCCESS in 3m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/be85138fab3a4b43808cb864764d9026 : SUCCESS in 3m 16s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6d73e7098a854ceeac0bb68395a61db9 : SUCCESS in 5m 52s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/46a25622a58644b9b867f387d5d5b3f8 : SUCCESS in 2m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cb29487d6e4463a858118450273d3e8 : SUCCESS in 4m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f422a5ab922b408689d357cca9b1497a : SUCCESS in 10m 04s","accounts_in_message":[],"_revision_number":9},{"id":"a11258e3a93146bf1e9099cfaab095bdfe8d16e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 10:27:57.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"e26bfed7c397165dfba0b52896cb93a9324cd0b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 13:48:43.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8a0965feeb546808eb94b91593afd91\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/42c3179cbf9c44e69fae54b8136f96c2 : SUCCESS in 34m 56s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/53dde9f4919c4b65a6df6fd0f1f4bfc7 : SUCCESS in 30m 44s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/29eade980c534faa805e8291945d674e : SUCCESS in 21m 34s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/239e76b2c0354c24a3ed203b11e1c81a : SUCCESS in 1h 14m 11s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ad7bed04c60f4711ab8e9301a366750c : SUCCESS in 29m 15s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4385be10a2034bf6803eac79bee30b93 : SUCCESS in 34m 50s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/24f8c5a9b3f64e51a0d371b728067cc7 : SUCCESS in 1h 18m 35s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/b9e3e6f727a24a6ab74b684df2138206 : SUCCESS in 2h 08m 55s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/cc66fcec2c79491797038c0a9b2c689b : SUCCESS in 1h 06m 15s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/25e9746c4e5640f68daebbac54eb5718 : SUCCESS in 1h 30m 32s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7d3f745917e245c9bac0ce5f0504a030 : SUCCESS in 1h 15m 34s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/f574003a3d0e4955852d07542fd7d19d : SUCCESS in 2h 01m 11s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/e69d5eaedf604ed5bf800d16e893c002 : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1053dbdd85d74ffb846495e83b48474d : SUCCESS in 5m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1bc349c374a54609a47454bf8b2d9e4f : SUCCESS in 3m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4d551a6d0c5147c6a8a07bdb96d9ab13 : SUCCESS in 3m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/030be93ad9844c9891886b9736213711 : SUCCESS in 6m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a636a7877c654c649a718ecb02a845e7 : SUCCESS in 9m 31s","accounts_in_message":[],"_revision_number":9},{"id":"f3180d34ffe6a575005c01dbe4d0dd1302850e8b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 13:48:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"d1d35e0ce8bd40020bb0e6d3e9c7ad06363a87e5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 13:52:24.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3a11ee9488440ffb058b1464a881e08\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdf1e38e48414d6fb6907e6d89c7bc7e : SUCCESS in 51s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1a47a3d7114d459d828387582d56bae4 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"f1d7da742d058d893f6abf3c0a21efaf940187b4","revisions":{"e24c7dd67d453d16922dd2d888e616a29978952c":{"kind":"REWORK","_number":1,"created":"2026-07-20 09:17:33.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/44/997944/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/997944/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/997944/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/997944/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/997944/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/997944/1"}}},"commit":{"parents":[{"commit":"cac8a805451839ec05447aabf0258603614bfc78","subject":"Updated ansible-lint in lint-requirements.txt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cac8a805451839ec05447aabf0258603614bfc78"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-20 08:53:38.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-20 09:17:30.000000000","tz":120},"subject":"Drop support for CentOS Stream 10","message":"Drop support for CentOS Stream 10\n\nChange-Id: I5973943fe83bbb00cfc8623c0cd4d4b98b369315\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e24c7dd67d453d16922dd2d888e616a29978952c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e24c7dd67d453d16922dd2d888e616a29978952c"}]},"branch":"refs/heads/master"},"71b512e83953bd196c14702c2c45f9bc4443987c":{"kind":"REWORK","_number":2,"created":"2026-07-20 11:38:03.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/44/997944/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/997944/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/997944/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/997944/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/997944/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/997944/2"}}},"commit":{"parents":[{"commit":"cac8a805451839ec05447aabf0258603614bfc78","subject":"Updated ansible-lint in lint-requirements.txt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cac8a805451839ec05447aabf0258603614bfc78"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-20 08:53:38.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-20 11:38:03.000000000","tz":0},"subject":"Drop support for CentOS Stream 10","message":"Drop support for CentOS Stream 10\n\nChange-Id: I5973943fe83bbb00cfc8623c0cd4d4b98b369315\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/71b512e83953bd196c14702c2c45f9bc4443987c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/71b512e83953bd196c14702c2c45f9bc4443987c"}]},"branch":"refs/heads/master"},"2622dabfbf86e74d41dc47f5e51c7fdd5c96f191":{"kind":"REWORK","_number":3,"created":"2026-07-27 15:08:27.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/44/997944/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/997944/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/997944/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/997944/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/997944/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/997944/3"}}},"commit":{"parents":[{"commit":"5d77668da8167435e4ac860392994f19ec733a46","subject":"Merge \"Rework tar normalization to a shared function\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5d77668da8167435e4ac860392994f19ec733a46"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-20 08:53:38.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-27 15:08:27.000000000","tz":0},"subject":"Drop support for CentOS Stream 10","message":"Drop support for CentOS Stream 10\n\nChange-Id: I5973943fe83bbb00cfc8623c0cd4d4b98b369315\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2622dabfbf86e74d41dc47f5e51c7fdd5c96f191"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2622dabfbf86e74d41dc47f5e51c7fdd5c96f191"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"cac8a805451839ec05447aabf0258603614bfc78","ours":"71b512e83953bd196c14702c2c45f9bc4443987c","theirs":"5d77668da8167435e4ac860392994f19ec733a46","merge_strategy":"recursive","contains_conflicts":true}},"589be34b73ed6a2a07eb632d9a6ef820c9096b2c":{"kind":"REWORK","_number":4,"created":"2026-07-27 15:08:50.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/44/997944/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/997944/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/997944/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/997944/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/997944/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/997944/4"}}},"commit":{"parents":[{"commit":"5d77668da8167435e4ac860392994f19ec733a46","subject":"Merge \"Rework tar normalization to a shared function\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5d77668da8167435e4ac860392994f19ec733a46"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-20 08:53:38.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-27 15:08:50.000000000","tz":0},"subject":"Drop support for CentOS Stream 10","message":"Drop support for CentOS Stream 10\n\nChange-Id: I5973943fe83bbb00cfc8623c0cd4d4b98b369315\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/589be34b73ed6a2a07eb632d9a6ef820c9096b2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/589be34b73ed6a2a07eb632d9a6ef820c9096b2c"}]},"branch":"refs/heads/master"},"1509a39a382520572d46ec5b4a0a3a26f918dc18":{"kind":"REWORK","_number":5,"created":"2026-07-28 06:43:02.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/44/997944/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/997944/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/997944/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/997944/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/997944/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/997944/5"}}},"commit":{"parents":[{"commit":"5d77668da8167435e4ac860392994f19ec733a46","subject":"Merge \"Rework tar normalization to a shared function\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5d77668da8167435e4ac860392994f19ec733a46"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-20 08:53:38.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-28 06:42:55.000000000","tz":120},"subject":"Drop support for CentOS Stream 10","message":"Drop support for CentOS Stream 10\n\nChange-Id: I5973943fe83bbb00cfc8623c0cd4d4b98b369315\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1509a39a382520572d46ec5b4a0a3a26f918dc18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1509a39a382520572d46ec5b4a0a3a26f918dc18"}]},"branch":"refs/heads/master"},"65d2f443c12cb70059337025d492e98542de7b81":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-07-31 08:47:39.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/44/997944/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/997944/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/997944/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/997944/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/997944/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/997944/6"}}},"commit":{"parents":[{"commit":"495b6b8fe437a06f32575b7308192ffc4f76f7b9","subject":"Merge \"Move PIP_* env variables to ARG and config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/495b6b8fe437a06f32575b7308192ffc4f76f7b9"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-20 08:53:38.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-31 08:47:39.000000000","tz":0},"subject":"Drop support for CentOS Stream 10","message":"Drop support for CentOS Stream 10\n\nChange-Id: I5973943fe83bbb00cfc8623c0cd4d4b98b369315\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/65d2f443c12cb70059337025d492e98542de7b81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/65d2f443c12cb70059337025d492e98542de7b81"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"5d77668da8167435e4ac860392994f19ec733a46","ours":"1509a39a382520572d46ec5b4a0a3a26f918dc18","theirs":"495b6b8fe437a06f32575b7308192ffc4f76f7b9","merge_strategy":"recursive","contains_conflicts":false}},"3e55f21365156d67af32421dddee5e1d7d551acb":{"kind":"REWORK","_number":7,"created":"2026-09-09 14:07:10.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/44/997944/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/997944/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/997944/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/997944/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/997944/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/997944/7"}}},"commit":{"parents":[{"commit":"e30a8df63633177915b95b0c94a03abf6c0b86fc","subject":"Updated from check-sources","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e30a8df63633177915b95b0c94a03abf6c0b86fc"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-20 08:53:38.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-09-09 14:07:03.000000000","tz":120},"subject":"Drop support for CentOS Stream 10","message":"Drop support for CentOS Stream 10\n\nChange-Id: I5973943fe83bbb00cfc8623c0cd4d4b98b369315\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3e55f21365156d67af32421dddee5e1d7d551acb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3e55f21365156d67af32421dddee5e1d7d551acb"}]},"branch":"refs/heads/master"},"87792b780f5817b311c7edf783a70b978f6b6dbf":{"kind":"REWORK","_number":8,"created":"2026-09-09 14:29:55.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/44/997944/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/997944/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/997944/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/997944/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/997944/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/997944/8"}}},"commit":{"parents":[{"commit":"e30a8df63633177915b95b0c94a03abf6c0b86fc","subject":"Updated from check-sources","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e30a8df63633177915b95b0c94a03abf6c0b86fc"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-20 08:53:38.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-09-09 14:29:55.000000000","tz":0},"subject":"Drop support for CentOS Stream 10","message":"Drop support for CentOS Stream 10\n\nChange-Id: I5973943fe83bbb00cfc8623c0cd4d4b98b369315\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/87792b780f5817b311c7edf783a70b978f6b6dbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/87792b780f5817b311c7edf783a70b978f6b6dbf"}]},"branch":"refs/heads/master"},"f1d7da742d058d893f6abf3c0a21efaf940187b4":{"kind":"REWORK","_number":9,"created":"2026-09-10 11:20:18.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/44/997944/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/997944/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/997944/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/997944/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/997944/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/997944/9"}}},"commit":{"parents":[{"commit":"e30a8df63633177915b95b0c94a03abf6c0b86fc","subject":"Updated from check-sources","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e30a8df63633177915b95b0c94a03abf6c0b86fc"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-20 08:53:38.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-09-10 11:20:18.000000000","tz":0},"subject":"Drop support for CentOS Stream 10","message":"Drop support for CentOS Stream 10\n\nChange-Id: I5973943fe83bbb00cfc8623c0cd4d4b98b369315\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f1d7da742d058d893f6abf3c0a21efaf940187b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f1d7da742d058d893f6abf3c0a21efaf940187b4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-1"],"failing_atoms":["label:Backport-Candidate\u003d1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
