)]}'
{"id":"openstack%2Fkolla~998119","triplet_id":"openstack%2Fkolla~master~Icffea19cf9c1279dd7e133c5588efd5818249b75","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-07-21 13:31:03.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}},"hashtags":[],"change_id":"Icffea19cf9c1279dd7e133c5588efd5818249b75","subject":"neutron: Move tap-as-a-service to be a plugin","status":"MERGED","created":"2026-07-21 10:54:09.000000000","updated":"2026-07-23 09:17:58.000000000","submitted":"2026-07-21 15:10:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"998119","meta_rev_id":"c26589670f20f2f97399d73a2e6bc4172d9e6d27","_number":998119,"virtual_id_number":998119,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-21 15:10:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-07-21 12:04:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2026-07-23 09:17:58.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2026-07-23 08:34:45.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":2,"date":"2026-07-21 12:03:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2026-07-21 13:31:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":-1,"date":"2026-07-21 12:04:10.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2026-07-21 11:53:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-21 12:03:22.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2026-07-21 12:04:10.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2026-07-21 13:29:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-23 08:34:45.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2026-07-23 09:17:58.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"c8cb76f33ca44f37a09a699468d7a69134c70e6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-07-21 10:54:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"22ee95ad717a80a1877d0a1cef01f8ceb3b4958a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-07-21 11:00:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7de57de29c54fb484cda498b7923ca405aacc8b6","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-07-21 11:53:04.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"48718fa1b70a6fea87200966317e91ea3e37ada7","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-07-21 12:03:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b9bb46db2d13292cb9bcaaef12b50744ec4f9a67","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-07-21 12:04:10.000000000","message":"Patch Set 1: Backport-Candidate-1 Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8a61c850465127752ab4e39ba7566d8b02fd64c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 13:29:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0c3261f224340dca349ebc46c24e34b\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/433cc8b46b67496eae90e0acc1a33502 : SUCCESS in 43m 53s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/b1c324f685ce4903948f511adf62d2ed : SUCCESS in 23m 50s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/1a9863829a234dbd8427b3597712413a : SUCCESS in 26m 16s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/39f8e0dbc9274058bd12e4476a5cee7f : SUCCESS in 25m 46s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/2093580345ad4be69cd0eed8fc9662db : SUCCESS in 36m 10s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/39f297817e5f4ab7b82fa0c7a7efe4c3 : SUCCESS in 25m 37s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/02cef43b7add4f51bf9481fe3539ddf9 : SUCCESS in 35m 26s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/6d5ea51935944e0abeacfa833dd58a87 : TIMED_OUT in 2h 33m 38s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/a8f00e61a5474b98b88a6bf12a4b1051 : SUCCESS in 1h 10m 00s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/85a99dd357114e2ba25bf897d706d5ba : SUCCESS in 1h 35m 11s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/a4b5d35998c7439eab6e5cda45e3b141 : SUCCESS in 1h 11m 51s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/e638a0d214694c14a641b81867d3b189 : SUCCESS in 1h 32m 04s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/31e48c67c4b8469db3b885d5f09819a1 : SUCCESS in 1h 06m 56s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/2f4e7a7cc1db44269575e843a1f34e54 : SUCCESS in 1h 35m 04s\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/8faf7bb301104431897539d08c9a4d59 : SUCCESS in 1h 04m 06s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/294996fbe67f4128ad64369ca76e44cd : SUCCESS in 1h 09m 21s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/8d5abd054f7b4b559e49599c0caec028 : SUCCESS in 1h 48m 11s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/2f0f24489a974a709cbc24895e85298f : SUCCESS in 1h 37m 43s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/70bd7176f1ab4264b01845d6aacdcc3a : SUCCESS in 1h 37m 13s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/8d7ba5355456441ab40129906a4cead7 : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5eac809f49244e0b9aca9d897d04b524 : SUCCESS in 6m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/271552d31d8e493bbd00f7f0ee06fda3 : SUCCESS in 3m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d11893e0c02246909da03102c6c54000 : SUCCESS in 4m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a77354ad32f24e57b8510337e8c70a36 : SUCCESS in 6m 44s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/84f9e58b775144899c6830f93ccf9dd9 : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42f4450c27f444d0800bbc4e704f5737 : SUCCESS in 6m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/faa9d8a625b745c489383b40df96ba9f : SUCCESS in 10m 13s","accounts_in_message":[],"_revision_number":1},{"id":"9ff1dd0f36a92fc77489cadef3d7f057e3c00b73","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 13:29:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2766de6a6594702bff4f9dcc88e1a39\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/f2a4e94c1b1f439fa5679e00008a928e : SUCCESS in 36m 58s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/f8ae37c8c7ea42298c81329a75818385 : SUCCESS in 38m 24s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/6a5c607e517b48e9b6444383470fe95c : SUCCESS in 38m 19s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/3a24804403584a3d899a80c029d38203 : SUCCESS in 38m 19s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/069c26395b244d4d9a29189c4f06315f : FAILURE in 1h 26m 41s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/93524e0cc735483693743e01672972ed : FAILURE in 1h 32m 14s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ac212ca1ed3e4f62970b4f5fbce299e3 : SUCCESS in 4m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/318573b1817343538b4370a22dc32420 : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/c35808a6eda74e55bc48730a3a674910 : SUCCESS in 5m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5943fb45ac65bee9b67650141bfc62d9ee6e5cce","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-07-21 13:31:03.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"885e67e83326441aa409b99b641771a9054fdbe5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 13:31:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1a3173dcc673878649c69d1a741992f91caebd7e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 15:10:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3cb309fb6a7c4c6b87c705a57977bead\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/5062921680d641029ee4bbcb0e244e2e : SUCCESS in 27m 57s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/d9886ce690064671b6c212d27a3a343e : SUCCESS in 12m 55s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/350fdb2fda15443580f8ad3d37642115 : SUCCESS in 25m 46s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/5540a8a7b33348e0805a375b858f9462 : SUCCESS in 37m 41s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ac7e7d14b7534f5da5e2a18325e6cd01 : SUCCESS in 15m 15s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6dc5448321f84e9db81cc8bc911f824f : SUCCESS in 26m 00s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/fdca1974a0934d9cb4b5b78cf5dfaac6 : SUCCESS in 1h 08m 23s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/d6cfa87742094b4387dbf01f5bf6d6cb : SUCCESS in 1h 37m 50s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/f932dbce09284bb5a6a78b3a859a7c3b : SUCCESS in 1h 11m 29s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/5367b9efc2c34db1879137edecb4d9fc : SUCCESS in 1h 38m 12s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/750944af8a964a15853943d838573756 : SUCCESS in 1h 04m 49s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/bb952d9db161477493eb9f0813a20b4f : SUCCESS in 1h 33m 58s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/afe326e71c9e43adb66426996ca52d29 : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c9df5afc01f43f4a92a87b0a0c51b78 : SUCCESS in 4m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5311985643ca4325b58688502a5f1157 : SUCCESS in 2m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dcd22044080e484584bae33e790bb9c3 : SUCCESS in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f070b87d22c94a56b754e308376d4521 : SUCCESS in 6m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bf0da65c32104b03b73b712c1bbdba33 : SUCCESS in 8m 32s","accounts_in_message":[],"_revision_number":1},{"id":"53702aad598123449e6fbfdd3e1c5fb3abd15360","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 15:10:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0ed0f9b4f89adca11d91c83d7d853c6726041496","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 15:11:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f61bffa17f7f41bbbe5f58f2ab4deff5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f23e336737f342a89b0550022b37f8d3 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1},{"id":"c89dfb6b4d3d6cad244a4ba3b02402347bd00779","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-07-23 08:34:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c26589670f20f2f97399d73a2e6bc4172d9e6d27","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-07-23 09:17:58.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f8929a9e11081fa5e4738285d47a0558ac3e1e08","revisions":{"f8929a9e11081fa5e4738285d47a0558ac3e1e08":{"kind":"REWORK","_number":1,"created":"2026-07-21 10:54:09.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/19/998119/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/19/998119/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/998119/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/998119/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/998119/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/19/998119/1"}}},"commit":{"parents":[{"commit":"ad9728d3ea90d454ab283b40f37469401ba785ce","subject":"copr: Debug periodic job by removing no_log","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ad9728d3ea90d454ab283b40f37469401ba785ce"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-21 10:53:23.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-21 10:54:07.000000000","tz":120},"subject":"neutron: Move tap-as-a-service to be a plugin","message":"neutron: Move tap-as-a-service to be a plugin\n\nNeutron has changed policy definitions and it\u0027s fixed in\nunreleased version of tap-as-a-service - and we\u0027re installing\nthat package from pypi instead of from tarballs.\n\nChange-Id: Icffea19cf9c1279dd7e133c5588efd5818249b75\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f8929a9e11081fa5e4738285d47a0558ac3e1e08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f8929a9e11081fa5e4738285d47a0558ac3e1e08"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-1"],"failing_atoms":["label:Backport-Candidate\u003d1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
